)]}'
{"id":"openstack%2Fmanila~876270","triplet_id":"openstack%2Fmanila~master~I3446464ff16a53a0cb08374ec51929cc498b0ee7","project":"openstack/manila","branch":"master","topic":"bug/2007719","attention_set":{},"removed_from_attention_set":{"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2024-07-11 20:42:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3446464ff16a53a0cb08374ec51929cc498b0ee7","subject":"share transfer DB API to use context.session","status":"ABANDONED","created":"2023-03-03 08:16:26.000000000","updated":"2024-07-11 20:42:35.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1d31e5487bcb8774a42f19539322f39653d8288c","_number":876270,"virtual_id_number":876270,"owner":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-06 02:39:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-03 09:16:54.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-03 09:29:02.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-03-03 09:31:26.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-03-03 17:03:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"34f4e6a8a330c9b39d6fb0fa82a19d1231479106","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-03-03 08:16:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35563c159717131e6982081523809223f8294ab4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-03-03 08:17:10.000000000","message":"Topic set to bug/2007719","accounts_in_message":[],"_revision_number":1},{"id":"4c401646f79197ac5c5fdfab43b1b02b76bd55b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-03-03 08:25:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c8e6fac36d54059be9cd0cdbce23c08e91d908c8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-03 09:16:54.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-70-876270-1/ns5_manila : FAILURE in 1h 0s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"942a48afeae9d7039ba7162ee2d8d4370561ca31","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-03 09:29:02.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/70/876270/2/check/DellEMC_Unity_Manila_dhss_true/5f99735/DellEMC_Unity_Manila_dhss_true/1348 : FAILURE in 1h 03m 32s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/70/876270/2/check/DellEMC_Unity_Manila_dhss_false/394c65b/DellEMC_Unity_Manila_dhss_false/1449 : SUCCESS in 1h 02m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"d1d34bb3d017da61349933740093623d2c95c150","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-03 09:31:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/70/876270/2/thirdparty-check/pure-devstack-manila-tempest-aio/ad08458/ : FAILURE in 1h 05m 18s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/70/876270/2/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/574fdde/ : SUCCESS in 1h 04m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6dcb1da1f743b8757ec38b7df9ee85c6af68f09a","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-03-03 09:37:31.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-70-876270-2/ns5_manila : FAILURE in 58m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"179f864760c434ff7b3a0a733f7bb27f4ef63a36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 17:03:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a031d801960443fba30d613d07ef261\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c14bc732f0c461d946d63ec1d51cee1 : SUCCESS in 10m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f485a904c7c941cda93fb5376df8f0b8 : SUCCESS in 21m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9f3359023cc4d52bc780606f2db20b3 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/591b141c5f6347e69b2b53c478ee46ad : SUCCESS in 10m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79ea1151eb084facae163d7273d8b53b : SUCCESS in 10m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da7504ab05804210a277f2df85ba01f6 : SUCCESS in 5m 33s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/279d5de150f248c79cdbd08d1dcb72f5 : FAILURE in 1h 15m 39s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4dc8768da1d2473497dae64e42e53ff6 : SUCCESS in 2h 16m 11s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3b24b6fb1f234954875f8c6e373f1663 : SUCCESS in 1h 20m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a6157e667b2c46d581f0b97343c0bd31 : SUCCESS in 1h 21m 28s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/fd736e51e51c4204acca533f16f326e7 : FAILURE in 2h 08m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/040bb149bc1a483294f652506119fe11 : SUCCESS in 1h 01m 03s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/57b27580ed454dde9633bea3585e5192 : SUCCESS in 5m 30s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a77ad239977d4251918c3eda10951fd3 : SUCCESS in 6m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d9be7bf5192049599bd75c81509f4344 : SUCCESS in 24m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f24c603a014c4d77ae7b014350feae53 : SUCCESS in 28m 23s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/37db4621bd054766bad88bb135c004eb : SUCCESS in 57m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/2686252d813d49448c02adddb1e802a5 : SUCCESS in 39m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a07ba79b98e64f83b6d2d5c2cc3b8aaf : SUCCESS in 33m 52s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/5d027a5e78e744fba638554c26523751 : SUCCESS in 1h 41m 02s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/93f967068cf94c94a8a80b5a561f37aa : SUCCESS in 1h 22m 23s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/6e31f1c10a104a1b99e5ec7b06eb8524 : SUCCESS in 1h 04m 36s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/dfbb2b6b663b4a9f9ef0edf7750a1fef : SUCCESS in 32m 52s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/09a82833763d4c578da0ddd119591f8b : SUCCESS in 46m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"20c0f934142e5fb15ae0dc6e52a0dcb0ea65ed06","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-03-06 01:00:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ec23af981fcb192757019f4eb1cfb4286f70651c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-06 02:11:59.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/70/876270/2/thirdparty-check/pure-devstack-manila-tempest-aio/e09ac3a/ : FAILURE in 1h 06m 24s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/70/876270/2/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/569d669/ : SUCCESS in 1h 10m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c4a749a04b9b8472be10b2e67dad1ab31347ce23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 02:39:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0c66183933248799447efa7813f2363\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f616a150e6d4a45aaed50d411e35add : SUCCESS in 8m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a9d4a671cd84698a1f64205a1364230 : SUCCESS in 15m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5b2d9e7133b4ed482514a858e20af83 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea630c00cd4a4713870f4dc66c668d7c : SUCCESS in 9m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd5ba034397e422d87bdf0e8a7228f9e : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd21d6fa60b34f28a1efc1b5552e4755 : SUCCESS in 4m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1836384120af4b72af58232d41c4fcfe : SUCCESS in 1h 28m 37s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/edb66541a5504528b66be365e1356145 : SUCCESS in 1h 36m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2008f208884e4640b6c32b4c5281772e : SUCCESS in 1h 20m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/48d127ef63404a9cb464168108c1eded : SUCCESS in 51m 44s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b94a86aa216c44beb241d8faab9a4450 : SUCCESS in 1h 16m 16s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2b42870df44645cfb027ed7079b0cfb6 : SUCCESS in 1h 12m 15s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e81e198c3ebf4c6da62a8a243d246ba3 : SUCCESS in 3m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/55339f2909354916821b0edc92cdd80b : SUCCESS in 4m 31s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e180981c38f742eb8859f6e4b7e54c9a : SUCCESS in 36m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/823d4c080f084fedb7d621eb9aca61d8 : SUCCESS in 29m 27s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/35cfe1f811e44b68849bdab319c1c93d : SUCCESS in 35m 40s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/2c8009d5a91947fd80fb39d6b4463d04 : SUCCESS in 38m 43s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/26366dc9251f405388ce27f3bbfd5b7e : TIMED_OUT in 1h 03m 10s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/8aca0de54b734630a806334c600ac17a : SUCCESS in 58m 07s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/15ba66ba3bed41318404017ca35e02fd : SUCCESS in 47m 31s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/e42f00872a394b809cabc084e3f44be1 : SUCCESS in 45m 24s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/5ec648e628b141ca92e7c0ec35eeaf83 : SUCCESS in 29m 47s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7ebd18b560d94bea8e7543dc4547e95c : SUCCESS in 48m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2e7cdc7120de2846b462090157356f3f5c87f75b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-03-06 04:03:07.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/70/876270/2/check/DellEMC_Unity_Manila_dhss_true/57b8504/DellEMC_Unity_Manila_dhss_true/1354 : FAILURE in 1h 11m 06s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/70/876270/2/check/DellEMC_Unity_Manila_dhss_false/02a9b83/DellEMC_Unity_Manila_dhss_false/1455 : FAILURE in 1h 23m 08s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"1d31e5487bcb8774a42f19539322f39653d8288c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-07-11 20:42:35.000000000","message":"Abandoned\n\nAlready addressed in: https://review.opendev.org/c/openstack/manila/+/885123","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"08a1ee41e90d3c23ef7e8729a7ac1b63bd4b32e0","revisions":{"20f8c47f7e4f07bfb5c10e00541c60babf2397dc":{"kind":"REWORK","_number":1,"created":"2023-03-03 08:16:26.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/70/876270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/70/876270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/876270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/876270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/876270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/70/876270/1"}}},"commit":{"parents":[{"commit":"f03e84ae6212ed80e3b1e56ff289cf13634e9a80","subject":"Merge \"[NetApp] Fix DHSS mode fails to create share on a new pool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f03e84ae6212ed80e3b1e56ff289cf13634e9a80"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-03-03 08:12:05.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-03-03 08:14:38.000000000","tz":480},"subject":"share transfer DB API to use context.session","message":"share transfer DB API to use context.session\n\nCloses-Bug: #2007719\n\nChange-Id: I3446464ff16a53a0cb08374ec51929cc498b0ee7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/20f8c47f7e4f07bfb5c10e00541c60babf2397dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/20f8c47f7e4f07bfb5c10e00541c60babf2397dc"}]},"branch":"refs/heads/master"},"08a1ee41e90d3c23ef7e8729a7ac1b63bd4b32e0":{"kind":"REWORK","_number":2,"created":"2023-03-03 08:25:14.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/70/876270/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/70/876270/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/876270/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/876270/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/70/876270/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/70/876270/2"}}},"commit":{"parents":[{"commit":"f03e84ae6212ed80e3b1e56ff289cf13634e9a80","subject":"Merge \"[NetApp] Fix DHSS mode fails to create share on a new pool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f03e84ae6212ed80e3b1e56ff289cf13634e9a80"}]}],"author":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-03-03 08:12:05.000000000","tz":480},"committer":{"name":"haixin","email":"haix09@chinatelecom.cn","date":"2023-03-03 08:24:44.000000000","tz":480},"subject":"share transfer DB API to use context.session","message":"share transfer DB API to use context.session\n\nCloses-Bug: #2007719\n\nChange-Id: I3446464ff16a53a0cb08374ec51929cc498b0ee7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/08a1ee41e90d3c23ef7e8729a7ac1b63bd4b32e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/08a1ee41e90d3c23ef7e8729a7ac1b63bd4b32e0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
