)]}'
{"id":"openstack%2Fmanila~877642","triplet_id":"openstack%2Fmanila~stable%2Fzed~Ia098bacc7d53f878d736a324847edd5c02e40631","project":"openstack/manila","branch":"stable/zed","topic":"remove-tripleo","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-03-18 06:18:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia098bacc7d53f878d736a324847edd5c02e40631","subject":"Remove TripleO job","status":"MERGED","created":"2023-03-17 04:43:24.000000000","updated":"2023-03-18 06:19:52.000000000","submitted":"2023-03-18 06:18:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877642-remove-tripleo","cherry_pick_of_change":877641,"cherry_pick_of_patch_set":1,"meta_rev_id":"c704f64048b95344820e49d96e9737f4ecff8de0","_number":877642,"virtual_id_number":877642,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-18 06:18:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2023-03-17 04:44:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2023-03-17 04:44:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-17 04:44:50.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-03-17 05:55:21.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-03-17 07:10:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e8c7c9e439681aabd4d8b02cff40a30069eda25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-17 04:43:24.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.1.","accounts_in_message":[],"_revision_number":1},{"id":"7aeb6b98c5aa1d3c595a4eb68f4721479806281f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-17 04:43:30.000000000","message":"Topic remove-tripleo-stable/zed removed","accounts_in_message":[],"_revision_number":1},{"id":"07f1e279b35a67af4ba6b25a4ce6bc3ade1c8f6f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-17 04:43:34.000000000","message":"Topic set to remove-tripleo","accounts_in_message":[],"_revision_number":1},{"id":"435a6f3e34839548e99badbbc35b3a595096432d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-03-17 04:44:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"66d1e4c2b53212594f551377963e2c82c45a423a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-17 05:55:21.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/42/877642/1/thirdparty-check/pure-devstack-manila-tempest-aio/406b072/ : SUCCESS in 1h 03m 43s\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/42/877642/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/12f651a/ : SUCCESS in 1h 10m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb6c27fd2e7e978db5f9bd2a68bb7f3f2c83da8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 07:10:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/085c35de31a54eff88a59e98b62c06d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a23ad5a896bc4404a339c700e3865da6 : SUCCESS in 13m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ccefd859c9a4cb4966393501a87fbba : FAILURE in 33m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b40327ce3604e86be8aa6d6dbe41c9f : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98b2d2c07459473985d53ae1e9261468 : FAILURE in 20m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e59504385054fc6b34e654bc363c307 : SUCCESS in 7m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b216b584b87d462abc699b4be2198588 : FAILURE in 20m 31s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9a8a2a838d2943d9ab3b07d083d718b4 : SUCCESS in 1h 27m 07s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7bfeaceb12504099a9cad194af0f5a4a : FAILURE in 2h 24m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e40bcb28af8c4fc08303e1d89b2dced2 : SUCCESS in 1h 15m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/79cdf743b33e4cffb0e4f635b952dbb5 : FAILURE in 1h 36m 14s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b30ac6ced9ef4abe9c332d92ee542a67 : FAILURE in 1h 38m 31s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f267cad46ae54070bdf9da6b866d5b7f : SUCCESS in 1h 09m 57s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/cdbfa28725d54c38916ef2503403fdb2 : SUCCESS in 7m 10s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e043211918024d74b4c0faf5a76e500e : SUCCESS in 4m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6fe3c4573ad44ed29d339062fb6dc01e : SUCCESS in 33m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3f19c40b43df41efb9ecd686b13c5f29 : SUCCESS in 34m 46s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d7b24b717d6d4cd8bb213b82c1373f23 : SUCCESS in 40m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/1cbf8e1436c548d6a3fe3bf033b64d35 : SUCCESS in 33m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f7e60e54661b4cedbc9ed1225db38bae : SUCCESS in 28m 52s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/1dfa1f2a46b04e4a93b8f72681fee7e8 : SUCCESS in 56m 52s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/73f25d38843c40cb9f135fe09c270c85 : SUCCESS in 1h 02m 36s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/51d5589b30664b429c65ddabec142cbc : FAILURE in 1h 03m 52s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/42bb035d19fb4f99aa1d51eecc55e0bf : SUCCESS in 30m 36s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/fd33cd7f58004d11bcd554f0e537123f : SUCCESS in 39m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d59281bbb2ba479821bb48882d436a93501bf90f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-03-18 02:37:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d187f0def9cb0ac75c8a0d8b49e64fbdfd62bb5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-18 05:24:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b97594410e294888b6c8fbf580fa434c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d71935420be94e16b929a9cc8effbb53 : SUCCESS in 10m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05303572006340e98989ea9854bb0df2 : SUCCESS in 25m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ea407b56a5b47e99d83f06148653363 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ddb2394ea4147b59e63942121a013d5 : SUCCESS in 10m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f92605a614b44e5bade57618d21b104 : SUCCESS in 9m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/baa3cdd3fd954462a5e0a06baab9e9b8 : SUCCESS in 9m 32s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/710f0f8c60c2486983ebec4df252d0cd : SUCCESS in 1h 13m 54s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5c6e40ddbe11421ca9e10a4c5015f70b : FAILURE in 2h 39m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8a97737d324c425ba41b6d378ceb05b3 : SUCCESS in 1h 21m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/282a38016c274c56a5bb8fb13548fb3b : SUCCESS in 1h 24m 53s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/2e84e03ca2e34189963b2db48b39f400 : FAILURE in 1h 57m 12s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7fca2551accb40c4a0836e40e3f26153 : SUCCESS in 1h 26m 26s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/66e30188688e4d7788c4218ab50564bf : SUCCESS in 5m 08s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/61d66cc083284ba38c85eb3fd9fcb00e : SUCCESS in 5m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d3f7cd753588458eb961624a0fad782c : SUCCESS in 30m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1198a445365b470d931cff9c6a4944cf : SUCCESS in 35m 56s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ad793f69c7fd4ca4b880d81ce17f6621 : SUCCESS in 55m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/33edc257efcf418c9a4df804ab1ab0d1 : SUCCESS in 43m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/39d7b6b072954199b5e1b269ee694d0b : FAILURE in 33m 48s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/e73d41c903324b6abdff743b551d8cd4 : SUCCESS in 1h 19m 10s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3e5e0bba444c45338ef6268f9968eecf : SUCCESS in 59m 10s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/1a1b6272379e48119eeeae3dec7cace5 : FAILURE in 1h 05m 46s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/89103c4e52914eb78eb3ce3759f2b62f : SUCCESS in 35m 27s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/51deb2bd83e343d68741cb0a3249d171 : SUCCESS in 38m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"98dc7104dc0a6c95a034521e2954506d47cbef2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-18 05:25:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"741b5e51223f400b88b5a97b7d8a69f78478b8ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-18 06:18:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30f989e3de6749418f6915477f1dc94c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a769bf41a6484098a1aca7462188e4a1 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b6a48569b98487ab119e0db66ffa597 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e5863801c274cac9ddb759e252af0cc : SUCCESS in 8m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec7be2c257454967b432688482469cc6 : SUCCESS in 9m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/caa9f6ca97834bee99bf526903a4f46f : SUCCESS in 51m 28s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ab94ead74a3243f08541563c9bc3336d : SUCCESS in 21m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1415236d71464d0abdca13dcfb60c39a : SUCCESS in 31m 52s","accounts_in_message":[],"_revision_number":1},{"id":"f2e04265ff93dd7cc52a0da38fc7738aed20120b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-18 06:18:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c704f64048b95344820e49d96e9737f4ecff8de0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-18 06:19:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae925bcde3b14a0d897daecb094758b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b6991ea60e841acb66d76338d31801a : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"512ea447e30f87f433677f0a379f791452b1788c","revisions":{"512ea447e30f87f433677f0a379f791452b1788c":{"kind":"REWORK","_number":1,"created":"2023-03-17 04:43:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/877642/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/42/877642/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/42/877642/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/42/877642/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/42/877642/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/42/877642/1"}}},"commit":{"parents":[{"commit":"19be7bb13ab988712b4aa57f0d9e73eecb2fb0af","subject":"update capacity filter during extend share","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/19be7bb13ab988712b4aa57f0d9e73eecb2fb0af"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-14 05:25:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-17 04:43:24.000000000","tz":0},"subject":"Remove TripleO job","message":"Remove TripleO job\n\nThis job has actually attracted no interest and has been kept\nexperimental. Now TripleO project is being deprecated so we should\ndrop this unused job.\n\nChange-Id: Ia098bacc7d53f878d736a324847edd5c02e40631\n(cherry picked from commit 729aad4b2b0ac3cbff9e143a858f8a59851d7d8f)\n(cherry picked from commit 72ae324ed3acf5b27b073bf3a2bf561a2c9f0b23)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/512ea447e30f87f433677f0a379f791452b1788c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/512ea447e30f87f433677f0a379f791452b1788c"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
