)]}'
{"id":"openstack%2Fmanila~877643","triplet_id":"openstack%2Fmanila~stable%2Fyoga~Ia098bacc7d53f878d736a324847edd5c02e40631","project":"openstack/manila","branch":"stable/yoga","topic":"remove-tripleo","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-04-06 20:05:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia098bacc7d53f878d736a324847edd5c02e40631","subject":"Remove TripleO job","status":"MERGED","created":"2023-03-17 04:51:56.000000000","updated":"2023-04-06 20:07:30.000000000","submitted":"2023-04-06 20:05:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877643-remove-tripleo","cherry_pick_of_change":877642,"cherry_pick_of_patch_set":1,"meta_rev_id":"da4f263079df959dbf1499ad71272d36405a8975","_number":877643,"virtual_id_number":877643,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-06 20:05:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-04-06 18:19:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-04-06 18:19:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-17 06:00:23.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-03-17 07:08:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-06 14:48:09.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2023-04-06 18:19:09.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"27889da823018887e532ccaa58d6d35b51eb2e77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-17 04:51:56.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"cf0b835b86dfa67ba146e7c7313ef6da9d0b65d5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-17 04:52:03.000000000","message":"Topic remove-tripleo-stable/yoga removed","accounts_in_message":[],"_revision_number":1},{"id":"3925b55218da28954007f2db61d1343a83ec3a0a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-17 04:52:07.000000000","message":"Topic set to remove-tripleo","accounts_in_message":[],"_revision_number":1},{"id":"407fb0212f6d14433dc5b346e5545b99354ae6de","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-03-17 06:00:23.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/43/877643/1/thirdparty-check/pure-devstack-manila-tempest-aio/a57800c/ : SUCCESS in 1h 05m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e647d301d1210223efc3c4623b442c3703d58faf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 07:08:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1bc1978318404a33bcea59aeec0f9f94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6148b763f7104c1799a9f2c8387ed6e4 : SUCCESS in 13m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9497ac7149c3435080326c378ece49c5 : SUCCESS in 31m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eb9110f9c9e444aae85cfee897fafe1 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7c43336ee024ec192edb3531da050e8 : SUCCESS in 17m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfc883d62e4d43e1bb1a1b35055d9015 : FAILURE in 18m 40s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/30be973170f942fda828cda9400a7e10 : FAILURE in 1h 22m 45s\n- manila-tempest-plugin-generic-scenario-stable https://zuul.opendev.org/t/openstack/build/778c257fe7944b768b92fd615b145c08 : SUCCESS in 2h 14m 30s (non-voting)\n- manila-tempest-plugin-generic-stable https://zuul.opendev.org/t/openstack/build/ccd91b739c5540aa92cd4f230ac5e8cc : SUCCESS in 1h 22m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native-stable https://zuul.opendev.org/t/openstack/build/b116735be6304fec83d16362dbbd5f3a : FAILURE in 1h 29m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-stable https://zuul.opendev.org/t/openstack/build/3185d64962924b49acb734c5cc7db172 : FAILURE in 1h 09m 28s (non-voting)\n- manila-tempest-plugin-zfsonlinux-stable https://zuul.opendev.org/t/openstack/build/2e5b7f8a0b1247519d7b9bcb5760d412 : SUCCESS in 1h 18m 23s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5e7a5ed6302b4124b5158007107bfdb2 : SUCCESS in 5m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/62f6010779d345589c699941494181f1 : SUCCESS in 5m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3ef2f601bfd24e808d43c18204e90304 : SUCCESS in 38m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/caea424c29a44bf29fe08aec4ef1d403 : SUCCESS in 40m 13s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/95109e2c814245db85da2d57212ddbb7 : SUCCESS in 52m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/83e65a21199b42a2b112ee7a606784d0 : SUCCESS in 45m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0fedcf763229464dbd59ed64e3d1652e : SUCCESS in 25m 58s (non-voting)\n- manila-tempest-plugin-lvm-fips-stable https://zuul.opendev.org/t/openstack/build/00979395ab21481ea66b283bd8890748 : SUCCESS in 1h 21m 48s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/9c3205ef7505467aacb71cdac383a343 : SUCCESS in 43m 56s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/87eb0e32092543dc9572b79ce2a14a5b : SUCCESS in 31m 33s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8f9f70a4e36941f5ae72bed8021db833 : SUCCESS in 41m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3edc99380a1b63e68b0e078b80b2db3142b158e","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-04-06 14:48:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e12c4b0463254fe02bcba3b354d345e33b59a9dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 16:45:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e959e6e422a49a2b07cf4d9c3282a42\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8aed6aa755343b880b969eca0f1b4dd : SUCCESS in 10m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38a949c2d7ce4a1d93f99127a1050f81 : SUCCESS in 16m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0540010edfb4fddad1d444bb8ef58c5 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17c6add20c1a40c6b9d10237fc1a80a0 : SUCCESS in 14m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a94653cfaa9b4635b05cc9245d4a6043 : SUCCESS in 11m 04s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/9222c840295048e5a8a414625629388f : SUCCESS in 1h 10m 51s\n- manila-tempest-plugin-generic-scenario-stable https://zuul.opendev.org/t/openstack/build/41c8adc9d2914d8ebad9949cde8a714f : FAILURE in 1h 56m 31s (non-voting)\n- manila-tempest-plugin-generic-stable https://zuul.opendev.org/t/openstack/build/3090da4410a84de88281723aff6c93dc : SUCCESS in 1h 47m 02s (non-voting)\n- manila-tempest-plugin-cephfs-native-stable https://zuul.opendev.org/t/openstack/build/d49b662e9684438995f11ae510b4da82 : SUCCESS in 1h 26m 18s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-stable https://zuul.opendev.org/t/openstack/build/5567899420394362b47944300ff51b3b : FAILURE in 1h 11m 45s (non-voting)\n- manila-tempest-plugin-zfsonlinux-stable https://zuul.opendev.org/t/openstack/build/d6648fdc75174f628bbbc5a4fb760273 : SUCCESS in 1h 15m 58s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3df49f50a4014d309d1448820d4b634a : SUCCESS in 4m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/343e881f55ff4d0a8cb77bec928c4165 : SUCCESS in 5m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/32631108f0eb4ec5bcf799110ad5f619 : SUCCESS in 33m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0241950684a441e680c5a3fc61fd9915 : SUCCESS in 42m 47s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/274ad557b73f4b4db00ba8b41795154d : SUCCESS in 1h 04m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/88d285ddf5a64d2789c05f48b5ecfe08 : SUCCESS in 38m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ed827d6e4e9c4c728072431158c5f2d6 : SUCCESS in 29m 43s (non-voting)\n- manila-tempest-plugin-lvm-fips-stable https://zuul.opendev.org/t/openstack/build/68a5da3c52bd41d0bfd7e59e17fcd68e : SUCCESS in 1h 06m 24s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/24d72492983f4f4a9d02f7dbc9c21c3f : SUCCESS in 59m 44s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/34952b9c4cb641f48d2093f42e436c78 : SUCCESS in 40m 18s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/34a8b87f162b421eac5970508657f398 : SUCCESS in 48m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"55f51b2095be4a7a4fc5f37f37b6098e8911bef7","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-04-06 18:19:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b5d0a2d534b054c6486cff329d190eda7047d02f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 18:19:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"50c985be88c60f4837f46663c8fb1bf12e7f8893","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 20:05:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20796b56498b4542a46f777fdce469ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffd00d17427c4219a6e60d2613db8d88 : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70baf681d337447a8623d35dff854b46 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72e479a6a7ec4a44a75453bc6b6594e7 : SUCCESS in 14m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8d47ce3ba3f4c93b3faa712287dc32b : SUCCESS in 15m 10s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/c074d20d806e4c61ba3734c11afe2ab5 : SUCCESS in 1h 45m 46s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/13cd3f8a5f5f4a01bcd12a9c29350717 : SUCCESS in 37m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/adf75cec2325468baa81ff780427a866 : SUCCESS in 38m 41s","accounts_in_message":[],"_revision_number":1},{"id":"682b9bfbcef0816d056a76928d3b5b1224f485af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 20:05:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"da4f263079df959dbf1499ad71272d36405a8975","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 20:07:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32a4cbd078e2429595429893107c195c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f76d7f4fa44784a486e792b8827950 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"74a06d77be53ed28765f4452c2a4380a69d5edb8","revisions":{"74a06d77be53ed28765f4452c2a4380a69d5edb8":{"kind":"REWORK","_number":1,"created":"2023-03-17 04:51:56.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/43/877643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/43/877643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/43/877643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/43/877643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/43/877643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/43/877643/1"}}},"commit":{"parents":[{"commit":"3d6fcc99458b07fa6f8e2983038a8c035b6dc98a","subject":"[Cephfs] Fix erroneous share mode override on ensure_shares","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3d6fcc99458b07fa6f8e2983038a8c035b6dc98a"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-14 05:25:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-17 04:51:56.000000000","tz":0},"subject":"Remove TripleO job","message":"Remove TripleO job\n\nThis job has actually attracted no interest and has been kept\nexperimental. Now TripleO project is being deprecated so we should\ndrop this unused job.\n\nChange-Id: Ia098bacc7d53f878d736a324847edd5c02e40631\n(cherry picked from commit 729aad4b2b0ac3cbff9e143a858f8a59851d7d8f)\n(cherry picked from commit 72ae324ed3acf5b27b073bf3a2bf561a2c9f0b23)\n(cherry picked from commit 512ea447e30f87f433677f0a379f791452b1788c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/74a06d77be53ed28765f4452c2a4380a69d5edb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/74a06d77be53ed28765f4452c2a4380a69d5edb8"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
