)]}'
{"id":"openstack%2Fmanila~879772","triplet_id":"openstack%2Fmanila~master~Ida4cc08fff5de340e37f9a675333c011e2fbdbd0","project":"openstack/manila","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-05-05 03:21:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ida4cc08fff5de340e37f9a675333c011e2fbdbd0","subject":"db: Migrate \"async operation\" APIs to enginefacade","status":"MERGED","created":"2023-04-06 17:03:42.000000000","updated":"2023-05-05 06:43:33.000000000","submitted":"2023-05-05 03:21:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879772-sqlalchemy-20","meta_rev_id":"9eeb97e8e85ca071639cf10ab86d9a097b042c10","_number":879772,"virtual_id_number":879772,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-05 03:21:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-05-04 21:17:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-04-27 23:35:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-05-04 21:17:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-06 18:53:16.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-06 19:22:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-06 19:22:35.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-06 19:25:35.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-04-06 20:13:47.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-04-14 21:05:58.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-04-15 07:57:02.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-04-27 23:35:37.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-05-04 21:17:41.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"6f3e96feed89e031d74f01cba730197906a967d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-06 17:03:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a6fa7fd352c897d3adec6570e84cfdeec5c2b640","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-04-06 18:53:16.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-72-879772-1/ns5_manila : SUCCESS in 1h 3m 52s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"ee43ea62f94b8422b770b8d2f30cb0e3fe2b65f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 19:22:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15b5a43d302242f49955560f10d9a9a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/157fd0c10e2e4e43a3b0357aa8375d72 : SUCCESS in 9m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cb686fc6a0e41379da0970ca7034027 : SUCCESS in 24m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1696df421b947c9b05243f2a716b39c : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3a3b7c28c2544279d22457ca2e22286 : SUCCESS in 11m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9993ca541d2d4d149ac03a0fc95decad : SUCCESS in 8m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1ca4a72a94a64671b687bfd2434580b1 : SUCCESS in 1h 22m 59s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b56929c7ef6b4567b10dfa3e2441ab35 : SUCCESS in 2h 10m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/abe6eb26474a4eed8cea46d332ab431e : SUCCESS in 1h 26m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native https://zuul.opendev.org/t/openstack/build/a22095ce631c4f1993f38556bca7f319 : SUCCESS in 1h 29m 20s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/4c6839f6053743539c5d5658a90c78b1 : FAILURE in 1h 40m 14s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/00c02fcc7e974fec855947b7c17b01a8 : SUCCESS in 1h 15m 29s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/fe52b59e40be44aba52ccedbe0be7996 : SUCCESS in 4m 17s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/430e15fff64c4d46b2db9e06723f8776 : SUCCESS in 4m 34s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/62b03c8fbf764f6c960998f489e55bc7 : SUCCESS in 37m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8ee4f5d30972434785ef76f57798944d : SUCCESS in 37m 50s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a4b7a4a4eb4f481287f595cba09a862c : FAILURE in 59m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/bcd06f5322fc4e58851236acc2ce4542 : SUCCESS in 37m 28s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d638fb0f94244428851090eaaf34d11c : SUCCESS in 26m 50s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/2c499e2e90b1404b9ab747d1262dfa1d : SUCCESS in 55m 23s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/bd7cbcddda294cb390ef054c11e59d94 : SUCCESS in 1h 00m 23s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/4e6fa4db0d314454b247d7cd0e2f6bf3 : FAILURE in 15m 00s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e00e17320b0e4549a93843ecc6d2ff40 : SUCCESS in 43m 55s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/63d2c11bebfa4b6caa629a7d38b15b51 : SUCCESS in 46m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04bcb73c8b383656eecd479990aba0c5a44bd4cc","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-04-06 19:22:35.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/72/879772/1/check/DellEMC_PowerMAX_Manila/fbf4ff4/DellEMC_PowerMAX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"e41cc52cf4699f58fd706a8800d3008258ed2963","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-04-06 19:25:35.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/72/879772/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/1ad4d84/ : SUCCESS in 1h 10m 48s","accounts_in_message":[],"_revision_number":1},{"id":"fe7d11428de47e5d3cecf9399365f80719c83f32","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-04-06 20:13:47.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-72-879772-1/NFS/ : FAILURE in 2h 39m 45s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"8617ce9fa2ad6b823b136fef8af771af6784b6b7","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-04-08 01:11:46.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-72-879772-1/NFS/ : SUCCESS in 3h 16m 37s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"70842cb46f8ac2629512b1c8f1b7bede16e4508d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-04-14 16:02:29.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"637b9a6dc119868ba20096dfb19c3906d4a6b03b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 18:14:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9517e4e814f740ac99ba227dd93b9200\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e72c507d6fcd408db20892924a3e04c8 : SUCCESS in 9m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d56b7a37d3ad4807a9dd4a49fedaa862 : SUCCESS in 21m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a98bd855a3b44cc9b9635b4f4a48b7c : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c4aa9057f824058bcb41bb24c746c12 : SUCCESS in 9m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36c3f498572146d99fbc28d2271337d3 : SUCCESS in 8m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a6eeddcbdfeb4fa19b8c6c747258763f : SUCCESS in 1h 40m 52s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/581a9ed4bf3e440ab72d3e9800a5cdd3 : FAILURE in 2h 05m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/63af13af56d441eda3ade6100b2f927c : SUCCESS in 1h 18m 57s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ff2155f42a94415e9cb88fff9810eb9e : SUCCESS in 1h 20m 15s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9a8470ae5b024319b0aa146f7bcf135f : FAILURE in 1h 36m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/58021a2293ef4668bc9c5a5fa95525b7 : SUCCESS in 1h 20m 48s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3bf45b565ee44629a890bd42a5ea8e71 : SUCCESS in 5m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ae1466a9010d4e05b2cb08deeab24925 : SUCCESS in 3m 59s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6a0ed70958a9456f964ac2949f7d2ced : SUCCESS in 31m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5fe5e13560ee468e8be09a7fce9d4247 : SUCCESS in 39m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ca15a5dd29be4b499e3af0147465c0e4 : SUCCESS in 54m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/1aa194017a764d0a94fdd38bea1b1e51 : SUCCESS in 34m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/779a9753f286420e96799224aa92f356 : FAILURE in 28m 02s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/ce141e3f216b433c9cf23ea565076e8c : SUCCESS in 1h 00m 22s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/bd89f6e41e7e4521b8bf3cb7a8c9408d : SUCCESS in 55m 04s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3f85b627a6b34e98a8bd0d0c2caf6f11 : SUCCESS in 1h 19m 50s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/758ec51b840248f4b77f77d51c665c75 : SUCCESS in 38m 13s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/098935f2862b42fc81bb3348455b030d : SUCCESS in 41m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ec0ed386c5d3ad4c8b006833cd98037586b42519","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-04-14 18:14:42.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/72/879772/2/check/DellEMC_PowerMAX_Manila/0c36671/DellEMC_PowerMAX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"36a4917ada66a7d775847c8610b91451567205c8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-04-14 18:38:01.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-72-879772-2/ns5_manila : SUCCESS in 1h 4m 39s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"b7845d6894337f144b7a07b0383f913cbd686b5c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-04-14 18:41:03.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/72/879772/2/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/5f7ea8c/ : SUCCESS in 1h 13m 27s","accounts_in_message":[],"_revision_number":2},{"id":"ddfeb69de92fe0a7a00e6ec56515c675aad2edb8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-04-14 18:42:41.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-72-879772-2/NFS/ : SUCCESS in 1h 39m 48s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"f17dea95c2cd7cbaaa901517a4ea3651824bc337","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-04-14 21:05:58.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/72/879772/2/upstream-check/manila-tempest-plugin-ontap-dhss/0465f92/ : SUCCESS in 1h 24m 23s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/72/879772/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/d2fdd7b/ : SUCCESS in 1h 28m 29s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"792ee15ebbf8f30212fb04af75ebe938432691b5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-04-15 07:57:02.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/72/879772/2/check/DellEMC_Unity_Manila_dhss_true/821a45d/DellEMC_Unity_Manila_dhss_true/23 : SUCCESS in 3h 45m 35s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/72/879772/2/check/DellEMC_Unity_Manila_dhss_false/47b3165/DellEMC_Unity_Manila_dhss_false/22 : SUCCESS in 1h 38m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"b0bbbf4feacb1b901a16bb865bf9c67a37bc3d4c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-04-27 23:35:37.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2fb9393860c3ee18ab66657cb619b05814cd3209","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-05-04 21:17:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4f2a1d3b4be7fde41b54a716fcc4182b61a1b579","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 21:18:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9747c227da8d6e57998cde23957851609e484b2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 22:41:14.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/879771 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"5bcd4f4fc94cb5c642c06ed2134b4c5e47df2b02","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-04 22:54:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"04d18881f2594ad67fd44111e383cdde4ec23fdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 00:51:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0caeeceb5be749f5b7723c4ea9b027f9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31fb1eaa510d40b4b6cc9ad0db88a97c : SUCCESS in 12m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/229b9f2b27d24d32bc84f1edfe1312f2 : SUCCESS in 23m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9c4483bfb194dd9ae15bdc99f164b7c : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca3ce3c66f2d4221b0b41f2cb0089329 : SUCCESS in 11m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45b92e46d2b64b2f800b7c71c44d78e7 : SUCCESS in 11m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9a91bed1183d41069f512aed78ad79d9 : SUCCESS in 50m 44s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/adcc8daeea10433f8bfeefcf284a953f : SUCCESS in 1h 55m 03s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c66f5dc13644474184107f21a1b4f2c1 : SUCCESS in 1h 13m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/738ad85f35eb4e81afb0a3caceddd8d8 : FAILURE in 1h 02m 17s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6d37114b98c04f398f89975a70943cfe : FAILURE in 1h 11m 53s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cbc800f4ebde40d7877ac4cef7704726 : FAILURE in 1h 00m 55s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/75aeda379a6b4beeb1cda7feb1ecc009 : SUCCESS in 6m 43s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c5890f121f684aa39ff2bebf704262e3 : SUCCESS in 4m 52s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f811980009224f2a86bbc5276cefb010 : SUCCESS in 29m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/45a723851af04ab4b56d09a583d5ad13 : SUCCESS in 36m 39s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/76bec990235c4e0c9b77379a2488c500 : SUCCESS in 43m 47s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/7f8f04861c2549dcb65d3944cee76b52 : SUCCESS in 30m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d89b850876ef4c30b2bb9d6e93c64cab : SUCCESS in 23m 55s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/f3db5d68de6f4edb8ad0208c57581c89 : SUCCESS in 58m 03s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/ca1511ccb043467aa08eadec6c4e15aa : SUCCESS in 51m 36s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fa020a3ee1c8495d9a0b4c916d18e7dc : SUCCESS in 55m 48s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d674075ebcaf4ef08f6fa19e302e4660 : SUCCESS in 34m 22s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c3495ca6fec34de4875fa0f4e17af4e4 : SUCCESS in 37m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4d9dffa5edbdbeec5aa7f48a8c10c6f7b5b114a7","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-05-05 00:51:29.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/72/879772/2/check/DellEMC_PowerMAX_Manila/e1bfe77/DellEMC_PowerMAX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"9ffe283f19aac581ef014efc2c5a506ae702c46d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 01:46:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f4826664bad22dbec7ebf49b3f3a9aa9d7a9c281","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-05 02:37:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/72/879772/2/upstream-check/manila-tempest-plugin-ontap-dhss/27de779/ : SUCCESS in 1h 42m 51s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/72/879772/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/cff07c1/ : SUCCESS in 1h 45m 41s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"b9313691bf7a9ef97bf3a64f165d577897bc37d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 03:21:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27ef09d8dc9d4485ae9df560a8506f20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ac391c10cea4fa8990338d1dcede2d1 : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53c7b6121de8405cacb0d0e1cd4219c5 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ad97376dc81428d896d7a4e9cd2f8ba : SUCCESS in 9m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3883a01fdc8143e698754e5096ab2071 : SUCCESS in 9m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2f96868a9b6d412c952f51ae6571f3a8 : SUCCESS in 1h 01m 58s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a6064ce50d59430498848616b45d886d : SUCCESS in 27m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6734ae750b964bada2613a14dc8741ef : SUCCESS in 28m 08s","accounts_in_message":[],"_revision_number":2},{"id":"7fa545e3f9db801dba6765635763409f87ce1510","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 03:21:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9003d8ae683549e6df232af9186b234fd848e913","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-05 03:23:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bb1de73dadc45d59544f03b7d0ed9c7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c278297e5934d1fbfd02f86eba07935 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2},{"id":"9eeb97e8e85ca071639cf10ab86d9a097b042c10","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-05 06:43:33.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/72/879772/2/check/DellEMC_Unity_Manila_dhss_true/72ad916/DellEMC_Unity_Manila_dhss_true/78 : SUCCESS in 3h 30m 57s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/72/879772/2/check/DellEMC_Unity_Manila_dhss_false/9832a97/DellEMC_Unity_Manila_dhss_false/77 : SUCCESS in 1h 24m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cd3b7355afdbcb389b5b0c1c6f95eb05d137ec70","revisions":{"c1adfe80a82f373e4d3d6e451edb50e7521aa536":{"kind":"REWORK","_number":1,"created":"2023-04-06 17:03:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/72/879772/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/72/879772/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/879772/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/879772/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/879772/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/72/879772/1"}}},"commit":{"parents":[{"commit":"e761253e335460c190d95ff3295640fdddb2a48a","subject":"db: Migrate remaining quota APIs to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e761253e335460c190d95ff3295640fdddb2a48a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-06 15:37:48.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-06 16:34:06.000000000","tz":60},"subject":"db: Migrate \"async operation\" APIs to enginefacade","message":"db: Migrate \"async operation\" APIs to enginefacade\n\nThis one is easy. All nice and self-contained.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ida4cc08fff5de340e37f9a675333c011e2fbdbd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c1adfe80a82f373e4d3d6e451edb50e7521aa536"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c1adfe80a82f373e4d3d6e451edb50e7521aa536"}]},"branch":"refs/heads/master"},"cd3b7355afdbcb389b5b0c1c6f95eb05d137ec70":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-04-14 16:02:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/72/879772/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/72/879772/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/879772/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/879772/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/72/879772/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/72/879772/2"}}},"commit":{"parents":[{"commit":"a9fc22841c6ffe99e047f60a9a8dc6440919ed29","subject":"db: Migrate remaining quota APIs to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a9fc22841c6ffe99e047f60a9a8dc6440919ed29"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-06 15:37:48.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-04-14 15:59:36.000000000","tz":60},"subject":"db: Migrate \"async operation\" APIs to enginefacade","message":"db: Migrate \"async operation\" APIs to enginefacade\n\nThis one is easy. All nice and self-contained.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ida4cc08fff5de340e37f9a675333c011e2fbdbd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cd3b7355afdbcb389b5b0c1c6f95eb05d137ec70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cd3b7355afdbcb389b5b0c1c6f95eb05d137ec70"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
