)]}'
{"id":"openstack%2Fmanila~881313","triplet_id":"openstack%2Fmanila~stable%2Fwallaby~I94b2576d808967517733bdcbb81444290e31b518","project":"openstack/manila","branch":"stable/wallaby","topic":"bug/1993829","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-03-06 11:31:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I94b2576d808967517733bdcbb81444290e31b518","subject":"[NetApp] catch errors on getting vserver when reusing share server","status":"ABANDONED","created":"2023-04-23 04:04:50.000000000","updated":"2024-03-07 22:59:37.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":881312,"cherry_pick_of_patch_set":1,"meta_rev_id":"42cf3dcb20b6b5371efa0ac6ec833005bfa2d71c","_number":881313,"virtual_id_number":881313,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-04-23 06:13:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-23 04:04:50.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"CC"},{"updated":"2023-04-23 05:24:24.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-23 06:13:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-06 11:31:50.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"1bf9de7a15283945838f966f7f2362b0a2ef5d8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-04-23 04:04:50.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"8036e7c29fa6cb6b4cb1e833b2e6c89a37753c5a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-04-23 04:04:55.000000000","message":"Topic bug/1993829-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"6480516964a95d2e7f9c2410e463893dfdcff3f4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-04-23 04:04:57.000000000","message":"Topic set to bug/1993829","accounts_in_message":[],"_revision_number":1},{"id":"d7bf8f3ee9aa31b0a4b167c691662cadeb12f9ed","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-04-23 05:24:24.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-13-881313-1/ns5_manila : SUCCESS in 57m 2s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"18f5b80d32f53c21b9475c5f223a2bfcf211c1eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-23 06:13:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1eb4c9ac93904235b5ff695d64c49063\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e82a54c9c9443aab94eab312cb673a6 : SUCCESS in 9m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd21b70a9d30412abd180df408c91e8f : SUCCESS in 18m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d59df985237402aa868e99252f66529 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2565cceeaf724475aad0823fab369a27 : SUCCESS in 8m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82f13f95b8f84d0da7443b93139b9f03 : SUCCESS in 8m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9f659437f9740b1923054aa0658ce45 : SUCCESS in 8m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93589b672bbd4a358dec663c56851633 : SUCCESS in 6m 53s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/403854539d17473b9df4b08f7fccf67a : SUCCESS in 56m 14s\n- manila-tempest-plugin-generic-scenario-wallaby https://zuul.opendev.org/t/openstack/build/0e33313947a6406faa5bc63723cf6b79 : SUCCESS in 2h 03m 21s (non-voting)\n- manila-tempest-plugin-generic-wallaby https://zuul.opendev.org/t/openstack/build/72b9625babf6491fac124aa978736dc1 : SUCCESS in 1h 11m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native-wallaby https://zuul.opendev.org/t/openstack/build/7aee1579bb21497ea9dbbf816a384a5e : FAILURE in 35m 20s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-wallaby https://zuul.opendev.org/t/openstack/build/290ae745268947e2904b35e4f118d496 : FAILURE in 48m 38s (non-voting)\n- manila-tempest-plugin-zfsonlinux-wallaby https://zuul.opendev.org/t/openstack/build/78b2608f14f545f8b949d230c73efd03 : SUCCESS in 57m 12s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/655693fc49ae45a0bbfb2b1e2dd4433f : SUCCESS in 4m 50s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8999b50a802f4fe7a8b836ba4b4ef222 : SUCCESS in 4m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss-wallaby https://zuul.opendev.org/t/openstack/build/97249f855e704d6da3a95f026a3153b8 : SUCCESS in 33m 44s\n- manila-tempest-plugin-dummy-dhss-wallaby https://zuul.opendev.org/t/openstack/build/7aac9fa55841471d98697b43b259e59a : SUCCESS in 32m 18s\n- manila-tempest-plugin-container-wallaby https://zuul.opendev.org/t/openstack/build/cb8c0a22c1b54069b472adaeb3ccb909 : SUCCESS in 39m 10s (non-voting)\n- manila-tempest-plugin-glusterfs-native-wallaby https://zuul.opendev.org/t/openstack/build/5abdf4da21544f6a9dabe41353ebb528 : SUCCESS in 34m 51s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs-wallaby https://zuul.opendev.org/t/openstack/build/ea4f2979d8894eaeb49040c1e905771f : SUCCESS in 22m 32s (non-voting)\n- manila-tempest-plugin-lvm-fips-wallaby https://zuul.opendev.org/t/openstack/build/2bcba09750294185b97b29cbb8edbcac : FAILURE in 42m 43s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/1000803e1a264d6ea7550546510d0c7d : SUCCESS in 44m 26s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/532cb07c79c242b6a191f2bec35e0778 : SUCCESS in 32m 18s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/63be91dafa8e4650acaa326f1207d699 : SUCCESS in 27m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e39509485bfd37f4803ba91a6945c7e99a71018","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-06 11:31:50.000000000","message":"Abandoned\n\nstable/wallaby branch of openstack/manila is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/wallaby if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1},{"id":"42cf3dcb20b6b5371efa0ac6ec833005bfa2d71c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-03-07 22:59:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b89c482a5f01ef0c9443c3574651b459e82f57a6","revisions":{"b89c482a5f01ef0c9443c3574651b459e82f57a6":{"kind":"REWORK","_number":1,"created":"2023-04-23 04:04:50.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/13/881313/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/881313/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/881313/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/881313/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/881313/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/881313/1"}}},"commit":{"parents":[{"commit":"1274f07b336d597d6c07ffdf612a84da01d1754e","subject":"Merge \"Fix DriverFilter/GoodnessWeigher string evaluations\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1274f07b336d597d6c07ffdf612a84da01d1754e"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2022-10-21 18:04:44.000000000","tz":120},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-04-23 04:04:50.000000000","tz":0},"subject":"[NetApp] catch errors on getting vserver when reusing share server","message":"[NetApp] catch errors on getting vserver when reusing share server\n\nfixes e.g. \u0027Vserver name is absent in backend details\u0027\n\nCloses-Bug: #1993829\nChange-Id: I94b2576d808967517733bdcbb81444290e31b518\n(cherry picked from commit 0ac8d1cc89161c8173f90d2c5331193798c9f80e)\n(cherry picked from commit 2507e9d9e78003bd06f61a5253c20284581c63e8)\n(cherry picked from commit f9e70ac786aa4327fdd62e25022f9f326444e649)\n(cherry picked from commit ac4ee0b28f141deb3557ccf1522617f799b7b689)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b89c482a5f01ef0c9443c3574651b459e82f57a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b89c482a5f01ef0c9443c3574651b459e82f57a6"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
