)]}'
{"id":"openstack%2Fmanila~883175","triplet_id":"openstack%2Fmanila~master~Ib36c81c213a374a918378854ce0a89ce70acf1d0","project":"openstack/manila","branch":"master","topic":"bug/1922512","attention_set":{},"removed_from_attention_set":{"33648":{"account":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"last_update":"2023-07-27 19:53:45.000000000","reason":"\u003cGERRIT_ACCOUNT_33648\u003e replied on the change","reason_account":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"}},"35586":{"account":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"last_update":"2023-08-02 22:26:20.000000000","reason":"Change was submitted"},"33301":{"account":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"last_update":"2023-08-02 22:26:20.000000000","reason":"Change was submitted"},"33756":{"account":{"_account_id":33756,"name":"Luisa","email":"luisarfamaral@gmail.com","username":"rfluisa"},"last_update":"2023-08-02 22:26:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib36c81c213a374a918378854ce0a89ce70acf1d0","subject":"[NetApp] Fix NetApp driver create from snapshot","status":"MERGED","created":"2023-05-17 13:27:55.000000000","updated":"2023-08-02 22:27:28.000000000","submitted":"2023-08-02 22:26:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883175-bug/1922512","meta_rev_id":"9154381fb1b87026efd7d077319ccb2517d7e869","_number":883175,"virtual_id_number":883175,"owner":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-02 22:26:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"recommended":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"all":[{"value":1,"date":"2023-07-27 19:53:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"value":2,"date":"2023-08-01 22:02:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-07-27 05:37:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"value":1,"date":"2023-07-28 01:42:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2023-08-02 03:54:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":0,"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"value":1,"date":"2023-08-02 21:23:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33756,"name":"Luisa","email":"luisarfamaral@gmail.com","username":"rfluisa"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35147,"name":"Raffaela de Castro Cunha","email":"raffaelacunha@gmail.com","username":"raffaelacunha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-17 13:27:55.000000000","updated_by":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"reviewer":{"_account_id":35147,"name":"Raffaela de Castro Cunha","email":"raffaelacunha@gmail.com","username":"raffaelacunha"},"state":"CC"},{"updated":"2023-05-17 14:49:03.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-17 15:01:42.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-05-17 16:39:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-17 16:40:09.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-17 16:58:54.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-05-17 18:41:25.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-06-22 18:06:53.000000000","updated_by":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"reviewer":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"state":"REVIEWER"},{"updated":"2023-07-03 20:47:53.000000000","updated_by":{"_account_id":33756,"name":"Luisa","email":"luisarfamaral@gmail.com","username":"rfluisa"},"reviewer":{"_account_id":33756,"name":"Luisa","email":"luisarfamaral@gmail.com","username":"rfluisa"},"state":"CC"},{"updated":"2023-07-25 18:58:47.000000000","updated_by":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"reviewer":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"state":"CC"},{"updated":"2023-07-26 14:46:55.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-07-26 15:59:14.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-07-27 05:37:02.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-07-27 19:53:45.000000000","updated_by":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"reviewer":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"state":"REVIEWER"},{"updated":"2023-07-28 01:42:05.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-07-28 02:53:29.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-01 22:02:23.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2023-08-02 03:54:19.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"950a007b62ad9457811f0476baa5f8a5249787d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"date":"2023-05-17 13:27:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7957639cae91828a43e1c60763b9e29fcd87cebf","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-05-17 14:49:03.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-75-883175-1/ns5_manila : SUCCESS in 1h 20m 40s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"1d9e84a76522207f8debde1203dbd5941ec7afe8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-05-17 15:01:42.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-75-883175-1/NFS/ : SUCCESS in 1h 33m 26s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"f7a1918b73a47f0369e4eb4f6a7617fe2fc68a39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 16:39:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fe6cfe9bb0c4981b1f0de5b422ba9f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1849e23bd5e141f59b7a0f7979d07a85 : SUCCESS in 12m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d24682db5d9475db1a798be69cc4a3f : SUCCESS in 25m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5c8a092f2cf418c8c8c53c373f1c6aa : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb059a6586724b5db47372ad5b92292d : SUCCESS in 10m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d576e33b1c744a6bc40db7549aa5d73 : SUCCESS in 9m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea6030101a6b4980af7be5abf13f6390 : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bd46ab21ec44b2fbed86cc48a2a076a : SUCCESS in 8m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/009a95403a0e44f6815f131e77b2d099 : SUCCESS in 1h 11m 14s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0f2f7d7cacbd4337b5a95433f2ce28d7 : TIMED_OUT in 3h 03m 19s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3899d519b41545e6ac89191658931e83 : SUCCESS in 1h 16m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6efa6ce3f7c244d68a85fdd58fd3dd20 : SUCCESS in 1h 05m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/04167b6a54a947b3b6800a3b9ec11518 : FAILURE in 7m 40s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d2df7c9503f24dd2b46c06a6d589c705 : FAILURE in 1h 07m 39s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4acb5e8f1c81406d8eb18009a5e11329 : SUCCESS in 4m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1cd8ee4423804ece8cb26acdef112bd5 : SUCCESS in 4m 38s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/88af6a997559439fb4b0ccf273b2d85f : SUCCESS in 35m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/682f86f54f164da7858827cd97657ddc : SUCCESS in 33m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/01a50cdd00f941038d5eed36dc364c55 : SUCCESS in 44m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/274eb343bc364edab390c0508974ce1f : SUCCESS in 42m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6a5c5dba382a458eada0308e603f3398 : SUCCESS in 30m 57s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/aa6c1e3b97f04712a3c610632ce1fbcb : FAILURE in 14m 24s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/b0807782d94c4c8d86c73c01e97f884f : SUCCESS in 57m 12s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3eadcab860964028958f5b30f8ba90e1 : SUCCESS in 45m 49s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/4e0430b642f446f5bc8e673878e51891 : SUCCESS in 42m 04s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/0c56d9c6149b4476820d301859fb69ea : SUCCESS in 55m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"067431557485c62e70cd4cc8761eb0c892f10d86","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-05-17 16:40:09.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/75/883175/1/check/DellEMC_PowerMAX_Manila/00fb619/DellEMC_PowerMAX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"562cf6919c7adc1c6476965823540ce67117590c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-17 16:58:54.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/75/883175/1/check/DellEMC_Unity_Manila_dhss_true/4126102/DellEMC_Unity_Manila_dhss_true/124 : SUCCESS in 3h 30m 44s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/75/883175/1/check/DellEMC_Unity_Manila_dhss_false/21d2858/DellEMC_Unity_Manila_dhss_false/136 : SUCCESS in 1h 43m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"0538060571df4f15708bedb9946194a8708fcdc4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-17 18:41:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/75/883175/1/upstream-check/manila-tempest-plugin-ontap-dhss/c53c04d/ : FAILURE in 1h 28m 25s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/75/883175/1/upstream-check/manila-tempest-plugin-ontap-dhss-rest/3f48fe4/ : FAILURE in 2h 00m 20s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/75/883175/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/d0684b7/ : SUCCESS in 1h 48m 22s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/75/883175/1/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/bb8da9d/ : FAILURE in 1h 54m 00s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"3255b57e2e199220055c031916c4a1f05471de7d","author":{"_account_id":33756,"name":"Luisa","email":"luisarfamaral@gmail.com","username":"rfluisa"},"date":"2023-07-03 20:47:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83950fc7d0314f5d6cc9b4bcadfdeaea016b20e1","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2023-07-04 12:18:16.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0e55f8513739ff71aa5035e0925ef902fd6391a","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-07-25 18:58:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e3aa1fa21e0884007ba0a5ee3fe99487a8a2f73d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"date":"2023-07-26 13:22:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"e1c4eb03899013153f17d049610026cbaba825f2","author":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"date":"2023-07-26 13:23:50.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"02a8bd05e3f3b38b17ce0eb7f3b63ab4eee3ad0b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-26 14:12:27.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-75-883175-2/ns5_manila : FAILURE in 49m 4s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"dcfe077fe6b3bce3c9e540b871e6a6e4df7ade4b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-26 14:46:55.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/75/883175/2/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/e4fa3ad/ : SUCCESS in 1h 22m 22s","accounts_in_message":[],"_revision_number":2},{"id":"7f473c94ceb8db75e1e8c2a5328e8c526e6afeff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 15:27:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c7aeedc43d149dd92f86ba29ffa4b84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4b3d82dfcf54bb6a6c3dd6a0267baec : SUCCESS in 9m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4795f7bf646a47ee858d57bf2381c93b : SUCCESS in 24m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/408c60197cb141079b40f0473ae2ea52 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fd9f7fecc6940f39d2ae44b515067de : SUCCESS in 10m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/188be75f6e5343a385d8cc220c55caba : SUCCESS in 9m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25d959dc421141b0bd7f06381b6dc9d8 : SUCCESS in 9m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ca13b0d28fa45b5a74ac7f6ad0178c4 : SUCCESS in 5m 55s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/14d756712f8345ce86504cedadaf104c : SUCCESS in 1h 33m 44s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3c87db5b12fc4c5aab371e4bbbfc7d5c : FAILURE in 30m 36s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ceef6e22253a4ba3806c01291a7ec3d0 : FAILURE in 37m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9a6192bf73b8460dab25d005ab035fb7 : SUCCESS in 1h 02m 39s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9ced023c2b054b7395b1fdb0d304da2b : SUCCESS in 1h 02m 40s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f5f9ab875639449b8146d94f3f0363d1 : SUCCESS in 1h 11m 54s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/86774a40e3a944c4959655913e42c8f5 : SUCCESS in 5m 05s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5448c98361524d64a02cdf19ffbd96a3 : SUCCESS in 5m 18s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8e252e02a3f64e27a3d44893f49cf53c : SUCCESS in 30m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e0340542b30f4d79aacfacf34ff5da2e : SUCCESS in 32m 26s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fb0d9b1a0df2472397e2a0aa636af1d2 : FAILURE in 45m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/ba157795c09246a48fc669c187356ace : SUCCESS in 26m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/30a32dda3c964111b67d8b9ce63f0b6a : SUCCESS in 19m 32s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/dee1cb765bda412591e37dca7c250d77 : SUCCESS in 1h 52m 37s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a014fc712d854dee9c8112ff84a2fe9f : SUCCESS in 54m 16s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/49241c12b3994657ad8943fb62f8cb80 : SUCCESS in 1h 06m 59s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/41f02513fdbe47188c7ce1eb3c49d637 : SUCCESS in 37m 49s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b8c5bade030642f6a26a9194840340fa : SUCCESS in 38m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a7970078d50ffd231f7048c9a2bd24c63cddc785","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-26 15:45:12.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-75-883175-2/NFS/ : SUCCESS in 2h 21m 56s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"320a936edf9c8279987258a146100c0f7f0f2cc4","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-07-26 15:59:14.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/75/883175/2/check/DellEMC_PowerStore_Manila_dhss_false/4849f9c/DellEMC_PowerStore_Manila_dhss_false/16 : FAILURE in 31m 03s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"5f24a3e50766b51fdc92a7bace9e327722a3264b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-26 18:47:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/75/883175/2/upstream-check/manila-tempest-plugin-ontap-dhss/2317c53/ : RETRY_LIMIT in 8m 28s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/75/883175/2/upstream-check/manila-tempest-plugin-ontap-dhss-rest/a3cd348/ : TIMED_OUT in 3h 16m 55s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/75/883175/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/abf9310/ : FAILURE in 1h 37m 18s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/75/883175/2/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/444689d/ : FAILURE in 2h 48m 10s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"d3bda820f0e27d6d7a6860e2b6f5a969eb849ff8","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-27 00:16:44.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/75/883175/2/check/DellEMC_Unity_Manila_dhss_true/d9d13a2/DellEMC_Unity_Manila_dhss_true/278 : SUCCESS in 3h 00m 00s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/75/883175/2/check/DellEMC_Unity_Manila_dhss_false/3d7b51f/DellEMC_Unity_Manila_dhss_false/296 : SUCCESS in 1h 10m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"2132875ba137cfd011bdcb15e91c35a821ce0212","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-07-27 05:37:02.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e1bb2940b2c8b89a0c4c78014b853ca50e6ab581","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-27 07:56:43.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/75/883175/2/upstream-check/manila-tempest-plugin-ontap-dhss/30af49a/ : RETRY_LIMIT in 8m 25s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/75/883175/2/upstream-check/manila-tempest-plugin-ontap-dhss-rest/c18629b/ : RETRY_LIMIT in 9m 28s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/75/883175/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/ca51d96/ : RETRY_LIMIT in 10m 27s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/75/883175/2/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/f135d05/ : FAILURE in 2h 18m 25s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"c6235e48073afeeaf247024b464abbeb1960d3f1","author":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"date":"2023-07-27 19:53:45.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dc19133c9d9045cc89258a51c750b05a9e9efefb","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-07-28 01:42:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5995013c9020a9412896433bac5b85b1b902cc9b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-28 02:53:29.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/75/883175/2/check/DellEMC_PowerFlex_Manila_dhss_false/4c807d2/DellEMC_PowerFlex_Manila_dhss_false/99 : ABORTED in 18m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"65ac60d470d1ad91091f29e4ed97dbf7aa406cae","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-08-01 22:02:23.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9d5f3d96f0b7776ec9c155ea7d28fa2ed444cba0","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-08-02 03:54:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"09cfb1f03eb76d9e1a235ef7f24a2ff0a2723639","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-08-02 21:23:38.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a1d75a66cf60a5437406fbe6b600e78a6b8dbd30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 21:24:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c73250c25bac1f2d8a87eb4070fad971f88e7748","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 22:26:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7d23933745741c0955c42377402aa01\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4d9d1054f6e40368cefbdd1849d4f30 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/659f11577876486cbfa882cc10f6e1a1 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0eb73aed9a304641adbf640ffd4c6251 : SUCCESS in 9m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e7cb9ac62c14ab787fa201f4aad62c5 : SUCCESS in 7m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/301b37044baf423eb11dbc275c3b7a57 : SUCCESS in 11m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d51d1f8dcc7431da5ce1be0ecb93477 : SUCCESS in 5m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/23dbb5e4d65845eab6e0a855ab084ca1 : SUCCESS in 56m 11s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d00e819b3bf349d08c94ab2d320d0971 : SUCCESS in 25m 09s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/902a435e6d8d478faf06213cc3a93681 : SUCCESS in 32m 19s","accounts_in_message":[],"_revision_number":2},{"id":"b43532bf89df5091501d1a791ef6db3f5462230e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 22:26:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9154381fb1b87026efd7d077319ccb2517d7e869","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 22:27:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e814f274272947c9abd90cdfc5afde3b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2922a896a9064796b27b53319718aedd : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bfa3d0fccc2464b8caa806bb8dc6050 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"69efb9ae5329cf5487273cd4281219880b90070e","revisions":{"1b580ea2e1955895cc66fb271dd8f5babb849a9e":{"kind":"REWORK","_number":1,"created":"2023-05-17 13:27:55.000000000","uploader":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"ref":"refs/changes/75/883175/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/75/883175/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/883175/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/883175/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/883175/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/75/883175/1"}}},"commit":{"parents":[{"commit":"9a7738a6b31686109b6a62b7ca437ee877b803d1","subject":"Merge \"db: Migrate \"availability zone\" APIs to enginefacade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9a7738a6b31686109b6a62b7ca437ee877b803d1"}]}],"author":{"name":"Raffaela Cunha","email":"raffaelacunha@gmail.com","date":"2022-10-13 19:45:41.000000000","tz":0},"committer":{"name":"Renan Vitor","email":"renanpiranguinho@gmail.com","date":"2023-05-17 13:24:26.000000000","tz":0},"subject":"[NetApp] Fix NetApp driver create from snapshot","message":"[NetApp] Fix NetApp driver create from snapshot\n\nChanges were done in create_share_from_snapshot\nmethod so that a scoped account does not need to get the\nsource and the destination cluster names and does not have the\npermissions.\n\nCloses-Bug: #1922512\nChange-Id: Ib36c81c213a374a918378854ce0a89ce70acf1d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1b580ea2e1955895cc66fb271dd8f5babb849a9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1b580ea2e1955895cc66fb271dd8f5babb849a9e"}]},"branch":"refs/heads/master"},"69efb9ae5329cf5487273cd4281219880b90070e":{"kind":"REWORK","_number":2,"created":"2023-07-26 13:22:55.000000000","uploader":{"_account_id":35586,"name":"Renan Vitor","email":"renanv@netapp.com","username":"renan_vitor"},"ref":"refs/changes/75/883175/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/75/883175/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/883175/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/883175/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/75/883175/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/75/883175/2"}}},"commit":{"parents":[{"commit":"9a7738a6b31686109b6a62b7ca437ee877b803d1","subject":"Merge \"db: Migrate \"availability zone\" APIs to enginefacade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9a7738a6b31686109b6a62b7ca437ee877b803d1"}]}],"author":{"name":"Raffaela Cunha","email":"raffaelacunha@gmail.com","date":"2022-10-13 19:45:41.000000000","tz":0},"committer":{"name":"renanpiranguinho","email":"renanpiranguinho@gmail.com","date":"2023-07-26 13:22:24.000000000","tz":0},"subject":"[NetApp] Fix NetApp driver create from snapshot","message":"[NetApp] Fix NetApp driver create from snapshot\n\nChanges were done in create_share_from_snapshot\nmethod so that a scoped account does not need to get the\nsource and the destination cluster names and does not have the\npermissions.\n\nCloses-Bug: #1922512\nChange-Id: Ib36c81c213a374a918378854ce0a89ce70acf1d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/69efb9ae5329cf5487273cd4281219880b90070e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/69efb9ae5329cf5487273cd4281219880b90070e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
