)]}'
{"id":"openstack%2Fmanila~883243","triplet_id":"openstack%2Fmanila~master~I1f010c4ff9faf837bd50b6c2823cbc526f3a42df","project":"openstack/manila","branch":"master","topic":"bug/2019846","attention_set":{},"removed_from_attention_set":{"18816":{"account":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"last_update":"2023-05-26 12:36:02.000000000","reason":"Change was submitted"},"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-05-26 11:23:32.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-05-25 15:38:46.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2023-05-26 12:36:02.000000000","reason":"Change was submitted"},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2023-05-26 12:36:02.000000000","reason":"Change was submitted"},"33038":{"account":{"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"},"last_update":"2023-05-26 12:36:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1f010c4ff9faf837bd50b6c2823cbc526f3a42df","subject":"Report PortLimitExceeded error to customer","status":"MERGED","created":"2023-05-16 10:18:51.000000000","updated":"2023-05-26 12:37:10.000000000","submitted":"2023-05-26 12:36:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883243-bug/2019846","meta_rev_id":"8e10fc95735694d73f1c98c164259f4be6448560","_number":883243,"virtual_id_number":883243,"owner":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-26 12:36:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":2,"date":"2023-05-26 11:23:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-05-25 15:38:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":1,"date":"2023-05-26 11:23:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-16 10:21:10.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":33038,"name":"Chuan Miao","email":"chuan137@gmail.com","username":"chuan137"},"state":"REVIEWER"},{"updated":"2023-05-16 10:21:10.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2023-05-16 10:21:10.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2023-05-16 10:21:10.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-05-16 10:21:10.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"REVIEWER"},{"updated":"2023-05-16 11:42:42.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-16 11:51:32.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-05-16 12:39:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-17 02:21:13.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-05-17 18:29:52.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-17 20:44:40.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"}],"messages":[{"id":"16c39495efdea0f2b7f07e60d43bda3342f75d8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-05-16 10:18:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a214c2ba0808c9a8ae772d1fd79295c83704c675","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-05-16 10:22:27.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f3f9083add5b4bc3ad1aa5fe8d01b6541d32717f","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-05-16 10:59:34.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ff32b23fe6773290c2e66ca4239798bb37b88544","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-05-16 11:42:42.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-43-883243-1/ns5_manila : SUCCESS in 1h 23m 14s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"263e63064ada971946a4dd4de939eef287c2ace3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-05-16 11:51:32.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-43-883243-1/NFS/ : SUCCESS in 1h 32m 22s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"e0a19b001b2ee24994cc91e191ccf3a332a73c8b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-05-16 12:03:50.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-43-883243-2/ns5_manila : SUCCESS in 1h 21m 52s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"599f3a14ef07f9abd9ce4aa8f5261fc86fa8ee70","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-05-16 12:18:05.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-43-883243-2/NFS/ : SUCCESS in 1h 43m 54s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"201b3d539c07a7fc61087a6a8ada7a83c048bb2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 12:39:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f43376103c4241ef9bc14da979358a52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/499f299134e347a3ab2323dc5066ab09 : SUCCESS in 9m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32388e0807284bdfa887de12f271ae27 : FAILURE in 22m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cde112073e6d466887a2decf93bca093 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80e9230f46d242898e44cf9e3d5e0365 : SUCCESS in 10m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c28aed834f6c46fa91ab3697bf1453c5 : SUCCESS in 9m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5fadc721d7441c08e4833059a8dacd3 : SUCCESS in 9m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d976919dd68047b99b3cac06d4f68195 : FAILURE in 33m 06s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/728962da696542e58c5cc3fa4b2a2430 : FAILURE in 2h 14m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d08bc1c949cc4e46af52f22de1d80340 : SUCCESS in 1h 09m 41s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c1b002fe239944cca3577dbead32b225 : FAILURE in 57m 53s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/d3c49881329e4a49a266bd197c4490f0 : FAILURE in 7m 35s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4da361458a8a4806a5c6abbd9552cb65 : SUCCESS in 1h 22m 07s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/2075a0c14fb7434d8d7394007d92d418 : SUCCESS in 4m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3f76b2d8692948e3b8f9a458677167c9 : SUCCESS in 4m 16s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/281b3b383b864661969cda0914f00d9c : SUCCESS in 36m 51s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ee7e88dc712942f79dc1a2f6f5eac4b2 : SUCCESS in 33m 13s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/27fe64e1535a4a87b531ac267698bd5b : SUCCESS in 46m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/6cc13f4c305e4a34bfc004020fa1a72f : SUCCESS in 36m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/94fbc128455345f9ba322ee6780451a5 : SUCCESS in 22m 11s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/26af12d0b9c14ed7abd276e58c44dafb : SUCCESS in 1h 24m 30s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/53fdea431b7d4c5a9f50f04d88d5d967 : SUCCESS in 59m 50s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/63fa3141def94038bb3071a1445765a5 : SUCCESS in 1h 01m 01s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6becfb7a46644369ab1efd414e1d9853 : SUCCESS in 37m 17s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/aae575c845bc4561a9f9c7c58b52a4a5 : SUCCESS in 37m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"41261f20878fc8cf2ac3fcc673795836a2401193","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-17 02:21:13.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/43/883243/2/check/DellEMC_Unity_Manila_dhss_true/3c8d985/DellEMC_Unity_Manila_dhss_true/121 : SUCCESS in 3h 43m 34s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/43/883243/2/check/DellEMC_Unity_Manila_dhss_false/068ea37/DellEMC_Unity_Manila_dhss_false/132 : SUCCESS in 1h 06m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"75f53d4f3b949778ae843e886f5c27b2fa4fe757","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-05-17 15:22:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"df546df845104f8d764a8223b0472db44841fdc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 18:29:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5ba87fad82446728c312831af397e38\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0960a81122604670bd2c310872cf718d : SUCCESS in 9m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8c0232d3bba4df390cf584c07d6a54f : FAILURE in 26m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79d36c99b07444c388ea39f1f0460579 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a77a165ea594c39bb14d1f4d134b2b7 : SUCCESS in 9m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81e64a859ee54c3c92ca4ebb03e6a588 : SUCCESS in 9m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c96d232ab1754c7a8dc650b878e4997f : SUCCESS in 11m 14s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3ca459e133134b7cb667e9d5e27b2e9e : SUCCESS in 1h 22m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/787bf4e75eea47ccba2086d264d988f4 : FAILURE in 3h 01m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c8d06ded8f834a88a65b7cf67ad08486 : FAILURE in 2h 53m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e81af868f05a4f73b900b04dc6942670 : SUCCESS in 1h 25m 13s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ddb004018ea542afae1b41067081a032 : FAILURE in 14m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f774db1bb6c84966a65af5f2b1009477 : SUCCESS in 1h 21m 26s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9b137ca8c98147c188f8c64885e9921d : SUCCESS in 5m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3325bad5230147b18f44b6eabe8a474c : SUCCESS in 4m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1a9917cfc0284b4786095098c14d9bd2 : SUCCESS in 35m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c49672d30d0a4c74850daadc2cc3d462 : SUCCESS in 46m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/83dfe68a238549e689269c7fcaac2b10 : FAILURE in 1h 00m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/9fab31d6131145dc997fd9a2fe6a187a : SUCCESS in 42m 15s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/25bdb7a6dfa142d087be8db6c03e4e9a : SUCCESS in 33m 04s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/3afd834f06e04714a41a4bd21c25fe0d : FAILURE in 30m 30s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4da5c6e7e51441b1afaf7b6e803c2f1f : SUCCESS in 1h 20m 43s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2cb3172aa19746b18166ef6aa7d001b6 : SUCCESS in 1h 01m 11s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/a5da27047945474db2792d44a2b0766c : SUCCESS in 55m 17s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/25c12b73b5c845ee921670fea46d0ef2 : SUCCESS in 43m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c315fb1270abcb221a57c78e9853b534757384bb","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-05-17 18:29:52.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/43/883243/2/check/DellEMC_PowerMAX_Manila/6ae69b1/DellEMC_PowerMAX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":2},{"id":"0251c71884b49828533b87d922944d7aed8fc877","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-17 20:29:38.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6bdd0faa88d93fdb4a1907b72effa27e16720fca","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-17 20:44:40.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/883243/2/upstream-check/manila-tempest-plugin-ontap-dhss/5695e18/ : SUCCESS in 2h 13m 40s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/883243/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/e12dae7/ : FAILURE in 2h 13m 36s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"40e4b5a1e98560cda52899ec7ce1fbdf44039f7a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-17 22:15:25.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/43/883243/2/check/DellEMC_Unity_Manila_dhss_true/8afaf00/DellEMC_Unity_Manila_dhss_true/125 : SUCCESS in 3h 45m 41s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/43/883243/2/check/DellEMC_Unity_Manila_dhss_false/8fd0525/DellEMC_Unity_Manila_dhss_false/137 : SUCCESS in 1h 11m 47s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"c26cb8ffb6e19572fbffdb6ee32698d2862ea449","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-05-19 09:37:44.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8a26d430087de8539102117b2059742437e1314a","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-05-19 09:37:47.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"50f018982a6de2f8ee54e0c5bea77067751571dc","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-05-19 10:58:53.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-43-883243-3/ns5_manila : SUCCESS in 1h 20m 38s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"cae8e935293fe6b911f94a223c4a30ab23ca27ff","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-05-19 11:05:40.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-43-883243-3/NFS/ : SUCCESS in 1h 27m 37s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"b0ebd43565bd1ea9daefa0202abf70b94f95e802","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 12:52:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68c36aa2796d43cc9be556fe99da3a61\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a684d2d499b4cb78e8caf32f149e573 : SUCCESS in 10m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9ea42a572e0459ebcace84162938004 : SUCCESS in 25m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/088c33b71d9d459a8d9ed97797103be9 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b332af8845f34ec890aead365120835f : SUCCESS in 9m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d82b955131684b9abadcecaa4939aa91 : SUCCESS in 8m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/adf1e3d1aa8e400993fb47a90a353cfb : SUCCESS in 14m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5e23cae9ef5f49a3bac3f4dd7e0282aa : SUCCESS in 1h 11m 33s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3c863d3fedf0460083d417229e810436 : SUCCESS in 2h 15m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/edda8b5f77404461b9eb7b088a4924b8 : TIMED_OUT in 3h 07m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b0935209ef19408a8d38548ca1dc2e3e : SUCCESS in 1h 17m 15s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/b5793fdd2793484180da4f8d2ef11d78 : SUCCESS in 1h 18m 47s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/03996ef83b71425bae470db448e891e8 : FAILURE in 59m 14s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d8236a50f8b143d3b667e61ec0413ba3 : SUCCESS in 4m 43s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1ed0790542aa43acb2c8fdb30fe635a6 : SUCCESS in 4m 40s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7ec3f7dc173b4e309a67cd420633a09a : SUCCESS in 34m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/62314789dfb04494812a40245f74d2c8 : SUCCESS in 38m 59s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/100f0ed507854a21b05970922a2c3fb2 : SUCCESS in 1h 03m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/a992cacd584b4a76b68d75100d3d05f1 : SUCCESS in 35m 07s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a53df747c99243f386a437729ca9dfb3 : SUCCESS in 17m 49s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/2bf3b2ba862e4e33bc8fa65711acc8e3 : FAILURE in 24m 36s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/bb41235b644c4a08b8e56a1c41e46ea3 : FAILURE in 39m 21s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5c99ef2c2c2d4f16953f8ab4eeed5dad : SUCCESS in 54m 19s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/83b739068bfb4657a8a0638baeb788b8 : SUCCESS in 25m 51s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/a7c0dc3473984cd48c4b266dab6b55b2 : SUCCESS in 33m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d989e103752e3eb1ec65316676fcf6f350194dd1","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-05-19 12:52:26.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/43/883243/3/check/DellEMC_PowerMAX_Manila/cab0216/DellEMC_PowerMAX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":3},{"id":"9a26fdc47792591255091a44cc9766f9733068b3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-19 13:50:25.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/43/883243/3/check/DellEMC_Unity_Manila_dhss_true/a2b9782/DellEMC_Unity_Manila_dhss_true/129 : SUCCESS in 4h 12m 23s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/43/883243/3/check/DellEMC_Unity_Manila_dhss_false/89cde64/DellEMC_Unity_Manila_dhss_false/141 : SUCCESS in 1h 29m 50s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"ae482c5dd10812436e60bd44ad39a72b22c4b2d2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-19 15:44:10.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/883243/3/upstream-check/manila-tempest-plugin-ontap-dhss/58d751b/ : FAILURE in 2h 50m 50s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/883243/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/c9b0e13/ : FAILURE in 1h 34m 08s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"da3319b2b85a7dbf6ae5b4dda47954705345f1bf","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-20 00:36:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"edfd7f633b7476c44f8504d9bb9c42042ba5884e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-05-21 10:32:55.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ec1948197f1044767ecf6b9a0c3a19b3ff90453a","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-05-21 10:33:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4186bbc38ebc6eb72cf3bea3d2e2acef09df53fd","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-05-21 11:51:50.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-43-883243-4/ns5_manila : SUCCESS in 1h 18m 23s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"bc12a12165befa74c95cee2c21273d0a514bf839","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-05-21 11:56:20.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-43-883243-4/NFS/ : SUCCESS in 1h 23m 2s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"1daad999482193fceed849c5ec40807b6349c69f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-21 13:42:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4abf11fd0c24100942ffa2177ba1ca8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8c7404b1ff8488490d7d87b905ce327 : SUCCESS in 8m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6622ee580e8940a9ad453d73d01dca8d : SUCCESS in 16m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdf10f8f89b2402ab494234ccee5ef7c : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f989b22442f459f9a38333126071361 : SUCCESS in 9m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ea84fc6d5db40308f9acc44afb04990 : SUCCESS in 8m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e50378fa3bf4e7b8003bed0ba6092bf : SUCCESS in 9m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb5a7f5900ec42cb9b118b41ebe29b17 : SUCCESS in 6m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/52596a02c2b149c5a01feacb07729cd9 : SUCCESS in 1h 00m 09s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f5ad05586f5c49f9a565b78991ea41e8 : TIMED_OUT in 3h 02m 35s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e8f8a5735aef440ea0e1a173ba569897 : SUCCESS in 1h 19m 39s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b0604c22d9284ac68e602dbcb52c213c : SUCCESS in 56m 04s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/37496794816d4fbabd1f232c0c466a31 : SUCCESS in 56m 58s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6a860795636a4d72b7688c7514c5fd72 : SUCCESS in 57m 28s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4f2012bf793647baa36605aa8a5c44c7 : SUCCESS in 4m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f48da824639e4d8e81ca63ca80373f94 : FAILURE in 5m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c928cddbf55f40f18a86707473b2a0fd : SUCCESS in 33m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/71b3f48b7eca42f2b6fc2206684079d7 : SUCCESS in 45m 22s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/040c195551c744e7a7313ab536014aa5 : SUCCESS in 56m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/fe2f3433195b4e63b8c3e07c50a6d99e : SUCCESS in 25m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0a873e7e344446a184d790fc9b139206 : SUCCESS in 16m 19s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/263589aae3e341cdaa7c610ee6058e65 : FAILURE in 9m 59s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/6f274ed3d7104ad48f9719a8abeecba2 : SUCCESS in 46m 02s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3710a80e2f504b49829ed56a7c31c1c4 : SUCCESS in 47m 44s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0df0e421e4424dd69a653b72e5f25ae5 : SUCCESS in 22m 10s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c4bc8e2aeb4c47f19fc7ed1e18f1dff8 : SUCCESS in 38m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"03f2f9f9bbc9daf6ad6d62a312cf487c4c642307","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-05-21 13:42:57.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/43/883243/4/check/DellEMC_PowerMAX_Manila/c1bda4f/DellEMC_PowerMAX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":4},{"id":"9c10fc8b4be75ea2018ccf93f847caac99db82d9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-05-21 14:38:40.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/43/883243/4/check/DellEMC_Unity_Manila_dhss_true/17fbd71/DellEMC_Unity_Manila_dhss_true/133 : SUCCESS in 4h 05m 28s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/43/883243/4/check/DellEMC_Unity_Manila_dhss_false/1b13c7b/DellEMC_Unity_Manila_dhss_false/145 : SUCCESS in 1h 32m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"aac7fde4a61a385d938a8a9abd4370d5ed260236","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-05-21 15:21:33.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/883243/4/upstream-check/manila-tempest-plugin-ontap-dhss/fdb9fc6/ : SUCCESS in 1h 11m 09s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/43/883243/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/9dc7e97/ : SUCCESS in 1h 38m 05s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"02abbdb9b94deb57f851a0501192aa379bd6fa15","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-05-25 15:38:46.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c72c641d3e6b482f64b2be7f8f0a4e1fa31d9654","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-05-26 11:23:32.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8e05e22e1261d45dbeab0a5e52d68b3bc60c130a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 11:24:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0ecf1552ce5ed1d1312b9a00bc202dadae0cf2d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 12:36:02.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4758a483f60b486684adb501de103001\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fee28bbc52184748a24f1dfae2daaa2a : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56dd96956e364119b71ef7ad7c6a5be2 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a0cc550479b4d9ca4d096fb63689b72 : SUCCESS in 10m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a69d9c1d2eb4c9b9d105b713f55670b : SUCCESS in 10m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/013536627fc64183966af1f0df437b01 : SUCCESS in 10m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9a0a48d14ca450d8fb88558a6fe26d5 : SUCCESS in 8m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f42976fdc781406d87aaa525c4d97a6e : SUCCESS in 1h 11m 28s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/444637536c8f49a284c95e13764411c3 : SUCCESS in 29m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/55366df149d0417294aecaeefabdef50 : SUCCESS in 41m 41s","accounts_in_message":[],"_revision_number":4},{"id":"b5d98748f72250abaaa666b270ec8ffc95878604","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 12:36:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"8e10fc95735694d73f1c98c164259f4be6448560","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 12:37:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/917f15313c59466bb9c696dd50cc4f21\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa18cdb3b5d44aa59517317bb0dcacdd : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb975d32e86e4d83ada6457b834a9eaa : SUCCESS in 48s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a372e5309821720308d13d8b597be551769e8e4f","revisions":{"bffe82b1d45050776be4a9a1fdc26682cbf31952":{"kind":"REWORK","_number":1,"created":"2023-05-16 10:18:51.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/43/883243/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/43/883243/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/43/883243/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/43/883243/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/43/883243/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/43/883243/1"}}},"commit":{"parents":[{"commit":"9762d8772695a754bea3dc3cd878b46ce20e7cf8","subject":"Merge \"db: Migrate \"driver private data\" APIs to enginefacade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9762d8772695a754bea3dc3cd878b46ce20e7cf8"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-05-16 10:09:07.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-05-16 10:15:44.000000000","tz":0},"subject":"Report PortLimitExceeded error to customer","message":"Report PortLimitExceeded error to customer\n\nWhen share creation fails due to missing ports quota (\n\u0027manila.exception.PortLimitExceeded: Maximum number of ports exceeded.\u0027),\nerror message propogated to user is not useful. Improve user message in\nthis scenario.\n\nCloses-bug: #2019846\nChange-Id: I1f010c4ff9faf837bd50b6c2823cbc526f3a42df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bffe82b1d45050776be4a9a1fdc26682cbf31952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bffe82b1d45050776be4a9a1fdc26682cbf31952"}]},"branch":"refs/heads/master"},"cb9a8695fe5395deb0bf2650e6bd5a16f2fef7d1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-05-16 10:22:27.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/43/883243/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/43/883243/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/43/883243/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/43/883243/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/43/883243/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/43/883243/2"}}},"commit":{"parents":[{"commit":"9762d8772695a754bea3dc3cd878b46ce20e7cf8","subject":"Merge \"db: Migrate \"driver private data\" APIs to enginefacade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9762d8772695a754bea3dc3cd878b46ce20e7cf8"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-05-16 10:09:07.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-05-16 10:21:35.000000000","tz":0},"subject":"Report PortLimitExceeded error to customer","message":"Report PortLimitExceeded error to customer\n\nWhen share creation fails due to missing ports quota (\n\u0027manila.exception.PortLimitExceeded: Maximum number of ports\nexceeded.\u0027), error message propogated to user is not useful.\nImprove user message in this scenario.\n\nCloses-bug: #2019846\nChange-Id: I1f010c4ff9faf837bd50b6c2823cbc526f3a42df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cb9a8695fe5395deb0bf2650e6bd5a16f2fef7d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cb9a8695fe5395deb0bf2650e6bd5a16f2fef7d1"}]},"branch":"refs/heads/master"},"2bba8318c363e7d5b6f603b68afcdc02857d1aeb":{"kind":"REWORK","_number":3,"created":"2023-05-19 09:37:44.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/43/883243/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/43/883243/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/43/883243/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/43/883243/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/43/883243/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/43/883243/3"}}},"commit":{"parents":[{"commit":"b3951d06dd5b0a6770e8b90f26d471ac38504159","subject":"Merge \"db: Migrate \"share network\" APIs to enginefacade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b3951d06dd5b0a6770e8b90f26d471ac38504159"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-05-16 10:09:07.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-05-19 09:37:35.000000000","tz":0},"subject":"Report PortLimitExceeded error to customer","message":"Report PortLimitExceeded error to customer\n\nWhen share creation fails due to missing ports quota (\n\u0027manila.exception.PortLimitExceeded: Maximum number of ports\nexceeded.\u0027), error message propogated to user is not useful.\nImprove user message in this scenario.\n\nCloses-bug: #2019846\nChange-Id: I1f010c4ff9faf837bd50b6c2823cbc526f3a42df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2bba8318c363e7d5b6f603b68afcdc02857d1aeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2bba8318c363e7d5b6f603b68afcdc02857d1aeb"}]},"branch":"refs/heads/master"},"a372e5309821720308d13d8b597be551769e8e4f":{"kind":"REWORK","_number":4,"created":"2023-05-21 10:32:55.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/43/883243/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/43/883243/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/43/883243/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/43/883243/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/43/883243/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/43/883243/4"}}},"commit":{"parents":[{"commit":"b3951d06dd5b0a6770e8b90f26d471ac38504159","subject":"Merge \"db: Migrate \"share network\" APIs to enginefacade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b3951d06dd5b0a6770e8b90f26d471ac38504159"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-05-16 10:09:07.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-05-21 10:32:52.000000000","tz":330},"subject":"Report PortLimitExceeded error to customer","message":"Report PortLimitExceeded error to customer\n\nWhen share creation fails due to missing ports quota (\n\u0027manila.exception.PortLimitExceeded: Maximum number of ports\nexceeded.\u0027), error message propogated to user is not useful.\nImprove user message in this scenario.\n\nCloses-bug: #2019846\nChange-Id: I1f010c4ff9faf837bd50b6c2823cbc526f3a42df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a372e5309821720308d13d8b597be551769e8e4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a372e5309821720308d13d8b597be551769e8e4f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
