)]}'
{"id":"openstack%2Fmanila~887115","triplet_id":"openstack%2Fmanila~master~I146bc09e4e8a39797e22458ff6860346e11e592e","project":"openstack/manila","branch":"master","topic":"bp/allow-locking-shares-against-deletion","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-08-21 21:13:49.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-08-23 23:27:37.000000000","reason":"Change was submitted"},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2023-08-22 02:51:34.000000000","reason":"\u003cGERRIT_ACCOUNT_30407\u003e replied on the change","reason_account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}},"31721":{"account":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"last_update":"2023-08-23 23:27:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I146bc09e4e8a39797e22458ff6860346e11e592e","subject":"Resource Locks: Support for share deletion lock","status":"MERGED","created":"2023-06-27 20:09:24.000000000","updated":"2023-08-24 15:39:30.000000000","submitted":"2023-08-23 23:27:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":95,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887115-bp/allow-locking-shares-against-deletion","meta_rev_id":"ae64b9195900c47c71dfd46e4098951e8c6f4710","_number":887115,"virtual_id_number":887115,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-23 23:27:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-08-21 21:13:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-08-24 15:39:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":2,"date":"2023-08-22 02:51:34.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-08-23 22:10:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-27 21:33:34.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-27 21:46:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-27 21:46:27.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-27 21:55:34.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-06-28 04:41:16.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-06-28 13:28:16.000000000","updated_by":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"reviewer":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"state":"CC"},{"updated":"2023-07-06 03:11:00.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-07-08 03:50:30.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-08 04:21:22.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-07-14 19:20:01.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2023-07-27 18:54:00.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-08-10 13:42:06.000000000","updated_by":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"reviewer":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"state":"REVIEWER"},{"updated":"2023-08-17 12:42:42.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2023-08-18 10:01:07.000000000","updated_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2023-08-24 15:39:30.000000000","updated_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"}],"messages":[{"id":"7224d581a73e5593a590683d1115666cd18c2f19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-06-27 20:09:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"989db8647dd7aef72da17b457b70eb49a5a9af8b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-06-27 21:33:34.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-15-887115-1/ns5_manila : SUCCESS in 1h 23m 41s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"a7f1cb89c18ac846923fd07acfdf173b0ad62023","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-27 21:46:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6c3e0d019814d1fb22657b21ab3f548\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b39310908ac45c48263934ecb110009 : SUCCESS in 9m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66c4d131488c4da499ad19f7117eca34 : FAILURE in 26m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c348b83cb094b928fabdfd9abc72b10 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30c14ba931e94e8f82be65c0593ad8f4 : SUCCESS in 11m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c355a75d4cb4b228ecf28312bf492ed : SUCCESS in 12m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f1cbf7b49204559a5fcd85699ef4d1d : SUCCESS in 9m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/787d6898a7b84d31b6b798a980eb78f9 : SUCCESS in 1h 28m 37s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/93441355234e48408be62d52ceb1c966 : FAILURE in 32m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ac883c1c10e24d9cad02c5ecc4545950 : FAILURE in 56m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/798730ca7b5842c8954c1fd599270e6c : SUCCESS in 1h 30m 50s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/de93cb12d8ae4bdea090d165a6736072 : SUCCESS in 1h 20m 40s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d9700f37458642b1b6fab2550f6c1e81 : SUCCESS in 1h 04m 32s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a97ed27e0c7841639a7ead114f078a91 : SUCCESS in 4m 52s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8c81ca665dd8486b831e6a7f68f6b43b : SUCCESS in 4m 46s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5432f5b987144015bd5d80f4071aabe3 : SUCCESS in 26m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/04e81b2e49744930b0f475ec7c347f51 : SUCCESS in 40m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/32a7c375dc8b41378bdbdca775564d03 : FAILURE in 1h 04m 44s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/3fefdecc3874484183d834640009d21a : SUCCESS in 26m 59s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/71bfded5cf87486aa4af31f7943d38e7 : SUCCESS in 35m 09s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/219f4c964c0245c0a0c21bb3ff80d1ac : SUCCESS in 1h 21m 01s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3c23e3e569244a7b950ade1c7d243e8b : SUCCESS in 48m 19s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c9a746646e8b4c0db2cc41e0f0df635c : SUCCESS in 1h 05m 21s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/75dec4bd06a247e3808c45d4c68f7648 : SUCCESS in 47m 18s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/0a186787d3a64193a47ba8daad681e66 : SUCCESS in 37m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1d486a84aed352607072025faa3fa42844f4545","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-06-27 21:46:27.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/15/887115/1/check/DellEMC_PowerMAX_Manila/69772fd/DellEMC_PowerMAX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"35d8864322aca95553a4be934c22ad6055af555d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-06-27 21:55:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/887115/1/upstream-check/manila-tempest-plugin-ontap-dhss/91cf9e4/ : RETRY_LIMIT in 7m 42s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/887115/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/488d0c6/ : RETRY_LIMIT in 8m 41s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"09a4cc0a7dfb4def67011133d817a73e220448d2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-06-28 04:41:16.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/887115/1/check/DellEMC_Unity_Manila_dhss_true/e078fb4/DellEMC_Unity_Manila_dhss_true/191 : SUCCESS in 3h 46m 33s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/1/check/DellEMC_Unity_Manila_dhss_false/f503453/DellEMC_Unity_Manila_dhss_false/205 : SUCCESS in 1h 13m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"921bb4f662aa6c83853d6d989dc222c405e837bd","author":{"_account_id":13425,"name":"Simon Dodsley","email":"simon@everpuredata.com","username":"sdodsley"},"date":"2023-06-28 13:28:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"63181aaaa32519d81c1652ffca0e0fb630d724a2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-06 03:11:00.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-887115-1/NFS/ : FAILURE in 12h 10m 47s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"7e8e67954f32032765f8457918f7cef191b98df7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-07-08 03:17:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"eb0f411dc1a44fa702a95a60addbfa1e4282da10","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-08 03:19:41.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-15-887115-2/ns5_manila : FAILURE in 1m 19s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"282bedd73b054052011396525a661b5c6156134f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-08 03:50:30.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila https://elab-os-logsrv.delllabs.net/15/887115/2/check/DellEMC_PowerFlex_Manila/334e5d4/DellEMC_PowerFlex_Manila/13 : FAILURE in 32m 23s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"c6c336c66825935790089c5c016b17dcaf135187","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-08 04:21:22.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/15/887115/2/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/c34942d/ : FAILURE in 1h 01m 56s","accounts_in_message":[],"_revision_number":2},{"id":"3e94f98a75d0c2d6308e800ccd7611524cfe9a3f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-08 04:40:15.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-887115-2/NFS/ : SUCCESS in 1h 22m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"c35fb1901a155c23fd11779a73ae0a9c7032ae7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-08 04:53:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/327070caa8eb4ac38f70d2f343c9a95e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/285e0eb11f1041b8a5b9a9c12c10a068 : SUCCESS in 9m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a82d15029a87468b9dc116fea3f9c73a : FAILURE in 33m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/943cd4ea958c4877919a319d99221b38 : SUCCESS in 18m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/760b3f29d0b7470484b880cbe10c4e3c : SUCCESS in 11m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0db8d6192fc4469bedff59cfe3601d0 : SUCCESS in 17m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/295efea63380490aaea9efade33e674b : SUCCESS in 9m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c8ce68fb0cc449de851a4865d0ffb7df : SUCCESS in 58m 33s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/191be7161b5f4b5792e5ffc42adc35d3 : FAILURE in 30m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e71ee63263594a82ba8f00e3ff9857e5 : FAILURE in 1h 06m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3adf898bfbd24256b7c81c045b056b8a : SUCCESS in 1h 20m 36s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/cd267cb9d48a48998a0d500f636dc756 : SUCCESS in 1h 26m 56s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e9349623e38c4ffabc491f5ab4063c2c : FAILURE in 29m 05s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/af91530853f3403f8f3227d3b6710db0 : SUCCESS in 4m 55s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1ba7e72afd7d402e82d9580d2d180807 : SUCCESS in 4m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8e900382adf24b17a9d19bc83fbf042d : SUCCESS in 42m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/86fe270a235c46a4ab00f8cd822c3f42 : SUCCESS in 50m 21s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/acfe9fa99ed34bb09231c3c56bc0b10c : FAILURE in 29m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/32bd24c111014c1e894a819a7b84b548 : SUCCESS in 38m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9687a115fa724ab1b5c9155a2df4162d : SUCCESS in 35m 48s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/d2777bd44d0f4c9a8da6876272a67586 : FAILURE in 1h 27m 48s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/2e1fec88349e42c891d684a9ba5975ff : SUCCESS in 1h 18m 35s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/937b0fcda11a452faec4c77fd701ecba : SUCCESS in 1h 06m 05s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/78986c1ccb6e42f7a83bb44b020e728c : SUCCESS in 57m 14s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e39c0bfff9fa4893ba36370365887a2c : SUCCESS in 31m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2e984930401a422b4a30db19e06125b2a4e1431d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-08 05:25:34.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila https://elab-os-logsrv.delllabs.net/15/887115/2/check/DellEMC_PowerFlex_Manila/6c83e41/DellEMC_PowerFlex_Manila/14 : FAILURE in 32m 05s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"44cffde09e214f19c6b84e11e20ca0b06bfde088","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-08 08:04:36.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/887115/2/upstream-check/manila-tempest-plugin-ontap-dhss/59fbd2a/ : RETRY_LIMIT in 8m 39s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/887115/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/0d6e2ca/ : TIMED_OUT in 3h 09m 57s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"5b7bbf610b97df66c60ce8679381d97ee4c79aae","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-08 08:10:17.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/887115/2/check/DellEMC_Unity_Manila_dhss_true/0dbb7e1/DellEMC_Unity_Manila_dhss_true/216 : SUCCESS in 3h 31m 27s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/2/check/DellEMC_Unity_Manila_dhss_false/61457a8/DellEMC_Unity_Manila_dhss_false/229 : SUCCESS in 57m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"955ef477ae30156f5db4922c0512d2046b756456","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-07-11 00:50:21.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"433447529492fad34b6b60a9c0f600c34b626b40","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-11 00:52:15.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-15-887115-3/ns5_manila : FAILURE in 1m 25s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"e07b6f46fd95fff52cf9fb6ed3f543f2974604dc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-11 01:15:43.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila https://elab-os-logsrv.delllabs.net/15/887115/3/check/DellEMC_PowerFlex_Manila/00fa743/DellEMC_PowerFlex_Manila/26 : FAILURE in 25m 05s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"96c676a21ccb8be523c0b67bb855a03f9c55f210","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-11 01:37:00.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/887115/3/check/DellEMC_Unity_Manila_dhss_true/fb5d10c/DellEMC_Unity_Manila_dhss_true/224 : FAILURE in 46m 21s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/3/check/DellEMC_Unity_Manila_dhss_false/e6bbabc/DellEMC_Unity_Manila_dhss_false/237 : FAILURE in 34m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"86686f8281c54d9b8f550466bd356f7dbd8a7aeb","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-11 02:04:15.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/15/887115/3/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/cf6cfc1/ : FAILURE in 1h 12m 17s","accounts_in_message":[],"_revision_number":3},{"id":"f1e6571bcd9365e256f813f9d875863a48625e79","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-11 02:14:46.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-887115-3/NFS/ : SUCCESS in 1h 24m 3s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"79b16bcd930dddc9272723ccf927de2dbf6e1923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 02:17:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0142cab1f2a54769bb213cda08f555d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5537da5330cd498f94f7eb698b47aee5 : SUCCESS in 12m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d11513228d147178be432867374a955 : FAILURE in 27m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7ce7adcc4d34b9bb1af0234a992bf7f : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/477055144ece4afa8fd4b1a091c86549 : FAILURE in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6c60543b7ab4658b1e2ee16cad955d2 : FAILURE in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9a90c70db424d4998a6d01dc12f8e41 : SUCCESS in 8m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f606282235fd4965bc7b63cde9849605 : FAILURE in 36m 10s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bae441278c684e009e078872f3d4dd7b : FAILURE in 31m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3899cf40b36849018ee15584e50e41ef : FAILURE in 56m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/05f66d678da04c51963c63e0a2871f2b : SUCCESS in 1h 18m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/60b80bdfb1554ecf94c95f41a597bf6c : FAILURE in 24m 15s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7f3fa645e1ec4324ab22ce0eb8249025 : FAILURE in 19m 32s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/0339d60c70184d6ab89144bcc3c94e35 : SUCCESS in 4m 43s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cf23e2127765454ba1cef62aa4762189 : SUCCESS in 4m 25s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/afacb3d3c09c4528b918b026d21bb6c2 : SUCCESS in 37m 00s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/70cda0d6377447f998faf99dc6d7c5ed : SUCCESS in 42m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/491573ee79e747c8b4e7e040ca1bc107 : FAILURE in 28m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/b887b00f9a024a57904ded0b87a824dc : FAILURE in 23m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ab1f9f8aa5374986a5ad5458a2934a25 : FAILURE in 23m 53s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/02ca25bf274d40a8bea32c5bfe3003c3 : FAILURE in 44m 27s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/5008d6f166464ce7b1a028f26a23c19a : SUCCESS in 1h 01m 39s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b4c87769109e421e89cce9e32b81f813 : SUCCESS in 50m 55s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b4c3b340930842d2940818d5bbc41ac6 : SUCCESS in 43m 34s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/cfdc3cc54980449999a85bca9f4b2e00 : SUCCESS in 47m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"208f51d9b126f3e61a8fe515d701a2a07d25caf0","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-07-14 19:20:01.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4706c3c63b77e5f323ffafb04dbe470b60d6782c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-07-14 21:40:55.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"dcc71a22fae679d1bba6e1931cdd2f20b224de9e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-14 22:46:21.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/15/887115/4/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/41f7d4a/ : FAILURE in 1h 02m 50s","accounts_in_message":[],"_revision_number":4},{"id":"e70e39ecbae5533746edb8e789015a639eb0f58c","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-14 23:03:40.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-15-887115-4/ns5_manila : FAILURE in 1h 22m 15s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"1575d4962217c5688de40867bb228c7cddfead34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-14 23:23:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2db52f05f2784f4b8c70cc78a5d1e325\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fa28dea6c8b4260b23a2b742e04fbab : SUCCESS in 10m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87b769a3db6348bebd8a8b86a1c55179 : FAILURE in 23m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ddec05b0c854f05ae46456af4ae56ea : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6f4eddba8c24cd0b59bb557d22e35ba : SUCCESS in 8m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fdbf3b00916448499dc554325335783 : SUCCESS in 9m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2482073611414281b75c773b528e5b38 : SUCCESS in 7m 27s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1789a5d672b14721992a3c5bd35a6297 : SUCCESS in 1h 41m 21s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5a964e79bac04b4987c8c1b551e468de : FAILURE in 50m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/82a795538777425c8aaa52ba8c0b222b : FAILURE in 30m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/901d25d6cb01459a80164c1dfa1310e1 : SUCCESS in 1h 06m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/7778860fa5e04aaf9660617168cca5d7 : SUCCESS in 48m 52s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e7234427a0f447de84120ea19a162f00 : FAILURE in 13m 32s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/76f499faa525462694c3efb6104e0ffa : SUCCESS in 5m 43s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2764656af8314fb495d9783088862b1e : SUCCESS in 4m 44s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/954cefb491be4b55b4f17f5afb870e7f : SUCCESS in 28m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/11e5a7b2a9da4d1e82702c6a6ff7c4fb : SUCCESS in 36m 06s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/14ca2eda196b41a1ae5ce069a192ef28 : FAILURE in 18m 11s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/e171afc345cb4cb6bf05d79b0b36d31d : SUCCESS in 31m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/19f169646dc6439f8b69049c3b2e4b4d : SUCCESS in 20m 35s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/ca3b17d05dc642f69f2c4cd38275d93e : SUCCESS in 1h 07m 04s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/bfa14e1043a74c1199085d1984d3dc13 : SUCCESS in 52m 43s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/68298c3ef00345b79bb5103823032b0e : SUCCESS in 50m 31s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6cc6af801b704fe4b60ddd8e18b65a7f : SUCCESS in 29m 03s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f875bb0265fb4e62bb6bae4486645b1e : SUCCESS in 33m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4469ace0ee28000320baf0327f47228b44d4c91f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-15 00:15:01.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/887115/4/check/DellEMC_Unity_Manila_dhss_true/0d9c041/DellEMC_Unity_Manila_dhss_true/239 : SUCCESS in 2h 33m 48s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/4/check/DellEMC_Unity_Manila_dhss_false/aeeda1f/DellEMC_Unity_Manila_dhss_false/255 : SUCCESS in 1h 02m 58s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"6f1e2f5cfb7327a31f1ec3b116bb5d226e2717bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-07-15 01:22:21.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"f01178ad2b10ea68b705e95083aedba6841de422","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-15 02:33:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/726eb700437242d1b7d4c6b7cb89df6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/406ff3b96b114cb1ab0aab05a33ca0b4 : SUCCESS in 8m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cba288c54794e3d857a381b4b38e603 : FAILURE in 16m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9ef70eed38d46598df429faa2dba8c6 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5fe52cf9c99407499daa8cac4b101cd : SUCCESS in 8m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/befc00e262f84d3e81a98a404a9f053f : SUCCESS in 9m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/73cfe71e0b8040b68669066b655d3d2b : SUCCESS in 7m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1a5ef8f388724a3f99896536a9214eda : SUCCESS in 56m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2bd5b433a176445794b3a2bd0c3f24da : FAILURE in 26m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/230bb90cf4cf4ecc862137d12cda4e1e : FAILURE in 29m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d0615ed4164849a7bfd6eb54877a8150 : SUCCESS in 1h 03m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a555bfa5791e44bfa985a316edd4c027 : SUCCESS in 59m 42s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/251688bf92dd406da5d707923021325d : FAILURE in 28m 46s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/22cbaf47d2b94616b653dc3f7c43d5dc : SUCCESS in 4m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/59feaf7f1b324ca58d04f93909bec7f1 : SUCCESS in 5m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4433fb58f18744ef83cd6cfee745ceb3 : SUCCESS in 26m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e03bf2a0c9934bb2a8c1085c49c447a3 : SUCCESS in 27m 47s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cc5dd8ace73c460f8af429256d501f31 : FAILURE in 19m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/246e7daf93a949dda3a123eedb3411f8 : SUCCESS in 39m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/cd7510288296414bac6ee49775b1924e : SUCCESS in 29m 38s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/905846d372f94c05afbbc107a8f5c44a : FAILURE in 1h 01m 41s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/240d2a7046364d2a8643f05649c12d60 : SUCCESS in 1h 02m 19s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6d581edbd2404fb69a5f4d37b368d85a : SUCCESS in 1h 04m 33s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/db0c183aaad443628eb9a2c79450452e : SUCCESS in 43m 51s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1fcb6264d99b4fecb09dc8d313968293 : SUCCESS in 46m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a8a028799fb67702f217ff1fa249faad8d7df5b8","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-15 02:38:27.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-15-887115-5/ns5_manila : SUCCESS in 1h 15m 38s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":5},{"id":"915e9a5e9dacf7a9c8c61781da8f80d091d2d829","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-15 02:39:27.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/15/887115/5/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/d8ec765/ : SUCCESS in 1h 14m 08s","accounts_in_message":[],"_revision_number":5},{"id":"a18c953726a02ed393d4466010196c3f89177e9d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-15 03:33:31.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/887115/5/check/DellEMC_Unity_Manila_dhss_true/f91b815/DellEMC_Unity_Manila_dhss_true/241 : FAILURE in 2h 10m 50s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/5/check/DellEMC_Unity_Manila_dhss_false/739a34b/DellEMC_Unity_Manila_dhss_false/257 : SUCCESS in 1h 01m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"be4b9c50f5a5c2b241e0596d2ea1f3303f54e291","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-15 05:46:52.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/887115/5/upstream-check/manila-tempest-plugin-ontap-dhss/01af6ff/ : TIMED_OUT in 3h 12m 22s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/887115/5/upstream-check/manila-tempest-plugin-ontap-no-dhss/311b50e/ : SUCCESS in 1h 39m 29s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"2a8e09fe082abddcbcd95f6a96b00925571048b8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-15 16:00:54.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-887115-5/NFS/ : SUCCESS in 1h 17m 22s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"311621ea9a8f82d99c6c291eee677f12dc484e82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-07-20 08:11:41.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"d0bd0ff7c04eee0400371791b542bed8968fe8af","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-07-20 08:13:07.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8081d67ce8994ac02292f02068822be85b7c8ce4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-20 08:14:32.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/6/check/DellEMC_PowerFlex_Manila_dhss_false/fe3d9c9/DellEMC_PowerFlex_Manila_dhss_false/63 : ABORTED in 2m 35s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"71b49242f38c127bfe1aa1772922b5558877ac10","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-20 08:59:34.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-15-887115-6/ns5_manila : FAILURE in 47m 23s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":6},{"id":"2758858a873899cb55850a58b2e0065ca4b3031b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-20 09:13:55.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/15/887115/6/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/34e888a/ : FAILURE in 1h 00m 44s","accounts_in_message":[],"_revision_number":6},{"id":"de8e783ef247de07d2feadda493d6439380b1167","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-20 09:39:00.000000000","message":"Patch Set 6:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-887115-6/NFS/ : FAILURE in 1h 26m 58s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"7f5d074146c956d8b8a795c4949d3b5d2b0b619f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 09:58:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94d7a28a1af14cfd9d00e4feca1b126f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fb4f7cfb84094254a07272fddfa3d633 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/012d8f76264f498bae512ad26b76a7d2 : SUCCESS in 8m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/805da39d5408489085dd53cb3255bc37 : SUCCESS in 23m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3e297f0ea744f5f8155ff947eaa6deb : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca0f5e2f8e7542d58e2a3a7f010e900d : SUCCESS in 10m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bab23cbc31a04855bb5f94e1d99d357a : SUCCESS in 10m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae7139ef9fe54051b0b48f1a73af4dae : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e9226c61ef7492c9181205e98716536 : SUCCESS in 7m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c48aaefbd61a4e0b9d56317178affc66 : SUCCESS in 1h 11m 17s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f778ae2ecf9d4768a52b2a41b39301e9 : FAILURE in 40m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/593aa649c035424eb9fb348920b32f65 : FAILURE in 53m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c55dae5a2bea439abb0b33fc1230bf39 : SUCCESS in 57m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/ffa4043023a244dbba44cc230c8342b7 : SUCCESS in 1h 31m 52s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6d46e47b311a47d4969cc69f9fd4d4a6 : FAILURE in 29m 57s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/06cbd8bcbd7f4932a65d1b703e771d11 : SUCCESS in 5m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/54502ed9f7164afa99730f23a44a937a : FAILURE in 28m 18s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5a5ff2a3b5bc4a9f8c84f931a41a2f8a : SUCCESS in 25m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/26f88fe453104715952b2a0ff9daae76 : SUCCESS in 53m 50s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d0b07f22056242c9a75a01914f9000c9 : FAILURE in 38m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/841fde2be2ee46af9e899c4d2f8e3a34 : SUCCESS in 41m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d1561effb2114202a5d710cb0e65c708 : SUCCESS in 23m 58s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/1ad085d6047f48d889e1745e60063954 : SUCCESS in 1h 39m 09s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3d709f91b4cb4d7697cc57270e27d54b : SUCCESS in 49m 17s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c17e1b46354f400292788910afdcc246 : SUCCESS in 1h 10m 53s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/8546cbb174aa4b12bc99a88bf667afe3 : SUCCESS in 56m 18s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/93bb4cd1cdc24e82b1f16dd8025d36ef : SUCCESS in 48m 46s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"0ca10ff53b0e18a883b7a641b3e37518ec5c51f5","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-07-20 09:59:13.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila https://elab-os-logsrv.delllabs.net/15/887115/6/check/DellEMC_PowerMAX_Manila/1fdf884/DellEMC_PowerMAX_Manila/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":6},{"id":"ae12e1df31f43f31ca5679b545be1e82a8244eed","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-20 12:11:14.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/887115/6/check/DellEMC_Unity_Manila_dhss_true/0699fa4/DellEMC_Unity_Manila_dhss_true/256 : SUCCESS in 3h 59m 16s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/6/check/DellEMC_Unity_Manila_dhss_false/6950f4b/DellEMC_Unity_Manila_dhss_false/274 : SUCCESS in 1h 10m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"a49e9e3ec06bdd7a9b5b7f514f05fb535a7a18f9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-20 12:49:26.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/6/check/DellEMC_PowerFlex_Manila_dhss_false/37c81b7/DellEMC_PowerFlex_Manila_dhss_false/65 : ABORTED in 2h 50m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"d5b2372bc27e5ff329350f55b6fa77fd613f3ca3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-20 12:56:59.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-887115-6/NFS/ : SUCCESS in 2h 2m 6s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"5183f576c7a0b754b76a2a1dcdacf11961e06726","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-20 13:14:07.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/887115/6/upstream-check/manila-tempest-plugin-ontap-dhss/598c043/ : TIMED_OUT in 3h 13m 40s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/887115/6/upstream-check/manila-tempest-plugin-ontap-no-dhss/0b26ed0/ : SUCCESS in 2h 51m 12s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"594752ca16665555554a8305e6e1542606fbf284","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-07-20 23:27:06.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"215a173914af3b363dbedcf30f0f3ad792497c6e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-20 23:57:48.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/7/check/DellEMC_PowerFlex_Manila_dhss_false/5a13843/DellEMC_PowerFlex_Manila_dhss_false/71 : FAILURE in 30m 25s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"f7e593758ad436b2b6a9642ad332379f116e2379","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-21 00:14:54.000000000","message":"Patch Set 7:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-15-887115-7/ns5_manila : FAILURE in 47m 21s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":7},{"id":"52b21c7fc495e4dff0496f0161092c446d66bb8d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-21 00:31:47.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/15/887115/7/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/30d8f7d/ : SUCCESS in 1h 02m 51s","accounts_in_message":[],"_revision_number":7},{"id":"1b717331e8e96c276339171a4cda5bf2f6287071","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-21 01:05:01.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-887115-7/NFS/ : SUCCESS in 1h 37m 36s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"2a44885c888ff3ecacdd077bd27c7d264c762e88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-21 02:05:59.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e77b127cbac43b696722a42c692bd37\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7d4d37e78b0347b7b2a2ad0b33112c48 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b30dca75fd2e4094b2c4d6cd78907414 : SUCCESS in 10m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e3be1c536bf410299ba0107e10722de : SUCCESS in 37m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6db8c5a68fdc450d871264edbb6b17e1 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aee6c87deaff4f919f6bfad6fdb1ac7d : SUCCESS in 17m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae9348ad73914a849f5531d5deea9930 : SUCCESS in 9m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee9dae04538d48479afc145d85e7de75 : SUCCESS in 14m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51c866728c204b7ba2e7e953c8403cb2 : SUCCESS in 5m 22s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/55f68148ec90455186ad50cd15f6f594 : SUCCESS in 1h 44m 42s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7618e1f0b9fc4edba53efcf915d1785d : FAILURE in 49m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/22a1cc3abd9c4ac497214f537c0833fb : FAILURE in 47m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d4e80a2630034bbfa30c465039aaefa9 : SUCCESS in 1h 15m 11s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/89705f06b2db40349918693e70a50f36 : FAILURE in 1h 16m 53s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/97dc400cb5054b00b62b00358d52bc25 : FAILURE in 1h 09m 41s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5e45ba1c196745c0a2fc3625aa22214c : SUCCESS in 5m 33s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/af5ab23808a54d51819109463c04d1d4 : FAILURE in 15m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8dfeefd2967d482fa473ec67e858769e : SUCCESS in 26m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3dbf7a5fd9b94142b3b20a27a58ba63c : SUCCESS in 33m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/40a7210c0d82449abcbd14d43cc3be49 : SUCCESS in 40m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/9af7af9276db4e8a8c4d13a330e1cbb5 : SUCCESS in 24m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ffe477d3d2454a7aafe7498c34a31602 : SUCCESS in 18m 38s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/8cad8a0e391047ec81d2fc27bc0d9848 : SUCCESS in 1h 04m 08s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/b093bdd0b17040f8b8924642419146f1 : SUCCESS in 2h 35m 16s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2ce3d789a98c4a459ab1838b36c13c41 : SUCCESS in 53m 47s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/34c6ae418f294f77bce82bba8b0e1f05 : SUCCESS in 26m 42s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/58442d6c6bc04364b616c7d0acdca7ef : SUCCESS in 43m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3c5003c6e012ff3c35ae43e1ff60f438db659260","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-21 02:37:04.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/7/check/DellEMC_PowerFlex_Manila_dhss_false/fa888d1/DellEMC_PowerFlex_Manila_dhss_false/72 : FAILURE in 30m 49s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"0265a554c5bc3e9911b263b044fe4ffc58b18108","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-21 03:06:37.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/887115/7/check/DellEMC_Unity_Manila_dhss_true/ca0fcbd/DellEMC_Unity_Manila_dhss_true/262 : SUCCESS in 2h 25m 08s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/7/check/DellEMC_Unity_Manila_dhss_false/85d0e04/DellEMC_Unity_Manila_dhss_false/280 : SUCCESS in 1h 13m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"37db67713265a5377969a747b3b30f695db2f23c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-21 03:40:18.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/887115/7/upstream-check/manila-tempest-plugin-ontap-dhss/d978b55/ : RETRY_LIMIT in 8m 20s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/887115/7/upstream-check/manila-tempest-plugin-ontap-no-dhss/481e1c6/ : SUCCESS in 1h 33m 05s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"73516f8487579be17f468a6408107f3853e3f201","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-07-26 22:55:33.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"b64a6065914fa1ce87dbc02ee26c617c6c2c9620","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-26 23:45:32.000000000","message":"Patch Set 8:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-15-887115-8/ns5_manila : FAILURE in 49m 27s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":8},{"id":"8c735a2ca90b6a1a11f8f67deadf2364eb6fba7f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-27 00:05:01.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/15/887115/8/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/e4757e1/ : FAILURE in 1h 08m 27s","accounts_in_message":[],"_revision_number":8},{"id":"89542cc2f17f910c9e4afa5c840a3864847f6028","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-27 00:41:24.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-887115-8/NFS/ : SUCCESS in 1h 45m 32s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"fa950b9191cede84c96983129e1768eb065ce143","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 00:53:36.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f9dccbed20f471684b80898c57b3135\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d0e3242a68a248e79936cae61a17b94d : FAILURE in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cde8d59a73b34d2ba02c71ffa4a4be53 : SUCCESS in 10m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa241da291b34b0b952f64d59a68692b : SUCCESS in 24m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fc7951843e145c8aaa09538365e469b : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e42dde2b2dc4b4a8a9e53046321672a : SUCCESS in 9m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7122847f1c744d12b68188ce4ecb0b05 : SUCCESS in 11m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf20df76d05f4490b0f0e3b356834dbd : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb255868756749b4b891da7ba568f5d5 : SUCCESS in 6m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/959a10fb9e854de9bdb20dc24ffd9b28 : SUCCESS in 1h 50m 55s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bac5f949b9aa4a8f9914210c9905f7bd : FAILURE in 50m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b7e98925cbf84f9f9c864695ab253719 : FAILURE in 29m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3f419d9a19354b82a7c8bb0296c52779 : SUCCESS in 1h 19m 09s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8a720b0983ac419e9296c60681132f7b : SUCCESS in 1h 21m 45s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/15fe867aaf7b4b27a0e81458c5588c56 : FAILURE in 49m 41s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/dbb1480127f14fb9a9d428c2bce708aa : SUCCESS in 5m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9f462837cb4940078d86f96cec9ae2b5 : FAILURE in 24m 08s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3752540283434b9586f2e742cc3334e5 : SUCCESS in 25m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f8867ebdc52542f8a57936af33cdc3b5 : SUCCESS in 48m 29s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/29eb4e6630e34c888ef08cd3db74aa94 : FAILURE in 37m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/89bfe63d16a84234ad21a13a5c7d7c63 : SUCCESS in 23m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/81afd355486944d5a1c1caad9a118eaa : SUCCESS in 17m 30s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/a94327ae269f46f993a4b964be77d048 : SUCCESS in 1h 17m 39s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/88359f2463b54e6696846f3a88765f07 : SUCCESS in 47m 03s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1902d8dd88a44c729bfa206cd15d325d : SUCCESS in 1h 02m 46s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/45ccf17f606348c68a4edaaf7dce5035 : SUCCESS in 26m 35s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1c264970da2341a0a4f26e91285f9a5e : SUCCESS in 48m 07s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e3ac0e4d25936e8d58b00da4d123751bd22da7e2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-27 14:08:48.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/887115/8/check/DellEMC_Unity_Manila_dhss_true/e88c596/DellEMC_Unity_Manila_dhss_true/282 : SUCCESS in 3h 04m 15s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/8/check/DellEMC_Unity_Manila_dhss_false/2321ed0/DellEMC_Unity_Manila_dhss_false/304 : SUCCESS in 58m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"a36727f28b041b1618955885a7c18f28ebdfb00b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-07-27 16:46:35.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"941bf632160173cb47e20c2848cb56de2be85c7a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-07-27 16:48:12.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy pure-devstack-manila-tempest-aio-jammy : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"d546c14f599590b6fe7d857b06aea9aa802056ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-27 18:21:59.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36bde82129234c3984deb4484ccbea77\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/db3a5f0e5bb842a3b71edd42f44b8ae8 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a152f799f69443d8f9c94772d503aa6 : SUCCESS in 10m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0d707de3eaa4d2db012447b4f0c0558 : SUCCESS in 23m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e25566950fc4d4194c4f3b3a3bbb5af : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0cf585483f4461b9eb2fca588350c8e : SUCCESS in 11m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/291478c9f13a436489d12c32fe65f047 : SUCCESS in 9m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bff0500d799f4d049dfa84c151c34b90 : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/086a0d6444d446989c29322478adc0a6 : SUCCESS in 6m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b88b0e4f84414f8198b5dcad47a1141b : SUCCESS in 1h 28m 02s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2b2628454701455bb64b1abc16b2dac6 : FAILURE in 52m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/257f46c9b469446e942704e1a8991fa8 : FAILURE in 37m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/cabb1203524a477aa64212e48633765d : SUCCESS in 55m 07s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0344753f9711440484b067278d1bcb96 : FAILURE in 1h 27m 43s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/12c813c950544183905470e622baa64d : SUCCESS in 1h 08m 07s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/25b833b7d7c747beb0e72c032eefc5bf : SUCCESS in 5m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/463aabf9d9ea4a298b2a5096529c5e5d : FAILURE in 20m 31s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/04abe18b214b4735bb0149ae85f54799 : SUCCESS in 25m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a40dd041a8dc41779a56288ef7eecc49 : SUCCESS in 38m 59s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6c338773911544dbb705673c14186f9e : FAILURE in 56m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/2b343168653e4b0d8f38313d75489386 : SUCCESS in 31m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/70f34138fdf247a7ba86508ed1d395e8 : SUCCESS in 18m 08s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/ff0a9615003d4701a75fdc7714642ab7 : SUCCESS in 1h 05m 40s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d3dfcee1c3594f95875df6d9cad97d6a : SUCCESS in 1h 10m 00s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4484d449077940ee9265cfdb17a506dd : SUCCESS in 1h 02m 14s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/71ace1fd2d79434bb75f9ab3861df118 : SUCCESS in 25m 01s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7c354d747a0449d7bca77a14ea099750 : SUCCESS in 39m 11s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"03e633cad6af2482bf2eebb456627988c19b2d72","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-07-27 18:54:00.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/9/check/DellEMC_PowerStore_Manila_dhss_false/62e174b/DellEMC_PowerStore_Manila_dhss_false/24 : FAILURE in 31m 42s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"9b38b8094412199de5905a4e1b09d09e8c1f4d8f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-07-27 19:00:51.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-887115-9/NFS/ : SUCCESS in 1h 57m 41s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"d32e02426efa88d41ec835a1183cb05fe1ab1e71","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-07-27 20:21:12.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-15-887115-9/ns5_manila : SUCCESS in 2h 24m 58s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":9},{"id":"f513e8c969e1dcab09fb93718433e200934a4ed9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-07-27 21:32:01.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/887115/9/upstream-check/manila-tempest-plugin-ontap-dhss/2e648b4/ : SUCCESS in 2h 19m 56s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/887115/9/upstream-check/manila-tempest-plugin-ontap-no-dhss/640d871/ : TIMED_OUT in 3h 08m 53s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"90e7110530259e88b1de02487ffb07f48c732272","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-07-28 06:40:20.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/887115/9/check/DellEMC_Unity_Manila_dhss_true/a1d8d2a/DellEMC_Unity_Manila_dhss_true/288 : SUCCESS in 3h 02m 28s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/9/check/DellEMC_Unity_Manila_dhss_false/a9e6b22/DellEMC_Unity_Manila_dhss_false/311 : SUCCESS in 1h 15m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"0a6aff6c2dbf6008268ba45b15ff897ba246b6b0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-07-28 08:15:02.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/9/check/DellEMC_PowerFlex_Manila_dhss_false/f045792/DellEMC_PowerFlex_Manila_dhss_false/106 : FAILURE in 30m 32s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"1f7569b11934a40a71156153bbe9abd3eee191fd","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2023-08-10 13:42:06.000000000","message":"Patch Set 9: Code-Review-1\n\n(23 comments)","accounts_in_message":[],"_revision_number":9},{"id":"9f3e6a2ce0af88a8543ff69d5529f46e5a345abc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-12 19:03:30.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"b2ef3b539a769fda70b1a696f2e11d38fc2c9da7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-12 19:03:38.000000000","message":"Patch Set 10:\n\n(24 comments)","accounts_in_message":[],"_revision_number":10},{"id":"456b5ef8effc1ef0159f5571722fd85d3ffb8a3e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-12 19:07:30.000000000","message":"Patch Set 10:\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"4151361ab7582803bc9767cb2dacbb175305e9f1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-12 19:16:27.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/10/check/DellEMC_PowerFlex_Manila_dhss_false/80769d4/DellEMC_PowerFlex_Manila_dhss_false/214 : FAILURE in 12m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"e82e52c31278832a2891cf2d89eb45d171744c82","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-12 19:27:07.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/887115/10/check/DellEMC_Unity_Manila_dhss_true/78c139f/DellEMC_Unity_Manila_dhss_true/362 : FAILURE in 23m 19s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/10/check/DellEMC_Unity_Manila_dhss_false/8494ca1/DellEMC_Unity_Manila_dhss_false/387 : FAILURE in 13m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"03821efb7bc01143192533fbd409fe4477196464","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-12 19:32:04.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"0dd57595a409fce8b63e5f05ac62289c6c332f0e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-12 19:44:55.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/11/check/DellEMC_PowerFlex_Manila_dhss_false/a935829/DellEMC_PowerFlex_Manila_dhss_false/215 : FAILURE in 12m 35s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"f0f07b437508e08d34ec5bc16c4aec359f130b1c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-12 19:54:16.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/887115/11/check/DellEMC_Unity_Manila_dhss_true/f9dd4c7/DellEMC_Unity_Manila_dhss_true/363 : FAILURE in 21m 55s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/11/check/DellEMC_Unity_Manila_dhss_false/10de0ad/DellEMC_Unity_Manila_dhss_false/388 : FAILURE in 12m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"00df6fcfb98300463a7a71da61a73c5e385ea1c1","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-08-12 20:18:28.000000000","message":"Patch Set 10:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-15-887115-10/ns5_manila : FAILURE in 1h 14m 26s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":10},{"id":"5805de912c21653a75d51066f50378711d3dec92","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-12 20:40:27.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/15/887115/11/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/7346c23/ : FAILURE in 1h 07m 29s","accounts_in_message":[],"_revision_number":11},{"id":"830f511985d1741b10c313bd36cf3d24bd610bc3","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-08-12 20:47:42.000000000","message":"Patch Set 11:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-15-887115-11/ns5_manila : FAILURE in 1h 15m 8s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":11},{"id":"b075378edec12f6fe5a76334d06926016a2b18d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-12 21:02:13.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e44294e9a0164a078d58b0a1d459f32f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9b5d0e41b3eb47dbbab7e9d303cf96bb : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8e7cbddc15a46568e7ae434545049b2 : SUCCESS in 11m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cf99593f21b4748ae412adbfe1e3d37 : SUCCESS in 19m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdb8a836a3be47a3978ae24d23456f31 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6479ddf9362f41e18d1efc6f69aba543 : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd0c59055c32405ca839fcd84269bd63 : SUCCESS in 11m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e61bff07b67544948308e5239b6c8437 : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdeb187248a1414eb8e5bbc8bd9941f6 : SUCCESS in 6m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/55790f8a24f546bfb8f29d40c8f9708c : SUCCESS in 1h 24m 24s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/70c2a6517d6a4887bc9d34d4c4fec35a : FAILURE in 31m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/77778ebc3e3a49dc8ddde2789ffe703b : FAILURE in 31m 15s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/360cd2873d1d46ecb330cc48a169a0c5 : SUCCESS in 48m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/372756a94b2b4e5aa6d74bb3c61316d2 : FAILURE in 31m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d0d37bff1ed449909a8daadafc7164ee : FAILURE in 53m 55s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/56994eb3456e44228dd30f449093ca0f : SUCCESS in 4m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/42839fc30df34ec3953dd8a5e6a2f03b : FAILURE in 15m 27s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d47ef526d7f04b489bdb6eadc262a7e5 : FAILURE in 23m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5d2a3f9256314315a38306334ea623b4 : FAILURE in 22m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f0f032103b7443419baa332284bdf086 : SUCCESS in 40m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/c512859a88cf42a2b877e2e0602b0df9 : FAILURE in 13m 43s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/325304574cd443aba3c242cb920699df : FAILURE in 12m 12s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/5be8dc24c61f406a84ee06963a87980d : FAILURE in 22m 01s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/f9fa9c8098d0476fa7a60185fd0570d4 : SUCCESS in 39m 47s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/406bd7fceae64d43bb253057531a2de3 : SUCCESS in 58m 46s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1920de2984dd4ba9af381d6abaac50c0 : FAILURE in 23m 57s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/bb22ba99a93f4be4bce69b14ab465bee : FAILURE in 22m 30s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"3688609006e0876bf6be6b41accb0d451e810216","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-15 17:53:56.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"aeb2dfbffc6800d802186583db493c30bb309587","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 19:28:18.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c47c299d13ca476b8d665ccc325b7a87\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/26b9752d5ea24ab394e2f60834d231d3 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1cefa4a50be4ffaa8c6df0e984c9433 : SUCCESS in 10m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3091242ad4054af4854a8cad567b01a0 : SUCCESS in 27m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf28e4c859d14b5fb5b3aecbb6360f27 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc046d12e2c64903a2ff0ed815fdca18 : SUCCESS in 9m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9f85548180d4d1fb7762b63f493ab7c : SUCCESS in 9m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/092e1ab4729444cf8d1b24c8d0cf1c2f : SUCCESS in 9m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8cea5e47c4e441e86d4943df468a590 : SUCCESS in 6m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6000076f551b4e1194d67a2c4d39e68c : SUCCESS in 1h 21m 24s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/dc77ea72c1e44da1abf4c19af36e3319 : FAILURE in 1h 05m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6bc2e220b56445aa98aad81d5e531134 : FAILURE in 44m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/65fc3547d43046cb93b5fe0d84b4e042 : FAILURE in 1h 21m 28s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6ac2a42a3a584252b14a87ecb84c6653 : FAILURE in 17m 43s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f833b8a361014a57a6837e86d35684d6 : SUCCESS in 1h 00m 16s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a295ea225970431fb7170f5dd0bfc1cb : SUCCESS in 5m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/329078558ec34cc4bb53a3ca59185301 : FAILURE in 19m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9508af510a5e4c12b714cb83d5c362d5 : SUCCESS in 44m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3daad43889314fdab081c2fa196e97f5 : SUCCESS in 45m 27s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/94561d80026047cfada8be01031f8c52 : SUCCESS in 44m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/3d538ea5bdea44c781350def2ec6125a : FAILURE in 15m 32s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6f1ff058863948ff8cd61b3ad8eec90a : FAILURE in 37m 01s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/047890309cd64b58b2f39e30cc7ee3b1 : FAILURE in 21m 34s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a9a33a364f6e40b0bd1927ea4c26a7ab : FAILURE in 49m 00s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/78b97db3b5c147218048fa3a427d6fb1 : SUCCESS in 1h 08m 22s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/45a22da88a524edb8be16b1f69beee97 : FAILURE in 47m 21s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3d50f426448b4b66a7f5bc2492624fdf : FAILURE in 35m 15s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"98f11ca1b1b6df00dc6407e0e70e68624f674234","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-15 19:38:42.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/11/check/DellEMC_PowerStore_Manila_dhss_false/f739942/DellEMC_PowerStore_Manila_dhss_false/73 : FAILURE in 7m 54s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":11},{"id":"87af957b3c5bf57158e1fd18f2517561e21c655e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-15 20:04:35.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/887115/11/check/DellEMC_Unity_Manila_dhss_true/5a3b013/DellEMC_Unity_Manila_dhss_true/376 : FAILURE in 20m 22s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/11/check/DellEMC_Unity_Manila_dhss_false/09b4ab6/DellEMC_Unity_Manila_dhss_false/401 : FAILURE in 12m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"4a3a8e82485cb653b2163e0ae14d5a8a0b585fb4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-15 22:39:35.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/887115/11/upstream-check/manila-tempest-plugin-ontap-dhss/655cd9b/ : FAILURE in 2h 33m 48s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/887115/11/upstream-check/manila-tempest-plugin-ontap-no-dhss/64b3e43/ : TIMED_OUT in 3h 10m 27s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"b456f843a83c9a12c0a105cbe8f91347767c3a44","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-08-17 12:42:42.000000000","message":"Patch Set 11: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":11},{"id":"c6062eef578948b3c694b35116a3c361d1236e38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-17 22:05:13.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"fe34a103f35ba5ce6243ee6731c34b57520259f9","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-17 22:05:58.000000000","message":"Patch Set 11:\n\n(8 comments)","accounts_in_message":[],"_revision_number":11},{"id":"8a4f1ae84a97afcef88f611a3a31f426de856d1a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-17 22:32:16.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/887115/12/check/DellEMC_Unity_Manila_dhss_true/87d628a/DellEMC_Unity_Manila_dhss_true/411 : FAILURE in 26m 45s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/12/check/DellEMC_Unity_Manila_dhss_false/d07e2fc/DellEMC_Unity_Manila_dhss_false/436 : FAILURE in 12m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"5bebe278bd0fbf7abaacd177a68ed01a8010ee90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 23:03:23.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4616a5af28343eeadb86f554aff16a5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d026714902194b7daef8c2d3dbe7f3bd : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66b21969b45d42168af88bff829785c2 : SUCCESS in 10m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f2f25c4b81d477497ec1351194ce379 : SUCCESS in 24m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ff8e03d4dd445c38fbeaaec2707c086 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c81ba2f087d46b3817109a2dd90cc99 : SUCCESS in 9m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6869a78c1694639962993ba573df255 : SUCCESS in 9m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9bd6d38231f441d08143688a7f1017b6 : SUCCESS in 10m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/04426357c7ef40dba5f807797cc2bd15 : RETRY_LIMIT in 1m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46bcaa3f3f1240cdb2f0bf3ff8344504 : SUCCESS in 7m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e3d85cea7de54575aba448c9c793912a : SUCCESS in 52m 10s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/eaccc590b5a34c82b79b2ed477399413 : FAILURE in 36m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/143bb64e8f5d4234b2c3a2ddb248a9f2 : FAILURE in 27m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e296531b5b9a4061a8bbd99dc2a5e28c : SUCCESS in 56m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/9271f3573f694c6ab617aae720d988ef : FAILURE in 16m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7806b2d4ff72449da7dfa9e3e1b579b6 : SUCCESS in 49m 42s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f31734c4c5ca4166ad66610e36f60171 : SUCCESS in 5m 11s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e2cdbb5b4ff2467c8a5cb38bc9087e75 : FAILURE in 18m 23s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/096a1cf7201342c49f77c0fc41ecbea5 : SUCCESS in 27m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/05e2ff1444044fe5a8b151791b886338 : SUCCESS in 28m 35s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4c59086c43ee43a8bebc0f09b6813540 : SUCCESS in 42m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/ee0ae581c4f548a889a2fd1efa7c7f6e : FAILURE in 17m 11s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/da604a30b1f14166912880d07a0b601d : FAILURE in 18m 40s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/cf9261f913d84b2e82a2044666d1def8 : FAILURE in 18m 44s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/98f54f13d8674b5b9a6814c9a62445a5 : SUCCESS in 40m 56s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9f6d6db4d1e14238b8cb0f2f1114c268 : SUCCESS in 44m 59s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2a6b872eef7043348c3f81b53f2f0563 : FAILURE in 20m 46s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ce28ab812100435ba185be88e68a922d : FAILURE in 22m 05s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"fc9f35e5fff927f9783ad4797b2e8cd685209b51","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-17 23:12:17.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/15/887115/12/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/4b6ea01/ : SUCCESS in 1h 05m 32s","accounts_in_message":[],"_revision_number":12},{"id":"79176b3b7c71a00726e925bf0ca3e3a56b0b5fac","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-17 23:17:41.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/12/check/DellEMC_PowerStore_Manila_dhss_false/160bf48/DellEMC_PowerStore_Manila_dhss_false/87 : FAILURE in 14m 02s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":12},{"id":"ed42e8b1f085897d263e09cc5fac8b8bf6ebb80b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-08-17 23:23:07.000000000","message":"Patch Set 12:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-15-887115-12/ns5_manila : SUCCESS in 1h 17m 18s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":12},{"id":"6dc4f25fb3cf2c31bfc832d330187dadc12b86c0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-17 23:28:21.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/887115/12/check/DellEMC_Unity_Manila_dhss_true/d4df625/DellEMC_Unity_Manila_dhss_true/413 : FAILURE in 24m 41s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/12/check/DellEMC_Unity_Manila_dhss_false/b1fd7af/DellEMC_Unity_Manila_dhss_false/438 : FAILURE in 12m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"19efd52331373b920c49adcb408e5e2fd3f061ee","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-18 02:19:12.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/887115/12/upstream-check/manila-tempest-plugin-ontap-dhss/886d9a0/ : TIMED_OUT in 3h 15m 02s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/887115/12/upstream-check/manila-tempest-plugin-ontap-no-dhss/e6bba9c/ : FAILURE in 1h 34m 46s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"5d006ef638770b3406d0f748cdb6a625ccba4f3f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-08-18 02:30:49.000000000","message":"Patch Set 12:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-887115-12/NFS/ : FAILURE in 2h 8m 42s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":12},{"id":"90ab071048e4bef189dd607776e5793c2b44489e","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-08-18 10:01:07.000000000","message":"Patch Set 12: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":12},{"id":"625f0239eece795cec145e8352f6931cf2c1bfa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-18 17:47:47.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"0923fa1473b9ae26d4039ebf5e47888e82d6304c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-18 17:51:57.000000000","message":"Patch Set 13:\n\n(6 comments)","accounts_in_message":[],"_revision_number":13},{"id":"979d45264f948ede7ccd48f71b2a5992da681463","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-18 18:15:14.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/887115/13/check/DellEMC_Unity_Manila_dhss_true/112ecce/DellEMC_Unity_Manila_dhss_true/427 : FAILURE in 27m 08s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/13/check/DellEMC_Unity_Manila_dhss_false/56f208e/DellEMC_Unity_Manila_dhss_false/453 : FAILURE in 12m 14s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"c3ddc50f0d1ac3c1e9a6823f90b4f4c9b99bb82a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-18 18:49:26.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/15/887115/13/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/5f4bfc2/ : SUCCESS in 59m 32s","accounts_in_message":[],"_revision_number":13},{"id":"06901aebe8e7e5d73ff0b349fe3d58af39f5d081","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-08-18 19:00:09.000000000","message":"Patch Set 13:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-15-887115-13/ns5_manila : SUCCESS in 1h 11m 51s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":13},{"id":"10c6aaab199e72c9ce2b76de07efa85dd64fcb89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 19:11:49.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e23175260c564233b720b8c0895d62d8\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/68799e7c55d7486b8ec84e5e7c90fb66 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf4fd9c2210e4b91bffb2da12aa26e1f : SUCCESS in 8m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/292c60337242406690d10dfc58ccba11 : SUCCESS in 19m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9043ae3f1b4b41768ff369d96fe13aae : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b79ab1cc2b94676bb15544e87300b74 : SUCCESS in 9m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7de68a2546e74dd5b5ce8e633ba1f7ec : SUCCESS in 9m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39737e5d89e641a89656de431f771124 : SUCCESS in 10m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/533e7f972f7947d29943862b30fa8aac : RETRY_LIMIT in 1m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0484b4d4592d487597fa7dccb7b0fbea : SUCCESS in 7m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6e0f39f1cc9b421783105a080147017a : FAILURE in 59m 01s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d94f59d3374c4f6488e1bc88879ba28e : FAILURE in 30m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9d044621bb894642890bdc1fa4026be9 : FAILURE in 42m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3668b547ef744562abb1d5806473905b : SUCCESS in 1h 10m 28s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/76fcdbd915034b0e80cd8e222d022a5c : FAILURE in 16m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4d2dd223b35b488282cafdbe60a3545c : SUCCESS in 1h 16m 46s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4f68f0b699a648f1a83adb4cc7204e02 : SUCCESS in 4m 43s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/05f503e00f994607a343a3a0d03f9b59 : FAILURE in 26m 53s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6e228c9d58fe4d1199586730cacda179 : SUCCESS in 24m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3497af98c44747d29d02026952850429 : SUCCESS in 41m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ffc570b59e5f4ee6b719857848234cc3 : SUCCESS in 53m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/7ea571e3901b4179b6ca0b4b57c10521 : FAILURE in 18m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2a3975f14e1749ba86690ee2b7adb0d4 : FAILURE in 29m 37s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/0b9480c73ecc482bbe18c7f82f4946cc : FAILURE in 34m 02s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4d7af0d99c9b4cd586d89a2bc5ea0211 : SUCCESS in 39m 35s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4372a9871b03403eb9a07674dcd0a757 : SUCCESS in 40m 22s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/81288dc2ab0642f1b1b2ad6a99895a0f : FAILURE in 21m 29s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8f32c90a282b449a990078f4b0c01298 : FAILURE in 23m 35s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"0e1644f533e1dad935f3c6fb27d234dca2201667","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-18 19:22:29.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"d52a12dddd0cbe876672a1fa640fe462bb62ccf0","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-08-18 19:28:29.000000000","message":"Patch Set 13:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-15-887115-13/NFS/ : SUCCESS in 1h 40m 20s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":13},{"id":"c444c38f808fbed46f41e93c01a4a39bf9e21f88","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-18 20:30:33.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/15/887115/13/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/fa4e8b8/ : FAILURE in 1h 07m 05s","accounts_in_message":[],"_revision_number":13},{"id":"16dbef35ffc02a186a274fbc9d5bfff1b650b8aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 20:44:33.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fee40eb703f4457f9bf0239fea425620\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/398222bd91d749558c8bae2ae9e367a3 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83a1735bcb974ad89192698df90b2b21 : SUCCESS in 11m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65fe9d8a71db47f39d38f4cb4f908ac1 : FAILURE in 26m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ce9138c4e3b40da86cc6b5f24572c09 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80c61ab77a4e46e18d52633c444acf28 : SUCCESS in 10m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d41ce0614cd4748909287ab50b9a834 : SUCCESS in 12m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6127fd75d0484775a5a381c3187e8112 : SUCCESS in 11m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/49674d3f5b6b45b4a9ef95e6fe457e51 : RETRY_LIMIT in 1m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cf55f13a5c040539155fb98aee1b524 : SUCCESS in 5m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dba4dd14c2274e96990da86c5bbb41d7 : SUCCESS in 57m 43s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/17813fa870024153bd9332b0585c525f : FAILURE in 33m 36s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3fc1a9d4162a47ee8b31a92d321fa56e : FAILURE in 31m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/595c6e5fcf3d42f09789e3842da34f51 : SUCCESS in 55m 31s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/6c10bd49208e4860b1fca1d7833e682d : FAILURE in 16m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/04930233c0c14743894f4fcda0722290 : SUCCESS in 1h 16m 32s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9ee9e125a1974c2f91f262688ae11b34 : SUCCESS in 5m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e516547fa6f343bc93f7a1a6fd9f4928 : FAILURE in 28m 02s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cec3656731a34dfabe84ca5067cccc10 : SUCCESS in 34m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b0c06cd88be34bf28206bb192cd82cfe : SUCCESS in 45m 23s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/837116e8ff7d4241990de7ad32bad03c : FAILURE in 58m 52s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/8fb210b31f2a4851838ef44713824cba : FAILURE in 17m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/99023a0d646f4d998fc4202062f46904 : FAILURE in 28m 14s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/b864760bbb074acd8f1380027f309ecf : FAILURE in 33m 03s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/5e4e7adcbd4c43e49e856fba1d88df36 : SUCCESS in 41m 32s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dd0108ae9ea94a789dc1004df1dfe416 : SUCCESS in 1h 06m 19s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d5801ae0e14c4d0eb3fa075fbc6e7454 : FAILURE in 39m 18s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/936a19214d864dcf96bae58822c4c188 : FAILURE in 19m 44s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"3391acee77ad41eddf59d5450de6ad574d338d31","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-18 20:58:30.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/13/check/DellEMC_PowerStore_Manila_dhss_false/b0ebbc1/DellEMC_PowerStore_Manila_dhss_false/93 : FAILURE in 13m 41s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":13},{"id":"7d9dca0bffb7bfea15758bb4aa20c4e269c235de","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-18 21:05:46.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/15/887115/13/check/DellEMC_Unity_Manila_dhss_true/19c9816/DellEMC_Unity_Manila_dhss_true/429 : FAILURE in 20m 56s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/15/887115/13/check/DellEMC_Unity_Manila_dhss_false/f0d8d7b/DellEMC_Unity_Manila_dhss_false/455 : FAILURE in 12m 06s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"e1ef96849e62415f6ac6408c174f1be2258f0dbf","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-19 00:00:03.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/887115/13/upstream-check/manila-tempest-plugin-ontap-dhss/05254cd/ : TIMED_OUT in 3h 14m 28s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/15/887115/13/upstream-check/manila-tempest-plugin-ontap-no-dhss/bdb8982/ : SUCCESS in 2h 57m 56s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"36c5b7df23a30d6ad003d54e147a184f1ab17543","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-08-21 07:01:21.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"56151da88016f3e115677c83ac6bdc14397fe51a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-21 15:19:47.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"54388fab7ecaaf2f4474aa5408ed11d65e58f8ac","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-08-21 21:13:49.000000000","message":"Patch Set 13: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"a8fa676adff67fe9fa5a761db3a771704948fc06","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-08-22 02:51:34.000000000","message":"Patch Set 13: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"000ccb846a8e321c9695b7c221b2787d110335ea","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-08-23 22:10:16.000000000","message":"Patch Set 13: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"79a694df64c0fb6877acf2b062c9acb8c7977468","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 22:10:47.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"9521bae1ae340308767fac19d320c2c53c813f35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 23:27:36.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d08406e16095476ba2743defdc928b1d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/99b6954cb31343829204bfe6f35e394e : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96be05fbcea84294954862fba9ee0ab2 : SUCCESS in 9m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efbed48466054e42b1a5f0e88df1d554 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d463ddbc4cb4e05bd1a9139d223a04d : SUCCESS in 8m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d9f55ca9c0c44d28efaa77e68595eca : SUCCESS in 8m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ff65e6e52c346009e7e3efccfa9dabf : SUCCESS in 9m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a86da92557645e6a97e47ad720668b4 : SUCCESS in 5m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fb7b24ed7f6d444b9c273d25905b7a75 : SUCCESS in 1h 12m 33s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8ee0b3d8308c4adfb06422fbe2dbad85 : SUCCESS in 29m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/41022513c3ec4d57b81d42185e0630c7 : SUCCESS in 35m 06s","accounts_in_message":[],"_revision_number":13},{"id":"eb056696ac79495159e4bcdfba0d089a0fdf8c6a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 23:27:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"8067e527c44affc8a785186f5509470b3bd8b8c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 23:28:44.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a333654ae90412bb324ff9628898057\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6e486a61927544819ed65721af600e5d : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6729c137c0bc478199383f63148cc2ce : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c1949901f964910b69b99802daa5921 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":13},{"id":"ae64b9195900c47c71dfd46e4098951e8c6f4710","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2023-08-24 15:39:30.000000000","message":"Patch Set 13: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"f641577d8a872ff381fe1c465e67b198c052b0be","revisions":{"f544b873ec1ca6684d6307254ab46bd27810a440":{"kind":"REWORK","_number":1,"created":"2023-06-27 20:09:24.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/15/887115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/15/887115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/15/887115/1"}}},"commit":{"parents":[{"commit":"5daf9bd6ca4c2799b7f90f6d12eaff27f80cf36d","subject":"Stop overriding auth/identity options of tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5daf9bd6ca4c2799b7f90f6d12eaff27f80cf36d"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-06-27 20:09:00.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-06-27 20:09:07.000000000","tz":-420},"subject":"[WIP] Resource Locks API","message":"[WIP] Resource Locks API\n\nAdd CRUD APIs for resource locks; they\u0027re not useful\nyet; a follow up patch will add support for using them\nto prevent share and access rule deletion.\n\nTodo:\n- Unit tests\n- API Ref; docs\n- Reno\n\nChange-Id: I146bc09e4e8a39797e22458ff6860346e11e592e\nImplements: bp/allow-locking-shares-against-deletion\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f544b873ec1ca6684d6307254ab46bd27810a440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f544b873ec1ca6684d6307254ab46bd27810a440"}]},"branch":"refs/heads/master"},"07bd0f55bf9e43165a35ec90041312f78c64746c":{"kind":"REWORK","_number":2,"created":"2023-07-08 03:17:51.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/15/887115/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/15/887115/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/15/887115/2"}}},"commit":{"parents":[{"commit":"5daf9bd6ca4c2799b7f90f6d12eaff27f80cf36d","subject":"Stop overriding auth/identity options of tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5daf9bd6ca4c2799b7f90f6d12eaff27f80cf36d"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-06-27 20:09:00.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-07-08 03:17:33.000000000","tz":-420},"subject":"[WIP] Resource Locks API","message":"[WIP] Resource Locks API\n\nAdd CRUD APIs for resource locks; they\u0027re not useful\nyet; a follow up patch will add support for using them\nto prevent share and access rule deletion.\n\nTodo:\n- Unit tests\n- API Ref; docs\n- Reno\n\nChange-Id: I146bc09e4e8a39797e22458ff6860346e11e592e\nImplements: bp/allow-locking-shares-against-deletion\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/07bd0f55bf9e43165a35ec90041312f78c64746c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/07bd0f55bf9e43165a35ec90041312f78c64746c"}]},"branch":"refs/heads/master"},"f0973e31034b8d77c13ed18148ceafa70fca7f63":{"kind":"REWORK","_number":3,"created":"2023-07-11 00:50:21.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/15/887115/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/15/887115/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/15/887115/3"}}},"commit":{"parents":[{"commit":"5daf9bd6ca4c2799b7f90f6d12eaff27f80cf36d","subject":"Stop overriding auth/identity options of tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5daf9bd6ca4c2799b7f90f6d12eaff27f80cf36d"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-06-27 20:09:00.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-07-11 00:50:11.000000000","tz":-420},"subject":"[WIP] Resource Locks API","message":"[WIP] Resource Locks API\n\nAdd CRUD APIs for resource locks with support\nfor preventing deletion of shares (applies to\nsoft-deletions and unmanage operations as well).\n\nTodo:\n- Unit tests\n- API Ref; docs\n- Reno\n\nChange-Id: I146bc09e4e8a39797e22458ff6860346e11e592e\nImplements: bp/allow-locking-shares-against-deletion\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f0973e31034b8d77c13ed18148ceafa70fca7f63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f0973e31034b8d77c13ed18148ceafa70fca7f63"}]},"branch":"refs/heads/master"},"478b45cb8eb812b67b46c200176d7f94177a15de":{"kind":"REWORK","_number":4,"created":"2023-07-14 21:40:55.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/15/887115/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/15/887115/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/15/887115/4"}}},"commit":{"parents":[{"commit":"5daf9bd6ca4c2799b7f90f6d12eaff27f80cf36d","subject":"Stop overriding auth/identity options of tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5daf9bd6ca4c2799b7f90f6d12eaff27f80cf36d"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-06-27 20:09:00.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-07-14 21:40:32.000000000","tz":-420},"subject":"[WIP] Resource Locks API","message":"[WIP] Resource Locks API\n\nAdd CRUD APIs for resource locks with support\nfor preventing deletion of shares (applies to\nsoft-deletions and unmanage operations as well).\n\nTodo:\n- Unit tests\n- API Ref; docs\n- Reno\n\nChange-Id: I146bc09e4e8a39797e22458ff6860346e11e592e\nImplements: bp/allow-locking-shares-against-deletion\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/478b45cb8eb812b67b46c200176d7f94177a15de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/478b45cb8eb812b67b46c200176d7f94177a15de"}]},"branch":"refs/heads/master"},"47bbf83d304cb3e03787d88e846058cfff4b6a66":{"kind":"REWORK","_number":5,"created":"2023-07-15 01:22:21.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/15/887115/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/15/887115/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/15/887115/5"}}},"commit":{"parents":[{"commit":"5daf9bd6ca4c2799b7f90f6d12eaff27f80cf36d","subject":"Stop overriding auth/identity options of tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5daf9bd6ca4c2799b7f90f6d12eaff27f80cf36d"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-06-27 20:09:00.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-07-15 01:21:15.000000000","tz":-420},"subject":"[WIP] Resource Locks API","message":"[WIP] Resource Locks API\n\nAdd CRUD APIs for resource locks with support\nfor preventing deletion of shares (applies to\nsoft-deletions and unmanage operations as well).\n\nTodo:\n- Unit tests\n- API Ref; docs\n- Reno\n\nChange-Id: I146bc09e4e8a39797e22458ff6860346e11e592e\nImplements: bp/allow-locking-shares-against-deletion\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/47bbf83d304cb3e03787d88e846058cfff4b6a66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/47bbf83d304cb3e03787d88e846058cfff4b6a66"}]},"branch":"refs/heads/master"},"c8746a7679115ba7f11daba12e68682bc55be8e5":{"kind":"REWORK","_number":6,"created":"2023-07-20 08:11:41.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/15/887115/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/15/887115/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/15/887115/6"}}},"commit":{"parents":[{"commit":"5daf9bd6ca4c2799b7f90f6d12eaff27f80cf36d","subject":"Stop overriding auth/identity options of tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5daf9bd6ca4c2799b7f90f6d12eaff27f80cf36d"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-06-27 20:09:00.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-07-20 08:11:17.000000000","tz":-420},"subject":"Resource Locks: Support for share deletion lock","message":"Resource Locks: Support for share deletion lock\n\nAdd CRUD APIs for resource locks with support\nfor preventing deletion of shares (applies to\nsoft-deletions and unmanage operations as well).\n\nChange-Id: I146bc09e4e8a39797e22458ff6860346e11e592e\nImplements: bp/allow-locking-shares-against-deletion\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c8746a7679115ba7f11daba12e68682bc55be8e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c8746a7679115ba7f11daba12e68682bc55be8e5"}]},"branch":"refs/heads/master"},"a9e676a6dc58f3747c990f6e0f96ce96f0ccc39e":{"kind":"REWORK","_number":7,"created":"2023-07-20 23:27:06.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/15/887115/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/15/887115/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/15/887115/7"}}},"commit":{"parents":[{"commit":"5daf9bd6ca4c2799b7f90f6d12eaff27f80cf36d","subject":"Stop overriding auth/identity options of tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5daf9bd6ca4c2799b7f90f6d12eaff27f80cf36d"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-06-27 20:09:00.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-07-20 23:26:52.000000000","tz":-420},"subject":"Resource Locks: Support for share deletion lock","message":"Resource Locks: Support for share deletion lock\n\nAdd CRUD APIs for resource locks with support\nfor preventing deletion of shares (applies to\nsoft-deletions and unmanage operations as well).\n\nChange-Id: I146bc09e4e8a39797e22458ff6860346e11e592e\nImplements: bp/allow-locking-shares-against-deletion\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a9e676a6dc58f3747c990f6e0f96ce96f0ccc39e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a9e676a6dc58f3747c990f6e0f96ce96f0ccc39e"}]},"branch":"refs/heads/master"},"f0dc2e4bf6e4391277ec03d0a4e4354d1ac6fdf6":{"kind":"REWORK","_number":8,"created":"2023-07-26 22:55:33.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/15/887115/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/15/887115/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/15/887115/8"}}},"commit":{"parents":[{"commit":"0e7812657b3ca90e618087b9fbb26816e60a90c3","subject":"Add count info in \u0027snapshot list\u0027 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0e7812657b3ca90e618087b9fbb26816e60a90c3"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-06-27 20:09:00.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-07-26 22:54:18.000000000","tz":-420},"subject":"Resource Locks: Support for share deletion lock","message":"Resource Locks: Support for share deletion lock\n\nAdd CRUD APIs for resource locks with support\nfor preventing deletion of shares (applies to\nsoft-deletions and unmanage operations as well).\n\nChange-Id: I146bc09e4e8a39797e22458ff6860346e11e592e\nImplements: bp/allow-locking-shares-against-deletion\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f0dc2e4bf6e4391277ec03d0a4e4354d1ac6fdf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f0dc2e4bf6e4391277ec03d0a4e4354d1ac6fdf6"}]},"branch":"refs/heads/master"},"278802545217e628720a375b9baeb55148b17dfc":{"kind":"REWORK","_number":9,"created":"2023-07-27 16:46:35.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/15/887115/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/15/887115/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/15/887115/9"}}},"commit":{"parents":[{"commit":"0e7812657b3ca90e618087b9fbb26816e60a90c3","subject":"Add count info in \u0027snapshot list\u0027 API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0e7812657b3ca90e618087b9fbb26816e60a90c3"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-06-27 20:09:00.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-07-27 16:46:35.000000000","tz":0},"subject":"Resource Locks: Support for share deletion lock","message":"Resource Locks: Support for share deletion lock\n\nAdd CRUD APIs for resource locks with support\nfor preventing deletion of shares (applies to\nsoft-deletions and unmanage operations as well).\n\nChange-Id: I146bc09e4e8a39797e22458ff6860346e11e592e\nImplements: bp/allow-locking-shares-against-deletion\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/278802545217e628720a375b9baeb55148b17dfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/278802545217e628720a375b9baeb55148b17dfc"}]},"branch":"refs/heads/master"},"7f839638a637c2847cb7b78bb99a73fab41d77c2":{"kind":"REWORK","_number":10,"created":"2023-08-12 19:03:30.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/15/887115/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/15/887115/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/15/887115/10"}}},"commit":{"parents":[{"commit":"824fdb3270e64782cfdeb6bf27ea107ef7c1e58f","subject":"Merge \"[DOC] Add Snapshot Metadata Documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/824fdb3270e64782cfdeb6bf27ea107ef7c1e58f"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-06-27 20:09:00.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-12 19:03:10.000000000","tz":-420},"subject":"Resource Locks: Support for share deletion lock","message":"Resource Locks: Support for share deletion lock\n\nAdd CRUD APIs for resource locks with support\nfor preventing deletion of shares (applies to\nsoft-deletions and unmanage operations as well).\n\nChange-Id: I146bc09e4e8a39797e22458ff6860346e11e592e\nImplements: bp/allow-locking-shares-against-deletion\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7f839638a637c2847cb7b78bb99a73fab41d77c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7f839638a637c2847cb7b78bb99a73fab41d77c2"}]},"branch":"refs/heads/master"},"3c2678c4ec4111ee22ec4047ac7480f5e7cf982b":{"kind":"REWORK","_number":11,"created":"2023-08-12 19:32:04.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/15/887115/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/15/887115/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/15/887115/11"}}},"commit":{"parents":[{"commit":"824fdb3270e64782cfdeb6bf27ea107ef7c1e58f","subject":"Merge \"[DOC] Add Snapshot Metadata Documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/824fdb3270e64782cfdeb6bf27ea107ef7c1e58f"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-06-27 20:09:00.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-12 19:31:41.000000000","tz":-420},"subject":"Resource Locks: Support for share deletion lock","message":"Resource Locks: Support for share deletion lock\n\nAdd CRUD APIs for resource locks with support\nfor preventing deletion of shares (applies to\nsoft-deletions and unmanage operations as well).\n\nChange-Id: I146bc09e4e8a39797e22458ff6860346e11e592e\nImplements: bp/allow-locking-shares-against-deletion\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3c2678c4ec4111ee22ec4047ac7480f5e7cf982b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3c2678c4ec4111ee22ec4047ac7480f5e7cf982b"}]},"branch":"refs/heads/master"},"090c2f73c4393083a4a71e39314625ea592b0e0a":{"kind":"REWORK","_number":12,"created":"2023-08-17 22:05:13.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/15/887115/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/15/887115/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/15/887115/12"}}},"commit":{"parents":[{"commit":"824fdb3270e64782cfdeb6bf27ea107ef7c1e58f","subject":"Merge \"[DOC] Add Snapshot Metadata Documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/824fdb3270e64782cfdeb6bf27ea107ef7c1e58f"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-06-27 20:09:00.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-17 22:05:01.000000000","tz":-420},"subject":"Resource Locks: Support for share deletion lock","message":"Resource Locks: Support for share deletion lock\n\nAdd CRUD APIs for resource locks with support\nfor preventing deletion of shares (applies to\nsoft-deletions and unmanage operations as well).\n\nChange-Id: I146bc09e4e8a39797e22458ff6860346e11e592e\nImplements: bp/allow-locking-shares-against-deletion\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/090c2f73c4393083a4a71e39314625ea592b0e0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/090c2f73c4393083a4a71e39314625ea592b0e0a"}]},"branch":"refs/heads/master"},"f641577d8a872ff381fe1c465e67b198c052b0be":{"kind":"REWORK","_number":13,"created":"2023-08-18 17:47:47.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/15/887115/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/15/887115/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/15/887115/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/15/887115/13"}}},"commit":{"parents":[{"commit":"824fdb3270e64782cfdeb6bf27ea107ef7c1e58f","subject":"Merge \"[DOC] Add Snapshot Metadata Documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/824fdb3270e64782cfdeb6bf27ea107ef7c1e58f"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-06-27 20:09:00.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-18 17:47:25.000000000","tz":-420},"subject":"Resource Locks: Support for share deletion lock","message":"Resource Locks: Support for share deletion lock\n\nAdd CRUD APIs for resource locks with support\nfor preventing deletion of shares (applies to\nsoft-deletions and unmanage operations as well).\n\nChange-Id: I146bc09e4e8a39797e22458ff6860346e11e592e\nImplements: bp/allow-locking-shares-against-deletion\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f641577d8a872ff381fe1c465e67b198c052b0be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f641577d8a872ff381fe1c465e67b198c052b0be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
