)]}'
{"id":"openstack%2Fmanila~890841","triplet_id":"openstack%2Fmanila~stable%2Fzed~Ia815720cedda2f5c70205ffda5c765364cee8f8c","project":"openstack/manila","branch":"stable/zed","topic":"bug/2012246","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-10-16 15:47:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia815720cedda2f5c70205ffda5c765364cee8f8c","subject":"[coordination] backend_url should be secret","status":"MERGED","created":"2023-08-08 23:59:40.000000000","updated":"2023-10-16 15:50:04.000000000","submitted":"2023-10-16 15:47:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890841-bug/2012246","cherry_pick_of_change":878152,"cherry_pick_of_patch_set":1,"meta_rev_id":"5c154b1ef8ae62581a929f1f674af5806b038b3b","_number":890841,"virtual_id_number":890841,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-16 15:47:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-10-16 13:40:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-10-16 13:40:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-08 23:59:40.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-08-09 01:03:19.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-08-09 03:03:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-16 13:40:00.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"810716a206060ef3ec6055342fb7f18426d8be7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-08 23:59:40.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.1.","accounts_in_message":[],"_revision_number":1},{"id":"eb927d6efefd8f50e7b1b67452831646cb96ca10","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-09 00:01:15.000000000","message":"Topic bug/2012246-stable/zed removed","accounts_in_message":[],"_revision_number":1},{"id":"5fc602037bead1a80355b5f6f9a2a02c0675d7ce","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-09 00:01:17.000000000","message":"Topic set to bug/2012246","accounts_in_message":[],"_revision_number":1},{"id":"715b2ff899a52df835d928e2e276de179067d703","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-09 01:03:19.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/41/890841/1/thirdparty-check/pure-devstack-manila-tempest-aio/334a153/ : SUCCESS in 1h 02m 39s","accounts_in_message":[],"_revision_number":1},{"id":"bfdbbbbf757cc4bef8e6e1de3354f5f244d833ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 03:03:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30f0a10f48c54163bef1b91bd5e952fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88ce5356124c4276953218c29c0f5bd7 : SUCCESS in 9m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95b06dbad10f4508ae7b71543fb6ae1c : SUCCESS in 18m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bb1638a6b334040ae8f8ef66c34ef3b : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3983f6b933e4a97971ab3bf6e03fd0e : SUCCESS in 8m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31233f9207d74b48a5bb85e12eb0dc5e : SUCCESS in 9m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54bbc2e4223f4812a561ac30e5161fd7 : SUCCESS in 11m 07s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/abeb15fe040645698d86c36c5be2f80d : SUCCESS in 57m 32s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/84011e690aa04352b0d565472bd7780e : FAILURE in 2h 55m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/564e7cb32cc14075bd54cb2932d9e5d6 : SUCCESS in 1h 37m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/03918d766bcb4a9f88fac28b25732a1e : SUCCESS in 1h 12m 42s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/e873c6e34aaf4042980050ee1e14ace2 : FAILURE in 12m 40s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a8211a71d9c649d1bfe697b9e577113d : FAILURE in 21m 02s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9de54497f85d4e26ae3e3175bd2c76f7 : SUCCESS in 5m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/64e95f538ec84499982268e106d33486 : SUCCESS in 4m 58s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/acae11f52f7143128daebcbbcdfdbf26 : SUCCESS in 27m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fbe035aef5d449c18f4737be24cfb266 : SUCCESS in 33m 31s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4670969440914c70a24cb21d72d43c48 : FAILURE in 17m 51s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/40dbf517ce964114bf855370f64a00ae : SUCCESS in 38m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fc17c1e5be9c47b1a598334766e8c860 : SUCCESS in 18m 40s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/11f296ab7c114ab5a254cc794057d022 : FAILURE in 14m 01s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/39c293f4316e4c398252d48b476ad7b0 : SUCCESS in 48m 36s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/21ab81917c8245d79ae13589a94ca81f : FAILURE in 50m 59s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/25e3dbeeaf264d388a05ac993da1bc63 : SUCCESS in 28m 03s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/5b5b08c90ad84b5ba91861f7bebd414d : TIMED_OUT in 2h 05m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad941f532c4d5233c02df18d512bcf1752c7e176","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-10-16 13:40:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"21932dae4dd6c6fcf26cc3ee7b4a24b7e3e678da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 13:40:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"96d036a0cf67ae5d6dd2a7f85018f171cf888522","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 15:47:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ee848ce97fb4354ab8b4cc508a58425\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24737477af5644ca8d508ed7f683b4b6 : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b2aba952b5e4e5fa8e5b22d7414d234 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c50cd39650504ee08780aa44b6c6b4dc : SUCCESS in 10m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f2684b355fa4173b1cd60ce37565c0a : SUCCESS in 8m 46s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2a39cac94164400983c736bae428433b : SUCCESS in 37m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/80f86319eaac405586a5510a9638241a : SUCCESS in 36m 32s","accounts_in_message":[],"_revision_number":1},{"id":"92b7392272dd67b8d91feb9109aab67900d5715d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 15:47:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5c154b1ef8ae62581a929f1f674af5806b038b3b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 15:50:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8a3710b76854a7793de1885e5a6a300\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69473b5f7f3e4b27bef125cd2909c198 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98ea6ac7688be7376a3d3512c69b5905c9738a01","revisions":{"98ea6ac7688be7376a3d3512c69b5905c9738a01":{"kind":"REWORK","_number":1,"created":"2023-08-08 23:59:40.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/41/890841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/41/890841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/41/890841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/41/890841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/41/890841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/41/890841/1"}}},"commit":{"parents":[{"commit":"ecd14b4e3cf4028c15345accf583638eda94e1c7","subject":"Fix duplicate entries in share_server_backend_details","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ecd14b4e3cf4028c15345accf583638eda94e1c7"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-20 09:31:52.000000000","tz":540},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-08 23:59:40.000000000","tz":0},"subject":"[coordination] backend_url should be secret","message":"[coordination] backend_url should be secret\n\nThe backend_url option can sometimes contain secrets.\n\nFor example when redis coordination backend is used and authentication\nis enabled in redis, the plain redis password is put as an URL element.\n\n[coordination]\nbackend_url\u003dredis://:password@127.0.0.1:6379\n\nCloses-Bug: #2012246\nChange-Id: Ia815720cedda2f5c70205ffda5c765364cee8f8c\n(cherry picked from commit 8ec46875665e93daf252a9feaf9d36d354c3660c)\n(cherry picked from commit 5da7be292664194d9845d1ae77fba821f97ff7fe)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/98ea6ac7688be7376a3d3512c69b5905c9738a01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/98ea6ac7688be7376a3d3512c69b5905c9738a01"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
