)]}'
{"id":"openstack%2Fmanila~890842","triplet_id":"openstack%2Fmanila~stable%2Fyoga~Ia815720cedda2f5c70205ffda5c765364cee8f8c","project":"openstack/manila","branch":"stable/yoga","topic":"bug/2012246","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-11-13 11:55:08.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-11-13 13:25:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia815720cedda2f5c70205ffda5c765364cee8f8c","subject":"[coordination] backend_url should be secret","status":"MERGED","created":"2023-08-09 00:00:03.000000000","updated":"2023-11-13 13:26:50.000000000","submitted":"2023-11-13 13:25:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890842-bug/2012246","cherry_pick_of_change":890841,"cherry_pick_of_patch_set":1,"meta_rev_id":"5804d45394a85827c410c8d49b5f8e21237bd9e5","_number":890842,"virtual_id_number":890842,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-13 13:25:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-11-13 11:55:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-11-13 11:55:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-09 00:00:03.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-08-09 00:02:41.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-08-09 01:37:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-06 16:02:00.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"81ef0d49ccd26085b0c90eb1a74a576ba982e5fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-09 00:00:03.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"99b96d9906e782726bef0941a031398f7d24623d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-09 00:01:07.000000000","message":"Topic bug/2012246-stable/zed-stable/yoga removed","accounts_in_message":[],"_revision_number":1},{"id":"42a56d3fb920d28ef0e5f95ccd17e01ee5f5dc5e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-09 00:01:09.000000000","message":"Topic set to bug/2012246","accounts_in_message":[],"_revision_number":1},{"id":"8bc544dfd0492ad57ee18c307386571a8ff8f463","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-09 00:02:41.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio pure-devstack-manila-tempest-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"adaf1f63f9b89406dbcb6a84c16f8a5a7bdf4a27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 01:37:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ba2e241c38c491693021097ea1c0596\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/797c1c3bdd674e9c8f268e9a3e7162e5 : SUCCESS in 9m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2193798f85084e3f82b21204b915ccfa : SUCCESS in 18m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aa7c4e10c4d4f5ba493320103e81ba2 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df8e92f270944805b599917fe311b37d : SUCCESS in 9m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc857c63f03545a2b478945fbbbab905 : SUCCESS in 9m 35s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/f531d720e71743158c8c09a73a495e17 : SUCCESS in 1h 19m 56s\n- manila-tempest-plugin-generic-scenario-stable https://zuul.opendev.org/t/openstack/build/b6d32eb861fe4a61812c15bcac5728a9 : SUCCESS in 1h 34m 40s (non-voting)\n- manila-tempest-plugin-generic-stable https://zuul.opendev.org/t/openstack/build/9eace4a612154975aad5996bb5dd3602 : SUCCESS in 1h 01m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native-stable https://zuul.opendev.org/t/openstack/build/b2f0232e11c149adb71a8b98635cd10b : SUCCESS in 1h 10m 02s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-stable https://zuul.opendev.org/t/openstack/build/dfd56126eab7429ea9eae7f11bb86c67 : FAILURE in 1h 32m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux-stable https://zuul.opendev.org/t/openstack/build/bff8814d9e514ef7b16250cec161e607 : FAILURE in 20m 48s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/596466273bba41afb3f7b6b2485ddd4c : SUCCESS in 4m 34s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/17961dc5bf014994b93dc918d3491009 : SUCCESS in 4m 15s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0cf17723814c4f6ab520e74eaf55f040 : TIMED_OUT in 1h 06m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6ac8c80bbfc0448fa643e5fecc76dc56 : SUCCESS in 29m 54s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/20c272299e594c04b41add23e60b071c : FAILURE in 20m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/f5350e937ba7455daff54ae847cabd9c : SUCCESS in 28m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d61b99d034a94a5993f3a161288f116a : SUCCESS in 20m 33s (non-voting)\n- manila-tempest-plugin-lvm-fips-stable https://zuul.opendev.org/t/openstack/build/d8465686cbec467ba97f9aa9b82105cc : FAILURE in 10m 51s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/dd2aa420c21a4b009c046a84c875b8a4 : SUCCESS in 48m 52s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0909674740124068b72b8176940d6670 : SUCCESS in 28m 14s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7090ddcb44694dbc9bc70a8d89c7974e : SUCCESS in 33m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29b15e4efe86c05d7ae7408205566621364a24c5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-06 08:35:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d61b8b3abdec5913474310b10770b1e13a64f72e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 11:56:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ab17e3b8e4d47408f5aad8736157b5e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a62645a7ce94d09b50d2b98d6cb2371 : SUCCESS in 8m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/816250c5e80a41a6b9d30f5520a7880e : SUCCESS in 25m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f556a0f1b6fe42598f286af91542cafe : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3fe6dab081f4dcd9d79e19ddccaeb0d : SUCCESS in 8m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35d95a39662e444da5b0e21535661043 : SUCCESS in 11m 22s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/bc851ded44d8435f840c57c1bcc5b8e9 : SUCCESS in 1h 25m 36s\n- manila-tempest-plugin-generic-scenario-stable https://zuul.opendev.org/t/openstack/build/ec4a9c31a30f472192895470a2017e72 : TIMED_OUT in 3h 06m 24s (non-voting)\n- manila-tempest-plugin-generic-stable https://zuul.opendev.org/t/openstack/build/234873a6b0d04153b3dafa25f5dcb352 : SUCCESS in 1h 16m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native-stable https://zuul.opendev.org/t/openstack/build/16a53f07990a434f8cafe8b650530b23 : SUCCESS in 1h 29m 53s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-stable https://zuul.opendev.org/t/openstack/build/113525e98d264698992ef1ec831c3166 : SUCCESS in 1h 40m 13s (non-voting)\n- manila-tempest-plugin-zfsonlinux-stable https://zuul.opendev.org/t/openstack/build/a2d260f16f3e4ab7a45402da6072995a : SUCCESS in 1h 10m 23s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f3be6c53aa3c4c1b88299b9d2bfc7008 : SUCCESS in 6m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4ef770c5ae4a45b0924f2bdd134b0592 : SUCCESS in 4m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ffec15026d884c7f87577bb4e392f1a9 : SUCCESS in 36m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5a3f6fc2d52747e68ab0e3f499c8bb94 : SUCCESS in 38m 12s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8bc2bb13f39b452185497ae20c34c2c3 : SUCCESS in 1h 05m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/282f0ffb72384456b332ef6c25ae37be : SUCCESS in 44m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0831e13ab0ef463b8cedf08b0e967bf5 : SUCCESS in 23m 31s (non-voting)\n- manila-tempest-plugin-lvm-fips-stable https://zuul.opendev.org/t/openstack/build/f3652fd285d74457b5db8e504d9f0b0b : SUCCESS in 1h 31m 43s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d302d29398434960849a3ce9805b1407 : SUCCESS in 1h 01m 08s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2b976f0860b344f59b1f53c6ff208d22 : SUCCESS in 48m 18s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f0fef67df7ba47bead03a6b30e748663 : SUCCESS in 45m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4ee31d7f34753875906027926b212745b61b160","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-11-13 11:55:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e758a45fc63a9566eb886635e6db2c83ef35f20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 11:55:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74f658eb6b49da1c730c38c6365d7fce3d725079","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 13:25:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c49bce10ab104d1ebdf4267c164b4f90\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93f3e4300ca645ce8f07281ac8377aee : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acd459b9888e4758b4b30a321b947f76 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ea407b0297045d6812f2360d058cd92 : SUCCESS in 12m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7078d01152774aa4af62e197f0ded59a : SUCCESS in 10m 42s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/54626f5574c948848bf3b2a55350844e : SUCCESS in 1h 25m 15s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c5f0ae4d7de346feace7889670cd56fd : SUCCESS in 35m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8dbfdaf0ee78456f920f825ed0fff278 : SUCCESS in 30m 29s","accounts_in_message":[],"_revision_number":1},{"id":"ccf56704fdfe2d92246915ff1a7009598365bd22","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 13:25:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5804d45394a85827c410c8d49b5f8e21237bd9e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 13:26:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8302a7d86f64469b86a528998217daa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40ffd3b29ebf47f294fa1c1f69f952b1 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b73958504c3638882ddd992c91317d70b0a2d444","revisions":{"b73958504c3638882ddd992c91317d70b0a2d444":{"kind":"REWORK","_number":1,"created":"2023-08-09 00:00:03.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/42/890842/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/42/890842/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/42/890842/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/42/890842/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/42/890842/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/42/890842/1"}}},"commit":{"parents":[{"commit":"3ba4d99288a372e5d9f7cf9f108399af92041553","subject":"Fix Manila API error message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3ba4d99288a372e5d9f7cf9f108399af92041553"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-20 09:31:52.000000000","tz":540},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-09 00:00:03.000000000","tz":0},"subject":"[coordination] backend_url should be secret","message":"[coordination] backend_url should be secret\n\nThe backend_url option can sometimes contain secrets.\n\nFor example when redis coordination backend is used and authentication\nis enabled in redis, the plain redis password is put as an URL element.\n\n[coordination]\nbackend_url\u003dredis://:password@127.0.0.1:6379\n\nCloses-Bug: #2012246\nChange-Id: Ia815720cedda2f5c70205ffda5c765364cee8f8c\n(cherry picked from commit 8ec46875665e93daf252a9feaf9d36d354c3660c)\n(cherry picked from commit 5da7be292664194d9845d1ae77fba821f97ff7fe)\n(cherry picked from commit 98ea6ac7688be7376a3d3512c69b5905c9738a01)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b73958504c3638882ddd992c91317d70b0a2d444"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b73958504c3638882ddd992c91317d70b0a2d444"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
