)]}'
{"id":"openstack%2Fmanila~890844","triplet_id":"openstack%2Fmanila~stable%2Fwallaby~Ia815720cedda2f5c70205ffda5c765364cee8f8c","project":"openstack/manila","branch":"stable/wallaby","topic":"bug/2012246","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-11-28 14:09:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia815720cedda2f5c70205ffda5c765364cee8f8c","subject":"[coordination] backend_url should be secret","status":"MERGED","created":"2023-08-09 00:00:46.000000000","updated":"2023-11-28 14:10:49.000000000","submitted":"2023-11-28 14:09:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890844-bug/2012246","cherry_pick_of_change":890843,"cherry_pick_of_patch_set":1,"meta_rev_id":"25c293a4cee878e2d485f6b4984924a18d960288","_number":890844,"virtual_id_number":890844,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-28 14:09:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-11-28 13:02:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-11-28 13:02:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-09 00:00:46.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-08-09 01:29:54.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-09 02:12:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-27 15:36:09.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2023-11-28 13:02:14.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"f43d9ab3e4434085c9bd2fcc7887227d0aa3c0f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-09 00:00:46.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"cd320ec7d9fcff069de0a7a24d9c60dbf4178a3f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-09 00:00:52.000000000","message":"Topic bug/2012246-stable/zed-stable/yoga-stable/xena-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"987546e6cde4d63227f8619ae126fd502566fc9d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-09 00:00:54.000000000","message":"Topic set to bug/2012246","accounts_in_message":[],"_revision_number":1},{"id":"2e487e8e9e5cfa8d6b3e6198bce7cef35818d731","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-08-09 01:29:54.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-44-890844-1/ns5_manila : SUCCESS in 1h 6m 17s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"098a53ad9e9fdc2211a26b491e20a9c316785fc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 02:12:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/842bec60bc4f47949cb2bd7cdcbda18b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00a7a2599ea14650869ea241fb6b75aa : SUCCESS in 9m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22da41f3a46b49ff9d9de53c7df983ef : SUCCESS in 17m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f6453c125dc4a078d89f596990c36cd : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63216dd19b2043498ec743184a51a456 : SUCCESS in 8m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01166b143c744c959f36bca4c8135b74 : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/184cfab5e2f0462c9209a9be14547b41 : SUCCESS in 8m 57s (non-voting)\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/5cd7b7c9e4b14e7fb3a34faf92c794cd : SUCCESS in 50m 14s\n- manila-tempest-plugin-generic-scenario-wallaby https://zuul.opendev.org/t/openstack/build/58524e3f85c0453d896270015a4cdfbe : SUCCESS in 2h 04m 21s (non-voting)\n- manila-tempest-plugin-generic-wallaby https://zuul.opendev.org/t/openstack/build/9b65c03181fc4b8b8ae54121e42a17f8 : SUCCESS in 1h 05m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native-wallaby https://zuul.opendev.org/t/openstack/build/d441cd55a3674dafb8c8ca4d60e0fbeb : FAILURE in 53m 16s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-wallaby https://zuul.opendev.org/t/openstack/build/aa70185e92924408bdaa9cc1742e319e : FAILURE in 32m 42s (non-voting)\n- manila-tempest-plugin-zfsonlinux-wallaby https://zuul.opendev.org/t/openstack/build/fda2b800844645619c12ffa3d9faaaab : FAILURE in 53m 00s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/86a87a92f74e4d83861b5bfe8784d73e : SUCCESS in 4m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5ff591876fa64dbc85fa673cda68154c : SUCCESS in 4m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss-wallaby https://zuul.opendev.org/t/openstack/build/daf52ac0b5bd4bf287be20df83598550 : SUCCESS in 39m 10s\n- manila-tempest-plugin-dummy-dhss-wallaby https://zuul.opendev.org/t/openstack/build/14fcfdbd3e3b4143a109bbf1ea72d184 : SUCCESS in 31m 35s\n- manila-tempest-plugin-container-wallaby https://zuul.opendev.org/t/openstack/build/430b6e2ad1954d5ab72c7c75881e38bd : SUCCESS in 51m 57s (non-voting)\n- manila-tempest-plugin-glusterfs-native-wallaby https://zuul.opendev.org/t/openstack/build/feedec2363694c869da60b98ff5c9bbc : SUCCESS in 22m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs-wallaby https://zuul.opendev.org/t/openstack/build/952ed4c146d944ff9b8b8c009728fa9c : SUCCESS in 15m 58s (non-voting)\n- manila-tempest-plugin-lvm-fips-wallaby https://zuul.opendev.org/t/openstack/build/2a5ce0c6fe6d4819b522dffd15746552 : FAILURE in 31m 17s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/32354abc50134229b1515e1edcdb5a5c : SUCCESS in 51m 16s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/9b9eae7f45254076b5a4ac1350e11ad5 : SUCCESS in 34m 39s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/52ee18e70f17481183495ac7cb98ded1 : SUCCESS in 44m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"858109c8a26f4ae20d1d8ba64c5ff52ed14bb410","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-11-28 13:02:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"389d57d136bf0bccbf58e8091a14aa792bac18a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 13:02:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c143ae4dee73754ed330738ceba1f7389e505a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 14:09:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2954ed758e354443a18024617ccf4361\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78988044299946d28133f90a8cc8a5e7 : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07119a0bbd0847f0b74bd358b1451f06 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/911b5a91d2fd47a082307a557660390f : SUCCESS in 9m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91105eaddd744e3eb7a070e87b327179 : SUCCESS in 10m 02s\n- manila-tempest-plugin-lvm-wallaby https://zuul.opendev.org/t/openstack/build/1d1bbee520e34503b5563affa1d074fb : SUCCESS in 1h 06m 11s\n- manila-tempest-plugin-dummy-no-dhss-wallaby https://zuul.opendev.org/t/openstack/build/bfe3fe3c81474ae7b40586ee7f526889 : SUCCESS in 17m 23s\n- manila-tempest-plugin-dummy-dhss-wallaby https://zuul.opendev.org/t/openstack/build/552d0ae8e3b24a8e8db881bac087431b : SUCCESS in 16m 06s","accounts_in_message":[],"_revision_number":1},{"id":"f49ef307b9d0b18559d18c05989112822a63b785","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 14:09:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"25c293a4cee878e2d485f6b4984924a18d960288","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 14:10:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d8620ea0df0450a89225465ab8b5bec\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c1cc7d544434dfea536dcd580a12b95 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"733be6ee2261053d6864a4ac65c1cdf4838db2a8","revisions":{"733be6ee2261053d6864a4ac65c1cdf4838db2a8":{"kind":"REWORK","_number":1,"created":"2023-08-09 00:00:46.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/44/890844/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/44/890844/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/890844/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/890844/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/44/890844/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/44/890844/1"}}},"commit":{"parents":[{"commit":"6386142c0db5e361b9360ec40e9dd55d2acd59a4","subject":"Merge \"Check project permissions for share replicas\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6386142c0db5e361b9360ec40e9dd55d2acd59a4"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2023-03-20 09:31:52.000000000","tz":540},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-09 00:00:46.000000000","tz":0},"subject":"[coordination] backend_url should be secret","message":"[coordination] backend_url should be secret\n\nThe backend_url option can sometimes contain secrets.\n\nFor example when redis coordination backend is used and authentication\nis enabled in redis, the plain redis password is put as an URL element.\n\n[coordination]\nbackend_url\u003dredis://:password@127.0.0.1:6379\n\nCloses-Bug: #2012246\nChange-Id: Ia815720cedda2f5c70205ffda5c765364cee8f8c\n(cherry picked from commit 8ec46875665e93daf252a9feaf9d36d354c3660c)\n(cherry picked from commit 5da7be292664194d9845d1ae77fba821f97ff7fe)\n(cherry picked from commit 98ea6ac7688be7376a3d3512c69b5905c9738a01)\n(cherry picked from commit b73958504c3638882ddd992c91317d70b0a2d444)\n(cherry picked from commit 60ae6a7b208fabec9590bc9c5a0abfc78d325bf9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/733be6ee2261053d6864a4ac65c1cdf4838db2a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/733be6ee2261053d6864a4ac65c1cdf4838db2a8"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
