)]}'
{"id":"openstack%2Fmanila~891019","triplet_id":"openstack%2Fmanila~master~I4b8e9344c36aa874b95fce3e0989d34b690159d6","project":"openstack/manila","branch":"master","topic":"ceilometer-integration-share-type-info","attention_set":{},"removed_from_attention_set":{"36240":{"account":{"_account_id":36240,"name":"Mariusz Adamski","display_name":"Mariusz Adamski","email":"mariusz.adamski@ovhcloud.com","username":"mariuszadamski-ovh","status":"OVH Group"},"last_update":"2023-08-30 14:01:34.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-08-18 17:07:41.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"6413":{"account":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"last_update":"2023-08-30 14:01:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4b8e9344c36aa874b95fce3e0989d34b690159d6","subject":"Include share_type information in notifications","status":"MERGED","created":"2023-08-10 09:21:43.000000000","updated":"2023-08-30 16:07:31.000000000","submitted":"2023-08-30 14:01:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891019-ceilometer-integration-share-type-info","meta_rev_id":"6734e6cdba2f64210fd5cbc5285c8042aa306bc6","_number":891019,"virtual_id_number":891019,"owner":{"_account_id":36240,"name":"Mariusz Adamski","display_name":"Mariusz Adamski","email":"mariusz.adamski@ovhcloud.com","username":"mariuszadamski-ovh","status":"OVH Group"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-30 14:01:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-08-29 18:36:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-08-18 17:07:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-08-29 18:36:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-10 09:54:17.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-10 10:24:09.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-08-10 10:37:29.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-10 10:42:33.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-08-10 10:56:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-10 12:02:14.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-08-15 23:04:15.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-08-18 08:24:01.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-08-18 11:23:03.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-08-22 19:54:22.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2023-08-29 18:36:43.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"f6cb9c3dbb13583b1fd3f251b3af91dc9d8efe4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36240,"name":"Mariusz Adamski","display_name":"Mariusz Adamski","email":"mariusz.adamski@ovhcloud.com","username":"mariuszadamski-ovh","status":"OVH Group"},"date":"2023-08-10 09:21:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15cbe1ff88040f39103d063862f7efd31bdabcd9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-10 09:54:17.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/891019/1/check/DellEMC_PowerFlex_Manila_dhss_false/cb9ab6c/DellEMC_PowerFlex_Manila_dhss_false/184 : FAILURE in 32m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"1b4acfcbef49111735e0d22a5752222cc2d40cd6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-10 10:24:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/19/891019/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/e7b0e6b/ : SUCCESS in 1h 01m 33s","accounts_in_message":[],"_revision_number":1},{"id":"49895d8f093e70add49bcd7f393a5602dc3e88b2","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-08-10 10:37:29.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-19-891019-1/ns5_manila : SUCCESS in 1h 15m 5s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"acb006fab26d4bdb25b1aae5c4a42a27bbb0d31b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-08-10 10:42:33.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-19-891019-1/NFS/ : SUCCESS in 1h 20m 27s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"e6a5322817b3d953a6fd01f7687470dbd14be0e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 10:56:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa3500946f8f43b18fd82857c0d01051\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdf8cd3dd394465a8e9708a93fce7632 : SUCCESS in 10m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09e11f47b3a34f4d95d3872005918a4d : FAILURE in 24m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5bf2abeea26462bbb5b5b0def581c59 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05b7aa42fbf744cb9bf6750637a2a20b : FAILURE in 8m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf73c49f570a45598d8d612c4647dc63 : FAILURE in 10m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0900a341f954a79bd74acdbf66d6bc4 : FAILURE in 9m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/67de850a810b430ab100162ce5df13bd : SUCCESS in 1h 15m 24s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b1718e795efd4803b2e955a2ca603254 : FAILURE in 39m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6943d697fdc94b5cbba17bac330b1a0a : FAILURE in 45m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/4abc02ff019b480f8de47f100e2adcf7 : SUCCESS in 59m 28s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8c43205b29394779a5a9a7e2de0d30bb : SUCCESS in 1h 20m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7452bfa0f4ad447985f509886bbc2f42 : SUCCESS in 1h 08m 39s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/35353d7c1fe64369bd15e0fb51fb3abe : SUCCESS in 5m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/37f5d6757e104bfd8172574e7398b744 : SUCCESS in 5m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5077f081553d4293bb6542ff560aff8f : SUCCESS in 22m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/82fbbbec74d4429ca391ee2dbdeeb8a2 : SUCCESS in 44m 27s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6d86000a0834417fa449616c78be39e4 : FAILURE in 36m 43s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/872ea83a5c06414195c7a0306ca44aad : SUCCESS in 47m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b93ac9846baa45cd9d15c3497e27d213 : SUCCESS in 20m 37s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/0fe0d571f0a24c26b2e7b3f43418fc34 : SUCCESS in 1h 27m 35s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/bcbe6bc2399045b9b39770502fc60b3d : SUCCESS in 38m 44s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8d8ac5f6b3a44395bd22867c334a691c : SUCCESS in 1h 00m 41s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/8a872bae8695448f9eef7d1d32a78127 : SUCCESS in 50m 12s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/db6e635ffb85407ebbd05b949b35b646 : SUCCESS in 40m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"967a6dd398aed08cb6bd5f34df18b08838b23321","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-10 12:02:14.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/891019/1/check/DellEMC_Unity_Manila_dhss_true/851f632/DellEMC_Unity_Manila_dhss_true/344 : SUCCESS in 2h 40m 16s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/891019/1/check/DellEMC_Unity_Manila_dhss_false/a3c8ba0/DellEMC_Unity_Manila_dhss_false/368 : SUCCESS in 1h 17m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"7778c1d55febd5479ce05f4f4bcbbbe6e36ffd9f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-15 23:04:15.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"71ce9c6a0d4a11753b7c3d7d9193140c7f6cde83","author":{"_account_id":36240,"name":"Mariusz Adamski","display_name":"Mariusz Adamski","email":"mariusz.adamski@ovhcloud.com","username":"mariuszadamski-ovh","status":"OVH Group"},"date":"2023-08-16 06:45:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7750a2a1037b96d6fd1c37f32b84444e0e31eae3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36240,"name":"Mariusz Adamski","display_name":"Mariusz Adamski","email":"mariusz.adamski@ovhcloud.com","username":"mariuszadamski-ovh","status":"OVH Group"},"date":"2023-08-18 06:54:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"32e352ddf22726cf06155f2189aed11cc8dfbe8d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-18 07:20:15.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/891019/2/check/DellEMC_Unity_Manila_dhss_true/5f0c753/DellEMC_Unity_Manila_dhss_true/416 : FAILURE in 25m 30s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/891019/2/check/DellEMC_Unity_Manila_dhss_false/d16e6c5/DellEMC_Unity_Manila_dhss_false/441 : FAILURE in 12m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"ffbdea802a4500b09571a6367694b43779f5bf64","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-18 08:02:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/19/891019/2/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/50ebefc/ : SUCCESS in 1h 06m 25s","accounts_in_message":[],"_revision_number":2},{"id":"f16387f3200ef8767dde62c7bdf4ac30a5baffa6","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-08-18 08:05:11.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-19-891019-2/NFS/ : SUCCESS in 1h 10m 24s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"467817f6b1f0341adee87879b20c0faacc24409b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-08-18 08:07:44.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-19-891019-2/ns5_manila : SUCCESS in 1h 12m 48s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"16a9f5cdff06899e1273ed0786ac3fc7d385541f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 08:09:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc2f12a9fc6a49aa94d9584adf3fee65\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e3caba1ed4b4846a59eaf4a8999dcd4 : SUCCESS in 10m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0dae415343e43978d72052c7f742446 : SUCCESS in 25m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b30ac73c6a749c28b70ef2b51515173 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b013790ba424bf18caa3b82c0b8915f : SUCCESS in 8m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c215218a4cd842148bbeeace703da5e7 : SUCCESS in 8m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f424ed7bc05444e29d89cf73576462ac : SUCCESS in 9m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5fb9c3feb70d4c719282848c19c65f99 : RETRY_LIMIT in 1m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b699ced9ea540699a997ce8ff193663 : SUCCESS in 6m 38s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9cc597dd954246b1afe77cdd38607cfe : SUCCESS in 1h 00m 35s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1e88ffada998434689a7a1591e9dfde5 : FAILURE in 47m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/556b3aa9223640f4bd79f9c1449df54e : FAILURE in 42m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f20f165d05974cd9b854a3124b87d9b1 : SUCCESS in 52m 22s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/5db5a2451d45437f964fb12d98232242 : FAILURE in 30m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/77a68b644af146e0b1fae3152e902c22 : SUCCESS in 1h 14m 15s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/971e5e15ef234b7397e3755b60c33726 : SUCCESS in 4m 49s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5dc24fdc9aa24f66b7235f1be87c9af6 : SUCCESS in 5m 26s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/395505c6f7074c46b45428fdb9f879d1 : SUCCESS in 39m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/96c71eeffc2141d6a77dd1aaf017c4fd : SUCCESS in 44m 38s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/60e78187f7264d4f825ad842aa1f8a5a : SUCCESS in 59m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/a2c1f8f166354093bdb2e59326448218 : FAILURE in 23m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/51a18e1ace4b42e985bbf3d251798243 : FAILURE in 28m 07s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/b89f5043128b47d5b6ba8a7c6d88d361 : FAILURE in 23m 21s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/bc00c85fb0da448cbd095b26b225ae2e : SUCCESS in 1h 03m 38s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1f59b8edd5bf4defa33a50bed1d27aaa : SUCCESS in 1h 05m 16s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/174bdf7293cd4d5c9d2558dccf060d8e : FAILURE in 27m 59s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/2190a013db5f4c24976905a0e452fb32 : FAILURE in 29m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d6b75ac0ea51f6de1b846f082c99e9c87918a38","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-18 08:24:01.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/891019/2/check/DellEMC_PowerStore_Manila_dhss_false/a2649f2/DellEMC_PowerStore_Manila_dhss_false/89 : FAILURE in 14m 17s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"301abc7d01decfcbcb9983f892a56a72a0bd947c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-18 08:35:29.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/891019/2/check/DellEMC_Unity_Manila_dhss_true/837a53d/DellEMC_Unity_Manila_dhss_true/417 : FAILURE in 25m 45s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/891019/2/check/DellEMC_Unity_Manila_dhss_false/c49f0e7/DellEMC_Unity_Manila_dhss_false/442 : FAILURE in 12m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"da999ee74e56a9a4dbfe33797f966d3a1bd03244","author":{"_account_id":36240,"name":"Mariusz Adamski","display_name":"Mariusz Adamski","email":"mariusz.adamski@ovhcloud.com","username":"mariuszadamski-ovh","status":"OVH Group"},"date":"2023-08-18 08:49:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b403bb103b0171801c9b25a1d96e3546cd279880","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-18 11:23:03.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/891019/2/upstream-check/manila-tempest-plugin-ontap-dhss/39872d7/ : TIMED_OUT in 3h 12m 19s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/891019/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/bebe602/ : FAILURE in 1h 29m 50s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"e38a0f0a586558a1f851297d28f7b631c7b50ad0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-18 17:07:41.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4d2cc874ce347de62bf0a67f2a0da48ad74d998b","author":{"_account_id":36240,"name":"Mariusz Adamski","display_name":"Mariusz Adamski","email":"mariusz.adamski@ovhcloud.com","username":"mariuszadamski-ovh","status":"OVH Group"},"date":"2023-08-23 15:40:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8a5aee43d11b6162674619dfcb5d893a67a442a3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-23 18:54:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"64033a723e71f74ec9176cf3c65e108d96f145b3","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-08-29 18:36:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7bec3374177a8664663788e75ff4781c273c8414","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 18:37:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"62712a90153e54e554198fcbbab7e6307d92054e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 19:33:49.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9a0f35a086741f2a5be95403d44c3aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86b6b23db47643d081506c6770db1fc1 : FAILURE in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbc363bfc3b547b4bdf9de24bbcf3dba : FAILURE in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/675f29e72a084822be473e334b798062 : FAILURE in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d5b8cff2074446393c571d54f9fd9cc : FAILURE in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62a63c40ea5444ceb199bfb7a3e92d38 : FAILURE in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9dbe8e79d5b4eb3a017d37a7267139c : SUCCESS in 6m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/be7b9eaaf4f94cfda233c15626bf5519 : FAILURE in 55m 24s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b0ff787a5c5b49a48856d76587a570d5 : SUCCESS in 28m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a81a6df2d7c14b9796769139cba689f9 : SUCCESS in 32m 38s","accounts_in_message":[],"_revision_number":2},{"id":"8f8a29aba99119b8ccf66d45a87b8e6aeaedb26a","author":{"_account_id":36240,"name":"Mariusz Adamski","display_name":"Mariusz Adamski","email":"mariusz.adamski@ovhcloud.com","username":"mariuszadamski-ovh","status":"OVH Group"},"date":"2023-08-30 08:53:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bf9882583f82e862a2a70db6f24332789e01a1c4","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-08-30 11:10:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8b2c0c26a48f02c715191db6aace621bf23bfbf7","author":{"_account_id":36240,"name":"Mariusz Adamski","display_name":"Mariusz Adamski","email":"mariusz.adamski@ovhcloud.com","username":"mariuszadamski-ovh","status":"OVH Group"},"date":"2023-08-30 12:37:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8dabbd635fa9dfd331ce0921cbfd7a6489d13b3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 12:55:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/744bb5e4121242b08261311957ac6f3b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af8f6bc34afb4c89983f839ec38e85f1 : SUCCESS in 10m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbfd3f8b54e84dbb827d460944eb5986 : SUCCESS in 26m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de31516faf924545994bd890441f695c : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c3b199926424275a88ff039f8533c34 : SUCCESS in 10m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0457a45362f84a48b71f06c54e56475c : SUCCESS in 11m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21de9ca8fd864047b6d962f307a00db2 : SUCCESS in 9m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/207b39025da4469ca4a506cb1c9d2cb1 : RETRY_LIMIT in 2m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e25ce93c18e048dfbcf430a6360e67d8 : SUCCESS in 7m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7ef738cff044466ab9e84542fdcb496a : SUCCESS in 1h 03m 11s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0c3d40d83a77453c8a854038b67cc213 : FAILURE in 36m 36s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a052134653dd40ecb9fde584f47af422 : FAILURE in 44m 30s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e1e72a11ac6f4678a94c943af6140b9b : SUCCESS in 52m 15s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/eb2da0666089435c8d8d5225de83f24e : SUCCESS in 1h 39m 19s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a41299ebe3054fa8b6a693d5394d6af3 : FAILURE in 41m 17s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/880c1ebc5c2147c692eefec6a31388e1 : SUCCESS in 4m 22s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/525dce6fec554e938dfb50ea454e8c8e : SUCCESS in 5m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/be5c5a4f7fb54964aebd55aa378820eb : SUCCESS in 34m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/12c82988f3524fb1bf26dd4e4c714047 : SUCCESS in 41m 29s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/29a9c8e63885430e9b14117aecc0f773 : SUCCESS in 56m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/77aec05d120941759f08ccd11e5a597a : FAILURE in 28m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/353eec0401c1491a87b9629d246c3b59 : FAILURE in 30m 32s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/f2801dddf9714511beafaff758b43f1e : SUCCESS in 1h 24m 38s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4d701ef8eda24155b3fce071f3a2bfab : SUCCESS in 1h 05m 53s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9daa12feb7684bedab24d97b66de7ae7 : SUCCESS in 57m 03s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/84a2c0b704814a26a70b4a28ca369ecf : SUCCESS in 30m 41s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c67ab605f65b42f99689317b6e634b45 : SUCCESS in 50m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e7fc635a3f95b60ec0951fe71da34247b98f8897","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 12:56:07.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5d6bb17cb616a727dd119918ae62d8daf105ec21","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-08-30 13:14:18.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/19/891019/2/check/DellEMC_Unity_Manila_dhss_true/0ae5659/DellEMC_Unity_Manila_dhss_true/486 : FAILURE in 18m 25s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/891019/2/check/DellEMC_Unity_Manila_dhss_false/2937599/DellEMC_Unity_Manila_dhss_false/512 : FAILURE in 12m 19s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"1bffc124663e116a9cb41900d6bfe91df3899123","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-08-30 13:31:45.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/891019/2/check/DellEMC_PowerFlex_Manila_dhss_false/26b4b3e/DellEMC_PowerFlex_Manila_dhss_false/272 : SUCCESS in 35m 51s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"fa711c1f7a62c83b24815c672a63b68276bb7116","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-08-30 13:34:51.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/19/891019/2/check/DellEMC_PowerStore_Manila_dhss_false/4d9dc1c/DellEMC_PowerStore_Manila_dhss_false/120 : SUCCESS in 38m 58s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"d283ef9f87b8c341b8fcab508c94ca86ff17e482","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 14:01:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab439fea04b0458ca18a2fc8133f34ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef63a17cafd04ec5b5c18be4c8c50bdb : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ba54fedb2da46659e4bbec4fc17c96a : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/503c4a8baa4f48fe9a7543ee0cae6823 : SUCCESS in 10m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67bc4fa1602b48218ee3df9697c5b841 : SUCCESS in 16m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/003cd312d13745b9ada8255ada12be6b : SUCCESS in 10m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b30cc7c5fb114da6a4ff72fed4ff3cd9 : SUCCESS in 7m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1afb1b0b39db4a349f86e36e487bcf48 : SUCCESS in 1h 04m 14s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d3a40c66a0514d378d7dceb32f1df636 : SUCCESS in 34m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/42b115837ebb4758be7fa912421c81eb : SUCCESS in 45m 09s","accounts_in_message":[],"_revision_number":2},{"id":"c6b0c29edfa009daf8225839419f2f876c112c5f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 14:01:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e5557a8f9b4d41b5da51bc81a5d45ccd6bbc452b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 14:02:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7c0fdc7a2c2437caf0a50932e9c591d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e7e975303db43068aec8b68fca0dca7 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d2e2d59511041a1abbd1c104bff8825 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2},{"id":"6734e6cdba2f64210fd5cbc5285c8042aa306bc6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-30 16:07:31.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/891019/2/upstream-check/manila-tempest-plugin-ontap-dhss/08e2bea/ : FAILURE in 2h 30m 01s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/19/891019/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/c15d150/ : TIMED_OUT in 3h 09m 56s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"94b71a5faf5fe161e2a74b36f99d6a0bd17b32a5","revisions":{"ea6bf3fcc719521d093c6b28527d6acbad55d4e0":{"kind":"REWORK","_number":1,"created":"2023-08-10 09:21:43.000000000","uploader":{"_account_id":36240,"name":"Mariusz Adamski","display_name":"Mariusz Adamski","email":"mariusz.adamski@ovhcloud.com","username":"mariuszadamski-ovh","status":"OVH Group"},"ref":"refs/changes/19/891019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/19/891019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/891019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/891019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/891019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/19/891019/1"}}},"commit":{"parents":[{"commit":"364000c140d6e12d59a8e04fb0398aef1d6badaf","subject":"Merge \"NetApp: Improve REST API cover and fix internals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/364000c140d6e12d59a8e04fb0398aef1d6badaf"}]}],"author":{"name":"Mariusz Adamski","email":"mariusz.adamski@ovhcloud.com","date":"2023-08-09 14:35:08.000000000","tz":120},"committer":{"name":"Mariusz Adamski","email":"mariusz.adamski@ovhcloud.com","date":"2023-08-09 14:35:08.000000000","tz":120},"subject":"Include share_type information in notifications","message":"Include share_type information in notifications\n\nFor the purposes of billing, it\u0027s useful to provide the information\nabout share type in oslo.messaging notifications.\n\nDifferent share types can have different quality of service, limits,\nquotas or even be backed by entirely different backends, so it makes\nsense to bill shares of different types differently. For example,\na customer may want to choose to create a share with higher performace\nvia an appropriate share type and thus be charged more for such a share.\n\nChange-Id: I4b8e9344c36aa874b95fce3e0989d34b690159d6\nSigned-off-by: Mariusz Adamski \u003cmariusz.adamski@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ea6bf3fcc719521d093c6b28527d6acbad55d4e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ea6bf3fcc719521d093c6b28527d6acbad55d4e0"}]},"branch":"refs/heads/master"},"94b71a5faf5fe161e2a74b36f99d6a0bd17b32a5":{"kind":"REWORK","_number":2,"created":"2023-08-18 06:54:27.000000000","uploader":{"_account_id":36240,"name":"Mariusz Adamski","display_name":"Mariusz Adamski","email":"mariusz.adamski@ovhcloud.com","username":"mariuszadamski-ovh","status":"OVH Group"},"ref":"refs/changes/19/891019/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/19/891019/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/891019/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/891019/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/19/891019/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/19/891019/2"}}},"commit":{"parents":[{"commit":"364000c140d6e12d59a8e04fb0398aef1d6badaf","subject":"Merge \"NetApp: Improve REST API cover and fix internals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/364000c140d6e12d59a8e04fb0398aef1d6badaf"}]}],"author":{"name":"Mariusz Adamski","email":"mariusz.adamski@ovhcloud.com","date":"2023-08-09 14:35:08.000000000","tz":120},"committer":{"name":"Mariusz Adamski","email":"mariusz.adamski@ovhcloud.com","date":"2023-08-18 06:53:56.000000000","tz":120},"subject":"Include share_type information in notifications","message":"Include share_type information in notifications\n\nFor the purposes of billing, it\u0027s useful to provide the information\nabout share type in oslo.messaging notifications.\n\nDifferent share types can have different quality of service, limits,\nquotas or even be backed by entirely different backends, so it makes\nsense to bill shares of different types differently. For example,\na customer may want to choose to create a share with higher performace\nvia an appropriate share type and thus be charged more for such a share.\n\nChange-Id: I4b8e9344c36aa874b95fce3e0989d34b690159d6\nSigned-off-by: Mariusz Adamski \u003cmariusz.adamski@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/94b71a5faf5fe161e2a74b36f99d6a0bd17b32a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/94b71a5faf5fe161e2a74b36f99d6a0bd17b32a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
