)]}'
{"id":"openstack%2Fmanila~891756","triplet_id":"openstack%2Fmanila~stable%2F2023.1~I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd","project":"openstack/manila","branch":"stable/2023.1","topic":"bug/1982808-stable/2023.1","attention_set":{},"removed_from_attention_set":{"31721":{"account":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"last_update":"2023-09-21 16:56:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd","subject":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up","status":"MERGED","created":"2023-08-18 18:37:52.000000000","updated":"2023-09-21 16:57:36.000000000","submitted":"2023-09-21 16:56:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891756-bug/1982808-stable/2023.1","cherry_pick_of_change":849849,"cherry_pick_of_patch_set":7,"meta_rev_id":"223d3b56144c875c55204976e6a2c2dc4fa74e34","_number":891756,"virtual_id_number":891756,"owner":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-21 16:56:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-09-21 16:09:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-09-21 15:58:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-09-21 16:09:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-18 19:39:24.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-08-18 21:07:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-19 00:24:23.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-09-21 15:58:31.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-09-21 16:09:38.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"83f2f4832057c2ad672b1c3ba48d6579e6953c7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"date":"2023-08-18 18:37:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0a7896f53c412ec77b002554cc25ea035768a9f7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-18 19:39:24.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/56/891756/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/87963cb/ : SUCCESS in 59m 42s","accounts_in_message":[],"_revision_number":1},{"id":"8783a1a19edc6dd4dcbdd54924e1613ec4f4e2ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 21:07:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7633b4654a764b2dbee67212514ae03a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/169f58ca612242489b340657ee73adcb : SUCCESS in 9m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/479b1b8b32b244e88f642da60eed1c94 : SUCCESS in 25m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5b563ea72cc44e3b8498d8759d653e3 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4dde456dce654dd3b9e1f2e32273367a : SUCCESS in 13m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0d71b476f7fc43d3bf22753ffa0926ad : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e623fcbad82f4cfcb2b8aeab18da46b3 : SUCCESS in 6m 42s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8be81c9e8d7a422992d7e93c67ec3ed4 : SUCCESS in 1h 57m 36s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/96b2e92261cf426ba843c6be8e06d9ef : FAILURE in 2h 24m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/02c0f9419bd645779a8d729a0364a93c : SUCCESS in 1h 07m 41s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7d4aefa93955403fac360fc4ada99148 : FAILURE in 26m 13s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/3625672821c643c1b7e900608324b2a6 : FAILURE in 22m 43s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e55fcd9dae724bfdad89152adac58f80 : FAILURE in 18m 17s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9f0c85e1bf1742d58f1bde46d951b0d4 : SUCCESS in 4m 16s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a669bd731d364ebc988e6764ada726e1 : SUCCESS in 3m 59s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/66da1c7a072e42898c9ede86f497ce3b : SUCCESS in 25m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2a98a2b023504b01a1174949754bb4b3 : SUCCESS in 29m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0b7cfaa1653f4ef1bb53a5c0915c399e : FAILURE in 14m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/79dfe61004e24cec90e3d23d3c4a99a6 : SUCCESS in 24m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/83f1e13276804ee5b9fd27f7931fccdf : SUCCESS in 28m 35s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/b85bdcde2f8f4b258f4194653c3ca689 : SUCCESS in 1h 02m 14s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/dd4ab5fbbfa647988821b7a2734f6e21 : SUCCESS in 1h 03m 29s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/42d927b159bb4c4da38b200192f76224 : SUCCESS in 1h 05m 59s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c032666a104c424fa108b75ea605e8f6 : SUCCESS in 30m 48s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/fb94626c4b704148bfa33e5c40ff1ead : SUCCESS in 50m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0acd87bd021894f8da7147f4c04eed023bb1a795","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-19 00:24:23.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/891756/1/upstream-check/manila-tempest-plugin-ontap-dhss/721e853/ : TIMED_OUT in 3h 15m 44s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/891756/1/upstream-check/manila-tempest-plugin-ontap-dhss-rest/c4208eb/ : POST_FAILURE in 2h 55m 15s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/891756/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/89391c0/ : SUCCESS in 2h 49m 01s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/891756/1/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/694e590/ : POST_FAILURE in 2h 18m 02s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"e1a5a46244f17898b96fa9586472da7a3b57fd47","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-21 15:58:31.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c11db8b9e420e3533bac4548e5925e940fefa9d","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-21 16:09:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a430f1b9a8724320b90ac4f7005f627d73611386","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 16:10:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85147047dbd4b8d8c7b2e8b74603a5f58aae85c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 16:56:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e413105fbb3a4db9beb2f7160101737b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d54402ed3344a4690de4e365340cfe8 : SUCCESS in 10m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e893f5acfa74b28869fa6d6244d6cf6 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42a34a97d6b74ac88f92b54838a713ac : SUCCESS in 8m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/524ee6386dc2409ea298ea794c1e7475 : SUCCESS in 9m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1b6596f88114329a931350a3381ba22 : SUCCESS in 6m 36s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/68501a8e35b8481cafed43195936e02d : SUCCESS in 40m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a06631c44cea4134a3934aacee69fa6f : SUCCESS in 38m 17s","accounts_in_message":[],"_revision_number":1},{"id":"bd39bacfedef2dec4f81bc9119463ab17c979fbd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 16:56:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"223d3b56144c875c55204976e6a2c2dc4fa74e34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 16:57:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/556233bd1a7b4619bc3e8a8a2d3f0b24\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deced815c2c64c42ae24c8bf111e2d91 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e61fcff187774516a90f38b1d1db67ae : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e85a081f1c48fa02f728e077e09be08587c7c1e1","revisions":{"e85a081f1c48fa02f728e077e09be08587c7c1e1":{"kind":"REWORK","_number":1,"created":"2023-08-18 18:37:52.000000000","uploader":{"_account_id":31721,"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","username":"felipefutty"},"ref":"refs/changes/56/891756/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/56/891756/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/891756/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/891756/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/891756/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/56/891756/1"}}},"commit":{"parents":[{"commit":"a4971f2cc5fa6bd81a1c245d0b06765208cffd04","subject":"Merge \"[coordination] backend_url should be secret\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a4971f2cc5fa6bd81a1c245d0b06765208cffd04"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2022-07-19 08:33:58.000000000","tz":-180},"committer":{"name":"Felipe Rodrigues","email":"felipefuty01@gmail.com","date":"2023-08-18 18:37:52.000000000","tz":0},"subject":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up","message":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up\n\nReplica promote is retaining unneeded snapshots from previous\nSnapMirror relationships and increasing the amount of space\nconsumed from snapshots in the storage system.\n\nThis patch fixes the issue by calling the snapmirror release\noperation after resync completes its transferring, which allows\nthe SnapMirror software to properly cleanup unneeded resources.\n\nCloses-Bug: #1982808\nChange-Id: I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd\n(cherry picked from commit 38b13bb40b879adcc06cef0b0fa629e4b00f7122)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e85a081f1c48fa02f728e077e09be08587c7c1e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e85a081f1c48fa02f728e077e09be08587c7c1e1"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
