)]}'
{"id":"openstack%2Fmanila~891757","triplet_id":"openstack%2Fmanila~stable%2Fzed~I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd","project":"openstack/manila","branch":"stable/zed","topic":"bug/1982808-stable/2023.1-stable/zed","attention_set":{},"removed_from_attention_set":{"36178":{"account":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"last_update":"2023-09-22 16:56:46.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-09-22 16:11:40.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}},"hashtags":[],"change_id":"I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd","subject":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up","status":"MERGED","created":"2023-08-18 18:49:47.000000000","updated":"2023-09-22 16:58:06.000000000","submitted":"2023-09-22 16:56:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891757-bug/1982808-stable/2023.1-stable/zed","cherry_pick_of_change":891756,"cherry_pick_of_patch_set":1,"meta_rev_id":"388b4365f976babeb349066b3d3e299357c52c2f","_number":891757,"virtual_id_number":891757,"owner":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-22 16:56:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-09-22 11:37:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-09-22 16:11:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-21 20:06:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2023-09-22 16:11:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-18 20:02:56.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-08-18 21:10:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-21 16:04:56.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-09-21 20:06:03.000000000","updated_by":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"reviewer":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"state":"REVIEWER"},{"updated":"2023-09-22 11:37:10.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"2ae4b20e2555db67189646d2c31ac20f9413ce2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-08-18 18:49:47.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.1.","accounts_in_message":[],"_revision_number":1},{"id":"0b1bd58be6d0ce7bfe97cce1125308b0f93387a2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-18 20:02:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/57/891757/1/thirdparty-check/pure-devstack-manila-tempest-aio/d080a2d/ : FAILURE in 1h 07m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d3215a1c7f355a18a0f1860b366153e6d5af32db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 21:10:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2009354967674ddea42a0ae63d6944ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e923c2a6836640058c7d61e2077a2fec : SUCCESS in 9m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5000baa2e23446fa368c5e352388eb4 : SUCCESS in 20m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baa0444ee1cf4cadad1c6ba88654a1b3 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76bdf2c28616487185cd117fb26449ae : SUCCESS in 12m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27f133a5177a416ca7ce55c506ed6a13 : SUCCESS in 11m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c29bf4a7451474babd89aeeec0da935 : SUCCESS in 9m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87568931a491432f922a4f941e4992da : SUCCESS in 6m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d8624c8e852c46a5982cd02babddb96c : SUCCESS in 1h 04m 53s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3aa0c015bf2241f6ab15b068d528070c : SUCCESS in 2h 15m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2cc88c6f04d84d879ebfdc27241bbb93 : SUCCESS in 1h 46m 38s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/af6fdbbaaf584168b56d39725ec802aa : SUCCESS in 1h 08m 30s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0845fdce0f034d89bc2633363d415ef3 : FAILURE in 21m 59s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/222d9ca8ce9846738f1bf9782621ea29 : FAILURE in 29m 18s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b5cb527e8539448a9e37990380beb439 : SUCCESS in 5m 03s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e92809fde337404cb824b333aa432cb2 : SUCCESS in 5m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7d8f1c1704754426924143f14438cc18 : SUCCESS in 25m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8fdcec8ac46243a9863f6aea5bae7a61 : SUCCESS in 27m 17s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/dc2ae9d8758047239486cc08c0c7e7e1 : FAILURE in 19m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/ff2aa3c45c0944aca8fba3295284beb8 : SUCCESS in 26m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c85bf9c1e9384a029a1d83b66dd70b25 : SUCCESS in 29m 51s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/24f1a6afba27494e9c167ffd3ce29272 : SUCCESS in 1h 21m 41s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a331a7a6e2784c1fb412be42b59e737d : SUCCESS in 1h 08m 11s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/e6da538d330b44538394c33a814e16e8 : FAILURE in 48m 39s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ed32fff89cb741288f9f9363e714730c : SUCCESS in 40m 18s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/d6e18bf4144f476a8a4f46adfcda1445 : SUCCESS in 41m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bcef45fe36f8d072f5315a9b0820e05fe6c84cfe","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-21 16:04:56.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c274150d44900d0e48beea040f630aa8fff4281e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-09-21 18:17:35.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"cdec048254f5262adf515eefcea38a1b59f86084","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-09-21 18:21:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c371f313d6f0f91cec4366cb7fa74f53afe1b866","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-21 19:33:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/57/891757/2/thirdparty-check/pure-devstack-manila-tempest-aio/78a5d2d/ : SUCCESS in 1h 15m 16s","accounts_in_message":[],"_revision_number":2},{"id":"eb608dab4497203b98a7c69ec33c3721897e3071","author":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"date":"2023-09-21 20:06:03.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1fdc1232d9a787c7c1489cff8c6ae647208a7211","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 20:52:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c902efb6715d44fa8c5388bb7419479e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05dfff738cd0423d97a8c85fed160732 : SUCCESS in 11m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f5da71bb2b343dc9def2453a9c0c618 : SUCCESS in 22m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d621239b5b68484f992025f02cb7fb1f : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62a1e5ee40bc446ebd6841e4015f4509 : SUCCESS in 9m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd5cdca279fa4f7c9f49336743e906f6 : SUCCESS in 10m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a0885bbd95147b88f48cd0e5fc3bd41 : SUCCESS in 8m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40e5df4b39ed4cc7b4eaf620a1804e05 : SUCCESS in 8m 00s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/eeea0561df4a4fdcb306354cc9e102e7 : SUCCESS in 2h 29m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5b74035ca5b34eeb8459bc987c6bb4a0 : SUCCESS in 1h 10m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/519c6de22c40481db66784d0dc45406c : SUCCESS in 1h 06m 58s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a191c6dc34a443d781841a137507c89a : SUCCESS in 1h 33m 43s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/13ac06ae57524c7ea43ae06fda38b025 : FAILURE in 32m 49s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f74672b3fb8c4f2cb4a15124009ac89f : SUCCESS in 5m 53s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7cadb984343d4e58b9f4e15eabb30160 : SUCCESS in 5m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1638b94ad62a4df492a347721b93d5cf : SUCCESS in 37m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c1e25c5770ff43f2ac8bcaad1d1e9447 : FAILURE in 41m 02s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7a705f0ba54142fc9502bc4d72d7e225 : FAILURE in 30m 00s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/8d032ae0c0d745c9a14ee458d1bc4b44 : SUCCESS in 26m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0a8f9aa5ac2b471e93a692bcbb98be1b : SUCCESS in 32m 29s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/73e130532c834c268821892fb1732715 : SUCCESS in 38m 46s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/6b9ff6544e964f56b7175bf789aa5130 : SUCCESS in 1h 09m 44s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/af7b205298f44ae3b674b4450222c01b : FAILURE in 53m 03s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1de7b71f469b4e448e271b9974abcd86 : SUCCESS in 41m 01s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/38f1468db5a14b11846904eb093a2537 : SUCCESS in 30m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"12a2ed983a403479de8b3ee4afd185285e3b8a5c","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-09-22 05:12:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dcf7523d2dda130c02f39c12a286ede6e426ec0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 06:46:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/841e270b28554a4e98fbb8057f41c403\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e6c821b43754436b3dcc605d2031196 : SUCCESS in 12m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c8874c2e1ff40d7b94e18b3444ea81e : SUCCESS in 20m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/584e6583d53340af86a5293dcbcb05d2 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4100cced53e740df9a5c6dcf3551af4c : SUCCESS in 7m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/982af49e88244dbfb0d5e8d074636e29 : SUCCESS in 10m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f479d8191dc48b895201124583c0a3d : SUCCESS in 7m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22a142ba3cde43cfa5d6d587cf16ccec : SUCCESS in 6m 49s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/466d172376a9444589f630b38d028b5f : SUCCESS in 1h 26m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b24fb0c0dea44621ba5475cab6c209f2 : SUCCESS in 1h 04m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c249773907744c9b86d908f7352dbea9 : SUCCESS in 1h 10m 23s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/1caabe7f260d46eaa5cb593e5e28cf79 : SUCCESS in 1h 32m 30s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9653dc28920147ab825a6fffaee722e5 : FAILURE in 17m 11s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ec14236416cf47d19c985456e8d69c2d : SUCCESS in 4m 22s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c64542a911f44108acd05075bc515065 : SUCCESS in 5m 26s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3d682d85fa4f448083da0d0dfe829891 : SUCCESS in 32m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/33992ad3737b4417bb2a3e1e814f1f9b : SUCCESS in 31m 46s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1762a33fbe9b41da8dd5102f50c1b0a2 : FAILURE in 22m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/efc3d9975a9b43debbbf4b2cec69bf7e : SUCCESS in 28m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/39fbd11014ee4370adc1b2314d4e58c4 : SUCCESS in 20m 31s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/47e23d6e2d5a4559909cd529f620343a : SUCCESS in 54m 39s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/71a0b43c5bb14b2da8c36a9fd5c53b8f : SUCCESS in 1h 03m 15s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/77543a8abd654bf1bc7452b6c89fc119 : FAILURE in 1h 01m 20s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/8e03fe430a8547d192e3459b502712c2 : SUCCESS in 23m 56s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/9270aa50c0a241dbaf1a4222dacf003d : SUCCESS in 45m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0b043a8a49db80e4f9d54662c0c1f17540ea81c8","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-22 11:37:10.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7215a6cfe2f0b4f8b7ba82da3fcaf2c91eb2acb5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-22 16:11:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a4a008a2fd1a22c7cfb7e243b941d1c8dcdd621f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 16:12:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"483ba7454b3dc3f7329794b7ff82a2ef02b306e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 16:56:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25cec266679d441385b2cd3c10e147f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50cd6117cdf94b4289504e098ce77c16 : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ced6497ee6342099ac0a85e9094a343 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afc689c0202645b587ef40751aa5cb2e : SUCCESS in 9m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ffc9a02bd31476da5a82903f1bbecee : SUCCESS in 11m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5f059fc8a484cacbdaf59d09b9a6b4c : SUCCESS in 5m 30s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/995d8f97626e4d4c9b60236fe792256d : SUCCESS in 26m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/55e611a81aa0446eb89d46f109a51d86 : SUCCESS in 40m 08s","accounts_in_message":[],"_revision_number":2},{"id":"dfabc79a5c9a75ea08be2dd2deef76f0820bb2c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 16:56:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"388b4365f976babeb349066b3d3e299357c52c2f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 16:58:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e75150550670476caf38c8bbd8938427\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/486f8a1f743740bdb59776c36bd6ed9e : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/050757fc14c541e0a068f142df4c7277 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6db352411413818ddd6c02762fda5d67e0219361","revisions":{"6bf0132f14ed62d3b2590f29d1641cd8c09f8e6e":{"kind":"REWORK","_number":1,"created":"2023-08-18 18:49:47.000000000","uploader":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"ref":"refs/changes/57/891757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/57/891757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/57/891757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/57/891757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/57/891757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/57/891757/1"}}},"commit":{"parents":[{"commit":"ecd14b4e3cf4028c15345accf583638eda94e1c7","subject":"Fix duplicate entries in share_server_backend_details","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ecd14b4e3cf4028c15345accf583638eda94e1c7"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2022-07-19 08:33:58.000000000","tz":-180},"committer":{"name":"Saravanan Manickam","email":"manicsaran@gmail.com","date":"2023-08-18 18:49:47.000000000","tz":0},"subject":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up","message":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up\n\nReplica promote is retaining unneeded snapshots from previous\nSnapMirror relationships and increasing the amount of space\nconsumed from snapshots in the storage system.\n\nThis patch fixes the issue by calling the snapmirror release\noperation after resync completes its transferring, which allows\nthe SnapMirror software to properly cleanup unneeded resources.\n\nCloses-Bug: #1982808\nChange-Id: I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd\n(cherry picked from commit 38b13bb40b879adcc06cef0b0fa629e4b00f7122)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6bf0132f14ed62d3b2590f29d1641cd8c09f8e6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6bf0132f14ed62d3b2590f29d1641cd8c09f8e6e"}]},"branch":"refs/heads/stable/zed"},"6db352411413818ddd6c02762fda5d67e0219361":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-09-21 18:17:35.000000000","uploader":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"ref":"refs/changes/57/891757/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/57/891757/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/57/891757/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/57/891757/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/57/891757/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/57/891757/2"}}},"commit":{"parents":[{"commit":"ecd14b4e3cf4028c15345accf583638eda94e1c7","subject":"Fix duplicate entries in share_server_backend_details","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ecd14b4e3cf4028c15345accf583638eda94e1c7"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2022-07-19 08:33:58.000000000","tz":-180},"committer":{"name":"Saravanan Manickam","email":"manicsaran@gmail.com","date":"2023-09-21 18:17:35.000000000","tz":0},"subject":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up","message":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up\n\nReplica promote is retaining unneeded snapshots from previous\nSnapMirror relationships and increasing the amount of space\nconsumed from snapshots in the storage system.\n\nThis patch fixes the issue by calling the snapmirror release\noperation after resync completes its transferring, which allows\nthe SnapMirror software to properly cleanup unneeded resources.\n\nCloses-Bug: #1982808\nChange-Id: I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd\n(cherry picked from commit 38b13bb40b879adcc06cef0b0fa629e4b00f7122)\n(cherry picked from commit e85a081f1c48fa02f728e077e09be08587c7c1e1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6db352411413818ddd6c02762fda5d67e0219361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6db352411413818ddd6c02762fda5d67e0219361"}]},"branch":"refs/heads/stable/zed","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
