)]}'
{"id":"openstack%2Fmanila~891758","triplet_id":"openstack%2Fmanila~stable%2Fyoga~I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd","project":"openstack/manila","branch":"stable/yoga","topic":"bug/1982808","attention_set":{},"removed_from_attention_set":{"36178":{"account":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"last_update":"2023-10-04 02:39:39.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-10-02 17:10:52.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"35677":{"account":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"last_update":"2023-10-04 02:39:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd","subject":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up","status":"MERGED","created":"2023-08-18 18:50:58.000000000","updated":"2023-10-04 02:41:54.000000000","submitted":"2023-10-04 02:39:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891758-bug/1982808","cherry_pick_of_change":891757,"cherry_pick_of_patch_set":1,"meta_rev_id":"abb9319d3b14132c27d49dd335271781f1bec67b","_number":891758,"virtual_id_number":891758,"owner":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-04 02:39:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-10-02 17:10:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2023-10-02 17:10:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-18 20:05:18.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-08-18 20:44:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-21 16:02:50.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2023-09-21 16:04:11.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2023-09-22 17:30:25.000000000","updated_by":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"reviewer":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"state":"REVIEWER"},{"updated":"2023-09-22 19:45:34.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"082757813da4d0292a63c24d87acac30edf55462","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-08-18 18:50:58.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"bf77fd2361e35fd8367f2b3db3227da66b7547be","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-18 20:05:18.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/58/891758/1/thirdparty-check/pure-devstack-manila-tempest-aio/1cfa865/ : FAILURE in 1h 09m 50s","accounts_in_message":[],"_revision_number":1},{"id":"49d09e03e0b7cff5c4ce4ca383657e75ef561a17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 20:44:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b17663204a347b6a1da020410c3cee6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f53bacb0afa485fafcd75064a571ff5 : SUCCESS in 9m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ffc1a8141ac941aeb469949bd369fc15 : SUCCESS in 18m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4352471eaa1849a580add14bf9e3f5e9 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8832bfe0ee074d958b944b8814a17135 : SUCCESS in 9m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56dd19e856d94be39503caaee639b7f9 : SUCCESS in 8m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/071180fd48024a7dae3de76797c7d4bf : SUCCESS in 5m 20s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/28160bc73f5340e088c1b69fe6cbd85e : SUCCESS in 1h 25m 24s\n- manila-tempest-plugin-generic-scenario-stable https://zuul.opendev.org/t/openstack/build/537b858b87c64f0eabf6cfaf8bf02579 : FAILURE in 1h 44m 31s (non-voting)\n- manila-tempest-plugin-generic-stable https://zuul.opendev.org/t/openstack/build/9ef097a0ba16498187edbebd6c56c4b2 : SUCCESS in 1h 02m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native-stable https://zuul.opendev.org/t/openstack/build/4d78f10a07184edaa251ce5dfd8e3046 : SUCCESS in 1h 05m 44s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-stable https://zuul.opendev.org/t/openstack/build/96b5d8af6b32488c8dce8ccdfc94ec47 : SUCCESS in 1h 20m 19s (non-voting)\n- manila-tempest-plugin-zfsonlinux-stable https://zuul.opendev.org/t/openstack/build/e190e7f2bf574eedb7ba0541daa6e092 : FAILURE in 22m 05s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6f730f45e0964f57afc13872ed49c471 : SUCCESS in 4m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e7a2082bdb8d4805a349f322474ad999 : SUCCESS in 4m 57s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fe5b47cfb115472085af281a2ca69984 : SUCCESS in 36m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/938e2c8ba41d40609782662d74a2e672 : SUCCESS in 38m 36s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e364227e34eb4085901735d4aa1dc83b : FAILURE in 17m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/ccc08c5923df428e90900b3d9b6df356 : SUCCESS in 34m 10s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/76ef68c48b854e0ba07ce1e7a26c63d3 : SUCCESS in 31m 29s (non-voting)\n- manila-tempest-plugin-lvm-fips-stable https://zuul.opendev.org/t/openstack/build/ac4b5861621b4665a70ea992bbf6b638 : SUCCESS in 1h 21m 51s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/e49db7636aac4eacaa2a55162f026697 : SUCCESS in 1h 04m 36s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/f2afffca86234f0e8713a9b6229bf7b1 : SUCCESS in 24m 37s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b2e5cc59e21b4af9ba4b87d213454de5 : SUCCESS in 34m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"233cd4e38c5849e46ec0318d155d47c49b773054","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-21 16:02:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"80c13a36cc9f9352c2e8c0e2975b506b50b8d511","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-21 16:04:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d55420ac3717529913011106a03c5fe966101a0a","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-21 16:09:15.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b2ff6974bf42c3fa6d1279d2c1a2a5acffb59e01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-09-21 18:18:34.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c1c1fb038519629c86067f78022b5130237eb775","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-21 18:21:06.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio pure-devstack-manila-tempest-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"477151fb428dd13c6ecfdaccc4e4c91b692c7f67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 20:41:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cec25c72ce24e40a4c04fb35b556e6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d0b5f4c231348b389a0401e3aa55337 : SUCCESS in 10m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b4a5b35d7e74800ad32699be9bbe58e : SUCCESS in 22m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5506d6c44504ee4a55b8ff57534d6f2 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83f2271f802b41c5993ccf3be7810517 : SUCCESS in 9m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69681cb6e61845ffa2a384fbc0f449fd : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fbda87effe44f81a1cc8991598bc5a3 : SUCCESS in 8m 09s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/3056d444d37447549c58501caed142b3 : SUCCESS in 48m 25s\n- manila-tempest-plugin-generic-scenario-stable https://zuul.opendev.org/t/openstack/build/a09dd8ee24fe4bf9967a0302832227dc : FAILURE in 2h 16m 16s (non-voting)\n- manila-tempest-plugin-generic-stable https://zuul.opendev.org/t/openstack/build/e349a667d5b04ef6b65db38b1eff9068 : SUCCESS in 1h 13m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native-stable https://zuul.opendev.org/t/openstack/build/2c0b94c3c46843febbc900da608e0f14 : SUCCESS in 1h 21m 16s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-stable https://zuul.opendev.org/t/openstack/build/f943ebc49eca4b839df4b3b06f4bd32f : FAILURE in 2h 05m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux-stable https://zuul.opendev.org/t/openstack/build/99b03b0fde354077814e092e80d84a19 : FAILURE in 15m 29s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6679a620b32a4bff8c44140b60fff979 : SUCCESS in 4m 34s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/52a5361daecc4a86bf3d5cee9cb3c790 : SUCCESS in 5m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3fe4c32c472b4c85bf98745ca1e3b7f4 : SUCCESS in 34m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/732ea4db588a42308bce9f3cf847bc21 : SUCCESS in 34m 25s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/11aff060b2b643eea0f8a046fbabc9a9 : FAILURE in 22m 00s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/b31d904980404e1593cb358a5350bb4f : SUCCESS in 42m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/43148cf361824692b3781dca7e032e6d : SUCCESS in 21m 41s (non-voting)\n- manila-tempest-plugin-lvm-fips-stable https://zuul.opendev.org/t/openstack/build/a357989ba942499c997a30e43f53c91c : SUCCESS in 40m 58s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/07c83592ab98458c81cb4273e1af1c0a : SUCCESS in 1h 07m 06s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/09509fc2290b4d7e92ec0922f5979795 : SUCCESS in 33m 50s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/db8d95b08bf84e2db0b0ef028513f524 : SUCCESS in 40m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1d70b423334f27f65e251fd6afabca813ba1696e","author":{"_account_id":35677,"name":"Thiago José de Andrade Alvoravel","display_name":"Thiago Alvoravel","email":"thiago.andrade@fit-tecnologia.org.br","username":"thiagoalvoravel"},"date":"2023-09-22 17:30:25.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3e17842079e7a4303408b43ce4c79a341de89b07","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-22 19:45:34.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"375f57d185c96411193095ed6461ac1795f2805f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-09-28 15:15:57.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0563be1054b6a162c17994a398585688e63150af","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-09-28 15:16:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"945f20f7c13814c40dcecef6a40cbc2ab1a9f42a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-28 16:43:39.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/58/891758/3/thirdparty-check/pure-devstack-manila-tempest-aio/d737b65/ : SUCCESS in 1h 25m 34s","accounts_in_message":[],"_revision_number":3},{"id":"0aca544bc77cd22fd88b99ea7d803f7850d41133","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 16:55:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f6f906c086444ca29165f86d70ac09b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a50f7170fd9a47a0afb8d22f396978e3 : SUCCESS in 8m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/899fd014058545b4958cab0dfd5a3e13 : SUCCESS in 16m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18d2595d4f4b4c3ebac7880346e92979 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c9e730750e94105862048b716f0f5a0 : SUCCESS in 9m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/986ae510d4b34297b5e389ae8badbe04 : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0303b28ff0d74160ab998d78e836948c : SUCCESS in 6m 03s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/e8a0945892c048cd90352c08f10b6b86 : SUCCESS in 31m 33s\n- manila-tempest-plugin-generic-scenario-stable https://zuul.opendev.org/t/openstack/build/4fd435a838da4752a7aa6e2d89cea520 : SUCCESS in 1h 37m 41s (non-voting)\n- manila-tempest-plugin-generic-stable https://zuul.opendev.org/t/openstack/build/d5da9bbf8b70471c93fa80adcdaa9388 : SUCCESS in 1h 33m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native-stable https://zuul.opendev.org/t/openstack/build/a35476ee747b4a3e8133c6886013c84c : SUCCESS in 1h 32m 05s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-stable https://zuul.opendev.org/t/openstack/build/bbe127129c854adab14d7bb4b897de62 : SUCCESS in 1h 21m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux-stable https://zuul.opendev.org/t/openstack/build/7210940a238442e0a9d7506062d3f2a3 : SUCCESS in 1h 20m 56s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6fa6df0620e144639812fd7b4b6b3f34 : SUCCESS in 4m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/71b8a3e2bb9545da98e81ff93783b5a7 : SUCCESS in 5m 44s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/624e86daac1b471296c9b9d06c032a3c : SUCCESS in 24m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9da595966f9f45d28ff33605880ad677 : FAILURE in 56m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a98f5ba5c27c46f195f6fe564f09b22a : FAILURE in 1h 06m 10s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/3dfe4b613ea2437fb77790f1ecdb7c5f : SUCCESS in 29m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/89d307ad313b4c1893c876d292d3520e : SUCCESS in 36m 19s (non-voting)\n- manila-tempest-plugin-lvm-fips-stable https://zuul.opendev.org/t/openstack/build/338f43a17cfc4d4f84b9f9e7e65f469d : SUCCESS in 33m 52s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/7e7932eeddc64f5897381bade6ac0205 : SUCCESS in 53m 26s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ea66eb42490e45638bb5a91760294018 : SUCCESS in 43m 33s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/dcbb3baebdb24dbc848017592a711aa9 : SUCCESS in 49m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ac37f6808b26c0ab0e72e310ba5c4338d48235ea","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-09-29 10:20:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"03c9ff8c1586eabaa86469c7082e94d1b2c3b27b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 12:22:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43ee40f4aa7a4e3089a1b04d1c412008\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2048927fd9e4416ba617f00c2e85750a : SUCCESS in 10m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80e8f6043b734c9388c50331cb89b8ee : SUCCESS in 18m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f70bd3267044f73bc7831e0fc31f78a : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a45df50eb864d50956fc7bd5cebdebe : SUCCESS in 7m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46758efa6bf94264a72373ad18bb16b9 : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a250285eb7bc4945b6350dcc72ff2664 : SUCCESS in 6m 49s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/f34e961e948b4264873a709c13a84aa2 : SUCCESS in 30m 02s\n- manila-tempest-plugin-generic-scenario-stable https://zuul.opendev.org/t/openstack/build/df3e769da4e2449a9e021a15f279bb22 : SUCCESS in 1h 55m 33s (non-voting)\n- manila-tempest-plugin-generic-stable https://zuul.opendev.org/t/openstack/build/e5cefeeeb1d34a169843943c3c484fc5 : SUCCESS in 1h 08m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native-stable https://zuul.opendev.org/t/openstack/build/77ccb4e3c22b4fa08f94f84efe5bf006 : FAILURE in 1h 21m 30s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-stable https://zuul.opendev.org/t/openstack/build/74c6ee316e8e453bb8aca2f4f9f7a2fa : FAILURE in 1h 34m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux-stable https://zuul.opendev.org/t/openstack/build/145e300e07f648cdb10892f2b0fb49d1 : SUCCESS in 1h 39m 20s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/bb4a9f420e1749d4a99d399b4de43358 : SUCCESS in 4m 48s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8ad7e282a674468aac3a4d0fc862cb41 : SUCCESS in 5m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7819e545b70e4f81b1157bf74d0de899 : SUCCESS in 24m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fdcccbc163e84aa7841dcd11967c027b : SUCCESS in 31m 48s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c3bfe67d870f4ee6afc7dc6abff3aa9e : SUCCESS in 1h 02m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/797e88b145ff47f99bf7fc59ada23ebb : SUCCESS in 27m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/16de892ff653403e81e8ac567694d68e : SUCCESS in 19m 14s (non-voting)\n- manila-tempest-plugin-lvm-fips-stable https://zuul.opendev.org/t/openstack/build/9083321b64664ec8aa98ffe006382962 : SUCCESS in 51m 00s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/61a545b2dd1c4e5ebde784146051ca92 : SUCCESS in 1h 04m 52s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/baaed321b9c3416f817ce0fa747195eb : SUCCESS in 31m 43s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/d9643d85cf8a4999b5a6a3b3c73bc1fe : SUCCESS in 42m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"74ce8d2caf62fd56fd8d18a0288529dac85b91e6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-02 17:10:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b7f80cdf053e6240b0e27781a34a85d5f189fbba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 17:11:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"71838e1b069be1c99955fe39134b85716bf71b56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 18:26:17.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8a4ac264bec478b9410c453189b35be\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f48d088928c4ae5a71fa04d3aeb61ce : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f492d8e3b68491c98de2d19ea7802e4 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82d509645ea84bb0ad1e4e3cf6917930 : SUCCESS in 8m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/562a0b6de3554abb93063dda0a4a4c27 : SUCCESS in 9m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/903d74088c014868837f0f03acea0542 : SUCCESS in 6m 27s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/c74ff7fa44f046c0becfcdb25c46dfc6 : SUCCESS in 50m 19s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d7bac72fb3384b2d9a8d72bb65490e39 : FAILURE in 34m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/84a30f3b5b8444108cf1d07e31eaea63 : SUCCESS in 46m 00s","accounts_in_message":[],"_revision_number":3},{"id":"3d124fd209d3a93f7cb36102a7598307db4a9478","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-03 23:49:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8dfaa7cffd75b97374c2780100be6c55b65021d6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-03 23:49:13.000000000","message":"Topic bug/1982808-stable/2023.1-stable/zed-stable/yoga removed","accounts_in_message":[],"_revision_number":3},{"id":"0cc8bc49181325e152321c6056d9aedf3e7328f8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-03 23:49:19.000000000","message":"Topic set to bug/1982808","accounts_in_message":[],"_revision_number":3},{"id":"8bb560fe8a895a1bdd3e4c23b274918995265936","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 01:47:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/293f1f7a59d74580ad738c9329de9c06\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf1c4cbc5efb4b8980db98025986a179 : SUCCESS in 8m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7363d9c70aa140afa5e9907f0fa88d94 : SUCCESS in 18m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce0a2ca5c221400b9df462da6e7d73d2 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bc2bc1a0c37473f88b026924b38e070 : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0781491ed5d74ac49af12ae4642ef883 : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcbf58323c5e4754b56de52344921c4a : SUCCESS in 7m 29s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/67c93f1ea4574f918eac0412608e5f6c : SUCCESS in 47m 11s\n- manila-tempest-plugin-generic-scenario-stable https://zuul.opendev.org/t/openstack/build/c9689b50b5db494d84265ff8f922d95a : FAILURE in 1h 53m 43s (non-voting)\n- manila-tempest-plugin-generic-stable https://zuul.opendev.org/t/openstack/build/c7ccd43c0bf84ba8be9b750f6ea77d1e : SUCCESS in 1h 01m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native-stable https://zuul.opendev.org/t/openstack/build/71f3886eef844feea0b6b8544dfe97b8 : SUCCESS in 1h 19m 56s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-stable https://zuul.opendev.org/t/openstack/build/c31a63527ba9408dbdbcf1a92c4fede5 : SUCCESS in 1h 37m 59s (non-voting)\n- manila-tempest-plugin-zfsonlinux-stable https://zuul.opendev.org/t/openstack/build/800d02ad111540f0999a501ab4629c57 : SUCCESS in 1h 00m 40s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/478c4e19b460424686f4dbc82ce8c0d6 : SUCCESS in 5m 33s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e3de5ad5db644b1cb724b9fd2d9f08bc : SUCCESS in 4m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b2ba3bf94d3d4074a440abefac09a404 : SUCCESS in 32m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d18ed868a74c4d318ac6cd97181f3792 : SUCCESS in 42m 44s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fce15c00838b43bdb19ba774976a0f34 : FAILURE in 1h 02m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/0a0506a036714f559511ecada452d277 : SUCCESS in 26m 16s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/178eb907ba774721a15e938368cf733c : SUCCESS in 31m 36s (non-voting)\n- manila-tempest-plugin-lvm-fips-stable https://zuul.opendev.org/t/openstack/build/a8537ca9d05346a2abdfb8f58dd5e08f : SUCCESS in 35m 36s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/7884e64f8e3046b79859b3a34afb6201 : SUCCESS in 1h 02m 00s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e7622eecb5674f10895ef009dddcd299 : SUCCESS in 44m 04s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/12907984336242d5bfde4f18b6c7c516 : SUCCESS in 25m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"473f98a82b510cdf9998d31874f604bdf8161490","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 01:48:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a3e2ccb9213a56502661cb1d98ebb5435ece1dd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 02:39:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/986daac650164784bbc4943798a0e8f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2292e2bac5774bebb9243d64cae8bb71 : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57cae3359b8d4c2c8045f4d7bca6a558 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/389af3a6bc30481bbfc423565bf0fc09 : SUCCESS in 8m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75b32b596498464e9f4424224718beab : SUCCESS in 10m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cdfb1a8de3d4f108a2878aa148ea215 : SUCCESS in 6m 51s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/49552c488bbc496589a613a181ede7c2 : SUCCESS in 42m 39s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/74d4124a75f44b4d96be24ad64c3fc54 : SUCCESS in 30m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9ce807de442b47fca18e71929954d605 : SUCCESS in 46m 33s","accounts_in_message":[],"_revision_number":3},{"id":"4285aea667c49046ab97f5cdd45ab39d6accd980","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 02:39:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"abb9319d3b14132c27d49dd335271781f1bec67b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-04 02:41:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ee5b3f18680461a9e52789f69ed5e81\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b15f2b670ea243939eb6ea27e81db8fd : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d6532ba79b1482dbcf485ae191b9408 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"65bcc27100abbbf838ff02735594da5bae3b05c1","revisions":{"de588e25c09be1197011a5a20a84e8bd6170f968":{"kind":"REWORK","_number":1,"created":"2023-08-18 18:50:58.000000000","uploader":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"ref":"refs/changes/58/891758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/58/891758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/58/891758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/58/891758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/58/891758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/58/891758/1"}}},"commit":{"parents":[{"commit":"3ba4d99288a372e5d9f7cf9f108399af92041553","subject":"Fix Manila API error message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3ba4d99288a372e5d9f7cf9f108399af92041553"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2022-07-19 08:33:58.000000000","tz":-180},"committer":{"name":"Saravanan Manickam","email":"manicsaran@gmail.com","date":"2023-08-18 18:50:58.000000000","tz":0},"subject":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up","message":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up\n\nReplica promote is retaining unneeded snapshots from previous\nSnapMirror relationships and increasing the amount of space\nconsumed from snapshots in the storage system.\n\nThis patch fixes the issue by calling the snapmirror release\noperation after resync completes its transferring, which allows\nthe SnapMirror software to properly cleanup unneeded resources.\n\nCloses-Bug: #1982808\nChange-Id: I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd\n(cherry picked from commit 38b13bb40b879adcc06cef0b0fa629e4b00f7122)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/de588e25c09be1197011a5a20a84e8bd6170f968"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/de588e25c09be1197011a5a20a84e8bd6170f968"}]},"branch":"refs/heads/stable/yoga"},"1dfb34268e913283a77e3e90bd4d1569f7e7392e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-09-21 18:18:34.000000000","uploader":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"ref":"refs/changes/58/891758/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/58/891758/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/58/891758/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/58/891758/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/58/891758/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/58/891758/2"}}},"commit":{"parents":[{"commit":"3ba4d99288a372e5d9f7cf9f108399af92041553","subject":"Fix Manila API error message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3ba4d99288a372e5d9f7cf9f108399af92041553"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2022-07-19 08:33:58.000000000","tz":-180},"committer":{"name":"Saravanan Manickam","email":"manicsaran@gmail.com","date":"2023-09-21 18:18:34.000000000","tz":0},"subject":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up","message":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up\n\nReplica promote is retaining unneeded snapshots from previous\nSnapMirror relationships and increasing the amount of space\nconsumed from snapshots in the storage system.\n\nThis patch fixes the issue by calling the snapmirror release\noperation after resync completes its transferring, which allows\nthe SnapMirror software to properly cleanup unneeded resources.\n\nCloses-Bug: #1982808\nChange-Id: I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd\n(cherry picked from commit 38b13bb40b879adcc06cef0b0fa629e4b00f7122)\n(cherry picked from commit e85a081f1c48fa02f728e077e09be08587c7c1e1)\n(cherry picked from commit 6bf0132f14ed62d3b2590f29d1641cd8c09f8e6e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1dfb34268e913283a77e3e90bd4d1569f7e7392e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1dfb34268e913283a77e3e90bd4d1569f7e7392e"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"},"65bcc27100abbbf838ff02735594da5bae3b05c1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-09-28 15:15:57.000000000","uploader":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"ref":"refs/changes/58/891758/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/58/891758/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/58/891758/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/58/891758/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/58/891758/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/58/891758/3"}}},"commit":{"parents":[{"commit":"3ba4d99288a372e5d9f7cf9f108399af92041553","subject":"Fix Manila API error message","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3ba4d99288a372e5d9f7cf9f108399af92041553"}]}],"author":{"name":"Fernando Ferraz","email":"sfernand@netapp.com","date":"2022-07-19 08:33:58.000000000","tz":-180},"committer":{"name":"Saravanan Manickam","email":"manicsaran@gmail.com","date":"2023-09-28 15:15:57.000000000","tz":0},"subject":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up","message":"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up\n\nReplica promote is retaining unneeded snapshots from previous\nSnapMirror relationships and increasing the amount of space\nconsumed from snapshots in the storage system.\n\nThis patch fixes the issue by calling the snapmirror release\noperation after resync completes its transferring, which allows\nthe SnapMirror software to properly cleanup unneeded resources.\n\nCloses-Bug: #1982808\nChange-Id: I516fb3575e30d18d971d6a1b7f3b9ad7120c3bbd\n(cherry picked from commit 38b13bb40b879adcc06cef0b0fa629e4b00f7122)\n(cherry picked from commit e85a081f1c48fa02f728e077e09be08587c7c1e1)\n(cherry picked from commit 6db352411413818ddd6c02762fda5d67e0219361)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/65bcc27100abbbf838ff02735594da5bae3b05c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/65bcc27100abbbf838ff02735594da5bae3b05c1"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
