)]}'
{"id":"openstack%2Fmanila~892349","triplet_id":"openstack%2Fmanila~stable%2Ftrain~I737dbc5138c81e6708b9ebae7e7c11622e124a3d","project":"openstack/manila","branch":"stable/train","topic":"bug/2024578","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-12-21 14:08:24.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I737dbc5138c81e6708b9ebae7e7c11622e124a3d","subject":"skip periodic update on replicas in \u0027error_deleting\u0027","status":"ABANDONED","created":"2023-08-22 17:49:34.000000000","updated":"2023-12-21 14:08:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":892348,"cherry_pick_of_patch_set":1,"meta_rev_id":"521d17eb6900d8351aac2cdbd42069e3dc5928ae","_number":892349,"virtual_id_number":892349,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-22 18:18:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-22 17:49:34.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"CC"},{"updated":"2023-08-22 18:18:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-22 19:46:18.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"eb24f5b3d9072b74f6b974b9f2d1c60f1365b995","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-22 17:49:34.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"0438a5c18a669d001dd62b6b5870bfbff919df24","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-22 17:50:05.000000000","message":"Topic bug/2024556-stable/2023.1-stable/zed-stable/yoga-stable/xena-stable/wallaby-stable/victoria-stable/ussuri-stable/train removed","accounts_in_message":[],"_revision_number":1},{"id":"752ed596f3c034265db64fad3bee924349391e47","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-22 17:50:08.000000000","message":"Topic set to bug/2024556","accounts_in_message":[],"_revision_number":1},{"id":"c94801d80b5cdcda2c5263a6c4d90b083439e51d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 18:18:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cff98c9ad8c2481c916e35abac954625\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e23eeb6c64d94ecc987479e8c436f868 : SUCCESS in 9m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb3b856543ed44f3a73a824ef51368dd : SUCCESS in 12m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/241657b8ce924987a2e96cb420b0274e : FAILURE in 7m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2b470a3f636442749764655b45383509 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/766efbfd212e41038574ca11b7d758d2 : SUCCESS in 7m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a16e83956c6c486ab57f216406c714fd : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f50d52eba4d54f51ab82ab65e0b216c0 : SUCCESS in 5m 32s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/171394641b474939a7c08460fd174592 : SUCCESS in 6m 18s\n- manila-tempest-dsvm-mysql-generic https://zuul.opendev.org/t/openstack/build/1574c63a50c94fdcaf25ae363853b506 : FAILURE in 16m 22s (non-voting)\n- manila-tempest-dsvm-postgres-container https://zuul.opendev.org/t/openstack/build/d4b9be3fad524cccab36233b5160aae5 : FAILURE in 7m 52s (non-voting)\n- manila-tempest-dsvm-postgres-zfsonlinux https://zuul.opendev.org/t/openstack/build/a92ba1596164483a8bcac86e3a05e431 : FAILURE in 18m 48s (non-voting)\n- manila-tempest-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b9ddcbaddf624b7fa736e975b376cd8a : FAILURE in 7m 05s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-native https://zuul.opendev.org/t/openstack/build/5e3395a601824db4b88953ea80044f49 : FAILURE in 18m 17s (non-voting)\n- manila-tempest-minimal-dsvm-cephfs-nfs https://zuul.opendev.org/t/openstack/build/a3f4fb14a26f4ea48e031aef78645e95 : FAILURE in 15m 20s (non-voting)\n- manila-tempest-dsvm-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0bec4182019341759f8e5876f9ea79f6 : FAILURE in 20m 10s (non-voting)\n- manila-tempest-minimal-dsvm-dummy https://zuul.opendev.org/t/openstack/build/8ec1b3a60f9b41ceb5ff67daae339f08 : FAILURE in 15m 38s\n- manila-tempest-minimal-dsvm-dummy-py2 https://zuul.opendev.org/t/openstack/build/f997c8c520e34564863f922646c4d14c : FAILURE in 19m 00s\n- manila-tempest-minimal-lvm-ipv6-only https://zuul.opendev.org/t/openstack/build/f17fbdf76ce745fca0d9d64f25ce532a : FAILURE in 17m 03s\n- manila-grenade https://zuul.opendev.org/t/openstack/build/ea42a6ceb7f549e09a125e5408a9abe2 : FAILURE in 20m 13s (non-voting)\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/01c28e1c0feb47e1b10c7bd627e8785e : FAILURE in 3m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"563ddca2c09fb85b116dddb704c313223743670e","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-08-22 19:46:18.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-49-892349-1/ns5_manila : FAILURE in 27m 6s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":1},{"id":"6f55c1e613130e814fa63e6a10c0ff48a5d3ceff","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-22 21:38:35.000000000","message":"Topic bug/2024556 removed","accounts_in_message":[],"_revision_number":1},{"id":"183d855471fd4c8bf69cb6e347434538844f9845","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-08-22 21:38:39.000000000","message":"Topic set to bug/2024578","accounts_in_message":[],"_revision_number":1},{"id":"521d17eb6900d8351aac2cdbd42069e3dc5928ae","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-12-21 14:08:24.000000000","message":"Abandoned\n\nTrain is about to transition to End of Life. Open patches needs to be abandoned before branch deletion.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9fc94735052a40027eb468b158e55a0f01e83a95","revisions":{"9fc94735052a40027eb468b158e55a0f01e83a95":{"kind":"REWORK","_number":1,"created":"2023-08-22 17:49:34.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/49/892349/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/892349/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/892349/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/892349/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/892349/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/892349/1"}}},"commit":{"parents":[{"commit":"d3419ed14ba7780b478deec4028d1f0859c5e801","subject":"Merge \"[RBAC] Return 404 if share is inaccessible\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d3419ed14ba7780b478deec4028d1f0859c5e801"}]}],"author":{"name":"Maurice Escher","email":"maurice.escher@sap.com","date":"2023-06-21 12:56:50.000000000","tz":120},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-08-22 17:49:34.000000000","tz":0},"subject":"skip periodic update on replicas in \u0027error_deleting\u0027","message":"skip periodic update on replicas in \u0027error_deleting\u0027\n\nChange-Id: I737dbc5138c81e6708b9ebae7e7c11622e124a3d\nCloses-Bug: #2024578\n(cherry picked from commit ca90c87c478e2edacc8bbe01be1e5c3c6e309baa)\n(cherry picked from commit 9338a3afec2b32f312c5787eb8637cfa92429c42)\n(cherry picked from commit c14f4f0940c7ee1676591f9b06139ec749094c1b)\n(cherry picked from commit 4597939d9e689b2bc941c4b5ad4b2e84f2778698)\n(cherry picked from commit 13a0d80ce61b73c599d922575a3a61a3a70757c2)\n(cherry picked from commit 4a7dcb01071fdc44d8faa4e86dac665d495dca44)\n(cherry picked from commit f5e7cb7b0f3c6d076ac75dde7579cbc9244f5f7d)\n(cherry picked from commit 2bb306614fcc5b724852c3b562133c6866b5997c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9fc94735052a40027eb468b158e55a0f01e83a95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9fc94735052a40027eb468b158e55a0f01e83a95"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
