)]}'
{"id":"openstack%2Fmanila~893256","triplet_id":"openstack%2Fmanila~stable%2F2023.1~Ifb660688c590fb1c32cbe0ae03e0ea444f19cbb8","project":"openstack/manila","branch":"stable/2023.1","topic":"cherrypick-bug/2030976-gn14cght4g","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-09-01 12:54:53.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2023-09-14 01:31:35.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2023-09-14 01:31:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifb660688c590fb1c32cbe0ae03e0ea444f19cbb8","subject":"Add default to read_deleted in context\u0027s from_dict","status":"MERGED","created":"2023-08-31 14:31:03.000000000","updated":"2023-09-14 01:34:57.000000000","submitted":"2023-09-14 01:31:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"893256-cherrypick-bug/2030976-gn14cght4g","cherry_pick_of_change":893223,"cherry_pick_of_patch_set":1,"meta_rev_id":"7ec4ee2c27a49f86dc93984223cef257dbaca006","_number":893256,"virtual_id_number":893256,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-14 01:31:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-09-01 12:54:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-09-14 00:44:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2023-09-14 00:44:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}],"CC":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-31 14:31:03.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"state":"CC"},{"updated":"2023-08-31 15:35:50.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-08-31 16:54:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-31 18:49:50.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-08-31 19:13:35.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2023-08-31 19:13:35.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2023-09-14 00:44:24.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"a970da57cc3c55d44f7274343121260fa54f309e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-31 14:31:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"86ab47f49d4b2e5f06aa3265bdbf1bca56db7f12","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-08-31 15:35:50.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/56/893256/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/dd4b642/ : SUCCESS in 1h 02m 28s","accounts_in_message":[],"_revision_number":1},{"id":"159328a1efaf7db1d95eee30f6cb17c3563a98d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 16:54:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f70c4182b2a4c3a939cc71222c27f2e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d977e0ecd6b9434a87c163b60701d743 : SUCCESS in 10m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14c50a13c81f461cb220b44cd651b9fc : SUCCESS in 21m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e2a254a33174cc5adb6494c0c0e24b0 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34261f486160496f989c55cc9169001e : SUCCESS in 12m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef5546dcc6764689b644c16b063260a9 : SUCCESS in 9m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ee1688f6500f4e7d98bf370e60bed4c0 : SUCCESS in 1h 19m 01s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b5a1089041a04ef6827866fce24c0107 : SUCCESS in 2h 17m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d8d0e86aa397463b982369ba42fd984f : SUCCESS in 57m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8adb8f3de3494335891eb5f158799d0c : FAILURE in 21m 01s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/0025fe99ffd049cb8f4ac15578a41ef6 : FAILURE in 10m 41s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/84d5cf91057548599b32100b096d715f : FAILURE in 21m 06s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e496037ef1174a1fb8146fa51160c23f : SUCCESS in 4m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/61aaa5e2130e45c6b69f7e056095edfb : SUCCESS in 5m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/08da4306e08d4f58b7e7798b43b99eba : SUCCESS in 32m 14s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/bae9c853651a4a0f9ad61c7e301b5845 : SUCCESS in 28m 31s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/64e502640ded4c43a6e90428cadbaea2 : FAILURE in 31m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/5e082ff3dabf475e95775485f22f6b4e : SUCCESS in 31m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f6ef6baca7734e39ab809d9b0619031d : SUCCESS in 21m 49s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/00984bc228b544ca96397b1fdf690513 : SUCCESS in 1h 02m 55s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/1dca9d6e66aa494d978b6d17c3a53f5c : SUCCESS in 1h 09m 39s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/1b171086254346cb817b68e61f7e63ee : SUCCESS in 1h 01m 18s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d924db9666904077ba7b90db7b9de05f : SUCCESS in 33m 30s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/9f1183f6e1f24c699c32176bb568b01c : SUCCESS in 37m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d66040d3fe10cb6c74349ae0a38225b05abbe8f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-08-31 18:49:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/893256/1/upstream-check/manila-tempest-plugin-ontap-dhss/b77e2c3/ : FAILURE in 1h 42m 26s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/56/893256/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/f3e46c5/ : SUCCESS in 1h 54m 40s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"26429c70c6a0a9773bc80d4a9ead11e3570af1af","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-08-31 19:13:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd492cbd417db165f58d65613ab0edcfc87537eb","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-01 12:54:53.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"97d02069e52d5c6a6d9272ef7ef1064ae2a9d0ce","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-14 00:44:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"720dc3fc23c6f2fd40ccc69d0f5cefe7c28d113c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 00:44:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e8c3ff401b8b697f30daeaa8c9f3ef939678837","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 01:31:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b5e6f428b6c4468a5398dc7672c635b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8926c3c816c04dd097a0ec68e684cc15 : SUCCESS in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84c22179e0d346cea880e17d19b8dabf : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ee49342c99148c9a02e108ddfb945ef : SUCCESS in 10m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9693ee747d14dbea5399b15a49ebc5f : SUCCESS in 10m 14s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e374a9941ea046e2b4beda09ecc39dd1 : SUCCESS in 31m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d151febe497241d88a0ecdc204a9d607 : SUCCESS in 41m 53s","accounts_in_message":[],"_revision_number":1},{"id":"897d6ec05103791f88a59f1ba3b885f904646217","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 01:31:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7ec4ee2c27a49f86dc93984223cef257dbaca006","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-14 01:34:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b01460b1c418412c93036db3d9a64b45\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/977fa41556f344f8941c295cfb82c3b9 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5992498e24dc1a7b0b2faacfb33722080eace10a","revisions":{"5992498e24dc1a7b0b2faacfb33722080eace10a":{"kind":"REWORK","_number":1,"created":"2023-08-31 14:31:03.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/56/893256/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/56/893256/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/893256/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/893256/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/56/893256/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/56/893256/1"}}},"commit":{"parents":[{"commit":"a4971f2cc5fa6bd81a1c245d0b06765208cffd04","subject":"Merge \"[coordination] backend_url should be secret\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a4971f2cc5fa6bd81a1c245d0b06765208cffd04"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2023-08-30 15:30:45.000000000","tz":0},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2023-08-31 14:31:03.000000000","tz":0},"subject":"Add default to read_deleted in context\u0027s from_dict","message":"Add default to read_deleted in context\u0027s from_dict\n\nthis is follow-up to Ic94323658c89df1c1ff32f511ca23502317d0f00\n\ncurrently the defaults in __init__ and from_dict of RequestContext\nfor read_deleted property are different:\n- in __init__ it is \u0027no\u0027\n- in from_dict, it is effectively None.\nAnd at the same time, this property is not allowed to be None.\n\nAs a result, the new _sanitize_context in oslo.messaging can not create\na sanitized context copy with error\n\nread_deleted can only be one of \u0027no\u0027, \u0027yes\u0027 or \u0027only\u0027, not None\n\nRelated-Bug: #2030976\nChange-Id: Ifb660688c590fb1c32cbe0ae03e0ea444f19cbb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5992498e24dc1a7b0b2faacfb33722080eace10a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5992498e24dc1a7b0b2faacfb33722080eace10a"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
