)]}'
{"id":"openstack%2Fmanila~893732","triplet_id":"openstack%2Fmanila~stable%2F2023.1~I8271fbb56b02fd112b250d71264105346ee3cec7","project":"openstack/manila","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-01-16 13:59:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8271fbb56b02fd112b250d71264105346ee3cec7","subject":"[DNM] Testing IPv6 job configuration on stable branches","status":"ABANDONED","created":"2023-09-05 13:05:51.000000000","updated":"2024-01-16 13:59:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a71c1527f34853346f771d0c35a060d57a1cb195","_number":893732,"virtual_id_number":893732,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-05 15:51:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-05 13:24:46.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-09-05 15:51:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cd5120902fbfcdbf0262112decc8b48889b95707","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-05 13:05:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ed43172e14c76be938343c68fe27f788905f0dc6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-05 13:24:46.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"94f6819962d270a798d2ae3f44253a00d58b2c8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 15:51:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b1c1ff161b24c4a8be91686a9337a92\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a90a1e9a9cc43d28e52f68c3f66ba9a : SUCCESS in 11m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c7bd1357c944ce0a2a2d41155560fe1 : SUCCESS in 24m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb66cab56b6c430b8b5ea487bb3b56df : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c3011f584884088baf10d9ba7c3be66 : SUCCESS in 10m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/926d9b16da8a4e2c9bb866f91d08da68 : SUCCESS in 11m 26s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d2d12fca7df74706b0671a15e03d6b2e : SUCCESS in 2h 20m 24s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/432ceafe78b04a62a70912654ed8d61f : SUCCESS in 58m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/0883c9a29ebc47f594dea9a604731c44 : FAILURE in 15m 17s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/f72627e7b1c84386afc06ae0186cd3b6 : FAILURE in 19m 36s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5f15f9245f0947df935b18ddba99b46b : FAILURE in 28m 22s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f992bedb6c214fa395df4da056628e90 : SUCCESS in 5m 46s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b6a84829ab164e19a4158709f17c42da : SUCCESS in 5m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/22bb819f5aa843b994020c9f82e1aa1d : SUCCESS in 27m 38s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d6b1dfe41fdf4ac694d253dae3140f45 : FAILURE in 29m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3ebd0eacda1d45ebb11a86c85b8c8f68 : FAILURE in 19m 37s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/1178a4fb3cfb435791d3aa336b762d02 : FAILURE in 12m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2e228684b373413aaa90694f52b2d6a7 : FAILURE in 12m 28s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/615ce167ddfe4b1b8af825e77f47414a : SUCCESS in 1h 00m 54s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/c49ee46a63e9462280dc05565d3d2aa8 : SUCCESS in 1h 03m 05s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/4b21c93ee3f94990bf8160cd38a65853 : FAILURE in 1h 10m 00s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/f9315f221d944ad0bdb4a61d5d9c3c67 : SUCCESS in 42m 40s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/5c0b71c442ee4d0385ffb65949881e53 : SUCCESS in 44m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a71c1527f34853346f771d0c35a060d57a1cb195","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-01-16 13:59:41.000000000","message":"Abandoned\n\nChange used for testing. Not needed anymore.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fca2a4f1bf8630bd5e198b6856a3f18d47251a75","revisions":{"fca2a4f1bf8630bd5e198b6856a3f18d47251a75":{"kind":"REWORK","_number":1,"created":"2023-09-05 13:05:51.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/32/893732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/32/893732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/32/893732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/32/893732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/32/893732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/32/893732/1"}}},"commit":{"parents":[{"commit":"a4971f2cc5fa6bd81a1c245d0b06765208cffd04","subject":"Merge \"[coordination] backend_url should be secret\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a4971f2cc5fa6bd81a1c245d0b06765208cffd04"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-05 13:05:15.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-09-05 13:05:15.000000000","tz":-180},"subject":"[DNM] Testing IPv6 job configuration on stable branches","message":"[DNM] Testing IPv6 job configuration on stable branches\n\nDepends-On: https://review.opendev.org/c/openstack/manila-tempest-plugin/+/893651\nChange-Id: I8271fbb56b02fd112b250d71264105346ee3cec7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fca2a4f1bf8630bd5e198b6856a3f18d47251a75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fca2a4f1bf8630bd5e198b6856a3f18d47251a75"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
