)]}'
{"id":"openstack%2Fmanila~894841","triplet_id":"openstack%2Fmanila~stable%2F2023.1~I9a4e5763927b7585a8fbd6b0004d6a123dcd7fae","project":"openstack/manila","branch":"stable/2023.1","topic":"bug/1957075","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-09-29 19:25:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9a4e5763927b7585a8fbd6b0004d6a123dcd7fae","subject":"[NetApp] Fixed replica promotion to updated autosize attributes","status":"MERGED","created":"2023-09-13 00:32:33.000000000","updated":"2023-09-29 19:27:19.000000000","submitted":"2023-09-29 19:25:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"894841-bug/1957075","cherry_pick_of_change":833793,"cherry_pick_of_patch_set":4,"meta_rev_id":"91690e7b2b047746c83945f6d726799cb16a5440","_number":894841,"virtual_id_number":894841,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-29 19:25:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-09-29 18:35:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-09-29 18:35:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-13 00:32:33.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":33648,"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","username":"nahimsouza"},"state":"CC"},{"updated":"2023-09-13 01:36:05.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-09-13 02:05:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-13 05:18:04.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-09-29 18:35:10.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"62c2bd58332bd0b319efc2ba3aa4472f2924a227","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-13 00:32:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0c229d737bec0ecfe2b6adc05040b123288ed190","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-13 00:33:09.000000000","message":"Topic bug/1957075-stable/2023.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"e0b8f3dd49f807003c1fbd088c64a91ff7bdf047","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-13 00:33:11.000000000","message":"Topic set to bug/1957075","accounts_in_message":[],"_revision_number":1},{"id":"95febc0ec39b364b04b6e0df5128ce86f6d07523","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-13 01:36:05.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/41/894841/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/26168b5/ : FAILURE in 1h 00m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e9e54911484ea808ebf6b466136f5b458eb96546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 02:05:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f4fac1cf32d4b2cbfde4654baddd9ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dff38589463345ee9bf1675a8f148fd9 : SUCCESS in 9m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8351b451a6aa49acbbeb25092c3757c8 : FAILURE in 19m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/921b622f94374cfbb4502af0a7810249 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d86fb93cfda740c4928095c80dbdec1e : SUCCESS in 10m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66b5a05630a54ce7af4a00df41d87512 : SUCCESS in 10m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed875186253f4c5c953fa5a423cbdc05 : SUCCESS in 5m 13s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/665c05ba7e3747bb903ee6b29b5efda7 : SUCCESS in 1h 30m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/64a396bf98484ce5b5b9e292e15dd8c1 : SUCCESS in 58m 55s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a80324b95858452586363302ad8e79ea : FAILURE in 9m 50s (non-voting)\n- manila-tempest-plugin-cephfs-nfs https://zuul.opendev.org/t/openstack/build/8292e3967de34bb3b9ce424c398f43a3 : FAILURE in 8m 42s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/266b12c76c314f39b5d773296b844919 : FAILURE in 15m 00s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/c2184aa15bab495c9ce16c6adfc8a200 : SUCCESS in 4m 06s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ab3cbc59527d470e97a4f7b20fb22aef : SUCCESS in 5m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0e623d2f741743c6b07dfefc7aa9e0cc : SUCCESS in 27m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a27f4aaf8654411085f27a1fb99c8861 : SUCCESS in 49m 16s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a5230d08b62a4ca89319fb5c90c6ed56 : FAILURE in 16m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/2438ffff709b4e609df20d06ca48580f : FAILURE in 23m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4bea9c65c2f0460bae312b71a115ce50 : FAILURE in 11m 00s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/6c8620dfb2f247f5a6df6e7dba3577e5 : SUCCESS in 52m 16s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4c6277b43e1a46ecb9e0dcc0fec52015 : SUCCESS in 49m 58s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/c341f459800f4e23a07054d8d78d5c66 : SUCCESS in 1h 06m 09s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/4441ec62b4f54c699fb05b8c68fa7736 : SUCCESS in 42m 47s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/57fc5dfdbe9d4eaf91cc2ed6d5fa0a0d : SUCCESS in 29m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a89aebbf4e46b4ae16cdf860c856eed156b0c947","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-13 05:18:04.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/41/894841/1/upstream-check/manila-tempest-plugin-ontap-dhss/2e3d6e1/ : FAILURE in 2h 25m 20s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/41/894841/1/upstream-check/manila-tempest-plugin-ontap-dhss-rest/4a207e9/ : FAILURE in 2h 31m 36s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/41/894841/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/40a70f5/ : FAILURE in 1h 48m 55s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/41/894841/1/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/034f539/ : TIMED_OUT in 3h 09m 57s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"d2658250e407bd63c110c703b748d9c63eb3e941","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-29 18:35:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d371efc1e69016a7b40b453bb14387d63d9015e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 18:35:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9f19cf6b849080325da6c340cfe9805ea4b9da3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 19:25:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e9a6e6e611e4bf690c11326dbaa4adb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e98b1e6fc90f4900a1fc0b53e0c0d2e0 : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab341052e47a4da7953b0c933c757b88 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8166436148f24bde8784cb979c0b05d3 : SUCCESS in 9m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46431424c2c94329893a738fc7c226e3 : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9bb26eb138c4797a43ce0a6715209c4 : SUCCESS in 6m 44s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c34a2971bf8c4c40b8c600b844feb55a : SUCCESS in 25m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1f0be59289b54372b69ab873bd6c5010 : SUCCESS in 41m 54s","accounts_in_message":[],"_revision_number":1},{"id":"47000425bb7697d47e3232cbc30bf192c400186d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 19:25:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"91690e7b2b047746c83945f6d726799cb16a5440","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 19:27:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35f4b15ac2c54d11b141a6a5c66178aa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d4b9eeb468c4d95bd3723c7f9a5a502 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ee7521b453b4f68a76d02edabe7df23 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b4ac0d89d160dd8e284327802ad891560be8548b","revisions":{"b4ac0d89d160dd8e284327802ad891560be8548b":{"kind":"REWORK","_number":1,"created":"2023-09-13 00:32:33.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/41/894841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/41/894841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/41/894841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/41/894841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/41/894841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/41/894841/1"}}},"commit":{"parents":[{"commit":"a4971f2cc5fa6bd81a1c245d0b06765208cffd04","subject":"Merge \"[coordination] backend_url should be secret\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a4971f2cc5fa6bd81a1c245d0b06765208cffd04"}]}],"author":{"name":"Nahim Alves de Souza","email":"nahimsouza@outlook.com","date":"2022-03-15 13:37:28.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-13 00:32:33.000000000","tz":0},"subject":"[NetApp] Fixed replica promotion to updated autosize attributes","message":"[NetApp] Fixed replica promotion to updated autosize attributes\n\nWhen a volume is created in NetApp ONTAP, it has a few autosize\nattributes that are set by default. The values of the attributes\nare defined according to the volume type (DP/RW).\n\nDuring the replica promotion, the types are swaped between source\nand destination volumes, but the autosize values were not being\nupdated. This patch fixes this behavior, calling an autosize reset\nafter promoting the replica.\n\nCloses-Bug: #1957075\nChange-Id: I9a4e5763927b7585a8fbd6b0004d6a123dcd7fae\n(cherry picked from commit 8896534bb4860d6bf4ad7d6f016b9a38e85c9840)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b4ac0d89d160dd8e284327802ad891560be8548b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b4ac0d89d160dd8e284327802ad891560be8548b"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
