)]}'
{"id":"openstack%2Fmanila~896000","triplet_id":"openstack%2Fmanila~master~I62afec521c5cdfdd67ab83da40e69e6a2688c737","project":"openstack/manila","branch":"master","topic":"bug/2004230","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-10-16 21:16:43.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-10-16 21:16:43.000000000","reason":"Change was submitted"},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2023-09-26 08:19:38.000000000","reason":"\u003cGERRIT_ACCOUNT_30407\u003e replied on the change","reason_account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}}},"hashtags":[],"change_id":"I62afec521c5cdfdd67ab83da40e69e6a2688c737","subject":"Fix policy check in metadata APIs","status":"MERGED","created":"2023-09-21 04:35:53.000000000","updated":"2023-10-17 01:53:38.000000000","submitted":"2023-10-16 21:16:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896000-bug/2004230","meta_rev_id":"d84bac4e49ddd3960e4ecb07b159b40118b547a6","_number":896000,"virtual_id_number":896000,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-16 21:16:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-10-16 18:11:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2023-10-16 18:11:46.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2023-10-16 18:12:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-21 04:37:19.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-09-21 05:09:12.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-21 06:49:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-21 07:18:39.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-09-22 02:52:49.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-09-22 05:33:03.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-09-22 20:00:45.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-09-25 15:31:22.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2023-09-25 15:31:22.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"a83072fb953bb9582adb0364912403dbbd0ca7f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-21 04:35:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58f53046ba5bfa2ce726b1431349db0109f90f31","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-21 04:37:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy pure-devstack-manila-tempest-aio-jammy : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"24693abbb8fac33cdd8bfcdd2300b5f6c76d2bc1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-21 05:09:12.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/00/896000/1/check/DellEMC_PowerFlex_Manila_dhss_false/05280fe/DellEMC_PowerFlex_Manila_dhss_false/435 : FAILURE in 31m 00s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"fab1080d9bccb54e46e9a9989030cd7835489c52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 06:49:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8303519b935e4e68afab360ba7e9c6b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12606cc00a274d689ba8c5eaaa828fda : SUCCESS in 9m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24bfc12d68ec458886bf09c666fc6fc9 : SUCCESS in 28m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b3c101c8dde439f9b890af7ea0c80a6 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9003931a379482d939455df4b1318a4 : SUCCESS in 10m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00867e0b590f4d5e934040d1ba1ca56d : SUCCESS in 12m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3dd55fbce3b44c181bf2b8728e79acc : SUCCESS in 9m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1464a519c7cd4c4081c86841cb369a43 : FAILURE in 7m 53s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a684dd1e3cdf4be58dec81ef1cfc2fa3 : SUCCESS in 7m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a2489ba1562c4810830b0a78f7136952 : SUCCESS in 39m 39s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4a0572598c56458cbb5460c7e092a90f : SUCCESS in 2h 07m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/9521a5adb06a47cfb1bb60d5b687a15d : SUCCESS in 1h 17m 03s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ea7e0e2e7f9b4ba3982f172b9d1cf668 : FAILURE in 1h 05m 59s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/58d9a0f78f434346adbad9fb425845e8 : FAILURE in 1h 29m 27s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/00dbb07c8cd148668926e4dedd6760e2 : SUCCESS in 1h 06m 47s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/7b77c396d3db4cada50431d792378d22 : SUCCESS in 3m 57s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0cdbf208fc70464aa17245fa8fa03d43 : SUCCESS in 5m 25s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a2fdd24c428a4a659063902c63df0064 : FAILURE in 22m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/eb90985a3c6f4f4990fe6e0388c1f554 : FAILURE in 40m 24s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cf4bc8e9374e423c8f55729f6c550f47 : FAILURE in 1h 03m 15s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/dc9d5d56f6fd49b9aaa48518525ddba3 : FAILURE in 11m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0de97914a56b4be8aeb9c9b084f8c0f3 : FAILURE in 23m 11s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/2c17b9787b8344e4a553353f386c20ef : SUCCESS in 36m 00s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3a30b3cdd20e4ce7861a84675beb39f8 : FAILURE in 52m 19s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8a0cdc3f527a43bbaed77339fe001b86 : FAILURE in 51m 16s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ec245d2ddbf04f8d9db2b9664ade097c : SUCCESS in 45m 11s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f29e9e1f4de24b1da45d348a406b0d23 : SUCCESS in 37m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"618811cd8c783523b9f4573afe32343373db25fa","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-09-21 07:18:39.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-00-896000-1/NFS/ : FAILURE in 2h 42m 25s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"c6c2e45b3a400a365893ce824473739dfedd93a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-22 00:33:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0271f1489e5d6d381ee85edb3b29c3a01736eaf7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-22 01:03:24.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/00/896000/2/check/DellEMC_PowerFlex_Manila_dhss_false/e2d91c4/DellEMC_PowerFlex_Manila_dhss_false/459 : SUCCESS in 29m 46s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"d228b1239fc64af3d1f1d8f9fd56e2d7fa657f67","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-22 01:45:36.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/00/896000/2/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/76f28c4/ : SUCCESS in 1h 11m 05s","accounts_in_message":[],"_revision_number":2},{"id":"144fa9335f35b79fe2f4b1b9f7afc777bc377fce","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-09-22 02:03:56.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-00-896000-2/NFS/ : SUCCESS in 1h 30m 18s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"154716885acaa3491d6a5d5e90ad2f3d3cbcc14d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 02:14:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09eedf568a8043a3aaac37a640a596ed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7719160805347a78b22643dcfb4e41c : SUCCESS in 9m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1680cc6c27247dc96e0d4855004eda3 : SUCCESS in 23m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e9169ee8a0644a2a1f5f3f8c8fcd97e : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9909c14cf8ba4c789ddd8c903e23c843 : SUCCESS in 11m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eadda30afb0f4e4ab29b2425ad3f28b6 : SUCCESS in 11m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00feee1a7fd841069e615cae9a08a5b2 : SUCCESS in 11m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c3e2ccbdc5c4f3ea85ca3865d7db615 : FAILURE in 8m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23c5c4ce79324a4f96bc6c03049053bc : SUCCESS in 6m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1aac288ca88c4114bde158ee1baa6754 : SUCCESS in 42m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/898fcadcdf0245cbabaa52f250b082b6 : SUCCESS in 1h 39m 48s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/477a002d7c464a3d869cdef337569a0b : SUCCESS in 1h 00m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/dce29950c7db4a06aa26e11adecd2846 : SUCCESS in 53m 37s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ce61db3ac7a64ee3b33a5f8c125090ba : FAILURE in 1h 31m 08s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c8e2d81451ca4d39a5b7aa1ad62e8129 : FAILURE in 1h 15m 03s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/624a3187b23946ed88a93b019007e681 : SUCCESS in 4m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/58ffdc605c2e4ad4ae3749c7a8f883ac : SUCCESS in 4m 37s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/24909ca5ed3648a49ea67b0550a0d212 : SUCCESS in 39m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c1ce4bd8c2e249f895039df3a7b1db5f : SUCCESS in 40m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/74c5223765884b2a8a88bdea8d510755 : SUCCESS in 54m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/fdc991fcda994748b284660e6ee6096e : FAILURE in 12m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/15cdc579cdbc464f81915583be06035e : FAILURE in 20m 31s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/c787fdbe6880420e959e577cd4d5abdf : SUCCESS in 55m 57s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/7f913bcc4efe488b979d4f27e41258bb : SUCCESS in 1h 00m 19s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1c6fb9a28b8c4d048ac21c973b7dd6eb : SUCCESS in 1h 02m 48s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/913604b5fdc749ffa00b8d4b8c9e45da : SUCCESS in 41m 38s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8ca9d725aa0e45e09856abef5fc65a50 : SUCCESS in 50m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0a3a1a3b136d9d0752d69480fa1bb8aeab407715","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-22 02:45:50.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/00/896000/2/check/DellEMC_PowerFlex_Manila_dhss_false/d568000/DellEMC_PowerFlex_Manila_dhss_false/460 : SUCCESS in 31m 14s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"8cabf9652319aef5cd27ae95ca0d987baa15c22f","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-22 02:52:49.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/00/896000/2/check/DellEMC_PowerStore_Manila_dhss_false/2737a0f/DellEMC_PowerStore_Manila_dhss_false/197 : SUCCESS in 38m 14s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"09060ed8049e77b8a34d029303dd053705773987","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-22 05:33:03.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/896000/2/upstream-check/manila-tempest-plugin-ontap-dhss/758da88/ : TIMED_OUT in 3h 17m 29s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/896000/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/1168875/ : FAILURE in 2h 37m 56s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"bfab8fffa722ab18702282084e2b6710f5294dba","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-09-22 20:00:45.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/00/896000/2/check/DellEMC_Unity_Manila_dhss_true/b300a1d/DellEMC_Unity_Manila_dhss_true/638 : SUCCESS in 3h 16m 42s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/00/896000/2/check/DellEMC_Unity_Manila_dhss_false/80f6204/DellEMC_Unity_Manila_dhss_false/673 : SUCCESS in 1h 00m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"ce191952233ab094716eeefe9d4de5107bd5f61b","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2023-09-26 08:19:38.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e041e1c6a9446e866757ab66549556638486dabe","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-28 12:47:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f7fd5deaa6693edbc59219f2bb6aa1a2af86169b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 12:48:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c2bdaab6b484a545a7f373550f475c8d4c5b501f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 13:40:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f756648e0cf44daa98c0a58fcafc4d5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceb218ba7243415c9040a6c570a40a1a : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adce8f8ed1da4340831190c2cad31eb8 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5adf3140abfa4245a90450dd79ea7b35 : SUCCESS in 9m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29fb2fc14dee46269a1fc8a9e1f03a70 : SUCCESS in 8m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e18df7f9ecb64f6c98fb79a126b139e3 : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d16cc4acba9447c92e9ed6a2316f51f : SUCCESS in 5m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4982504658234a6788d5289bfc088c09 : SUCCESS in 42m 26s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ed840feece4445ffa285b61752a71f3b : SUCCESS in 30m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/be8fcccc622649d09a97a3b7b74b7144 : SUCCESS in 46m 51s","accounts_in_message":[],"_revision_number":2},{"id":"398a88f67402db0bb3f92b21e4968da25a6bc0ca","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-13 23:57:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3bf524470acb6af11ecea7ea6a1ad276c22aa815","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-14 01:27:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b2dc32a4a38492fad0e16aa8ef5e788\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33c17fbc94f543aa96be774d866b1ea8 : SUCCESS in 10m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/579ad9d7ecd74f2490cc50022aa2f913 : SUCCESS in 22m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95768a9bbdbc466bb8387bdfb1475928 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3343d17073044d58904b931721cf07b7 : SUCCESS in 11m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87903f3b166a4e7997e99912c35c8536 : SUCCESS in 12m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1169d0aa1ae944fda8905e306ee9675f : SUCCESS in 13m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6af0fe156b434c158b415a46e81e9bef : SUCCESS in 8m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5a5578368823478d8461050591a12f65 : SUCCESS in 6m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/04d22e7bfdff40c48961c558dc5bcde9 : SUCCESS in 33m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4521be16df064f1299be951914a80adb : SUCCESS in 1h 28m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bdbcdcb068714f88aebde4fe065a08b1 : SUCCESS in 46m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7058e7f7ef8d451bac13fd2c0ffff453 : SUCCESS in 1h 06m 51s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8684ba6b081e4bfbb05cff91b9ffcbd3 : SUCCESS in 1h 08m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1fdf83fd2ae84c81a913992899bc20f9 : SUCCESS in 48m 50s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/866ebe9da99c48bfa337e36094dd1248 : SUCCESS in 4m 26s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c65ac3dad1c0400f9e198920994ca697 : SUCCESS in 4m 26s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/aa71ab0675ce45f7abf32f41080e394c : SUCCESS in 23m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2602662648d0409ab7949888c458073e : SUCCESS in 43m 31s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2f592b1ad1084b23bfa765bbdda8c2fa : FAILURE in 50m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/7b54322d7b27444fac13a7cc5c06b8ad : FAILURE in 9m 18s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/58362af4463b4871ad0d456e4d9ca85d : FAILURE in 22m 55s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/68c5847e57444f889db501c82cf5188b : SUCCESS in 57m 45s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/9139de403d0d442abfe5beaa0e0f9bd1 : SUCCESS in 1h 01m 51s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5428e13914b64290bc39158139fbc5a7 : SUCCESS in 1h 00m 07s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/62396f479c9c4da490131dbb729f005b : SUCCESS in 41m 28s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/33ede021d0b44769a88c5b22604b03ac : SUCCESS in 54m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c4e4245bdb017f8852e17ec44e845f25f0f8bc1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-14 01:27:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9eb58ac752b64cd344e040f9738a3926906423e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-14 02:01:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed63fee6e9794980884b6e8ddc690d48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08744886312145bfad5430c1ec090deb : SUCCESS in 8m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c537cbfec7ed4d0791b2debbf77b357e : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2051aa86aec48c49f362eecc3d621b7 : SUCCESS in 9m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90b249c349a049d0a01029b08227d623 : SUCCESS in 10m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8d7928d90c54c16a72601b2e01ea1be : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68dfbc6f80e64460809c77c7c28d70a2 : SUCCESS in 5m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/1882e804424e41cfbd24aaac8f75ebe6 : SUCCESS in 30m 49s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7082fc58ba9745beba83b2010c35ce6f : SUCCESS in 23m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c3b6979903384b53b28355a502eeeae1 : SUCCESS in 28m 29s","accounts_in_message":[],"_revision_number":2},{"id":"f3aabf846f9443c37fa4ff932c5c88d1e04a3c68","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-14 03:11:53.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/896000/2/upstream-check/manila-tempest-plugin-ontap-dhss/0f38475/ : RETRY_LIMIT in 7m 36s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/896000/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/945cc67/ : FAILURE in 1h 43m 28s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"8e7fb68895a7fb2f5e3dd266f14b25311f739106","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-16 18:11:46.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"08d126b304cada7e9cce04aaa4b101bc17666e44","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-16 18:12:14.000000000","message":"Patch Set 3: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6395f98c03f7a5f136d463f40400dcbb4c98fd68","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-16 19:00:25.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/00/896000/3/check/DellEMC_PowerFlex_Manila_dhss_false/b8ef3a3/DellEMC_PowerFlex_Manila_dhss_false/556 : SUCCESS in 31m 44s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"5bb418944c367bb11b747a72c7ad46ed1ff938f7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-16 19:16:12.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/00/896000/3/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/b08ecd3/ : FAILURE in 1h 03m 29s","accounts_in_message":[],"_revision_number":3},{"id":"48f0b334717fc3d2d01a2be36931275a710991a3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-16 19:34:27.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-00-896000-3/NFS/ : SUCCESS in 1h 12m 11s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"be871afcdcf9cbe930b7ff14a4fe7912f365f50a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 20:33:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b4d7c8e405f4c648fe647d7592b2230\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d934dc6be846c19ee3bd00252ab6d8 : SUCCESS in 10m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c09825d436c41229ed1ccd2fa5c3611 : SUCCESS in 19m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13ba14061fce4ebdac54a8f753521eaf : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2dc940e39c1546f7b430c93327815299 : SUCCESS in 9m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/094d893197a04fd191491a786093a4e5 : SUCCESS in 8m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07d873292e234e39895c7db0d14b80c9 : SUCCESS in 9m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/809a182373a647ba9cb9584f1b5d22ea : SUCCESS in 9m 55s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa1df703ce3d444b8bcd4f51318bbad8 : SUCCESS in 6m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4e5d0b926a69490bacd9816a558f80e1 : SUCCESS in 34m 56s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/28ee0ae5207348308f31e84fa0d9f5a3 : SUCCESS in 2h 15m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/aa6aa1ae2c694633a366fa756cace614 : SUCCESS in 1h 01m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3d117c755f0343d99d97e796467f866e : SUCCESS in 1h 10m 57s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/2c78ec4961734c4cb9c519f2e6743dd0 : SUCCESS in 1h 08m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f901593003e8457fa4683fc91dd72133 : SUCCESS in 1h 21m 12s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1c60866a4984465d9f0d75d1a7f6199b : SUCCESS in 5m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/08b85a4e9a8c4b3b84dbb9c2fea6c91c : SUCCESS in 5m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/13a807be2200478c905957fb38a8bf79 : SUCCESS in 43m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1143386a634e421fa544a60dad40e7fb : SUCCESS in 40m 49s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c66435b740b04b1d83b0f51960e823a9 : SUCCESS in 52m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/7b3242027a4e44ff992548ac2b8118e2 : FAILURE in 27m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1c1dc908e4814f60a27cb89351a9839f : FAILURE in 22m 18s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/0c145dc256d74214a810f1b6191c0925 : SUCCESS in 55m 39s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4c161111d468450b92563b31e70b6311 : SUCCESS in 1h 08m 25s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1b341d47d75d47bb9991a340eaf58273 : SUCCESS in 42m 39s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/a8b2197bf3164b82a53c13678284440b : SUCCESS in 27m 36s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/4a8b1ec593fc4e6a80c09809ed6a67bc : SUCCESS in 31m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e03d9c113a2c04db5f05c37d0ed9b3745b65e2fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 20:33:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f6f9fe7f80eafca838f56694575f998225beb247","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-16 21:05:58.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/00/896000/3/check/DellEMC_PowerFlex_Manila_dhss_false/7485665/DellEMC_PowerFlex_Manila_dhss_false/560 : SUCCESS in 32m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"5a615cfb9327d281d8e5fa19445b025e0f67f2c7","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-10-16 21:16:37.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/00/896000/3/check/DellEMC_PowerStore_Manila_dhss_false/9678975/DellEMC_PowerStore_Manila_dhss_false/246 : SUCCESS in 38m 15s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"0109e980290a6735cd73e38961926950b666a432","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 21:16:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81d41d914c0a4358ad3cc974c4f1627c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43f79ee330e94ccc8cfda377cc0bd5d3 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fa7961d2cb94a2cbaa39117a550da7b : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a38f3365a4c4d9a9f0c91637eb1b4f4 : SUCCESS in 12m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92181715861341f3a7e56d7cd1211866 : SUCCESS in 10m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f90abe2ae2b4a9fb88e9c77aad9e7d2 : SUCCESS in 11m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/590d01479f1142bb8b02248762406ae9 : SUCCESS in 6m 45s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f62457aaeffa4ba89074e5e3df1f4f58 : SUCCESS in 28m 59s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b7aa5bdc2c1c43c1942615e00704f4cd : SUCCESS in 37m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/628d1f0077eb4f9d8f3cc5b6ef35cbff : SUCCESS in 39m 33s","accounts_in_message":[],"_revision_number":3},{"id":"51dc3400e6709dd8daa9de03c1225a9fdc4e8984","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 21:16:43.000000000","message":"Change has been successfully merged\n\n2 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":3},{"id":"cd65ced906f22573d2258447294702258e0fdd12","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 21:17:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6441a51d40f40a8af4847c04920e823\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11e36f7983904929a8d3215b55a0f357 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60e6dbc0cb4f44c7b353aca696bfccc9 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3},{"id":"37fcdb08187581e980288b6dc8b6ede0e379aa4c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-16 23:22:40.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/896000/3/upstream-check/manila-tempest-plugin-ontap-dhss/54711c9/ : RETRY_LIMIT in 7m 36s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/00/896000/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/fccc325/ : SUCCESS in 2h 47m 57s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"d84bac4e49ddd3960e4ecb07b159b40118b547a6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-17 01:53:38.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/00/896000/3/check/DellEMC_Unity_Manila_dhss_true/47dbbf8/DellEMC_Unity_Manila_dhss_true/663 : SUCCESS in 3h 39m 18s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/00/896000/3/check/DellEMC_Unity_Manila_dhss_false/4bdbfc5/DellEMC_Unity_Manila_dhss_false/700 : SUCCESS in 57m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ad29f8a06513342fb63d665a36e16549ecf3aefc","revisions":{"b646ba8a4a354ddcc3a5099259d9fa6dfe3759f8":{"kind":"REWORK","_number":1,"created":"2023-09-21 04:35:53.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/00/896000/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/00/896000/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/00/896000/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/00/896000/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/00/896000/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/00/896000/1"}}},"commit":{"parents":[{"commit":"d12530d70eae6502dd0ecb51d6f77bd3e834f496","subject":"[rbac] Pull up policy checks on share APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d12530d70eae6502dd0ecb51d6f77bd3e834f496"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-21 04:35:09.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-21 04:35:09.000000000","tz":-420},"subject":"Fix policy check in metadata APIs","message":"Fix policy check in metadata APIs\n\nWhen a user doesn\u0027t have access to a non-public\nresource, the appropriate response is HTTP 404,\nnot HTTP 403.\n\nChange-Id: I62afec521c5cdfdd67ab83da40e69e6a2688c737\nCloses-Bug: #2004230\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b646ba8a4a354ddcc3a5099259d9fa6dfe3759f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b646ba8a4a354ddcc3a5099259d9fa6dfe3759f8"}]},"branch":"refs/heads/master"},"267a67dafe2fcf0d23aa962fa39ba9748da94feb":{"kind":"REWORK","_number":2,"created":"2023-09-22 00:33:18.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/00/896000/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/00/896000/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/00/896000/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/00/896000/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/00/896000/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/00/896000/2"}}},"commit":{"parents":[{"commit":"d12530d70eae6502dd0ecb51d6f77bd3e834f496","subject":"[rbac] Pull up policy checks on share APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d12530d70eae6502dd0ecb51d6f77bd3e834f496"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-21 04:35:09.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-22 00:31:23.000000000","tz":-420},"subject":"Fix policy check in metadata APIs","message":"Fix policy check in metadata APIs\n\nWhen a user doesn\u0027t have access to a non-public\nresource, the appropriate response is HTTP 404,\nnot HTTP 403.\n\nChange-Id: I62afec521c5cdfdd67ab83da40e69e6a2688c737\nCloses-Bug: #2004230\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/267a67dafe2fcf0d23aa962fa39ba9748da94feb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/267a67dafe2fcf0d23aa962fa39ba9748da94feb"}]},"branch":"refs/heads/master"},"ad29f8a06513342fb63d665a36e16549ecf3aefc":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-10-16 18:11:46.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/00/896000/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/00/896000/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/00/896000/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/00/896000/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/00/896000/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/00/896000/3"}}},"commit":{"parents":[{"commit":"22e568d1d374ec62ada26a03b85dcc5432fbeadf","subject":"Merge \"NetApp API failed. Reason - 18177:Relationship is in use by SnapMirror in local cluster\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/22e568d1d374ec62ada26a03b85dcc5432fbeadf"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-21 04:35:09.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-16 18:11:46.000000000","tz":0},"subject":"Fix policy check in metadata APIs","message":"Fix policy check in metadata APIs\n\nWhen a user doesn\u0027t have access to a non-public\nresource, the appropriate response is HTTP 404,\nnot HTTP 403.\n\nChange-Id: I62afec521c5cdfdd67ab83da40e69e6a2688c737\nCloses-Bug: #2004230\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ad29f8a06513342fb63d665a36e16549ecf3aefc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ad29f8a06513342fb63d665a36e16549ecf3aefc"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
