)]}'
{"id":"openstack%2Fmanila~896540","triplet_id":"openstack%2Fmanila~stable%2F2023.2~I5b1f1ce517efed000f17b1e0901e183a1913ba9f","project":"openstack/manila","branch":"stable/2023.2","topic":"bug/2004230","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-10-11 11:52:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5b1f1ce517efed000f17b1e0901e183a1913ba9f","subject":"[rbac] Pull up policy checks on share/snapshot APIs","status":"MERGED","created":"2023-09-27 15:27:04.000000000","updated":"2023-10-11 11:52:56.000000000","submitted":"2023-10-11 11:52:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896540-bug/2004230","cherry_pick_of_change":884446,"cherry_pick_of_patch_set":4,"meta_rev_id":"86115cb76b5f4ac10eb45df012d767862f97c5a0","_number":896540,"virtual_id_number":896540,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-11 11:52:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-10-11 10:59:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-10-11 10:59:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-27 15:54:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-27 19:20:25.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-10-11 10:59:23.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"ac8dd7067f53139a4ae658364e5b18cafe028c0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-27 15:27:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1b32a6491ec51f0b1c8629107b7068a25812aae8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-27 15:27:09.000000000","message":"Topic bug/2004230-stable/2023.2 removed","accounts_in_message":[],"_revision_number":1},{"id":"ddfdbc7a35cf08e71a6e79eb8bb37ceb46b2fd4f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-27 15:27:12.000000000","message":"Topic set to bug/2004230","accounts_in_message":[],"_revision_number":1},{"id":"40bee6aba199137c310ae6fd7dda93a78ee37e45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-27 15:54:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20ca38f1602e49698d1723ea74c518f4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbb4311a8ca94b60a5c2c1be67ed4b26 : SUCCESS in 9m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93d6c2ae5fa046c48d3106e2fcec95f8 : SUCCESS in 25m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a7364b2a93a4d8083cadeafecb8fade : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15a35dd44a0e48f183fa27b47d33d730 : SUCCESS in 12m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd1515d364244e36b7f32f39eb21ea90 : SUCCESS in 9m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7abf0447b0b74a93bef8371ff6083422 : SUCCESS in 13m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f8b94396da3425ea56d3a7e7d246569 : FAILURE in 8m 23s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8fc3b298f8b845ba8e1ba5823d7ba113 : SUCCESS in 5m 12s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aceb32475b874447b0b309875a8f1bcd : FAILURE in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b387d2be5c254aed325da0f69b234a224f3f4fe","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-27 19:20:25.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/896540/1/upstream-check/manila-tempest-plugin-ontap-dhss/71f63c4/ : TIMED_OUT in 3h 24m 13s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/896540/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/a09094e/ : SUCCESS in 1h 52m 23s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"7050ef3baca6dc79540ebd6e7e2309c0044825dc","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-10 18:23:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"224a3f619b9d9ab803fa7d1fb7ba0ed327d0c6a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 20:06:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80a122a1373344c2ab5839feaaf20519\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28af2618daf54d11ac5c430c99185b3d : SUCCESS in 10m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ccec42bd00e4fd290ba3c77e05eb173 : SUCCESS in 25m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa243fb2ea7b46d68a33980030d61993 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ed83195a5ed4fe6891b1612c136baf0 : SUCCESS in 11m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a1a540c676547888fb250264341f3f3 : SUCCESS in 9m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e18d5113a3324336aa814ddf0a26206d : SUCCESS in 10m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae26bc4800884154bce6b002c3122419 : SUCCESS in 8m 04s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9df90615b4b448849a88b5e02d9e0cb9 : SUCCESS in 27m 53s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ded5b9e3c1a747ce8ccddeeaf4308a9b : SUCCESS in 1h 40m 29s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fd9fd36ddb8f4a9d82edf9ec8fc98275 : SUCCESS in 1h 03m 02s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/dcfa0fc6fd2647bba1e29d297251eaa2 : SUCCESS in 57m 31s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/17a764783b2d4b7590fb0c3852254d92 : SUCCESS in 1h 05m 27s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/dbb6b91ae6f346628f74db2edc932bcc : SUCCESS in 1h 13m 13s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/105848c19fa64029bc09f4349a17a457 : SUCCESS in 4m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b1670c85b7e94f748a5d36b57f10d889 : FAILURE in 5m 29s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5dee1fc45b48471caea349d91ee9ba69 : SUCCESS in 40m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4acdf149dc7a4d4aa72c1897da88989b : SUCCESS in 45m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/21bc2fe39dad4139b53e6c31d3e594ef : SUCCESS in 56m 46s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/0107d3c8549947efa2c04f6f99a79d2b : FAILURE in 23m 57s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8d421ac701ed454ab1673be68bcaa7b3 : FAILURE in 22m 31s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/57df850e87c747f190a5845ded8cb453 : SUCCESS in 58m 32s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/bb16bdac0658485f96c0a9b186e6f661 : SUCCESS in 1h 06m 51s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/17dc9907694a4254b423da981195a9e2 : SUCCESS in 1h 03m 42s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/faba7f1336f046f09ea025778841ccc8 : SUCCESS in 43m 35s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8150a2e49b574acb9a5faa368e76cd4a : SUCCESS in 48m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d5cfca830d3af04dfc1fb6665d33cc93194aff28","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-10 23:13:48.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/896540/1/upstream-check/manila-tempest-plugin-ontap-dhss/b8f0785/ : FAILURE in 2h 52m 14s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/40/896540/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/301e164/ : SUCCESS in 3h 06m 42s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"c76c2f3d520dd2fe00b32b4b8902663bb6388fdd","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-10-11 10:59:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"590d4a8dc1a42ff53fceedefa53a8fc8f58a9e63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 10:59:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e8df8167e0ef64d40cb006987dd5024104debc87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:52:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8c3ed89da03495ca8443abcc4b853ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3bb3d2a5e6741659a3e088df2c1cac7 : SUCCESS in 10m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfb398c66f6d4f1c9e56c3b71bc2d79f : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ba284e0804f443f9d256f33f420b528 : SUCCESS in 10m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17af0afbeb9b43dc8dab2f6244339af7 : SUCCESS in 9m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c8768d16c755496b9b38c767fcd5fa8d : SUCCESS in 11m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/612e0f093f224d079732aebf28c3685e : SUCCESS in 38m 49s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9ccc837c15b644238d617be849187c84 : SUCCESS in 40m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/98685b9ca7ca49b3adcf3d893a1ba7a2 : SUCCESS in 32m 59s","accounts_in_message":[],"_revision_number":1},{"id":"86115cb76b5f4ac10eb45df012d767862f97c5a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:52:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"318140e2505a946f7523e47dc6cf5b4ac34a7813","revisions":{"318140e2505a946f7523e47dc6cf5b4ac34a7813":{"kind":"REWORK","_number":1,"created":"2023-09-27 15:27:04.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/40/896540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/40/896540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/896540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/896540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/40/896540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/40/896540/1"}}},"commit":{"parents":[{"commit":"a1451aa2055b404566d8bc8afb1dcb1f9c71993d","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a1451aa2055b404566d8bc8afb1dcb1f9c71993d"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-05-25 22:20:31.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-27 15:27:04.000000000","tz":0},"subject":"[rbac] Pull up policy checks on share/snapshot APIs","message":"[rbac] Pull up policy checks on share/snapshot APIs\n\nRBAC enforcement in manila happens in stages:\n1) Does user have access to the API\n2) Does user have access to the resource\n3) Is user permitted to perform the API action on the resource\n\nIf (1) fails, user gets a HTTP 403, if (2) fails,\nthey get a HTTP 404; if (3) fails, they get a HTTP 403.\n\nMore often than not, (2) prevents \"existence\" detection\nof resources that don\u0027t belong to the user; except in\ncase of \"public\" resources (e.g.: shares can be \"public\").\n\nIn some share API methods, policy checks for (1) are\nhappening after a bunch of processing. This leads to\nsome inconsistency.\n\nFix these occurrences to ensure a consistent user\nexperience.\n\nChange-Id: I5b1f1ce517efed000f17b1e0901e183a1913ba9f\nRelated-Bug: #2004230\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n(cherry picked from commit 190876809f2bda4153166e055c9de9bd1786f2a8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/318140e2505a946f7523e47dc6cf5b4ac34a7813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/318140e2505a946f7523e47dc6cf5b4ac34a7813"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
