)]}'
{"id":"openstack%2Fmanila~896759","triplet_id":"openstack%2Fmanila~master~Ie857caf8a439ca6b252baa3c1fbfb3f0e62d7562","project":"openstack/manila","branch":"master","topic":"bug/1996907","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-09-29 20:08:28.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"36178":{"account":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"last_update":"2023-10-03 16:31:14.000000000","reason":"\u003cGERRIT_ACCOUNT_36178\u003e replied on the change","reason_account":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-10-13 19:55:39.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2023-10-13 20:43:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie857caf8a439ca6b252baa3c1fbfb3f0e62d7562","subject":"NetApp API failed. Reason - 18177:Relationship is in use by SnapMirror in local cluster","status":"MERGED","created":"2023-09-28 10:09:52.000000000","updated":"2023-10-13 20:44:46.000000000","submitted":"2023-10-13 20:43:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"896759-bug/1996907","meta_rev_id":"29a19a643fa7887b9aaa5f8b8e58cecd999e5284","_number":896759,"virtual_id_number":896759,"owner":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-13 20:43:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-10-04 12:44:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2023-10-03 16:31:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},{"value":2,"date":"2023-10-13 19:55:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},{"value":1,"date":"2023-10-13 19:55:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":35002,"deleted":true}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-28 10:09:52.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":35002,"deleted":true},"state":"CC"},{"updated":"2023-09-28 11:25:46.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-09-28 13:17:13.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-28 13:29:31.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-09-28 14:47:04.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-09-28 15:19:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-28 15:52:43.000000000","updated_by":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"reviewer":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"state":"REVIEWER"},{"updated":"2023-09-28 15:54:23.000000000","updated_by":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"reviewer":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"state":"CC"},{"updated":"2023-09-28 17:40:20.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-09-28 21:10:51.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2023-10-04 12:44:51.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"6d64325f0efb732e88a7051abaca0d8bb2f6728a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2023-09-28 10:09:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"783712f31b1335c3049b372d0b3d69c436a91b4e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-28 11:25:46.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/59/896759/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/f42ce49/ : SUCCESS in 1h 14m 43s","accounts_in_message":[],"_revision_number":1},{"id":"6df10df7a32b47f24970d8bc62ac8563e242ff89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2023-09-28 11:40:24.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6d6677068cddbf7b55e1efe2fd2be5d16f35d9ef","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-28 11:41:55.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy pure-devstack-manila-tempest-aio-jammy : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"8b08206136bfa0e89232dbbc37b47aa0548a03e5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2023-09-28 11:49:39.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2},{"id":"79d3f7ce8902fa898abd53d508b508a7b5aab4f9","tag":"autogenerated:gerrit:restore","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2023-09-28 12:07:37.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"222f281a78ee41d5669a03e0837afdf51d31ffce","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-28 12:14:56.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/29e4bbf9f8e24154b074c98f8702271d : RETRY_LIMIT in 44s","accounts_in_message":[],"_revision_number":2},{"id":"8670dfed709f07b54aa19b9a56a004641e89d347","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2023-09-28 13:02:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fa4ac1dc4595b0adc26b33f318c995d837e756d6","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-28 13:05:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy pure-devstack-manila-tempest-aio-jammy : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"08b372a1ce212f63add57bc6354733540bc8389c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-28 13:17:13.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/59/896759/2/check/DellEMC_PowerFlex_Manila_dhss_false/c50a9d0/DellEMC_PowerFlex_Manila_dhss_false/505 : FAILURE in 21m 38s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"508c3226d80d61b5ea54cdc819811523aff2ebf5","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-28 13:29:31.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/59/896759/2/check/DellEMC_PowerStore_Manila_dhss_false/2d7edb9/DellEMC_PowerStore_Manila_dhss_false/216 : SUCCESS in 35m 56s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"304b082715f9b7aa1a4bce5ddfb6b0f51bf4e8f2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-09-28 14:47:04.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-59-896759-1/NFS/ : SUCCESS in 1h 25m 33s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"05e7e7a7590143991e66e8602f7492c5f02629da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 15:19:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb1112cec09c4be5a219b53794944732\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f285a60cfe624768ab000ab31ec3a40d : SUCCESS in 11m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7814434ff27497687e3cfd7f8623294 : FAILURE in 26m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e1a613afede431fb2a025bd3dd370f4 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b12263269c74f4da1dad83ae521a59d : SUCCESS in 10m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af0a2247ddd64d40a4a89334b2d20920 : SUCCESS in 7m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9e5ce1a9e47048248efd53244bbf2a22 : SUCCESS in 9m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2637c48b35345008d0ac2630239be1b : SUCCESS in 9m 04s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/03eff250925b430bb7a39f0029b8bf97 : SUCCESS in 44m 49s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0d3580854ea9402e87cb24679f9c3125 : FAILURE in 2h 00m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c86555a705ec47c7bb460f3469556ab0 : FAILURE in 3h 03m 34s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/bdee6ac5f7184084bfdb0bbc465c643f : POST_FAILURE in 1h 09m 09s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/cf410294cb724a1b983925b5109b2c44 : FAILURE in 1h 25m 15s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8f5d7b71ca2f47768a1b3accae1c1fb9 : SUCCESS in 1h 27m 03s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9c742d3ad80f4dbc84d8d61df749ca6f : SUCCESS in 4m 53s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fb594d0040714ab693987867c3726947 : SUCCESS in 5m 11s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/32ba7dbf47954938b200eb6994549859 : SUCCESS in 39m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4cf50cde78fe4a7496f7af606647e9ae : SUCCESS in 34m 52s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a65f3f4e228d428c939b39441e078b1a : SUCCESS in 41m 43s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/28e17fe81a294d248fb6a58050362a28 : FAILURE in 15m 43s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c04e8a7e81f14138ba1a1493c33fdc64 : FAILURE in 10m 47s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/9c0e13e21c26421c8de7e833550d4c96 : SUCCESS in 53m 05s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/2a23fa0886d746478b2e153b7b62a207 : SUCCESS in 1h 02m 15s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/87227fa7dc0b49e99ad5e13e2cf78443 : SUCCESS in 1h 03m 05s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/24624a38db0e43498c5539f63b47636f : SUCCESS in 33m 37s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/60a9b3afe1ef4e60984235eae3a71fb8 : SUCCESS in 48m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf2ced62407302ea2f50c41aa458f8ac46fb77e7","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-09-28 15:31:01.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-59-896759-2/NFS/ : SUCCESS in 1h 39m 30s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"f5c151462377f3bb959b73ba0d2000127220bd1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2023-09-28 15:41:02.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"496d6b0c1efe7d836e7641f13e9c2e827b01c9d4","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-09-28 15:52:43.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a727b7b5210336c0f351a421f31ad912365b5c5d","author":{"_account_id":33301,"name":"Caique Mello","email":"caiquemellosbo@gmail.com","username":"MelloCaique"},"date":"2023-09-28 15:54:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"605c3512fcb01774a413fd3df78a011a2ae53371","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-28 16:15:12.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/59/896759/3/check/DellEMC_PowerFlex_Manila_dhss_false/2ee326a/DellEMC_PowerFlex_Manila_dhss_false/511 : SUCCESS in 30m 41s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"80d5da065c972e44adc8b1186f4ba8a3d6372f7c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-28 16:42:57.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/59/896759/3/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/06210bf/ : SUCCESS in 59m 54s","accounts_in_message":[],"_revision_number":3},{"id":"ea2e9caa57ed75e793deb65146beb9aafc64912b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 17:27:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa3a3e5d4e3b49339003ae00b7e00684\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45538bddd3cf4850b412fe860981543c : SUCCESS in 9m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/121565bd71e94814aaaa161b88b4743f : SUCCESS in 25m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93076276318f4ba3bdefad2a7565228f : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c5905b942b24e609b6952958a373a84 : SUCCESS in 11m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e01a154b5b19433894ce5e49cd1b9400 : SUCCESS in 10m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00684d2fd5d04158bfbcee3640f3fefa : SUCCESS in 12m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/39dafb26966b45c88a7aa391120e4722 : SUCCESS in 11m 35s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d256d62d07e04a9fa922e7df8e96ce78 : SUCCESS in 36m 34s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/494068486e1c4df288140b5338b296ae : FAILURE in 1h 43m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4daab17404764765b9dab40028bd644c : SUCCESS in 1h 02m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d03189d7a1df4ed1bd519e2af23816ce : SUCCESS in 54m 03s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/59b76387775846dfb348bac9eb41a17b : SUCCESS in 1h 39m 55s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4f1b45c9d7e6409592b55897f02e541c : FAILURE in 1h 16m 08s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b94fe5e2a9344d368a6c1de80072225c : SUCCESS in 5m 26s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c850a91e5f6c493aae2076ae13d2ef6b : SUCCESS in 5m 33s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ce312ae3f2204150a7010c3636eb4d0c : SUCCESS in 32m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d910fa098d9e4f1e9659499f281b1391 : SUCCESS in 29m 56s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cd535c3a677c49ce89f6a99e137bc6f1 : FAILURE in 1h 01m 51s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/78b3028ee559458e852e704cee592522 : FAILURE in 25m 40s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/6f39e6b7b0224c708e1ae8892a0c7a69 : FAILURE in 22m 43s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/72f660263e3940278b1b0fe6daac2269 : SUCCESS in 56m 14s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d7844e6ecb224537a8bb5eebe6766479 : SUCCESS in 1h 00m 35s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/16556f7d5b7141029ecb371fb8d028b7 : SUCCESS in 1h 04m 59s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1daa3685d6b142219bd5ebfba71d6dad : SUCCESS in 44m 39s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3b2980a3ae784d8b8b0ff4b4a0dfe521 : SUCCESS in 52m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1f2929371d3ecb237b0644acb4159ae1512a1afc","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-28 17:40:20.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/3/upstream-check/manila-tempest-plugin-ontap-dhss/8b9404f/ : RETRY_LIMIT in 9m 46s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/3/upstream-check/manila-tempest-plugin-ontap-dhss-rest/7c390d7/ : RETRY_LIMIT in 9m 35s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/2fe0a11/ : RETRY_LIMIT in 11m 28s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/3/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/a092579/ : RETRY_LIMIT in 10m 19s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"5825ae3b695de1ed58f76ff5919199b70d9a33a8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-09-28 18:21:00.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-59-896759-3/NFS/ : SUCCESS in 1h 29m 28s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"647ce7d230e775b901de9643e9738fbdd3f36ce4","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-28 18:37:14.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/59/896759/3/check/DellEMC_PowerStore_Manila_dhss_false/820b126/DellEMC_PowerStore_Manila_dhss_false/222 : SUCCESS in 37m 11s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"08e6deb39cb92d256b951186bd8d50819af60c3b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-28 18:55:47.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/59/896759/3/check/DellEMC_PowerFlex_Manila_dhss_false/60edc97/DellEMC_PowerFlex_Manila_dhss_false/516 : SUCCESS in 31m 45s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"fe1a83f5ffcb6e2dad313fa6050d6f621542eff1","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-28 21:10:51.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cf86da5739aa38429888fc6c65b33b12d4a7df3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2023-09-29 08:41:01.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"eaf5f1a0e1535580a8965b6caef4f6bee9f7f0d7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-29 09:51:04.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/59/896759/4/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/42af24b/ : SUCCESS in 1h 07m 38s","accounts_in_message":[],"_revision_number":4},{"id":"a7a60ee488d14d6400946482f00e6a7f66daaa1a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-29 10:10:14.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/59/896759/4/check/DellEMC_PowerFlex_Manila_dhss_false/12caf38/DellEMC_PowerFlex_Manila_dhss_false/521 : SUCCESS in 31m 13s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"cdffdd730a5470bfed55b2f7e50fc86f3632310e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 10:28:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee9c5b34d9074b56b9c8e9cd340a623c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/327350142c8745eab520202bcc8338c4 : SUCCESS in 10m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff4a28352a174f9f987a4e95b45026c8 : FAILURE in 26m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d1376c8a72943a3ad5dda122b7292b8 : FAILURE in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49b30f6fd25642b8afbbab87018d8834 : SUCCESS in 12m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf496668e66845ab84a51a4fce523f10 : SUCCESS in 10m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca8eef20367746eaa6bcb6e9b284f9b9 : SUCCESS in 10m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/efb376b2c5e642b59b14d76e70adbac3 : SUCCESS in 7m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee68747bf5924b4eae3b5827ed013ebc : SUCCESS in 9m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d8cf4b4fedda45a19ad2bfdb96f823fd : SUCCESS in 34m 28s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e234778f17e54545a084b0ed303264fc : SUCCESS in 1h 45m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d82a5e6d8135426e871d4a9a13623b79 : SUCCESS in 1h 14m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3e2fb849ab8d4df09aebd16ad1621a83 : SUCCESS in 1h 05m 06s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/123938981de74e1ab6803585e1e6794e : SUCCESS in 1h 38m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/66d1e538ae894313b024985577a31aac : SUCCESS in 1h 01m 11s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f51d763e7eee4f0c9049f20f7effef07 : SUCCESS in 5m 25s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e1d02a43dff244efbacdf5c7d74d6ee7 : SUCCESS in 5m 14s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/54825f77309e4e84b3cfb27bc4df6686 : SUCCESS in 27m 33s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/979fb22cebe54e1e9212e1aa5cd449e8 : SUCCESS in 47m 34s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/87da7524deb04fdd8ba68ae65a1dd973 : FAILURE in 1h 00m 01s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/ed724a2372a3468b9cf9de1535c50c27 : FAILURE in 11m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c5abf00b5e65423293b258ce8169d9ce : FAILURE in 17m 52s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/ff3b8a37b13647a09c7d3c677ff02949 : SUCCESS in 52m 58s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4b7609fa43e243688d87e9a66d1ffc48 : FAILURE in 1h 14m 38s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/56fbff24d8074bd2af5859e9bae37f97 : SUCCESS in 1h 06m 12s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e39299c374b1464496055372bcd1ee6b : SUCCESS in 48m 47s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/4fbd789df5b348f5abf2c68082dc1b72 : SUCCESS in 55m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"926d593e13a67e145aac598626ff1673d701709d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-09-29 11:19:44.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-59-896759-4/NFS/ : SUCCESS in 1h 45m 54s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"d3eb957d74ad2201c406eea0c958a98ef84a0ff1","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2023-09-29 13:07:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"142a2d7bf294652ac795e72897d3cdb935ff5d86","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-29 13:10:58.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy pure-devstack-manila-tempest-aio-jammy : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":4},{"id":"51d7f1a3af062a8db21c475238175a2af3605938","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 15:33:29.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/69e4ae17215e43079c8cde2e4c5e544a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3736cb7d3fd4ca685e103961406c9f8 : SUCCESS in 10m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88e14b68e53a43a1939071a51c16426a : SUCCESS in 27m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6238b582784424e9f4f6c3815f12872 : FAILURE in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45e681bbd6cf4fddb06f41ecd46153d0 : SUCCESS in 10m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2cf673bdbcc347cd9c0868f6c0b9b121 : SUCCESS in 9m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6c2e1aabd93459993dada2dd97cf317 : SUCCESS in 10m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3ff50d113394346ad188cf1548d2966 : SUCCESS in 9m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de32df69f29c4a1bb9d78719aa42ad65 : SUCCESS in 7m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e643ce6a7d11438eb9bddaa2af8a8cef : SUCCESS in 42m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/876b1edb4bfc4c02abf1a8b078eeee52 : SUCCESS in 2h 19m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/4157bba9389a4216bd0f1b44ab248321 : SUCCESS in 1h 10m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b7f0315366254e7c959f8ee8bfd642c0 : SUCCESS in 1h 17m 53s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c7fb8f765d4b4aa99ac34a8b0724d7e0 : SUCCESS in 1h 43m 37s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ede1c144efe04bc0802d576180f4bcad : FAILURE in 57m 44s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6e235db608924def97c209d62f54156e : SUCCESS in 5m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6b00ba2fbfce41218c55b8e9c5505729 : SUCCESS in 5m 02s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/85d5d57a45cd4877b62e8f145c572769 : SUCCESS in 46m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/072f742a61494f668751977418de2df0 : SUCCESS in 30m 26s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/697052b99bc047eb9b459523abf53dcf : SUCCESS in 56m 48s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/079f3c94a0584d8bb25120d953d02521 : FAILURE in 9m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/48408fd69d604b2d8dba33443a38158d : FAILURE in 12m 17s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/a6cde20f975b43229bc3c5f3d5385191 : SUCCESS in 57m 44s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/775dd8f70e534dc9ab3ad6bd1806c830 : SUCCESS in 45m 24s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/13d8a04790b540089faa9b0b04c9caef : SUCCESS in 53m 41s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/4236f75dd9da4f5ab6115b9466e9aa8a : SUCCESS in 46m 48s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7d6848b7ed6c4d67a95031ee2caabf77 : SUCCESS in 32m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"55220b684d76a68fba3dd7e5980d84abadb43bcb","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-09-29 20:08:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"49a4f754c95bec987a8e33f25a38e0c2f95197da","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2023-09-30 06:59:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cfdcdaa406752f2feb88eb63d55c724810be1bc9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-30 08:09:37.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/59/896759/4/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/0dc7168/ : SUCCESS in 1h 07m 26s","accounts_in_message":[],"_revision_number":4},{"id":"56540539f5fc1cb583d514c359c27668fbff75a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-30 09:17:58.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/acd5012811ee4c8fb9a7f6ffc612bfd7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/344e1e6a73994af490d8d42b05c18cfa : SUCCESS in 9m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3c09b0751df4cf9bc06d4fcfe5b8fe2 : SUCCESS in 22m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afa2d1618fc443559f54ea9528a0dbb2 : FAILURE in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bcd6b6f0dda42ee8dafa33e93428e0c : SUCCESS in 10m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f15ef08cf1714deb9c95192566c2d2c0 : SUCCESS in 9m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/245b0440bf234f629fbe83a297a4e43d : SUCCESS in 10m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dd4532d5883344d38079d2e39f99eac5 : SUCCESS in 7m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42dca91d57ce44bfb4ce67006160dc57 : SUCCESS in 7m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/396f867579a74360bfe3e51f82f0bcd5 : SUCCESS in 44m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f9091133cdbd4b4c9ecbf5cb0c8c7dda : SUCCESS in 2h 12m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b6242fe5f9e0490da8a505473681be7f : SUCCESS in 53m 02s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/05cf4f11055c40528d5540565bfdd1e7 : SUCCESS in 1h 15m 41s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/1ac220a5d38b4532a10b10a0467622d3 : SUCCESS in 1h 33m 41s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/630bb46ad364490489a4fd381f282223 : FAILURE in 53m 43s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1ba7264cbd5d443bb39aded3e9784733 : SUCCESS in 5m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ffe680a9090543d4b9f197edc317fb0e : SUCCESS in 4m 26s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/85ba8fd9984c4e5ab48a130783503c0b : SUCCESS in 37m 48s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ad57b2a3c0554b5daae13ed25dc52bad : SUCCESS in 43m 45s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/14de29865a6f4d22a3496b2fc1c8722a : SUCCESS in 59m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/e1693cd3a7c3484b9319056ad0a6dc85 : FAILURE in 22m 49s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8adde67657a74b93bad71daab50e5eae : FAILURE in 10m 36s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/35139f94f57540f68f0eeae700f335fd : SUCCESS in 53m 42s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/655ac0cd395a4558bec12b91efb13893 : SUCCESS in 1h 03m 42s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e7ae9349a68a4f98827f008bf1e96743 : SUCCESS in 42m 34s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/7e7f7bcf3772478a92329832ba5cb8c4 : SUCCESS in 24m 15s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c97c278a25aa4766b788e87c7ebe9e3f : SUCCESS in 47m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"0f8cd8516d0efb5c73b6f7313bb671ea6a2ddbba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2023-09-30 13:42:43.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"c5ea6d6e16953efd95b9c9c3ffd5770ce9d7be18","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-30 14:13:12.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/59/896759/5/check/DellEMC_PowerFlex_Manila_dhss_false/d6d0d8c/DellEMC_PowerFlex_Manila_dhss_false/537 : SUCCESS in 30m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"b60fb23ece3665cef7b8b9a1d19075f7567bd81f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-30 14:47:22.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/59/896759/5/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/c6677c9/ : FAILURE in 1h 02m 14s","accounts_in_message":[],"_revision_number":5},{"id":"549be5258a6a0cd4d6d107195deccc60aadc0d62","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-09-30 15:26:36.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-59-896759-5/NFS/ : SUCCESS in 1h 43m 31s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"49f62782cc50c35029b096d3bb3e41302f0c8349","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-30 16:53:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d441995c12c146e6a232524d59d41dac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d765e93dc3243108de4927a9ce4fcde : SUCCESS in 10m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3dd785c6b97c43758e89832ee414ba04 : FAILURE in 20m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c248b30b245143ecac110cb0a6f89ec8 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9ffd3af8709479688b72c451e6c8b3a : SUCCESS in 9m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8becdf6741947ab84c24e7e2ac503bc : SUCCESS in 10m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60ff89ca32d94de2934bb116480cdda4 : SUCCESS in 9m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/09f8562b6e9a474e8bcc4d7ec51d9628 : SUCCESS in 9m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e80ce1e3427492cbc5c6a0c0f6b9057 : SUCCESS in 6m 31s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d927ca2e3e0d40c6a21fd4775547716b : SUCCESS in 42m 19s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9e98082f707e429a86eba92e4862ad29 : SUCCESS in 1h 23m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7ac0e983ccf844ba91a785ed6de9863e : TIMED_OUT in 3h 03m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/48d6258a5aba47179f8f9a28446cd104 : FAILURE in 1h 11m 42s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/3e8cd177108e4ceabd4725ad3818de15 : SUCCESS in 59m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e6f06feab3a74b949e16b9d390e8a3b6 : FAILURE in 1h 13m 28s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4e7b93647424407f8ab73b97c490aef5 : SUCCESS in 4m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bce579df02c143aea6f3475f3f6eabd1 : SUCCESS in 4m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d05c4bfa65d940fb84b52bb6a09bc49b : SUCCESS in 23m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/ce260a46ca064686bd2a97c333cffaa6 : SUCCESS in 42m 52s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/92ca09e76f0c40a388a8aa0a0f099ab0 : SUCCESS in 36m 16s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/25484457d92841bbb35edce328da58dd : FAILURE in 10m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/932b68c3d86045d7be69f754344006cd : FAILURE in 23m 26s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/7b6bd89c84cb4f0f8cf1e0b4701591b2 : SUCCESS in 58m 26s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/257b80d4596641daa09eb71f55adc5b3 : SUCCESS in 38m 42s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/166b05af462b40188561138480f396ea : SUCCESS in 38m 02s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/061e19c9abd74efe9af9ddd58e402576 : SUCCESS in 41m 33s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/dacabb936e084cd1ba08699ad1427265 : SUCCESS in 28m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0f9e75b97c1faf99c6fb21a23b6d975d3ba42a07","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-30 17:26:03.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/59/896759/5/check/DellEMC_PowerFlex_Manila_dhss_false/f541bce/DellEMC_PowerFlex_Manila_dhss_false/538 : SUCCESS in 32m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"8a3c9161b4004b79aac0353e44d5e8033e0487a3","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-30 17:33:40.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/59/896759/5/check/DellEMC_PowerStore_Manila_dhss_false/194170c/DellEMC_PowerStore_Manila_dhss_false/229 : SUCCESS in 39m 55s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"1328e90e06df57afb8d7a5a54f0244c00b44bee3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-30 20:14:23.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/5/upstream-check/manila-tempest-plugin-ontap-dhss/b0d70a5/ : POST_FAILURE in 2h 07m 02s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/5/upstream-check/manila-tempest-plugin-ontap-dhss-rest/bfc2d58/ : TIMED_OUT in 3h 19m 08s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/5/upstream-check/manila-tempest-plugin-ontap-no-dhss/308595f/ : SUCCESS in 2h 45m 00s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/5/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/ce30f00/ : TIMED_OUT in 3h 10m 45s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"e649d678615f5840cfb4044ded8f09d58c9867ec","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2023-10-01 11:08:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8e218c9df3f87b496e4891f10c0fc47e9ecdb87f","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2023-10-01 12:49:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0fd4772a81489f4b9bd34c18e2a99b8a41eb0786","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-01 14:22:48.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/5/upstream-check/manila-tempest-plugin-ontap-dhss/8ed4e42/ : FAILURE in 1h 33m 16s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/5/upstream-check/manila-tempest-plugin-ontap-dhss-rest/9544fde/ : TIMED_OUT in 3h 12m 59s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/5/upstream-check/manila-tempest-plugin-ontap-no-dhss/844f9b9/ : TIMED_OUT in 3h 10m 03s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/5/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/5f9900f/ : TIMED_OUT in 3h 08m 33s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"0cc6e41cabeb1384904045e0af13ed45f20bdc63","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2023-10-02 07:12:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e1e9bfc2636fec10f917cb54faad7241385443e0","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2023-10-02 07:45:50.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f9dbc1f23d270cd9bee4269e6816e82c8863fe0d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-02 10:34:08.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/5/upstream-check/manila-tempest-plugin-ontap-dhss/c3c6571/ : RETRY_LIMIT in 10m 00s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/5/upstream-check/manila-tempest-plugin-ontap-dhss-rest/c2bdeed/ : TIMED_OUT in 3h 20m 10s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/5/upstream-check/manila-tempest-plugin-ontap-no-dhss/06f4a46/ : RETRY_LIMIT in 11m 39s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/5/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/eb2271b/ : RETRY_LIMIT in 11m 40s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"637306677166bb33d90918210fa5037bc6c0de81","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2023-10-02 12:01:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"541ef16621d6880bbbfed8aa0c94eb8df96d1ba5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-02 15:19:29.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/5/upstream-check/manila-tempest-plugin-ontap-dhss/a6076e3/ : FAILURE in 2h 38m 04s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/5/upstream-check/manila-tempest-plugin-ontap-dhss-rest/93aa8dd/ : TIMED_OUT in 3h 15m 50s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/5/upstream-check/manila-tempest-plugin-ontap-no-dhss/859c58d/ : SUCCESS in 2h 58m 51s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/5/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/7b75a0c/ : TIMED_OUT in 3h 08m 33s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"b9b7f87a834336b8972f2d515634b1573a7ca420","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2023-10-03 03:20:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"487130b89fe16138ae051ad55941e6750b1022b9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-03 06:41:10.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/5/upstream-check/manila-tempest-plugin-ontap-dhss/21444ff/ : TIMED_OUT in 3h 18m 59s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/5/upstream-check/manila-tempest-plugin-ontap-dhss-rest/3d40011/ : TIMED_OUT in 3h 14m 42s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/5/upstream-check/manila-tempest-plugin-ontap-no-dhss/68acc5a/ : SUCCESS in 1h 53m 18s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/59/896759/5/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/a5d9bc0/ : TIMED_OUT in 3h 08m 43s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"594cee421a71751c077d1cbd995bb80def2dbffb","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2023-10-03 09:13:49.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"15b2660c4e97e594439bd902975983483bdbd2f7","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2023-10-03 16:31:14.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bb651c922e4dad9ffe83f9e17b675eef3acd9593","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-10-04 12:44:51.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"64d24b1143957cf2c6959605ae5afb367365dfa0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-11 00:02:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2204dd32dfaa95e66fe9c865ea0eb1f50d02f7a2","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2023-10-12 11:01:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fe1c7d3fbdeb15bd4ae6080c53140318d5ccc642","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-13 19:55:39.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"dc542bcdab54a72eecb3fbfc82d87cad39f16175","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 19:56:15.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"cb7867acd6f2599268d0b574df2d8dc430a730d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 20:43:37.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/262b2a5e7ead4927bfa561630b1e259f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1da8c9831caf45589d60cb83ff6377c6 : SUCCESS in 10m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc5e0239b16e4dedbabb34c2f1abd41a : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f9d5e11021645f09e8192f1b9218c6e : SUCCESS in 10m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdbedaff37534dfcad2fe77b9779d8c9 : SUCCESS in 9m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/daa126ccdb1943c4971408328a1900a2 : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61a213ff636d4255b2d8b79d46de7c9f : SUCCESS in 6m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/939471f501b44ddda0f98f3f0677f525 : SUCCESS in 39m 42s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/09de6b9ed8a840b394b2346a999f8cfb : SUCCESS in 22m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2d872da85059413992ca514f2f16f516 : SUCCESS in 42m 31s","accounts_in_message":[],"_revision_number":5},{"id":"0100c1faf4100eb5e21d2eb9bf5f35569e8647cc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 20:43:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"29a19a643fa7887b9aaa5f8b8e58cecd999e5284","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 20:44:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a39c7dead9e7498ba116d271922d04f8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1656425655d4cb58e548228e4ace8f8 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76105f75362d43fd811dd556e9e9d230 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"1f9d51fdba24c218940b5eeacd3175b6b0767635","revisions":{"71cb4b877a735877f4bd9243696d9ade8320411c":{"kind":"REWORK","_number":1,"created":"2023-09-28 10:09:52.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/59/896759/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/59/896759/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/59/896759/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/59/896759/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/59/896759/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/59/896759/1"}}},"commit":{"parents":[{"commit":"a45ae5fcbd3d711342ca8fdda0b2f36579740bd3","subject":"Merge \"Use pre-commit for \u0027pep8\u0027 tox target\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a45ae5fcbd3d711342ca8fdda0b2f36579740bd3"}]}],"author":{"name":"LuizSantos","email":"luizfelipe.looze@gmail.com","date":"2023-06-05 14:46:00.000000000","tz":0},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2023-09-28 10:09:46.000000000","tz":330},"subject":"NetApp API failed. Reason - 18177:Relationship is in use by SnapMirror in local cluster","message":"NetApp API failed. Reason - 18177:Relationship is in use by SnapMirror in local cluster\n\nChanged the client method \u0027get_snapmirror_destinations\u0027 instead of \u0027get_snapmirrors\u0027\n\nCloses-Bug: #1996907\nChange-Id: Ie857caf8a439ca6b252baa3c1fbfb3f0e62d7562\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/71cb4b877a735877f4bd9243696d9ade8320411c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/71cb4b877a735877f4bd9243696d9ade8320411c"}]},"branch":"refs/heads/master"},"2efdfe054c156374e6ef51327ba20b7ef574d52a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-09-28 11:40:24.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/59/896759/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/59/896759/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/59/896759/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/59/896759/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/59/896759/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/59/896759/2"}}},"commit":{"parents":[{"commit":"a45ae5fcbd3d711342ca8fdda0b2f36579740bd3","subject":"Merge \"Use pre-commit for \u0027pep8\u0027 tox target\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a45ae5fcbd3d711342ca8fdda0b2f36579740bd3"}]}],"author":{"name":"LuizSantos","email":"luizfelipe.looze@gmail.com","date":"2023-06-05 14:46:00.000000000","tz":0},"committer":{"name":"Gireesh Awasthi","email":"gawasthi2010@gmail.com","date":"2023-09-28 11:40:24.000000000","tz":0},"subject":"NetApp API failed. Reason - 18177:Relationship is in use by SnapMirror in local cluster","message":"NetApp API failed. Reason - 18177:Relationship is in use by SnapMirror in local cluster\n\nChanged the client method \u0027get_snapmirror_destinations\u0027 instead of \u0027get_snapmirrors\u0027\n\nCloses-Bug: #1996907\nChange-Id: Ie857caf8a439ca6b252baa3c1fbfb3f0e62d7562\n(cherry picked from commit 63b0a52683a3a109cedfc6241eb0d9e0f2cedf4c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2efdfe054c156374e6ef51327ba20b7ef574d52a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2efdfe054c156374e6ef51327ba20b7ef574d52a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"41201991daab74e370fdca67354218530a460b00":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-09-28 15:41:02.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/59/896759/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/59/896759/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/59/896759/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/59/896759/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/59/896759/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/59/896759/3"}}},"commit":{"parents":[{"commit":"a45ae5fcbd3d711342ca8fdda0b2f36579740bd3","subject":"Merge \"Use pre-commit for \u0027pep8\u0027 tox target\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a45ae5fcbd3d711342ca8fdda0b2f36579740bd3"}]}],"author":{"name":"LuizSantos","email":"luizfelipe.looze@gmail.com","date":"2023-06-05 14:46:00.000000000","tz":0},"committer":{"name":"Gireesh Awasthi","email":"gawasthi2010@gmail.com","date":"2023-09-28 15:41:02.000000000","tz":0},"subject":"NetApp API failed. Reason - 18177:Relationship is in use by SnapMirror in local cluster","message":"NetApp API failed. Reason - 18177:Relationship is in use by SnapMirror in local cluster\n\nChanged the client method \u0027get_snapmirror_destinations\u0027 instead of \u0027get_snapmirrors\u0027\n\nCloses-Bug: #1996907\nChange-Id: Ie857caf8a439ca6b252baa3c1fbfb3f0e62d7562\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/41201991daab74e370fdca67354218530a460b00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/41201991daab74e370fdca67354218530a460b00"}]},"branch":"refs/heads/master","description":"Edit commit message"},"210c1c0fda5f081aeddeafacd80b3c1385b2f965":{"kind":"REWORK","_number":4,"created":"2023-09-29 08:41:01.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/59/896759/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/59/896759/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/59/896759/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/59/896759/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/59/896759/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/59/896759/4"}}},"commit":{"parents":[{"commit":"a45ae5fcbd3d711342ca8fdda0b2f36579740bd3","subject":"Merge \"Use pre-commit for \u0027pep8\u0027 tox target\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a45ae5fcbd3d711342ca8fdda0b2f36579740bd3"}]}],"author":{"name":"LuizSantos","email":"luizfelipe.looze@gmail.com","date":"2023-06-05 14:46:00.000000000","tz":0},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2023-09-29 08:40:57.000000000","tz":330},"subject":"NetApp API failed. Reason - 18177:Relationship is in use by SnapMirror in local cluster","message":"NetApp API failed. Reason - 18177:Relationship is in use by SnapMirror in local cluster\n\nChanged the client method \u0027get_snapmirror_destinations\u0027 instead of \u0027get_snapmirrors\u0027\n\nCloses-Bug: #1996907\nChange-Id: Ie857caf8a439ca6b252baa3c1fbfb3f0e62d7562\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/210c1c0fda5f081aeddeafacd80b3c1385b2f965"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/210c1c0fda5f081aeddeafacd80b3c1385b2f965"}]},"branch":"refs/heads/master"},"1f9d51fdba24c218940b5eeacd3175b6b0767635":{"kind":"REWORK","_number":5,"created":"2023-09-30 13:42:43.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/59/896759/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/59/896759/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/59/896759/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/59/896759/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/59/896759/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/59/896759/5"}}},"commit":{"parents":[{"commit":"a45ae5fcbd3d711342ca8fdda0b2f36579740bd3","subject":"Merge \"Use pre-commit for \u0027pep8\u0027 tox target\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a45ae5fcbd3d711342ca8fdda0b2f36579740bd3"}]}],"author":{"name":"LuizSantos","email":"luizfelipe.looze@gmail.com","date":"2023-06-05 14:46:00.000000000","tz":0},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2023-09-30 13:42:39.000000000","tz":330},"subject":"NetApp API failed. Reason - 18177:Relationship is in use by SnapMirror in local cluster","message":"NetApp API failed. Reason - 18177:Relationship is in use by SnapMirror in local cluster\n\nChanged the client method \u0027get_snapmirror_destinations\u0027 instead of \u0027get_snapmirrors\u0027\n\nCloses-Bug: #1996907\nChange-Id: Ie857caf8a439ca6b252baa3c1fbfb3f0e62d7562\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1f9d51fdba24c218940b5eeacd3175b6b0767635"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1f9d51fdba24c218940b5eeacd3175b6b0767635"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
