)]}'
{"id":"openstack%2Fmanila~896764","triplet_id":"openstack%2Fmanila~stable%2F2023.2~I62afec521c5cdfdd67ab83da40e69e6a2688c737","project":"openstack/manila","branch":"stable/2023.2","topic":"bug/2004230","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-10-11 10:59:43.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-10-11 11:52:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I62afec521c5cdfdd67ab83da40e69e6a2688c737","subject":"Fix policy check in metadata APIs","status":"MERGED","created":"2023-09-28 15:55:08.000000000","updated":"2024-01-09 20:17:48.000000000","submitted":"2023-10-11 11:52:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"896764-bug/2004230","cherry_pick_of_change":896000,"cherry_pick_of_patch_set":2,"meta_rev_id":"3a95c9df8101290511a3cd82ebcdaf1084558cc7","_number":896764,"virtual_id_number":896764,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-11 11:52:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-10-11 10:59:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-10-11 10:59:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-28 16:18:21.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-09-28 18:22:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-28 18:39:37.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-10-05 19:44:53.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"CC"},{"updated":"2023-10-11 10:59:43.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"b0049c53d59b1ccaa9735b69441a04423742b948","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-28 15:55:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5cd88a9e070bf3a5458c0bd88baf64bac91b8a8d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-28 15:57:39.000000000","message":"Topic bug/2004230-stable/2023.2 removed","accounts_in_message":[],"_revision_number":1},{"id":"91e9eb96224d751025799e91bba80df4b26596d6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-09-28 15:57:42.000000000","message":"Topic set to bug/2004230","accounts_in_message":[],"_revision_number":1},{"id":"8874c51e4f0c8850641db33a53ce38b715d9d70f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-28 16:18:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/dc82a3ea7aba4bb592f19860560dfc08 : RETRY_LIMIT in 44s","accounts_in_message":[],"_revision_number":1},{"id":"7c82406316cc890d23bc1e28e5262441e5600e0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 18:22:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf9a3b316333418aa6d7c146dadd1f83\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55ab98c14b3847e0bf631f62cddba95d : SUCCESS in 8m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60fb019e0f144dcbbf10c564b64a4fb7 : SUCCESS in 23m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65295a68f27c46d9a34526f0dceaa529 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86f710d1afa1426e83dd04046c47d15d : SUCCESS in 9m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/359933550a78461e9123d2eaef51b8f8 : SUCCESS in 11m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f09aacb419524a149d823813b5400b85 : SUCCESS in 11m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a8863ecd633a4de9becc5300cf994b45 : FAILURE in 10m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/550699f903934082906f61b14a717479 : SUCCESS in 6m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5dd9f7b9f9fa4039a6fd2adf21b26d91 : SUCCESS in 43m 38s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2cf47580eba54f0fb8b0ac7ddec50465 : SUCCESS in 2h 21m 04s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d3e8602d784c4b309a874784fc158f7c : SUCCESS in 1h 05m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3eeab05fb3c44db2ac10a7319d2b815b : SUCCESS in 51m 08s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7716979562bc4a4a94cfc73fb16febe6 : FAILURE in 1h 10m 53s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cf1072cb4cfe490495446d78a5384e2c : SUCCESS in 1h 04m 17s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f5a698eca77f4df797e1bbc5193274bf : SUCCESS in 5m 26s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/1ad47850e2f746fcaf43da04e6f85b4f : SUCCESS in 5m 02s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/db81052bf32041e0923450af3619e9a6 : SUCCESS in 28m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/afba11f7307d4dd0946b147b2f2ccf02 : SUCCESS in 30m 04s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7273e677aac844838b97728a987fcf84 : SUCCESS in 1h 01m 59s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/ed7d249ce4fa4012b9985a23fef39297 : FAILURE in 11m 54s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d5957683b24d40b89fe9478042a42428 : FAILURE in 24m 30s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/e1a9a91ba37a47d099afbeb7fba50a7e : SUCCESS in 55m 19s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/71b2d532b9544e69890885325da803d9 : SUCCESS in 47m 46s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a1ebc3ed91b344f7a9408d9e9bf70f25 : SUCCESS in 1h 07m 48s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0f7cab4bbcfa4c7db012ac45fe7d6698 : SUCCESS in 41m 21s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/09fb977693354a399c4bee04030a3fbb : SUCCESS in 50m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"faf126d23070644573ee8ffde78d847ee3f9d5d2","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-28 18:39:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/64/896764/1/upstream-check/manila-tempest-plugin-ontap-dhss/27f4227/ : RETRY_LIMIT in 8m 45s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/64/896764/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/f252810/ : RETRY_LIMIT in 9m 34s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"7c7e3af3ffad87731f71eb4ec7eb0954f377dc1e","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-10-05 19:44:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b955ea148c74d40666a6398ba202eb4681eaff7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-05 22:45:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6f1ad2975e584c04760aae821f49e3bee23b10af","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-05 22:45:56.000000000","message":"Topic bug/2004230-stable/2023.2 removed","accounts_in_message":[],"_revision_number":2},{"id":"2ea7abae9b936f5c7c995db437346f07f4b1ab65","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-05 22:46:02.000000000","message":"Topic set to bug/2004230","accounts_in_message":[],"_revision_number":2},{"id":"4967b378e99dd11acb9df0e2ba4816376feb84fd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-05 22:46:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"710181d20bcecd9549229cf60d29dc7597bd9cf2","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-05 22:47:38.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy pure-devstack-manila-tempest-aio-jammy : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"d29697821577f06d3325188142416cde7d01a660","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 00:44:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8170d7f16de45f19ec734863a8be016\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49c89af9c39148e8a72ef918235458b3 : SUCCESS in 11m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1a097855d7a4ad2aa77ae699006f90b : FAILURE in 25m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a05e459a7a034ec180a2e03927dc31c6 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56f19a31f5bc4c3f80b43571787e3271 : SUCCESS in 10m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e3d9d2433a34b69bd13a44805f3ae32 : SUCCESS in 10m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b47ea8d74d394966aa8a8cfaf8a48cfa : SUCCESS in 12m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/14c1864705e94282a77cfc76542adecf : FAILURE in 8m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fabd6b8c208f41e387d5c20108baea4c : SUCCESS in 7m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9e3516cc6f6249df9275aacf92632842 : SUCCESS in 38m 43s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a274834fa2514b88baf3773605e8ac12 : SUCCESS in 1h 54m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/efea3f22d64c47bbb259bc476bb23188 : SUCCESS in 1h 34m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ea40209c97ae48159d4e559a43587d7f : SUCCESS in 1h 05m 32s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8bf8da4f89e544df846e40816ad71727 : SUCCESS in 1h 30m 00s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/39acb85c4da0472d8f8e9c724653be93 : SUCCESS in 1h 14m 12s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1d875026e3b84838b95a5247ac8c2c1f : SUCCESS in 4m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5a470bcdf66e4fc5b2cb5b9312a89183 : SUCCESS in 4m 59s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e8d6a6fda5f6427b9ae8379914615f87 : SUCCESS in 31m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/121a541d88e84aae8bb9c93a53a1fa75 : SUCCESS in 38m 10s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b268754c90764d59977b482c0e99391c : SUCCESS in 1h 02m 24s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/ad49fc129e76449db4d3f929afcbcd97 : FAILURE in 21m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/fd234eb674a6499c9ebe1ebeb011645e : FAILURE in 23m 44s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/1ebdb7f0598c4c4bb2228bd27cb5700f : SUCCESS in 45m 16s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3b0c849def714dcaa82e0dbdefd1aad1 : SUCCESS in 43m 44s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ef97ce8da5474718ad07d800e076f350 : SUCCESS in 1h 00m 14s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b73167ae75fb40789287b425be033165 : SUCCESS in 41m 16s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/0e718f7eb421412daa06170572d445fd : SUCCESS in 47m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2627044133bde318b6fcdfd36840e0e1f0473029","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-06 03:12:25.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/64/896764/2/upstream-check/manila-tempest-plugin-ontap-dhss/44897e7/ : FAILURE in 2h 26m 39s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/64/896764/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/a2bcca9/ : RETRY_LIMIT in 8m 40s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"9384c3abaf7d47a3857e2ab542162c8e2ce51e10","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-10-11 10:59:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cbd15a7d71294d3dfb801f1093e42fbd40feb4c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:00:21.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2108eed5e89290477a46bfe60f63d332170dcf14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:52:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e8ce8da79d14b6fac7f4eea54165d84\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9e6f2a8b3154ead928520dd35bc133f : SUCCESS in 11m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c88cb10390d24451989c08538bfb7d85 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fe42a4e12ed425880237706c9ce4735 : SUCCESS in 10m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a8fea9570934cb394d329fb685c1915 : SUCCESS in 10m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27520bcd16d84af79d77bb2baf3d106e : SUCCESS in 12m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0709e9b0b97e4f178ca815c3119e29fe : SUCCESS in 7m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a93f68e21aff4fe6b46aec854873ce17 : SUCCESS in 41m 20s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2e0cf0ae39824e6ba8793c8d58fc4e80 : SUCCESS in 35m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0da9bb8a3b83428d9940565acf37e6ed : SUCCESS in 35m 51s","accounts_in_message":[],"_revision_number":2},{"id":"cb637d02c464fc0c9ebfe447ff7cb043fa069ca9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:52:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"245e047c945b585e715ee992fd5b91f035c8163d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-16 18:13:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a95c9df8101290511a3cd82ebcdaf1084558cc7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-01-09 20:17:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1ee34c740bfe8f76d4c586beebc78d4b26459185","revisions":{"6decce5cdb22968ece41bd74d6d6a9bd615c3d56":{"kind":"REWORK","_number":1,"created":"2023-09-28 15:55:08.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/64/896764/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/64/896764/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/64/896764/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/64/896764/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/64/896764/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/64/896764/1"}}},"commit":{"parents":[{"commit":"a1451aa2055b404566d8bc8afb1dcb1f9c71993d","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a1451aa2055b404566d8bc8afb1dcb1f9c71993d"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-21 04:35:09.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-28 15:55:08.000000000","tz":0},"subject":"Fix policy check in metadata APIs","message":"Fix policy check in metadata APIs\n\nWhen a user doesn\u0027t have access to a non-public\nresource, the appropriate response is HTTP 404,\nnot HTTP 403.\n\nChange-Id: I62afec521c5cdfdd67ab83da40e69e6a2688c737\nCloses-Bug: #2004230\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6decce5cdb22968ece41bd74d6d6a9bd615c3d56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6decce5cdb22968ece41bd74d6d6a9bd615c3d56"}]},"branch":"refs/heads/stable/2023.2"},"1ee34c740bfe8f76d4c586beebc78d4b26459185":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2023-10-05 22:45:52.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/64/896764/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/64/896764/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/64/896764/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/64/896764/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/64/896764/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/64/896764/2"}}},"commit":{"parents":[{"commit":"f94e16a97887ef1267524d8a9f8527dac2f36fa7","subject":"Merge \"Fix share network delete procedure\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f94e16a97887ef1267524d8a9f8527dac2f36fa7"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-21 04:35:09.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-05 22:45:52.000000000","tz":0},"subject":"Fix policy check in metadata APIs","message":"Fix policy check in metadata APIs\n\nWhen a user doesn\u0027t have access to a non-public\nresource, the appropriate response is HTTP 404,\nnot HTTP 403.\n\nChange-Id: I62afec521c5cdfdd67ab83da40e69e6a2688c737\nCloses-Bug: #2004230\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n(cherry picked from commit 267a67dafe2fcf0d23aa962fa39ba9748da94feb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1ee34c740bfe8f76d4c586beebc78d4b26459185"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1ee34c740bfe8f76d4c586beebc78d4b26459185"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
