)]}'
{"id":"openstack%2Fmanila~896906","triplet_id":"openstack%2Fmanila~master~I2dd898daeccb2fe2ac6485a90585f314ede0cd9c","project":"openstack/manila","branch":"master","topic":"trivial","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-10-02 21:42:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2dd898daeccb2fe2ac6485a90585f314ede0cd9c","subject":"tests: Add \u0027reason\u0027 argument to exceptions","status":"MERGED","created":"2023-09-29 07:33:31.000000000","updated":"2023-10-03 00:00:32.000000000","submitted":"2023-10-02 21:42:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896906-trivial","meta_rev_id":"3f45041e7e7443a990dac0dabb94b4ce38d8f63e","_number":896906,"virtual_id_number":896906,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-02 21:42:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2023-10-02 17:07:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2023-10-02 17:07:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-29 08:03:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-29 08:05:18.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-29 08:36:09.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-09-29 08:39:39.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-09-29 08:41:35.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-09-29 11:22:06.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-10-02 17:07:40.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"42b3959e4006872e194dd33be38d2b7b27409b7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-29 07:33:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a2b0feff519b16e1056aaa3a6b6be622eaca65d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 08:03:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f585c615b9ae41f7ad46a4262b4b55e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2b20e3fc4b64f8f81ad4b92996e7856 : SUCCESS in 10m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4380c72015644bfb93c8e5027fdedea0 : SUCCESS in 24m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1d076fe7a76453b86cd22982b824b6c : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/784940a00eab44059911c030837ebf57 : SUCCESS in 9m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a15937f4b2248e1b841068d900a583b : SUCCESS in 18m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf7cc55fe3d74a658d83964b498f2587 : SUCCESS in 7m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/144118e377544c33a0e2313de2ac493a : SUCCESS in 8m 16s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6e4832edec8d4cc993a84b4e12abe82c : SUCCESS in 5m 22s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fcf53b12ed8c4257b2fc5207bf59b41e : SUCCESS in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"642384c0fbe07b189c385ad0c988dc43c2a5ac9e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-29 08:05:18.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/06/896906/1/check/DellEMC_PowerFlex_Manila_dhss_false/60bef1a/DellEMC_PowerFlex_Manila_dhss_false/517 : SUCCESS in 31m 31s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"bda93e342b10df23ed2be04c9aa4d7d8a40602c7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-29 08:36:09.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/06/896906/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/ee4b006/ : SUCCESS in 59m 59s","accounts_in_message":[],"_revision_number":1},{"id":"72a2002ea894760ce195b95f55dde4eb95b52cac","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-29 08:39:39.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/06/896906/1/check/DellEMC_PowerStore_Manila_dhss_false/3522683/DellEMC_PowerStore_Manila_dhss_false/223 : SUCCESS in 36m 00s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"8de2e7f7284e2420b6feccb376723b8a630b16d8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-09-29 08:41:35.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-06-896906-1/NFS/ : SUCCESS in 1h 7m 46s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"c40706cec1c2d4bafce7a2cb4e679720919068aa","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-29 11:22:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/06/896906/1/upstream-check/manila-tempest-plugin-ontap-dhss/f992747/ : TIMED_OUT in 3h 17m 16s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/06/896906/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/cedb60d/ : RETRY_LIMIT in 10m 11s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"d7b266f4153ffbb6dacab2cdf984b32e97456cf5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-02 17:07:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"70cc1c6b523d346c46ffe09c0062107ea2ffd67c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 17:08:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"87e81c44c6d999402afe2b5397950a79fbf7ba87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 17:49:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5fdbb6bbda747d8a8494fe34532d6d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d4b01b920da46898f81313d8c72c9fd : SUCCESS in 12m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f670b9bd59645bf8534fc38f94e6cea : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f21f9f14ef034dccb54412823c43d6cd : FAILURE in 21m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f363072759b403c9f9ee68856c90410 : FAILURE in 22m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b930b435209c41c1b1dc155ae06f8004 : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":1},{"id":"9b87f3aab6f9572579cbad7892fe656e94b31a30","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-02 20:58:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af174386739c1f7915856a37e5d1c7fd9953f25d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 21:24:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/147ab607e3a64dd280a633ce424226f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c17751199d9e406abfc8b811d92ca0e2 : SUCCESS in 10m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3200fc320aee49838453aa3b27cd92e5 : SUCCESS in 25m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec00d9e226b04d2aa420f65f1b90ba14 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff82feaea22b4da3ba06b2af41ada616 : SUCCESS in 15m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b31834c049f4b80809563be92aeb4ef : SUCCESS in 14m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ccf25f6b78dc4352aa619b52cf2e3d50 : SUCCESS in 15m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e10a93b313d140d2939003ac965903c8 : SUCCESS in 8m 34s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/38d11bc5f2874c2983f285ff9e4f0996 : SUCCESS in 5m 25s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5ef1d323c5304aaca1f53483797188b3 : SUCCESS in 5m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"766d43b16e4619f64fbf0f802ad8a26263f5684f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 21:25:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0f94a6ee8a83868308e869ffe8e3df30475e61bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 21:42:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe2a6cfc0bba498392590c0e8969bc08\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/895f5169fb3a4fa593dbd626e175cbab : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e367d973dd84db39f01ca91c346554c : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/576a33d53d28436ab8553bd437e67600 : SUCCESS in 9m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05d554c549d2439c8255f984b755ceca : SUCCESS in 9m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30cb909120794e4385ce54ee50147cf7 : SUCCESS in 9m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a460793eed3fa0631d7667d70788e789bc0fe3e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 21:42:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"23e0432f5fe6d4a12a831a7a8ee03bb35bdff83e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 21:43:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d5e4f03ee554159aa814577740e41a6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba224f39a3524cbfb55c6f668952feea : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1},{"id":"6f28499769f6d584fcf5550852ffb8770ffb92c6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-02 21:56:19.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/06/896906/1/check/DellEMC_PowerFlex_Manila_dhss_false/955a5c1/DellEMC_PowerFlex_Manila_dhss_false/541 : SUCCESS in 31m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"793d58d52fd99668d3f983e8c761fe91388be81a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-10-02 22:03:31.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/06/896906/1/check/DellEMC_PowerStore_Manila_dhss_false/516cc24/DellEMC_PowerStore_Manila_dhss_false/230 : SUCCESS in 38m 24s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"3f45041e7e7443a990dac0dabb94b4ce38d8f63e","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-03 00:00:32.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/06/896906/1/upstream-check/manila-tempest-plugin-ontap-dhss/858298b/ : SUCCESS in 2h 16m 49s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/06/896906/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/5dc7010/ : SUCCESS in 2h 28m 40s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f500c5877adb7df6d4f2a7b852405e38b735334e","revisions":{"f500c5877adb7df6d4f2a7b852405e38b735334e":{"kind":"REWORK","_number":1,"created":"2023-09-29 07:33:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/06/896906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/06/896906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/896906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/896906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/06/896906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/06/896906/1"}}},"commit":{"parents":[{"commit":"398722be59ab94418a86983cb350811cca662d2b","subject":"Merge \"[RBAC] Enforce check for share updates\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/398722be59ab94418a86983cb350811cca662d2b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-29 07:32:30.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-29 07:32:30.000000000","tz":60},"subject":"tests: Add \u0027reason\u0027 argument to exceptions","message":"tests: Add \u0027reason\u0027 argument to exceptions\n\nThis resolves some warnings that were getting spewed out at the\nbeginning of a test run.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I2dd898daeccb2fe2ac6485a90585f314ede0cd9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f500c5877adb7df6d4f2a7b852405e38b735334e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f500c5877adb7df6d4f2a7b852405e38b735334e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
