)]}'
{"id":"openstack%2Fmanila~896932","triplet_id":"openstack%2Fmanila~master~Ia60e8f87a69efe2fc35e00707b6e391cf17422b6","project":"openstack/manila","branch":"master","topic":"sqlalchemy-20","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-29 12:17:57.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ia60e8f87a69efe2fc35e00707b6e391cf17422b6","subject":"db: Add tests for \"share instance\" APIs","status":"NEW","created":"2023-09-29 11:19:38.000000000","updated":"2023-09-29 21:04:14.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"09a9bf307abaadb7edf6f8ca4e1ba7c1e67d57da","_number":896932,"virtual_id_number":896932,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-09-29 16:34:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-29 11:54:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-29 12:17:57.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-09-29 13:26:08.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-09-29 15:55:09.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-09-29 17:00:46.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-09-29 17:50:51.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"803cdafa14584fab82485565bc6adaa790d295ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-29 11:19:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d089d6c1e558dc8f5dc6827e83238a32d8b8dec4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 11:54:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6c9b2be5eae485490b0ddcd16c0ff5d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70201b7be2544cbc912cfd642b8861b4 : SUCCESS in 10m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19b9ed420ba941369f3043e1025951e9 : SUCCESS in 25m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8649b91baac54e5289385d1ab3248a66 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16ba47ab1f8f4940b056e11597119c3e : SUCCESS in 9m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/488b0001229d4d799dab48cfe5de61ec : SUCCESS in 11m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9f28d419cee4eee802967025f9cf0f3 : SUCCESS in 11m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/58e0c83f00514548877d172dd3f8e0df : SUCCESS in 8m 53s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/380768940f524e9a892c7638d593ebdc : SUCCESS in 5m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2ca5fdccf8e6417d8d3617d42bae504c : SUCCESS in 6m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a03f06dbc5550b5a8b55329fb24362575d2e114f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-29 12:17:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/97c3799bff834f75a125d678277d67ff : RETRY_LIMIT in 1m 27s","accounts_in_message":[],"_revision_number":1},{"id":"4ad638b07fe61463760f7ce54ade03e2531106b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-29 12:22:23.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"8526f8c06a508061f09829a08d0cb060ac228b93","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-29 12:27:40.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy pure-devstack-manila-tempest-aio-jammy : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"7900380ab08e2bb3ca0475bb981e4d602869be65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 12:46:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bbe3ab921b154b71b346d1c956dcaf61\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2270127cdbf043fca0fe8cc548366aea : SUCCESS in 7m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/debbc40584524e4bbde92b8b57594d75 : SUCCESS in 20m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce98c3619b564b918786b3bf2b6eccfc : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a1004e8580048c7b462bcf099df2676 : SUCCESS in 10m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6a74fe244cf45f1b12e041ad53fcb55 : SUCCESS in 11m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6b6dda061234313881101bbed37a08a : SUCCESS in 10m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a5c9129196340f98bcfbe80f4720c94 : SUCCESS in 9m 26s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5a17d6b74e414804bacfd63e8a5261e1 : SUCCESS in 5m 34s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/93ab2afc059c445daab569e5bb268994 : SUCCESS in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"235c2473005330d709206eb32704943297e2c411","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-29 13:26:08.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/32/896932/2/check/DellEMC_PowerStore_Manila_dhss_false/558d4c5/DellEMC_PowerStore_Manila_dhss_false/225 : SUCCESS in 38m 42s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"89abf2b6fdea948c03a964184393b925e041dec9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-29 15:55:09.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/896932/2/upstream-check/manila-tempest-plugin-ontap-dhss/2dbf6a8/ : FAILURE in 2h 47m 33s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/896932/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/ce2628f/ : FAILURE in 3h 05m 34s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"c32a67589e9c4c9b5e852ac205bcfe0bb5bd7224","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-29 15:55:56.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"569698dfeaf875c2a317534801a463391912cd68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-29 16:34:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9782f6a36ebf44fd806c1647733e4c91\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a2905cfb32d4a54a6dc525a7e5a83c6 : SUCCESS in 9m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9594d097c504872ac3008cf42c07475 : FAILURE in 36m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f594a80f879e45e184238f69b48a7786 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a4639e514f9417a9d75c4f0d29161cb : SUCCESS in 11m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bce2e6cd14a542ae8e166c4450a7defc : SUCCESS in 9m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d52d4111f1949b188fc97d6dc877523 : SUCCESS in 16m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c6e3333a9a344589c68bf518e274755 : SUCCESS in 11m 02s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b16c01d9aaa04d79abd9258bf9ab1f3b : SUCCESS in 6m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4362b60922a746cc97c2f488cd37e1bf : SUCCESS in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e621c80f82dabedf93c89a436610c02cea71e5bd","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-09-29 17:00:46.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-896932-1/NFS/ : SUCCESS in 1h 40m 44s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"bbfa0fb4b2c0d9f0d16cf3883baa968f376852b3","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-09-29 17:04:42.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/32/896932/3/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/5149c4c/ : SUCCESS in 59m 13s","accounts_in_message":[],"_revision_number":3},{"id":"0ecef6e7205776978e19d379ff4eb0f5f030fa3b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-09-29 17:14:39.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/32/896932/3/check/DellEMC_PowerStore_Manila_dhss_false/3b19591/DellEMC_PowerStore_Manila_dhss_false/227 : SUCCESS in 39m 50s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"f0dd468a8bfe62d87a7c6c5be97a00354cb755a8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-09-29 17:50:51.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/32/896932/3/check/DellEMC_PowerFlex_Manila_dhss_false/42275c1/DellEMC_PowerFlex_Manila_dhss_false/533 : SUCCESS in 35m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"3e7bc6395485181e85668666853d62df66d0272c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-09-29 18:35:09.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-896932-2/NFS/ : SUCCESS in 1h 15m 7s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"1da17b31fa8ad4da6a9b3adf2b32575d85a3ec5b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-09-29 19:50:31.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/896932/3/upstream-check/manila-tempest-plugin-ontap-dhss/f4c05a1/ : TIMED_OUT in 3h 13m 58s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/32/896932/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/f996f4f/ : FAILURE in 2h 41m 34s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"09a9bf307abaadb7edf6f8ca4e1ba7c1e67d57da","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-09-29 21:04:14.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-32-896932-3/NFS/ : SUCCESS in 1h 44m 11s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8768a4b452b2d57c0353af77ba220de87f7856b2","revisions":{"082a261fa93a2bec76cf1f38745afdf48c572159":{"kind":"REWORK","_number":1,"created":"2023-09-29 11:19:38.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/32/896932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/32/896932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/32/896932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/32/896932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/32/896932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/32/896932/1"}}},"commit":{"parents":[{"commit":"378d82e45d9a05340863f403ed417f6b0cf898a8","subject":"db: Remove \u0027get_session\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/378d82e45d9a05340863f403ed417f6b0cf898a8"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-29 08:16:33.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-29 11:06:28.000000000","tz":60},"subject":"db: Add tests for \"share instance\" APIs","message":"db: Add tests for \"share instance\" APIs\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ia60e8f87a69efe2fc35e00707b6e391cf17422b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/082a261fa93a2bec76cf1f38745afdf48c572159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/082a261fa93a2bec76cf1f38745afdf48c572159"}]},"branch":"refs/heads/master"},"f0c011af15f25eb7bc336e47fd4c86d4a06b880b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-09-29 12:22:23.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/32/896932/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/32/896932/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/32/896932/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/32/896932/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/32/896932/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/32/896932/2"}}},"commit":{"parents":[{"commit":"95c74a7141bcf75dec97f8345b3c8668caca435d","subject":"db: Migrate \"share instance\", \"share replica\" APIs to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/95c74a7141bcf75dec97f8345b3c8668caca435d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-29 08:16:33.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-29 12:22:13.000000000","tz":60},"subject":"db: Add tests for \"share instance\" APIs","message":"db: Add tests for \"share instance\" APIs\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ia60e8f87a69efe2fc35e00707b6e391cf17422b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f0c011af15f25eb7bc336e47fd4c86d4a06b880b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f0c011af15f25eb7bc336e47fd4c86d4a06b880b"}]},"branch":"refs/heads/master"},"8768a4b452b2d57c0353af77ba220de87f7856b2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-09-29 15:55:56.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/32/896932/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/32/896932/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/32/896932/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/32/896932/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/32/896932/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/32/896932/3"}}},"commit":{"parents":[{"commit":"75071729837606156bca80014caafaf8661d6eed","subject":"db: Migrate \"share instance\", \"share replica\" APIs to enginefacade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/75071729837606156bca80014caafaf8661d6eed"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-29 08:16:33.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-09-29 15:55:39.000000000","tz":60},"subject":"db: Add tests for \"share instance\" APIs","message":"db: Add tests for \"share instance\" APIs\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ia60e8f87a69efe2fc35e00707b6e391cf17422b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8768a4b452b2d57c0353af77ba220de87f7856b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8768a4b452b2d57c0353af77ba220de87f7856b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
