)]}'
{"id":"openstack%2Fmanila~897964","triplet_id":"openstack%2Fmanila~stable%2Fyoga~I58dcd9716cf95d0d696c13a4c831df787726bcda","project":"openstack/manila","branch":"stable/yoga","topic":"bug/2024658","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-10-13 14:02:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I58dcd9716cf95d0d696c13a4c831df787726bcda","subject":"Fix duplicate entries in share_server_backend_details","status":"MERGED","created":"2023-10-11 14:46:48.000000000","updated":"2023-10-13 14:04:05.000000000","submitted":"2023-10-13 14:02:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897964-bug/2024658","cherry_pick_of_change":888118,"cherry_pick_of_patch_set":2,"meta_rev_id":"3d0bd95a0b001fd958759a6bd163638200a5a491","_number":897964,"virtual_id_number":897964,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-13 14:02:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-10-13 13:11:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2023-10-12 09:54:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-10-13 13:11:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-11 14:46:48.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"CC"},{"updated":"2023-10-11 14:48:58.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-10-11 16:58:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-12 09:54:18.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2023-10-13 13:11:12.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"01f2820a5f6f90d31d52260325b03f700210542c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-11 14:46:48.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"2dc364aa3f2a9a3c4494dd55c9c08ce6f85544e9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-11 14:46:53.000000000","message":"Topic bug/2024658-stable/yoga removed","accounts_in_message":[],"_revision_number":1},{"id":"8d949a9cb9ffbffe5a814d80c74903f39812e806","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-11 14:46:56.000000000","message":"Topic set to bug/2024658","accounts_in_message":[],"_revision_number":1},{"id":"12c1299aac7e3dd55ec392c05ddd02f7fbac2991","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-11 14:48:58.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio pure-devstack-manila-tempest-aio : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"a71c16b245aade4162e03e3d4cef937d98e1ae9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 16:58:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ff6a9b0300e483c91a91ed8cd43e9aa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0607c2e3adfe49e7be76a82c9b280be3 : SUCCESS in 11m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f5b8b277dac44cabbc6d81a9fffeb53 : SUCCESS in 20m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb3990afeea94f2bbd6fb605b2d30157 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fe733d25cf44a0cbccdb5931f202fde : SUCCESS in 8m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6479cb70ee24beab525578aa5fd2be7 : SUCCESS in 9m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0171c1728a3e48549130fdbc50c3b449 : SUCCESS in 8m 32s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/3c5c8692ed944c3d805192f09af1010d : SUCCESS in 47m 51s\n- manila-tempest-plugin-generic-scenario-stable https://zuul.opendev.org/t/openstack/build/9eab7a86bd3f4c32b82ef9418b87649f : FAILURE in 1h 52m 00s (non-voting)\n- manila-tempest-plugin-generic-stable https://zuul.opendev.org/t/openstack/build/c35d71e2379f4d84b923f254f8bc2c3c : SUCCESS in 1h 42m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native-stable https://zuul.opendev.org/t/openstack/build/7f2c57e8e5064574b48d728f0083504c : SUCCESS in 2h 02m 42s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-stable https://zuul.opendev.org/t/openstack/build/53a4fbb9f6c8424786da84e5c7fa4d06 : FAILURE in 1h 47m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux-stable https://zuul.opendev.org/t/openstack/build/d77b0cc6803c48aebb3306b37b8717d8 : SUCCESS in 1h 13m 02s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/db9fd341f0cc44e3b4f2c736f6c5e06b : SUCCESS in 5m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/660048eb0c874151b5e2fc9f1b41a780 : SUCCESS in 8m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/454dbf15fa424cdf934189473bb6ff71 : SUCCESS in 32m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f1344a1e348349d594ce44188476d8ff : SUCCESS in 37m 03s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5ada64b174e84082b96065add25fbfff : FAILURE in 59m 15s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/2904201db9cc40968e8f4cd0b84d18a1 : SUCCESS in 33m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c38ba2a7db144a4c98e61e9a8e39b3b2 : SUCCESS in 21m 39s (non-voting)\n- manila-tempest-plugin-lvm-fips-stable https://zuul.opendev.org/t/openstack/build/3b713b387bc74de28a0ea6a5f90934bb : SUCCESS in 46m 55s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/9625bbe8b50746a59259d7a30d151ea3 : SUCCESS in 55m 08s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/34e089edb7f54a4da7e4311e79214ef3 : SUCCESS in 35m 53s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ab9d7442d43347c9a9f3b6f71f4bdecf : SUCCESS in 36m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5a88377f8779c07b9471aa1366739f3d5f11a559","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-10-12 09:54:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3530216705558321dc9fe909b73df668f49c496a","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-10-13 13:11:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01001c8e91635c5a4b231139622da94d32c9d1cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 13:11:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c259553158198f8d1ad800288de34752143d96f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 14:02:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4674e469c34544b9add5c1108f670d94\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9679bed3afa24bebadcd35280acccc41 : SUCCESS in 9m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f15201d74a3b40db96a19bf462fcd187 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27d574ef715041d1974d68417fdcfb4d : SUCCESS in 8m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8bc4c5dc57d74c12afd79d2a8dde90a0 : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4ae9b2e4372445184adc03c4e2cd7ed : SUCCESS in 8m 14s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/93a6e5fe6f904649a7348c4e55741ca6 : SUCCESS in 46m 40s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/83c61f088d0744fb9557fb5dcc4373e0 : SUCCESS in 39m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/34e0cfe3363e467a94a8df4f27414788 : SUCCESS in 43m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b8891f72e2c72bf09479ef2fb80381c949562455","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 14:02:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3d0bd95a0b001fd958759a6bd163638200a5a491","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 14:04:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa53a0ba3a2744328c38880830db0130\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b4e82177b46412f90517b6c9d46b29f : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6308491029dd4d33a486cc7452399f2f : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dd0492623c2f25b551fa9b29635adb18da8582c1","revisions":{"dd0492623c2f25b551fa9b29635adb18da8582c1":{"kind":"REWORK","_number":1,"created":"2023-10-11 14:46:48.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/64/897964/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/64/897964/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/64/897964/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/64/897964/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/64/897964/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/64/897964/1"}}},"commit":{"parents":[{"commit":"f776fe77adffd41194120092b053b360ea1d3d46","subject":"Merge \"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f776fe77adffd41194120092b053b360ea1d3d46"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-06-22 12:36:31.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-11 14:46:48.000000000","tz":0},"subject":"Fix duplicate entries in share_server_backend_details","message":"Fix duplicate entries in share_server_backend_details\n\nshare_server_backend_details_set() add entries in db table without\nchecking existing entries with given combinaton of share_server_id\nand key. This causes duplicate records. Fix it by validating presence\nof share server id and key.\n\nCloses-bug: #2024658\nChange-Id: I58dcd9716cf95d0d696c13a4c831df787726bcda\n(cherry picked from commit 37278df338ab4d2753ca004e556a197f847be0ce)\n(cherry picked from commit 5fddd2db19c138bdb4057a58f1e631e97ba9c8f2)\n(cherry picked from commit ecd14b4e3cf4028c15345accf583638eda94e1c7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dd0492623c2f25b551fa9b29635adb18da8582c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dd0492623c2f25b551fa9b29635adb18da8582c1"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
