)]}'
{"id":"openstack%2Fmanila~897965","triplet_id":"openstack%2Fmanila~stable%2Fxena~I58dcd9716cf95d0d696c13a4c831df787726bcda","project":"openstack/manila","branch":"stable/xena","topic":"bug/2024658","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-10-13 18:56:42.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-10-13 19:40:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I58dcd9716cf95d0d696c13a4c831df787726bcda","subject":"Fix duplicate entries in share_server_backend_details","status":"MERGED","created":"2023-10-11 14:47:16.000000000","updated":"2023-10-13 19:41:40.000000000","submitted":"2023-10-13 19:40:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897965-bug/2024658","cherry_pick_of_change":897964,"cherry_pick_of_patch_set":1,"meta_rev_id":"7364c38f57a3ccc523a25b0d1ea15f53ec586818","_number":897965,"virtual_id_number":897965,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-13 19:40:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-10-13 18:56:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2023-10-12 09:54:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-10-13 18:56:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}],"CC":[{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-11 14:47:16.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"CC"},{"updated":"2023-10-11 16:42:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-11 16:42:22.000000000","updated_by":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-12 09:54:30.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2023-10-13 17:33:39.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"14f632211e2495059c19ced24f2c8b6b4af51c5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-11 14:47:16.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"f6be4bc92aecf31906e24f7966de0e6a9cdfe4f8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-11 14:47:22.000000000","message":"Topic bug/2024658-stable/xena removed","accounts_in_message":[],"_revision_number":1},{"id":"cc71c523cf555bb99da4659d33da98ed1223118f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-11 14:47:25.000000000","message":"Topic set to bug/2024658","accounts_in_message":[],"_revision_number":1},{"id":"e977ec2acb5a5f794103f4b1a20ab309b9fb3e38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 16:42:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a784e15745964652b1f5df96dcef4f6c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73615d5dd7864377925223cb3b2e75fc : SUCCESS in 10m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abe13136336240a7837338f67ad4bd89 : SUCCESS in 19m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aab24f11d13948309feffb8f140eee03 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81fc4bfceac24d81a9a0e4b45e5abc2e : SUCCESS in 8m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c87e2007ac8487aaa8a299050895c30 : SUCCESS in 8m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfed771ebacf4078a027fdf627e8bbf4 : SUCCESS in 9m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/296fad9b0aec49938e0bc144413f7fde : SUCCESS in 9m 50s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/7d8a0f87cf204e9c8932dc3aaf9ef384 : SUCCESS in 44m 22s\n- manila-tempest-plugin-generic-scenario-stable https://zuul.opendev.org/t/openstack/build/2daa4ca6cab14c92b294940bff1c27c3 : FAILURE in 1h 52m 33s (non-voting)\n- manila-tempest-plugin-generic-stable https://zuul.opendev.org/t/openstack/build/d3e49a4add994e4888ecb50837526193 : SUCCESS in 1h 20m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native-stable https://zuul.opendev.org/t/openstack/build/829e12e6581a4c58a84a61d7ee2bc3ec : SUCCESS in 1h 29m 05s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-stable https://zuul.opendev.org/t/openstack/build/4bbc1d369e774822aca112532601804c : SUCCESS in 1h 11m 19s (non-voting)\n- manila-tempest-plugin-zfsonlinux-stable https://zuul.opendev.org/t/openstack/build/1d37755c4565448eb2b977f4fb21a8fb : SUCCESS in 1h 20m 21s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9c44105794934e38be32e664a2d497d1 : SUCCESS in 4m 56s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/27ccad8d5e7c43c4861ba8665c51a8eb : SUCCESS in 6m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8c8de6a3952c44029f3203659ad53da1 : SUCCESS in 34m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/058d87c6b88f4874844f652393ad4b65 : SUCCESS in 37m 24s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5ffb240a2fd84306be7b4f54e5ffcd75 : SUCCESS in 37m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/eadadf6878104f01810905cf129151c4 : SUCCESS in 39m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/465ccf52373a47d2bcf4cfd494db6d85 : SUCCESS in 31m 19s (non-voting)\n- manila-tempest-plugin-lvm-fips-stable https://zuul.opendev.org/t/openstack/build/4017f08378cc4779b4070d718cb61553 : SUCCESS in 42m 46s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/33d72333a0c94f22861212d626ebd406 : FAILURE in 26m 14s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/3fe612d3961648da96e93c1e5e6d7462 : SUCCESS in 28m 58s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/4f3555cea03045a0b3a7189fb38adca7 : SUCCESS in 40m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f41d3a661038ecd40ecae5bf8ad658af47a62a82","author":{"_account_id":12016,"name":"DellEMC PowerMAX CI","email":"emc.vmax.ci@emc.com","username":"emc-vmax-ci","tags":["SERVICE_USER"]},"date":"2023-10-11 16:42:22.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerMAX_Manila_xena https://elab-os-logsrv.delllabs.net/65/897965/1/check/DellEMC_PowerMAX_Manila_xena/974f41c/DellEMC_PowerMAX_Manila_xena/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerMAX CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerMAX_CI","accounts_in_message":[],"_revision_number":1},{"id":"b73a37fe837ef0fbadb369f90acd36040dc06a56","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2023-10-12 09:54:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3e19140865cb8b3220946cdaec4b420b9cf73d8f","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-10-13 18:56:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a8a12d852296c9cf69417fafa7edf0c385b0772","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 18:57:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ada8221d1efac2b8d505908a29ca87593859a12a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 19:40:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a85f8d0951f2499d937c9e405c6b46a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a3384068f81409a85feff44a600b2bc : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd9fc78ea19949cdbd3217e1b6eecab4 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fd5751d5de34fadb14d1b4cb143ee47 : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb7f7409ec2449c9ac93c0deb7af4511 : SUCCESS in 9m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4847ab7b06da4552b8d60540b909fe79 : SUCCESS in 6m 17s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/46159fe621514cefa64c4e330ddad161 : SUCCESS in 37m 36s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bab75e970ba44845879f08e968ce2028 : SUCCESS in 38m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d6f56cf5b36a47f2a113f1b7ee3a16d8 : SUCCESS in 31m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ff5b032e446633c9e0c5bdfbf89ddbdc499dcff5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 19:40:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7364c38f57a3ccc523a25b0d1ea15f53ec586818","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 19:41:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37d20d77f2914bb28644be92a61fcdcc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ab1ee3cbced46098ce8e8673d8f3a11 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/552a287a79184addb48bab9e31fca5d9 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a759ed6936f1c75e73b1f0f703b4c11551aca835","revisions":{"a759ed6936f1c75e73b1f0f703b4c11551aca835":{"kind":"REWORK","_number":1,"created":"2023-10-11 14:47:16.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/65/897965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/65/897965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/65/897965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/65/897965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/65/897965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/65/897965/1"}}},"commit":{"parents":[{"commit":"64de01b89df5ecd56876c38dfdac350192c06b26","subject":"Merge \"NetApp ONTAP: Fix SnapMirror snapshots not being cleaned up\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/64de01b89df5ecd56876c38dfdac350192c06b26"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2023-06-22 12:36:31.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-11 14:47:16.000000000","tz":0},"subject":"Fix duplicate entries in share_server_backend_details","message":"Fix duplicate entries in share_server_backend_details\n\nshare_server_backend_details_set() add entries in db table without\nchecking existing entries with given combinaton of share_server_id\nand key. This causes duplicate records. Fix it by validating presence\nof share server id and key.\n\nCloses-bug: #2024658\nChange-Id: I58dcd9716cf95d0d696c13a4c831df787726bcda\n(cherry picked from commit 37278df338ab4d2753ca004e556a197f847be0ce)\n(cherry picked from commit 5fddd2db19c138bdb4057a58f1e631e97ba9c8f2)\n(cherry picked from commit ecd14b4e3cf4028c15345accf583638eda94e1c7)\n(cherry picked from commit dd0492623c2f25b551fa9b29635adb18da8582c1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a759ed6936f1c75e73b1f0f703b4c11551aca835"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a759ed6936f1c75e73b1f0f703b4c11551aca835"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
