)]}'
{"id":"openstack%2Fmanila~897967","triplet_id":"openstack%2Fmanila~stable%2F2023.1~I62afec521c5cdfdd67ab83da40e69e6a2688c737","project":"openstack/manila","branch":"stable/2023.1","topic":"bug/2004230","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-11-16 23:10:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I62afec521c5cdfdd67ab83da40e69e6a2688c737","subject":"Fix policy check in metadata APIs","status":"MERGED","created":"2023-10-11 14:48:25.000000000","updated":"2023-11-16 23:11:13.000000000","submitted":"2023-11-16 23:10:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897967-bug/2004230","cherry_pick_of_change":896764,"cherry_pick_of_patch_set":2,"meta_rev_id":"7cc596c4533149619aec1a1f721d7f114cb09fc0","_number":897967,"virtual_id_number":897967,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-16 23:10:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-11-16 21:47:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-11-16 21:47:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-11 15:48:30.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-10-11 16:51:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-11 20:11:45.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-11-16 21:47:22.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"d2946222f4911ffdea481b0b33eaec63eae7ddd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-11 14:48:25.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"b56accbc234da803975af774ef2c95628642fe22","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-11 14:48:30.000000000","message":"Topic bug/2004230-stable/2023.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"013ce3ee385c6b111ec1f82d8fae6e09c034a83a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-11 14:48:33.000000000","message":"Topic set to bug/2004230","accounts_in_message":[],"_revision_number":1},{"id":"474797cd6699cbd4c07823094fcdf927f314d01e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-11 15:48:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/67/897967/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/385316e/ : FAILURE in 59m 01s","accounts_in_message":[],"_revision_number":1},{"id":"bed83c35c9c991305614af7e3cd6903768bf60a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 16:51:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccc9dfc6744b41c3beacbf7ba2898467\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06a54450b6574b9c90b667e17ea045e4 : SUCCESS in 9m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a195f4709e5427fbda904a6db310520 : SUCCESS in 20m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af4405e18bcc4652aaa5bf4f1300bbed : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b44b6ea52f2b4b649fac9589e508dfbc : SUCCESS in 9m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8d0792025974905803326bb90422465 : SUCCESS in 10m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03200fb907c8495191cf08d5f03e44f5 : SUCCESS in 7m 59s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f9dcdb31884b4caca3de188f6eed819d : SUCCESS in 1h 50m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5f00522372c64f04832976b30512637b : SUCCESS in 1h 18m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6c3ce26af04643389361d0bb07696b9e : FAILURE in 24m 47s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a39356c8b252483da6aa31905ef9d869 : FAILURE in 26m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e93854fc04d04eba826ca0ff83c37998 : SUCCESS in 1h 07m 47s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ea8d88daa5894d3889c3c7de07b9d2d5 : SUCCESS in 5m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/92bb67b3ec45440d9669d421e5add24f : SUCCESS in 4m 36s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/fd4ec4e5b6ac4378bc92b94828ef164e : SUCCESS in 31m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/398a063b2ba14a909909311566724b27 : SUCCESS in 42m 35s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a856c54881424a6aa31c625c87327b33 : SUCCESS in 45m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/12f9a28efcb6446685b61a28e39896dd : FAILURE in 10m 37s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d7c53de749ee46edae47582b16151706 : FAILURE in 21m 09s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/cc665ae3ffe34328a8e2432ab37246e8 : SUCCESS in 49m 54s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/bd8b741244234572a5eb9600208c5d11 : SUCCESS in 1h 02m 32s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/ca31cb5e55604a669308f3a1e83679d2 : SUCCESS in 53m 14s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6d88931ba729449392858ce4e441104e : SUCCESS in 38m 47s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/bf989d4d7d4940b8b16838ba95bfa9e3 : SUCCESS in 42m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34cf3387b6d287e53bc913e8c87a6d7275ee2603","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-11 20:11:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/67/897967/1/upstream-check/manila-tempest-plugin-ontap-dhss/b07f80c/ : TIMED_OUT in 3h 18m 03s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/67/897967/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/17d5e0d/ : FAILURE in 2h 41m 40s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"c2d5b0c0bc09070b8651ce75a2c52317ba8ccc81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-16 18:13:59.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"286c6bf810c886df9e59e63e5cb8a2d67fd8bfb5","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-16 18:15:25.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy pure-devstack-manila-tempest-aio-jammy : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"4e0843c50f3cee482d00fe569cca07313e09e434","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 19:53:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56cfae9781e547958f0ef48d032992c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37b456c178654460b0830bbef39e8e81 : SUCCESS in 9m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/283669a003b44ac682ca36878ea54a70 : SUCCESS in 18m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77c2f509f2574f97b186be497ef09972 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efb6e89e771b4744b05de23fd64b64b7 : SUCCESS in 8m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27e078d639334d2fb3114a9bc95cc4b2 : SUCCESS in 9m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbb629e28c79485084cda508b2d04310 : SUCCESS in 7m 26s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f9433d609159491084a1d0e898cdb7fe : SUCCESS in 1h 36m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d13421283d544178ad1e3577f0a8769e : SUCCESS in 1h 34m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/5570c956fdee43be82166ce8653f85db : FAILURE in 28m 48s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/54e920cf0c2c44d9beea8e1ea7c3fb51 : FAILURE in 25m 55s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5ab5f08d89a846f3b40ca9f7e583f9f9 : SUCCESS in 1h 01m 03s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6cd5e647a2ad498b9fb61d16b3d8cbb9 : SUCCESS in 5m 22s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8e9d64fdc6874425ba0d4159650e8faa : SUCCESS in 4m 49s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0d919102dee24c3b9a321e090a16da3f : SUCCESS in 40m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/447eb94dcdfe4f4cbc74683219f1ef0d : SUCCESS in 28m 54s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/05138acf023148a8a03e0cd15455aec9 : SUCCESS in 44m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/ee1f7857cdbe46d89917a6fd2a0ba028 : FAILURE in 24m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ba994f44d7c44ef4863318c6bb578dae : FAILURE in 21m 54s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/353725fdf1da443a99c790d1b4a3ea43 : SUCCESS in 55m 30s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/5bd6005733404a3387ff3bbe1570b463 : SUCCESS in 47m 12s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/0e3af47646ee4600a430736e53b9bc48 : SUCCESS in 52m 39s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6b76ccc896b5487080ce85a4cc8b2d36 : SUCCESS in 47m 48s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/6f2155e534814b5e868c0c14df715592 : SUCCESS in 45m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"72a77fa0f417c9603444c60d7166eecfd3756ab6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-16 23:07:51.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/67/897967/2/upstream-check/manila-tempest-plugin-ontap-dhss/6f41b6f/ : TIMED_OUT in 3h 12m 28s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/67/897967/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/f33f821/ : SUCCESS in 2h 41m 41s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"19160c7960d2074c907d9d9a83be327073cf26fb","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-11-16 21:47:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c4342a2d7d0495f64627e375326efb3513067b0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 21:47:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cbd581765303efbae5b394c7eb6df2bea4964471","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 23:10:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ebe92b9925c41248718974af0239469\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90c13a91de784c1c8a58e02887e392d3 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb181d1a902c4fbd94f98bc9551b744e : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36b33293ea224fc4a6dec716238a3e25 : SUCCESS in 9m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fe91a7d61044ca78c5059f4f54bc37d : SUCCESS in 6m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/26c459bb3bdb4cbd9e248b94a1ae9488 : SUCCESS in 1h 15m 23s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/90a14809b391462686a7359c740be9d1 : SUCCESS in 22m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/95d173327c18487b8b2c6b2e6c5ddfb3 : SUCCESS in 28m 13s","accounts_in_message":[],"_revision_number":2},{"id":"07fbaf9e667be9e001df343f94962f8db4330d27","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 23:10:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7cc596c4533149619aec1a1f721d7f114cb09fc0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 23:11:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/840ff4a21e994a87943b140cc4566306\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9ec2cf504a2457d90a8eac5642e722c : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ddd0ecebe144f5380ef52ac82270e51 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d5c2eb1124a9f44b3ad71c6cfdb01f7fa80ee88d","revisions":{"355f1ec9381c00f694ebe5228321c07ed3f2a97a":{"kind":"REWORK","_number":1,"created":"2023-10-11 14:48:25.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/67/897967/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/67/897967/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/67/897967/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/67/897967/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/67/897967/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/67/897967/1"}}},"commit":{"parents":[{"commit":"d40e43f39dd34afcbeac656c5a3d2b235b7da9f5","subject":"Merge \"Respect provided \"host\" when plugging ports\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d40e43f39dd34afcbeac656c5a3d2b235b7da9f5"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-21 04:35:09.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-11 14:48:25.000000000","tz":0},"subject":"Fix policy check in metadata APIs","message":"Fix policy check in metadata APIs\n\nWhen a user doesn\u0027t have access to a non-public\nresource, the appropriate response is HTTP 404,\nnot HTTP 403.\n\nChange-Id: I62afec521c5cdfdd67ab83da40e69e6a2688c737\nCloses-Bug: #2004230\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n(cherry picked from commit 267a67dafe2fcf0d23aa962fa39ba9748da94feb)\n(cherry picked from commit 1ee34c740bfe8f76d4c586beebc78d4b26459185)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/355f1ec9381c00f694ebe5228321c07ed3f2a97a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/355f1ec9381c00f694ebe5228321c07ed3f2a97a"}]},"branch":"refs/heads/stable/2023.1"},"d5c2eb1124a9f44b3ad71c6cfdb01f7fa80ee88d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-16 18:13:59.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/67/897967/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/67/897967/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/67/897967/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/67/897967/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/67/897967/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/67/897967/2"}}},"commit":{"parents":[{"commit":"d40e43f39dd34afcbeac656c5a3d2b235b7da9f5","subject":"Merge \"Respect provided \"host\" when plugging ports\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d40e43f39dd34afcbeac656c5a3d2b235b7da9f5"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-09-21 04:35:09.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-16 18:13:59.000000000","tz":0},"subject":"Fix policy check in metadata APIs","message":"Fix policy check in metadata APIs\n\nWhen a user doesn\u0027t have access to a non-public\nresource, the appropriate response is HTTP 404,\nnot HTTP 403.\n\nChange-Id: I62afec521c5cdfdd67ab83da40e69e6a2688c737\nCloses-Bug: #2004230\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n(cherry picked from commit ad29f8a06513342fb63d665a36e16549ecf3aefc)\n(cherry picked from commit 1ee34c740bfe8f76d4c586beebc78d4b26459185)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d5c2eb1124a9f44b3ad71c6cfdb01f7fa80ee88d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d5c2eb1124a9f44b3ad71c6cfdb01f7fa80ee88d"}]},"branch":"refs/heads/stable/2023.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
