)]}'
{"id":"openstack%2Fmanila~897969","triplet_id":"openstack%2Fmanila~stable%2F2023.1~Ifab3ec9de2abeb1af8e6499feca0236e1ee22fcf","project":"openstack/manila","branch":"stable/2023.1","topic":"bug/1967760","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-11-16 19:59:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifab3ec9de2abeb1af8e6499feca0236e1ee22fcf","subject":"Shorten snapshots names in CephFS drivers","status":"MERGED","created":"2023-10-11 16:18:29.000000000","updated":"2023-11-16 20:00:46.000000000","submitted":"2023-11-16 19:59:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897969-bug/1967760","cherry_pick_of_change":896321,"cherry_pick_of_patch_set":1,"meta_rev_id":"490349c47dfde597068e09f364a5b217f76032f5","_number":897969,"virtual_id_number":897969,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-16 19:59:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-11-16 18:27:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-11-16 18:27:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-11 16:18:29.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"CC"},{"updated":"2023-10-11 16:49:01.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-10-11 18:54:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-11 21:56:57.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-11-16 18:27:14.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"83f9876b92b87aa54705063fea9891aca4a124c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-11 16:18:29.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"d5485ed9e3c133e25ca0a194f93ee9c7aedee0bc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-11 16:18:34.000000000","message":"Topic bug/1967760-stable/2023.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"2a217dc84583431be3338e3608d3e40af055c5f9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-11 16:18:37.000000000","message":"Topic set to bug/1967760","accounts_in_message":[],"_revision_number":1},{"id":"3c4f385fb09ca9eabd2c079b284ff09b7d87db59","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-11 16:49:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/69/897969/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/99b941e/ : FAILURE in 29m 22s","accounts_in_message":[],"_revision_number":1},{"id":"d9666139e22b7c2b828b479041d88fa3fe7073b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 18:54:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0be03149660849198fb89481fb6b880e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/369f7cb5831c45e1b21f65c10ff7a11f : SUCCESS in 9m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5d348d1a7454d38ab5bec9fd953cb5c : SUCCESS in 21m 13s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf43330d724b4ccd93b075225f35fdcf : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/457bad5a9884457e98e149df649c3cbf : SUCCESS in 9m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/504b5bbb3410493ca3487346c21d2b42 : SUCCESS in 11m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d09143ea154249d7a6e8f02704ddcd6c : SUCCESS in 8m 52s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c983c6fc0d8644529e478046abd0cedb : SUCCESS in 2h 29m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c662afafc4c540f1b9fe9095b344460c : SUCCESS in 1h 15m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/1ecc59d9d7c94f83aa297593c12f5d4c : FAILURE in 15m 27s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c8bba054b5524c18b3c16fc17dc6a0de : FAILURE in 26m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d82a38c22007440f85ace9ac3a352de3 : SUCCESS in 1h 18m 02s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/892f4370962d42a1887544b84f8044e6 : SUCCESS in 5m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7e2247859aa24371833f48513f8e3652 : SUCCESS in 5m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/43946f59c8914f149ee24b88bafe81f6 : SUCCESS in 43m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/36bdcd9bed8b43d4b53c0820ca470464 : SUCCESS in 42m 12s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2bbfc7e0592f4b2f9ae23f0241559a8d : SUCCESS in 51m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/f098a473a12e42fdb503d8d181f29d1d : FAILURE in 24m 32s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/66c301e474974621a34b30d06657823c : FAILURE in 26m 41s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/c668b5dd9eb647a1969dd106dd31de5d : SUCCESS in 55m 11s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/f0d024a222894102b0ee1087022dbd5f : SUCCESS in 1h 07m 18s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/93594670e6d24079a45bc967dd174c60 : SUCCESS in 1h 12m 13s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/efa56b4a19af4e2f8d0f54a2654989cd : SUCCESS in 39m 21s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7fa70a78b9a144acaf64e37fc26674eb : SUCCESS in 56m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"302a2333c2804e047b253fb7938b033093607051","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-11 21:56:57.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/897969/1/upstream-check/manila-tempest-plugin-ontap-dhss/5f05130/ : FAILURE in 2h 56m 32s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/69/897969/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/c700c10/ : FAILURE in 3h 01m 32s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"e8126d767ec95e12114049b5257babb64da50585","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-13 22:07:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"09074c72ae69742e7d4f3b8f9df5b11e7979eae4","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-11-16 18:27:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ccc5d46d788d2fc4feeb138193470e213ca81ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 18:27:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fdc231a1f25c54e9ab44849b5e3cae50d40eaecb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 19:59:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d34a144bed6a4fd7863f441206267b1c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6e94f40f22b4b3abfafb733aff89cb3 : SUCCESS in 9m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba76f697556a4d408f9fc79c2ef8deb5 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/772dc11724cc481d91ca647fefd494c5 : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a6089867c4e4c2f9c8b38b18c74ae1b : SUCCESS in 7m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/45085bc8ffab42ca83c05053ccf853c9 : SUCCESS in 1h 23m 08s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a44492e4e0ef4fce882f71f11f42a2ee : SUCCESS in 36m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/38319a1c0f2a4b578b866cdf85701d68 : SUCCESS in 33m 37s","accounts_in_message":[],"_revision_number":1},{"id":"b9d2f8acd89146e47088ed62e3d6bee731ede9a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 19:59:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"490349c47dfde597068e09f364a5b217f76032f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 20:00:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6ec02fa11d04b87baca2856636734b8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/205547d50cbe444ea361394c37a2874c : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10ac0600479841e1af05921fdeea2b67 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3be5df4e1fe5feac43c3c1524eb00c4441eec476","revisions":{"3be5df4e1fe5feac43c3c1524eb00c4441eec476":{"kind":"REWORK","_number":1,"created":"2023-10-11 16:18:29.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/69/897969/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/69/897969/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/69/897969/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/69/897969/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/69/897969/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/69/897969/1"}}},"commit":{"parents":[{"commit":"6781f22bf1d78b89f500f538244d0991152f8e6d","subject":"[RBAC] Enforce check for share updates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6781f22bf1d78b89f500f538244d0991152f8e6d"}]}],"author":{"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","date":"2022-09-19 08:31:13.000000000","tz":0},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-11 16:18:29.000000000","tz":0},"subject":"Shorten snapshots names in CephFS drivers","message":"Shorten snapshots names in CephFS drivers\n\nThe name set to snapshots in the CephFS drivers\nis generated appending the snapshot id with the\nsnapshot instance id.\n\nThis is a problem because Ceph has a limitation\nin the name of the subvolumes (shares) being created.\n\nWhat happens then is, when trying to create a snapshot,\nthe snapshot gets created with a truncated name and it cannot\nbe accessed later.\n\nThis change drops the snapshot instance id from the\nassigned name in order to avoid this limitation.\n\nCloses-Bug: #1967760\n\nChange-Id: Ifab3ec9de2abeb1af8e6499feca0236e1ee22fcf\n(cherry picked from commit c86523d5909a2d61e03fb29f4c0c71b67cf18205)\n(cherry picked from commit 4d4d4bdb118968b1cb089fa8b415c8354874e3a7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3be5df4e1fe5feac43c3c1524eb00c4441eec476"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3be5df4e1fe5feac43c3c1524eb00c4441eec476"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
