)]}'
{"id":"openstack%2Fmanila~898488","triplet_id":"openstack%2Fmanila~stable%2Fyoga~Ic3be777b238a467d1b7bd1daa6aa088dedb095b0","project":"openstack/manila","branch":"stable/yoga","topic":"bug/1955627","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-02-20 14:09:52.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic3be777b238a467d1b7bd1daa6aa088dedb095b0","subject":"[RBAC] Enforce check for share updates","status":"ABANDONED","created":"2023-10-16 22:15:31.000000000","updated":"2024-02-20 14:09:52.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":897962,"cherry_pick_of_patch_set":1,"meta_rev_id":"02d5a8a3d124f6222f455e652195b75d18f27579","_number":898488,"virtual_id_number":898488,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-02-08 05:53:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-16 23:21:18.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-10-17 00:48:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-20 14:09:52.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"cc08a7a02695722175918b618b5580de7c563939","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-16 22:15:31.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"c77f0222e26d49210f777f263b9db4874a863b50","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-16 22:15:39.000000000","message":"Topic bug/1955627-stable/yoga removed","accounts_in_message":[],"_revision_number":1},{"id":"1983a1360b695f44493250e9c9cb2b3d99ae4a06","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-16 22:15:42.000000000","message":"Topic set to bug/1955627","accounts_in_message":[],"_revision_number":1},{"id":"ba29c3fa5d413e199ba398eb65d41b9f50e89983","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-16 23:21:18.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/88/898488/1/thirdparty-check/pure-devstack-manila-tempest-aio/b5caf4e/ : SUCCESS in 1h 03m 52s","accounts_in_message":[],"_revision_number":1},{"id":"2ac0dfd95fcb68385bdeeffecf91eb611fd72f19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-17 00:48:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7141529c8b44159b345f62c010fb04f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa783731b61d496eb4031efc371f25ea : SUCCESS in 10m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/007f4651ec854675be2bf2dff458d772 : SUCCESS in 22m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae495a5df7e146788f7490f412024b0c : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef47604121df4e60b4855a354d3ab445 : SUCCESS in 8m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb0b9cb9671b47a1bcc58d736f900da1 : SUCCESS in 10m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e291bfa60e14454b87f79b769cec8fdf : SUCCESS in 6m 09s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/1799764145054f6391dd37181f9a5c25 : SUCCESS in 34m 48s\n- manila-tempest-plugin-generic-scenario-stable https://zuul.opendev.org/t/openstack/build/147cf0ae651a4894a5a1b41f2cfe899f : FAILURE in 1h 56m 54s (non-voting)\n- manila-tempest-plugin-generic-stable https://zuul.opendev.org/t/openstack/build/7a148cd1858941da8e9a36bb6d3615e2 : SUCCESS in 1h 33m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native-stable https://zuul.opendev.org/t/openstack/build/104f0f3e135049808e016e5ea39fd346 : SUCCESS in 1h 11m 09s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-stable https://zuul.opendev.org/t/openstack/build/f6ad49a5d7ea43878298ff9695b3c9b6 : FAILURE in 2h 00m 47s (non-voting)\n- manila-tempest-plugin-zfsonlinux-stable https://zuul.opendev.org/t/openstack/build/52077c61c67348198c2020b96649f3a3 : FAILURE in 1h 28m 25s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f601ac85ece74db8b8b7e1e6aaea2146 : SUCCESS in 5m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ecb97b7668b0483b96da4093d2218873 : SUCCESS in 5m 07s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ec7870cd8a044055a28226033fd9aca7 : FAILURE in 26m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cd22e3e014164613879e01f7f9edc466 : FAILURE in 30m 03s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ca2d1a9b48f34bfcb45cc8bae535321e : SUCCESS in 1h 05m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/89f2c84e3e254fa3b22c359d1130de7f : SUCCESS in 30m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/284ebeef697d4343a3c9bf05ce4c9d67 : SUCCESS in 30m 20s (non-voting)\n- manila-tempest-plugin-lvm-fips-stable https://zuul.opendev.org/t/openstack/build/038cc65a7e744c9cbd458a6da628ce0e : SUCCESS in 38m 08s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/16cf128c0b8941358746d95882ec90a9 : SUCCESS in 1h 03m 09s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/8b0ff007adbc4c31af27cdaa55c545ad : SUCCESS in 31m 21s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3513ce91f15848dcb3e61e1445fd1031 : SUCCESS in 46m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"97be77d94e136c108e1ac1bcfbece7e717e25741","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-12-19 22:41:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"105da350e857396fb2d6cc9e0c0f03738aeb72b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-20 00:55:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c11d5523f00e4545bafbe13f9ff94220\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/541f16cb9763434f8223164dfe624881 : SUCCESS in 11m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6930ba6b5f274335aa32e6ebfaa6ab8b : SUCCESS in 16m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f733a539c7f45d198aaca9122cc845e : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48562e77447341b59cf8728707f94847 : SUCCESS in 7m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d492484a3a2459b92bf432c4c709412 : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f78982c6713a4bcc9af5398080e934db : SUCCESS in 7m 43s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/f65b1d8e796e480291a797f72e18ee5e : SUCCESS in 1h 24m 22s\n- manila-tempest-plugin-generic-scenario-stable https://zuul.opendev.org/t/openstack/build/63e3cd1a1f724ea0812216441902c39e : FAILURE in 2h 08m 24s (non-voting)\n- manila-tempest-plugin-generic-stable https://zuul.opendev.org/t/openstack/build/643f7acb0c0345ff83e165bc826f8681 : FAILURE in 1h 51m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native-stable https://zuul.opendev.org/t/openstack/build/7175aa0460fc418284aae2ec8e514758 : SUCCESS in 1h 27m 20s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-stable https://zuul.opendev.org/t/openstack/build/c7ebad90c45d43a4bf4ca1b41b468e7a : FAILURE in 1h 47m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux-stable https://zuul.opendev.org/t/openstack/build/fd7c13f3ce734bf99a42534a972333f2 : FAILURE in 1h 05m 53s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6f5f28ef36ef4b12b3dec68331d36805 : SUCCESS in 4m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/de8a167d79bd44e2bab04bcf03e95ca8 : SUCCESS in 4m 29s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/600b0d274e07494a916f80f85c4bbf4e : FAILURE in 24m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dbbc4094143c461ba11335f9694616fe : FAILURE in 39m 33s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/460085043718400c9297d16197dad9ac : FAILURE in 51m 05s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/5a781b3b132a4e52814f944f94431908 : SUCCESS in 34m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d580259565d54132a1859516cc64db21 : SUCCESS in 19m 56s (non-voting)\n- manila-tempest-plugin-lvm-fips-stable https://zuul.opendev.org/t/openstack/build/d2ec30b2a34e44ac8388e36f5ac80478 : SUCCESS in 58m 36s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/310deaaf114b4bf3a68a5d4e7b3b68f0 : SUCCESS in 53m 55s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/a61120941e314bdabede8d42b4fdcfc5 : SUCCESS in 33m 01s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f047a52f50c247b09d713769003ed7ce : SUCCESS in 33m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e1a1afd0d3e77e72915ae52f87b4e2ceb7669fca","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-02-06 05:58:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7421f34b0f0c2a3864553880de51b909992cd13c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-06 08:08:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc8307138bcf450eba294fcd16d38c06\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6795507fcbb24b4ca6ae91884433d37a : SUCCESS in 10m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d0966c3e6074b1fbefbf2935cb9b092 : SUCCESS in 18m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d98f5de378354e9e836f5ae97805a97c : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe8fd6fd310b488a9d2c7197eaab9604 : SUCCESS in 9m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/add7c6edc60644d8a15415e546dfc604 : SUCCESS in 10m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b33082aa158340e7a93b72cf63e4e6f8 : SUCCESS in 6m 42s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/a1af0770c2a0458790eaa37181d7a00e : SUCCESS in 1h 22m 59s\n- manila-tempest-plugin-generic-scenario-stable https://zuul.opendev.org/t/openstack/build/5e5588c257ea4959857b860cfaddfa63 : FAILURE in 2h 04m 43s (non-voting)\n- manila-tempest-plugin-generic-stable https://zuul.opendev.org/t/openstack/build/6b8606aa97814a39b05d308b88e2aaab : SUCCESS in 1h 03m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native-stable https://zuul.opendev.org/t/openstack/build/11253d96e7ca4c46bb775105504968c7 : SUCCESS in 1h 10m 02s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-stable https://zuul.opendev.org/t/openstack/build/2e1896094d6b4801bb64552436c5ca30 : FAILURE in 1h 06m 28s (non-voting)\n- manila-tempest-plugin-zfsonlinux-stable https://zuul.opendev.org/t/openstack/build/8f414faa6e104dfda14619d2a2f573c8 : FAILURE in 1h 18m 02s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/23714a7bd8a846828bc75c7648b378a4 : SUCCESS in 4m 17s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d03e1fe861bb4dddb883dde1c95ab620 : SUCCESS in 4m 48s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0212796f9f584ae5bca057723b598154 : FAILURE in 29m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/090362cfe419454c9142c26172b53e92 : FAILURE in 40m 13s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b07968731c8641c6963e451c9da8aafe : SUCCESS in 1h 02m 52s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/8d60c43508c64e708d0c48ee1ccc0564 : SUCCESS in 51m 04s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9c82f04bb0104c95949ea93c7f24c21f : SUCCESS in 28m 43s (non-voting)\n- manila-tempest-plugin-lvm-fips-stable https://zuul.opendev.org/t/openstack/build/3da4a05899044b5c84c53aa4a75a98d0 : SUCCESS in 1h 19m 08s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/05b2cd31570249369a4e1ef943874016 : SUCCESS in 54m 36s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/71c29e7266894ecb9d038e4c13c6d338 : SUCCESS in 37m 24s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/61cecbad6e264abda5ff2448cc4f6dfa : SUCCESS in 41m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0137295c07ef849768c7e1c9eeb605b99858fa6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-02-08 05:03:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7487aa8156283f8e7d0085411f4a5fbe086eb501","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 05:53:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fffa9af1f8034d6da0340673d0677cce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68837a92cb2546a883510bb62b466da4 : SUCCESS in 11m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e31b2a604bad4a53abe5ce4033e72b73 : SUCCESS in 14m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37aa9c01d9264f709906fd149b7b8ee2 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31c2e7aa19094451b9964f7d2370f51e : SUCCESS in 10m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df53621df4b24ec0b9f26b877ec096fa : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a96e3f72a624407b139af136a6cb1a1 : SUCCESS in 7m 27s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/f02e49cf3e3d4f458c8247dc9f57892a : FAILURE in 13m 45s\n- manila-tempest-plugin-generic-scenario-stable https://zuul.opendev.org/t/openstack/build/cdafaebf5a184b21b862c3ebd64955dc : FAILURE in 27m 04s (non-voting)\n- manila-tempest-plugin-generic-stable https://zuul.opendev.org/t/openstack/build/beb46da821ad4225b3954912206f036b : FAILURE in 27m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native-stable https://zuul.opendev.org/t/openstack/build/e2cb366fbe144240a415127c65888ea4 : FAILURE in 19m 48s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-stable https://zuul.opendev.org/t/openstack/build/9ebf359cb083405c83b1898e29249c06 : FAILURE in 15m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux-stable https://zuul.opendev.org/t/openstack/build/6ea24331ccbe439ea84d7437d5e9368a : FAILURE in 14m 26s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6f67194af883470cb3c94ee8796ee7b4 : SUCCESS in 4m 18s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c26387f98da34766a4a08ec4609f43b5 : SUCCESS in 4m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d6131d44831e4941b9c22863dd811ee8 : SUCCESS in 23m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5fb80b50d8074c049018627ecc29b68c : SUCCESS in 32m 39s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7e62d9e439884a5c91b07c773e6a0ed3 : FAILURE in 27m 51s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/97bd7e3eedf54dff8e5aba42cfc1a607 : SUCCESS in 44m 06s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8bcf062f0325437e85c9d3fb706a2608 : SUCCESS in 30m 39s (non-voting)\n- manila-tempest-plugin-lvm-fips-stable https://zuul.opendev.org/t/openstack/build/74d9b605a1b8447da21961bbe42cfd94 : FAILURE in 18m 22s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/e66322e1c8974f1e9c7c5b1c8c79271d : FAILURE in 30m 53s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/705e3984713e4b1cb897318d5a1794ad : FAILURE in 17m 27s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3d1bd4c2a27240268db882c3e3c437b4 : FAILURE in 20m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0ace01d953cf47b5981dbd983ff9faf01f68436e","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-02-08 23:18:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"895ad31405639efaeb3df497eb66bb1f83efd82b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 00:05:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f9a72999c224e52905343425a2eb6b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7587ee470ff44dd380081942f46c0a9b : SUCCESS in 9m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00ad15ae9ddb4d53a11b9a41a01e0e54 : SUCCESS in 22m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/714980815abc4ecfa605858f8a06dfbf : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61c31268433149108987761af9d6fd17 : SUCCESS in 7m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26a29bcdb8474a4187b7232f5e30703b : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c41c9c40eabd4a7a83fd9b804a147bf0 : SUCCESS in 5m 29s\n- manila-tempest-plugin-lvm-stable https://zuul.opendev.org/t/openstack/build/320b4c018a2540d3b0ec5a62ca61f55a : FAILURE in 15m 59s\n- manila-tempest-plugin-generic-scenario-stable https://zuul.opendev.org/t/openstack/build/91a1f17d37c94800a85ef4baad0c1f11 : FAILURE in 17m 37s (non-voting)\n- manila-tempest-plugin-generic-stable https://zuul.opendev.org/t/openstack/build/584751dc0aca4eada80440f1616c591c : FAILURE in 20m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native-stable https://zuul.opendev.org/t/openstack/build/2cdbcc1b960c4394881cccc89c1fa2f9 : FAILURE in 18m 30s (non-voting)\n- manila-tempest-plugin-cephfs-nfs-stable https://zuul.opendev.org/t/openstack/build/7028ecce91714830a13f66abc64ac885 : FAILURE in 14m 19s (non-voting)\n- manila-tempest-plugin-zfsonlinux-stable https://zuul.opendev.org/t/openstack/build/97e92a223c5f4a358dbafd2ce439f750 : FAILURE in 17m 11s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3b532f676ce44d208dfc6d3a918c77fe : SUCCESS in 4m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4d66e9132402433caf9a357166ecc419 : SUCCESS in 6m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8e63f7a5565a44d89a264e1aa3ed123e : SUCCESS in 28m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3a4106a0a0884021a439330b8fe5bf52 : SUCCESS in 34m 29s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cfa91e8b470b47f690413911046338f0 : FAILURE in 19m 38s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/f8744dd33db4481d822c7872f0331831 : SUCCESS in 40m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9e8f644e58144dc4839f7f7711fe9d69 : SUCCESS in 20m 31s (non-voting)\n- manila-tempest-plugin-lvm-fips-stable https://zuul.opendev.org/t/openstack/build/9b48de78cd5b45c393a6dcc307c80bb2 : FAILURE in 29m 08s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/f171f55206854d5bbc4f258ae0b9db3f : FAILURE in 28m 57s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/102eec1f37b24436898879f5aa1d08b6 : FAILURE in 31m 13s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/43637671cf0c4aa0b1a574a0d03f8eff : FAILURE in 28m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"02d5a8a3d124f6222f455e652195b75d18f27579","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-20 14:09:52.000000000","message":"Abandoned\n\nstable/yoga branch of openstack/manila is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/yoga if you want to further work on this patch.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1c18e6cbc1c2ad448262f50bd817cf98fc17b5e9","revisions":{"3d6f685931a29ee53c37e5b87751d224de1623d8":{"kind":"REWORK","_number":1,"created":"2023-10-16 22:15:31.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/88/898488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/88/898488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/88/898488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/88/898488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/88/898488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/88/898488/1"}}},"commit":{"parents":[{"commit":"dd0492623c2f25b551fa9b29635adb18da8582c1","subject":"Fix duplicate entries in share_server_backend_details","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dd0492623c2f25b551fa9b29635adb18da8582c1"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-09-15 20:56:44.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-16 22:15:31.000000000","tz":0},"subject":"[RBAC] Enforce check for share updates","message":"[RBAC] Enforce check for share updates\n\nWhen a user has access to the APIs to reset status,\ntask state or replica state but doesn\u0027t have access to\nthe share, they must be prevented from performing\nthose actions. This enforcement allows granular control\nof these actions and the resources themselves.\n\nChange-Id: Ic3be777b238a467d1b7bd1daa6aa088dedb095b0\nCloses-Bug: #1955627\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n(cherry picked from commit 55edb00cc12ade9df333e9c0a3472e68d797b8f6)\n(cherry picked from commit f1aa5f0744f21cdae9394da00f36cc57ce5e09f0)\n(cherry picked from commit 6781f22bf1d78b89f500f538244d0991152f8e6d)\n(cherry picked from commit 27a89a6a1cfa0f75d270e46c933aa24cd9708c27)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3d6f685931a29ee53c37e5b87751d224de1623d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3d6f685931a29ee53c37e5b87751d224de1623d8"}]},"branch":"refs/heads/stable/yoga"},"1c18e6cbc1c2ad448262f50bd817cf98fc17b5e9":{"kind":"REWORK","_number":2,"created":"2024-02-08 05:03:42.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/88/898488/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/88/898488/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/88/898488/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/88/898488/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/88/898488/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/88/898488/2"}}},"commit":{"parents":[{"commit":"dd0492623c2f25b551fa9b29635adb18da8582c1","subject":"Fix duplicate entries in share_server_backend_details","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dd0492623c2f25b551fa9b29635adb18da8582c1"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2022-09-15 20:56:44.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2024-02-08 05:03:26.000000000","tz":-480},"subject":"[RBAC] Enforce check for share updates","message":"[RBAC] Enforce check for share updates\n\nWhen a user has access to the APIs to reset status,\ntask state or replica state but doesn\u0027t have access to\nthe share, they must be prevented from performing\nthose actions. This enforcement allows granular control\nof these actions and the resources themselves.\n\nChange-Id: Ic3be777b238a467d1b7bd1daa6aa088dedb095b0\nCloses-Bug: #1955627\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n(cherry picked from commit 55edb00cc12ade9df333e9c0a3472e68d797b8f6)\n(cherry picked from commit f1aa5f0744f21cdae9394da00f36cc57ce5e09f0)\n(cherry picked from commit 6781f22bf1d78b89f500f538244d0991152f8e6d)\n(cherry picked from commit 27a89a6a1cfa0f75d270e46c933aa24cd9708c27)\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1c18e6cbc1c2ad448262f50bd817cf98fc17b5e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1c18e6cbc1c2ad448262f50bd817cf98fc17b5e9"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
