)]}'
{"id":"openstack%2Fmanila~898992","triplet_id":"openstack%2Fmanila~master~I73c2706c9d889f8d08057a543b2a963740bb3f33","project":"openstack/manila","branch":"master","topic":"bug/2035572","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2023-10-23 12:53:21.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2023-11-09 16:35:32.000000000","reason":"\u003cGERRIT_ACCOUNT_30002\u003e replied on the change","reason_account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},"32594":{"account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"last_update":"2023-11-02 16:26:29.000000000","reason":"\u003cGERRIT_ACCOUNT_32594\u003e replied on the change","reason_account":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-11-09 17:55:39.000000000","reason":"Change was submitted"},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2023-11-09 17:55:39.000000000","reason":"Change was submitted"},"6413":{"account":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"last_update":"2023-11-09 17:55:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I73c2706c9d889f8d08057a543b2a963740bb3f33","subject":"Don\u0027t fail remove-export in NFS-Ganesha","status":"MERGED","created":"2023-10-20 22:39:30.000000000","updated":"2023-11-09 17:56:51.000000000","submitted":"2023-11-09 17:55:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898992-bug/2035572","meta_rev_id":"5cc1c23cee16eefd1e8eb3912439248111e09c64","_number":898992,"virtual_id_number":898992,"owner":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-09 17:55:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-10-23 12:53:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2023-11-02 16:26:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":2,"date":"2023-11-09 16:35:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2023-10-30 01:24:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"value":1,"date":"2023-11-09 16:35:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"}],"reviewers":{"REVIEWER":[{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-20 23:10:55.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-21 00:08:39.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-10-21 00:09:17.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-10-21 01:54:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-21 12:46:56.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-10-21 14:10:17.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-10-22 22:57:45.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2023-10-22 22:57:45.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":6413,"name":"Victoria Martinez de la Cruz","email":"victoria@redhat.com","username":"vkmc"},"state":"REVIEWER"},{"updated":"2023-10-22 22:57:45.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2023-10-22 22:57:53.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"state":"REVIEWER"},{"updated":"2023-10-30 01:24:10.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-10-31 18:07:58.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"}],"messages":[{"id":"5599dbfeca2a8c17b8b4d4b6e0dceffffe110418","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-20 22:39:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"89d61b1394cb622a6d7756a46cd2e08486d4ed5d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-20 23:10:55.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/92/898992/1/check/DellEMC_PowerFlex_Manila_dhss_false/75569a4/DellEMC_PowerFlex_Manila_dhss_false/587 : SUCCESS in 31m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"dcb3cffb39a86b55b87c6f97510ca2f1a2f9a891","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-21 00:08:39.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-92-898992-1/NFS/ : SUCCESS in 1h 28m 51s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"df8c49c298f07379c9887e297fad9f9bb1270392","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-21 00:09:17.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/92/898992/1/check/DellEMC_Unity_Manila_dhss_true/dfb798c/DellEMC_Unity_Manila_dhss_true/689 : FAILURE in 1h 29m 32s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/92/898992/1/check/DellEMC_Unity_Manila_dhss_false/bb4d334/DellEMC_Unity_Manila_dhss_false/726 : SUCCESS in 1h 02m 43s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d70e7bc5e5979d1015a44a7b8789ea9ae73a49c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-21 01:54:48.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0be9cac0c51245f0aa70d8c15ab40028\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1805a3e303304ed1a776bafda7ba3562 : SUCCESS in 8m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/225dcb0464f944d399b330df5a6eaad7 : SUCCESS in 23m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38dfbcd32de142efb51252c3cda30476 : FAILURE in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd1e310867624d819c204fc5d8b1c7c6 : SUCCESS in 9m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/afe3e02a9a5e4004a558369bbfedbb2a : SUCCESS in 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c56e2b2e35f247209257435b376e84dd : SUCCESS in 7m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c7a068c02c0b42aa9a230936e3c53259 : SUCCESS in 35m 11s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/624879f9c7474543a5886f805f2df55f : TIMED_OUT in 3h 07m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/55be932f70804c8393c9045f307b48c7 : SUCCESS in 1h 12m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e11652638ab94bc080330e37b8272bc4 : SUCCESS in 1h 07m 22s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/5d047bb273684c6c81658277bcb88946 : SUCCESS in 1h 23m 45s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/dbd05a1584fa4d43b371c5ed733d3019 : SUCCESS in 1h 08m 57s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6b8ea0c9b8c74f24b09765b71778afdf : SUCCESS in 4m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7f5d8728ea244fdd9db6c79699bc778b : SUCCESS in 4m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b32eb72f36ce4659aaf8c198106bd5e4 : SUCCESS in 26m 42s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1f849ff6250043c5a12b265b1e279521 : SUCCESS in 27m 19s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/00af77309bdd4211ae6ce71fa738092f : SUCCESS in 32m 57s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/ce0ec58abcc849329f7994d5d97e4079 : FAILURE in 12m 57s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9869bae9d32041698fcc8a058e59a7df : FAILURE in 11m 15s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/20a2fc9945964a5593eeaf45b5b3a9ec : SUCCESS in 37m 30s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/202ec44686984d5caaf097bf1fc70a44 : SUCCESS in 44m 59s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0ceb49f78fd84c7c91d7925c73aff971 : SUCCESS in 42m 26s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0c7e6f1fb109447c9563b8b39db5655a : SUCCESS in 26m 30s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/d2db2ff12474432a980dcdb2fce539c3 : SUCCESS in 28m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5df4a81d637bd8c3a03e800192b3d8b9ee4cec77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-21 09:56:42.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9165cc9e2f6fa1e8b007bed24ead74d3da58560f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-21 10:29:18.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/92/898992/2/check/DellEMC_PowerFlex_Manila_dhss_false/209aeac/DellEMC_PowerFlex_Manila_dhss_false/588 : SUCCESS in 32m 20s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"f7743ee805186a56fd8e8758707ca052185220d1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-21 11:28:58.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/92/898992/2/check/DellEMC_Unity_Manila_dhss_true/f8560d2/DellEMC_Unity_Manila_dhss_true/690 : FAILURE in 1h 32m 00s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/92/898992/2/check/DellEMC_Unity_Manila_dhss_false/095c997/DellEMC_Unity_Manila_dhss_false/727 : SUCCESS in 1h 09m 42s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"bd3f687732c1634f32829a3848c7072a53a87ff7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-21 12:15:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a836fc9448a948998535c7f3a758d3d1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcc935a5e98247a4914f92aa8ec3eca0 : SUCCESS in 11m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e1e79f9b9704d58a23d80c3379b884e : SUCCESS in 24m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af9b1361279c4af68f7c3dee4a82a775 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0b3f4f912444f30b743a8aaa102db9e : SUCCESS in 9m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d7ff45b9683048118cb56f16cf8eb9e9 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/567e4d51a0d34e0b8ea7e9cb0ed4bf80 : SUCCESS in 6m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a2c18cf4a36e47db859e5db585099be6 : SUCCESS in 26m 30s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9f94183f0b424df99b77bcbc2fee18ab : SUCCESS in 2h 13m 52s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c2f999dc7cb54c1c80ab90a5ede8ddae : SUCCESS in 1h 31m 57s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3f7cc6b14a2940e6aaef0240e437a41a : SUCCESS in 1h 08m 43s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8f9e01b090a94009aba38b0102d55191 : SUCCESS in 1h 29m 31s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c96720bce7834b59b3bc157d4de25450 : SUCCESS in 1h 19m 15s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/bcec44d163c04a19a84a2ba1f41df716 : SUCCESS in 4m 10s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ee8951479dca4058afea63e0dadf972e : SUCCESS in 7m 53s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8a40863a4a73477d8837cd6f1338e8ab : SUCCESS in 24m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0c8048f772fa4c47a9cde61e0d15c7bf : SUCCESS in 45m 46s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/22b21c0fa8c24fc7870515225c3e7aca : SUCCESS in 54m 47s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/d29323b14a3849a4a98a4ef8205785ad : FAILURE in 14m 37s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/01ac8932aa29419483cb1a62c870d948 : FAILURE in 23m 19s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/9d558bdf618041b4b10305f094ee2577 : SUCCESS in 51m 49s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0756e60af96041f7a20e70c0235b5b9d : SUCCESS in 39m 21s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/80a81d5d5c3e47e2a2373e454bbce527 : SUCCESS in 1h 26m 57s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/66212c0e26be4ceeb42812c000aa8123 : SUCCESS in 25m 33s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/4cfdd9bb6e2e48c3a56571b6fc491ba3 : SUCCESS in 47m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"741f0412a05a6c0fe205fd0f82290b2d649c2557","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-21 12:46:48.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/92/898992/2/check/DellEMC_PowerFlex_Manila_dhss_false/5a1681f/DellEMC_PowerFlex_Manila_dhss_false/589 : SUCCESS in 30m 36s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"879107d0bd3d19c2edb795350254bb6b044f16a2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-10-21 12:46:56.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/92/898992/2/check/DellEMC_PowerStore_Manila_dhss_false/75a153e/DellEMC_PowerStore_Manila_dhss_false/253 : SUCCESS in 30m 45s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"24aea3babf9eb4feafc534d198ad8aebf72d309b","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-21 14:10:17.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/898992/2/upstream-check/manila-tempest-plugin-ontap-dhss/0d46100/ : RETRY_LIMIT in 7m 40s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/92/898992/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/3e70530/ : FAILURE in 1h 52m 37s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"c26c09bfffefb83972e41a196af607cc0802a784","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-21 14:23:27.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/92/898992/2/check/DellEMC_Unity_Manila_dhss_true/5a823d5/DellEMC_Unity_Manila_dhss_true/691 : FAILURE in 2h 07m 15s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/92/898992/2/check/DellEMC_Unity_Manila_dhss_false/7d90793/DellEMC_Unity_Manila_dhss_false/728 : SUCCESS in 59m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"ac5cc2ff133df56061cab269be55c60af4310fd2","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-10-23 12:53:21.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"766d385711424e10cd6a7bcd44805ecd18b53dce","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-10-30 01:24:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"de835972dc879fdafcb2f4daf337eb9dea118d06","author":{"_account_id":32594,"name":"Ashley Rodriguez","email":"ashrod98@redhat.com","username":"ashrod98"},"date":"2023-11-02 16:26:29.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2745d882c0a0649700775a8fc633e146446b544e","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2023-11-09 16:35:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"203165794775865d6b4b8dbe5b2e197e8428b944","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 16:36:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3b79dd4d6ebd4adcd79d9cbc0d586416a975caf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:55:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69462dab956440869c36c51cd7cfd9e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4921c6ad0666444f9ce6a481fc86a999 : SUCCESS in 10m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a66fb7d6530b45819417c9a423f9f7e8 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e10808f379145c7ab2f4b9fc1521a94 : SUCCESS in 10m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/38a67f8075184792838f3b92e32eb918 : SUCCESS in 11m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/961f280622554c5db66989d1e6e79aaa : SUCCESS in 7m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2f42141c315249a1b738655d0962a927 : SUCCESS in 1h 19m 29s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2164082359684ea3967b79a509e93806 : SUCCESS in 45m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4bddd7831da743b8b6f26481b613e3ff : SUCCESS in 33m 13s","accounts_in_message":[],"_revision_number":2},{"id":"11fd17f9d401fab6ff6ee66ab77470646fee23a6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:55:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5cc1c23cee16eefd1e8eb3912439248111e09c64","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:56:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a34fe9f4c91248c5a14119155134de5a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3e4befe416d4e79a84c6cd6faaa8445 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58e7d7ac6a3b43daa6f1eb0225799d95 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"609b925a51b230c93d8ad408cf8ba351657dd83e","revisions":{"847a314277935991556755ff9c70d8e3b5bd5a17":{"kind":"REWORK","_number":1,"created":"2023-10-20 22:39:30.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/92/898992/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/92/898992/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/92/898992/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/92/898992/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/92/898992/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/92/898992/1"}}},"commit":{"parents":[{"commit":"8229f3e5c19b079edc3a7a23185e4bd8ce52d1ec","subject":"Fix BadRequest from share-server-migration-get-progress","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8229f3e5c19b079edc3a7a23185e4bd8ce52d1ec"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-20 22:25:01.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-20 22:39:05.000000000","tz":-420},"subject":"Don\u0027t fail remove-export in NFS-Ganesha","message":"Don\u0027t fail remove-export in NFS-Ganesha\n\nWhen dropping an export over DBUS, if there\u0027s a problem\nwith the removal, we can ignore the failure and move on.\nIf not, the share ends up in an \"error_deleting\" state.\n\nThere\u0027s not much that we can do here besides logging\nthe failure as an exception.\n\nChange-Id: I73c2706c9d889f8d08057a543b2a963740bb3f33\nCloses-Bug: #2035572\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/847a314277935991556755ff9c70d8e3b5bd5a17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/847a314277935991556755ff9c70d8e3b5bd5a17"}]},"branch":"refs/heads/master"},"609b925a51b230c93d8ad408cf8ba351657dd83e":{"kind":"REWORK","_number":2,"created":"2023-10-21 09:56:42.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/92/898992/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/92/898992/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/92/898992/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/92/898992/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/92/898992/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/92/898992/2"}}},"commit":{"parents":[{"commit":"8229f3e5c19b079edc3a7a23185e4bd8ce52d1ec","subject":"Fix BadRequest from share-server-migration-get-progress","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8229f3e5c19b079edc3a7a23185e4bd8ce52d1ec"}]}],"author":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-20 22:25:01.000000000","tz":-420},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-10-21 09:56:42.000000000","tz":0},"subject":"Don\u0027t fail remove-export in NFS-Ganesha","message":"Don\u0027t fail remove-export in NFS-Ganesha\n\nWhen dropping an export over DBUS, if there\u0027s a problem\nwith the removal, we can ignore the failure and move on.\nIf not, the share ends up in an \"error_deleting\" state.\n\nThere\u0027s not much that we can do here besides logging\nthe failure as an exception.\n\nChange-Id: I73c2706c9d889f8d08057a543b2a963740bb3f33\nCloses-Bug: #2035572\nSigned-off-by: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/609b925a51b230c93d8ad408cf8ba351657dd83e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/609b925a51b230c93d8ad408cf8ba351657dd83e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
