)]}'
{"id":"openstack%2Fmanila~899149","triplet_id":"openstack%2Fmanila~stable%2F2023.2~I3d9a3bbc81f18d60a0710d9336ad6f97176875be","project":"openstack/manila","branch":"stable/2023.2","topic":"bug/2037580","attention_set":{},"removed_from_attention_set":{"12404":{"account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"last_update":"2023-11-10 01:29:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3d9a3bbc81f18d60a0710d9336ad6f97176875be","subject":"Add service_network_host option for port host","status":"MERGED","created":"2023-10-25 02:27:23.000000000","updated":"2023-11-10 01:30:13.000000000","submitted":"2023-11-10 01:29:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899149-bug/2037580","cherry_pick_of_change":897077,"cherry_pick_of_patch_set":5,"meta_rev_id":"ff62a0ed4b234216963c73223a7ec6031563f491","_number":899149,"virtual_id_number":899149,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-10 01:29:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2023-11-10 00:10:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2023-11-10 00:10:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-25 03:46:58.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-10-25 04:53:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-25 05:02:32.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-11-10 00:10:22.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"06d7cfea119ae26561918dceea8720fafced507e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-10-25 02:27:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c344cb3466f46f296ac7b4473214a908751e6796","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-10-25 02:27:34.000000000","message":"Topic bug/2037580-stable/2023.2 removed","accounts_in_message":[],"_revision_number":1},{"id":"42c3d6d883393283a8efc4f758e1b4b74894a519","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-10-25 02:27:41.000000000","message":"Topic set to bug/2037580","accounts_in_message":[],"_revision_number":1},{"id":"35e31586d7195cdc6fea46432c17dd96374a56db","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-25 03:46:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/49/899149/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/c2645ae/ : SUCCESS in 1h 17m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9f3f5b87f12f7a3f4f015d934783cbb1a7ff0397","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 04:53:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f65b0287fae54c48a1bc027f188f856e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c626d51489a4725aa4b392772b79184 : SUCCESS in 11m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ab7ebb12584406dba66c476b05f81e1 : SUCCESS in 25m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/511164c7faf84da6b637a903426150fc : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c83151db5c334c0fa1c9eb31946ced00 : SUCCESS in 11m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad5f1fc0342c46bb9277cc16bdd2442b : SUCCESS in 11m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9bcbbfa54dc42f6a88cc05398c32be9 : SUCCESS in 13m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9425d5e9473b4fd9aa2fb1dfc8b1a3a4 : SUCCESS in 10m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b28bf3e4aa740b98f8d3316bd32c511 : SUCCESS in 8m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/add04c1fd0e541a5a9937e607303667d : SUCCESS in 40m 17s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c2be0d47296e4a5999c87d2a779233e0 : SUCCESS in 2h 08m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a434d6af543e4c84abd81bc22a91f7d6 : SUCCESS in 1h 19m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/864d73de88e04daaa9dcf9aee4d1400e : SUCCESS in 1h 20m 11s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/763f6aa896104972af2803ca98d7b927 : SUCCESS in 1h 37m 43s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/07a3c9718e964164adfb11682a44f60b : SUCCESS in 56m 48s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/2645404fe1e74bd3840a5a3f37bbc6e7 : SUCCESS in 5m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ee5d1052f7234614918789a5ac7a10b3 : SUCCESS in 4m 22s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/aaa3586f9d784876a4f45f5f489e77ff : SUCCESS in 37m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3e0de814c6584356ab2a27c435bbeb0c : SUCCESS in 45m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c4b368a060254759b6bc0f0174f7905c : SUCCESS in 52m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/2437cfc5447c4a17bf65b16ebf223ad8 : FAILURE in 21m 15s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2c02a06e2c12490faaa14c4d4eb64b87 : FAILURE in 11m 35s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/29a5a9aeb7a24a2b88eba87f9ddb9c9e : SUCCESS in 41m 32s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/ca0fd697f9ce463d9b945d293c276bb0 : SUCCESS in 46m 26s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/980ee9e1a3b142f9b8131f318879f480 : SUCCESS in 48m 48s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/3e08e2592e3d483397735da687368265 : SUCCESS in 31m 12s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b6d17f59127d4c70bbc82dbdc58c03cd : SUCCESS in 35m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3530c01d376f84b92f333712fb511fd785501852","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-25 05:02:32.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/899149/1/upstream-check/manila-tempest-plugin-ontap-dhss/e4ab623/ : RETRY_LIMIT in 7m 43s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/899149/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/6dab08e/ : RETRY_LIMIT in 8m 26s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"17c5fba2042cf463db8e8f574efd28fad7bf0b99","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-11-10 00:10:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18359ccdddb959be77a044c3fd1f969d68d5d698","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 00:10:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74eabd62bd328c65572cbc603d6399dfe0c2bef0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 01:29:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ada5cba9a36d47bb9a7412c57cbb0f9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8876470c447c48319977c844a2d4a290 : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82d9e417f4c547dda67884d2ab94099b : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/891a24489eaa4479bf329072c5879804 : SUCCESS in 10m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d6f061dfc2243e3837934e6082f1a17 : SUCCESS in 10m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38ea31a369d749b78e8260daf5f4831d : SUCCESS in 9m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11c04c089213452a9274722d9ae0d5b1 : SUCCESS in 6m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cc0d3deb1cc7439cbdb4750b79578dc5 : SUCCESS in 1h 12m 50s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/77247c6c5d404cc38832865186a93771 : SUCCESS in 22m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3095df2c0dcc4c47b443c54b3ca25967 : SUCCESS in 40m 22s","accounts_in_message":[],"_revision_number":1},{"id":"8d547581386eef63ae60c3fea777ae7e32e2f738","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 01:29:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ff62a0ed4b234216963c73223a7ec6031563f491","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 01:30:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ca704b8ec8841eb892f12db55a9a3f9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07600813954f4dfe906e59e9bd0596dd : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7e8d00922564e848af716fb26578d9d : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"455881d0252e3e2f67ca41879e5871eba6b6df12","revisions":{"455881d0252e3e2f67ca41879e5871eba6b6df12":{"kind":"REWORK","_number":1,"created":"2023-10-25 02:27:23.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/49/899149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/899149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/899149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/899149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/899149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/899149/1"}}},"commit":{"parents":[{"commit":"51b860a01a8b8be4c61cd080efe7befa3d80e36f","subject":"Merge \"NetApp API failed. Reason - 18177:Relationship is in use by SnapMirror in local cluster\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/51b860a01a8b8be4c61cd080efe7befa3d80e36f"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2023-10-02 18:49:15.000000000","tz":480},"committer":{"name":"Rico Lin","email":"ricolin@ricolky.com","date":"2023-10-25 02:27:23.000000000","tz":0},"subject":"Add service_network_host option for port host","message":"Add service_network_host option for port host\n\nNetwork port host can be totally different than manila host,\nSo it\u0027s weird to force binding host to only manila host.\n`service_network_host` allow we specify host for network ports\nwhen `dhss` is True.\n\nChange-Id: I3d9a3bbc81f18d60a0710d9336ad6f97176875be\n(cherry picked from commit 129dc9b28ba776e47b552d7bddb0a63a38745082)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/455881d0252e3e2f67ca41879e5871eba6b6df12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/455881d0252e3e2f67ca41879e5871eba6b6df12"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
