)]}'
{"id":"openstack%2Fmanila~899151","triplet_id":"openstack%2Fmanila~stable%2Fzed~I3d9a3bbc81f18d60a0710d9336ad6f97176875be","project":"openstack/manila","branch":"stable/zed","topic":"bug/2037580","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-11-17 17:53:29.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"12404":{"account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"last_update":"2023-11-17 19:24:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3d9a3bbc81f18d60a0710d9336ad6f97176875be","subject":"Add service_network_host option for port host","status":"MERGED","created":"2023-10-25 02:28:59.000000000","updated":"2023-11-17 19:25:36.000000000","submitted":"2023-11-17 19:24:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899151-bug/2037580","cherry_pick_of_change":899150,"cherry_pick_of_patch_set":2,"meta_rev_id":"afa9deef2e9a69bfffbf8ca051c0b58a0cf3681e","_number":899151,"virtual_id_number":899151,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-17 19:24:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2023-11-17 17:53:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2023-11-17 17:53:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-25 03:33:24.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-10-25 05:15:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-17 17:53:29.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"}],"messages":[{"id":"29719b4ecd9266c383a6e10a0acb90c6cb704e39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-10-25 02:28:59.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.1.","accounts_in_message":[],"_revision_number":1},{"id":"c99e0d5ff260237852be602bef0113b3a884065f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-10-25 02:29:04.000000000","message":"Topic bug/2037580-stable/zed removed","accounts_in_message":[],"_revision_number":1},{"id":"9cd83e7e73af53283cf9a9be708c5b543c89deff","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2023-10-25 02:29:09.000000000","message":"Topic set to bug/2037580","accounts_in_message":[],"_revision_number":1},{"id":"b4bf915225933bad1903359e0e32fe6152e7c430","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-25 03:33:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/51/899151/1/thirdparty-check/pure-devstack-manila-tempest-aio/bd2558d/ : FAILURE in 1h 03m 31s","accounts_in_message":[],"_revision_number":1},{"id":"e72835381d3c56569f3692234c7f7ab0257c7073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 05:15:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/399e0ece296a485f8f8c95c491b79ff2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec602ac3e0984f2f83c8a02c8b97c8e9 : SUCCESS in 11m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b018afa880f340a7a345c2a2a0e80c91 : SUCCESS in 19m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ba91ce13dfe4d559a2cb75958624138 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70f7c5e4c7614c9eaf5f99964d0e3eab : SUCCESS in 9m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e63a02660c424b0b988c3841fbd0e57e : SUCCESS in 10m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88117178715e4a199efc0785b397ebcc : SUCCESS in 10m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50fcc8967b874b659f2a657104bc191d : SUCCESS in 9m 28s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/cbe79a1020a847849cb10b1686f649e1 : SUCCESS in 2h 26m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/dbc4fbce2eab4ed58dec6ea3601d7d08 : FAILURE in 2h 11m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/febcf5d973284589ba06930c1688b5b8 : SUCCESS in 1h 17m 18s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/cd37fbe688e34ba99e43169d98037c79 : FAILURE in 1h 18m 11s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1be426a623534baa9ddbdaad6b2ddd37 : FAILURE in 1h 01m 56s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b593d026a491494ba0fa75c400491510 : SUCCESS in 6m 05s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d8613ac59da242a696564a3d0ffb4904 : SUCCESS in 4m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7d15f9a9e65b42508a8531ec63fd00f2 : SUCCESS in 36m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/967a99167d344d2583a0f83b0ca77b20 : SUCCESS in 42m 21s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/fb5b0d99b8bc49f2b01eb6d6468df8be : SUCCESS in 53m 05s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/e241f0bfc83848bf83e42a14e6d1937a : SUCCESS in 46m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ffdda75c8c5f4b3bb9579b92b28983d3 : SUCCESS in 35m 32s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/2d3492a1f50646e19751d6f255428cf0 : SUCCESS in 55m 08s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/bd87f9d257d04034b00649bfab3fbac6 : SUCCESS in 59m 25s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/622047841a69405bb90e36809e09f858 : FAILURE in 1h 08m 00s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b2643026b6e7470b8d600760b797fc10 : SUCCESS in 26m 48s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c9896e37b4514f699888eaacb18c3835 : SUCCESS in 27m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b72e9303fc598cd7382937509c3a5c45f26c68b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-11-14 21:22:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1be8276d35839ec34ea1a6d8509e08bf3f66e8a5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-11-14 21:22:07.000000000","message":"Topic bug/2037580-stable/zed removed","accounts_in_message":[],"_revision_number":2},{"id":"5a629ef3241876ada8b0f842b26a7d14c89804cc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-11-14 21:22:12.000000000","message":"Topic set to bug/2037580","accounts_in_message":[],"_revision_number":2},{"id":"439c9b1c7617ca42d8f413bf8d532dd0913bc6ac","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-14 22:40:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio http://openstack-logs.purestorage.com/51/899151/2/thirdparty-check/pure-devstack-manila-tempest-aio/0fb6a4b/ : FAILURE in 1h 17m 19s","accounts_in_message":[],"_revision_number":2},{"id":"e32c42f4c6612b274c7d15fb9f5e1a1ed95b2310","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 23:39:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a12d47173e0c4049868da6c258803d02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47400a98977146e2be0c8ad7169cb012 : SUCCESS in 10m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8497b23a60148b49186a14a9fa425cc : SUCCESS in 17m 56s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c667bf88a0a484f86fcaed5cc1727f2 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5806a39635c542afb19090e545e77e19 : SUCCESS in 9m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9aeb8ad7d4643da864af5d80e1895d0 : SUCCESS in 9m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b87977631dd74d31bcd5468e470e24ae : FAILURE in 18m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d741d3686684b92bc391305e5d6c6ff : SUCCESS in 7m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/af53907b073342409f77bb2c0ad701d6 : SUCCESS in 1h 06m 07s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/47ea5bc184754f8191a89e90019614c7 : FAILURE in 2h 11m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d7ff4cb9ae7f4ff3b985d2bfde53c2f9 : SUCCESS in 58m 33s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/710c59117db345a68c308d24c99dd867 : FAILURE in 21m 06s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7d41424aca684602a7f1343bf49c4589 : FAILURE in 11m 40s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/382c8ef83ec249e4802936ce280b89e3 : SUCCESS in 1h 24m 58s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a4948cf55cca444b9215e2cb8daa2ccd : SUCCESS in 4m 49s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e258860e7df24e79b50d167fe2aff8ae : SUCCESS in 5m 14s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d519b50355af4729bf5453c97a90496c : SUCCESS in 26m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/70f5515ec8954029a8b602eb062e66c8 : SUCCESS in 51m 46s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ed5d4d2b23d04fe984cd3e9c5f83adc4 : SUCCESS in 51m 51s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/9d467e64b8d24f6cb3c305cedc02ce51 : SUCCESS in 37m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/84b273838bb4402687de03a61a924f98 : SUCCESS in 24m 31s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/4dd8f85d7dd048cf990c8fc489aac3bf : SUCCESS in 59m 49s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/03bcdc9ad6e8444b8ebfb7a518d0210e : SUCCESS in 1h 06m 02s (non-voting)\n- manila-grenade-skip-level https://zuul.opendev.org/t/openstack/build/8bb1577cbf61465e8a6bfc02e1c936e0 : FAILURE in 1h 03m 29s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/a9c178e002084f09bb65d8ed479ab4a7 : SUCCESS in 42m 24s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b56fd07ef7dd4dd38600a2c109b6b37a : SUCCESS in 32m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"32e24e8e4a8cee01e07100d9133d2b0c834435c5","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-11-17 17:53:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7131e25aabffebf6ea2e2e35c35f9abebe11694a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 17:54:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d66121db67bc259c268e08748e851e5f0278a728","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 19:24:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e4f6028c37e439f86305ec21cec30fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/960f90bf6abc412b9b5a97aadc08dd45 : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1cf87ca010746f68b86af2911e2bffd : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ac97ed564bc43d7bdf4fa65666137ff : SUCCESS in 10m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed66acf2a119493a9612a6d6062b1bd6 : SUCCESS in 8m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e34f1cb444014fe598eef66f916dca3e : SUCCESS in 6m 51s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/78352b7a4e4b424b9c883a4f0736646a : SUCCESS in 1h 29m 53s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9883eaf6de464cb7a8c864f1d0f0eae8 : SUCCESS in 31m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d9f1dbc27a644e40a1138f7b34ce4829 : SUCCESS in 36m 54s","accounts_in_message":[],"_revision_number":2},{"id":"6e9cd3023269a687e868e78a480888d94c85e2bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 19:24:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"afa9deef2e9a69bfffbf8ca051c0b58a0cf3681e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-17 19:25:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/060164a85360415e94dcb24bbaf3f194\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fef10f7b7b64418a3683884e928431b : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5872e43830e34995a299f8814d96d8c3 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5034acd8284d3208b9ae2a0147f5c410fe74cd79","revisions":{"6f91f015280f155464502de6e0e03aeccc7a2223":{"kind":"REWORK","_number":1,"created":"2023-10-25 02:28:59.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/51/899151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/51/899151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/899151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/899151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/899151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/51/899151/1"}}},"commit":{"parents":[{"commit":"bfd0a6b5a0ae02b1880fff892bb553752c324de6","subject":"Merge \"[RBAC] Enforce check for share updates\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bfd0a6b5a0ae02b1880fff892bb553752c324de6"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2023-10-02 18:49:15.000000000","tz":480},"committer":{"name":"Rico Lin","email":"ricolin@ricolky.com","date":"2023-10-25 02:28:59.000000000","tz":0},"subject":"Add service_network_host option for port host","message":"Add service_network_host option for port host\n\nNetwork port host can be totally different than manila host,\nSo it\u0027s weird to force binding host to only manila host.\n`service_network_host` allow we specify host for network ports\nwhen `dhss` is True.\n\nChange-Id: I3d9a3bbc81f18d60a0710d9336ad6f97176875be\n(cherry picked from commit 129dc9b28ba776e47b552d7bddb0a63a38745082)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6f91f015280f155464502de6e0e03aeccc7a2223"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6f91f015280f155464502de6e0e03aeccc7a2223"}]},"branch":"refs/heads/stable/zed"},"5034acd8284d3208b9ae2a0147f5c410fe74cd79":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-11-14 21:22:01.000000000","uploader":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"ref":"refs/changes/51/899151/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/51/899151/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/899151/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/899151/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/51/899151/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/51/899151/2"}}},"commit":{"parents":[{"commit":"bfd0a6b5a0ae02b1880fff892bb553752c324de6","subject":"Merge \"[RBAC] Enforce check for share updates\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bfd0a6b5a0ae02b1880fff892bb553752c324de6"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2023-10-02 18:49:15.000000000","tz":480},"committer":{"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","date":"2023-11-14 21:22:01.000000000","tz":0},"subject":"Add service_network_host option for port host","message":"Add service_network_host option for port host\n\nNetwork port host can be totally different than manila host,\nSo it\u0027s weird to force binding host to only manila host.\n`service_network_host` allow we specify host for network ports\nwhen `dhss` is True.\n\nChange-Id: I3d9a3bbc81f18d60a0710d9336ad6f97176875be\n(cherry picked from commit 129dc9b28ba776e47b552d7bddb0a63a38745082)\n(cherry picked from commit 455881d0252e3e2f67ca41879e5871eba6b6df12)\n(cherry picked from commit 07c584cf7ef2339d25eb771895df688b0d1693dc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5034acd8284d3208b9ae2a0147f5c410fe74cd79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5034acd8284d3208b9ae2a0147f5c410fe74cd79"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
