)]}'
{"id":"openstack%2Fmanila~899295","triplet_id":"openstack%2Fmanila~master~I0af426254be886e31553275feb23b9ec90ec2cc8","project":"openstack/manila","branch":"master","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-09-04 07:21:08.000000000","reason":"Change was submitted"},"33728":{"account":{"_account_id":33728,"name":"Ashley Rodriguez","email":"ashrodri@redhat.com"},"last_update":"2024-09-04 07:21:08.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-09-04 06:23:22.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2024-09-03 09:05:32.000000000","reason":"\u003cGERRIT_ACCOUNT_30407\u003e replied on the change","reason_account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2024-08-31 05:40:32.000000000","reason":"\u003cGERRIT_ACCOUNT_32919\u003e replied on the change","reason_account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"}}},"hashtags":[],"change_id":"I0af426254be886e31553275feb23b9ec90ec2cc8","subject":"Add json and yaml formatting to manila-manage","status":"MERGED","created":"2023-10-25 18:34:17.000000000","updated":"2024-09-04 07:22:18.000000000","submitted":"2024-09-04 07:21:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899295","meta_rev_id":"14880ab54c425e507297f1c60411d0ffc3fb953f","_number":899295,"virtual_id_number":899295,"owner":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-04 07:21:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2024-09-04 06:23:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2024-08-31 05:40:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":2,"date":"2024-09-03 09:05:32.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":1,"date":"2024-09-04 06:23:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33728,"name":"Ashley Rodriguez","email":"ashrodri@redhat.com"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":33728,"name":"Ashley Rodriguez","email":"ashrodri@redhat.com"}],"CC":[{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-25 19:05:49.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-25 19:47:37.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-10-25 20:24:19.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-10-25 21:02:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-25 21:33:58.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-12-13 20:14:31.000000000","updated_by":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-12-15 21:10:18.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-12-15 23:02:42.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-01-16 13:44:53.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2024-01-16 13:44:53.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2024-01-16 13:44:53.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-01-16 13:44:53.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":33728,"name":"Ashley Rodriguez","email":"ashrodri@redhat.com"},"state":"REVIEWER"}],"messages":[{"id":"f3e14f8df8abcfb6bafa1ac8fd7f7b0268fa4679","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-10-25 18:34:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2cf037aefa6a2d7f89404aaa59cbbeb7174e2a27","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-25 19:05:49.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/95/899295/1/check/DellEMC_PowerFlex_Manila_dhss_false/dbde1f8/DellEMC_PowerFlex_Manila_dhss_false/594 : SUCCESS in 31m 16s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"dc4b6f50b8415a2116713584f6cd1a24081b5303","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-25 19:47:37.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/95/899295/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/532500e/ : SUCCESS in 1h 12m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f5432eb90a3669a775aefc91b986a8e3d6f6bde1","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-25 20:24:19.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-95-899295-1/NFS/ : SUCCESS in 1h 49m 42s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"e1071aff70fafcad006a6f64bb3e4acf1e8221db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 21:02:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d2d12b9bcba4626af3dc41d151356d7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42bc42ac763e4e63b2f3df69944e36ce : SUCCESS in 10m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4ea0a0c7f4641d28d558c1aac7cb359 : FAILURE in 25m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5694c60b8022499588c0eaa724e5a8ba : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/256db0d297e641ae8497587d323f8b3a : FAILURE in 10m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c63af5cd8bf4974bfbcada76a89fe0d : FAILURE in 7m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/51b595ee32364ca5adf0b93f0e873951 : SUCCESS in 45m 13s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f806aaea37ab4fdf846bc61154f466e3 : SUCCESS in 2h 22m 00s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a2f9943e74934cd5a24ca4f81f10084d : SUCCESS in 1h 26m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e2a5792294264696ab138dcb37c79800 : SUCCESS in 49m 14s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b3b9c8b722da416b87eda1efbb866e30 : SUCCESS in 1h 54m 55s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c7be63bbf6b0457c835b520825b951bd : FAILURE in 1h 13m 10s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4a718d651b704712abd7fa343f0f1589 : SUCCESS in 4m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c3f1513216ca4eb89520950393971d65 : SUCCESS in 4m 22s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6fdb7190c18d4d6fb1e792f41b147c35 : SUCCESS in 35m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5574cab7abdd47d2926d7200f3a04cf6 : SUCCESS in 49m 16s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3a1baf9fe931483180a4f0954180f1be : FAILURE in 54m 55s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/0efa94aae4b743d597d9e73c513c73f3 : FAILURE in 22m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/080424e50334459b86f678e4293353c2 : FAILURE in 23m 00s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/70392ea347c6410682dde6b7a5b0765f : SUCCESS in 56m 17s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/ef4e76c10ed9432586abd99874b42528 : SUCCESS in 58m 32s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2117f82bf18245eb9bd3fdfb3f519516 : SUCCESS in 42m 30s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/df20a02e5f5d455688bf57891fd63792 : SUCCESS in 43m 48s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/0ef42bd592604ce691c53b6b402be12e : SUCCESS in 47m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d7b8ed5a183ab483c0355b974af48f3a834ba20","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-25 21:33:58.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/95/899295/1/check/DellEMC_Unity_Manila_dhss_true/8d633be/DellEMC_Unity_Manila_dhss_true/696 : SUCCESS in 2h 59m 25s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/95/899295/1/check/DellEMC_Unity_Manila_dhss_false/b93df73/DellEMC_Unity_Manila_dhss_false/734 : SUCCESS in 1h 01m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d2744554579d1a58c967559db5b11418580e21d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-12-13 20:12:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"22365f518a767912849557a0bbbde465d10ae1d9","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-12-13 20:14:31.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-95-899295-2/ns5_manila : FAILURE in 1m 1s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":2},{"id":"81ae72042190159aaf99346e19649eaebc80b836","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-12-13 20:20:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6a441b62c3a61f6e85687c84053f7d866c59ac6b","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-12-13 20:36:59.000000000","message":"Patch Set 3:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-95-899295-3/ns5_manila : FAILURE in 1m 4s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":3},{"id":"56b86589f2a0d4296bb251dac66fd802686ef26d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-12-13 20:48:29.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/95/899295/3/check/DellEMC_PowerFlex_Manila_dhss_false/4aa7836/DellEMC_PowerFlex_Manila_dhss_false/689 : SUCCESS in 27m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"d5206f13c3d18d934dede76f06552d04f8b47603","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-13 20:49:45.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/95/899295/3/check/DellEMC_Unity_Manila_dhss_true/4a6f129/DellEMC_Unity_Manila_dhss_true/755 : FAILURE in 28m 29s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/95/899295/3/check/DellEMC_Unity_Manila_dhss_false/b0eb700/DellEMC_Unity_Manila_dhss_false/807 : FAILURE in 18m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"3e24cc0f23526b76758554fa18160fc00b3d8271","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-12-13 21:27:36.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-95-899295-2/NFS/ : SUCCESS in 1h 14m 25s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"ace2cd1c25530c3e343a70ed88e8e9cb18f3d080","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-12-13 21:33:50.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/95/899295/3/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/e659fdd/ : FAILURE in 1h 10m 51s","accounts_in_message":[],"_revision_number":3},{"id":"58c03c7257e643c0c01474cccdcb9f9281eb94ab","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-12-13 21:58:39.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-95-899295-3/NFS/ : SUCCESS in 1h 15m 24s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"ffb3b07c69fe0e4e00446160c8321b6a752b492a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-13 23:32:07.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/44671ba923f14f47b305c1a15da30eb9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4650cfa49d854c4791025acfda64d155 : SUCCESS in 10m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75fcbd81527b43deb898fca9aab1b278 : FAILURE in 26m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbcd79ce8b2f476c97b5beaa3bcb26b2 : FAILURE in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/498a2784a4ca45fa95bb3ef29861e653 : SUCCESS in 10m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9bbd6da6ebdc47f89765307f8e186eca : SUCCESS in 10m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3252c6eb36e6448a84c5f2ec3922c1f4 : SUCCESS in 8m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7539c19268e041b580f3db9d01519121 : SUCCESS in 1h 14m 15s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b5aac98ceb524e428a272889269ffe63 : SUCCESS in 1h 48m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/62b856c89d1a42488a27dbc6b893d982 : TIMED_OUT in 3h 03m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e90d5a8fbd4b45d294e78396083ba782 : SUCCESS in 1h 07m 42s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4b6af0e4b15748b79e96b12568067c30 : SUCCESS in 1h 05m 41s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e6bc06971e1a43b98c7a3c993a556e9c : FAILURE in 55m 51s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b3a1876fe55240a7875e3d99ef2707c5 : SUCCESS in 4m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d626f8e033d443de81b26b2b592f64a9 : SUCCESS in 4m 45s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a3b3c9490437458a864a2f2fbedb7b1c : SUCCESS in 27m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7981a6b597174c39b47a1d185631119e : SUCCESS in 39m 30s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4cfc69863cd74854a8ce80f12d2a3382 : SUCCESS in 49m 54s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/50065676cf0c4569b3516d413c513805 : FAILURE in 11m 04s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f43380a9b2534b2d8f9f3aa3a0944303 : FAILURE in 21m 57s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/52d9fc27ac1e49f5b8845c077a90512d : SUCCESS in 53m 02s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/175b59d3df904b1680e86a3640052385 : SUCCESS in 49m 25s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/29a75e6e4253458c8541c99c2652228c : SUCCESS in 42m 24s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/a2bf61413c91490f8efd1617269ee00c : SUCCESS in 42m 51s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/912bf411f93a4433a04b1c2a630c6cd8 : SUCCESS in 47m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d9e9dc8f4764f12f650791b10765279a42ebb462","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-12-15 18:22:37.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"91a4fde903a520e0d0444f68670740956516418d","author":{"_account_id":15386,"name":"Nexenta CI","email":"openstackci@nexenta.com","username":"nexentaci","tags":["SERVICE_USER"]},"date":"2023-12-15 18:24:22.000000000","message":"Patch Set 4:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* NexentaStor5_Manila_NFS http://104.254.242.106/refs-changes-95-899295-4/ns5_manila : FAILURE in 1m 6s\n\nLeave a comment with run-nexentaci to trigger a recheck. For more information about Nexenta CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Nexenta_CI","accounts_in_message":[],"_revision_number":4},{"id":"d09eed9a1c822a6259aa512ebf12111cdab204a1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-12-15 18:34:12.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/95/899295/4/check/DellEMC_PowerFlex_Manila_dhss_false/3f40d44/DellEMC_PowerFlex_Manila_dhss_false/693 : FAILURE in 11m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"284ea1a1beb177b3745c2323c807d876e030677b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-15 18:54:34.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/95/899295/4/check/DellEMC_Unity_Manila_dhss_true/9c073a7/DellEMC_Unity_Manila_dhss_true/759 : FAILURE in 31m 40s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/95/899295/4/check/DellEMC_Unity_Manila_dhss_false/fb3de98/DellEMC_Unity_Manila_dhss_false/811 : FAILURE in 19m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"15a49bfac0e7600e0a5d3e44f610673b8be94e77","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-12-15 19:41:55.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-95-899295-4/NFS/ : SUCCESS in 1h 18m 57s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"5e9f6e1f82c6be9779db0b3ce3404e93979090be","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-12-15 19:45:14.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/95/899295/4/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/201771b/ : FAILURE in 1h 19m 47s","accounts_in_message":[],"_revision_number":4},{"id":"b0c6f4dfa6676b335915f3c0420e321169ddae1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-15 20:42:56.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8ddd654f2ac4f9c9f448e1d5e1743a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd4ff4e66566421ab9330c161416bacc : SUCCESS in 8m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6515d2036be4d00a1ee985552eee56a : SUCCESS in 22m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/818719920edb411382591e422a5f07e5 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8ea96f7f4804ff2b06877fc93149713 : SUCCESS in 8m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3b16769d81e4f468b2c336ebd339a7c : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b74588a384a04cbebc6870fbd1fe8cf7 : SUCCESS in 5m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0ed10b1f2a8d4a4ea221107688ac0d40 : SUCCESS in 53m 38s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ce27c5326e8341f6acfb87b7db1d4673 : SUCCESS in 2h 14m 21s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/af913e682824457fa19d2b7ca5b0a06b : SUCCESS in 1h 19m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a6a8d5a48e4a45e6b25054fdf6cfac2d : SUCCESS in 1h 11m 24s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/26003f76fe4545f2b7dba5a947324a6b : SUCCESS in 1h 35m 53s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/44b82961ef0a4c0dbd8ca1002e7827de : FAILURE in 20m 09s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d439bbfe7f2b460b8c9ab86b407eefa5 : SUCCESS in 4m 33s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cb85bba67d5e42b7957164a99c188e6a : SUCCESS in 4m 41s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/571c5e001d834aada21b18099b6401fc : SUCCESS in 39m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b461139a5e964f57a07ca5ca7c91942d : SUCCESS in 39m 04s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9785ce07334b4587aa6e43819b3b0607 : SUCCESS in 37m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/d2b9d377d9ee451288cc772b45215035 : FAILURE in 23m 54s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/ff537aea01ad4e0088b56364b8b040d9 : FAILURE in 23m 21s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/9dbfb5f3c04c42eb85ad6d660c4ba105 : SUCCESS in 1h 26m 14s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/2e05c7f75cf14142b688405d6232b9e9 : SUCCESS in 1h 07m 41s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a32bd89f3e404292a196d27383096c22 : SUCCESS in 1h 06m 02s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/961f2c4e1bde4db79e36809aa14896df : SUCCESS in 28m 08s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/97a2d2128db34b3da56d89b37029f652 : SUCCESS in 33m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"91aa81cffe770931dfc76fd58572ccfafb152ad1","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-12-15 21:10:18.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/95/899295/4/check/DellEMC_PowerStore_Manila_dhss_false/ca85001/DellEMC_PowerStore_Manila_dhss_false/291 : FAILURE in 27m 07s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"3271238261865a411a1125337be1cb435955c5b6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-12-15 21:14:04.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/95/899295/4/check/DellEMC_PowerFlex_Manila_dhss_false/f1f99de/DellEMC_PowerFlex_Manila_dhss_false/694 : SUCCESS in 30m 52s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"28801ab44ecf70a69afbfb9f31c9a072234e63d1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-12-15 21:15:26.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/95/899295/4/check/DellEMC_Unity_Manila_dhss_true/711b9fe/DellEMC_Unity_Manila_dhss_true/760 : FAILURE in 32m 14s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/95/899295/4/check/DellEMC_Unity_Manila_dhss_false/c1c604f/DellEMC_Unity_Manila_dhss_false/812 : FAILURE in 19m 05s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"36065f146a02f2bbfbb625bd5bb0df2dcf0cae84","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-12-15 23:02:42.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/95/899295/4/upstream-check/manila-tempest-plugin-ontap-dhss/5155b66/ : FAILURE in 2h 11m 33s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/95/899295/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/63dfde7/ : RETRY_LIMIT in 12m 26s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"58c00a928592884491d29c9efe547be4e902b686","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-01-16 13:54:34.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b84cbdb4cd353bcafff3c8a6b4c4be08a8123c2b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-01-26 22:01:40.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"97cd49a3f2a4d8912fc4aa9c963cef6fc1bbd76a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-01-27 00:24:05.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"015095b4cf020e4ac9d4bcd6c008d992c6902b7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-01-29 23:11:16.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"5da6877cfe4a418efd731a646f7da7bb475c12f9","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-01-29 23:11:58.000000000","message":"Patch Set 4: Workflow-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"7e3754b5ce715c5234971d1f2fcb8dc497a74163","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-01-29 23:12:03.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"f20dedb69e9a1012f1ef071414617000d6d9fb4a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-01-30 01:41:40.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-95-899295-5/NFS/ : SUCCESS in 1h 22m 8s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"ad5cd8b9fcd3f76a2cecdad6ccc31b2f2bee5af7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 02:18:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/73c49526077b444b8a56d4eaf3005e9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8bb21df7ce4467db82e008eb0c0be68 : SUCCESS in 12m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f513ae5c0d34431b6e5cb39baca6c06 : FAILURE in 24m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddb9a3aa47304fd69352cf5823c228f0 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d5acb99dff544368cb16e5e657f0b21 : FAILURE in 9m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3238c94b77d741fc881cdbafe419d721 : FAILURE in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acdc737b77eb4223930e5cb87860ffd8 : SUCCESS in 6m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ef891dfce4484cbca5c0b7d6cc66aaca : SUCCESS in 1h 10m 39s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b6f9369a805d4160b87738f4f9359b7e : FAILURE in 2h 59m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/65497f07b9f9441ba2ca23e3cd7826ba : SUCCESS in 57m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/69ff34bd8b8248ccb130d9fd9268c280 : SUCCESS in 54m 08s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/aa0998259b9f4026acb341decb5900d2 : SUCCESS in 1h 16m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/492cf9448eff4cbeb5f08e28784f5fd7 : SUCCESS in 1h 03m 06s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a3b50169c93747f18f819522f27c5861 : SUCCESS in 4m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ebc556fedc0045df943c932516051d1b : FAILURE in 4m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7049b5aabe1e4a9186ba580e7da0224f : SUCCESS in 27m 12s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7c5f89a559c74240a908bc0eee9b3f0f : SUCCESS in 42m 24s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/95da570c739444dfb0d7ee689d14f6c7 : FAILURE in 54m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/aa8a7e803d5c45ca9f89dccf93b818be : FAILURE in 10m 37s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b7db21f528e94c4ebe2a931a713733d2 : FAILURE in 21m 27s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/b1b121842d7044cda40e875c50a5a979 : SUCCESS in 54m 48s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/e2af19ff1b04482ca2731c4365044382 : SUCCESS in 1h 02m 29s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0af93b82dadb4a60bc228228651216ef : SUCCESS in 1h 04m 29s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/32a02cd42e6644998c1f793902021036 : SUCCESS in 24m 15s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/a952e4d6ecbd4d59a7abadd8a6bf212d : SUCCESS in 42m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"84d6ac37020c284f4c4397e998d11a4e9dc8303f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-08-30 16:41:45.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":6},{"id":"2289dc6490ed6e0d88ea8369822cf909e765aac2","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-08-30 16:43:02.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"1cbc9693dd22ac60fa4079ec3d75c9ab9cb213f0","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-30 17:10:20.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/d9700667b78142caa10b05efd3fb8640\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/68e2b30b4db34717a02b4baa43277335 : FAILURE in 28m 04s","accounts_in_message":[],"_revision_number":6},{"id":"0121bb84da1e247ab47b0fd2cd1a2b0bee8bd49c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-30 17:14:02.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/95/899295/6/check/DellEMC_PowerFlex_Manila_dhss_false/53b16e1/DellEMC_PowerFlex_Manila_dhss_false/1598 : SUCCESS in 32m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"0a3a73d80fd34a1e824846d2ce8abf50b36856af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 18:54:38.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6d7143a9b1c4bd480b9c6b30fa13faa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7090268dc4624f7caee4d36785b45433 : SUCCESS in 9m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ebd443176eac430c8a3276912ab0b468 : SUCCESS in 26m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b2cf2bc2f6b4ff48578ed9628bb611a : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efbaa6130e764530a7fc33cd0325a07a : SUCCESS in 11m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/69e3e3e29fec4d57bc0f0823fdb38ed6 : SUCCESS in 8m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac849d061f484025a9a6f99c72437615 : SUCCESS in 10m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/410561048e8d4ffcbdf38c6dd37f9b83 : SUCCESS in 10m 49s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2dd2ecc7aae24576929182ceb606fd56 : FAILURE in 19m 05s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2ab354a1cf9c445fb2bfa2882d9a5e1e : SUCCESS in 2h 10m 03s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b75434e79aeb47eebf9ada1a64613e8a : FAILURE in 57m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/1bacb3e21314494db6a57f3189d32224 : SUCCESS in 1h 04m 32s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/74788a0f7aee4859bcfcbbd41fce7961 : SUCCESS in 58m 02s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ec5fa994b4a341f4b07e028df7831bc6 : SUCCESS in 1h 04m 00s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/73bfd82b98a541e2b5a92f7e88b6d991 : SUCCESS in 4m 15s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/49463431a20545f3af832d89951f2879 : SUCCESS in 4m 20s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/62ad73db00aa40f482915200a13a12ee : SUCCESS in 28m 19s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1f0b30e2f452410186a2c7147dba0058 : SUCCESS in 43m 12s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6cb0114a35a24c4ca50922913053d853 : FAILURE in 43m 55s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/5033c4becc3747a2b792267db42bd937 : FAILURE in 19m 48s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e6edbeab788d40e089aa3ddb78d0df92 : SUCCESS in 58m 02s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b5bc321e3ffd44cb878fe995bbba4c50 : SUCCESS in 36m 12s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/124152883a004d45a628a67c6c404ed6 : SUCCESS in 39m 52s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/598fb05f8b0848ce9e3d9f9bb904039d : FAILURE in 4m 38s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/69d162aef1d94314a3f1f21d98f18d0d : SUCCESS in 7m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4a7695d7a5dc397716fd4e9eb156c8a4e96dd121","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-08-30 21:50:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ab9cb18c02cdb400c6591c555c261190ea30ce29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-31 00:53:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e7063b8d0cb41ec86be75e6e35df00c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c16f6f870084e6bb9cad2663295a45b : SUCCESS in 9m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/088dc7eaee8e44079a87e6df98f93499 : SUCCESS in 24m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01ee0417691b44d88641bac3cc1fc338 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20093f2e242345fe9a69073581f371ef : SUCCESS in 8m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe927ff9b4aa4b3daa2339bb1e100757 : SUCCESS in 7m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/323f169777ce4cf6bde0714311174a16 : SUCCESS in 10m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c11c9f3ee113414e88dadd161ffa410d : SUCCESS in 8m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/27af1a3d26324273b31ba5ac27f5c401 : SUCCESS in 1h 04m 37s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/86058b490e3f467198f7bb88a81f2018 : TIMED_OUT in 3h 02m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/96ded248e51345d7bc627a61e0a09438 : FAILURE in 56m 45s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/cd5cc7540915459b8bc17f41745ea863 : SUCCESS in 1h 01m 44s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/99aab95cca0e46c3a41a67aabedbfb4c : SUCCESS in 57m 14s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8a2d884fb3c94e128e156ff29299fe48 : FAILURE in 1h 02m 37s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/0485f38d97b7484ba03450cf00a62530 : SUCCESS in 3m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f0600923e03545e3b440a65a9e7e64a5 : SUCCESS in 3m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/af0df4b1eb404ecfa9b16d1690427b8c : SUCCESS in 35m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f5b0e646fd6941f18d07411b02857044 : SUCCESS in 42m 05s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d45ad23452584bd9a08c7d732f32696e : FAILURE in 48m 02s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/2bc5cf55eaab4c68b141625bb8a2cde5 : FAILURE in 15m 00s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/30c29eaf14db4bb1a28b11ce32639d10 : SUCCESS in 44m 41s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/82af29e847ca4e3e8f461c3c7dac5217 : SUCCESS in 21m 42s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7b7980f245cc4e1e9f4a04ce4cfcaf22 : SUCCESS in 25m 29s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/5bfaac85ebf64bd887ed82e80c921a91 : FAILURE in 4m 10s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/60ef302edddb48afb338526e7324168f : SUCCESS in 6m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1bf3f1b21bcbf99a25b3629b9f2dd9b6a598e37f","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-31 01:10:29.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/95/899295/6/upstream-check/manila-tempest-plugin-ontap-no-dhss/433da0a/ : RETRY_LIMIT in 11m 51s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"a21183d97b651b026d5f98fd6135e5d5f6c14b9f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-31 01:25:31.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/95/899295/6/check/DellEMC_PowerFlex_Manila_dhss_false/5871e33/DellEMC_PowerFlex_Manila_dhss_false/1604 : SUCCESS in 31m 50s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"4fec0ff8264ed235c12bf8850e141a62691eef24","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-31 01:28:00.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/95/899295/6/check/DellEMC_PowerStore_Manila_dhss_false/5069307/DellEMC_PowerStore_Manila_dhss_false/466 : SUCCESS in 34m 21s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"a3a6238f1c87f120cbcc31be94660564c4ba46f5","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-08-31 05:40:32.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"7b64a9f02bcac4c1237f7f3401eab8cd682dd184","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2024-09-03 09:05:32.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"bf7d57c9594d83f3327c424e59cccb09a6739c14","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-04 06:23:22.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"980ec34de0f577bfb51b8c2de85bc193340c176f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 06:23:52.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"fdfe7ab4411f822b468245de2a87e549a1f125e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 07:21:08.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25b1951871e841e4b1fc6e6867914a7b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3568d8896b8b4133884688b5f6e9051a : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7283c2ae84004da7acdbf8d44e2c2fae : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21a6515c15074cf99d002d5fdbd7a5b1 : SUCCESS in 10m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/837bc14e1e254ef3a763b17e4a20d976 : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cd8a877507b43218d223d6e14fc8b45 : SUCCESS in 8m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/eb90845b919248fea175e92096fccb81 : SUCCESS in 56m 10s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/57e8ebb3bdbf4a8aabd0e8d4a817b90f : SUCCESS in 42m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/194ad36a78e94e19a153fbec904b7570 : SUCCESS in 31m 56s","accounts_in_message":[],"_revision_number":6},{"id":"f982589a0258341c9cab1071cd8346a64d87f051","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 07:21:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"14880ab54c425e507297f1c60411d0ffc3fb953f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 07:22:18.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ecbb9a981164b07b7ba2f98da66d705\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/659aa7f52b4a4f209adbd44c4b52f063 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cb777847a224143842e26886c40c5c2 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"861a67193b5965149a83e66052980779458cecfc","revisions":{"d29c73111ca76f070f6aa6eabf879cc0bcd5647a":{"kind":"REWORK","_number":1,"created":"2023-10-25 18:34:17.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/95/899295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/95/899295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/95/899295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/95/899295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/95/899295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/95/899295/1"}}},"commit":{"parents":[{"commit":"ce5b5f329dd96a0b1c3ae7e39ab3942cec8be47d","subject":"Merge \"[Doc] Add a quick and dirty IRC guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ce5b5f329dd96a0b1c3ae7e39ab3942cec8be47d"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-10-25 18:33:34.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-10-25 18:33:34.000000000","tz":-180},"subject":"[WIP] Add json and yaml formatting to manila-manage","message":"[WIP] Add json and yaml formatting to manila-manage\n\nChange-Id: I0af426254be886e31553275feb23b9ec90ec2cc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d29c73111ca76f070f6aa6eabf879cc0bcd5647a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d29c73111ca76f070f6aa6eabf879cc0bcd5647a"}]},"branch":"refs/heads/master"},"c6c30d0fbecbb129f4d932c01e513dc86b3ab04c":{"kind":"REWORK","_number":2,"created":"2023-12-13 20:12:52.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/95/899295/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/95/899295/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/95/899295/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/95/899295/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/95/899295/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/95/899295/2"}}},"commit":{"parents":[{"commit":"ce5b5f329dd96a0b1c3ae7e39ab3942cec8be47d","subject":"Merge \"[Doc] Add a quick and dirty IRC guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ce5b5f329dd96a0b1c3ae7e39ab3942cec8be47d"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-10-25 18:33:34.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-12-13 20:09:49.000000000","tz":-180},"subject":"Add json and yaml formatting to manila-manage","message":"Add json and yaml formatting to manila-manage\n\nAdd the possibility to specify how the data will be displayed\nwhile issuing the manila-manage command. This can be helpful to\nautomations,\n\nChange-Id: I0af426254be886e31553275feb23b9ec90ec2cc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c6c30d0fbecbb129f4d932c01e513dc86b3ab04c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c6c30d0fbecbb129f4d932c01e513dc86b3ab04c"}]},"branch":"refs/heads/master"},"23a10e938870fb57ce62b00c1254d1a16ee2de5f":{"kind":"REWORK","_number":3,"created":"2023-12-13 20:20:56.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/95/899295/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/95/899295/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/95/899295/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/95/899295/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/95/899295/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/95/899295/3"}}},"commit":{"parents":[{"commit":"ce5b5f329dd96a0b1c3ae7e39ab3942cec8be47d","subject":"Merge \"[Doc] Add a quick and dirty IRC guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ce5b5f329dd96a0b1c3ae7e39ab3942cec8be47d"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-10-25 18:33:34.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-12-13 20:20:49.000000000","tz":-180},"subject":"Add json and yaml formatting to manila-manage","message":"Add json and yaml formatting to manila-manage\n\nAdd the possibility to specify how the data will be displayed\nwhile issuing the manila-manage command. This can be helpful to\nautomations,\n\nChange-Id: I0af426254be886e31553275feb23b9ec90ec2cc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/23a10e938870fb57ce62b00c1254d1a16ee2de5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/23a10e938870fb57ce62b00c1254d1a16ee2de5f"}]},"branch":"refs/heads/master"},"a52bdc1cc5ca0b8d572deda0a5cd60c6eb3409c5":{"kind":"REWORK","_number":4,"created":"2023-12-15 18:22:37.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/95/899295/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/95/899295/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/95/899295/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/95/899295/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/95/899295/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/95/899295/4"}}},"commit":{"parents":[{"commit":"ce5b5f329dd96a0b1c3ae7e39ab3942cec8be47d","subject":"Merge \"[Doc] Add a quick and dirty IRC guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ce5b5f329dd96a0b1c3ae7e39ab3942cec8be47d"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-10-25 18:33:34.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-12-15 18:22:28.000000000","tz":-180},"subject":"Add json and yaml formatting to manila-manage","message":"Add json and yaml formatting to manila-manage\n\nAdd the possibility to specify how the data will be displayed\nwhile issuing the manila-manage command. This can be helpful to\nautomations,\n\nChange-Id: I0af426254be886e31553275feb23b9ec90ec2cc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a52bdc1cc5ca0b8d572deda0a5cd60c6eb3409c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a52bdc1cc5ca0b8d572deda0a5cd60c6eb3409c5"}]},"branch":"refs/heads/master"},"ebe659c713253ea3699910839319fca3fb20040e":{"kind":"REWORK","_number":5,"created":"2024-01-29 23:11:16.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/95/899295/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/95/899295/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/95/899295/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/95/899295/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/95/899295/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/95/899295/5"}}},"commit":{"parents":[{"commit":"ce5b5f329dd96a0b1c3ae7e39ab3942cec8be47d","subject":"Merge \"[Doc] Add a quick and dirty IRC guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ce5b5f329dd96a0b1c3ae7e39ab3942cec8be47d"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-10-25 18:33:34.000000000","tz":-180},"committer":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2024-01-29 23:11:07.000000000","tz":-180},"subject":"Add json and yaml formatting to manila-manage","message":"Add json and yaml formatting to manila-manage\n\nAdd the possibility to specify how the data will be displayed\nwhile issuing the manila-manage command. This can be helpful to\nautomations,\n\nChange-Id: I0af426254be886e31553275feb23b9ec90ec2cc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ebe659c713253ea3699910839319fca3fb20040e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ebe659c713253ea3699910839319fca3fb20040e"}]},"branch":"refs/heads/master"},"861a67193b5965149a83e66052980779458cecfc":{"kind":"REWORK","_number":6,"created":"2024-08-30 16:41:45.000000000","uploader":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"ref":"refs/changes/95/899295/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/95/899295/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/95/899295/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/95/899295/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/95/899295/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/95/899295/6"}}},"commit":{"parents":[{"commit":"4bf505404ae77a230f79841797e0ee41d2ada414","subject":"Merge \"api: Add schemas for messages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4bf505404ae77a230f79841797e0ee41d2ada414"}]}],"author":{"name":"silvacarloss","email":"ces.eduardo98@gmail.com","date":"2023-10-25 18:33:34.000000000","tz":-180},"committer":{"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","date":"2024-08-30 16:41:35.000000000","tz":-180},"subject":"Add json and yaml formatting to manila-manage","message":"Add json and yaml formatting to manila-manage\n\nAdd the possibility to specify how the data will be displayed\nwhile issuing the manila-manage command. This can be helpful to\nautomations,\n\nChange-Id: I0af426254be886e31553275feb23b9ec90ec2cc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/861a67193b5965149a83e66052980779458cecfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/861a67193b5965149a83e66052980779458cecfc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
