)]}'
{"id":"openstack%2Fmanila~899320","triplet_id":"openstack%2Fmanila~master~I629669476e585a834673b8c8b49ad4b0270b877f","project":"openstack/manila","branch":"master","topic":"bug/2015328","attention_set":{},"removed_from_attention_set":{"33728":{"account":{"_account_id":33728,"name":"Ashley Rodriguez","email":"ashrodri@redhat.com"},"last_update":"2024-03-13 14:08:49.000000000","reason":"Change was submitted"},"36480":{"account":{"_account_id":36480,"name":"Gray Lutalo","display_name":"gray","email":"jogareko@gmail.com","username":"gray"},"last_update":"2024-03-13 14:08:49.000000000","reason":"Change was submitted"},"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-03-13 12:59:18.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"36850":{"account":{"_account_id":36850,"name":"Solomon Barine Akpuru","display_name":"Solly","email":"solobarine@gmail.com","username":"solly47"},"last_update":"2024-03-13 14:08:49.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-03-09 08:53:30.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"36853":{"account":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"last_update":"2024-03-13 14:08:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I629669476e585a834673b8c8b49ad4b0270b877f","subject":"Forbid resetting state of active replicas","status":"MERGED","created":"2023-10-25 23:04:54.000000000","updated":"2024-03-13 14:10:02.000000000","submitted":"2024-03-13 14:08:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":52,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"899320-bug/2015328","meta_rev_id":"161710c61a013e2f2517cebcbf0850166fffffc0","_number":899320,"virtual_id_number":899320,"owner":{"_account_id":36480,"name":"Gray Lutalo","display_name":"gray","email":"jogareko@gmail.com","username":"gray"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-13 14:08:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-03-13 12:59:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2024-03-09 08:53:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2024-03-13 12:59:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":33728,"name":"Ashley Rodriguez","email":"ashrodri@redhat.com"}],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":36850,"name":"Solomon Barine Akpuru","display_name":"Solly","email":"solobarine@gmail.com","username":"solly47"},{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"}],"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":33728,"name":"Ashley Rodriguez","email":"ashrodri@redhat.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-25 23:15:23.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-10-25 23:18:03.000000000","updated_by":{"_account_id":36480,"name":"Gray Lutalo","display_name":"gray","email":"jogareko@gmail.com","username":"gray"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-10-25 23:35:10.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-26 01:07:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-26 01:07:58.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-10-26 02:59:36.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-03-07 11:59:29.000000000","updated_by":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"reviewer":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"state":"CC"},{"updated":"2024-03-08 20:10:45.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-03-08 20:18:58.000000000","updated_by":{"_account_id":36850,"name":"Solomon Barine Akpuru","display_name":"Solly","email":"solobarine@gmail.com","username":"solly47"},"reviewer":{"_account_id":36850,"name":"Solomon Barine Akpuru","display_name":"Solly","email":"solobarine@gmail.com","username":"solly47"},"state":"CC"},{"updated":"2024-03-09 08:53:48.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":33728,"name":"Ashley Rodriguez","email":"ashrodri@redhat.com"},"state":"REVIEWER"},{"updated":"2024-03-09 08:53:48.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"d24a59296f8b85dc9450836ad1361be131476a01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36480,"name":"Gray Lutalo","display_name":"gray","email":"jogareko@gmail.com","username":"gray"},"date":"2023-10-25 23:04:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0b7beef4b4c7375e9ecbf55dc0cd51d8710b32c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36480,"name":"Gray Lutalo","display_name":"gray","email":"jogareko@gmail.com","username":"gray"},"date":"2023-10-25 23:06:26.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"29bb0308e93cab7c82ff4d17646bbc5cd4b8666e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-25 23:15:23.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/b68644de8fc443c5b35b4250c5bcbffc : RETRY_LIMIT in 44s","accounts_in_message":[],"_revision_number":2},{"id":"6165e8c0276bab12089e804072010ce5010d75b3","author":{"_account_id":36480,"name":"Gray Lutalo","display_name":"gray","email":"jogareko@gmail.com","username":"gray"},"date":"2023-10-25 23:18:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"823905f99cae32af329e7ccd7a83bb9577d520a6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-25 23:35:10.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/899320/2/check/DellEMC_PowerFlex_Manila_dhss_false/20bdf9d/DellEMC_PowerFlex_Manila_dhss_false/597 : SUCCESS in 28m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"b5fd99c10e3cb4b8e10c8dfa163ea1b12191c446","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-25 23:56:10.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d65b674aff6a329f78f8ec1b48a24096a5214b21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 01:07:38.000000000","message":"Patch Set 2: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/814a346514394fcea39e3aa92654351e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6314942f22942a6a1a15275fc7a8035 : SUCCESS in 10m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9b34084d06d49be83d76e0244ab161f : FAILURE in 18m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/292ac27dbb1f4593950921336e351531 : FAILURE in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f8efb4e6d884978ac9ff53e9cb38c62 : FAILURE in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/999fd4722a6446ebbee7862725edcf1c : FAILURE in 5m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5964ece6a2c743dfbaa21be0a201e8d9 : SUCCESS in 41m 40s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4fdd78725a28408799f50e26b6614cb4 : FAILURE in 2h 00m 28s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5edd4564435444a097148212cc02bd7d : SUCCESS in 1h 16m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/5e36566611f84d5993506db8e7da5f1f : SUCCESS in 53m 07s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/9eeecb74f5684a2086a7e580af4e6a7c : SUCCESS in 1h 38m 38s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/129883641d2643d3a7e18ab2c078c822 : SUCCESS in 58m 17s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/87075ad25ab04dd9b1414b5ef7245c58 : SUCCESS in 4m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/53f41a85d74546d8857ee276423c41c9 : FAILURE in 5m 29s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/886a83e683d44c9ca529dac489463d41 : SUCCESS in 25m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a46f469c55cb445bad4f632a1946f43d : SUCCESS in 34m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/65c08d8b6868440dbc1a767d271a6756 : FAILURE in 56m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/b8c5f6e2ec544b918383ee8bf3f37a7d : FAILURE in 11m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/82b2175902bc4b8a9fbe67be22775440 : FAILURE in 10m 17s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/5bb751dc8bdb49279654dc9531844211 : SUCCESS in 1h 02m 20s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d72f7f3d02da4d23885da437794922ab : SUCCESS in 52m 09s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0e0b96f2e69f4a3ba54b578668337084 : SUCCESS in 46m 20s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/a2ef48a19a9346b19a8ebceb3c081bbf : SUCCESS in 45m 56s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/2a53ef9ab9784c23913b12ca961a7b6f : SUCCESS in 38m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d9b1a692b8bc4f044458e52d83868fbf300345c8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-26 01:07:58.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-899320-2/NFS/ : SUCCESS in 1h 18m 10s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"ca99a23959fd333ec3d3156869006bc9a8e7dc2b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-26 01:09:46.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-899320-1/NFS/ : SUCCESS in 1h 49m 58s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"e78a308bc81dbf2f7d03f8fb6b6019776453c125","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-26 02:59:36.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/899320/2/check/DellEMC_Unity_Manila_dhss_true/9b67dff/DellEMC_Unity_Manila_dhss_true/698 : SUCCESS in 2h 58m 31s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/899320/2/check/DellEMC_Unity_Manila_dhss_false/0b9bc39/DellEMC_Unity_Manila_dhss_false/737 : SUCCESS in 1h 07m 39s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"8e324930ab7203983595a46a77cc5e2c03c7bd79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36480,"name":"Gray Lutalo","display_name":"gray","email":"jogareko@gmail.com","username":"gray"},"date":"2023-10-27 17:08:09.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ddf17aec70fad98355a5c22b776d7548fff67be1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-27 18:02:28.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/899320/3/check/DellEMC_PowerFlex_Manila_dhss_false/5441d66/DellEMC_PowerFlex_Manila_dhss_false/620 : SUCCESS in 32m 00s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"56bf746b0a91dbd5c5112fd97abf9327dab2a1d8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-27 18:18:39.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/20/899320/3/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/d3f72fb/ : SUCCESS in 1h 09m 40s","accounts_in_message":[],"_revision_number":3},{"id":"caa85602f11bb0e1d36fff524f4f9a1ce187e5c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 18:58:28.000000000","message":"Patch Set 3: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c042ac4cc39470eb7ac510552391d75\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76dd5efd308540ddb07fa616fa527c5d : SUCCESS in 9m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c570f482b2d41bc82c8ca6fef9b60af : FAILURE in 19m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b93e5abc54b34444bc68c7178e4396c1 : FAILURE in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/350faec64ece4bdea5b8ecfbedd856aa : FAILURE in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fae55a586c484c6e866e18fd8270248e : FAILURE in 4m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6496a13336854067a6e746894b8dc205 : SUCCESS in 30m 41s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bee97350695b4f498cca4947b5a4941a : SUCCESS in 1h 49m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/920683fb308346f1abcc5816783f9ca8 : SUCCESS in 58m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/083e1a0deda44507aff1401bbd227d7a : SUCCESS in 51m 18s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/67f04b5f6d3940a9b15b30feb8b97807 : SUCCESS in 1h 06m 23s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/49a9241a5d364325ba716621cd7f38c1 : SUCCESS in 47m 33s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ab30d213c0d840d79848f97313fb610a : SUCCESS in 4m 49s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/02a22c7780904e51ac3b40db84ecff80 : FAILURE in 5m 47s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5cd9865a7d364968b4edfbfb38292d2c : SUCCESS in 24m 50s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/907ff37bece84a2494d46ba5c8a1ae12 : SUCCESS in 34m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d5b1bcc4bb3543698e437e764d8c151d : SUCCESS in 34m 35s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/98f916ba6acc43a08528b2c01379b4ae : FAILURE in 13m 44s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/01476c3ca0fe49f8990190fcbe3c0a06 : FAILURE in 10m 52s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/95b6ba4554734dd9bd93eab13199ef7c : SUCCESS in 43m 22s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/35324b325568479a8ef54cfd773b49ce : SUCCESS in 39m 05s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/63e14a324c774b81904ca3cfbcd23387 : SUCCESS in 55m 50s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/5c51fccc2db34973a26a497f63969ced : SUCCESS in 28m 00s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/15e3a4d1bf42473396e78763c697e40f : SUCCESS in 36m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1296e3c0438f206d4b8e441df1b7c7e162fa1358","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-27 19:43:42.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"da1995c2070d95ee7773d2186f06835793e38002","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-27 19:50:44.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-899320-3/NFS/ : SUCCESS in 1h 24m 39s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"05cc4db34bbe8dfb30cbccf081add04f4427d234","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-28 19:25:08.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/899320/3/check/DellEMC_Unity_Manila_dhss_true/5adf6bf/DellEMC_Unity_Manila_dhss_true/715 : SUCCESS in 4h 13m 02s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/899320/3/check/DellEMC_Unity_Manila_dhss_false/f860767/DellEMC_Unity_Manila_dhss_false/754 : SUCCESS in 1h 11m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"36ca64cb4f9fffe509545cc05586bf053640495a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36480,"name":"Gray Lutalo","display_name":"gray","email":"jogareko@gmail.com","username":"gray"},"date":"2023-10-29 05:32:51.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d677752a8824bec6b4d3b73685d6351cb81c1126","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-29 06:07:04.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/899320/4/check/DellEMC_PowerFlex_Manila_dhss_false/4264d3f/DellEMC_PowerFlex_Manila_dhss_false/630 : SUCCESS in 33m 56s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"5810e06040c0376682cbdafb0d0749abb1c5c562","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36480,"name":"Gray Lutalo","display_name":"gray","email":"jogareko@gmail.com","username":"gray"},"date":"2023-10-29 06:21:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a0ae96677083e90b5cb8240815cb288ecb5edf86","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-29 07:13:41.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/899320/5/check/DellEMC_PowerFlex_Manila_dhss_false/f58d2bc/DellEMC_PowerFlex_Manila_dhss_false/632 : SUCCESS in 33m 29s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"b886464c23536173d1813c092ea6f3e6870df811","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-29 07:16:14.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-899320-4/NFS/ : SUCCESS in 1h 43m 0s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"98101c60cdb1e34a9e40b6e97e41a370550fccd8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-29 07:32:41.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/20/899320/5/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/870a5ee/ : FAILURE in 1h 09m 32s","accounts_in_message":[],"_revision_number":5},{"id":"29fbb0b52ba63edbc31999b5ceddcd262c2f1988","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-29 08:04:57.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7630e8bbe2f746a2b3459aaedac52675\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b00680b2a2d2445da18c3927b3226074 : SUCCESS in 11m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3182656da26446babef9b61337e253c5 : FAILURE in 23m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14bc659cfb964351bd6c79d37eb41719 : FAILURE in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08b76cb9a4fe4b12adfd6acd6c5191b0 : SUCCESS in 11m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a11698d138447949fd0c8d804dc7ca7 : SUCCESS in 6m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6833a69958c34e1b93fa732c70050651 : SUCCESS in 33m 03s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/05ad3d1fa6164309bbc7217b99aa7af1 : FAILURE in 1h 41m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0569e360fc444cfeb738a834170df7e1 : SUCCESS in 1h 00m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b00fadf3ffbb4614a0eced4547ed04a8 : SUCCESS in 55m 35s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/1a7ff05a4f0c4b51aa8a5ded2e6eaea5 : SUCCESS in 1h 06m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0b977f31b78e4d5e9c06996a09e6939b : FAILURE in 49m 04s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/435c4cdd28564b7e8829ee1552e92124 : SUCCESS in 4m 50s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f0ee4943efd34ddfa6fb4e28a2186c6a : SUCCESS in 5m 38s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5d68c2fea7c94e7287f23fabd021c88a : SUCCESS in 23m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8fa60d4537f341f39bb97122c7afe8d9 : SUCCESS in 30m 04s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c8d6cbdc55b04aaeb1c4c0a245fc406f : SUCCESS in 36m 54s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/89075f57e702465389ecdb357c987aea : FAILURE in 11m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/dfe14709e5534abbb4d791ac7f0f6441 : FAILURE in 13m 43s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/bc93b03b5dab4c1e9fe1a0ab8977bf44 : SUCCESS in 34m 08s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/5529bf61744f4319b2a373c1fab3815c : SUCCESS in 48m 35s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c2ea67bc88f84ce4ab89ee7b07ef7566 : SUCCESS in 43m 15s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/8ed5e19ae0b0432cacd381f842cd007b : SUCCESS in 32m 46s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8516df6690e24205a303640c301249d7 : SUCCESS in 36m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e2bfe8495cb614a640a46d81a9eeee4dc5829f02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36480,"name":"Gray Lutalo","display_name":"gray","email":"jogareko@gmail.com","username":"gray"},"date":"2023-10-29 08:14:57.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"49f50aa17b7538a3ae2b255c39b1ae5951571a90","author":{"_account_id":36480,"name":"Gray Lutalo","display_name":"gray","email":"jogareko@gmail.com","username":"gray"},"date":"2023-10-29 08:20:03.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"0343dcc34c4e1e52bc829ebec21527e5fdf51fa4","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-29 08:33:16.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-899320-5/NFS/ : SUCCESS in 2h 2s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"bfc7e0de5274af5b4e77742c8a4e9fb749588c13","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-29 08:45:45.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/899320/6/check/DellEMC_PowerFlex_Manila_dhss_false/749d589/DellEMC_PowerFlex_Manila_dhss_false/633 : SUCCESS in 30m 31s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"196b04032283cf448b228dc0fec15900952b7b29","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-29 09:29:24.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/20/899320/6/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/c8ad013/ : FAILURE in 1h 12m 14s","accounts_in_message":[],"_revision_number":6},{"id":"05a755d21fd62aa11d0b849eee29334b2e2ebfc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-29 10:07:16.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/832f8afc77c24d779fdc244581c78aca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5c89f301c7d4d4ba30eaafb5753578f : SUCCESS in 8m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eaaea2c743aa41869f97dab88c854ed9 : FAILURE in 23m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6840d85882d14fe7abd535e4b166313d : FAILURE in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a58430600c54815b069346dd21886df : SUCCESS in 10m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/656ff5174e8d475a98c13015ead13bf6 : SUCCESS in 7m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9a7c7d9ccbae4e74bfba2f7a2d012c95 : SUCCESS in 36m 08s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b45b0df23e334fd3bdfb543f67384872 : SUCCESS in 1h 50m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/76baa17d80e3465e8a56c54cc32815ad : SUCCESS in 1h 24m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/69b9eac1625e404fb52db849ccef8c81 : SUCCESS in 46m 53s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/530000d0978f4e0881c813981ce339ac : SUCCESS in 1h 01m 27s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/859c62cada914911947717f762316ed6 : SUCCESS in 55m 25s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ef63293f883b46a6a5e2f188ca9f230c : SUCCESS in 5m 15s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/aa4a470a1c3d44a7a9e3fd169bb1b085 : SUCCESS in 4m 44s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9bbe8acc88a848ba9024686f280810c5 : SUCCESS in 23m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2bd948a445a346b8845c589b37f48c80 : SUCCESS in 32m 16s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cbf7cface3764e0ba4fddbfaf51ed3ec : SUCCESS in 37m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/214a17a253644cc18fe2cd49d1848d49 : FAILURE in 10m 50s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d56d5a1b6fba40bca89d5ebc9691a2c7 : FAILURE in 11m 15s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/1d8e22b150dd4f07b3a2466569e8259e : SUCCESS in 33m 31s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/8245ebda205d44e5991d5c94424494a4 : SUCCESS in 40m 46s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/49a6554260074a61a38071d83b43eb6a : SUCCESS in 42m 55s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/f9b82d01a9b64fd2bfa9d97273fda5b8 : SUCCESS in 29m 24s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f2c1e56eb1dc4ffcab78a5fc4c3b4ff2 : SUCCESS in 36m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"79580132162007323ae383cfbaae0b2e78bd2c65","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-29 10:20:12.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-899320-6/NFS/ : SUCCESS in 2h 4m 53s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"4ae9e8dbd245d489a32bfedea807ecac5bc1cbc6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-29 22:37:04.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/899320/6/check/DellEMC_Unity_Manila_dhss_true/167b417/DellEMC_Unity_Manila_dhss_true/722 : SUCCESS in 4h 33m 58s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/899320/6/check/DellEMC_Unity_Manila_dhss_false/1673888/DellEMC_Unity_Manila_dhss_false/764 : SUCCESS in 58m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"712d60730f37c9a9e2233dbb092853db303a626c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-31 18:21:04.000000000","message":"Patch Set 6: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"eca78fae49f11258709e4625e724b284ace5ffab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36480,"name":"Gray Lutalo","display_name":"gray","email":"jogareko@gmail.com","username":"gray"},"date":"2023-11-02 04:37:21.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"da470ef19b25cc219f9c10c818da7c4869142af6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-11-02 05:07:28.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/899320/7/check/DellEMC_PowerFlex_Manila_dhss_false/4a7a5b9/DellEMC_PowerFlex_Manila_dhss_false/667 : SUCCESS in 29m 52s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"3a04e11a58de1359af6397566d9881a1fcab9365","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-02 05:48:41.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/20/899320/7/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/57f406a/ : FAILURE in 1h 09m 11s","accounts_in_message":[],"_revision_number":7},{"id":"f3a4fe4fa635102a8046b115fb812e33ce76485f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-02 05:59:21.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-899320-7/NFS/ : SUCCESS in 1h 17m 44s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"d92a173e5006adc7afcec1132af64439af7cabe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 06:14:13.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8d07d1f3e664ecca093bcfe7ea128e4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31a2f0865a2542a895a03651ca3b7613 : SUCCESS in 9m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89e4682d31c842c4a95a969c1cd3839a : FAILURE in 22m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4609d4d861ab46e79608c42c5f479e75 : FAILURE in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d02b96da6aa74103ae6d205093bb70d0 : SUCCESS in 10m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1eb2a0e448404f1c823ea2fe7aea35bc : SUCCESS in 7m 30s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a12d77f2bb0146859da2696e6938646e : SUCCESS in 1h 15m 31s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9747f04e55e145b481f4297877c0d44f : FAILURE in 1h 35m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8b9bc529525d40ec8aa100cc25871c8e : SUCCESS in 58m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c38847bddeba4b4ab960ed4c1fe98464 : SUCCESS in 50m 07s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/117a3ff95e384c33be93b3c0e13f0e55 : SUCCESS in 1h 16m 03s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/24b9d4b7d1a4403e9a9a2a76e0750475 : SUCCESS in 1h 05m 09s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3ec9c4f789134e878e95f1e388f144e7 : SUCCESS in 4m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/94ffbd81f597430fab20c7705bcf0e88 : SUCCESS in 4m 16s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0acd5ab2fd0f49f38552a4418b3efa6f : SUCCESS in 37m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1919ea8b93874059b535b74fea5d8160 : SUCCESS in 31m 45s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ac9703cc2fde48309e417a0e00afe002 : SUCCESS in 59m 02s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/a31f4d2c6a434faaa817e8c05d13eca8 : FAILURE in 23m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/620142c9c5654e6f84ecdf03983ddbfe : FAILURE in 13m 41s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/03dbff8dc8fb4856a563fede82da001b : SUCCESS in 57m 45s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a58b9d3dea824916962cf9b47d6294a2 : SUCCESS in 45m 35s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9512ef4069cd492a9617c9688a6677ec : SUCCESS in 1h 09m 05s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0559e9dc87064360b5d9476efd8f6540 : SUCCESS in 44m 29s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b1583adab85041ddbc2cdb4c9cf754ad : SUCCESS in 45m 04s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"052a707e5e9c85e4a5524fe2bf08091ededc300a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36850,"name":"Solomon Barine Akpuru","display_name":"Solly","email":"solobarine@gmail.com","username":"solly47"},"date":"2024-03-07 10:07:42.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"6c00355d0e5b942d09f2bff6185ac0f90105bf34","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-07 11:59:29.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"bd69e46a8f6b4e4a03f29613d107ed67da388c19","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-07 12:03:08.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-899320-8/NFS/ : SUCCESS in 1h 18m 20s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"d0c939cfb15e7917014dedee95177b8db4d2614c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 12:33:03.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/11a0eacac63d4f2ba2b14bd9b66d4c66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b792fc0728b843348936ed101a929dc7 : SUCCESS in 9m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d890a0b4cffa4a4f9b57e9119165c7f4 : SUCCESS in 23m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/820dcafad99145f08761f7725dc28b03 : FAILURE in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad01508e1db14529bb7c8390612e1378 : SUCCESS in 16m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84e3ab1371ec4b8fb10e051caadddb3d : SUCCESS in 9m 05s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/71ace744ae9446d8881b12df13d27cc6 : SUCCESS in 57m 48s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bf91be03d962464b83751253636a6f20 : SUCCESS in 2h 24m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/8fff2ef31a564d88b004dc1b028320b5 : SUCCESS in 1h 22m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/2485426c824e4d75aee86a3cc0152e45 : SUCCESS in 1h 07m 43s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/0ebcf43581fc403aaf7e80ed67cb43c0 : SUCCESS in 1h 26m 08s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d767c57700254244b9ed76ea1f7599d8 : SUCCESS in 1h 11m 38s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4e0a0b149b4c4b73b3e40812eb794118 : SUCCESS in 4m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/370babddc8294f8e81f32821b97245e0 : FAILURE in 4m 16s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c5856523238944e0b228e45c82cafbd6 : SUCCESS in 48m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/46177342535945bb9b1010165985d475 : SUCCESS in 39m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/24c5f035f2ed4e5da3d6f4df8829330a : SUCCESS in 53m 57s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/70cfeff8ef5b4a348ae2b386a9cf3697 : FAILURE in 16m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a3df79ff591d438ab23a653f8cb8b8f2 : FAILURE in 13m 10s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/d43c8c62f6014b3eaa9aefd9f7c645ae : SUCCESS in 1h 15m 57s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/11e4ea4d8d324364b653a9d3b2e72ed9 : SUCCESS in 1h 00m 11s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/73029b92ecf14f108323627e57b33dfd : SUCCESS in 59m 12s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/29fff07e77c64daa8fd784fbfe8d3201 : SUCCESS in 25m 06s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/14b2d0587c4047edbafc2396a3a19103 : SUCCESS in 34m 59s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4a29701fcd6925a9da43cc3a1fedb9b9365f2f3a","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-03-07 19:05:23.000000000","message":"Patch Set 8: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":8},{"id":"82d41194693d94e7b6b713ca6c4a910d9085e903","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36850,"name":"Solomon Barine Akpuru","display_name":"Solly","email":"solobarine@gmail.com","username":"solly47"},"date":"2024-03-08 08:13:55.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"676969433c6ddca2f5330d84124297a1fefd5db0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36850,"name":"Solomon Barine Akpuru","display_name":"Solly","email":"solobarine@gmail.com","username":"solly47"},"date":"2024-03-08 08:16:36.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1d65f5249fa69633c8524fcfac50d097fbc9c95e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-08 09:54:53.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-899320-10/NFS/ : SUCCESS in 1h 10m 36s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":10},{"id":"2416de05741d6e0dffa4ff3ccbfb240eabfc0984","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-08 09:58:19.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-899320-9/NFS/ : SUCCESS in 1h 44m 2s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"b33ad3d8a2a779c8c920ae2533d7f89b0fa8c18e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 10:32:45.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/570af82f742a4f6eb51ead250b56943a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4db92937d6b3416eb063950cc87c2cd3 : SUCCESS in 11m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10ec5df530ed47ed84eb638793073e72 : SUCCESS in 33m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3ee1c45b98349299606964430b3c30c : FAILURE in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fcbf4d164a74fbd9a0d315af1c6e086 : SUCCESS in 16m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e3cd91775aa4b6ebcce91f6cc5ee63b : SUCCESS in 7m 15s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/754fd477d65f458ab0d8676618215f0e : SUCCESS in 51m 41s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/fa4b24b6e4214a19b53f600f0e44163c : SUCCESS in 2h 10m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7065fdb43fef46a19fde414fe8151c0d : SUCCESS in 1h 39m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e0dda0b21727455fb55066c452c353a9 : SUCCESS in 1h 06m 58s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ae3afb62892f456a8532b91f5afd64f5 : FAILURE in 1h 17m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f79148b4bbac4e0380960e4460526380 : SUCCESS in 1h 02m 33s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e3c8e13290b94f32a8316a1828814be5 : SUCCESS in 3m 59s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8730515a2e8c425d847963597a281a78 : FAILURE in 3m 03s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3c73a5a1caab438f8e54863062bd498a : SUCCESS in 29m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8082416667f9426cb3060f011b52c243 : SUCCESS in 37m 47s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/549179c8b2a444cd9320779b4ce0e4e2 : FAILURE in 51m 04s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/de9ee60e8c8c4333815f765da3692e29 : FAILURE in 23m 32s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/29923ce6a6514fa6901d2a7545a233c5 : FAILURE in 24m 08s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/533f62a693c34d4aab4b7be689a0cd96 : SUCCESS in 1h 15m 43s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/38294375f83f408da65c43dbec784c67 : SUCCESS in 48m 29s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b95cf195c83246d5a3d6ace7caeec591 : SUCCESS in 53m 38s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ccaf968471d948168416d2fb4578f344 : SUCCESS in 39m 03s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8d50be2a4bc74ea190cd9501b3578a91 : SUCCESS in 41m 25s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b4f51589e1581014c7b0f4193827c2e29fe814cf","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-08 12:11:27.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"bcfc2f90c0a4a90ffb8dafcae27b3349948a2c56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36850,"name":"Solomon Barine Akpuru","display_name":"Solly","email":"solobarine@gmail.com","username":"solly47"},"date":"2024-03-08 13:03:41.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"944e8b45e0671c3287933311de6d2e59bfcff0c0","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-08 13:21:33.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"87fb7b46b24071965585f64c364e89fb1d48611d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-08 15:02:33.000000000","message":"Patch Set 11:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-899320-11/NFS/ : SUCCESS in 1h 58m 30s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":11},{"id":"49eb99092cb1389016cd785367d4ed9a636cf0b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36850,"name":"Solomon Barine Akpuru","display_name":"Solly","email":"solobarine@gmail.com","username":"solly47"},"date":"2024-03-08 15:06:24.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"77fbd3f0ce2704736bc2fa5b24e35fe0feb9fc4d","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-08 16:33:17.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"d2a916c8b5783f2fc97938cf76d344ed172a1330","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 17:01:54.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02ffd8607a4d40b58bd5529ccff83f25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c806627d23fc4b17ab163d483937f6cc : SUCCESS in 9m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/111d038078934dbaae9dff59a7354c64 : SUCCESS in 27m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d8da506d38b48fea3799efa91268618 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/282222e5813546ce88a2661d5fbd028f : SUCCESS in 14m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd0ac14c2d4549cb81ccddc877b8e6e3 : SUCCESS in 9m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/402d891d41ca48fd9626588bad724c93 : SUCCESS in 9m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/e2c4a7842bd54ec6bfa82ebbe35cd9f0 : SUCCESS in 1h 08m 45s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a1e575f3d28048108769bdfeaa23d070 : SUCCESS in 1h 52m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/faeb194d15c744c5a09f9815e2d0a33d : SUCCESS in 1h 23m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b4972dccadee48bf9e1fdbe6075e6262 : FAILURE in 1h 10m 56s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/83174dba269f462abfb2d60c9a243279 : FAILURE in 1h 10m 36s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8a61d119179a46aab0cfe67faac6097e : FAILURE in 1h 06m 58s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/70bf864de62f4068a41805d0956df78e : SUCCESS in 4m 05s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2325dc0a9ec9473c949c180a960ee800 : FAILURE in 2m 53s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d98bb26248a648fc8a40214907aff189 : SUCCESS in 31m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/dafcb5c4fd2a4b1bb12c765300be2da1 : SUCCESS in 37m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/5604dc7d6db648f99586b3b2088b8c95 : SUCCESS in 44m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/6a6b1b1417574ad48149e8e7e1842aff : FAILURE in 17m 20s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2555a565f25d4d77a3d7e0b173d22394 : FAILURE in 21m 05s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/fba356bb106c4d15972d2bbb0ce43f8c : SUCCESS in 1h 33m 56s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/a465c12fae4f488aad882f0ede5926ea : SUCCESS in 54m 58s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d220b7385d2f4155a0e1c1ba6ee8dbb4 : SUCCESS in 57m 49s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/8eeca0076aef469cbca4f305cc0bc6bc : SUCCESS in 35m 40s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/44e0a20372d2483292663488470b6ffb : SUCCESS in 42m 34s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"32750db54e7c7047f82a11d1310a1911e5655697","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-08 18:09:50.000000000","message":"Patch Set 12:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-899320-12/NFS/ : SUCCESS in 1h 35m 46s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":12},{"id":"c33621b8bd3cd5c1b5b61f97ee59ba4effcc59b3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-03-08 19:27:06.000000000","message":"Patch Set 12: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":12},{"id":"c6eea31876e04467a519c29cb3397bc3f369d3f4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-08 20:10:45.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/899320/12/upstream-check/manila-tempest-plugin-ontap-dhss/5c5c5c3/ : FAILURE in 1h 39m 21s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/899320/12/upstream-check/manila-tempest-plugin-ontap-no-dhss/96feec0/ : SUCCESS in 3h 08m 16s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":12},{"id":"ff2f68b4c661e93dbe5664af582de0b50e844b29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36850,"name":"Solomon Barine Akpuru","display_name":"Solly","email":"solobarine@gmail.com","username":"solly47"},"date":"2024-03-08 20:14:08.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"5e0298eb87452230771acbdaca16a850b0011fe2","author":{"_account_id":36850,"name":"Solomon Barine Akpuru","display_name":"Solly","email":"solobarine@gmail.com","username":"solly47"},"date":"2024-03-08 20:18:58.000000000","message":"Patch Set 12:\n\n(6 comments)","accounts_in_message":[],"_revision_number":12},{"id":"7f1ea4e318daf0b2c2a2df482d5d9786e8a90d1d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-03-08 20:30:31.000000000","message":"Patch Set 13: Code-Review+2\n\n(12 comments)","accounts_in_message":[],"_revision_number":13},{"id":"d57f52e70296ff4eb700488855974a96c13f5c55","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-08 22:28:20.000000000","message":"Patch Set 13:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-899320-13/NFS/ : SUCCESS in 2h 13m 50s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":13},{"id":"94321fdfa859a705db97c217783cced007b8b246","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 22:28:42.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55e21262e5674f4fa2041fb7cc5325b4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255fdf0084c749f384cbf8fe40c6b029 : SUCCESS in 9m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/747c8c1bd9ba4c93a7d2479597b060b4 : SUCCESS in 23m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13ba4fd38c0e4e6a82d5a69322abeb5e : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3890d19718744dd9e228b7c7e21e5c0 : SUCCESS in 10m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64e65d39f1e846778559a7c60723caca : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14cb29f0b42a437486b6e0ff60d71677 : SUCCESS in 7m 43s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d37afedfb4a945a0b04dc03f9a79c571 : SUCCESS in 1h 13m 09s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2d6eff71385e4c05932a86de9e8dd1a6 : SUCCESS in 2h 08m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5efe423cfbdc4e2f93a04b6e7d706ec2 : SUCCESS in 58m 05s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/062bdb2eaf1842aea631de3da42f94fb : SUCCESS in 55m 58s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7d7a5d8740ab4217807b37509e38a4f6 : SUCCESS in 1h 24m 19s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/80c41e26f8b44f1a84df656d802e4d47 : FAILURE in 1h 16m 21s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/54c4e508fe2e4785a1144e4baec8ace3 : SUCCESS in 4m 17s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5afe728fc524407bae2995110431d84b : FAILURE in 3m 14s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6bce45af02dc435e8db3795da1bdfc00 : SUCCESS in 41m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6995f61ab94a40a6b1e849900020f65d : SUCCESS in 30m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e1fcafbce96b4c068440ce80ad99e4e3 : FAILURE in 49m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/8439443afad04c41bf83bd291c5280e7 : FAILURE in 12m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9b7edb93d609406482ff1b79ea574aa1 : FAILURE in 19m 27s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/2e7798fde6c7406eba304fa8e844f269 : SUCCESS in 51m 57s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4da450b401da4cc3a7fa2c5c3f91fc0a : SUCCESS in 46m 57s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b7020f18eb8d47fe9e0ca5d503d42277 : SUCCESS in 1h 02m 49s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2eb4a7605a94404fa52dc45306ce62c6 : SUCCESS in 38m 58s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/53cdae2ec9b34e7eb096a3b204a043de : SUCCESS in 28m 40s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"8a64c1336179b9141a1973485fda233e7ae259c7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-09 02:55:54.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/899320/13/upstream-check/manila-tempest-plugin-ontap-dhss/3d6521c/ : TIMED_OUT in 4h 10m 26s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/899320/13/upstream-check/manila-tempest-plugin-ontap-no-dhss/d5ba366/ : TIMED_OUT in 4h 09m 55s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"79702c564390306fcdce19890b9081e04991f368","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36850,"name":"Solomon Barine Akpuru","display_name":"Solly","email":"solobarine@gmail.com","username":"solly47"},"date":"2024-03-09 06:43:08.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"aecfa2ee6243e67355d58e61756fec81c92d01f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-09 08:24:35.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f3d9b2aa32d441982fa9f5ef7c57f6e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d9ddd912c974e10a45c57ad1f0044a7 : SUCCESS in 9m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b94607a42725497498980d8e290663fa : SUCCESS in 24m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77910aec98f5475b9161acce52d4e479 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f2cb9f298244763b88b6f5df88f180c : SUCCESS in 8m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c64f43ff8cc04e7f80294634e41c0500 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/812d5be24c2c4120b394f73cf35288b0 : SUCCESS in 5m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/dd954cbc87a64d8dacb8eefb1ae2ae6b : SUCCESS in 45m 10s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/de78df8bc5034ee0ad9d487811292b0c : SUCCESS in 1h 29m 51s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b03c1cae62cb4619a7085d6eb2f422d6 : SUCCESS in 54m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/5e44bcd84b384fceae6ae71e216302bf : SUCCESS in 1h 08m 16s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8b2bf365f3ad4a64a7234a8d4834f709 : SUCCESS in 1h 35m 47s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/99f97c6255d54408be6e599d4ad02b9c : SUCCESS in 47m 01s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e2d73056649646efb000f0b9460aed87 : SUCCESS in 4m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c09bef4aef6b43539d2285369088096b : FAILURE in 2m 54s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c9ae048c0e184769abb81a05043a2c50 : SUCCESS in 37m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/92f97078cc4547608507ddda2c544de3 : SUCCESS in 42m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8988e6a8b9bb494b83e81413ee6ba9df : FAILURE in 56m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/12a43a4245414060a11ba463ac48dfa5 : FAILURE in 21m 11s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/099b9db553604d6e973975a2b4a54f40 : FAILURE in 28m 21s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/a450e50bdb094e919dc547388536e25e : SUCCESS in 1h 24m 46s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3f508fbac90043048fca0ce18a51431c : SUCCESS in 58m 49s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c8ad32398db543ee932d72d7f90e6b99 : SUCCESS in 58m 23s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/acb8cf24df66403299b78774ae6ce958 : SUCCESS in 43m 56s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e36b86bcdbed4975b4c823b072291d02 : SUCCESS in 43m 30s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a675b8aada331bcbbf2b874cb30e9a39da9d5c4d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-09 08:33:26.000000000","message":"Patch Set 14:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-899320-14/NFS/ : SUCCESS in 1h 49m 55s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":14},{"id":"2025c73afa0fec8f836ee079c1745bd3bf0f7f15","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-03-09 08:53:30.000000000","message":"Patch Set 14: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"e8428846f17efa4dc3316111455abcbfa00fe802","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-09 10:34:14.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/899320/14/upstream-check/manila-tempest-plugin-ontap-dhss/769fecb/ : SUCCESS in 2h 06m 08s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/20/899320/14/upstream-check/manila-tempest-plugin-ontap-no-dhss/ae50aef/ : SUCCESS in 1h 53m 17s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"7eb39f5e7e97eff6f15945f2841ee5565e0d0a57","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-03-13 12:59:18.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"78707e9e104b43d5909cf93043a454c7734977e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 12:59:49.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"1b54d315d1c55f58c7eebb3ae53c1afa56df8d04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 14:08:47.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/626f9de9c5744933b381552d083d306a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1556ee8700bf450cbdf6238208e3a07b : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14ab95f73c0c4e7f8f3d6ba5b40ee41a : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd364565dbe141b591e4c552b6534959 : SUCCESS in 9m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a916020cbc14e6d90aaaef6c547dd1b : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d54f39b4407242c6a94593b65fe572b0 : SUCCESS in 8m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f03d243b7a28453b80a640aca0d206fe : SUCCESS in 1h 07m 30s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bc4715e1429c4eb4ae67cff7958fd1c0 : SUCCESS in 31m 22s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0fbc96560b3848fd9023901e33383ea0 : SUCCESS in 39m 20s","accounts_in_message":[],"_revision_number":14},{"id":"dbd7d98912174447683addca07d2ab998a195695","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 14:08:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"161710c61a013e2f2517cebcbf0850166fffffc0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 14:10:02.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a144110f700d4e45a99834b8b0fc9e80\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f635e950fcf64c338480d8eac8d9a3f9 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85187a5599974247a650cab4a0dfe997 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"b7a1b5b2cf765b1570432b67363905251893ce42","revisions":{"3c6875be2dcd3d09c77e35fdf30c6ed77735d205":{"kind":"REWORK","_number":1,"created":"2023-10-25 23:04:54.000000000","uploader":{"_account_id":36480,"name":"Gray Lutalo","display_name":"gray","email":"jogareko@gmail.com","username":"gray"},"ref":"refs/changes/20/899320/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/899320/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/899320/1"}}},"commit":{"parents":[{"commit":"5d5ece51eac9fed78373d1a43f24a6112765df93","subject":"update irc dcoumentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5d5ece51eac9fed78373d1a43f24a6112765df93"}]}],"author":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-10-25 22:33:30.000000000","tz":0},"committer":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-10-25 22:33:30.000000000","tz":0},"subject":"reset replica state not forbidden for active replica","message":"reset replica state not forbidden for active replica\n\nIn a replication setup, users encountered a critical issue where they unintentionally reset the replica_state of an active replica while attempting to resolve errors on a non-active replica. This led to a situation with no active replica, causing data loss. Users expected server-side valdation to prevent such actions.\nThis commit implements the necessary validation in the codebase to ensure that the reset_replica_state action cannot be applied to active replicas, addressing the reported issue and improving data integrity in replication setups.\n\nCloses-Bug: #2015328\nChange-Id: I629669476e585a834673b8c8b49ad4b0270b877f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3c6875be2dcd3d09c77e35fdf30c6ed77735d205"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3c6875be2dcd3d09c77e35fdf30c6ed77735d205"}]},"branch":"refs/heads/master"},"26a60021565b74bdbc560b64a48acde6ae98a2d5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-10-25 23:06:26.000000000","uploader":{"_account_id":36480,"name":"Gray Lutalo","display_name":"gray","email":"jogareko@gmail.com","username":"gray"},"ref":"refs/changes/20/899320/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/899320/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/899320/2"}}},"commit":{"parents":[{"commit":"5d5ece51eac9fed78373d1a43f24a6112765df93","subject":"update irc dcoumentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5d5ece51eac9fed78373d1a43f24a6112765df93"}]}],"author":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-10-25 22:33:30.000000000","tz":0},"committer":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-10-25 23:06:26.000000000","tz":0},"subject":"reset replica state not forbidden for active replica","message":"reset replica state not forbidden for active replica\n\nIn a replication setup, users encountered a critical issue where they unintentionally reset the replica_state of an active replica while attempting to resolve errors on a non-active replica. This led to a situation with no active replica, causing data loss. Users expected server-side valdation to prevent such actions.\n\nCloses-Bug: #2015328\n\nChange-Id: I629669476e585a834673b8c8b49ad4b0270b877f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/26a60021565b74bdbc560b64a48acde6ae98a2d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/26a60021565b74bdbc560b64a48acde6ae98a2d5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"206d5a09e4b230b6e837763b4212fed3523e66bc":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-10-27 17:08:09.000000000","uploader":{"_account_id":36480,"name":"Gray Lutalo","display_name":"gray","email":"jogareko@gmail.com","username":"gray"},"ref":"refs/changes/20/899320/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/899320/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/899320/3"}}},"commit":{"parents":[{"commit":"9812983bbb2583b17e62f87f65214be9e379632d","subject":"Merge \"[Pure Storage] Fix connection issue on network failure.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9812983bbb2583b17e62f87f65214be9e379632d"}]}],"author":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-10-25 22:33:30.000000000","tz":0},"committer":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-10-27 17:08:09.000000000","tz":0},"subject":"reset replica state not forbidden for active replica","message":"reset replica state not forbidden for active replica\n\nIn a replication setup, users encountered a critical issue where they unintentionally reset the replica_state of an active replica while attempting to resolve errors on a non-active replica. This led to a situation with no active replica, causing data loss. Users expected server-side valdation to prevent such actions.\n\nCloses-Bug: #2015328\n\nChange-Id: I629669476e585a834673b8c8b49ad4b0270b877f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/206d5a09e4b230b6e837763b4212fed3523e66bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/206d5a09e4b230b6e837763b4212fed3523e66bc"}]},"branch":"refs/heads/master","description":"Rebase"},"5b09277eb409fb580abfe59bb493107205d6aa86":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2023-10-29 05:32:51.000000000","uploader":{"_account_id":36480,"name":"Gray Lutalo","display_name":"gray","email":"jogareko@gmail.com","username":"gray"},"ref":"refs/changes/20/899320/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/899320/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/899320/4"}}},"commit":{"parents":[{"commit":"1c44d9946012e78be451e317a021085bea7a1c98","subject":"Merge \"Change status and error handling for /shares API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1c44d9946012e78be451e317a021085bea7a1c98"}]}],"author":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-10-25 22:33:30.000000000","tz":0},"committer":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-10-29 05:15:02.000000000","tz":0},"subject":"reset replica state not forbidden for active replica","message":"reset replica state not forbidden for active replica\n\nIn a replication setup, users encountered a critical issue where they unintentionally reset the replica_state of an active replica while attempting to resolve errors on a non-active replica. This led to a situation with no active replica, causing data loss. Users expected server-side valdation to prevent such actions.\nThis commit implements the necessary validation in the codebase to ensure that the reset_replica_state action cannot be applied to active replicas, addressing the reported issue and improving data integrity in replication setups.\n\nCloses-Bug: #2015328\nChange-Id: I629669476e585a834673b8c8b49ad4b0270b877f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5b09277eb409fb580abfe59bb493107205d6aa86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5b09277eb409fb580abfe59bb493107205d6aa86"}]},"branch":"refs/heads/master"},"316ed8962129c6ba7b4232f64f76bbfc2eb322df":{"kind":"REWORK","_number":5,"created":"2023-10-29 06:21:55.000000000","uploader":{"_account_id":36480,"name":"Gray Lutalo","display_name":"gray","email":"jogareko@gmail.com","username":"gray"},"ref":"refs/changes/20/899320/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/899320/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/899320/5"}}},"commit":{"parents":[{"commit":"1c44d9946012e78be451e317a021085bea7a1c98","subject":"Merge \"Change status and error handling for /shares API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1c44d9946012e78be451e317a021085bea7a1c98"}]}],"author":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-10-25 22:33:30.000000000","tz":0},"committer":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-10-29 06:21:02.000000000","tz":0},"subject":"reset replica state not forbidden for active replica","message":"reset replica state not forbidden for active replica\n\nIn a replication setup, users encountered a critical issue where they unintentionally reset the replica_state of an active replica while attempting to resolve errors on a non-active replica. This led to a situation with no active replica, causing data loss. Users expected server-side valdation to prevent such actions.\nThis commit implements the necessary validation in the codebase to ensure that the reset_replica_state action cannot be applied to active replicas, addressing the reported issue and improving data integrity in replication setups.\n\nCloses-Bug: #2015328\nChange-Id: I629669476e585a834673b8c8b49ad4b0270b877f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/316ed8962129c6ba7b4232f64f76bbfc2eb322df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/316ed8962129c6ba7b4232f64f76bbfc2eb322df"}]},"branch":"refs/heads/master"},"41395c3aea1191db43f7847f74953e6f265fea84":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2023-10-29 08:14:57.000000000","uploader":{"_account_id":36480,"name":"Gray Lutalo","display_name":"gray","email":"jogareko@gmail.com","username":"gray"},"ref":"refs/changes/20/899320/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/899320/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/899320/6"}}},"commit":{"parents":[{"commit":"1c44d9946012e78be451e317a021085bea7a1c98","subject":"Merge \"Change status and error handling for /shares API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1c44d9946012e78be451e317a021085bea7a1c98"}]}],"author":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-10-25 22:33:30.000000000","tz":0},"committer":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-10-29 08:14:57.000000000","tz":0},"subject":"Forbid resetting state of active replicas","message":"Forbid resetting state of active replicas\n\nIn a replication setup, users encountered a critical\nissue where they unintentionally reset the replica_state\nof an active replica while attempting to resolve errors\non a non-active replica. This led to a situation with no\nactive replica, causing data loss. Users expected\nserver-side validation to prevent such actions.\n\nThis commit implements the necessary validation in the\ncodebase to ensure that the reset_replica_state action\ncannot be applied to active replicas, addressing the\nreported issue and improving data integrity in\nreplication setups.\n\nCloses-Bug: #2015328\nChange-Id: I629669476e585a834673b8c8b49ad4b0270b877f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/41395c3aea1191db43f7847f74953e6f265fea84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/41395c3aea1191db43f7847f74953e6f265fea84"}]},"branch":"refs/heads/master","description":"Edit commit message"},"587e2c4ab9dd41e3e760b820a37567fdc06e71e9":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-11-02 04:37:21.000000000","uploader":{"_account_id":36480,"name":"Gray Lutalo","display_name":"gray","email":"jogareko@gmail.com","username":"gray"},"ref":"refs/changes/20/899320/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/899320/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/899320/7"}}},"commit":{"parents":[{"commit":"165b140eababcdf7b811f9c47f582ec05893b2ab","subject":"Merge \"Fix typo in the release note for bug 2038607\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/165b140eababcdf7b811f9c47f582ec05893b2ab"}]}],"author":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-10-25 22:33:30.000000000","tz":0},"committer":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-11-02 04:37:21.000000000","tz":0},"subject":"Forbid resetting state of active replicas","message":"Forbid resetting state of active replicas\n\nIn a replication setup, users encountered a critical\nissue where they unintentionally reset the replica_state\nof an active replica while attempting to resolve errors\non a non-active replica. This led to a situation with no\nactive replica, causing data loss. Users expected\nserver-side validation to prevent such actions.\n\nThis commit implements the necessary validation in the\ncodebase to ensure that the reset_replica_state action\ncannot be applied to active replicas, addressing the\nreported issue and improving data integrity in\nreplication setups.\n\nCloses-Bug: #2015328\nChange-Id: I629669476e585a834673b8c8b49ad4b0270b877f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/587e2c4ab9dd41e3e760b820a37567fdc06e71e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/587e2c4ab9dd41e3e760b820a37567fdc06e71e9"}]},"branch":"refs/heads/master","description":"Rebase"},"6533938906c9a62f9f6a7dccb74a1e7c82c0ed75":{"kind":"REWORK","_number":8,"created":"2024-03-07 10:07:42.000000000","uploader":{"_account_id":36850,"name":"Solomon Barine Akpuru","display_name":"Solly","email":"solobarine@gmail.com","username":"solly47"},"ref":"refs/changes/20/899320/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/899320/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/899320/8"}}},"commit":{"parents":[{"commit":"a5532c3080b7457093744a5b49d2f3ed3a6dd0df","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a5532c3080b7457093744a5b49d2f3ed3a6dd0df"}]}],"author":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-10-25 22:33:30.000000000","tz":0},"committer":{"name":"Solly","email":"solobarine@gmail.com","date":"2024-03-07 10:05:49.000000000","tz":60},"subject":"Forbid resetting state of active replicas","message":"Forbid resetting state of active replicas\n\nIn a replication setup, users encountered a critical\nissue where they unintentionally reset the replica_state\nof an active replica while attempting to resolve errors\non a non-active replica. This led to a situation with no\nactive replica, causing data loss. Users expected\nserver-side validation to prevent such actions.\n\nThis commit implements the necessary validation in the\ncodebase to ensure that the reset_replica_state action\ncannot be applied to active replicas, addressing the\nreported issue and improving data integrity in\nreplication setups.\n\nCloses-Bug: #2015328\nChange-Id: I629669476e585a834673b8c8b49ad4b0270b877f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6533938906c9a62f9f6a7dccb74a1e7c82c0ed75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6533938906c9a62f9f6a7dccb74a1e7c82c0ed75"}]},"branch":"refs/heads/master"},"1ecce0c1a8228b91c7a1b47a222329a08fca82de":{"kind":"REWORK","_number":9,"created":"2024-03-08 08:13:55.000000000","uploader":{"_account_id":36850,"name":"Solomon Barine Akpuru","display_name":"Solly","email":"solobarine@gmail.com","username":"solly47"},"ref":"refs/changes/20/899320/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/899320/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/899320/9"}}},"commit":{"parents":[{"commit":"a5532c3080b7457093744a5b49d2f3ed3a6dd0df","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a5532c3080b7457093744a5b49d2f3ed3a6dd0df"}]}],"author":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-10-25 22:33:30.000000000","tz":0},"committer":{"name":"Solly","email":"solobarine@gmail.com","date":"2024-03-08 08:13:24.000000000","tz":60},"subject":"Forbid resetting state of active replicas","message":"Forbid resetting state of active replicas\n\nIn a replication setup, users encountered a critical\nissue where they unintentionally reset the replica_state\nof an active replica while attempting to resolve errors\non a non-active replica. This led to a situation with no\nactive replica, causing data loss. Users expected\nserver-side validation to prevent such actions.\n\nThis commit implements the necessary validation in the\ncodebase to ensure that the reset_replica_state action\ncannot be applied to active replicas, addressing the\nreported issue and improving data integrity in\nreplication setups.\n\nCloses-Bug: #2015328\nChange-Id: I629669476e585a834673b8c8b49ad4b0270b877f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1ecce0c1a8228b91c7a1b47a222329a08fca82de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1ecce0c1a8228b91c7a1b47a222329a08fca82de"}]},"branch":"refs/heads/master"},"d19b3da3e75f0624447d3c120fc6cf0d359459cc":{"kind":"REWORK","_number":10,"created":"2024-03-08 08:16:36.000000000","uploader":{"_account_id":36850,"name":"Solomon Barine Akpuru","display_name":"Solly","email":"solobarine@gmail.com","username":"solly47"},"ref":"refs/changes/20/899320/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/899320/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/899320/10"}}},"commit":{"parents":[{"commit":"a5532c3080b7457093744a5b49d2f3ed3a6dd0df","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a5532c3080b7457093744a5b49d2f3ed3a6dd0df"}]}],"author":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-10-25 22:33:30.000000000","tz":0},"committer":{"name":"Solly","email":"solobarine@gmail.com","date":"2024-03-08 08:16:14.000000000","tz":60},"subject":"Forbid resetting state of active replicas","message":"Forbid resetting state of active replicas\n\nIn a replication setup, users encountered a critical\nissue where they unintentionally reset the replica_state\nof an active replica while attempting to resolve errors\non a non-active replica. This led to a situation with no\nactive replica, causing data loss. Users expected\nserver-side validation to prevent such actions.\n\nThis commit implements the necessary validation in the\ncodebase to ensure that the reset_replica_state action\ncannot be applied to active replicas, addressing the\nreported issue and improving data integrity in\nreplication setups.\n\nCloses-Bug: #2015328\nChange-Id: I629669476e585a834673b8c8b49ad4b0270b877f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d19b3da3e75f0624447d3c120fc6cf0d359459cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d19b3da3e75f0624447d3c120fc6cf0d359459cc"}]},"branch":"refs/heads/master"},"f58bf29e572bd9a2fe6e71448dc2f0be6d470151":{"kind":"REWORK","_number":11,"created":"2024-03-08 13:03:41.000000000","uploader":{"_account_id":36850,"name":"Solomon Barine Akpuru","display_name":"Solly","email":"solobarine@gmail.com","username":"solly47"},"ref":"refs/changes/20/899320/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/899320/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/899320/11"}}},"commit":{"parents":[{"commit":"a5532c3080b7457093744a5b49d2f3ed3a6dd0df","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a5532c3080b7457093744a5b49d2f3ed3a6dd0df"}]}],"author":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-10-25 22:33:30.000000000","tz":0},"committer":{"name":"Solly","email":"solobarine@gmail.com","date":"2024-03-08 13:03:12.000000000","tz":60},"subject":"Forbid resetting state of active replicas","message":"Forbid resetting state of active replicas\n\nIn a replication setup, users encountered a critical\nissue where they unintentionally reset the replica_state\nof an active replica while attempting to resolve errors\non a non-active replica. This led to a situation with no\nactive replica, causing data loss. Users expected\nserver-side validation to prevent such actions.\n\nThis commit implements the necessary validation in the\ncodebase to ensure that the reset_replica_state action\ncannot be applied to active replicas, addressing the\nreported issue and improving data integrity in\nreplication setups.\n\nCloses-Bug: #2015328\nChange-Id: I629669476e585a834673b8c8b49ad4b0270b877f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f58bf29e572bd9a2fe6e71448dc2f0be6d470151"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f58bf29e572bd9a2fe6e71448dc2f0be6d470151"}]},"branch":"refs/heads/master"},"222adc48f083cc590a6e7e6183344568348248ed":{"kind":"REWORK","_number":12,"created":"2024-03-08 15:06:24.000000000","uploader":{"_account_id":36850,"name":"Solomon Barine Akpuru","display_name":"Solly","email":"solobarine@gmail.com","username":"solly47"},"ref":"refs/changes/20/899320/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/899320/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/899320/12"}}},"commit":{"parents":[{"commit":"a5532c3080b7457093744a5b49d2f3ed3a6dd0df","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a5532c3080b7457093744a5b49d2f3ed3a6dd0df"}]}],"author":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-10-25 22:33:30.000000000","tz":0},"committer":{"name":"Solly","email":"solobarine@gmail.com","date":"2024-03-08 15:06:08.000000000","tz":60},"subject":"Forbid resetting state of active replicas","message":"Forbid resetting state of active replicas\n\nIn a replication setup, users encountered a critical\nissue where they unintentionally reset the replica_state\nof an active replica while attempting to resolve errors\non a non-active replica. This led to a situation with no\nactive replica, causing data loss. Users expected\nserver-side validation to prevent such actions.\n\nThis commit implements the necessary validation in the\ncodebase to ensure that the reset_replica_state action\ncannot be applied to active replicas, addressing the\nreported issue and improving data integrity in\nreplication setups.\n\nCloses-Bug: #2015328\nChange-Id: I629669476e585a834673b8c8b49ad4b0270b877f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/222adc48f083cc590a6e7e6183344568348248ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/222adc48f083cc590a6e7e6183344568348248ed"}]},"branch":"refs/heads/master"},"b7ad6c3e90a13b8e1c9641c15b82c02bb9d009a5":{"kind":"REWORK","_number":13,"created":"2024-03-08 20:14:08.000000000","uploader":{"_account_id":36850,"name":"Solomon Barine Akpuru","display_name":"Solly","email":"solobarine@gmail.com","username":"solly47"},"ref":"refs/changes/20/899320/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/899320/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/899320/13"}}},"commit":{"parents":[{"commit":"a5532c3080b7457093744a5b49d2f3ed3a6dd0df","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a5532c3080b7457093744a5b49d2f3ed3a6dd0df"}]}],"author":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-10-25 22:33:30.000000000","tz":0},"committer":{"name":"Solly","email":"solobarine@gmail.com","date":"2024-03-08 20:12:59.000000000","tz":60},"subject":"Forbid resetting state of active replicas","message":"Forbid resetting state of active replicas\n\nIn a replication setup, users encountered a critical\nissue where they unintentionally reset the replica_state\nof an active replica while attempting to resolve errors\non a non-active replica. This led to a situation with no\nactive replica, causing data loss. Users expected\nserver-side validation to prevent such actions.\n\nThis commit implements the necessary validation in the\ncodebase to ensure that the reset_replica_state action\ncannot be applied to active replicas, addressing the\nreported issue and improving data integrity in\nreplication setups.\n\nCo-Authored-By: Solly \u003csolobarine@gmail.com\u003e\nCloses-Bug: #2015328\nChange-Id: I629669476e585a834673b8c8b49ad4b0270b877f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b7ad6c3e90a13b8e1c9641c15b82c02bb9d009a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b7ad6c3e90a13b8e1c9641c15b82c02bb9d009a5"}]},"branch":"refs/heads/master"},"b7a1b5b2cf765b1570432b67363905251893ce42":{"kind":"REWORK","_number":14,"created":"2024-03-09 06:43:08.000000000","uploader":{"_account_id":36850,"name":"Solomon Barine Akpuru","display_name":"Solly","email":"solobarine@gmail.com","username":"solly47"},"ref":"refs/changes/20/899320/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/899320/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899320/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/899320/14"}}},"commit":{"parents":[{"commit":"a5532c3080b7457093744a5b49d2f3ed3a6dd0df","subject":"Merge \"reno: Update master for unmaintained/xena\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a5532c3080b7457093744a5b49d2f3ed3a6dd0df"}]}],"author":{"name":"Gray Lutalo","email":"jogareko@gmail.com","date":"2023-10-25 22:33:30.000000000","tz":0},"committer":{"name":"Solly","email":"solobarine@gmail.com","date":"2024-03-09 06:40:42.000000000","tz":60},"subject":"Forbid resetting state of active replicas","message":"Forbid resetting state of active replicas\n\nIn a replication setup, users encountered a critical\nissue where they unintentionally reset the replica_state\nof an active replica while attempting to resolve errors\non a non-active replica. This led to a situation with no\nactive replica, causing data loss. Users expected\nserver-side validation to prevent such actions.\n\nThis commit implements the necessary validation in the\ncodebase to ensure that the reset_replica_state action\ncannot be applied to active replicas, addressing the\nreported issue and improving data integrity in\nreplication setups.\n\nCo-Authored-By: Solly \u003csolobarine@gmail.com\u003e\nCloses-Bug: #2015328\nChange-Id: I629669476e585a834673b8c8b49ad4b0270b877f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b7a1b5b2cf765b1570432b67363905251893ce42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b7a1b5b2cf765b1570432b67363905251893ce42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
