)]}'
{"id":"openstack%2Fmanila~899346","triplet_id":"openstack%2Fmanila~master~I8d4b30daae2fe8c88c30d93d402bf2e5a558f804","project":"openstack/manila","branch":"master","attention_set":{},"removed_from_attention_set":{"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2023-10-27 20:39:50.000000000","reason":"Change was submitted"},"36483":{"account":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"last_update":"2023-10-27 20:39:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8d4b30daae2fe8c88c30d93d402bf2e5a558f804","subject":"Change status and error handling for /shares API","status":"MERGED","created":"2023-10-26 00:11:20.000000000","updated":"2023-10-27 22:13:35.000000000","submitted":"2023-10-27 20:39:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"899346","meta_rev_id":"c69fcbb4b3198519333b70817e38d11836e3e055","_number":899346,"virtual_id_number":899346,"owner":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-27 20:39:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2023-10-27 20:11:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2023-10-27 18:11:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2023-10-27 20:11:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-26 00:42:44.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-26 01:24:02.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-10-26 02:19:48.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-10-26 02:27:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-26 05:24:55.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-10-26 05:40:09.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-10-26 22:52:59.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-10-26 23:21:54.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-10-27 20:11:30.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"}],"messages":[{"id":"22d993e62ed05e0a62e793fafbe110408d0e1d17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2023-10-26 00:11:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b4f71bb53b79ab968c79343e4dd52f8f4b07db3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-26 00:42:44.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/46/899346/1/check/DellEMC_PowerFlex_Manila_dhss_false/05ea3b0/DellEMC_PowerFlex_Manila_dhss_false/598 : SUCCESS in 31m 09s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"39a8d7e80f207b1d5350397ac818852c85aaa375","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-26 01:24:02.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/46/899346/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/098bb24/ : SUCCESS in 1h 10m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0d9308e2e8be6ea8cfffb7da1d479e5e9df5e9b3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-26 02:19:48.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c1b3aee24282206b3f7ee4190edebf31ba1a6dbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 02:27:56.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a55ebd33f954476839675930ae63e17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18034d87427e40aa9c636179b8664cea : SUCCESS in 10m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f72b6a63c774ee49f54215c9d2170bd : FAILURE in 21m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b927a6a2a1b148d99937c6e74000409c : FAILURE in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ddf3e4a0740e4fce8e7e31caee513d85 : SUCCESS in 9m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0750e6bc92774915a2469b4ff025a548 : SUCCESS in 9m 06s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ab5430951cf54342a67d7f4d90e62f0a : SUCCESS in 29m 59s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ae53bfc61f584ae183d8e3b06668d781 : FAILURE in 2h 11m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d645e688bf324aa8909eb8abd4b69bb1 : SUCCESS in 1h 02m 23s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/407c935397894fd3ba1b855be2558eb0 : SUCCESS in 1h 08m 44s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7317513acc2140038039b4579136e042 : SUCCESS in 1h 07m 55s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5bb1a35914714e20be2fa5428cb32c38 : FAILURE in 1h 13m 19s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6450a784a23646cd96e83a37b7eb539a : SUCCESS in 6m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/f3443f6c913e4fcf8b7c10f7b967af00 : SUCCESS in 4m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3f7a9e4f19684374b73bd9b857aaf444 : SUCCESS in 42m 41s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/29196890522a422b9ff929d6ac3cf58d : SUCCESS in 42m 16s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/21051d3e8d6c480fa70e6d2517012f6a : SUCCESS in 54m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/a8a103f9137b40d1a54b90db6f7443fc : FAILURE in 9m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0e00fef8eaab475690ecffe6263d37cb : FAILURE in 20m 09s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/2394bde5bfd643a8bb15472392fb6068 : SUCCESS in 55m 40s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/6494b0e5a3994ccfb6b1dcbbc3c039bd : SUCCESS in 1h 04m 47s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/95741c4001014e058de9b5b5c78cea0b : SUCCESS in 1h 03m 07s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2515808dfa604db08b3bc464dc79c595 : SUCCESS in 40m 56s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/16498381d8c545468b93cedeb10f0154 : SUCCESS in 33m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3472697badb501484a3af77989c2f8dcbfd88cb5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-26 05:24:55.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/46/899346/1/check/DellEMC_Unity_Manila_dhss_true/beae350/DellEMC_Unity_Manila_dhss_true/699 : SUCCESS in 2h 25m 18s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/46/899346/1/check/DellEMC_Unity_Manila_dhss_false/d6f4a2f/DellEMC_Unity_Manila_dhss_false/738 : SUCCESS in 1h 00m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"61c2276fee532c9a1052df46d587d630fa946512","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-26 05:40:09.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-46-899346-1/NFS/ : SUCCESS in 2h 50m 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"c343745d0ac6c77aad2a14433c4a7ffe1a21dfb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2023-10-26 20:23:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dee95fc87ec962fefa32c7ef0c0c2a72350d2c49","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-26 20:58:35.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/46/899346/2/check/DellEMC_PowerFlex_Manila_dhss_false/b33a891/DellEMC_PowerFlex_Manila_dhss_false/602 : SUCCESS in 34m 40s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"116f023c13ea66b97b14526e28a0260664ab262d","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-26 21:34:25.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/46/899346/2/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/ac7d4ea/ : FAILURE in 1h 09m 18s","accounts_in_message":[],"_revision_number":2},{"id":"935fff74e3167139ea47abab93e7cdf33e130446","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-26 22:16:01.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-46-899346-2/NFS/ : SUCCESS in 1h 52m 0s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"2518ee25b216465fb8c643ff180f861743c72f47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-26 22:42:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b833ff61f5b4c72ad654277b3e39ac2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16aa2db097e841ee99135c3b087e471a : SUCCESS in 11m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d358bf87cb404589bfda22b3d0d253cc : SUCCESS in 27m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46910fde93ab4daa8457cffa32f5a0d3 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c122cf94f704b5593b437eb040c4751 : SUCCESS in 10m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd47ec488d7a4725a63d062424ee8a1e : SUCCESS in 9m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/60d0fcd60af94409adcca3bd0a077e4b : SUCCESS in 43m 20s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/8608417c3bce4313abd849284afebac2 : FAILURE in 2h 13m 16s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ea0f42b15e6745c3bb7511fd9a36868e : SUCCESS in 1h 08m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/94b11983db9a48679f96aa246e46807e : SUCCESS in 1h 09m 34s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6d7d916a5362495b97337b7a525fad3c : FAILURE in 1h 06m 39s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2870172adf18463797c68b8f62016cd5 : SUCCESS in 50m 20s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ddad5a6526bd466c9ed67b45cb504797 : SUCCESS in 5m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/80acc240281347a490551aa6f38cb69b : SUCCESS in 3m 48s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/bd98709f7bc14e669016f689938e616c : SUCCESS in 29m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6a0a636f4fb74559b61be7c3d9da56bc : SUCCESS in 31m 50s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7680acd8ff6d4b7ab23da9e106a0bd94 : SUCCESS in 38m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/6db57ea7117346989d24ebf5d2739e93 : FAILURE in 11m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/cad90de486474101811ab5cb7594cbee : FAILURE in 10m 20s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/7884e0db4c644ed28f695da907c1641d : SUCCESS in 57m 12s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4e2000261b5f492d8d4429879cba8ef9 : SUCCESS in 49m 15s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/56e62e9f0a0f4fcc9954c483853268cc : SUCCESS in 1h 03m 12s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/a0f64898b31c401d83b98905cc005b7b : SUCCESS in 50m 23s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3076070829dd4a0d97d530397998c8de : SUCCESS in 36m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2e3eb09250b1d3f99db890de52e63fc9790e5114","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-26 22:52:59.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/46/899346/2/upstream-check/manila-tempest-plugin-ontap-dhss/130afca/ : RETRY_LIMIT in 8m 33s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/46/899346/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/da5e25e/ : RETRY_LIMIT in 8m 12s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"57c311abc3c58740e5ea15334e16cdb7204ea8d8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-26 23:15:08.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/46/899346/2/check/DellEMC_PowerFlex_Manila_dhss_false/7debeaa/DellEMC_PowerFlex_Manila_dhss_false/603 : SUCCESS in 32m 28s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"dc097729aec9f3e67be66b3e76505535a1c5b207","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-10-26 23:21:54.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/46/899346/2/check/DellEMC_PowerStore_Manila_dhss_false/292e88c/DellEMC_PowerStore_Manila_dhss_false/260 : SUCCESS in 39m 14s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"83c52707b0ea64e24aa5571849adb259f53f2f27","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-26 23:41:00.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/46/899346/2/check/DellEMC_Unity_Manila_dhss_true/e94f342/DellEMC_Unity_Manila_dhss_true/703 : SUCCESS in 3h 17m 04s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/46/899346/2/check/DellEMC_Unity_Manila_dhss_false/c260e32/DellEMC_Unity_Manila_dhss_false/742 : SUCCESS in 1h 06m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"8da6cefd9e9f4d107f26cf3f237bc1340b06a22e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2023-10-27 00:59:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b416260069e22562c260ce7970c4dc98e092d092","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-27 01:30:12.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/46/899346/3/check/DellEMC_PowerFlex_Manila_dhss_false/f01abad/DellEMC_PowerFlex_Manila_dhss_false/604 : SUCCESS in 30m 15s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"842b96f842a2f09b7fd910fe04741b1436b2b59e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-27 02:15:52.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/46/899346/3/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/5927fcb/ : FAILURE in 1h 13m 59s","accounts_in_message":[],"_revision_number":3},{"id":"2cf8e8bfe0b6a0237845cb3a4e02cf7556d86292","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-27 02:23:41.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-46-899346-3/NFS/ : SUCCESS in 1h 23m 39s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"8e00171f1cdb6e4f6bf3381d9ac8c40e901cd162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 02:47:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b59bbbfa2344d7c9869c4e001c08956\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76a5b67c99ae4c3da453e77154b9de59 : SUCCESS in 9m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2524031565b047358f7ed868ede23c25 : SUCCESS in 23m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ecba6eabcdf4d1abef903235f0fb43c : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62adecb0f11146e4900f873425bbcb76 : SUCCESS in 11m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b9b9f293470422eb755b810a5a84d2d : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a72e7c26c3914d8cab85a99a99a858a0 : SUCCESS in 6m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f1796843d1bb479da37e1d8080691f8f : SUCCESS in 34m 29s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a023e0cbc29c42229f83f9bdab221c05 : SUCCESS in 1h 46m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e2fe02cf011140b9a4dfc9aaf2a64a33 : SUCCESS in 1h 11m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ec118ae8a61047f882e175725ad70990 : SUCCESS in 1h 07m 40s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/3c03b09e5741463e93d8134fe70f235e : SUCCESS in 1h 29m 13s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/22282f13f6a64a1fb5ebc8bbf5f12b62 : SUCCESS in 1h 30m 29s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/41ad0f63a91f4004b7dabc302a8715c3 : SUCCESS in 5m 10s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/7678b79ce4ad4bd7b1a0d44bf12979a6 : SUCCESS in 5m 02s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/73538166e3c149c49ad2e90e182ebfe5 : SUCCESS in 25m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d04b451d80b3423ba9a46da72519b2c5 : SUCCESS in 42m 47s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f4743c8f4a01400ebfda13c94527a0ab : SUCCESS in 1h 01m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/5ce071774495465f9a22ec99921151a9 : FAILURE in 25m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9d425636f9fe488b9fc7b6fba1292187 : FAILURE in 22m 23s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/a44b1202e7174bb9b97d75603b68d7c6 : SUCCESS in 58m 14s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/b7caba734e8d41948dcc6629b14eb9e6 : SUCCESS in 1h 11m 28s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e07483d2fe054f149bc6b71e4f008913 : SUCCESS in 59m 42s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/879f905ab9544bc090d828b266e162fe : SUCCESS in 30m 37s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/4f048161b4874a7382d54b8f538cc74d : SUCCESS in 35m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cd7c768d4a9866ab5c44572aa0fa2b61250fc6da","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-27 03:02:54.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/46/899346/3/upstream-check/manila-tempest-plugin-ontap-dhss/24516be/ : RETRY_LIMIT in 8m 18s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/46/899346/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/424de75/ : RETRY_LIMIT in 7m 52s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"f3a2cea216110dd4bb9da2e434ab0c80f3a8ff8d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-27 03:17:50.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/46/899346/3/check/DellEMC_PowerFlex_Manila_dhss_false/e10ebe4/DellEMC_PowerFlex_Manila_dhss_false/605 : SUCCESS in 30m 07s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"68329e6362639e265195eea29477863c6bdd1d73","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-10-27 03:24:11.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/46/899346/3/check/DellEMC_PowerStore_Manila_dhss_false/50af7d3/DellEMC_PowerStore_Manila_dhss_false/261 : SUCCESS in 36m 29s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"e6a12cf7d52789292aebefe7b115efd5aee000a5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-27 04:00:31.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/46/899346/3/check/DellEMC_Unity_Manila_dhss_true/b8a82ab/DellEMC_Unity_Manila_dhss_true/704 : SUCCESS in 3h 00m 32s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/46/899346/3/check/DellEMC_Unity_Manila_dhss_false/b138796/DellEMC_Unity_Manila_dhss_false/743 : SUCCESS in 1h 03m 11s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"e5403945237c357654af55baeb17012122070bea","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-27 06:01:55.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c66cd83a5327b94707ebd7f791ce8e3f6b0ee664","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2023-10-27 12:33:28.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"26f7f8a50510d340c413cffef27150d1291f8b67","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-27 13:11:20.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/46/899346/4/check/DellEMC_PowerFlex_Manila_dhss_false/3164900/DellEMC_PowerFlex_Manila_dhss_false/611 : SUCCESS in 31m 59s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"14b497a495f01aa136281715ceb162d82c8f10aa","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-27 13:44:42.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/46/899346/4/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/aa255ff/ : FAILURE in 1h 09m 37s","accounts_in_message":[],"_revision_number":4},{"id":"23ba75ae0f792856b79afc71f5df3a317f4d1d46","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-27 13:55:14.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-46-899346-4/NFS/ : SUCCESS in 1h 21m 25s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"4c6634fa9fe939e62c0498bf43c5ace188fbc652","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 14:36:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2763876ea46b4efc864c9c0f8aee9cf9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80348fff8ba7468f8247ab967231138f : SUCCESS in 10m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fa8d108913446bbb4a6f8fc01d282cb : FAILURE in 23m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcf0eab60bf44e80b56bd02932961352 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/081aee86db874bf6af60ff1fe7c76bc1 : SUCCESS in 10m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3cdd8e3c0c834ed292cc83312fcc2329 : SUCCESS in 12m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70eacd966b0543bb9fa39db248a91fec : SUCCESS in 6m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8c0e8a6aa81e4ac1849c43bc7afcf5b1 : SUCCESS in 34m 31s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7e37e64df0df4bcf952b85d055548777 : SUCCESS in 1h 54m 47s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5049fa7249a445bcaca48a218b20c1dd : SUCCESS in 1h 02m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/d99fdfa480d54d489ae0c47301a2bc07 : FAILURE in 53m 58s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/2653c644d81a4a6da76deee9d1ca32ea : SUCCESS in 1h 16m 25s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/f4ce734f92024d51a1e35cdbbfb61635 : SUCCESS in 58m 24s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f2f9d0471b0741e29d6c058c65814016 : SUCCESS in 4m 32s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b6db47ee84a940858eaa2a7d1fb8d171 : SUCCESS in 4m 27s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0788636ab9434e06a7d4b8e52e1068ee : SUCCESS in 29m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/eabf54d7370b469994bc176f8bf9b95d : SUCCESS in 33m 46s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8190c29262b14925b82d74c419b09281 : SUCCESS in 45m 42s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/dfd8a67d4d224934959c32fdaa9632d7 : FAILURE in 13m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/d7ca365d44b8400a88ec619b50c9e95f : FAILURE in 11m 03s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/6c31dbe3a3c34dffbde32159a160b8f3 : SUCCESS in 39m 23s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/96993f75a06e450290211c36bee79530 : SUCCESS in 49m 44s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b8e0afe810544608ad3f9649971db712 : SUCCESS in 49m 21s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/63e9cee932df437a977f0d6fca6fb081 : SUCCESS in 34m 52s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/5efad0ab15d34c3db74b6ec81dd3360d : SUCCESS in 40m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"21bf1e568a37b29b0051be128102c0f7121d139c","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-27 14:46:51.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/46/899346/4/upstream-check/manila-tempest-plugin-ontap-dhss/b46d6d0/ : RETRY_LIMIT in 8m 21s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/46/899346/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/40eab06/ : RETRY_LIMIT in 9m 39s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"2a429a76195f5f15f8f43c14b7bd4dc35f9a54cf","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-10-27 15:15:07.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/46/899346/4/check/DellEMC_PowerStore_Manila_dhss_false/50d151a/DellEMC_PowerStore_Manila_dhss_false/264 : SUCCESS in 38m 44s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"a81aeb161a11e91e5885adbcc883412eab8a1a75","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-27 15:53:02.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/46/899346/4/check/DellEMC_PowerFlex_Manila_dhss_false/e33ea3d/DellEMC_PowerFlex_Manila_dhss_false/616 : SUCCESS in 31m 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"b00fb5cdc309caceeccc5af8ef0a5b92498edb88","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-27 18:11:22.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b05ff0b6735041876a0f7990e73281aa06f70899","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2023-10-27 18:53:26.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"7e05346a8918996b3b1028ebc6a06454bd7b1eb0","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-27 19:02:43.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/46/899346/4/check/DellEMC_Unity_Manila_dhss_true/3210553/DellEMC_Unity_Manila_dhss_true/708 : SUCCESS in 2h 59m 55s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/46/899346/4/check/DellEMC_Unity_Manila_dhss_false/c8b4e33/DellEMC_Unity_Manila_dhss_false/747 : SUCCESS in 1h 08m 38s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"90be8b4f1634b9ab7d37ee950fdbf3b27f09dfb4","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2023-10-27 20:11:30.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"36c4bbb935761ad04cb9527c4aa303f73014449f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 20:12:02.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"81ba2ace478878d04c3dde71be11be703d4ccef4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 20:39:50.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35211a1d8ea34f4d94f4181044bd0c72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76e20ceb5ed64478b21a84949d19413b : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6be7d78437a40938ef35a818a6828b3 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9429f55464314212a8e2edd4adaf81df : SUCCESS in 10m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b71c8e6b7cb94e538ed36ca5d0753301 : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8aef8a4a4c564487b4c52ab87cef5fae : SUCCESS in 6m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/85dcc241af684f21b63c0439efe79fde : SUCCESS in 27m 13s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/081ef26406734c61b8d93fcc89bbc0db : SUCCESS in 25m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a2a3cfa196b44df68f29f2975b215a39 : SUCCESS in 26m 01s","accounts_in_message":[],"_revision_number":4},{"id":"9682dc4e94bc822c58389c03a74a8ebf6c4744bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 20:39:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"ab5f925a05c547fe08f4625e291b1551585385a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 20:41:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89a0f87ff7ec4cbbb322953e0a22e62f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5161a361588d402e932617724f81197b : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f68fdaafbd884cc980ef8403d79ba757 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":4},{"id":"c69fcbb4b3198519333b70817e38d11836e3e055","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-27 22:13:35.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff","revisions":{"aaf315bae8e80092c896fb69dd909e589893ade6":{"kind":"REWORK","_number":1,"created":"2023-10-26 00:11:20.000000000","uploader":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"ref":"refs/changes/46/899346/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/46/899346/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/46/899346/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/46/899346/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/46/899346/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/46/899346/1"}}},"commit":{"parents":[{"commit":"5d5ece51eac9fed78373d1a43f24a6112765df93","subject":"update irc dcoumentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5d5ece51eac9fed78373d1a43f24a6112765df93"}]}],"author":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-26 00:11:03.000000000","tz":180},"committer":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-26 00:11:03.000000000","tz":180},"subject":"change status and error when /shares api fails to handle share_type in request","message":"change status and error when /shares api fails to handle share_type in request\n\nChange-Id: I8d4b30daae2fe8c88c30d93d402bf2e5a558f804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/aaf315bae8e80092c896fb69dd909e589893ade6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/aaf315bae8e80092c896fb69dd909e589893ade6"}]},"branch":"refs/heads/master"},"9002696777e35aaa66c4c7ec6faa8dd497f20247":{"kind":"REWORK","_number":2,"created":"2023-10-26 20:23:39.000000000","uploader":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"ref":"refs/changes/46/899346/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/46/899346/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/46/899346/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/46/899346/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/46/899346/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/46/899346/2"}}},"commit":{"parents":[{"commit":"5d5ece51eac9fed78373d1a43f24a6112765df93","subject":"update irc dcoumentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5d5ece51eac9fed78373d1a43f24a6112765df93"}]}],"author":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-26 00:11:03.000000000","tz":180},"committer":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-26 20:23:03.000000000","tz":180},"subject":"Change status and error handling for /shares API","message":"Change status and error handling for /shares API\n\nThis change modifies the status and error handling logic of the /shares API when it fails to handle the specified share_type in the request. The updated logic ensures that appropriate responses are generated to handle this scenario effectively.\n\nThe modifications include updating the error messages and status codes returned by the API to provide clearer information to the users. Additionally, the internal error handling mechanisms have been enhanced to improve error reporting and troubleshooting.\n\nCloses-Bug: #1944478\n\nChange-Id: I8d4b30daae2fe8c88c30d93d402bf2e5a558f804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9002696777e35aaa66c4c7ec6faa8dd497f20247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9002696777e35aaa66c4c7ec6faa8dd497f20247"}]},"branch":"refs/heads/master"},"908c8758139fd282cc5a1320b6d1b1a29b957c12":{"kind":"REWORK","_number":3,"created":"2023-10-27 00:59:42.000000000","uploader":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"ref":"refs/changes/46/899346/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/46/899346/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/46/899346/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/46/899346/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/46/899346/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/46/899346/3"}}},"commit":{"parents":[{"commit":"5d5ece51eac9fed78373d1a43f24a6112765df93","subject":"update irc dcoumentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5d5ece51eac9fed78373d1a43f24a6112765df93"}]}],"author":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-26 00:11:03.000000000","tz":180},"committer":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-27 00:59:10.000000000","tz":180},"subject":"Change status and error handling for /shares API","message":"Change status and error handling for /shares API\n\nThis change modifies the status and error handling logic of the /shares API when it fails to handle the specified share_type in the request. The updated logic ensures that appropriate responses are generated to handle this scenario effectively.\n\nThe modifications include updating the error messages and status codes returned by the API to provide clearer information to the users. Additionally, the internal error handling mechanisms have been enhanced to improve error reporting and troubleshooting.\n\nCloses-Bug: #1944478\n\nChange-Id: I8d4b30daae2fe8c88c30d93d402bf2e5a558f804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/908c8758139fd282cc5a1320b6d1b1a29b957c12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/908c8758139fd282cc5a1320b6d1b1a29b957c12"}]},"branch":"refs/heads/master"},"b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-10-27 12:33:28.000000000","uploader":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"ref":"refs/changes/46/899346/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/46/899346/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/46/899346/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/46/899346/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/46/899346/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/46/899346/4"}}},"commit":{"parents":[{"commit":"5d5ece51eac9fed78373d1a43f24a6112765df93","subject":"update irc dcoumentation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5d5ece51eac9fed78373d1a43f24a6112765df93"}]}],"author":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-26 00:11:03.000000000","tz":180},"committer":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-27 12:26:40.000000000","tz":180},"subject":"Change status and error handling for /shares API","message":"Change status and error handling for /shares API\n\nThis change modifies the status and error handling\nlogic of the /shares API when it fails to handle\nthe specified share_type in the request. The updated\nlogic ensures that appropriate responses are\ngenerated to handle this scenario effectively.\n\nCloses-Bug: #1944478\n\nChange-Id: I8d4b30daae2fe8c88c30d93d402bf2e5a558f804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
