)]}'
{"id":"openstack%2Fmanila~899476","triplet_id":"openstack%2Fmanila~master~I5fa9419cef7292834300c4c23b4a21b76eb00f27","project":"openstack/manila","branch":"master","topic":"jipheens","attention_set":{},"removed_from_attention_set":{"36488":{"account":{"_account_id":36488,"name":"Jipheens wahome","display_name":"JIPHEENS","email":"wahomejipheens@gmail.com","username":"jipheens-wahome"},"last_update":"2023-10-30 22:13:18.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5fa9419cef7292834300c4c23b4a21b76eb00f27","subject":"HTTP 500 raised by POST /shares API when failing to handle share_type in request","status":"ABANDONED","created":"2023-10-27 13:14:41.000000000","updated":"2023-10-30 22:13:18.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"70eba65eefeccce3eec172dc17eca7a94807dad2","_number":899476,"virtual_id_number":899476,"owner":{"_account_id":36488,"name":"Jipheens wahome","display_name":"JIPHEENS","email":"wahomejipheens@gmail.com","username":"jipheens-wahome"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-10-27 13:38:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2023-10-28 14:25:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-27 13:38:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-27 13:46:44.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-27 14:27:12.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-10-27 14:44:01.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-10-27 22:01:24.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-10-28 14:25:06.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"b50719cb2561953d8fa592a8762c51f49c4e4fa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36488,"name":"Jipheens wahome","display_name":"JIPHEENS","email":"wahomejipheens@gmail.com","username":"jipheens-wahome"},"date":"2023-10-27 13:14:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b4b689f4d549d999dcedd0d970272cba39b94dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 13:38:45.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5706a652aadd4628be1cc470592e6f0b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb1c21a073574285bcd5791c8a271583 : SUCCESS in 10m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3529c66eec974367922cf5829c29e052 : FAILURE in 21m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cf3629aceeb48fab744a47703867897 : FAILURE in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fa530c8005c42e9be086124d7be0cb3 : FAILURE in 10m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a763671dfb0443aa564d4905f42f165 : FAILURE in 9m 05s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/17f9182d76074bf78867a498156c9794 : SUCCESS in 4m 35s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/e2e75f62e13042b68d86c7e9f4832936 : SUCCESS in 4m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cced4abb63f4ac93cb551faa15726ba5dd8ff277","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-27 13:46:44.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/76/899476/1/check/DellEMC_PowerFlex_Manila_dhss_false/738cedc/DellEMC_PowerFlex_Manila_dhss_false/612 : SUCCESS in 31m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"00d8a4d3214bd5e81c161fd1d3a31999951b4212","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-27 14:27:12.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/76/899476/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/5f6f162/ : SUCCESS in 1h 10m 12s","accounts_in_message":[],"_revision_number":1},{"id":"f194d90e0578efb97c47542dddacb251863d265a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-27 14:44:01.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-76-899476-1/NFS/ : SUCCESS in 1h 28m 57s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"3a4c924d23616cf97edace21713aa549b9b0365b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-27 22:01:24.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/76/899476/1/check/DellEMC_Unity_Manila_dhss_true/2cc2c00/DellEMC_Unity_Manila_dhss_true/709 : SUCCESS in 2h 58m 41s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/76/899476/1/check/DellEMC_Unity_Manila_dhss_false/fa61d4f/DellEMC_Unity_Manila_dhss_false/748 : SUCCESS in 1h 00m 10s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"a03bd1b7f76b587f563741f84920b9095f193957","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-28 14:25:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"70eba65eefeccce3eec172dc17eca7a94807dad2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-30 22:13:18.000000000","message":"Abandoned\n\nIncorrect submission; please see comments on other patches you have submitted","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ef6b7bef5d1a03351c0db00945f499f7981189d","revisions":{"9ef6b7bef5d1a03351c0db00945f499f7981189d":{"kind":"REWORK","_number":1,"created":"2023-10-27 13:14:41.000000000","uploader":{"_account_id":36488,"name":"Jipheens wahome","display_name":"JIPHEENS","email":"wahomejipheens@gmail.com","username":"jipheens-wahome"},"ref":"refs/changes/76/899476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/76/899476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/899476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/899476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/76/899476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/76/899476/1"}}},"commit":{"parents":[{"commit":"9812983bbb2583b17e62f87f65214be9e379632d","subject":"Merge \"[Pure Storage] Fix connection issue on network failure.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9812983bbb2583b17e62f87f65214be9e379632d"}]}],"author":{"name":"jipheens-wahome","email":"wahomejipheens@gmail.com","date":"2023-10-27 13:07:21.000000000","tz":180},"committer":{"name":"jipheens-wahome","email":"wahomejipheens@gmail.com","date":"2023-10-27 13:14:17.000000000","tz":180},"subject":"HTTP 500 raised by POST /shares API when failing to handle share_type in request","message":"HTTP 500 raised by POST /shares API when failing to handle share_type in request\n\nChange-Id: I5fa9419cef7292834300c4c23b4a21b76eb00f27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9ef6b7bef5d1a03351c0db00945f499f7981189d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9ef6b7bef5d1a03351c0db00945f499f7981189d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
