)]}'
{"id":"openstack%2Fmanila~899477","triplet_id":"openstack%2Fmanila~master~I121d6dcbf06ec962f315e66ff25a6e203f9d8e7d","project":"openstack/manila","branch":"master","topic":"jipheens","attention_set":{},"removed_from_attention_set":{"36488":{"account":{"_account_id":36488,"name":"Jipheens wahome","display_name":"JIPHEENS","email":"wahomejipheens@gmail.com","username":"jipheens-wahome"},"last_update":"2023-10-30 22:12:35.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I121d6dcbf06ec962f315e66ff25a6e203f9d8e7d","subject":"HTTP 500 raised by POST /shares API when failing to handle share_type in request","status":"ABANDONED","created":"2023-10-27 13:41:54.000000000","updated":"2023-10-30 22:12:35.000000000","total_comment_count":5,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"e0409e4e5640932dd986159fbab25dcfafa2bee7","_number":899477,"virtual_id_number":899477,"owner":{"_account_id":36488,"name":"Jipheens wahome","display_name":"JIPHEENS","email":"wahomejipheens@gmail.com","username":"jipheens-wahome"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-10-27 15:48:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2023-10-28 14:27:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-27 14:18:26.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-27 14:51:45.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-10-27 15:39:43.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-10-27 15:48:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-28 02:21:43.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-10-28 14:27:35.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"ad80be2135eacf1dbb58862a994852b575b32271","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36488,"name":"Jipheens wahome","display_name":"JIPHEENS","email":"wahomejipheens@gmail.com","username":"jipheens-wahome"},"date":"2023-10-27 13:41:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3115c637adaf80046172dc449f67003ad9f07bc7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-27 14:18:26.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/77/899477/1/check/DellEMC_PowerFlex_Manila_dhss_false/3d73d69/DellEMC_PowerFlex_Manila_dhss_false/613 : FAILURE in 31m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"1255b6115191acdbec1f74ab321e4e69bc119628","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-27 14:51:45.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/77/899477/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/00f7d76/ : FAILURE in 1h 08m 35s","accounts_in_message":[],"_revision_number":1},{"id":"9562e486c88fbcb2aef7bae3dee3983c4f7db02b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-27 15:39:43.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-77-899477-1/NFS/ : FAILURE in 1h 54m 39s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"ce82e93b6af263af0d730852e3d607f9213e16a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 15:48:50.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f24ffd476a3c45df956c7c3a8d19fba0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df54067a45a9459bbafd2532b68a2350 : SUCCESS in 10m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a34fedeced734c93b3597c737131d98b : FAILURE in 19m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c69910e5209d4202820feb7d44210278 : FAILURE in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f324e2a5c51e47c4a461167c08a4639f : FAILURE in 9m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae648fb5934e431b9b8b081a9de968eb : FAILURE in 9m 20s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/46e9b7fcf97d44708809e06c7276205f : FAILURE in 37m 07s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/0f8e2d8fda5147938a3618e086c6a635 : SUCCESS in 2h 03m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2f8690c3a38d45e89f8cdca92e175d01 : FAILURE in 1h 23m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b672f925577a4e42b4f9cc63409e8e66 : FAILURE in 56m 42s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ace1c58c5a9a415da2707b1581f16f36 : FAILURE in 1h 19m 55s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/971201d43320426e9baa0baa123065ae : FAILURE in 1h 05m 43s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/22b7adacdd374bd586f0d2861090cb93 : SUCCESS in 5m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/342cbe79949d4d5fadd1d1060814ae12 : SUCCESS in 4m 10s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1e628417303c434f87981fe1d50bc109 : FAILURE in 26m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b4c00f2aed4f4bebabb00f530b7708b5 : FAILURE in 32m 13s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6f641dc3b498445da2eae711d93cd6ac : SUCCESS in 40m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/a9677d516f1a4fbca493216dc4c5bbcd : FAILURE in 11m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/f0bb104ca86a45ae886f13e09171c72a : FAILURE in 11m 27s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/f1126460e11f4f9290db98ff28cb9bda : FAILURE in 44m 40s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/8b6c633cd6b747e1be483657a51d3f62 : FAILURE in 50m 50s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fdc9543b3f62460d8cfebcfe460c32ba : FAILURE in 54m 04s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/65eed835d70c4b4988c9c30b70a04621 : SUCCESS in 31m 48s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/2809c3508eb34e10a974fe5b10668bf2 : SUCCESS in 36m 13s (non-voting)\n\nWarning:\n  Comments left for invalid file manila/tests/api/v2/test_shares.py","accounts_in_message":[],"_revision_number":1},{"id":"78bb117159091f8d44058add1da4a9e3e0cb7fb6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-28 02:21:43.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/77/899477/1/check/DellEMC_Unity_Manila_dhss_true/23c6987/DellEMC_Unity_Manila_dhss_true/710 : FAILURE in 4h 20m 17s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/77/899477/1/check/DellEMC_Unity_Manila_dhss_false/a797f36/DellEMC_Unity_Manila_dhss_false/749 : FAILURE in 1h 06m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"d41801183080c542309c965cc9335b714da725dc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-28 14:27:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"32296eb580c751df7e6fd4ebfbbeaca5ab9e6b98","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-28 14:30:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0409e4e5640932dd986159fbab25dcfafa2bee7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-30 22:12:35.000000000","message":"Abandoned\n\nIncorrect submission; please see comments on other patches you\u0027ve submitted","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b0e9172a9fc3ba88b8ead32b03e3f932175e9a36","revisions":{"b0e9172a9fc3ba88b8ead32b03e3f932175e9a36":{"kind":"REWORK","_number":1,"created":"2023-10-27 13:41:54.000000000","uploader":{"_account_id":36488,"name":"Jipheens wahome","display_name":"JIPHEENS","email":"wahomejipheens@gmail.com","username":"jipheens-wahome"},"ref":"refs/changes/77/899477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/77/899477/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/77/899477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/77/899477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/77/899477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/77/899477/1"}}},"commit":{"parents":[{"commit":"9ef6b7bef5d1a03351c0db00945f499f7981189d","subject":"HTTP 500 raised by POST /shares API when failing to handle share_type in request","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9ef6b7bef5d1a03351c0db00945f499f7981189d"}]}],"author":{"name":"jipheens-wahome","email":"wahomejipheens@gmail.com","date":"2023-10-27 13:40:49.000000000","tz":180},"committer":{"name":"jipheens-wahome","email":"wahomejipheens@gmail.com","date":"2023-10-27 13:41:12.000000000","tz":180},"subject":"HTTP 500 raised by POST /shares API when failing to handle share_type in request","message":"HTTP 500 raised by POST /shares API when failing to handle share_type in request\n\nChange-Id: I121d6dcbf06ec962f315e66ff25a6e203f9d8e7d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b0e9172a9fc3ba88b8ead32b03e3f932175e9a36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b0e9172a9fc3ba88b8ead32b03e3f932175e9a36"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
