)]}'
{"id":"openstack%2Fmanila~899482","triplet_id":"openstack%2Fmanila~master~I343eb4e0b617f111979092ee54641afe41804254","project":"openstack/manila","branch":"master","topic":"jipheens","attention_set":{},"removed_from_attention_set":{"36488":{"account":{"_account_id":36488,"name":"Jipheens wahome","display_name":"JIPHEENS","email":"wahomejipheens@gmail.com","username":"jipheens-wahome"},"last_update":"2023-10-30 22:12:57.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I343eb4e0b617f111979092ee54641afe41804254","subject":"HTTP500 raised by POST /shares API when failing to handle share_type in request","status":"ABANDONED","created":"2023-10-27 13:57:29.000000000","updated":"2023-10-30 22:12:57.000000000","total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"41b073a018bbc3ee7e5a6509075d63321fe18859","_number":899482,"virtual_id_number":899482,"owner":{"_account_id":36488,"name":"Jipheens wahome","display_name":"JIPHEENS","email":"wahomejipheens@gmail.com","username":"jipheens-wahome"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-10-27 15:53:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2023-10-28 14:29:37.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-27 14:49:24.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-27 15:06:33.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-10-27 15:53:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-27 16:06:04.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-10-28 06:13:16.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-10-28 14:29:37.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"30f075535acffdaaa5ab75b6d682da167ad00e7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36488,"name":"Jipheens wahome","display_name":"JIPHEENS","email":"wahomejipheens@gmail.com","username":"jipheens-wahome"},"date":"2023-10-27 13:57:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf18e94db818f5d09920c1ae8c77762ba4769422","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-27 14:49:24.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/82/899482/1/check/DellEMC_PowerFlex_Manila_dhss_false/541244d/DellEMC_PowerFlex_Manila_dhss_false/614 : FAILURE in 30m 51s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"9cebc1cc99334380ac33688245025cf66cb9f16a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-27 15:06:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/82/899482/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/c30980f/ : FAILURE in 1h 06m 44s","accounts_in_message":[],"_revision_number":1},{"id":"660727d6bac627df7cdf1b2080c14e2611922998","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 15:53:53.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a66a5aaba6da4758af6837a8f8932635\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67f3da30e79d4fe3a1399b65e12f08b3 : SUCCESS in 10m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df8b896dad4c4742a25264d6b61b778d : FAILURE in 28m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cec028f9d4c6429582a6a4976b1762ef : FAILURE in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a78500b1a9d84f2799db2ff6427f170b : FAILURE in 10m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/858aec14523e416fbae8aa4f660275fd : FAILURE in 9m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5e53d622522d49b1883a36d7b1413bbb : FAILURE in 30m 21s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7c2280b2789e4b2facd827747a552c02 : SUCCESS in 1h 52m 46s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6d8789b0199f4ac4beba425cf8cd3fd9 : FAILURE in 1h 19m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7f61ac90840d44729c1dc3b59413f316 : FAILURE in 55m 23s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e7571fec4ada49b986a581db908372be : FAILURE in 1h 23m 56s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/62e52f5ae6cc4c618cdf404952ed9965 : FAILURE in 1h 04m 00s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/45d3053b70c84258b7cd72b39d8b1c06 : SUCCESS in 5m 11s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fff9834bde854fdaa24e960fad37a06d : FAILURE in 4m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/dc60a8dcb578442f973b5a1d7d8f0b3b : FAILURE in 32m 13s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/09e87f89739c43209bf8df79966f1034 : FAILURE in 35m 42s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6f5e2ce7a78742cda081edaac17c43cd : FAILURE in 19m 53s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/1e44c34be8f04d6da4560fb154b30a85 : FAILURE in 11m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/441be60ed0f84e6ca113371413c42a7e : FAILURE in 11m 19s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/93df7466389347c2b5ca801553fa0dc8 : FAILURE in 43m 44s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4be7a7b4f4174cc0b86f693bb24bb2a3 : FAILURE in 55m 29s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/11710c0773614f3fbe891fb38ae13e4d : FAILURE in 50m 15s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6996a8d3c00c419f8ec1f5e879d6d1d3 : SUCCESS in 32m 17s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/04b4b9f09b724e24b8423d8856d22560 : SUCCESS in 37m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0ddcc673206f50f72a640efd6113cfc12fda8dfb","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-27 16:06:04.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-82-899482-1/NFS/ : FAILURE in 1h 51m 0s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"619474b8b8a7bdea2937a1421bd2fc0688d60c32","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-28 06:13:16.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/82/899482/1/check/DellEMC_Unity_Manila_dhss_true/abcca9a/DellEMC_Unity_Manila_dhss_true/711 : FAILURE in 3h 51m 34s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/82/899482/1/check/DellEMC_Unity_Manila_dhss_false/f5ba9e6/DellEMC_Unity_Manila_dhss_false/750 : FAILURE in 1h 02m 59s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"f4cd585cb834fe12d70c7aa3a9e00dd8913839aa","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-28 14:29:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"41b073a018bbc3ee7e5a6509075d63321fe18859","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-30 22:12:57.000000000","message":"Abandoned\n\nIncorrect submission; please see comments on other patches you\u0027ve submitted","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f85fe228f581c6cc7499a8ad60fe42b48ca88e29","revisions":{"f85fe228f581c6cc7499a8ad60fe42b48ca88e29":{"kind":"REWORK","_number":1,"created":"2023-10-27 13:57:29.000000000","uploader":{"_account_id":36488,"name":"Jipheens wahome","display_name":"JIPHEENS","email":"wahomejipheens@gmail.com","username":"jipheens-wahome"},"ref":"refs/changes/82/899482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/82/899482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/899482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/899482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/82/899482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/82/899482/1"}}},"commit":{"parents":[{"commit":"b0e9172a9fc3ba88b8ead32b03e3f932175e9a36","subject":"HTTP 500 raised by POST /shares API when failing to handle share_type in request","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b0e9172a9fc3ba88b8ead32b03e3f932175e9a36"}]}],"author":{"name":"jipheens-wahome","email":"wahomejipheens@gmail.com","date":"2023-10-27 13:55:57.000000000","tz":180},"committer":{"name":"jipheens-wahome","email":"wahomejipheens@gmail.com","date":"2023-10-27 13:55:57.000000000","tz":180},"subject":"HTTP500 raised by POST /shares API when failing to handle share_type in request","message":"HTTP500 raised by POST /shares API when failing to handle share_type in request\n\nChange-Id: I343eb4e0b617f111979092ee54641afe41804254\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f85fe228f581c6cc7499a8ad60fe42b48ca88e29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f85fe228f581c6cc7499a8ad60fe42b48ca88e29"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
