)]}'
{"id":"openstack%2Fmanila~899513","triplet_id":"openstack%2Fmanila~master~I168564a5b054d17762ad668ebbe4f5e7b562197b","project":"openstack/manila","branch":"master","topic":"filter-message-fix","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-01-31 11:44:47.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"36483":{"account":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"last_update":"2024-01-31 13:00:22.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-01-30 23:43:43.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2024-01-31 08:26:40.000000000","reason":"\u003cGERRIT_ACCOUNT_30407\u003e replied on the change","reason_account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}},"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2024-01-31 13:00:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I168564a5b054d17762ad668ebbe4f5e7b562197b","subject":"Change misleading user message when user services are down","status":"MERGED","created":"2023-10-27 23:14:03.000000000","updated":"2024-02-01 21:43:59.000000000","submitted":"2024-01-31 13:00:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":53,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899513-filter-message-fix","meta_rev_id":"832512268a5eacc6c231d51e99653f2352bcf325","_number":899513,"virtual_id_number":899513,"owner":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-31 13:00:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-01-31 11:44:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2024-01-30 23:43:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2024-01-31 08:26:40.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-31 04:23:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2024-01-31 11:44:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-27 23:29:45.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2023-10-27 23:29:51.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2023-10-27 23:43:51.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-28 00:26:56.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-10-28 00:57:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-28 01:07:22.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-10-28 01:07:36.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-10-28 01:34:17.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-10-29 03:26:30.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-01-17 00:40:04.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2024-01-24 08:14:46.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2024-01-26 04:53:38.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2024-01-31 04:23:47.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"5b6f4818637a4a79a4d49868c7a567f38c64a6a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2023-10-27 23:14:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d86a4532dcfb4a3d011144e7d5ea57045472ad3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-27 23:29:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a2ba9a93b6e1c40d81c09616d2b9b9fbd7afba0","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-27 23:29:51.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"75e9e6269e071704b946bfcaccb84324923f05f4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-27 23:43:51.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/899513/1/check/DellEMC_PowerFlex_Manila_dhss_false/3bc9f53/DellEMC_PowerFlex_Manila_dhss_false/623 : SUCCESS in 29m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"8df95289c4ffd94387f15b37fab005b519b20883","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-28 00:26:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/13/899513/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/795f00d/ : FAILURE in 1h 10m 56s","accounts_in_message":[],"_revision_number":1},{"id":"74e91068dc21127bab8ab4de1be12a3a45c091cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-28 00:57:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7d874e4ff09431eb33353953dfcd38e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a01402dc8e444ed890dd78ef218b4a15 : SUCCESS in 10m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5ae982682f049f88dff9d16c6a6829c : FAILURE in 31m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2242f24eaada433296ae5a8263dcbf61 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f2f596c608c46d2a1114cce6a54bc50 : SUCCESS in 15m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea32e295007b4b1b9ddf2b585d5b7126 : SUCCESS in 10m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57a93833fed34bdbbfc5119062a8b7a7 : SUCCESS in 7m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b4fa731d7cda4f87834920f289e38b8f : SUCCESS in 33m 44s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f9a5f8fb017d475295e20b9016b4cfd9 : SUCCESS in 1h 42m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5928e68515404577b7cfe9e13f22b67d : SUCCESS in 59m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/71e37f98ead44f82a4b2d683d2fae716 : SUCCESS in 51m 19s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/0718bcc134ae4a7d88708eb81d73df76 : SUCCESS in 1h 06m 10s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d400d89abc14448b8c6a697f8e783e8d : SUCCESS in 57m 07s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8285b4a99c234cefa60faa2700f964e2 : SUCCESS in 5m 38s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/da1eaadcc6044fe5ba97aec595d39cae : SUCCESS in 5m 36s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cdd727e748374b7abef5097ad0265e64 : SUCCESS in 30m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d570a235d4b040608e357558f89b01d3 : SUCCESS in 34m 57s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ae34e5df19b64877bb11c724434ec293 : SUCCESS in 40m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/e5fbc359a61a4fde9d5e7c4a464b88c9 : FAILURE in 13m 07s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7a124b406aed4d3c81ea05e0ca8c01c7 : FAILURE in 13m 53s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/77fb784ec666492bb5f1f066c68993f0 : SUCCESS in 41m 32s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/258200edcbea4e5b9ed386dc0106d65d : SUCCESS in 53m 30s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/af69360a58de49e890b95cc9fe69f74d : SUCCESS in 51m 12s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/5a2ee7181e0e4a35815a02ef0d7a0685 : SUCCESS in 22m 53s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/9cc7a64720ba4099a0d5b71c43ca4517 : SUCCESS in 34m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c8fbcf43b723847655cbaece40a6bdb29595beb","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-28 01:07:22.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/899513/1/upstream-check/manila-tempest-plugin-ontap-dhss/5fcc091/ : RETRY_LIMIT in 7m 36s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/899513/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/c2e9b94/ : RETRY_LIMIT in 8m 14s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"6c129b00b8f5111eab00cef5cae29056ae7be271","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-28 01:07:36.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-899513-1/NFS/ : SUCCESS in 1h 53m 11s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"4b6c0edd8a3e8a63639e5f2951c94e3f7776dfa8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-28 01:31:55.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/899513/1/check/DellEMC_PowerFlex_Manila_dhss_false/6385c6f/DellEMC_PowerFlex_Manila_dhss_false/624 : SUCCESS in 33m 48s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"f34e15819a9a330eaead3bf4365366591a83bce1","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-10-28 01:34:17.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/899513/1/check/DellEMC_PowerStore_Manila_dhss_false/aa1a99e/DellEMC_PowerStore_Manila_dhss_false/266 : SUCCESS in 36m 11s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"29a86dd8fe0733d129c09cf92ce659db2cc669cc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-29 03:26:30.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/899513/1/check/DellEMC_Unity_Manila_dhss_true/cb02545/DellEMC_Unity_Manila_dhss_true/717 : SUCCESS in 4h 15m 00s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/899513/1/check/DellEMC_Unity_Manila_dhss_false/d07196b/DellEMC_Unity_Manila_dhss_false/756 : SUCCESS in 1h 09m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"888ac98a1c6adf446841d2b0394d1e35d4230a2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2023-10-29 19:20:42.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2f9c3a970966dd13805a48633a377172da163312","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2023-10-29 19:35:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ddddf9b5fb7ad2b053cbbc0119eed63c6e6ed4e6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-29 20:04:33.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/899513/3/check/DellEMC_PowerFlex_Manila_dhss_false/b1786b5/DellEMC_PowerFlex_Manila_dhss_false/635 : SUCCESS in 28m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"044e24e1b931b0445fd809b3afc3a4960182272f","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2023-10-29 20:14:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b6b14756316651b61ccf64b53232cb2c8e82fe56","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-29 20:49:45.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/13/899513/3/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/2efc6fc/ : FAILURE in 1h 12m 29s","accounts_in_message":[],"_revision_number":3},{"id":"e51670a64bb330595fb47455daf960d601ae995e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-29 21:07:08.000000000","message":"Patch Set 2:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-899513-2/NFS/ : FAILURE in 1h 46m 8s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"ba01ec0562b15985bcc354a7a86954d8e72bcbf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-29 21:11:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0966cbbf6814ae387d2fb6a263ef318\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f2ad14e3cc04d498523d4c160780abe : SUCCESS in 11m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11968b77cb5343e8af09c6de62c67dae : FAILURE in 21m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdc81df935344a2fb9f49f46a0a22b20 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc83519c884f40d6b856b8ba5105d325 : FAILURE in 13m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ddef931130b4980adf2d882c0dfd02d : FAILURE in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a3923464f604a0db8013eff5c1b16ad : SUCCESS in 6m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a8f4c4b0b7784f0a805edb6eb9b01b67 : SUCCESS in 26m 00s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d5d90b5a28814c61b50f4b7796f7210b : SUCCESS in 1h 33m 04s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a76dc64516d142aea1f5025027c3bf7d : SUCCESS in 1h 01m 32s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ae8fc98f18fe4fde97d4f31c33c58740 : SUCCESS in 51m 48s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e60a0d0879544a92a240c549585e43ef : SUCCESS in 1h 15m 28s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/90127f089a8a454491ebb13e060f4303 : FAILURE in 55m 12s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/cb75dc92ce9c4bffbd8494dbaef41028 : SUCCESS in 5m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3fb4e44758894d6cafd856c71bf5045f : SUCCESS in 5m 03s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/dca0add8e3c84008bea4419815c8207c : SUCCESS in 26m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/530e19bbbc664860ab35893dfee05533 : SUCCESS in 32m 48s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/48b0ec08d2424a6d83a13fdf19256504 : SUCCESS in 39m 53s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/a2dd25901bff4d4eb377d247cb0b2749 : FAILURE in 12m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/1a55725a837c4d25b8967273fca695d8 : FAILURE in 12m 46s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/84c63cc25e0d436db58cb9dc1b1160ef : SUCCESS in 35m 58s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/9a76150a454548c7a0b3dc71b4ae4bd2 : SUCCESS in 46m 50s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/762f115bbecf4fe0a136fd832616db83 : SUCCESS in 40m 51s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/cb59624fa98c4e27b3f3ecd326f52147 : SUCCESS in 31m 58s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/6b70345f72cc4c3e90b08f433bd34ad4 : SUCCESS in 30m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c41a62b61ad86fda70cacb2bf2f12b4e6a7cf5b1","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-29 21:36:30.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-899513-3/NFS/ : SUCCESS in 1h 45m 29s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"90d76a6dfb97b64aa21be98d1822ff4794be3290","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-30 03:02:17.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/899513/3/check/DellEMC_Unity_Manila_dhss_true/0ec99c2/DellEMC_Unity_Manila_dhss_true/723 : SUCCESS in 4h 25m 12s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/899513/3/check/DellEMC_Unity_Manila_dhss_false/4b4ae3d/DellEMC_Unity_Manila_dhss_false/766 : SUCCESS in 58m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"5bc1d9c64dd673d43f79188b0fdd1fabff901420","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-31 18:33:46.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"81ecef64eee185be9c9895577b11c63ac7ccad94","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2023-11-07 22:52:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"20c7f9005f2bd8546879ca4dcd1b270bc808aa9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2023-11-08 01:24:45.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"568deaff9188c03712028717637ff775e72d25b1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-08 02:39:10.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/13/899513/4/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/feabd0a/ : FAILURE in 1h 12m 33s","accounts_in_message":[],"_revision_number":4},{"id":"d30fdcb2f593446ebd80e9a1264fee8ea8d87f2a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-08 03:20:30.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-899513-4/NFS/ : SUCCESS in 1h 55m 23s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"e21b161ee69239b2f32cc40a2f3ffbe283e617c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 03:24:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cef034f4ce8f4c42b82477146894cf02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a623b5d2f8a84288accc499c4fe2f877 : SUCCESS in 10m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a77be19b463c4647b2ebb054eaa3944f : FAILURE in 25m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/116c11feacef4429a117ab00b250439a : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/584c4660bc6245d79fa1fd98895bf583 : FAILURE in 9m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/768e9ae06bb54ba6b3a78958ccc3c1a1 : FAILURE in 9m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86700764a81f4be1b35d2aa633958ca3 : SUCCESS in 6m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f94cfdfd35c24b9fb2914f5d02f7f9bf : SUCCESS in 1h 16m 42s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/cbce508ff1e94385b984ec7c7d4ae891 : FAILURE in 1h 58m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6584414749b549d881f27acab7f6dfe7 : SUCCESS in 1h 07m 44s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9a79714e543842a0b68bc905f8ba0ce7 : FAILURE in 57m 29s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4d0c8b70a7ca45f3aff3dd913c5cc922 : SUCCESS in 1h 31m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/65de90a566c546dea2dacb941d9535ef : FAILURE in 1h 04m 57s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d66593812245448a9655d040d56a2a09 : SUCCESS in 4m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/86c6268d494e4ee39911b9c99a9fdfc3 : SUCCESS in 5m 02s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/acfee7a2d7df4ae5afc1aa7188144a65 : SUCCESS in 26m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f77f49fd0b2944f5b963c51db267d3b3 : SUCCESS in 31m 56s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/78eee06dc3614ff7ab61ec812092fe57 : FAILURE in 47m 04s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/78c748c09c5f4f72a014fa9ec7f3bedd : FAILURE in 24m 08s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/7c82859805594dfba2c93fa44345583f : FAILURE in 24m 34s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/455a5fe917ba4361a341d67b4e0bf50f : SUCCESS in 1h 41m 45s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/66f627d1c91c4cd79ae78c79b638fc10 : SUCCESS in 1h 00m 10s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5696ad6a32cf4fcfa1e42da2ff5e9048 : SUCCESS in 1h 05m 36s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/03e7444ae5854ec385c12934b2aa10fa : SUCCESS in 46m 20s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/418a678498e64332a356acf74a3d6dae : SUCCESS in 52m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"915a2b6cd7b4726b6b8c9b2d27d0c7740291cacb","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-11-08 22:44:57.000000000","message":"Patch Set 4: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"c444a5873031da115ae8dc669c964d915e1dd2ec","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-11-08 22:49:35.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"051fe083bfb856234498450b93ea9d9d44999deb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2023-11-09 02:56:57.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"bb8c482f361ed7e37c9e242eef5ee0f8045313e1","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2023-11-09 02:58:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7366569d204edd0c1fdcbba71ef55384dd489577","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-09 04:07:33.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/13/899513/5/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/687cf8d/ : FAILURE in 1h 08m 56s","accounts_in_message":[],"_revision_number":5},{"id":"f19dd4015f9661b2021d7c31fa430277ee2d76a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 05:18:16.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14d06de7b9984f89a376dc764616ba55\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab2fcfab7fd540e9a3a74fc425810d19 : SUCCESS in 10m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc6a4f2e18f64ff1a6dbfefdb2c6b168 : FAILURE in 29m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06e16ad67cba4c98a3c85784383ffb67 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42dda0601f144e318afbacc11ef0c2b7 : FAILURE in 10m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dfe1c16da91d478897d6ee4487123902 : FAILURE in 10m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b9dd123f5ff04c56af9ef34480f4aba0 : SUCCESS in 6m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ff2ebf9cc3a346368be14d993d71fab3 : SUCCESS in 59m 57s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bed43bde9a8a48c99bdec98f5db3f24d : FAILURE in 2h 07m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a3fc079a292342c499220bca0322a098 : SUCCESS in 1h 13m 19s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/dc60de25f52d4c6d82087a420c725e96 : SUCCESS in 1h 18m 04s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/403f21feb6b04f6684d4167f69711e3c : SUCCESS in 1h 30m 48s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/bd090df0fff542b7b1a1ac09ca9e8e2c : FAILURE in 1h 14m 55s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/31729acd33434ae6ad4cca433a77034a : SUCCESS in 5m 12s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4932357bb60c4871872deaaa0211e968 : SUCCESS in 5m 40s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/38fdf4e9e3844f0496bdd5c8f658094c : SUCCESS in 43m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/62faf63e9f49410c80821f9bafcdd837 : SUCCESS in 30m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ae30302344194b388499df92bb32712d : SUCCESS in 44m 05s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/5bdd3cb887e44a50ac99f743c7a3b0ee : FAILURE in 11m 04s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/5f2a9ba22f64459ead470c464ebd0fde : FAILURE in 23m 45s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/8f898646730641649ebf31bafeab2748 : SUCCESS in 1h 21m 54s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/73b9c68159dd4957b41df3b5beeee8f0 : SUCCESS in 51m 57s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/63569e72f3494c4790eeeca616139e28 : SUCCESS in 47m 59s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d41d7cf1153047479d787f92104a5c5c : SUCCESS in 43m 57s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/414f522538ec486bb29a0e58410c9dfa : SUCCESS in 50m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"38c917dfab8cff8ef79e268a822c37ba0bf93f95","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-11-09 05:50:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"56375a74837c16bcc669e149d8abfbe78b9dc9aa","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-09 14:17:37.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-899513-5/NFS/ : SUCCESS in 2h 36m 16s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"beb5180c45886b26b785e804825a766cf73fa6dc","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2023-11-10 22:46:28.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"fec774d627da84637a2e0d9ef389a9c7995ff651","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-11-12 01:56:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d0e78c4dfb5f04a79371386e8be533d984ce8cd3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-12-20 22:01:15.000000000","message":"Topic set to bug/1886690","accounts_in_message":[],"_revision_number":5},{"id":"fcc4ec5499e9b450d1def48c7772932e758e7149","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2024-01-05 21:11:55.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"340335aca23d5d75f6d58cbeeca0acfc4d6fad6b","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2024-01-05 21:15:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0f72248b7241f404353fd87cc3ad20ce9a4ca115","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-05 21:46:21.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/899513/6/check/DellEMC_PowerFlex_Manila_dhss_false/b8bf6b3/DellEMC_PowerFlex_Manila_dhss_false/712 : SUCCESS in 34m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"b0e3adcd452e5bcb1a4deeea1bf37963e287ccb7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-01-05 22:14:13.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/13/899513/6/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/5efe290/ : FAILURE in 1h 00m 05s","accounts_in_message":[],"_revision_number":6},{"id":"1981c752d70b889a233112e935f8c654c2eb6ba0","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-01-05 23:08:52.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-899513-6/NFS/ : SUCCESS in 1h 56m 38s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"214679952e6ecce4c17c35225ec4eec55b8c2460","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-05 23:36:05.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/11e6f9d1bc384e89b465559d8207f290\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/454e175f99fa493e8f459d7f6dc1ef28 : SUCCESS in 13m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/757742310a324bc29e9ebe389971a1c2 : SUCCESS in 25m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e6c5f277d0749e39475473bfe355ba6 : FAILURE in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d190d1acacb4d3ea887130c32a1b77e : SUCCESS in 11m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3d70470109d49b892a1ac255a188fe6 : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd5e843a9ef946d3aa6067c0dddf2c8b : SUCCESS in 6m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/605f8e3cb46f4b64aa55209df1e61fd9 : SUCCESS in 1h 15m 37s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1e13f78411df44d4923ba4fde825fda9 : SUCCESS in 2h 18m 33s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/84544348f6364c5c946749cadb15a924 : SUCCESS in 1h 29m 02s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/bd591a2b2ee146348aa0e6149e8ce9f7 : SUCCESS in 1h 12m 03s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6c78b5836b914e269f04254ab0d5a8a4 : SUCCESS in 1h 41m 29s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b356b82beb3f48959e929d69940bead2 : FAILURE in 49m 58s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/841ce98fb063420f84650e9eb8ea3b7b : SUCCESS in 5m 45s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/486ebfa301574d0fa46335f13dd36a17 : SUCCESS in 5m 00s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8fc8e93ee79f4eaabf2156700320855c : SUCCESS in 41m 49s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b2cd17efd7da4fe5a7abead1958d49e0 : SUCCESS in 28m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e6e1f085060c4304a91b261d05fb7c9b : SUCCESS in 35m 13s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/5a40ceb1a999461ab53cbc3c51b4e4ac : FAILURE in 11m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/00f9dcab7c494001bbc1eb7a2db6e55c : FAILURE in 11m 42s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/6a130fb6fba04097a4a962da712df3f8 : SUCCESS in 51m 42s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/8aee77f61bd4484a959ff31437e54870 : SUCCESS in 42m 48s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/408c80f5de9c4919a329dcc230d4b900 : SUCCESS in 56m 04s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e0cd74ca479c49218c9b1d03b5bc748f : SUCCESS in 26m 57s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/287878845acc47519b03b59a1961baa3 : SUCCESS in 38m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c1846f906187e45f6840fc45cd5d52e4b0ece67a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2024-01-06 00:05:58.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"554ea61a43d73a14ab1fd01aa25f092ac142def7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-06 00:38:06.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/899513/7/check/DellEMC_PowerFlex_Manila_dhss_false/986fdf8/DellEMC_PowerFlex_Manila_dhss_false/713 : SUCCESS in 31m 51s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"004e066d7a5d742d78c48fd7814808cdb3da7230","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-01-06 01:20:14.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/13/899513/7/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/ec52a39/ : FAILURE in 1h 12m 13s","accounts_in_message":[],"_revision_number":7},{"id":"31840f15215fce378f5bef786bcca11d7fb86454","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-01-06 02:03:33.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-899513-7/NFS/ : SUCCESS in 1h 57m 13s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"18d31d8223b3d5423311b6268c4f9e3f24d61098","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-06 03:16:15.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/50ed99d637554d6fbeb23e56b8cae118\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8d3eaeb87e4450c8f17fa1733d8a6fe : SUCCESS in 9m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eab9a9b192a041bba62870b67134af9d : SUCCESS in 21m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ff117cf6e6949188a9e94ea75b5734b : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/663483680ad04184a7c1d587ff63b11e : SUCCESS in 10m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0baa34418ce9494f92ee7410fc440a8a : SUCCESS in 9m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04c64c612e9f401182a0c1deb22f88e1 : SUCCESS in 6m 44s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d4d5641d3ea74a3880f453f160763213 : SUCCESS in 1h 14m 43s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/cb6fcaedae79438686e00093f3bbd626 : TIMED_OUT in 3h 03m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/29fff72f4d164a0b95a2c71853bf3f96 : SUCCESS in 1h 23m 25s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/17b1b9d832d34e0fb8a5f173182f6a36 : SUCCESS in 1h 11m 09s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/4f245accdf2d408eb8b33907ce908a4c : SUCCESS in 1h 30m 10s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ba0df39f35cb4dc9a2541daf7c502070 : SUCCESS in 48m 50s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8a65b40dfe734e368cf22bb749ea9d01 : SUCCESS in 4m 28s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4b262846b53a48eabcd8a890b9f3c56d : SUCCESS in 4m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/434d52c49fad436b8472c85a9533a875 : SUCCESS in 37m 30s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/60c7a4431591412396f82974a18d131f : FAILURE in 40m 43s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e20bdb05b4234a17851c9047895bf9eb : SUCCESS in 52m 40s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/b3746cfb15624130aa7f562559e0c69f : FAILURE in 10m 04s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b8d197fcf1ef4d438d67025fb2a58117 : FAILURE in 25m 17s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/9da63ed0bf094d198184717928aa6a82 : SUCCESS in 1h 18m 16s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/32b53c4d80a549dd9549c92283e8509f : SUCCESS in 59m 57s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a6b364d51f2f4716b7fac3d72423310d : SUCCESS in 39m 25s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2b9f43fe72c746d296fdf57aaa022dcd : SUCCESS in 24m 32s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/eb982cf4a3f446ae9dc14cef20afd9a7 : SUCCESS in 34m 55s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9f557a3bfaac30f5fb97a3dd86378cdc991c02ce","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2024-01-11 21:55:51.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"80ef2bdc82ecb5f01d4f5c1ed9d0fff44f6010a7","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2024-01-12 01:08:46.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"cd79f16e5be51dafe10f4197ce5733beb6feaac7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2024-01-12 02:44:30.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"423c118cf9c198e05e52910b9fa15251bc7b8f4f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-01-12 03:55:11.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-899513-8/NFS/ : SUCCESS in 1h 10m 16s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"dc5042753cb33d44b3fda964948a4dc2d552ca42","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-12 05:44:50.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/899513/8/check/DellEMC_PowerFlex_Manila_dhss_false/60bcdd7/DellEMC_PowerFlex_Manila_dhss_false/717 : ABORTED in 3h 00m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"472a9e63cc165e0d9708086abfa534ecffa4ced0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 06:07:36.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96da4a965f464ddb9ecbd5299996cea3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6baf407752c24cbf99d96a8e551e133b : SUCCESS in 9m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d388e38feff4b44b8a1d715fcdeabcb : SUCCESS in 26m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b2eacc82d634590a93574d6cbb70159 : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fe442ffbd6e4e0f809dcd7ac0a4f9f7 : SUCCESS in 12m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/866afaadc9d3425793c8d0d229746d5f : SUCCESS in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6e6afaf140f430195d12a97ba770f5a : SUCCESS in 7m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f356cfe8d70c4458b54b38c51e39fb25 : SUCCESS in 1h 14m 27s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/49436d5a612c4e7baae3539562771d18 : FAILURE in 2h 36m 01s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/988446409dcb4764bb20f9e6142afc57 : SUCCESS in 1h 13m 58s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6c5cc00718d9492c960ddab6b1ecaf44 : SUCCESS in 1h 01m 12s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7f3f54ecb8bc49379facbe59be2ea70a : SUCCESS in 1h 21m 02s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/617aa9b5854d4b1cb99edd1c83476adf : SUCCESS in 1h 09m 54s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6637a72d1e8d4e2485bff6506e156431 : SUCCESS in 4m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/70679ea56b6d4b9b880b2ce1c9708000 : SUCCESS in 4m 24s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/598fa5e67d53477492f365146bafcb2f : SUCCESS in 52m 08s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a3f43dd4eb0c47b4882145db3e683dd1 : SUCCESS in 46m 36s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e68cd2f17ac74b6f90f73b80a3485091 : SUCCESS in 1h 01m 39s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/e99fce5e1254422f91b8693496a1ff56 : FAILURE in 24m 03s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c0ef4f3998a246baaf1977c3259a1e55 : FAILURE in 26m 55s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/dec388483d364701a861737695dc7576 : SUCCESS in 1h 11m 32s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/52e38a1af5e74e39bea7da19d8dc9810 : SUCCESS in 1h 06m 06s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2584bc23d15b4a78ab0987a2779771d3 : SUCCESS in 1h 18m 55s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/caa4e71a3fa24bf78fd3c1c584a7e5c7 : SUCCESS in 44m 32s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e1b1f1160f154209adc8d046ac7f6166 : SUCCESS in 55m 56s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"fefd75ccbbb72f207a8c81de134cca475338f3c7","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-01-12 06:24:13.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/899513/8/upstream-check/manila-tempest-plugin-ontap-dhss/0a11d64/ : RETRY_LIMIT in 9m 17s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/899513/8/upstream-check/manila-tempest-plugin-ontap-no-dhss/3b61e4c/ : RETRY_LIMIT in 11m 04s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"c2b63735031803724238ec70f78b675d19700205","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-01-12 06:41:54.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/899513/8/check/DellEMC_PowerStore_Manila_dhss_false/8ee871c/DellEMC_PowerStore_Manila_dhss_false/16 : SUCCESS in 34m 02s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"5ed5cf897d9e59541619dc90ed27cb5a249dbdb8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-12 09:07:54.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/899513/8/check/DellEMC_PowerFlex_Manila_dhss_false/9d4914f/DellEMC_PowerFlex_Manila_dhss_false/718 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"8ea2c37bb6d297e79a9d98fd94f9726bcd57795b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-01-12 19:35:44.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"d5c7c7d7da709f8f36b89308e5e554e919e094d9","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2024-01-12 23:55:04.000000000","message":"Patch Set 8:\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"a5633193daec9553d5fc886eb2f97d85e6a3a0ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2024-01-12 23:57:29.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"fc6324c024ac2945949f1df6d9db0cfdfd5b9a43","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-01-13 01:28:35.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-899513-9/NFS/ : SUCCESS in 1h 30m 47s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"f2139a077765cc837fbb49b6edd8863e65f248df","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-13 02:57:47.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/899513/9/check/DellEMC_PowerFlex_Manila_dhss_false/7cc99a4/DellEMC_PowerFlex_Manila_dhss_false/719 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"ff0e95049270fd73a2a5ff6a33ad50289e6dec63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-13 03:09:00.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a62277bca4b43ab8eae375f1a2819b6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44af8492a2fe4eb6958408d1382c071f : SUCCESS in 9m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7006c1c47aa34e178017ffd57fdf3525 : SUCCESS in 25m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79de2409ac824db29ffea2c223913c47 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5045e06fd3b742dbaf33d51c94993628 : SUCCESS in 9m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c9c1651bbe1c41d496124197910ed7e3 : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d1bf8ff4afc469e9de91a5122863e71 : SUCCESS in 7m 18s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/04f387619c004cb5bc4c9954a7213192 : SUCCESS in 1h 09m 15s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bfaed2eb06894809a8d183fc9508d8e1 : FAILURE in 3h 02m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/feb5178a99754838abac47276ca403ca : SUCCESS in 1h 30m 54s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8ddcb2f629974b18986ffbd912e41c9a : SUCCESS in 1h 06m 58s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/747e0811e667402882a07dae61ebc42c : SUCCESS in 1h 25m 14s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b7268a4f3053415bb611f25fe26acc52 : SUCCESS in 1h 13m 22s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e64f87ab11ff4ca0ade9d69d080c9052 : SUCCESS in 4m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0ccfc25819ed4f1e9d255653ae2855af : SUCCESS in 5m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9fc6f56b89994a94b10101714f501182 : SUCCESS in 39m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/edcb847dee5a498e9cffa5cdfeb473b1 : SUCCESS in 30m 59s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/088263414efb4cc2beb8b0ec7f7ca8ee : FAILURE in 1h 04m 17s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/4cd4aacbf3714e31a36f5e580d8b6d15 : FAILURE in 13m 54s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/039fa86f328e4560a3e8ddfa87d12864 : FAILURE in 22m 55s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/7eeb6d5df40049c5a830e1bae6fbd123 : SUCCESS in 1h 02m 36s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d2e82f22978b48d7a303108f060f439f : SUCCESS in 58m 52s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/474203cacdfe4df1b1aa8c8c3b9ae8d3 : SUCCESS in 50m 12s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/dca55baac4cf46c1a8c3c56bef42106b : FAILURE in 1h 01m 52s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/5eaf07ca9d2143c0ab809361687c76d4 : SUCCESS in 35m 45s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"b73c60bc852c213721c02a0ec0c41241c03c20c3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-01-13 03:36:06.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/899513/9/upstream-check/manila-tempest-plugin-ontap-dhss/62864f6/ : RETRY_LIMIT in 8m 39s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/899513/9/upstream-check/manila-tempest-plugin-ontap-no-dhss/bba779a/ : RETRY_LIMIT in 12m 51s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"e46470bf788020f908d556c7518aeaf8faa47072","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-01-13 03:42:03.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/899513/9/check/DellEMC_PowerStore_Manila_dhss_false/703a724/DellEMC_PowerStore_Manila_dhss_false/17 : SUCCESS in 32m 48s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"a7f4c0518321ab9b6d57c7abb97c660231bb09a4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-13 06:09:18.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/899513/9/check/DellEMC_PowerFlex_Manila_dhss_false/1400a9f/DellEMC_PowerFlex_Manila_dhss_false/720 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"033973bf1166b3b5331a101708217d75b41673ce","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-01-15 20:06:27.000000000","message":"Patch Set 9: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"0fff88fbba79050fee2fbd4a79106b49b37a8bd0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2024-01-17 00:19:50.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"182df24f7182c0d52d309e40f7946c093ff8be7c","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2024-01-17 00:21:17.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"c82107f820026e7b40f5794ea2ecced912a504b3","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-01-17 00:39:58.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"9ca230ab06d77d2f7d78a9df556c5926a3ac4a45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 01:58:39.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d8e76db9d14432582322e671627640b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d1d852051a84a58b2c09c9228262de7 : FAILURE in 6m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a67ffe65ef46441ea3a62183ec1ffa36 : SUCCESS in 27m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3463d6545ad4d3fb26b135d604a2f34 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65f411f0c1c74da5b44bbcea3053946c : SUCCESS in 9m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18d76c7289ec47689ebc2a2119936912 : SUCCESS in 8m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a4253cc3f52402cafa450e86a5e948e : SUCCESS in 7m 10s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f3ce25ca7a664c5e9439283070a34a75 : SUCCESS in 58m 58s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4422238096d8448e96b0656f69e10d28 : SUCCESS in 1h 36m 07s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c2cfc79bfa1d469a8f81bf1121245ef1 : SUCCESS in 55m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/0bb5f04ff8a94b52a468f38e5236ab0b : FAILURE in 1h 08m 51s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/420a495ca0ce4a088c1c3cd1e889b584 : SUCCESS in 1h 33m 02s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/db634bb6d1234958aa17d677f9ea0b51 : SUCCESS in 1h 18m 47s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a7dc581b3fb34c7489fc001c2bd37982 : SUCCESS in 4m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/52c60961052f4bb1b738c2279a4a9dd7 : SUCCESS in 4m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3d9880c34f6846eaaf6281475b68ce05 : FAILURE in 48m 05s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6a30d60b36444414b240bdf4839331b7 : SUCCESS in 45m 38s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e8ec08c3894749f48255b6590264342b : SUCCESS in 38m 21s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/50ffc8cb8e0649d6809b3ce755ae62e0 : FAILURE in 23m 15s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/e84eedcb59864e9ea7545d1130b874bd : FAILURE in 11m 00s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/06782410b62c4774a037df839e404b6d : SUCCESS in 1h 23m 28s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/420b2c7f20b94474984d613ae9d18098 : SUCCESS in 1h 05m 34s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c55f71f057e541ab811982164d512575 : SUCCESS in 59m 08s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/46283e22155343bcba9e0512193f8ef5 : SUCCESS in 41m 48s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7b0879b6774e4c63abfbfd6d66c56615 : SUCCESS in 48m 11s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c47f06c55c9da869e684c1908cd6d56c770f292e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-01-17 02:26:02.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-899513-10/NFS/ : SUCCESS in 2h 5m 53s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":10},{"id":"f83d30df1cada73d0df45f33529762b3ec5a5893","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-17 03:20:09.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/899513/10/check/DellEMC_PowerFlex_Manila_dhss_false/944e0cf/DellEMC_PowerFlex_Manila_dhss_false/724 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"6028a98e9cc90f7012eb0c3bbdb6fe2b49602804","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-01-21 20:40:00.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"8dd798f904791c3a909055d40c97f2f407613999","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 00:07:24.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0cd9dde31f143dc96f616930f09f257\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7c46321eed349beb70a2e4f6b00b067 : SUCCESS in 8m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51cb9e825c44493ea32b9798af71f61f : SUCCESS in 18m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28841c06e0bb4b85bdef721fbd4f72ff : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d34e3714f734cb78ed1e113e8c91dc3 : SUCCESS in 10m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/81aa6fd8b6484cefb9887fcdfd9fdfe4 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f42ed79138c48398e59d81e8f777a38 : SUCCESS in 6m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/49c5863e3f054e36a3fb5b70f1b1c73b : SUCCESS in 50m 27s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3b0fed5b42bd4831aaac83de9b82a7c3 : TIMED_OUT in 3h 02m 59s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/72df337b6d61467aa471c085ce711328 : SUCCESS in 1h 25m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/984d2b5fdb3844339cd8ca94efabb0c7 : SUCCESS in 1h 00m 55s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b93c08ba70404427817e7cb97fe46199 : SUCCESS in 1h 45m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/dadab06592b440639fd9ba6c6b39fc40 : FAILURE in 1h 09m 09s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/23439c2504704bd19b1d4f9b3e478f77 : SUCCESS in 5m 14s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5f73e0e71f7d4fdbaf9ffb0ea902812c : SUCCESS in 4m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a777f62bde9e46c68d774ed1f0d0b3d7 : SUCCESS in 39m 37s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3bc3118820c844119408715c48a6505d : SUCCESS in 47m 36s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d7961a8754ee4723bcac48dfe25ddf09 : SUCCESS in 55m 04s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/2b4045f9e05943d39e6c868e9cf85b93 : FAILURE in 23m 33s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/cc3184813f27453a81a57d66a5aa1a04 : FAILURE in 21m 41s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/6273fd6a55b744a9bbf016e8604f4987 : SUCCESS in 1h 19m 09s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0c5d4f43df8948de914872e1b3ea0f78 : SUCCESS in 1h 12m 13s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a93de225205442d18817d62e0a15629a : SUCCESS in 1h 06m 02s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/4d2adb19b01b4d92a936e9a2f9824845 : SUCCESS in 49m 18s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/cd72d882a8af4433a4c7872a2268a8f7 : SUCCESS in 48m 03s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f8beef60935255651a3ae1812ec3b80d4b6c0693","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-01-22 00:21:33.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/899513/10/upstream-check/manila-tempest-plugin-ontap-dhss/24e9223/ : RETRY_LIMIT in 7m 36s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/899513/10/upstream-check/manila-tempest-plugin-ontap-no-dhss/10b97a5/ : RETRY_LIMIT in 10m 35s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"fb7e55e87b2d7b737eff076f1de14c891ad0bdc9","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-01-22 00:49:00.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/899513/10/check/DellEMC_PowerStore_Manila_dhss_false/8c575a4/DellEMC_PowerStore_Manila_dhss_false/21 : SUCCESS in 41m 17s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"7b24cdc71ba27c91011dbbdc4c8004e0cedfb1cf","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-22 03:07:43.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/899513/10/check/DellEMC_PowerFlex_Manila_dhss_false/fb9c1f5/DellEMC_PowerFlex_Manila_dhss_false/733 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"b38d49b989cd5983c0f1547ba23d2c54ebdfab07","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2024-01-24 08:14:46.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"79b37fbb762cc9f2f0d59e623f4d768c9d071a6b","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-01-26 11:24:08.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"99a0b4cb4705f123f386b43014ebd637807e9b86","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2024-01-27 03:00:20.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"cc3dcbe2e0f5a5d0f1f203cdd0385cd9f3b50885","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2024-01-29 10:08:17.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"6a783fe9825b8ca557775a4905d1d4b5f4cf28d7","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-01-29 15:42:47.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"bd0f920936230b0104a104aa0a1c63d11352891c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2024-01-29 20:23:13.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"b750a4d3da99bee3091db356d7c28f21616de370","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2024-01-29 20:25:15.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"78337bbdff305ec8c40845d3b0598ffb4f392ccc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-01-29 20:30:28.000000000","message":"Topic filter-message-fix removed","accounts_in_message":[],"_revision_number":11},{"id":"9789db11ef6e8f22cd3116fbb272b75ec77cb85a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-01-29 20:30:37.000000000","message":"Topic set to bug/1886690","accounts_in_message":[],"_revision_number":11},{"id":"4845457706bfbe142b43f9dcefc5ca5342ed106d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-29 21:56:34.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/07d576526e8844b1a81f889b4af5e7a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/804c4f61856342159fe79119fbcdbaae : SUCCESS in 10m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfa48511b3724b9d995656d6c23e4bad : FAILURE in 24m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2ef56eb066e4c6e9a8611ce3bfa17f6 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34241b066b9e40f4b6440fab588cba90 : FAILURE in 10m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb6804eac6864930be87c613f02ca723 : FAILURE in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99da4fc394564d90ace5adf829d72b69 : SUCCESS in 6m 35s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fe92d662888249dd8ced0beb29544e92 : SUCCESS in 1h 12m 17s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7be8c85beb794d748ee75c71c11f5ec7 : SUCCESS in 1h 24m 19s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/979207e92efe44b9952c9c4b9d0892e4 : SUCCESS in 56m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c54dc6aef1f241c9881fbe8f193afb5c : SUCCESS in 1h 17m 39s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7472715eecf6400587249244b885ffd6 : SUCCESS in 1h 09m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/fed9cbfbe40848b8bc4e62c7ba552c0c : SUCCESS in 54m 10s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/92ffdef7ed894624bf25328387238555 : SUCCESS in 4m 29s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/fa6e0b905b2e45db9f5b9c213960b45d : SUCCESS in 4m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/eef91c826e62473088e84ba252d0f030 : SUCCESS in 41m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6deeb035aa0444ed8386ff9ae194f28b : SUCCESS in 43m 48s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4860d0ecc4f64fa3bfe3a7d56efab52d : FAILURE in 52m 41s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/7dfd9fddfa9c481e8d202aa2c6ead5e1 : FAILURE in 23m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4a7f4027f7cf488b81169623507d5863 : FAILURE in 24m 18s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/a7576a16048a4f979526ad5059a2f68a : SUCCESS in 1h 24m 11s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3d2d5b45414842328dd5958cc0b05921 : SUCCESS in 58m 21s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fa034b77b9c54a89bbf1f81f146dc880 : SUCCESS in 1h 05m 37s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/68e12b1a18854fe3a6ef0a039c3f0382 : SUCCESS in 26m 38s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8d9e844387374199883229ba9bc9dc60 : SUCCESS in 49m 46s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d67ac11c3c63da73cf0611735fc5db8a3ac95872","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-01-29 22:24:40.000000000","message":"Patch Set 11:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-899513-11/NFS/ : SUCCESS in 2h 1m 8s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":11},{"id":"364553fc7c2cd298b24022df45b0628b14ec764d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-29 23:23:30.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/899513/11/check/DellEMC_PowerFlex_Manila_dhss_false/1d3a10a/DellEMC_PowerFlex_Manila_dhss_false/749 : ABORTED in 3h 00m 02s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"b4d74d30e8e99a0c83f02981c9e6dcb1a7935cb0","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2024-01-30 08:56:32.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"0951d7d96744aca85b7d6cb534741b2cad7f089e","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-01-30 13:16:37.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"fa361d3663a86c778debde825d769ff20e98d5c1","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2024-01-30 20:26:27.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"784aa085320f93c23fd518d86aa4b0358ef16349","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2024-01-30 20:36:40.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"7d858cd0b9975bb04bfa29b7a58135d994c8c0f8","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2024-01-30 20:38:42.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"88088e45c4ee1ed4be1785cbecdbc8f256cedee1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2024-01-30 20:42:06.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"2b2b88699eb546edfb5187cd37bd76aa4c7cde2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-30 22:33:38.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c1d1f3d9ab014c40ad5668c411dbde4e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b40d1d8506fb43ec91b37b783874250c : FAILURE in 5m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f453f38d6fb4dcbbd984332f1388d11 : SUCCESS in 18m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d16ea4a17b84eec98a7ca736f58b711 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24a98c2f5dbc452c966baa85ce63352f : SUCCESS in 9m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/35ee1c32a3014de8b3e4f7a8666aeed1 : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d4ea7c75e234c57939e374fba9723d2 : SUCCESS in 7m 01s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/03815d078c8e4cd7b1a9d326f81c333d : SUCCESS in 1h 16m 39s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/9c2673c91989472a8352e8123aafd705 : SUCCESS in 1h 25m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ee0b715ad0bd453d8bdc7e1a5d7a227d : SUCCESS in 1h 24m 22s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/cea55858c2b44b2897a3d41da4bbc8b1 : SUCCESS in 1h 06m 15s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/17ab0f1ce9b64a8f8a244ff6b44afda5 : SUCCESS in 1h 44m 33s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/3dff8fbd4fd84d00af76beabe0d80246 : FAILURE in 1h 28m 01s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4c9f39db9cd24aa686f694f880479cc6 : SUCCESS in 4m 42s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/83578019815e4c12aa581b50d4e2d1b9 : SUCCESS in 4m 42s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/191241e1385c4f799e30995434ab215d : SUCCESS in 37m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/730f278ff6b94fe78dadd65100a6d4f3 : SUCCESS in 43m 12s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1e187289a0a746068f2cf273c27791bf : SUCCESS in 56m 09s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/018aa712458443acaa638a6d7ba730d8 : FAILURE in 25m 45s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a66c63ab95604a998220b04557bc6d90 : FAILURE in 26m 28s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/46b977cd0a8d4149af35acad4ada5b3d : SUCCESS in 1h 38m 14s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/7d8c76e0373745819b91db7fcc1a13b3 : SUCCESS in 1h 04m 53s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c11b79d62f714fe6baef43cf137374ad : SUCCESS in 1h 12m 16s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ebd1398fb89d4811a78e0cf449407594 : SUCCESS in 42m 11s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c6675172dc9c48eaa8951540f726372c : SUCCESS in 45m 08s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"fded44cd289f9faab64144585e337f3c4d2654b1","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-01-30 23:05:25.000000000","message":"Patch Set 12:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-899513-12/NFS/ : SUCCESS in 2h 28m 23s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":12},{"id":"83eefcd5b23709fd53310eb200ed144a9b8b0982","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-01-30 23:21:19.000000000","message":"Patch Set 13:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-899513-13/NFS/ : SUCCESS in 2h 14m 18s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":13},{"id":"d6fce09f4c259bb502dc0e0b0c69c74b6c7c0f65","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-01-30 23:43:43.000000000","message":"Patch Set 13: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"2bc35d672e9cc9822ba4c60a39d639566eefc509","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2024-01-31 04:23:47.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"cf5cda30b29e6b9724508dbb7c514811c705d7c4","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2024-01-31 08:26:40.000000000","message":"Patch Set 13: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"c363649ead917d338da0d4df84022e602214c2c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 10:56:01.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9cacd0e194e4ac8afc0fe2628e0e94b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd0441c33ed9450b8a6cf5b9ec8d9dfe : SUCCESS in 10m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c3bd882942d4ea2905391014e638286 : SUCCESS in 24m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e79a1b469914aa4a3beeab919dc9806 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6068a0b417e4d58b007d1a0fd6bfc7d : SUCCESS in 10m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f06c8db9a0c84421af0a2856f2a21d62 : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31d6df8fef1f41b687137d206c452b3f : SUCCESS in 6m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0323c61e34ee45d4abd63cd7704201da : SUCCESS in 43m 43s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/55277b284b734b159eb1ad6e3222077d : SUCCESS in 1h 32m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/7f2bd51f7a06400dacf9fd81cb76812e : FAILURE in 2h 18m 18s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c9262ce7a8fd4e94ac3baec221718a1d : SUCCESS in 1h 06m 11s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/0c42b87b0939419f95ae25affa6ca6f1 : SUCCESS in 1h 37m 17s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ddf44213f2d7469eb9d223ce79230472 : SUCCESS in 1h 16m 04s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b7ca9c290dfc42e89efbc3eb4ef22db1 : SUCCESS in 4m 25s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/909b06c702a44fa3a087592a45e30ac3 : FAILURE in 3m 32s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8d799afc382b46f7bf132ac6de5bf78f : SUCCESS in 42m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7735d165a40640978149f6104d7e6910 : SUCCESS in 46m 27s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e42adc80976a4b3cbfdaec6c8e05ea62 : SUCCESS in 46m 43s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/6afec748b1ba4c2e9cea93444005137a : FAILURE in 13m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/3a4a07d162e445db902a2bf4ac0cf7f3 : FAILURE in 22m 24s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/226d41ada1c441a1bbb99d1ae21dc36e : SUCCESS in 1h 34m 52s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/fda41e6b5ba34d0c9773bb0c3793b4c5 : SUCCESS in 1h 10m 29s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f9fdfc30afa94259b9a499cc63441390 : SUCCESS in 1h 04m 32s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/9d1882f53b684aae8f8826452ba3413a : SUCCESS in 24m 11s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/afa5542af3a448768e3c31103d45d244 : SUCCESS in 37m 48s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"7851ead30b32fe9cbd35418a4cdf2c44b7c63e6a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-01-31 11:08:40.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/899513/13/upstream-check/manila-tempest-plugin-ontap-dhss/d65a78a/ : RETRY_LIMIT in 7m 41s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/899513/13/upstream-check/manila-tempest-plugin-ontap-no-dhss/4e48c78/ : RETRY_LIMIT in 11m 15s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"6b2575273b3ff71f156a5b703234d5ad47751d87","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-01-31 11:26:42.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/899513/13/check/DellEMC_PowerFlex_Manila_dhss_false/86ec11d/DellEMC_PowerFlex_Manila_dhss_false/754 : SUCCESS in 30m 25s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":13},{"id":"7ce9418fc0a3b906c575ee839fffe3bb796484a0","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-01-31 11:29:37.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/899513/13/check/DellEMC_PowerStore_Manila_dhss_false/450fad3/DellEMC_PowerStore_Manila_dhss_false/30 : SUCCESS in 33m 20s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":13},{"id":"afaa9fcc8f4bdbcb75a1dbd4f9418d92a1b601cf","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-01-31 11:44:47.000000000","message":"Patch Set 13: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"d33414f0dfeb4118602fd1d8de25d226a29c02c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 11:45:18.000000000","message":"Patch Set 13: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":13},{"id":"696bc3c3081776767b3b58529ba62f8105666a42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 13:00:22.000000000","message":"Patch Set 13: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0da6f7b4a7724216ab10d38c45e5e2bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e26d7a26d7584f7aa6da2ef25365dfd9 : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e8fb501790840988fc0f71b1fe32abc : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2208766720a4942a3316b49a469d45d : SUCCESS in 11m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b2df1b2af3a4ec0984bd8b629faaf27 : SUCCESS in 8m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/820c1231cdc7475f8ff89cbc7441dcbe : SUCCESS in 6m 47s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c87bfb94094643b9b3127c63bebb5dcb : SUCCESS in 1h 09m 56s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/66adb07fab83402aae19c3c1f7f6ca2c : SUCCESS in 28m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3659d5c03e2045258347de1c13e93241 : SUCCESS in 34m 37s","accounts_in_message":[],"_revision_number":13},{"id":"eabb0c646adee8a534f25b57d0712fbcce9132dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 13:00:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":13},{"id":"3c92e137c27a402cbb44e20bfa1624d3c9546970","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-31 13:01:40.000000000","message":"Patch Set 13:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94266b301eaa47a586512a7d9fbcc16b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06f1e87cacff4b7bbc26beb142ae1a56 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbc3bba8d74a40c789cf96dda009ba23 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":13},{"id":"832512268a5eacc6c231d51e99653f2352bcf325","author":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"date":"2024-02-01 21:43:59.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"ecda09a14d0b31169b275ed67a9088a131fbef12","revisions":{"03809eef39927380909a2c84914049b32018466a":{"kind":"REWORK","_number":1,"created":"2023-10-27 23:14:03.000000000","uploader":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"ref":"refs/changes/13/899513/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/899513/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/899513/1"}}},"commit":{"parents":[{"commit":"b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff","subject":"Change status and error handling for /shares API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff"}]}],"author":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-27 23:09:22.000000000","tz":180},"committer":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-27 23:09:55.000000000","tz":180},"subject":"Change misleading user message when user services are down","message":"Change misleading user message when user\nservices are down\n\nenhance the user message when all of the share\nmanager services are down or are still\ninitializing\n\nCloses-Bug: #1886690\n\nChange-Id: I168564a5b054d17762ad668ebbe4f5e7b562197b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/03809eef39927380909a2c84914049b32018466a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/03809eef39927380909a2c84914049b32018466a"}]},"branch":"refs/heads/master"},"02f8271fdfe6fb04838cef44f38e52cd93fbf25d":{"kind":"REWORK","_number":2,"created":"2023-10-29 19:20:42.000000000","uploader":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"ref":"refs/changes/13/899513/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/899513/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/899513/2"}}},"commit":{"parents":[{"commit":"b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff","subject":"Change status and error handling for /shares API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff"}]}],"author":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-27 23:09:22.000000000","tz":180},"committer":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-29 19:17:10.000000000","tz":180},"subject":"Change misleading user message when user services are down","message":"Change misleading user message when user\nservices are down\n\nenhance the user message when all of the share\nmanager services are down or are still\ninitializing\n\nCloses-Bug: #1886690\n\nChange-Id: I168564a5b054d17762ad668ebbe4f5e7b562197b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/02f8271fdfe6fb04838cef44f38e52cd93fbf25d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/02f8271fdfe6fb04838cef44f38e52cd93fbf25d"}]},"branch":"refs/heads/master"},"39722ffe73e107b220b7c8728b9f67ed5c4e3ec1":{"kind":"REWORK","_number":3,"created":"2023-10-29 19:35:27.000000000","uploader":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"ref":"refs/changes/13/899513/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/899513/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/899513/3"}}},"commit":{"parents":[{"commit":"b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff","subject":"Change status and error handling for /shares API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff"}]}],"author":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-27 23:09:22.000000000","tz":180},"committer":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-29 19:34:54.000000000","tz":180},"subject":"Change misleading user message when user services are down","message":"Change misleading user message when user\nservices are down\n\nenhance the user message when all of the share\nmanager services are down or are still\ninitializing\n\nCloses-Bug: #1886690\n\nChange-Id: I168564a5b054d17762ad668ebbe4f5e7b562197b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/39722ffe73e107b220b7c8728b9f67ed5c4e3ec1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/39722ffe73e107b220b7c8728b9f67ed5c4e3ec1"}]},"branch":"refs/heads/master"},"ff6242e9064763ec639a5ca61bf44a78c87b18e2":{"kind":"REWORK","_number":4,"created":"2023-11-08 01:24:45.000000000","uploader":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"ref":"refs/changes/13/899513/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/899513/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/899513/4"}}},"commit":{"parents":[{"commit":"b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff","subject":"Change status and error handling for /shares API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff"}]}],"author":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-27 23:09:22.000000000","tz":180},"committer":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-11-08 01:24:29.000000000","tz":180},"subject":"Change misleading user message when user services are down","message":"Change misleading user message when user\nservices are down\n\nenhance the user message when all of the share\nmanager services are down or are still\ninitializing\n\nCloses-Bug: #1886690\n\nChange-Id: I168564a5b054d17762ad668ebbe4f5e7b562197b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ff6242e9064763ec639a5ca61bf44a78c87b18e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ff6242e9064763ec639a5ca61bf44a78c87b18e2"}]},"branch":"refs/heads/master"},"7ebe8a789239c53540e8cc6d271c7fd699339d52":{"kind":"REWORK","_number":5,"created":"2023-11-09 02:56:57.000000000","uploader":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"ref":"refs/changes/13/899513/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/899513/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/899513/5"}}},"commit":{"parents":[{"commit":"b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff","subject":"Change status and error handling for /shares API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff"}]}],"author":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-27 23:09:22.000000000","tz":180},"committer":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-11-09 02:52:58.000000000","tz":180},"subject":"Change misleading user message when user services are down","message":"Change misleading user message when user\nservices are down\n\nenhance the user message when all of the share\nmanager services are down or are still\ninitializing\n\nCloses-Bug: #1886690\n\nChange-Id: I168564a5b054d17762ad668ebbe4f5e7b562197b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7ebe8a789239c53540e8cc6d271c7fd699339d52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7ebe8a789239c53540e8cc6d271c7fd699339d52"}]},"branch":"refs/heads/master"},"5c9b9939c49eccaf7cb3a52993f2b44ebc0b67dc":{"kind":"REWORK","_number":6,"created":"2024-01-05 21:11:55.000000000","uploader":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"ref":"refs/changes/13/899513/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/899513/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/899513/6"}}},"commit":{"parents":[{"commit":"b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff","subject":"Change status and error handling for /shares API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff"}]}],"author":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-27 23:09:22.000000000","tz":180},"committer":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2024-01-05 19:26:15.000000000","tz":180},"subject":"Change misleading user message when user services are down","message":"Change misleading user message when user\nservices are down\n\nenhance the user message when all of the share\nmanager services are down or are still\ninitializing\n\nCloses-Bug: #1886690\n\nChange-Id: I168564a5b054d17762ad668ebbe4f5e7b562197b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5c9b9939c49eccaf7cb3a52993f2b44ebc0b67dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5c9b9939c49eccaf7cb3a52993f2b44ebc0b67dc"}]},"branch":"refs/heads/master"},"8f115742a738b522f1c0b29a3be40ca626f71469":{"kind":"REWORK","_number":7,"created":"2024-01-06 00:05:58.000000000","uploader":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"ref":"refs/changes/13/899513/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/899513/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/899513/7"}}},"commit":{"parents":[{"commit":"b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff","subject":"Change status and error handling for /shares API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff"}]}],"author":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-27 23:09:22.000000000","tz":180},"committer":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2024-01-06 00:05:34.000000000","tz":180},"subject":"Change misleading user message when user services are down","message":"Change misleading user message when user\nservices are down\n\nenhance the user message when all of the share\nmanager services are down or are still\ninitializing\n\nCloses-Bug: #1886690\n\nChange-Id: I168564a5b054d17762ad668ebbe4f5e7b562197b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8f115742a738b522f1c0b29a3be40ca626f71469"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8f115742a738b522f1c0b29a3be40ca626f71469"}]},"branch":"refs/heads/master"},"545cca221367fe703b2602302752344ec83da8e8":{"kind":"REWORK","_number":8,"created":"2024-01-12 02:44:30.000000000","uploader":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"ref":"refs/changes/13/899513/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/899513/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/899513/8"}}},"commit":{"parents":[{"commit":"b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff","subject":"Change status and error handling for /shares API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff"}]}],"author":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-27 23:09:22.000000000","tz":180},"committer":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2024-01-12 02:44:29.000000000","tz":0},"subject":"Change misleading user message when user services are down","message":"Change misleading user message when user\nservices are down\n\nenhance the user message when all of the share\nmanager services are down or are still\ninitializing\nremoved a duplicate test: test_create_share_non_admin\nCloses-Bug: #1886690\n\nChange-Id: I168564a5b054d17762ad668ebbe4f5e7b562197b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/545cca221367fe703b2602302752344ec83da8e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/545cca221367fe703b2602302752344ec83da8e8"}]},"branch":"refs/heads/master"},"2a35c65d1d613c1458a95b4c2e751340a785abf7":{"kind":"REWORK","_number":9,"created":"2024-01-12 23:57:29.000000000","uploader":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"ref":"refs/changes/13/899513/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/899513/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/899513/9"}}},"commit":{"parents":[{"commit":"b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff","subject":"Change status and error handling for /shares API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff"}]}],"author":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-27 23:09:22.000000000","tz":180},"committer":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2024-01-12 23:57:01.000000000","tz":0},"subject":"Change misleading user message when user services are down","message":"Change misleading user message when user\nservices are down\n\nenhance the user message when all of the share\nmanager services are down or are still\ninitializing\nremoved a duplicate test: test_create_share_non_admin\n\nCloses-Bug: #1886690\nChange-Id: I168564a5b054d17762ad668ebbe4f5e7b562197b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2a35c65d1d613c1458a95b4c2e751340a785abf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2a35c65d1d613c1458a95b4c2e751340a785abf7"}]},"branch":"refs/heads/master"},"de6743f39695d623c29674529d54403fd3bdbfc7":{"kind":"REWORK","_number":10,"created":"2024-01-17 00:19:50.000000000","uploader":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"ref":"refs/changes/13/899513/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/899513/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/899513/10"}}},"commit":{"parents":[{"commit":"b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff","subject":"Change status and error handling for /shares API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff"}]}],"author":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-27 23:09:22.000000000","tz":180},"committer":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2024-01-17 00:19:26.000000000","tz":0},"subject":"Change misleading user message when user services are down","message":"Change misleading user message when user\nservices are down\n\nenhance the user message when all of the share\nmanager services are down or are still\ninitializing\nremoved a duplicate test: test_create_share_non_admin\n\nCloses-Bug: #1886690\nChange-Id: I168564a5b054d17762ad668ebbe4f5e7b562197b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/de6743f39695d623c29674529d54403fd3bdbfc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/de6743f39695d623c29674529d54403fd3bdbfc7"}]},"branch":"refs/heads/master"},"0f150cc3fc38a90a2591ae1044b781e24663404f":{"kind":"REWORK","_number":11,"created":"2024-01-29 20:23:13.000000000","uploader":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"ref":"refs/changes/13/899513/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/899513/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/899513/11"}}},"commit":{"parents":[{"commit":"b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff","subject":"Change status and error handling for /shares API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff"}]}],"author":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-27 23:09:22.000000000","tz":180},"committer":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2024-01-29 20:23:02.000000000","tz":0},"subject":"Change misleading user message when user services are down","message":"Change misleading user message when user\nservices are down\n\nenhance the user message when all of the share\nmanager services are down or are still\ninitializing\nremoved a duplicate test: test_create_share_non_admin\n\nCloses-Bug: #1886690\nChange-Id: I168564a5b054d17762ad668ebbe4f5e7b562197b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0f150cc3fc38a90a2591ae1044b781e24663404f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0f150cc3fc38a90a2591ae1044b781e24663404f"}]},"branch":"refs/heads/master"},"77b6bac00a3c6c93bf3485be6757bfcd695b759d":{"kind":"REWORK","_number":12,"created":"2024-01-30 20:36:40.000000000","uploader":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"ref":"refs/changes/13/899513/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/899513/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/899513/12"}}},"commit":{"parents":[{"commit":"b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff","subject":"Change status and error handling for /shares API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff"}]}],"author":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-27 23:09:22.000000000","tz":180},"committer":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2024-01-30 20:36:28.000000000","tz":0},"subject":"Change misleading user message when user services are down","message":"Change misleading user message when user\nservices are down\n\nenhance the user message when all of the share\nmanager services are down or are still\ninitializing\nremoved a duplicate test: test_create_share_non_admin\n\nCloses-Bug: #1886690\nChange-Id: I168564a5b054d17762ad668ebbe4f5e7b562197b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/77b6bac00a3c6c93bf3485be6757bfcd695b759d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/77b6bac00a3c6c93bf3485be6757bfcd695b759d"}]},"branch":"refs/heads/master"},"ecda09a14d0b31169b275ed67a9088a131fbef12":{"kind":"REWORK","_number":13,"created":"2024-01-30 20:42:06.000000000","uploader":{"_account_id":36483,"name":"melaku alehegn","email":"melakualehegn34@gmail.com","username":"melakualehegn"},"ref":"refs/changes/13/899513/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/899513/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/899513/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/899513/13"}}},"commit":{"parents":[{"commit":"b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff","subject":"Change status and error handling for /shares API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b24ef91f2c6e1e63c6bae8e52b778bba22ff15ff"}]}],"author":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2023-10-27 23:09:22.000000000","tz":180},"committer":{"name":"melakualehegn","email":"melakualehegn34@gmail.com","date":"2024-01-30 20:41:58.000000000","tz":0},"subject":"Change misleading user message when user services are down","message":"Change misleading user message when user\nservices are down\n\nenhance the user message when all of the share\nmanager services are down or are still\ninitializing\nremoved a duplicate test: test_create_share_non_admin\n\nCloses-Bug: #1886690\nChange-Id: I168564a5b054d17762ad668ebbe4f5e7b562197b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ecda09a14d0b31169b275ed67a9088a131fbef12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ecda09a14d0b31169b275ed67a9088a131fbef12"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
