)]}'
{"id":"openstack%2Fmanila~899520","triplet_id":"openstack%2Fmanila~master~I6e05a01ce9aae4f402c21910117bf854e75ef0d8","project":"openstack/manila","branch":"master","attention_set":{},"removed_from_attention_set":{"36488":{"account":{"_account_id":36488,"name":"Jipheens wahome","display_name":"JIPHEENS","email":"wahomejipheens@gmail.com","username":"jipheens-wahome"},"last_update":"2023-10-30 22:14:16.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6e05a01ce9aae4f402c21910117bf854e75ef0d8","subject":"error 500 updates unittest and function","status":"ABANDONED","created":"2023-10-28 08:56:51.000000000","updated":"2023-10-31 20:22:43.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"21102116c2f83f8270886e43c7decb98bb6160a7","_number":899520,"virtual_id_number":899520,"owner":{"_account_id":36488,"name":"Jipheens wahome","display_name":"JIPHEENS","email":"wahomejipheens@gmail.com","username":"jipheens-wahome"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-10-28 10:38:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2023-10-29 17:27:18.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-28 09:27:57.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-28 10:11:24.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-10-28 10:38:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-28 11:10:53.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-10-29 10:02:00.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-10-29 17:27:18.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-10-31 20:22:43.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"}],"messages":[{"id":"382d4252e7cbf152f6fb337bc125a59aad53ca64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36488,"name":"Jipheens wahome","display_name":"JIPHEENS","email":"wahomejipheens@gmail.com","username":"jipheens-wahome"},"date":"2023-10-28 08:56:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df66040a9ad232fa82fdecab7b253738221421e3","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-28 09:27:57.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/899520/1/check/DellEMC_PowerFlex_Manila_dhss_false/9f8a172/DellEMC_PowerFlex_Manila_dhss_false/626 : FAILURE in 30m 51s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"d35bc91761131ca1fddcac1b40c7a3cc64f29a1f","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-28 10:11:24.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/20/899520/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/c73cf84/ : FAILURE in 1h 12m 30s","accounts_in_message":[],"_revision_number":1},{"id":"1821f75b6cb60ac0c175eb0aee42d72c667af9e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-28 10:38:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b2cac23829eb472fbb14e069479e828b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9489c4163db4a4d835b57ab4df3e140 : SUCCESS in 11m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57ed9a6ab8674286bc66e50381062366 : FAILURE in 21m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/032094c3d22241fabc17061f280cf2c3 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec07279b1d0d4ebc9aceb89c168914d1 : FAILURE in 11m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a54782f4aa4d489fa8b23521d5f3d6bf : FAILURE in 7m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/21bfcb44e02e49a59a07333217528193 : FAILURE in 34m 06s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d31a6e31f8084a38afa61780b394aa13 : SUCCESS in 1h 40m 13s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1b166f67a2814cbdb7485064914d149c : FAILURE in 53m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/abf02db72fd44ab89ca8cd7e8f7c43f3 : FAILURE in 56m 09s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/25128444a4e448488b3eccc6713fd564 : FAILURE in 1h 04m 12s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/2b11f81e1ffc4e2eb45e0e745fc6ff5b : FAILURE in 1h 03m 26s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d752f5cda506435a9227fb8631e8bc0d : SUCCESS in 5m 37s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2c788bc0277449b0b97081e30116da70 : SUCCESS in 5m 56s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c3a6b4e793244e5786b2884929c31a9c : FAILURE in 27m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c344d44b2bbb4fbda9fee4b2d04f53c2 : FAILURE in 33m 09s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/675250fb7ebb400491fcca13d07ae779 : FAILURE in 37m 40s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/15d3c1d891ca4d2ebf37bebe41661cbe : FAILURE in 15m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c0483b1057044ad585f18af2bbf13003 : FAILURE in 12m 35s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/314d68f6aea9461aa1b55879934892ea : FAILURE in 34m 44s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/7664f7035a9a4742bb0d74ad6046a55a : FAILURE in 52m 58s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f7ef77459ef74c748fb31a0509461e97 : FAILURE in 44m 06s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/05549dbd1a02440485dee0c05ba472f6 : SUCCESS in 33m 21s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/4a8f712963fb43279b51ea9fccb59dbc : SUCCESS in 37m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"94f40dcaaedf4a8f160a823857c9912510d7e1a7","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-28 11:10:53.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-20-899520-1/NFS/ : FAILURE in 2h 13m 42s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"c0678d02246bf9bda39c02bc14ccc2c617abc897","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-29 10:02:00.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/20/899520/1/check/DellEMC_Unity_Manila_dhss_true/5acd062/DellEMC_Unity_Manila_dhss_true/719 : FAILURE in 3h 17m 47s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/20/899520/1/check/DellEMC_Unity_Manila_dhss_false/d9d41a6/DellEMC_Unity_Manila_dhss_false/758 : FAILURE in 1h 09m 20s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"ef37640beb76ec5593b5f5cca590295fdee02f80","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-29 17:27:18.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"6e469a8b99e4d97a4df9adacee34ebee667e8af4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-29 17:27:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c03e6e5a97b3d22386eb93de5983e36588565a7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-30 22:14:16.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"21102116c2f83f8270886e43c7decb98bb6160a7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2023-10-31 20:22:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"72a5c3582069db44a2f83a99c0ac857dc4fd0408","revisions":{"72a5c3582069db44a2f83a99c0ac857dc4fd0408":{"kind":"REWORK","_number":1,"created":"2023-10-28 08:56:51.000000000","uploader":{"_account_id":36488,"name":"Jipheens wahome","display_name":"JIPHEENS","email":"wahomejipheens@gmail.com","username":"jipheens-wahome"},"ref":"refs/changes/20/899520/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/20/899520/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899520/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899520/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/20/899520/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/20/899520/1"}}},"commit":{"parents":[{"commit":"1c44d9946012e78be451e317a021085bea7a1c98","subject":"Merge \"Change status and error handling for /shares API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1c44d9946012e78be451e317a021085bea7a1c98"}]}],"author":{"name":"jipheens-wahome","email":"wahomejipheens@gmail.com","date":"2023-10-28 08:53:56.000000000","tz":180},"committer":{"name":"jipheens-wahome","email":"wahomejipheens@gmail.com","date":"2023-10-28 08:53:56.000000000","tz":180},"subject":"error 500 updates unittest and function","message":"error 500 updates unittest and function\n\nChange-Id: I6e05a01ce9aae4f402c21910117bf854e75ef0d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/72a5c3582069db44a2f83a99c0ac857dc4fd0408"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/72a5c3582069db44a2f83a99c0ac857dc4fd0408"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
