)]}'
{"id":"openstack%2Fmanila~899671","triplet_id":"openstack%2Fmanila~master~Iceb40430548c94afa7c3e0a8c73e1c0d2f26efb6","project":"openstack/manila","branch":"master","topic":"sqlalchemy-20","hashtags":[],"change_id":"Iceb40430548c94afa7c3e0a8c73e1c0d2f26efb6","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (create)","status":"ABANDONED","created":"2023-10-31 11:59:03.000000000","updated":"2024-03-25 10:27:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"c262149107e30fa15d5ed5354cb550e1e483f36b","_number":899671,"virtual_id_number":899671,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-03-21 18:20:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"reviewer_updates":[{"updated":"2023-10-31 12:30:13.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-10-31 13:10:28.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-10-31 13:24:07.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-10-31 13:52:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-31 14:04:01.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2023-10-31 14:28:54.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2023-10-31 15:14:24.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"97efc25f4c5dfa190c22ec9b96fbf43d3f31adc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-10-31 11:59:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f73ed1027b837d35d7bc0c711d2c65168633f55","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-10-31 12:17:31.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"4ebdbc3ba2f8bf5f143f3a7db8e08848e44705ca","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-31 12:30:13.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/71/899671/1/check/DellEMC_PowerFlex_Manila_dhss_false/ac7c818/DellEMC_PowerFlex_Manila_dhss_false/640 : SUCCESS in 30m 53s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"20f94542158c892e21184243aac9e55116a8251b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-10-31 13:10:28.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/71/899671/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/8f0648e/ : SUCCESS in 1h 09m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e396ea5b04a9477c20641b60cc845946e2f083d1","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-10-31 13:24:07.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-899671-1/NFS/ : SUCCESS in 1h 24m 41s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"9e9a183e56258ee66e61de448380964713f8121c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 13:52:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34da38238c10414db3badc16d095088d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbf6488424894c7d86becd0703292a30 : SUCCESS in 11m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/301898bc225a4c7fa1be42f53b740ac2 : SUCCESS in 27m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6463fe6f8c5c4180a12c4d77d81c3bc6 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dab8a388f00b4404bf60abe0ea186e8a : SUCCESS in 8m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab91a8de49d44b5ca3d7902c41102bb9 : SUCCESS in 9m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d04d98b56d1a497c88a0cdf9e4b70967 : SUCCESS in 1h 00m 31s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/1dd04fc97b0b4ae689bc4eb968b2b138 : SUCCESS in 1h 44m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2f404bbd1bb14892923fb321f197b76a : SUCCESS in 1h 05m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/44d8d735159e41caa2f540e8504f459d : SUCCESS in 53m 34s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/37e350f5b09940119b0bf5ecf99be616 : SUCCESS in 1h 18m 29s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/45bd771c346444e987fba08a0fde0317 : SUCCESS in 55m 32s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5b8298fae03f4d43a1c347df33cff9d1 : SUCCESS in 4m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6bc7f9272d584e429db25b18bf3581f5 : SUCCESS in 6m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/9e2ed30dcef74518a88d6e6515d87a7c : SUCCESS in 28m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1506f516dae148649bd96220fa02d9dc : SUCCESS in 36m 08s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d4acd358dc9d495d9a5f47ff689a0ae7 : SUCCESS in 45m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/cdda26b36d79480cbbc98a54e82210bb : FAILURE in 10m 53s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/22a6bb9361844dd2ab830d53ba5b1dce : FAILURE in 13m 31s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/56468d1c96124e0d8859cfc01176b52e : SUCCESS in 59m 40s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/6c2de789bf7346f09acbb77f63e35255 : SUCCESS in 52m 56s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/71001136bbb14c10b6d375603cfb665e : SUCCESS in 51m 30s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e97544c4431c4833a24e00d7ff8d17db : SUCCESS in 28m 33s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/d19f0e2667ff4701bb20c54fa03ff11a : SUCCESS in 35m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"efdb4841aa64ec66df0c1c0db314012d99c6b523","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2023-10-31 14:04:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/71/899671/1/upstream-check/manila-tempest-plugin-ontap-dhss/83b8f74/ : RETRY_LIMIT in 8m 47s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/71/899671/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/4e83290/ : RETRY_LIMIT in 9m 32s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"a84f98af8af40b4e53cd5a2aa0f1735898b57ab2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-10-31 14:23:01.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/71/899671/1/check/DellEMC_PowerFlex_Manila_dhss_false/9e1722f/DellEMC_PowerFlex_Manila_dhss_false/643 : SUCCESS in 30m 30s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"444a5af636375cc35ef0172759b7eb8ff407436a","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2023-10-31 14:28:54.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/71/899671/1/check/DellEMC_PowerStore_Manila_dhss_false/60674e3/DellEMC_PowerStore_Manila_dhss_false/269 : SUCCESS in 36m 20s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"11a871cc8b96f606a493ec4ef5f1423d8616ecbe","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-10-31 15:14:24.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/71/899671/1/check/DellEMC_Unity_Manila_dhss_true/65fe400/DellEMC_Unity_Manila_dhss_true/727 : SUCCESS in 3h 15m 02s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/71/899671/1/check/DellEMC_Unity_Manila_dhss_false/c101c3a/DellEMC_Unity_Manila_dhss_false/770 : SUCCESS in 1h 16m 24s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"dba5b59a70fe0d210e20936d43b4a6828892a4c0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"real_author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-21 16:24:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_15334\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"_revision_number":2},{"id":"95cb0bb2cc133e3373b3d8f691ef366a7b859d60","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-21 16:56:39.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/71/899671/2/check/DellEMC_PowerFlex_Manila_dhss_false/8812b01/DellEMC_PowerFlex_Manila_dhss_false/1028 : SUCCESS in 32m 09s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"e8092e527352492494c09deda36de060f311aa10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 18:20:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93e6cdebb70b402db66d0feb7185def2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c8e10133f784ed39a842778b5329745 : SUCCESS in 9m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c2c73da4ff340c6942dbc2e59750bb7 : SUCCESS in 24m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09465c6a1e7d41a0bfd56eea98e91eed : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03c2b76e226c45a38d036abd33f43d0b : SUCCESS in 12m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/91c5f7253918485a91f1a76c0c267bf6 : SUCCESS in 9m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/062e14ceae7742b7a2ede46dfd775a98 : SUCCESS in 10m 43s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/93ee0020b73d452dbc936254f7c8f7ea : SUCCESS in 1h 09m 52s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4741f1a2cd8e4c598d5de2fe856d6809 : SUCCESS in 1h 45m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/63070558a1e74b96b5a01bb7406ce98b : SUCCESS in 1h 01m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8902f9aea2f748d58ce42e1bdcf846d5 : SUCCESS in 1h 02m 50s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/af9cb39bc1b644fb8cccf9ff9702995a : FAILURE in 1h 38m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/56473bbefba8484eaa35651c74aa6c50 : SUCCESS in 1h 20m 44s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9950f05d0a9548c992c7ec36efd8b938 : SUCCESS in 5m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/804202f161cf4e229f44353e1e604e40 : FAILURE in 3m 28s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/77297d94e47642c58408d943a77f4264 : SUCCESS in 39m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f1f26a1894114fb48fc03685fba47080 : SUCCESS in 32m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ab35b305419849aaba62c8da2e279a00 : FAILURE in 41m 48s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/06ce1dcf2c964043ba108384caf86295 : SUCCESS in 1h 02m 37s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/89bccf3eb1ba4c3eb797088c2a570f6f : SUCCESS in 1h 28m 12s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/809f8e0aa14a4019a5dc0561b11e660c : SUCCESS in 57m 53s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/14e9a164f26048b583229d900e6c6b29 : SUCCESS in 28m 43s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/0ec66371d7584da8b15df0db9c82a906 : SUCCESS in 31m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"73826db38523c2b39b01e1d4665d37da10985a16","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-21 18:57:46.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/71/899671/2/check/DellEMC_PowerStore_Manila_dhss_false/f6917d2/DellEMC_PowerStore_Manila_dhss_false/125 : SUCCESS in 36m 38s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"89ddd0abb1ff3211d336458294b41455b69cf426","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-21 18:57:59.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/71/899671/2/upstream-check/manila-tempest-plugin-ontap-dhss/0293217/ : RETRY_LIMIT in 3m 38s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/71/899671/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/f859032/ : RETRY_LIMIT in 3m 14s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"8268815b63a354e2b522c61e013492247f9365f6","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-21 19:56:40.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/71/899671/2/check/DellEMC_Unity_Manila_dhss_true/fa63b82/DellEMC_Unity_Manila_dhss_true/971 : SUCCESS in 3h 32m 09s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/71/899671/2/check/DellEMC_Unity_Manila_dhss_false/543f9fc/DellEMC_Unity_Manila_dhss_false/1040 : SUCCESS in 1h 35m 33s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"ea86fce739fb9cc9b7f9399d638ef524f0ed672b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-21 21:58:41.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-71-899671-2/NFS/ : SUCCESS in 4h 19m 31s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"aaac506a0454f47ff10f90bbfe15e08feb263f36","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-21 22:10:49.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/71/899671/2/check/DellEMC_PowerFlex_Manila_dhss_false/c83805b/DellEMC_PowerFlex_Manila_dhss_false/1038 : SUCCESS in 31m 18s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"c262149107e30fa15d5ed5354cb550e1e483f36b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-03-25 10:27:26.000000000","message":"Abandoned\n\nWe\u0027ve figured the issue out","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cf77fbfde19d2b462cb8246cb05c7cc9178d31b1","revisions":{"822cee5c600caa8fa8d3a60ac1896c793f65a49b":{"kind":"REWORK","_number":1,"created":"2023-10-31 11:59:03.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/71/899671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/71/899671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/71/899671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/71/899671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/71/899671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/71/899671/1"}}},"commit":{"parents":[{"commit":"6a1ad1c2f6a52125c78a5a2db4cbb9c22df2e793","subject":"db: Remove unnecessary helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6a1ad1c2f6a52125c78a5a2db4cbb9c22df2e793"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-10-31 11:41:49.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-10-31 11:58:48.000000000","tz":0},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (create)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (create)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Iceb40430548c94afa7c3e0a8c73e1c0d2f26efb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/822cee5c600caa8fa8d3a60ac1896c793f65a49b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/822cee5c600caa8fa8d3a60ac1896c793f65a49b"}]},"branch":"refs/heads/master"},"cf77fbfde19d2b462cb8246cb05c7cc9178d31b1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-03-21 16:24:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"real_uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/71/899671/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/71/899671/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/71/899671/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/71/899671/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/71/899671/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/71/899671/2"}}},"commit":{"parents":[{"commit":"4dc5f6b73015f0540a470a11c321f7ef05ec11c4","subject":"Merge \"Fix the link to get the next share replica\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4dc5f6b73015f0540a470a11c321f7ef05ec11c4"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-10-31 11:41:49.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-21 16:24:14.000000000","tz":0},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (create)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (create)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Iceb40430548c94afa7c3e0a8c73e1c0d2f26efb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cf77fbfde19d2b462cb8246cb05c7cc9178d31b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cf77fbfde19d2b462cb8246cb05c7cc9178d31b1"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
