)]}'
{"id":"openstack%2Fmanila~899893","triplet_id":"openstack%2Fmanila~master~I7ff8fc412d81c9f90db3d0c3b790f76b5afc87bc","project":"openstack/manila","branch":"master","topic":"sqlalchemy-20","hashtags":[],"change_id":"I7ff8fc412d81c9f90db3d0c3b790f76b5afc87bc","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete c)","status":"ABANDONED","created":"2023-11-01 18:16:45.000000000","updated":"2024-03-25 10:26:33.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"c05065a437a32d9cd3c26f940ac3745ee01b5401","_number":899893,"virtual_id_number":899893,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-22 15:51:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-11-01 19:35:43.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-11-01 19:38:10.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-01 20:00:53.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-11-01 20:03:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-02 17:47:41.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"35117a21b9e426371401effde6652df3e28a2f0d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-01 18:16:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"711ceba9a7e37e0f955093c29c4d78ad6257cc7c","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-01 19:35:43.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/93/899893/1/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/d92012a/ : FAILURE in 1h 15m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ba7def0bb0771f68cbf1ee08d73b0bc7c0560046","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-11-01 19:38:10.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/93/899893/1/check/DellEMC_PowerFlex_Manila_dhss_false/3cd6cd4/DellEMC_PowerFlex_Manila_dhss_false/665 : SUCCESS in 30m 33s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"635b3c2ce4b90785fc688ad2f44e641d253233dc","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-01 20:00:53.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-93-899893-1/NFS/ : SUCCESS in 1h 18m 58s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"f44b41e83605807a9bb9f08fbb6fd1d53e656511","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 20:03:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8ef24d33d584f3f823b7fcbd8e17bdc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8751852d16824621b2bc2675b5280d44 : SUCCESS in 11m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86c6b8349c864f04b2e8fdb5a71e2586 : FAILURE in 22m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24d3cf68e6474da7b259cd310dfe7921 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52123bd4147945cea233709be33ea19c : SUCCESS in 9m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59e8b614ad6944e0b0937f255b271c55 : SUCCESS in 8m 58s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ef9df7b0156548e08465338e7f2bc851 : SUCCESS in 56m 41s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/af13fd799c174d8a8c932dcfd06d7f7a : SUCCESS in 1h 44m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d7075be96af04b44bc605a7f95059f6a : SUCCESS in 1h 20m 56s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/253e25fd804c4a73a0e0ebf255194e3f : SUCCESS in 55m 25s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f9af3cc2cf9e4549b03f03cfe1c8de9c : SUCCESS in 1h 11m 06s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d0428ceb778c46318ca9b300ab87afcd : SUCCESS in 57m 34s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/c275f9366d4942e1b04c814818ce91fa : SUCCESS in 5m 51s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/6b47f137f8634930a5564d66fdbc6d65 : SUCCESS in 5m 14s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/54ceeb3bbd16470f815e5431171808cf : FAILURE in 37m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8930502b5635462285815460a07446ac : FAILURE in 45m 22s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/871c884456844f1ab68e5a928d0b425d : SUCCESS in 41m 56s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/989d232308e745aa94fa94579ab5d7b9 : FAILURE in 22m 36s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/0bbfa33ecbbe428ba45a1675ebf66241 : FAILURE in 22m 29s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/205f5226f1034602ab91aefc2fb5b028 : SUCCESS in 1h 30m 08s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/6b3f00d1af114d0095b68c1ea8416601 : SUCCESS in 39m 32s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2bc5373362b9462290555607e96692ae : SUCCESS in 53m 11s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c16d34edb146485f91a8e471cd38d1ff : SUCCESS in 42m 24s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1265f6f3dd3249d9b90d257ac5b8f533 : SUCCESS in 44m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"45bd4bbc1e491421e2c7f2c69d18c7ffa487fe8a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-02 11:48:44.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"068ff85453706c129cccd19e74d705587e2a9155","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-02 12:00:22.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/862b0823c4ab4ac7b3b39f31f03cdcef : RETRY_LIMIT in 1m 00s","accounts_in_message":[],"_revision_number":2},{"id":"e3d3109e9bccb85e5e9f6f0a2a426ef71febac92","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-11-02 12:51:43.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/93/899893/2/check/DellEMC_PowerFlex_Manila_dhss_false/08979ee/DellEMC_PowerFlex_Manila_dhss_false/669 : SUCCESS in 31m 59s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"29cc74f1df078bf0ad7c3add982dd459b6163ebb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 13:39:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6bd72e7af9d457fb3c369157a89be20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6568816377c49e7af9fca4d1adc1492 : SUCCESS in 10m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/000443940a7047e79ac17aa9cb27ce31 : FAILURE in 20m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f3ab2ab5e854b6795aac2ff12567a76 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22f29d2813ef4ec293b8890d614b93a4 : SUCCESS in 9m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/16faff0a5cb74ff4a40b043f2e419126 : SUCCESS in 10m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/b5eb71ee1e5f40a6a6fc45ec004ea1e5 : SUCCESS in 50m 50s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/cdbd041974da45848391818197b2edb6 : SUCCESS in 1h 45m 12s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/832b5e5060014400b5c20e5f532944f6 : SUCCESS in 1h 22m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a3194add04ca40c5a3f6fe9dc89f6425 : FAILURE in 1h 34m 40s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/eeacdafdda26495f97cd9efa1aad8430 : SUCCESS in 1h 42m 14s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/88bb92fecff84759a1e5bf6add2b49eb : SUCCESS in 1h 03m 13s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f7a8a7adbf0a49758ec3211ef1049972 : SUCCESS in 5m 27s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b64b6078e6604ceba68ebe3dc2ee9ed4 : SUCCESS in 4m 38s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6a8cf0b1f5b2448b969345170bf1805b : FAILURE in 43m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/48a8d54b2c954007a1f752b52c0bb090 : FAILURE in 28m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/f3acc23d90f84d9db4dcc00278aba7f5 : SUCCESS in 40m 14s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/d835adf868fe4b90aa72f6a5de34563d : FAILURE in 26m 26s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/c68807c0f2e44f13b261df70a56b7cf4 : FAILURE in 13m 33s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/0618c0fcf8a54dc2b5b565df53264319 : SUCCESS in 1h 19m 48s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/ab09efa9c40848c59bc9c29465dd1190 : SUCCESS in 48m 40s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fabee115675c4732a58fe9250c6865f4 : SUCCESS in 1h 13m 50s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1c6ccfbb05f04428a523dc7601182891 : SUCCESS in 27m 59s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/579d657c385f47978a6d1ca8a4914367 : SUCCESS in 36m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ecd5a396505d547927bb8218e8dd0340a194f464","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-02 14:19:55.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-93-899893-2/NFS/ : SUCCESS in 1h 47m 1s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"8e10240e8cba33a65b32530516fd50ea8eda4028","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-11-02 17:47:41.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/93/899893/2/check/DellEMC_Unity_Manila_dhss_true/7ff1dd1/DellEMC_Unity_Manila_dhss_true/738 : SUCCESS in 3h 27m 51s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/93/899893/2/check/DellEMC_Unity_Manila_dhss_false/b706c2a/DellEMC_Unity_Manila_dhss_false/786 : FAILURE in 14m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"cc286c07c64009d3bf7616a337389e035bff45eb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-02 18:06:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"7bf31bc3a6315b5ebcf00300182c0ee9a4ceca7e","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-02 18:17:42.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/e8fea6e3eef34248abc0e88bb15f1709 : RETRY_LIMIT in 58s","accounts_in_message":[],"_revision_number":3},{"id":"3abce4d050c469d2e5c6294c388d81d889946748","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-11-02 19:16:09.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/93/899893/3/check/DellEMC_PowerFlex_Manila_dhss_false/7dcd592/DellEMC_PowerFlex_Manila_dhss_false/674 : SUCCESS in 30m 51s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"f317d1cbe20a0615d3d5977d6ef4b546b1291eb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 19:56:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e85e5c6c5751417fa731a72628a1318c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e063633bd984e9f8b56003699dc9512 : SUCCESS in 9m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bc897dfb2c14210a5bd0b96d42540d4 : SUCCESS in 28m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1c1e7b0c97543b49e52dcfd98c9f8b6 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4226406e769c41b2962a9def8ca6cb3a : SUCCESS in 11m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ac1f4ce66804086a50cd8939d4c32c4 : SUCCESS in 8m 54s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/24bfb500454f479bbc3f56ec5b319e5c : SUCCESS in 1h 13m 21s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/601169d9e3574685ba9871d28d63619e : FAILURE in 17m 19s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c761c76618894043a062429c56e36b1b : SUCCESS in 1h 43m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/0d341360415e4ac7aa3c03d38e60c58b : SUCCESS in 49m 57s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/773b4b1af96b436dafee1e01ccbe6fec : FAILURE in 1h 30m 07s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/33da59f4495a4d2ab25c6bda8952e4e1 : SUCCESS in 1h 28m 51s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/87ae67e2106c4f4ab45a8efd81d8277a : SUCCESS in 4m 27s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4fbae04f38d34f8286d7a2ec7437f9c5 : SUCCESS in 3m 59s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3d73a3763e274c40bb7b84025a4221b4 : FAILURE in 36m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fd99a96bd939485aa6841504e1f60e05 : FAILURE in 45m 54s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6841233846f04b4189f6e84ff7d63d70 : FAILURE in 46m 23s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/7ef34d3bd99d4a2eae82cbbe26e9db76 : FAILURE in 19m 25s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4b1a6982049d4f52846cb7ebed374e8f : FAILURE in 13m 05s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/4d503d4a3cc04b25800d6d9772e9c0a8 : SUCCESS in 1h 16m 12s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/4d33ea77799b48b8ac2602dc72ac28ed : SUCCESS in 54m 31s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1ed50c16eb154927abb6fe6e152876a0 : SUCCESS in 46m 21s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/a474a55d87a54a2cb086aa38543ad316 : SUCCESS in 32m 05s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/9d3593946c8b47ddab268483ed03ad4b : SUCCESS in 33m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6dab40982fc2048f7f33afd82661a5ff8da83513","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-02 20:03:16.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-93-899893-3/NFS/ : SUCCESS in 1h 20m 17s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"ef312f471cb1a87b48d25ded76bab596c4e817b5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-11-03 07:30:27.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/93/899893/3/check/DellEMC_Unity_Manila_dhss_true/1d09bee/DellEMC_Unity_Manila_dhss_true/743 : SUCCESS in 3h 22m 30s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/93/899893/3/check/DellEMC_Unity_Manila_dhss_false/9907309/DellEMC_Unity_Manila_dhss_false/791 : SUCCESS in 1h 02m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"e3cafe44246393bda1828fc6fb495045d3fd6a31","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-03-21 17:21:01.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"4f85cc587293a12b0f5555bb0be05a96aa3dd073","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 19:18:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c83012202604a758735739977530487\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6395702eb66c4c868a4763a2b050ef86 : SUCCESS in 8m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ca1547dd08a4bc29668b7def7c36302 : SUCCESS in 24m 09s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee4c6deb11da40d08e1778ce0f02c83c : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b645fff5e11946229967a2c2566e8311 : SUCCESS in 12m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e658fb9411e74d3999cb130d04e6599d : SUCCESS in 8m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9fb7fdf6aa5d4ea7b94401bbfbc21067 : SUCCESS in 11m 56s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/f18cc02e067340aa857ccc8d1c02f9a7 : SUCCESS in 59m 11s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2b887e8d55a548949e3b2ea33a42b58b : SUCCESS in 1h 46m 41s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/e74e7a222dda43d082a8b2db7d4b2e90 : SUCCESS in 1h 04m 08s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/308e3e95889a498a85b7f9e862e9e6ed : SUCCESS in 1h 05m 05s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/dea9556afb4048b7a8b8ed212f7b3819 : SUCCESS in 1h 43m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/02bc02b35eab46a3aade98d81be5dd17 : SUCCESS in 59m 57s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/39ea4e4ffd7d4e95b70abec16c8e3256 : SUCCESS in 4m 19s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dd7b9d0219c74ecda63c15c0b30336f7 : FAILURE in 3m 25s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/219c43e9d3e64f5c8527e5034c62c9a7 : FAILURE in 43m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c8443a4581364e0e95f549e0ab9b3bee : FAILURE in 39m 58s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/cdb58eaf754e4cf88e7ddca267051f34 : FAILURE in 50m 07s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/d39916d83ba5416ea4497381eb038b18 : SUCCESS in 1h 17m 45s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3fa58d356f96424f8af395709562629e : SUCCESS in 57m 23s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/87d58c8194594ab6b2fe90f3e4d6fde4 : SUCCESS in 59m 39s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/fc4815451ee64e14ac5b4ab401d8b065 : SUCCESS in 33m 34s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3ebf5a5e981a4915a9170fdc3a38b697 : SUCCESS in 42m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"68160245cc00349d93d0889b8ce1e1c9f8ba8e47","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-21 21:39:31.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/93/899893/4/check/DellEMC_PowerFlex_Manila_dhss_false/aa4396c/DellEMC_PowerFlex_Manila_dhss_false/1037 : SUCCESS in 31m 09s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"f4c0ce224c45402f3d76e5a61dcad29c1deb2e01","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-22 03:25:57.000000000","message":"Patch Set 4:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"a95dc2bb4895221120bdc3cca30dc7181dd6cc44","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-22 13:02:31.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"5f18d4bf859fe86356f41aace1e72d5a3b64a343","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-22 14:00:09.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"cd3ed8ecd4e8b4f52dfcd2a393440aae8e255faa","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-22 15:02:41.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-93-899893-5/NFS/ : SUCCESS in 1h 27m 52s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"2724be8811deb8ad967bab5f44cafefbe8ae759e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-22 15:51:31.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-93-899893-6/NFS/ : SUCCESS in 1h 16m 37s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"5b867dc9414443bebf9da591950c332bb0d0278c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 15:51:50.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/899954 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"b00245185a508575f6555c96aa8a8023238c859e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-22 20:09:36.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/93/899893/6/check/DellEMC_PowerFlex_Manila_dhss_false/d8c769e/DellEMC_PowerFlex_Manila_dhss_false/1058 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"4d9e7d5d6a2ed1fbea081d574adba70fe5d96593","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-22 21:33:53.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/93/899893/6/check/DellEMC_Unity_Manila_dhss_true/4119eeb/DellEMC_Unity_Manila_dhss_true/981 : SUCCESS in 2h 38m 29s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/93/899893/6/check/DellEMC_Unity_Manila_dhss_false/91e7248/DellEMC_Unity_Manila_dhss_false/1059 : SUCCESS in 1h 02m 54s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"c05065a437a32d9cd3c26f940ac3745ee01b5401","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-03-25 10:26:33.000000000","message":"Abandoned\n\nWe\u0027ve figured the issue out","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"cc284935043b857a0cd6ffcddd296aecb8072051","revisions":{"a40840a3c4208327d2d58b1670032ddae6b23cf9":{"kind":"REWORK","_number":1,"created":"2023-11-01 18:16:45.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/93/899893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/93/899893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/899893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/899893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/899893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/93/899893/1"}}},"commit":{"parents":[{"commit":"01201171849406b1370d2da7950521d4cd345363","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/01201171849406b1370d2da7950521d4cd345363"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-01 18:16:05.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-01 18:16:05.000000000","tz":0},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere. We also need to add \u0027share_delete\u0027 in here.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I7ff8fc412d81c9f90db3d0c3b790f76b5afc87bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a40840a3c4208327d2d58b1670032ddae6b23cf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a40840a3c4208327d2d58b1670032ddae6b23cf9"}]},"branch":"refs/heads/master"},"aa720d901dc565638bfda0b2dbb30746716f06a2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-11-02 11:48:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/93/899893/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/93/899893/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/899893/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/899893/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/899893/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/93/899893/2"}}},"commit":{"parents":[{"commit":"09174ae6d6006fa466717911910c122b05483d48","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/09174ae6d6006fa466717911910c122b05483d48"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-01 18:16:05.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-02 10:27:23.000000000","tz":0},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete c)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete c)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere. We also need to add \u0027share_delete\u0027 in here.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I7ff8fc412d81c9f90db3d0c3b790f76b5afc87bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/aa720d901dc565638bfda0b2dbb30746716f06a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/aa720d901dc565638bfda0b2dbb30746716f06a2"}]},"branch":"refs/heads/master"},"3f19d7528d06e5a656ff71d37add242d7f890f90":{"kind":"REWORK","_number":3,"created":"2023-11-02 18:06:38.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/93/899893/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/93/899893/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/899893/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/899893/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/899893/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/93/899893/3"}}},"commit":{"parents":[{"commit":"b74709fdbb128d4890748793bc844937926828f1","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b74709fdbb128d4890748793bc844937926828f1"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-01 18:16:05.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-02 18:05:09.000000000","tz":0},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete c)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete c)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere. We also need to add \u0027share_delete\u0027 in here.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I7ff8fc412d81c9f90db3d0c3b790f76b5afc87bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3f19d7528d06e5a656ff71d37add242d7f890f90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3f19d7528d06e5a656ff71d37add242d7f890f90"}]},"branch":"refs/heads/master"},"2019c8226c1b61109d573b5ce4e31bf2bcd47c42":{"kind":"REWORK","_number":4,"created":"2024-03-21 17:21:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/93/899893/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/93/899893/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/899893/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/899893/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/899893/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/93/899893/4"}}},"commit":{"parents":[{"commit":"d0f70a9a4c5f56786611d7b1e959ac05f56ce8ee","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d0f70a9a4c5f56786611d7b1e959ac05f56ce8ee"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-01 18:16:05.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-21 17:20:16.000000000","tz":0},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete c)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete c)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere. We also need to add \u0027share_delete\u0027 in here.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I7ff8fc412d81c9f90db3d0c3b790f76b5afc87bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2019c8226c1b61109d573b5ce4e31bf2bcd47c42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2019c8226c1b61109d573b5ce4e31bf2bcd47c42"}]},"branch":"refs/heads/master"},"e4efea1f6d3772c1e256cf84aab9c7a495c5d9e4":{"kind":"REWORK","_number":5,"created":"2024-03-22 13:02:31.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/93/899893/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/93/899893/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/899893/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/899893/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/899893/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/93/899893/5"}}},"commit":{"parents":[{"commit":"a57a60e82cf748d0da58b4e0b0ed233f6ea88507","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a57a60e82cf748d0da58b4e0b0ed233f6ea88507"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-01 18:16:05.000000000","tz":0},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-03-22 13:01:40.000000000","tz":-240},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete c)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete c)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere. We also need to add \u0027share_delete\u0027 in here.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I7ff8fc412d81c9f90db3d0c3b790f76b5afc87bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e4efea1f6d3772c1e256cf84aab9c7a495c5d9e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e4efea1f6d3772c1e256cf84aab9c7a495c5d9e4"}]},"branch":"refs/heads/master"},"cc284935043b857a0cd6ffcddd296aecb8072051":{"kind":"REWORK","_number":6,"created":"2024-03-22 14:00:09.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/93/899893/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/93/899893/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/899893/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/899893/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/93/899893/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/93/899893/6"}}},"commit":{"parents":[{"commit":"a57a60e82cf748d0da58b4e0b0ed233f6ea88507","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a57a60e82cf748d0da58b4e0b0ed233f6ea88507"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-01 18:16:05.000000000","tz":0},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-03-22 13:56:43.000000000","tz":-240},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete c)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete c)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere. We also need to add \u0027share_delete\u0027 in here.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I7ff8fc412d81c9f90db3d0c3b790f76b5afc87bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cc284935043b857a0cd6ffcddd296aecb8072051"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cc284935043b857a0cd6ffcddd296aecb8072051"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
