)]}'
{"id":"openstack%2Fmanila~899954","triplet_id":"openstack%2Fmanila~master~I77f55bbb701469cdb2a9999f10c703616d02d9cc","project":"openstack/manila","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-03-25 10:26:44.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"11816":{"account":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"last_update":"2024-03-25 10:26:50.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I77f55bbb701469cdb2a9999f10c703616d02d9cc","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)","status":"ABANDONED","created":"2023-11-02 11:48:44.000000000","updated":"2024-03-25 10:26:54.000000000","total_comment_count":17,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"804af9d70838122b82f5b02795175da9f727480c","_number":899954,"virtual_id_number":899954,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-03-22 17:15:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-02 12:00:21.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2023-11-02 13:23:10.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-02 13:29:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-02 15:15:02.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2023-11-03 04:07:56.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2023-11-06 13:24:17.000000000","updated_by":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"reviewer":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"state":"CC"},{"updated":"2024-03-12 00:04:09.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2024-03-22 17:26:00.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-03-22 17:37:23.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"}],"messages":[{"id":"0e5ff68fd8d610169e83d476ce64d4a3da66d03b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-02 11:48:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1998fe5bc5e76e0530494d84be0bb502d3f5179","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-02 12:00:21.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy finger://master.pure-opendev-ci.dev.purestorage.com:7979/050cd30916c04efaa6dbf8a1ed68fc6d : RETRY_LIMIT in 58s","accounts_in_message":[],"_revision_number":1},{"id":"35f1b6e2aad2e09c2d53679648f047643c772b64","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-02 12:08:06.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d5d940479ac0fe7d5cf933bd5aaeff6d22377e0a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-02 12:11:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e1fb5f1203397e1300b0a7942d1c40055b03896","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-02 12:11:50.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"6412940ec945e45ac6afae871d648c080993b6ea","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-11-02 13:23:10.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/54/899954/1/check/DellEMC_PowerFlex_Manila_dhss_false/c17f3d0/DellEMC_PowerFlex_Manila_dhss_false/670 : SUCCESS in 31m 22s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"29e654cc4ec7018c3139caaea59e0e64983aa628","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 13:29:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d0c451d601a846bfb8d8c8b3200aee4c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b0e3a0b1961498d8e1383ea94a3081c : SUCCESS in 10m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1586a35c8fb64c3a8723aaa9e8fb74bd : FAILURE in 16m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f3e70095f884a49abdd594dc3cd79fc : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56d131baf1d04259b077d8979cd68b3e : FAILURE in 12m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7fde54c8061d4c3da66a954109537503 : FAILURE in 8m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/618a342f89584909b2978b28300e47b7 : SUCCESS in 1h 18m 06s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3448ee78be11446785989ef1fda29d25 : SUCCESS in 1h 36m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fbb9eedc3cb9481ea105f287306068cc : SUCCESS in 1h 01m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/47b4caea75bf4c348a560b3f50f7c6ef : SUCCESS in 51m 05s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/291940b0b5a34aea848e11c583101098 : SUCCESS in 1h 11m 06s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/acde5ff3c3b545d58bd1df5e040b5a31 : SUCCESS in 1h 10m 52s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/389806637aa2400fbb4242ed6809b4c9 : SUCCESS in 5m 04s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4bf68c7cd2c844159cfaeec0f8412161 : SUCCESS in 4m 12s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/6313885f458b4983a7e7d7771d7b5882 : FAILURE in 22m 54s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3563834e59474ec2852e9f35c4ec4be9 : FAILURE in 33m 48s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/0e917ec268d4478cbd7c51d1919bd556 : FAILURE in 56m 52s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/f6573493ea754f1f851a0341813dddf4 : FAILURE in 24m 29s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4791b3a615a54ffe9969d64f8ac5e3af : FAILURE in 23m 33s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/4477b888d8794330bb303f8318869aad : SUCCESS in 1h 18m 32s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/3cd106346fa042209e3ae073d64cbfde : SUCCESS in 59m 06s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a06120719904b9f939aed8413f3acf5 : SUCCESS in 59m 51s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0f627c4da8714576b0d3410f63ec8921 : SUCCESS in 40m 42s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/d04c3b8214c24f8f8d99f5f01ef1193a : SUCCESS in 44m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0513fb834bba4ef82603b0ee389d30837a62b44","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-02 15:15:02.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-899954-1/NFS/ : SUCCESS in 2h 12m 8s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"7b98ded8e5b3384feb766317610357f968f02fc6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-02 18:06:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"89f7c1b215bd3b3429c787ae188bfb9a08ff33a0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-11-02 18:08:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b8b5f4b3954a8cb9962380ce9f3405424aa33fd0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2023-11-02 18:45:19.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/54/899954/2/check/DellEMC_PowerFlex_Manila_dhss_false/e2b5344/DellEMC_PowerFlex_Manila_dhss_false/673 : SUCCESS in 30m 36s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"499c0690dc15d9397912ecfad0a82ae805e9231a","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2023-11-02 19:23:27.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pure-devstack-manila-tempest-aio-jammy http://openstack-logs.purestorage.com/54/899954/2/thirdparty-check/pure-devstack-manila-tempest-aio-jammy/f43c3e0/ : FAILURE in 1h 12m 57s","accounts_in_message":[],"_revision_number":2},{"id":"6047f5a3397e07990d27e15bc6ef3796718326df","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2023-11-02 19:42:46.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-899954-2/NFS/ : SUCCESS in 1h 29m 46s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"245a7f8d5ef50b49bb06f1237e4baca86aa069f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 21:18:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/32fb23482ffd40f48854c38522375443\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7182b6330ee4ddc88bbf3776cf2fee1 : SUCCESS in 12m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ab39a4b5d90450fb1b328e2103fad4e : SUCCESS in 23m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/076e2a94d56a49febf9dc977a0732f15 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/883e1ea403b14965b685d4de4890066c : SUCCESS in 11m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8cf585b6973149eeae6fe1a59a081faa : SUCCESS in 9m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d398fbf110bb4ec1a92b57cbd24d70aa : SUCCESS in 1h 08m 02s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/81320b7cb495482998027a6231c006bb : FAILURE in 2h 57m 58s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5ada7c727b4c4b539c0626a38912589b : TIMED_OUT in 3h 03m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3b61d81eed8149078e89ca487dc200ce : SUCCESS in 54m 11s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7bd7c566c22c4c42a9db7b6aef2b203f : SUCCESS in 1h 40m 50s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/10d8a2f2077543a9b2ef9b4b4d7ea34f : SUCCESS in 52m 39s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ea358766cfae4ccd912f69231c210d7e : SUCCESS in 5m 09s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/325f4dbd9c734d48b0c741ca1593e3b5 : SUCCESS in 4m 54s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a8c8dc9cac794c6fa83e718c882712e7 : FAILURE in 27m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/443a935989b446e395d99251658b046c : FAILURE in 31m 54s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/18c04d42671243618b9b5f17b0f06049 : SUCCESS in 38m 19s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/156a73a8138c4ee199a3f308928dda16 : FAILURE in 23m 31s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/b42eba09f2964ed888f1bb54cf301989 : FAILURE in 25m 07s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/a289c17fd86c46718c902976a23a0315 : SUCCESS in 1h 23m 41s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/c0a5b25d79bf42c2bb5e2d87424ef7b4 : SUCCESS in 1h 07m 06s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b830851c877e4a2e8d27c6c1fa41ff14 : SUCCESS in 1h 02m 14s (non-voting)\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/3e07fc5f171e43f4b7dd6c1e2dbd02c8 : SUCCESS in 41m 26s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f8a8700168774e12a26671912b8522b7 : SUCCESS in 32m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a3b521d7c8b1b0b7f5e62a6fa552e8eca8173e1d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2023-11-03 04:07:56.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/54/899954/2/check/DellEMC_Unity_Manila_dhss_true/c1f816c/DellEMC_Unity_Manila_dhss_true/742 : SUCCESS in 2h 59m 58s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/54/899954/2/check/DellEMC_Unity_Manila_dhss_false/bc669ec/DellEMC_Unity_Manila_dhss_false/790 : FAILURE in 14m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"f8846607bc483ad063e2994c88d4fdf4048a7d3c","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2023-11-06 13:24:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0628f3555aaa692f2b1d56c05af18a45a8a64674","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-03-12 00:04:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ec5bb9815fcc240d7ca3553e333fcac96216feea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 02:30:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/afe6e60a45d548918b80ebcde6cbcc28\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bd9f7dee9c848e69caaf25412b580f1 : SUCCESS in 10m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c432d7daa0674f4cb3a746827e4b0be1 : FAILURE in 19m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06b7233cf1894d0b968a86f51e5501f0 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d949f3cd2b39425591806982c38cb50c : SUCCESS in 8m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa35f2a5dc084035bec81afd02c93504 : SUCCESS in 7m 41s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2f7400e291014d93a580e71a9a01d617 : SUCCESS in 1h 11m 46s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/925655f64d0f4fd0ae800431070b79bf : SUCCESS in 2h 12m 31s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3bcb722c3af348dd80568f2bc6d715e8 : SUCCESS in 2h 16m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/bd7c31c543394ce1b354622afa386ed1 : SUCCESS in 45m 08s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/c87c138b03e44b81bdea7afce039e37e : FAILURE in 1h 36m 34s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/98cf22e252a74d8782c0187a6560fa1f : SUCCESS in 1h 16m 14s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/08e99c92923a402587ce1a1626ea5094 : SUCCESS in 4m 20s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/660a2649f4c54552a53f0694acbf7417 : FAILURE in 3m 15s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8826906fffdb4c14ad96b2b8eba719e7 : FAILURE in 40m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/57699512f585490bb037d96ccf12fa63 : FAILURE in 44m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/18a6c6ba469041ebbf963aa26c255cc0 : FAILURE in 50m 30s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/8b4e76d41c84400999b76217c1703fee : FAILURE in 20m 51s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/8b23514f8b8f45f58d0a5b75724a7a1c : FAILURE in 21m 13s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/dcf780370c6b43cabdc6d1523125df3b : SUCCESS in 50m 36s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/83254735f5be45bea0bacd5e118a5e76 : SUCCESS in 39m 28s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/514703d237f84339b75af7015831d4b7 : SUCCESS in 49m 10s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/98066bdccf814926bbfb42a4af5b5a04 : SUCCESS in 23m 30s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7bcd7922948940d786617d239e32c4cb : SUCCESS in 32m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d927dc6ada40ced321563ee31698f3b95f86d97d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"real_author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-21 16:26:38.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_15334\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"_revision_number":3},{"id":"19dd1f4829162d026a9b3abc00f33211012a8eab","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-03-21 17:01:57.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"4e6b147f728bdb4edb5144a28e22037f7848e2e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 19:38:06.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2acac9b05f8845f590cbd7e5fafc1338\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1346b88eaf104454987f69e7076edfde : SUCCESS in 10m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5b97719161c4e9da0bf4c596ccd443e : SUCCESS in 27m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0b3dae8e14249beb08e27dc95f25486 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8017239ff00d4165ab50af3d1a002508 : SUCCESS in 9m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6aab8dab53f040b1a20c6a4c816f1afe : SUCCESS in 10m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/50e659e877fe4ea887b7cabc207095fa : SUCCESS in 10m 42s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/68d1515d9fcd40adbd99a258654ec666 : SUCCESS in 1h 14m 59s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/eb5e832930b748d2824c4bd820ba850a : FAILURE in 2h 28m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/f2f5b1c1ed19471fa5c6667586510a1a : SUCCESS in 1h 14m 53s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/5e8f36cee1034e57b5b28072725789cf : SUCCESS in 1h 02m 44s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d112606aa7a544799802e948962bc02c : FAILURE in 1h 17m 06s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a5dbfff9d7d5481fafba2f3d083ffd56 : SUCCESS in 1h 20m 41s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/2e327806de7a41079a95972b270f2567 : SUCCESS in 5m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a29b8ce873eb4350ace15e28f2fb2bfb : FAILURE in 3m 15s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c411c8bbc4d34ec2b89cf2d005ba0c43 : FAILURE in 41m 55s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/b12d123385864c98a1ca659afd52191e : FAILURE in 34m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/96b590642b084a259da752cec2914ffa : SUCCESS in 47m 09s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/82036e5be1fa43b4b170c4bdf647db4e : SUCCESS in 1h 44m 02s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/f135ec65b4b14ef6a2b83a8915d55cd1 : SUCCESS in 1h 04m 31s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/045a3118783043a2b0586f28a21c2783 : SUCCESS in 1h 06m 56s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/671046adf4404bbca81b54d3b3c8214e : SUCCESS in 29m 58s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/97fdb907e9484ab8a014377c70272833 : SUCCESS in 38m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"62991f5f24e252a278737f8258e78597f6bc2bde","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-21 21:08:22.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/54/899954/4/check/DellEMC_PowerFlex_Manila_dhss_false/6942204/DellEMC_PowerFlex_Manila_dhss_false/1036 : SUCCESS in 34m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"b59a2bac8bce41167e7ff8b45ddd9720e6963e92","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"real_author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-22 01:16:21.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_15334\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"_revision_number":5},{"id":"cca3a41edc291da33caac76a92cc12135ec064c3","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-22 02:02:04.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-899954-3/NFS/ : SUCCESS in 3h 46m 30s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"d56bfdda0487962a15a5222408e28514eeb0f194","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-22 02:19:02.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/54/899954/5/check/DellEMC_PowerFlex_Manila_dhss_false/c5db381/DellEMC_PowerFlex_Manila_dhss_false/1043 : SUCCESS in 31m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"387fbd2e3c933ada9cfff68b7ce00ddf01376e72","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-22 02:45:18.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6644dc282d8f303b6679f5134c70ff5089e0fb43","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-22 03:18:57.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"da3938bc4ddef67325105347162ba4e6a1bc8bc2","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-22 03:34:11.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-899954-4/NFS/ : SUCCESS in 2h 22m 19s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"67a565153b3637f14639ea02d5881d53a61b26c9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-22 03:46:55.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/54/899954/7/check/DellEMC_PowerFlex_Manila_dhss_false/4e92fe7/DellEMC_PowerFlex_Manila_dhss_false/1046 : SUCCESS in 27m 27s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"b1398a3d7266339a68f93e1cdeffb44eae46fa9f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-22 04:03:40.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"766dc90a0401beabe698cb75ded7d565437dc1b6","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-22 04:04:23.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-899954-5/NFS/ : SUCCESS in 1h 19m 32s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"75611f894e21fc2e93f8021c19db07342ee810ac","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-22 04:36:33.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-899954-7/NFS/ : SUCCESS in 1h 10m 37s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"391f5ab0a958b8069e0457596101a0bcf6256a24","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-22 04:48:47.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/54/899954/8/check/DellEMC_PowerFlex_Manila_dhss_false/c1937e6/DellEMC_PowerFlex_Manila_dhss_false/1048 : SUCCESS in 29m 40s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"806af9b6eaa78ab2446f8caedcf91c72278a6e0d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-22 04:54:17.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"86cb006ca1b8dabeb9240ced310f1ddd9276c7f9","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-22 04:55:13.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-899954-6/NFS/ : SUCCESS in 1h 54m 26s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"0f744c3b8bb99d3976e17515cea92de1b50eba8d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-22 05:24:44.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-899954-8/NFS/ : SUCCESS in 1h 20m 42s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"6ea5acb029e95717c76a964339e3edc0a7be7411","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-22 05:59:33.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/54/899954/9/check/DellEMC_PowerFlex_Manila_dhss_false/ea71779/DellEMC_PowerFlex_Manila_dhss_false/1051 : SUCCESS in 31m 42s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"6014d0e7912105acbcc47962cce4363775360d22","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-22 07:04:33.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-899954-9/NFS/ : SUCCESS in 1h 30m 31s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"b187cc733d9deb23c7584defeeb7c010f9ee6944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 07:13:59.000000000","message":"Patch Set 9: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8658e1577a294587b229e77984365427\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffdfdeb837814b989439c9dbd4a4dd71 : SUCCESS in 12m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdb5b88c375d4a16aa6f387abb8d4e66 : SUCCESS in 23m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5450f3dfc53a410593bdd7bfd1b66c3c : FAILURE in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/583838e6a8c442178d18c6ef83b38327 : SUCCESS in 12m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7df6481a7e804010a024fd926013f30b : SUCCESS in 8m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/442c78729f614dcba0054eb38efdaedf : SUCCESS in 12m 47s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/24be25c16a1245858b0060845f605ca7 : SUCCESS in 1h 11m 11s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/658d3711813e4e449414d111d25c2de7 : SUCCESS in 2h 13m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a8351933b31249c8abdd868545fef7f3 : SUCCESS in 1h 27m 46s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7b4a329ef3e848eba5fe31eddc37b027 : SUCCESS in 53m 13s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/17977f69945643898d7649298ddf5c7e : FAILURE in 1h 25m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/28ce2e70036e4864b68d7d9d1ce3b029 : SUCCESS in 1h 21m 02s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a67050ee17124eda9fb19af704ff6250 : SUCCESS in 4m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/18d420f8423b42b7919950deb9628725 : FAILURE in 2m 59s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/73c1793e51924008b38820edb4d4be6c : SUCCESS in 27m 17s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/21d7e6e1c2314aee834e92f013957579 : SUCCESS in 31m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/869d5ec242fb4160be120c6906f91735 : FAILURE in 48m 23s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/dba8b486e0ec46c687f4fbfb07ddefe0 : SUCCESS in 1h 11m 43s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/60abfe2c0e454f62a189e1b753da120f : SUCCESS in 58m 23s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/091cb39614344fce8c80274b592a2499 : SUCCESS in 39m 08s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/521e164fa3734023b360f7e05efd8354 : SUCCESS in 30m 31s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/d3a8bac435084a50b06b869616dd9fce : SUCCESS in 43m 40s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5cc1f52580472bbd85d002db1738cb869edf1e2b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-22 11:55:17.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"3092461ed7e81fb63c2ad1e06ab11b7df7fa6d45","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-22 12:20:11.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7b76378392178d4a1e10a1f08c5620722b710c31","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-22 12:54:11.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"b422290046d9c450db853e747c1c78bc6b6ea4be","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-22 12:55:25.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5645a027fd89a198dcb6bf5c3fab1f6b988321b5","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-22 13:40:47.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"de8a9fb1de53f9c126f4f9ab54455d2aaec4bf5d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-22 14:09:20.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"ab97483e8222d8ebd11d2fa9f49a09c7606b4e97","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-22 14:13:28.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c27554a83822ee9112aadc053de536f80e0347f1","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-22 14:13:35.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-899954-10/NFS/ : SUCCESS in 1h 38m 47s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":10},{"id":"b493fdf94345db7a77df9e37f532c18596657fca","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-22 14:48:07.000000000","message":"Patch Set 11:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-899954-11/NFS/ : SUCCESS in 1h 43m 16s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":11},{"id":"203091c58bb45a29c9e048317d7f4a01d363a5eb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-22 15:12:27.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"a1357befbad46d94d84524b5322b9565b714bd1b","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-22 15:13:09.000000000","message":"Patch Set 13:\n\n(4 comments)","accounts_in_message":[],"_revision_number":13},{"id":"24fd5b39d32919711949f59b9a0ff9d530e204cd","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-22 15:14:10.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"c6a4993166b2f1b2cdfb527e85f3ad0de24db187","author":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"date":"2024-03-22 16:14:22.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"be7ae87b3359f032e2b99b951de53e0efd083101","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-22 16:20:10.000000000","message":"Patch Set 12:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-899954-12/NFS/ : SUCCESS in 1h 15m 20s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":12},{"id":"4415bd5f970717b3218038712a72be4e59607d8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 17:15:04.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fffc1a4b8614e3c85c0bf2879896ce4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d934ee7def784b09a5385cd7e1b40cc4 : SUCCESS in 12m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7a9a05923de4dd9bcdf8af739ac5d71 : SUCCESS in 29m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/781fdc9bdbe9477fbe9753fd0db99e0a : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ec8882a2654486489218d5f87f3c032 : SUCCESS in 10m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/43126d6f6b0b4c50806457e59f76051b : SUCCESS in 8m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8de5e27a41694b1095f1681bed2b2199 : SUCCESS in 15m 31s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/de8195fa423347c29a49a756ff5d72c5 : SUCCESS in 59m 51s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3d17366a6acd44f59d007c8004ba09e3 : SUCCESS in 1h 59m 23s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5dd8ae90b2714f80b609ded10eb77b1c : SUCCESS in 1h 18m 12s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/136b06715aef454c90c4e94512f1c329 : SUCCESS in 1h 00m 56s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e6ad3806f2ba4f3fb831876968514140 : FAILURE in 1h 35m 47s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/48467b3ee2734ff89fb9efa788ca1ebd : FAILURE in 1h 16m 09s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/cb5354cc829847a9be893f29cc8446f3 : SUCCESS in 4m 53s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/20e71f66782240769cecc2f44928b0d9 : FAILURE in 4m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/ff5a9f95e2864fbd909f515bfeee8786 : SUCCESS in 27m 47s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/06dec1c8cbf440278c0945a049541ff0 : SUCCESS in 49m 02s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/aa1657ab86fd4ea9921306eff5a2db46 : SUCCESS in 41m 38s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/2dd8276c1a944cc594d7291b22377690 : SUCCESS in 1h 25m 41s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/fc331c0f51084a458332f0114cd577ca : SUCCESS in 57m 57s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/84aa3b7d13154309854b5c8a39a9d95b : SUCCESS in 55m 50s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b12d9f1f1e844b549991b761422527b5 : SUCCESS in 28m 51s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7b5dc693c68c472486d3ddafe7f61712 : SUCCESS in 31m 55s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"48e8b825c85f7b6204351e6010fcf7539171b963","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-22 17:26:00.000000000","message":"Patch Set 13:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/899954/13/upstream-check/manila-tempest-plugin-ontap-dhss/36f7f8d/ : RETRY_LIMIT in 4m 11s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/54/899954/13/upstream-check/manila-tempest-plugin-ontap-no-dhss/ce9d155/ : RETRY_LIMIT in 4m 14s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":13},{"id":"c94467e83f7a68f862e60077bfd1fa175346c3ab","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-22 17:37:23.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/54/899954/13/check/DellEMC_PowerStore_Manila_dhss_false/bfe1423/DellEMC_PowerStore_Manila_dhss_false/132 : FAILURE in 21m 57s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":13},{"id":"6ddec67bda84312ee5a2d67a527f1b4d7c94534b","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-22 19:17:47.000000000","message":"Patch Set 13:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-54-899954-13/NFS/ : SUCCESS in 1h 12m 56s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":13},{"id":"a560499ed114fd48d07fb801e20172e4e02c8cf2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-23 08:09:49.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/54/899954/13/check/DellEMC_PowerFlex_Manila_dhss_false/af11b93/DellEMC_PowerFlex_Manila_dhss_false/1062 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":13},{"id":"357db9c52a23095991dfc0e64155a93e2ee6c295","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-23 10:34:40.000000000","message":"Patch Set 13:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/54/899954/13/check/DellEMC_Unity_Manila_dhss_true/9cb29a6/DellEMC_Unity_Manila_dhss_true/985 : SUCCESS in 3h 29m 47s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/54/899954/13/check/DellEMC_Unity_Manila_dhss_false/68c8edb/DellEMC_Unity_Manila_dhss_false/1064 : SUCCESS in 1h 07m 17s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":13},{"id":"08287ca8d766f4bd25505daf78db22cdd8ee5e2e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-03-25 10:26:44.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"3d9c288fd009eb342a8b7b1a28b35b6ff206ed0d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-03-25 10:26:47.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":13},{"id":"62cd60058bf45004bd7300583030b38ec817d493","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-03-25 10:26:50.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":13},{"id":"804af9d70838122b82f5b02795175da9f727480c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-03-25 10:26:54.000000000","message":"Abandoned\n\nWe\u0027ve figured the issue out","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"299b36b3f894c2cf5c48b230f9683cfa63361166","revisions":{"09174ae6d6006fa466717911910c122b05483d48":{"kind":"REWORK","_number":1,"created":"2023-11-02 11:48:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/54/899954/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/54/899954/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/54/899954/1"}}},"commit":{"parents":[{"commit":"1c103e63ed99d33d493f7e1fc179d003e130a7ee","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete a)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1c103e63ed99d33d493f7e1fc179d003e130a7ee"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-02 10:27:08.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-02 10:27:23.000000000","tz":0},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere. We also need to add \u0027share_delete\u0027 in here.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\nChange-Id: I77f55bbb701469cdb2a9999f10c703616d02d9cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/09174ae6d6006fa466717911910c122b05483d48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/09174ae6d6006fa466717911910c122b05483d48"}]},"branch":"refs/heads/master"},"b74709fdbb128d4890748793bc844937926828f1":{"kind":"REWORK","_number":2,"created":"2023-11-02 18:06:38.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/54/899954/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/54/899954/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/54/899954/2"}}},"commit":{"parents":[{"commit":"1c103e63ed99d33d493f7e1fc179d003e130a7ee","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete a)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1c103e63ed99d33d493f7e1fc179d003e130a7ee"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-02 10:27:08.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-02 18:04:48.000000000","tz":0},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere. We also need to add \u0027share_delete\u0027 in here.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\nChange-Id: I77f55bbb701469cdb2a9999f10c703616d02d9cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b74709fdbb128d4890748793bc844937926828f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b74709fdbb128d4890748793bc844937926828f1"}]},"branch":"refs/heads/master"},"9c9fee84c997e1bb1d8d06bd0306d8daec61c8b3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-03-21 16:26:38.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"real_uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/54/899954/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/54/899954/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/54/899954/3"}}},"commit":{"parents":[{"commit":"fb60c1c9e00bc2ba991d943ed022526b44a02a99","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete a)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fb60c1c9e00bc2ba991d943ed022526b44a02a99"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-02 10:27:08.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-21 16:26:38.000000000","tz":0},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere. We also need to add \u0027share_delete\u0027 in here.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\nChange-Id: I77f55bbb701469cdb2a9999f10c703616d02d9cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9c9fee84c997e1bb1d8d06bd0306d8daec61c8b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9c9fee84c997e1bb1d8d06bd0306d8daec61c8b3"}]},"branch":"refs/heads/master","description":"Rebase"},"d0f70a9a4c5f56786611d7b1e959ac05f56ce8ee":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-03-21 17:01:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/54/899954/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/54/899954/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/54/899954/4"}}},"commit":{"parents":[{"commit":"a60c7ea218383b4d0b50316ae72592cefabcc9a7","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete a)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a60c7ea218383b4d0b50316ae72592cefabcc9a7"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-02 10:27:08.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-21 16:52:41.000000000","tz":0},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere. We also need to add \u0027share_delete\u0027 in here.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\nChange-Id: I77f55bbb701469cdb2a9999f10c703616d02d9cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d0f70a9a4c5f56786611d7b1e959ac05f56ce8ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d0f70a9a4c5f56786611d7b1e959ac05f56ce8ee"}]},"branch":"refs/heads/master"},"8044bb2e653b570100f4881501bdef47a45bdfa2":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-03-22 01:16:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"real_uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/54/899954/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/54/899954/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/54/899954/5"}}},"commit":{"parents":[{"commit":"647c325e5f149f26c4c83246ebd69619660c5851","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete a)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/647c325e5f149f26c4c83246ebd69619660c5851"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-02 10:27:08.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-03-22 01:16:21.000000000","tz":0},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere. We also need to add \u0027share_delete\u0027 in here.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\nChange-Id: I77f55bbb701469cdb2a9999f10c703616d02d9cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8044bb2e653b570100f4881501bdef47a45bdfa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8044bb2e653b570100f4881501bdef47a45bdfa2"}]},"branch":"refs/heads/master","description":"Rebase"},"797e70479296ce961cf4dd8266b7ab0155e94115":{"kind":"REWORK","_number":6,"created":"2024-03-22 02:45:18.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/54/899954/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/54/899954/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/54/899954/6"}}},"commit":{"parents":[{"commit":"647c325e5f149f26c4c83246ebd69619660c5851","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete a)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/647c325e5f149f26c4c83246ebd69619660c5851"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-02 10:27:08.000000000","tz":0},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-03-22 02:43:59.000000000","tz":-240},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere. We also need to add \u0027share_delete\u0027 in here.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\nChange-Id: I77f55bbb701469cdb2a9999f10c703616d02d9cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/797e70479296ce961cf4dd8266b7ab0155e94115"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/797e70479296ce961cf4dd8266b7ab0155e94115"}]},"branch":"refs/heads/master"},"79934e6833731eb1ec9454e99889e4b04291e68b":{"kind":"REWORK","_number":7,"created":"2024-03-22 03:18:57.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/54/899954/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/54/899954/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/54/899954/7"}}},"commit":{"parents":[{"commit":"647c325e5f149f26c4c83246ebd69619660c5851","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete a)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/647c325e5f149f26c4c83246ebd69619660c5851"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-02 10:27:08.000000000","tz":0},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-03-22 03:18:51.000000000","tz":-240},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere. We also need to add \u0027share_delete\u0027 in here.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\nChange-Id: I77f55bbb701469cdb2a9999f10c703616d02d9cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/79934e6833731eb1ec9454e99889e4b04291e68b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/79934e6833731eb1ec9454e99889e4b04291e68b"}]},"branch":"refs/heads/master"},"fa318991e2a682e09b470a069c22c9e43d17fbbd":{"kind":"REWORK","_number":8,"created":"2024-03-22 04:03:40.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/54/899954/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/54/899954/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/54/899954/8"}}},"commit":{"parents":[{"commit":"647c325e5f149f26c4c83246ebd69619660c5851","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete a)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/647c325e5f149f26c4c83246ebd69619660c5851"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-02 10:27:08.000000000","tz":0},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-03-22 03:59:04.000000000","tz":-240},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere. We also need to add \u0027share_delete\u0027 in here.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\nChange-Id: I77f55bbb701469cdb2a9999f10c703616d02d9cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fa318991e2a682e09b470a069c22c9e43d17fbbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fa318991e2a682e09b470a069c22c9e43d17fbbd"}]},"branch":"refs/heads/master"},"4fd62370b8aa8b93b936cc750d7a3517cbe7c767":{"kind":"REWORK","_number":9,"created":"2024-03-22 04:54:17.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/54/899954/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/54/899954/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/54/899954/9"}}},"commit":{"parents":[{"commit":"647c325e5f149f26c4c83246ebd69619660c5851","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete a)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/647c325e5f149f26c4c83246ebd69619660c5851"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-02 10:27:08.000000000","tz":0},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-03-22 04:52:40.000000000","tz":-240},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere. We also need to add \u0027share_delete\u0027 in here.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\nChange-Id: I77f55bbb701469cdb2a9999f10c703616d02d9cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4fd62370b8aa8b93b936cc750d7a3517cbe7c767"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4fd62370b8aa8b93b936cc750d7a3517cbe7c767"}]},"branch":"refs/heads/master"},"c55fc346b28ea3f24cd4f73d37d9c067412e5b95":{"kind":"REWORK","_number":10,"created":"2024-03-22 11:55:17.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/54/899954/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/54/899954/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/54/899954/10"}}},"commit":{"parents":[{"commit":"647c325e5f149f26c4c83246ebd69619660c5851","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete a)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/647c325e5f149f26c4c83246ebd69619660c5851"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-02 10:27:08.000000000","tz":0},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-03-22 11:55:09.000000000","tz":-240},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere. We also need to add \u0027share_delete\u0027 in here.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\nChange-Id: I77f55bbb701469cdb2a9999f10c703616d02d9cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c55fc346b28ea3f24cd4f73d37d9c067412e5b95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c55fc346b28ea3f24cd4f73d37d9c067412e5b95"}]},"branch":"refs/heads/master"},"a57a60e82cf748d0da58b4e0b0ed233f6ea88507":{"kind":"REWORK","_number":11,"created":"2024-03-22 12:54:11.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/54/899954/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/54/899954/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/54/899954/11"}}},"commit":{"parents":[{"commit":"647c325e5f149f26c4c83246ebd69619660c5851","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete a)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/647c325e5f149f26c4c83246ebd69619660c5851"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-02 10:27:08.000000000","tz":0},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-03-22 12:54:06.000000000","tz":-240},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere. We also need to add \u0027share_delete\u0027 in here.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\nChange-Id: I77f55bbb701469cdb2a9999f10c703616d02d9cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a57a60e82cf748d0da58b4e0b0ed233f6ea88507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a57a60e82cf748d0da58b4e0b0ed233f6ea88507"}]},"branch":"refs/heads/master"},"73a6698536ff7adee2f5025ccc67fad1f5af5c67":{"kind":"REWORK","_number":12,"created":"2024-03-22 14:09:20.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/54/899954/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/54/899954/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/54/899954/12"}}},"commit":{"parents":[{"commit":"647c325e5f149f26c4c83246ebd69619660c5851","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete a)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/647c325e5f149f26c4c83246ebd69619660c5851"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-02 10:27:08.000000000","tz":0},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-03-22 14:08:53.000000000","tz":-240},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere. We also need to add \u0027share_delete\u0027 in here.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\nChange-Id: I77f55bbb701469cdb2a9999f10c703616d02d9cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/73a6698536ff7adee2f5025ccc67fad1f5af5c67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/73a6698536ff7adee2f5025ccc67fad1f5af5c67"}]},"branch":"refs/heads/master"},"299b36b3f894c2cf5c48b230f9683cfa63361166":{"kind":"REWORK","_number":13,"created":"2024-03-22 15:12:27.000000000","uploader":{"_account_id":11816,"name":"mike bayer","display_name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","username":"zzzeek","status":"Red Hat"},"ref":"refs/changes/54/899954/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/54/899954/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/54/899954/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/54/899954/13"}}},"commit":{"parents":[{"commit":"647c325e5f149f26c4c83246ebd69619660c5851","subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete a)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/647c325e5f149f26c4c83246ebd69619660c5851"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-11-02 10:27:08.000000000","tz":0},"committer":{"name":"Mike Bayer","email":"mike_mp@zzzcomputing.com","date":"2024-03-22 15:12:20.000000000","tz":-240},"subject":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)","message":"WIP: db: Migrate \"share instance\", \"share replica\" APIs to enginefacade (delete b)\n\nThis will be rolled into [1] once I figure out what I\u0027m doing wrong\nthere. We also need to add \u0027share_delete\u0027 in here.\n\n[1] https://review.opendev.org/c/openstack/manila/+/880205\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\nChange-Id: I77f55bbb701469cdb2a9999f10c703616d02d9cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/299b36b3f894c2cf5c48b230f9683cfa63361166"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/299b36b3f894c2cf5c48b230f9683cfa63361166"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
