)]}'
{"id":"openstack%2Fmanila~911085","triplet_id":"openstack%2Fmanila~master~Ia40d1d9e313160ebced50af1a2d5fa78707cfa60","project":"openstack/manila","branch":"master","attention_set":{},"removed_from_attention_set":{"35058":{"account":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"last_update":"2024-03-13 08:25:41.000000000","reason":"\u003cGERRIT_ACCOUNT_35058\u003e replied on the change","reason_account":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-03-14 02:03:31.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia40d1d9e313160ebced50af1a2d5fa78707cfa60","subject":"Add user defined extra capabilities","status":"ABANDONED","created":"2024-03-05 10:59:44.000000000","updated":"2024-03-14 02:03:31.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e4a762ea4c697e144c1933d86ba0e7807042bb85","_number":911085,"virtual_id_number":911085,"owner":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-13 09:19:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-05 11:58:59.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-05 13:02:08.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-03-05 13:32:42.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-03-05 14:10:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-08 01:23:53.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"CC"},{"updated":"2024-03-08 01:23:58.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-03-13 08:24:40.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"1072dffe01c0f5b025ed60ffb7bdfb2c3c7e41e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2024-03-05 10:59:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8b21735fb7b0893bb6ee341c48a1897fdd1293b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-05 11:58:59.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/85/911085/1/check/DellEMC_PowerFlex_Manila_dhss_false/ac6164e/DellEMC_PowerFlex_Manila_dhss_false/931 : SUCCESS in 31m 50s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"3c809d3e31068dacf27ce539244455e19a569451","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-05 13:02:08.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-85-911085-1/NFS/ : SUCCESS in 1h 57m 10s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"11745ae62961c4ed307504473329c55cd87b94c9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-05 13:32:42.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/85/911085/1/check/DellEMC_Unity_Manila_dhss_true/8c49ad0/DellEMC_Unity_Manila_dhss_true/903 : SUCCESS in 2h 28m 48s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/85/911085/1/check/DellEMC_Unity_Manila_dhss_false/fa12284/DellEMC_Unity_Manila_dhss_false/960 : SUCCESS in 56m 46s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"aaf1c46d9ad95cdcd49a1f728026aad372bf7a85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-05 14:10:05.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ebad810f7b68492dbc375da19dff31ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49aa596f3fc5487f90fa5e2a29d353e6 : SUCCESS in 9m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bc045a198734bf8a381597d17e0a0ea : FAILURE in 23m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ff1a91febda48268930564b96acab75 : FAILURE in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae655a51204e431f94ae12295defc0ee : FAILURE in 11m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/04335da817ad49e7b3e68dd1608295d7 : FAILURE in 9m 39s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7e366a0d2a3e4fd2a16a0c34e97a4895 : SUCCESS in 1h 19m 55s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c842d4cebda64fb2b5a50589a30961df : FAILURE in 2h 09m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/bf02c6c705414216b07b6fbfbc588351 : TIMED_OUT in 3h 02m 49s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/de5f239c3a6c4dcc977ba218634412c5 : FAILURE in 1h 02m 21s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/bc70fa26020e47449030160b3b7fd49e : FAILURE in 1h 31m 20s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/5c670a1c5fc84bc6b49b889a245f9a32 : SUCCESS in 1h 20m 59s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/bc1f1faf2e614b28b79a865f3f86181e : SUCCESS in 6m 28s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/187d9656153249ec856d4f550db877d0 : FAILURE in 3m 35s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/95a7cc40be104b7b96a271c12430ea4a : SUCCESS in 35m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/af96965e8678478fb250689703bd4953 : SUCCESS in 36m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2cdb65e79ce44cf294bd5abb30da358e : SUCCESS in 45m 34s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/81622cc66aa345ebb9cb5a1f029dace2 : FAILURE in 21m 53s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2a871775ad3d43a09aafa4820bff65ed : FAILURE in 27m 43s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/7c058d8e37774ae483eb1561e7da142f : SUCCESS in 56m 30s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/49081cc6162f4807afc8a9b7fb0e1ab6 : SUCCESS in 1h 02m 34s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c6291194165f4c01b341ae4fd569cdd1 : SUCCESS in 1h 07m 29s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d500278d35e546a992c1a6ce08da768b : SUCCESS in 35m 09s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/a68985a1bdf243bab20755d786042da5 : SUCCESS in 44m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"646c039f941411c1ce682992ea9e0cf5b33e7051","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2024-03-06 03:12:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"6c950ec696dcf43e5dc3472b72bd161db6ddd9fb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-06 03:44:57.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/85/911085/2/check/DellEMC_PowerFlex_Manila_dhss_false/3246b99/DellEMC_PowerFlex_Manila_dhss_false/939 : SUCCESS in 31m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"0852c0676520c944a56569aa8b2475849ac98922","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 04:56:23.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/538c6fdb1b9540e0a98395ae4095dd30\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a0fd228801b4796830d0b5ffd192990 : SUCCESS in 11m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1b25b2dbd7e4802bfe70de821eb115c : FAILURE in 21m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/052e7cf7e7754ad2ae56376f2886ee7c : FAILURE in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d138a5a0f8c346d3b5ade3b4595d9327 : FAILURE in 9m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b36a7a8b02a4dcbb29615888767770b : FAILURE in 8m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9ad190cf5eeb4578be5217d999741dde : FAILURE in 1h 12m 58s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/4f7f3cc3151d4e0285d3f3cccfd215c9 : SUCCESS in 1h 34m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/109eb474e2d848898ecdbb8a8525c6ae : SUCCESS in 1h 19m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/637df2b0d5964a72b88f6e58256accba : SUCCESS in 1h 05m 20s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ab89e383556844fa832b7844058f556c : SUCCESS in 1h 35m 14s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1e82e95a9bea4b1cb782ad4ac8c31fa4 : SUCCESS in 56m 37s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/a10a9ea6753641779c2b8648db20d32f : SUCCESS in 5m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4259305e373c4672869717f04533881d : FAILURE in 3m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/dbdeb4118e46405b93e945115fe98019 : SUCCESS in 24m 35s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/386ec082d1e04e3787b077103088362e : SUCCESS in 29m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7508c396019e4cdeb285795ffe758c34 : SUCCESS in 1h 03m 00s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/ac83943ed13d4249be2c3c5157e4c2a8 : FAILURE in 23m 12s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/4423339227974fcfb907bde2899ae564 : FAILURE in 14m 20s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/547ee925ffdf49f6812574cc115cf76d : SUCCESS in 53m 48s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/e2d91f8d82ea4134979aac7a479e2d7d : SUCCESS in 59m 47s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/05cb5115336a47dca722211a0a46d294 : SUCCESS in 1h 03m 21s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/79f3ef4e6c6a47e79c649a1483fca822 : SUCCESS in 47m 43s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/db2ad10d85544812a4009e8d4779bff9 : SUCCESS in 42m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"30641f57114625fbac17643ba0cbac4de8c6309e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-06 05:20:43.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-85-911085-2/NFS/ : SUCCESS in 2h 7m 27s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"7061008620179f989b5fff75cecc483a726e4ada","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-06 06:00:30.000000000","message":"Patch Set 2:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/85/911085/2/check/DellEMC_Unity_Manila_dhss_true/d38cd16/DellEMC_Unity_Manila_dhss_true/908 : SUCCESS in 2h 47m 14s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/85/911085/2/check/DellEMC_Unity_Manila_dhss_false/c728c1e/DellEMC_Unity_Manila_dhss_false/965 : SUCCESS in 1h 03m 57s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"ad6ef39044a6da69e06a235719db0a5aae1e11de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2024-03-06 06:24:48.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"546bba6406b9316ca4780aadb54b38032fb37185","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-06 08:03:52.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-85-911085-3/NFS/ : SUCCESS in 1h 38m 44s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"57c7e950f78d80e5f883d7a571eb666952dc2583","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-06 08:20:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a394937bdc674bdabb92e9c0efdad825\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/774f3233a6c14b5486057041a4095c1d : SUCCESS in 9m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c869c6a546f045bd91be4ded3ae63a5e : FAILURE in 25m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7bf2920ace04234b4adfbb45a48cece : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22230ffbda034358a7b92a7655e4121e : FAILURE in 8m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d9141d39af84cd98a71ec1089137e3a : FAILURE in 8m 37s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/c78f2d7bc57944dbb979569fd0a9e98d : SUCCESS in 1h 12m 49s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/77ae32e16f304c9db093d40fbc475824 : SUCCESS in 1h 48m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/75d4758b6db947febb42f79ba34a3209 : SUCCESS in 58m 00s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ea14580aaf3b484db6cd5c24ef45320a : FAILURE in 1h 15m 47s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/703b2bd40dab44c98cf8766906dd2af4 : SUCCESS in 1h 31m 21s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/8da4d511bd994692a238ff030e6b101b : SUCCESS in 1h 13m 15s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4ff90fe904714b99918529c466f8633c : SUCCESS in 5m 14s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/d7ed6dc51bf245bdb9481db29d6fe2ab : FAILURE in 4m 14s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7b0478d708db422bb41045d7cb7ea3fa : SUCCESS in 39m 53s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2ab91118d0494a1e88d9a16ce49ea314 : SUCCESS in 43m 14s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3034364586fc4047912e2fa0d611680c : FAILURE in 32m 27s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/2ee87389e42f4216bb87fdff60f0f44a : FAILURE in 22m 11s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2e3a63f382ff4ba49c2221a79a0c97cf : FAILURE in 25m 38s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/0dd4e37196a44c28bbdc92736cdaa9e3 : SUCCESS in 59m 41s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/304f6f64fc114d6ebd680b7e23cf6ffd : SUCCESS in 1h 00m 42s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a3d2ece4838d45c9b991197432fad2f9 : SUCCESS in 40m 08s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/0e6339a1e2fa49ce8b61f30bdd0d7b75 : SUCCESS in 43m 57s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b267b9e0774a4e57b381163f7860996d : SUCCESS in 40m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c1dd16c1f9d0cb7b4c829c012417e48a53dd9fa5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-06 08:34:40.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/85/911085/3/check/DellEMC_Unity_Manila_dhss_true/cbed412/DellEMC_Unity_Manila_dhss_true/909 : SUCCESS in 2h 09m 35s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/85/911085/3/check/DellEMC_Unity_Manila_dhss_false/582226e/DellEMC_Unity_Manila_dhss_false/966 : SUCCESS in 53m 48s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"c2e964ffd5b143487e2dabd1f649e88c6a02f5de","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-03-08 01:23:53.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0ede025114eb67a957c49f1575136378dd8e0eb2","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-03-08 01:23:58.000000000","message":"Patch Set 3: Code-Review-1","accounts_in_message":[],"_revision_number":3},{"id":"fac3d2aa39aee0c3a483bf4adeef8ad115a43750","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2024-03-13 08:24:15.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"9565e434e18f340f7cdc9dfbd56d10ea7f77e68a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 08:24:31.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/manila for 911085,4","accounts_in_message":[],"_revision_number":4},{"id":"09a0634d3e76bed86384c8f78f1184c83c65b649","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-13 08:24:40.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/manila for 911085,4","accounts_in_message":[],"_revision_number":4},{"id":"011b326f3be5442c91abd23bb6b30cf79817cbb9","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2024-03-13 08:25:41.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"1e9c2d4c2ecaca3d2cd992c0dec95487bbb25cfe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2024-03-13 08:26:26.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nThe following files contain Git conflicts:\n* manila/share/manager.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"08a3f5c9b1f0aaa636eb5a2c95eb199d46bd6e4d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2024-03-13 08:26:32.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":5},{"id":"941fb6c3a7cb799725ac1cde5c334a0af4d07aca","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2024-03-13 08:31:13.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":5},{"id":"2527ca8ede1c3a8eb43d80c92a28984f645a278d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 09:19:33.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/61044e9254144423b799a7379d52825f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ab170233b8040a69ad07bfa1693c8c6 : FAILURE in 5m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b39310529c9d45728b2edfac1683a6e8 : SUCCESS in 17m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c70828f096f4cc2954de2764eeecca2 : FAILURE in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7795a670f98b469cab118ae6b1567119 : FAILURE in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b6bcfc37ee394b49a91d4fd950f468b4 : FAILURE in 6m 50s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/8ea327f5c7a24c1a9144cb2c0c5ed308 : FAILURE in 26m 35s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/baee6fa8ee8242b6a025c860b5df8e7b : FAILURE in 46m 36s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ae9b1da17b994d1dab066a42d45bf1ae : FAILURE in 34m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ed74c29538b648dd9df3f973c860d659 : FAILURE in 39m 49s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ce20fa1e30624bf2b4da21f4c89f9842 : FAILURE in 39m 01s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a03c2f42e4364e149047426a59ac5ac6 : FAILURE in 33m 32s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/bcb5cc21d9cd47759dd7d06e618538d0 : FAILURE in 4m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b3af4dbf3e3942a88271b81698376c46 : FAILURE in 3m 31s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/df5d119dc0074e98a6781984b70cdf67 : FAILURE in 10m 58s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/03fbd63d2f804a6784a0171a17ead311 : FAILURE in 11m 05s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/97543cf0cf694d7988b3b3cdee81e466 : FAILURE in 20m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/e6594d7ac011463c8e819efe1325c58c : FAILURE in 11m 05s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/da4b356a54934c1ca8983fd989007cdc : FAILURE in 11m 07s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/cf66aa6911c14a628cbb1da1459b551e : FAILURE in 40m 25s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/cc9774f0c6064ab4b84baf512f7b9f0e : FAILURE in 37m 25s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dd371ac2a0da4a83b2e6780613bd03ca : FAILURE in 37m 12s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ddd2ad17c698498a9c386e0aab67a776 : FAILURE in 40m 09s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f7a2df2a6a0345f0ae83f285c5de62b7 : FAILURE in 27m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0ae1e3b89a5bcb74010325507f4dc7c3abd013b9","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-13 09:51:43.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-85-911085-4/NFS/ : SUCCESS in 1h 27m 4s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"45176e95b7ac458b131bff7da1227ab002feddf9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-13 11:35:52.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/85/911085/5/check/DellEMC_PowerFlex_Manila_dhss_false/705f587/DellEMC_PowerFlex_Manila_dhss_false/988 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"5dee2b3616b7c3228a5850b8317f435892b5611f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-13 12:17:11.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/85/911085/5/check/DellEMC_Unity_Manila_dhss_true/6791f8a/DellEMC_Unity_Manila_dhss_true/940 : FAILURE in 40m 41s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/85/911085/5/check/DellEMC_Unity_Manila_dhss_false/9993ba9/DellEMC_Unity_Manila_dhss_false/1007 : FAILURE in 3h 50m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"e5f808d6455dea6ae07ed9186c38e11f4d067fe8","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2024-03-14 02:03:10.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":5},{"id":"52a58b90acc314211d9ef7e7eba7d76a0ebc9a73","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2024-03-14 02:03:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e4a762ea4c697e144c1933d86ba0e7807042bb85","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2024-03-14 02:03:31.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"eb40359d4ae27ae81abca352e9fae792d05bae25","revisions":{"ad5497fa98ff40b6dda5b207820dda0282eb2e1b":{"kind":"REWORK","_number":1,"created":"2024-03-05 10:59:44.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/85/911085/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/85/911085/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/911085/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/911085/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/911085/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/85/911085/1"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"niuke","email":"niuke@inspur.com","date":"2024-03-05 10:55:12.000000000","tz":480},"committer":{"name":"niuke","email":"niuke@inspur.com","date":"2024-03-05 10:57:48.000000000","tz":480},"subject":"Add user defined extra capabilities","message":"Add user defined extra capabilities\n\nProvide a mechanism that will allow sys admins who are managing manila\ninstallations with multiple backends to assign key/value pairs to\nbackends.\n\nFor example a sysadmin could specify the key \u0027service_level\u0027 with the\nvalues \u0027high\u0027, \u0027medium\u0027 and \u0027low\u0027 for different backends. The sys admin\nspecifies the \u0027service_level\u0027 for each backend in manila.conf as a json\nstring, for example:\n\nextra_capabilities\u003d\u0027{\"service_level\"\u003d\"high\"}\u0027\n\nor\n\nextra_capabilities\u003d\u0027{\"service_level\"\u003d\"medium\"}\u0027\n\nChange-Id: Ia40d1d9e313160ebced50af1a2d5fa78707cfa60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ad5497fa98ff40b6dda5b207820dda0282eb2e1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/ad5497fa98ff40b6dda5b207820dda0282eb2e1b"}]},"branch":"refs/heads/master"},"10e78591f008fea154ba003aa7b38e639396abf7":{"kind":"REWORK","_number":2,"created":"2024-03-06 03:12:57.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/85/911085/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/85/911085/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/911085/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/911085/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/911085/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/85/911085/2"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"niuke","email":"niuke@inspur.com","date":"2024-03-05 10:55:12.000000000","tz":480},"committer":{"name":"niuke","email":"niuke@inspur.com","date":"2024-03-06 03:12:41.000000000","tz":480},"subject":"Add user defined extra capabilities","message":"Add user defined extra capabilities\n\nProvide a mechanism that will allow sys admins who are managing manila\ninstallations with multiple backends to assign key/value pairs to\nbackends.\n\nFor example a sysadmin could specify the key \u0027service_level\u0027 with the\nvalues \u0027high\u0027, \u0027medium\u0027 and \u0027low\u0027 for different backends. The sys admin\nspecifies the \u0027service_level\u0027 for each backend in manila.conf as a json\nstring, for example:\n\nextra_capabilities\u003d\u0027{\"service_level\"\u003d\"high\"}\u0027\n\nor\n\nextra_capabilities\u003d\u0027{\"service_level\"\u003d\"medium\"}\u0027\n\nChange-Id: Ia40d1d9e313160ebced50af1a2d5fa78707cfa60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/10e78591f008fea154ba003aa7b38e639396abf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/10e78591f008fea154ba003aa7b38e639396abf7"}]},"branch":"refs/heads/master"},"79268f0a93ce9444615ba696522267a1958b547a":{"kind":"REWORK","_number":3,"created":"2024-03-06 06:24:48.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/85/911085/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/85/911085/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/911085/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/911085/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/911085/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/85/911085/3"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"niuke","email":"niuke@inspur.com","date":"2024-03-05 10:55:12.000000000","tz":480},"committer":{"name":"niuke","email":"niuke@inspur.com","date":"2024-03-06 06:23:38.000000000","tz":480},"subject":"Add user defined extra capabilities","message":"Add user defined extra capabilities\n\nProvide a mechanism that will allow sys admins who are managing manila\ninstallations with multiple backends to assign key/value pairs to\nbackends.\n\nFor example a sysadmin could specify the key \u0027service_level\u0027 with the\nvalues \u0027high\u0027, \u0027medium\u0027 and \u0027low\u0027 for different backends. The sys admin\nspecifies the \u0027service_level\u0027 for each backend in manila.conf as a json\nstring, for example:\n\nextra_capabilities\u003d\u0027{\"service_level\"\u003d\"high\"}\u0027\n\nor\n\nextra_capabilities\u003d\u0027{\"service_level\"\u003d\"medium\"}\u0027\n\nChange-Id: Ia40d1d9e313160ebced50af1a2d5fa78707cfa60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/79268f0a93ce9444615ba696522267a1958b547a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/79268f0a93ce9444615ba696522267a1958b547a"}]},"branch":"refs/heads/master"},"b0b69041f4701a212a952d626d604cbc37a58846":{"kind":"REWORK","_number":4,"created":"2024-03-13 08:24:15.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/85/911085/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/85/911085/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/911085/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/911085/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/911085/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/85/911085/4"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"niuke","email":"niuke@inspur.com","date":"2024-03-05 10:55:12.000000000","tz":480},"committer":{"name":"niuke","email":"niuke@inspur.com","date":"2024-03-13 08:22:43.000000000","tz":480},"subject":"Add user defined extra capabilities","message":"Add user defined extra capabilities\n\nProvide a mechanism that will allow sys admins who are managing manila\ninstallations with multiple backends to assign key/value pairs to\nbackends.\n\nFor example a sysadmin could specify the key \u0027service_level\u0027 with the\nvalues \u0027high\u0027, \u0027medium\u0027 and \u0027low\u0027 for different backends. The sys admin\nspecifies the \u0027service_level\u0027 for each backend in manila.conf as a json\nstring, for example:\n\nextra_capabilities\u003d\u0027{\"service_level\"\u003d\"high\"}\u0027\n\nor\n\nextra_capabilities\u003d\u0027{\"service_level\"\u003d\"medium\"}\u0027\n\nChange-Id: Ia40d1d9e313160ebced50af1a2d5fa78707cfa60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b0b69041f4701a212a952d626d604cbc37a58846"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b0b69041f4701a212a952d626d604cbc37a58846"}]},"branch":"refs/heads/master"},"eb40359d4ae27ae81abca352e9fae792d05bae25":{"kind":"REWORK","_number":5,"created":"2024-03-13 08:26:26.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/85/911085/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/85/911085/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/911085/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/911085/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/85/911085/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/85/911085/5"}}},"commit":{"parents":[{"commit":"d4fb4319f5d59db67a8f42975f27d719005935d4","subject":"Merge \"refactoring: modify function parameters and annotations to match\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d4fb4319f5d59db67a8f42975f27d719005935d4"}]}],"author":{"name":"niuke","email":"niuke@inspur.com","date":"2024-03-05 10:55:12.000000000","tz":480},"committer":{"name":"Ke Niu","email":"niuke@inspur.com","date":"2024-03-13 08:26:26.000000000","tz":0},"subject":"Add user defined extra capabilities","message":"Add user defined extra capabilities\n\nProvide a mechanism that will allow sys admins who are managing manila\ninstallations with multiple backends to assign key/value pairs to\nbackends.\n\nFor example a sysadmin could specify the key \u0027service_level\u0027 with the\nvalues \u0027high\u0027, \u0027medium\u0027 and \u0027low\u0027 for different backends. The sys admin\nspecifies the \u0027service_level\u0027 for each backend in manila.conf as a json\nstring, for example:\n\nextra_capabilities\u003d\u0027{\"service_level\"\u003d\"high\"}\u0027\n\nor\n\nextra_capabilities\u003d\u0027{\"service_level\"\u003d\"medium\"}\u0027\n\nChange-Id: Ia40d1d9e313160ebced50af1a2d5fa78707cfa60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/eb40359d4ae27ae81abca352e9fae792d05bae25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/eb40359d4ae27ae81abca352e9fae792d05bae25"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
