)]}'
{"id":"openstack%2Fmanila~911694","triplet_id":"openstack%2Fmanila~master~I6a0575c1ed86213010e50fe1b7a733cdf7fa1736","project":"openstack/manila","branch":"master","topic":"bug/1867030","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-04-09 16:18:43.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-04-05 15:12:21.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"36853":{"account":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"last_update":"2024-04-09 17:09:35.000000000","reason":"Change was submitted"},"36860":{"account":{"_account_id":36860,"name":"Mike Lutaaya","display_name":"mike-lu","email":"hafswabilan@gmail.com","username":"mike-lu"},"last_update":"2024-04-09 17:09:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6a0575c1ed86213010e50fe1b7a733cdf7fa1736","subject":"Add delete CLI helper to ShareCommands","status":"MERGED","created":"2024-03-07 11:17:43.000000000","updated":"2024-04-09 17:10:46.000000000","submitted":"2024-04-09 17:09:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":63,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911694-bug/1867030","meta_rev_id":"6066e9cb42a79da6b6922335b5ea4249c6f28fb7","_number":911694,"virtual_id_number":911694,"owner":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-09 17:09:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-04-09 16:18:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":2,"date":"2024-04-05 15:12:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2024-04-09 16:18:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},{"_account_id":36860,"name":"Mike Lutaaya","display_name":"mike-lu","email":"hafswabilan@gmail.com","username":"mike-lu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-07 11:22:50.000000000","updated_by":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2024-03-07 11:22:50.000000000","updated_by":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-03-07 12:53:30.000000000","updated_by":{"_account_id":36860,"name":"Mike Lutaaya","display_name":"mike-lu","email":"hafswabilan@gmail.com","username":"mike-lu"},"reviewer":{"_account_id":36860,"name":"Mike Lutaaya","display_name":"mike-lu","email":"hafswabilan@gmail.com","username":"mike-lu"},"state":"CC"},{"updated":"2024-03-07 14:04:52.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-03-07 14:41:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-07 17:26:25.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-03-13 12:36:58.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-13 14:24:27.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-03-13 16:29:59.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"}],"messages":[{"id":"e07a80c4d4ef1c6809fc9f00de74071e00c71513","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-07 11:17:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95368bc094e61ee5d990a76b87d55e7af2f9a4d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-07 12:25:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b26cdc21915191d48d48d62b26fd2a4f5f6fb56d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-07 12:29:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"56da1c60277dd2f1d5bf25f43a995ba565f779d8","author":{"_account_id":36860,"name":"Mike Lutaaya","display_name":"mike-lu","email":"hafswabilan@gmail.com","username":"mike-lu"},"date":"2024-03-07 12:53:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"064f56bc6fef34c97ab09eb404cdd04a85d78800","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-03-07 13:18:15.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c8988a8435bb46cae518e70fd69a2f0ec409cf09","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-07 14:04:52.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-2/NFS/ : SUCCESS in 1h 20m 4s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"a0bea7052917ddea6334f564aa11a7ce93823ebe","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-07 14:24:27.000000000","message":"Patch Set 1:\n\nBuild failed. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-1/NFS/ : FAILURE in 2h 9m 37s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"210d373d1d86a2b493d8fb2e412c1e9bd863b73d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 14:41:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbb507757c474d44938bc06f01955484\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e325a5c89eb466f8308dabee31c3e3b : SUCCESS in 10m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64211d2e645a4ddc8fab4504dc65cb92 : SUCCESS in 21m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35819b1e29984ab8b431a9b678635834 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02538913528b4ab6a523730449d18e3d : SUCCESS in 8m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/81f09bac7cb74e279f8b41e946e48d36 : SUCCESS in 12m 13s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bbb75457de0744c79738cd64fde0220d : SUCCESS in 58m 27s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e74dc0d9ed2c42b5b8a1790158f1e132 : SUCCESS in 2h 10m 25s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/c7079fee8415465bb0dac8b3f93ba41b : SUCCESS in 1h 20m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/332e266aae384376a2e5b23eb323a9a6 : SUCCESS in 1h 12m 30s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/598575fcd9684eac9b81179e3811e252 : FAILURE in 1h 25m 06s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0901768057a741419f6d8282f169a386 : SUCCESS in 1h 14m 17s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/30a37598a81347cdb715bcb69b430e75 : SUCCESS in 4m 41s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ef5967a1f2d948ee9a93ac2b620bffc5 : FAILURE in 3m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/5b640ecc8db2455da26c38d6271efb56 : SUCCESS in 36m 16s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/1f91575d22da4ef7afc5cf7fdecf1e85 : SUCCESS in 37m 51s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2b9c8e3eb64f4b298dce5850a457fa1b : SUCCESS in 56m 58s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/a52ae3c1f87b46b8b096d938489a36c4 : FAILURE in 18m 53s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/a4b65194a52d472085d64e649699a8ff : FAILURE in 19m 33s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/94c9fb3fb9814bc3a79c4b7d94907b9f : SUCCESS in 1h 05m 34s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/11c89777ab534904a2909bce84f49226 : SUCCESS in 48m 36s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a1fb3a2f6bf845eabab03de5be89da00 : SUCCESS in 1h 01m 36s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ae4667ee40db4bd580ba3cb78c0e2960 : SUCCESS in 41m 50s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b164bdc3492e4fd4ac0571db8307f047 : SUCCESS in 33m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4b779ffa4838634e67408608c5b14a3e3b0abed4","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-07 15:04:40.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-3/NFS/ : SUCCESS in 1h 49m 52s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"67a80edbf05f5eb82299eeee935d25a1c990bd56","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-07 17:26:25.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/94/911694/3/upstream-check/manila-tempest-plugin-ontap-dhss/4027cdb/ : FAILURE in 2h 25m 20s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/94/911694/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/200669e/ : SUCCESS in 2h 43m 27s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"5d25c9183824f8ee16f9dd189d019aa55ce57b55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-08 13:37:01.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"5f74322a1be780278368570e11c5f144cda909b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-08 15:22:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69d393f7ca6c4788bc9b7683dbb0d0fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25c28a280322461e89f2fd1142e3c130 : SUCCESS in 9m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d042cf0aa8c244889533a0a761ed4f9b : FAILURE in 25m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d517dc7085cf4ef992a4e203b17eea8e : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c628a2058bf64c46b30bd7607d077636 : SUCCESS in 12m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a0cdb214f7ea49d6b6f56208b2877bfa : SUCCESS in 8m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/fc2d0cd0a7a14a48b11c4e054b2a8af4 : SUCCESS in 1h 24m 43s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/982afaf05d00461599a844159cad56fa : SUCCESS in 1h 43m 05s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ae263578088a41298d22cd7b42f09d4d : SUCCESS in 1h 02m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/22239ef11fbb4f1fb4df0440909e89e8 : SUCCESS in 1h 07m 47s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/5ce96a97533a4acca2b5621b50df3491 : SUCCESS in 1h 22m 51s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9089596a68f345d3a444b97198bb8f81 : SUCCESS in 1h 14m 31s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/869d6451632a479e892f3bc6637b7992 : SUCCESS in 6m 21s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/4538e179bc97472f937acfd6aa2150af : FAILURE in 4m 31s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/04a439f3f8a947738f651d512faed596 : SUCCESS in 31m 03s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a0f679f7b92f4e1cadc8b9f02a7523f7 : SUCCESS in 38m 49s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/19c4312301a74740b5749f634f6b8507 : FAILURE in 35m 28s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/f663c49c6494468b8e04219e38af3da7 : FAILURE in 12m 22s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/9afe682b1f8e494b97062312374fc407 : FAILURE in 16m 24s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/2590cd8d8cde46e6bbb09f268f567a5f : SUCCESS in 1h 27m 15s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/b6d523da11b84b4b87efe564e8901725 : SUCCESS in 1h 01m 29s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/85d38fd043184c8ba0504b72fd3d77e0 : SUCCESS in 52m 44s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/573a5373ce9847a8be6af91bc6f791f4 : SUCCESS in 28m 45s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/56c63987f6844a9c9a0202324228b681 : SUCCESS in 31m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"650a35645f7f525f9eba9fee696480a1df93e298","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-08 16:42:58.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-4/NFS/ : SUCCESS in 2h 8m 55s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"78a549907b4d8f03608fae5f324a630941137cb3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-08 17:28:13.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/94/911694/4/upstream-check/manila-tempest-plugin-ontap-dhss/db28d09/ : SUCCESS in 1h 19m 36s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/94/911694/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/48f27bf/ : SUCCESS in 1h 49m 21s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"ec493d3295f399b4dcc5cb248b76d34c00ffe4e8","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-03-09 00:23:51.000000000","message":"Patch Set 4: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b69577f142874417ddca4b1d64a7bb7a479362ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-13 11:47:24.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"bccdb9c0c5778a7e55e9d52d9a1982a58570c6f4","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-13 11:47:37.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"7659ef83424ecd28adbde6e9843bba38d5092a29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-13 11:56:54.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fc6b6c90c1ce68a2a9b0262efec50d2d743b96d1","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-13 11:57:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5362169cb22ffdc47dc743f3964e098bd3bedb3e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-13 12:36:58.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/6/check/DellEMC_PowerFlex_Manila_dhss_false/79e7778/DellEMC_PowerFlex_Manila_dhss_false/990 : SUCCESS in 30m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"63961c5733f09fea246a629457247ac2018b80cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-13 13:49:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9ab3d89a97d40e68c4d0ffd482f611a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfbcddd2680e4b4b954d094a96342342 : SUCCESS in 9m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2544405907b4474898019ddb541ca38f : SUCCESS in 24m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e68e92af7d4142d0b2d08d8ea7726220 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9af73a293d4e41ba9dc1cd1d8825e37a : SUCCESS in 10m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5fecab6cb1a44cf19878881afbd833b3 : SUCCESS in 7m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d098ef7a810d4343a501b623daa0aaa9 : SUCCESS in 1h 09m 30s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f64040b0cb3548ac85562b38daa4fa4e : SUCCESS in 1h 49m 09s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/765b11836f1445bb83de26b940f31000 : SUCCESS in 1h 14m 27s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e3bce0947907496582581a724c1a82bd : SUCCESS in 1h 10m 42s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/5c25a24a82374539aa4c1f460fe02bf9 : FAILURE in 1h 28m 38s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4f42c4f4ce6d4615833c6d75774fa59c : FAILURE in 1h 04m 15s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f22ff9c3f59d4e2ebe36b92c9bad1286 : SUCCESS in 4m 23s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/eb0ffbaed7e44b6b948fd9b4a0fec9c9 : FAILURE in 4m 29s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2f6b5f458ef44c949c12dec2feab9fd2 : SUCCESS in 37m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9d540cf0ed614aa6960c8a733c05a045 : SUCCESS in 32m 43s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4d7d006e6e814cb1b9314e8cc4bb1432 : SUCCESS in 47m 44s (non-voting)\n- manila-tempest-plugin-glusterfs-native https://zuul.opendev.org/t/openstack/build/fc1f3ce25f5e4b23a4368b8931ac8261 : FAILURE in 23m 00s (non-voting)\n- manila-tempest-plugin-glusterfs-nfs https://zuul.opendev.org/t/openstack/build/2853648153454234a7a0f50a913bfbef : FAILURE in 24m 44s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/4e3a046318584e4c91c30fc2969c212a : SUCCESS in 1h 21m 26s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/271fdd5da8594925a5848ad00f7ea719 : SUCCESS in 1h 00m 34s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/16641d8af9f1441b8b686bc56e17baef : SUCCESS in 47m 30s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/39032af967dc4e26a5fb6f120a1150ee : SUCCESS in 36m 35s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/504f61253f0a49a983a70def182a46fc : SUCCESS in 40m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"543d217b8f9da1e040a65c1e6bdc7811ef5d253e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-13 14:20:32.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/6/check/DellEMC_PowerFlex_Manila_dhss_false/e80054f/DellEMC_PowerFlex_Manila_dhss_false/993 : SUCCESS in 29m 54s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"0e96eb670edbadf0750b0b05d4aca7455d7368f2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-13 14:24:27.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/6/check/DellEMC_PowerStore_Manila_dhss_false/6d52260/DellEMC_PowerStore_Manila_dhss_false/106 : SUCCESS in 34m 50s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"532ed656e7644eb8192fb6596f1d497ab36ce600","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-13 15:02:56.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-5/NFS/ : SUCCESS in 1h 38m 14s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"0262efdbcae9c8ba2cc0f080dabaf1d65771c467","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-13 15:05:22.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/94/911694/6/upstream-check/manila-tempest-plugin-ontap-dhss/8b6b6ef/ : FAILURE in 1h 15m 21s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"cfeaafd50a087f26f8db3afba0c1942261a826cf","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-13 15:56:13.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-6/NFS/ : SUCCESS in 1h 31m 30s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"4001c48085d5bdcd6d7d4d09be5a382a8999df3a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-13 16:29:59.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/94/911694/6/check/DellEMC_Unity_Manila_dhss_true/ca920a4/DellEMC_Unity_Manila_dhss_true/942 : SUCCESS in 2h 18m 24s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/6/check/DellEMC_Unity_Manila_dhss_false/def2983/DellEMC_Unity_Manila_dhss_false/1009 : SUCCESS in 56m 26s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"4115034a1b70d02c3f581e7110fe2bd4c500ea9c","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-03-13 18:15:38.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4289fbfbc50fd22bedb691f70f7ebfccd3eecdf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-16 07:59:22.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"d6372ed0a11294fdd774412ba1e42a456e5dc1d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-16 09:21:20.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b9d9296330b15d6977ad748b5a35b8740ad5f5f4","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-16 09:22:43.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"2d8c7aceb9f4afe9653824eccad285f52c82c4b6","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-16 09:23:07.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c05002bcb57611c66b96cd8654769ef6d6897e9f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-16 10:04:27.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-7/NFS/ : SUCCESS in 2h 4m 44s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"c25f060b4ea9980438c49a206149e9ff5ab27939","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-16 11:15:10.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-8/NFS/ : SUCCESS in 1h 53m 26s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"d1bfe22aea8e1fbb486aecaa3e77b69077bd80b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-16 11:21:27.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c56909a3e32648b8814cc95bd9a0854e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf84d3efab634dd28fe6fb488e395ffd : SUCCESS in 10m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0983dd30e6c5408c9430296fff173ef3 : FAILURE in 18m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83efb7850ac141f991519548c3ecd960 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41b6ff6be5c844ce85b458723f4ce5ee : FAILURE in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8368ddc0be574eaaab827e8f20d8fd24 : FAILURE in 4m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/38e07d99ccec4a809023cf9f6b02a723 : SUCCESS in 54m 39s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/c77d622876dd40139b83ecd40f62e597 : SUCCESS in 1h 48m 55s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/aeffa172af14456da06b764bdd8f3388 : SUCCESS in 1h 21m 37s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e92165fd76824f2f804fada4023c9763 : SUCCESS in 1h 11m 07s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/aaa88d8936ee457d8c793e7423697a4c : SUCCESS in 1h 30m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/e932ac20c63a47c7bc984092f0fe6739 : FAILURE in 1h 08m 06s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b0a70a84280b4961abefa7e5fcaa8aeb : SUCCESS in 4m 18s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5a4736d0ac5a4cbdb2c23f919b128005 : FAILURE in 3m 17s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4de429096a5e4cd0a3724a454225f847 : SUCCESS in 34m 39s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cd32b29604244a34867f2da3f04c9437 : SUCCESS in 27m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/203b99b909034a0a995cfd19c1f1a64f : SUCCESS in 39m 44s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/514cc962accb4cb897e555105a46f870 : SUCCESS in 1h 21m 51s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/b5fabf1036d84e7783262fa690e2de4e : SUCCESS in 59m 58s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/abc12996e9704012a57aa51ccf173205 : SUCCESS in 1h 00m 37s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/82286ef8fa584113b9a8b9cbb3ce8a93 : SUCCESS in 41m 18s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/9ccddc0633214a74945abca4f900a2fa : SUCCESS in 40m 35s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"258d0bbc58fd5d826830e8c73807166ac9b81999","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-16 12:37:04.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/94/911694/8/check/DellEMC_Unity_Manila_dhss_true/255b365/DellEMC_Unity_Manila_dhss_true/954 : SUCCESS in 3h 15m 26s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/8/check/DellEMC_Unity_Manila_dhss_false/6961215/DellEMC_Unity_Manila_dhss_false/1022 : SUCCESS in 58m 18s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"226b9c5760307786aa93bea3095f83b4e2b3f71a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-19 12:25:37.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"c5a149edbb55c55072ed0b63ac8e18116417dc2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-19 12:47:35.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"ecaf59d4e71d19c6cbcb519495f24ff15692e587","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-19 13:16:54.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/10/check/DellEMC_PowerFlex_Manila_dhss_false/c156fba/DellEMC_PowerFlex_Manila_dhss_false/1013 : SUCCESS in 29m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"6a72282324861c13c0b22c7be7dd4af225a6e943","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-19 13:56:10.000000000","message":"Patch Set 9:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-9/NFS/ : SUCCESS in 1h 30m 13s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":9},{"id":"cc0c34d6421393eb1d63da79199bbb1236295f2c","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-19 14:44:41.000000000","message":"Patch Set 10:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-10/NFS/ : SUCCESS in 1h 48m 43s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":10},{"id":"2a44f7c785dc3c3398628f0eb8ed7bfadadfeee9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-19 15:12:19.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/94/911694/10/check/DellEMC_Unity_Manila_dhss_true/e10a42c/DellEMC_Unity_Manila_dhss_true/958 : SUCCESS in 2h 24m 25s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/10/check/DellEMC_Unity_Manila_dhss_false/4b408e4/DellEMC_Unity_Manila_dhss_false/1026 : SUCCESS in 1h 00m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"94dd982a7550553f7d2c2a891460f50e1143cc81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-19 15:34:05.000000000","message":"Patch Set 10: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ae79e5c82c040179060b60006290dd4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/654151deb5774e4e9ee4ca991fb88b2a : SUCCESS in 9m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e76942c357249fb8331338393894481 : FAILURE in 24m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02ad9454c61a448387a71aa63e7db06a : FAILURE in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be6317c931444615bfa8c75048d8ada1 : FAILURE in 9m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9570eca355d34096ad5c1f3e3cd6261a : FAILURE in 9m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9b9be74678ea4f9a80f0c3b7b39a3153 : SUCCESS in 1h 04m 10s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/050806b3966749cbb09ea487cff3f6cc : FAILURE in 2h 38m 11s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/eba6cc3e751f44b3a8620e558c9a4588 : SUCCESS in 1h 26m 28s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/76784b80bc7e444088731de77e4cc1f5 : SUCCESS in 1h 00m 57s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d03f9e4381c24baa9e96d749e054a6d5 : SUCCESS in 1h 39m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/ddf44d646d354346a325a6d692e1150d : FAILURE in 1h 05m 48s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3dda4e08984a40f4be77e07047275ae1 : SUCCESS in 4m 39s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/bb50b96bf3c84052a85474eb00d2e2aa : FAILURE in 3m 05s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/2a73b374b18245c5af994fcccd057f4a : SUCCESS in 38m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/fe5a629b69c9403f9f5774c3533d3a0d : SUCCESS in 28m 01s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ce30c9677bfe4d41868b94f9b0a6115a : SUCCESS in 55m 33s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/6048b880f22149cfa7619e374d04b4aa : SUCCESS in 1h 23m 46s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/83ccef071f61493290564c611ad7b904 : SUCCESS in 49m 25s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e70b07deb4b4479ab0afe7bd731ee86a : SUCCESS in 49m 12s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ac9ae720c5d34e4da0e253f0682fd58b : SUCCESS in 40m 04s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/d32b3e4130a44599bb6cace10826ea4d : SUCCESS in 46m 11s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0a6cb5aa7aa3c7d72997165c00fb68f529d67d1d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-03-19 16:25:31.000000000","message":"Patch Set 10: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"6926c6cb7e5b5ecac84435f1a569c3db7f10bf7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-20 12:38:14.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"dc1252fb0c90775dd5989e7f15ab92d82fbdeb21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-20 12:41:30.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"9a4fbc4e0ee562526727c0e81b8dbabbc21cc604","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-20 12:42:05.000000000","message":"Patch Set 12:\n\n(4 comments)","accounts_in_message":[],"_revision_number":12},{"id":"fc19ec2d886fe2da578c3aafbe375da65c0e190a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-20 13:12:56.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/12/check/DellEMC_PowerFlex_Manila_dhss_false/e7c9871/DellEMC_PowerFlex_Manila_dhss_false/1019 : SUCCESS in 31m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":12},{"id":"c93c24a314b428c36c4e0a55c46c0528a588fcd5","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-20 14:13:07.000000000","message":"Patch Set 11:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-11/NFS/ : SUCCESS in 1h 34m 29s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":11},{"id":"be4c5cc5cb308e47ad790f5a7f3bfdfa59657a6f","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-20 14:41:34.000000000","message":"Patch Set 12:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-12/NFS/ : SUCCESS in 1h 32m 57s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":12},{"id":"caa2498f8a7af721bdd25cfb64fa046bc70e6b28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-20 14:43:12.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e428be4ce5c44c1f9e99d187e9656c11\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3df781936eb4331ad020517e4b4d789 : SUCCESS in 8m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/126dd0055f0d4a7f98738be192f6a7d6 : FAILURE in 21m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0aa5ac6fd73145e99c133796f7eaa89c : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/782e815b025f4772bfd92a300ed193c6 : FAILURE in 9m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0fbe85d50dfd47229518c221986132ef : FAILURE in 7m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4c80ca0b31784a089a481eec6016e6fb : SUCCESS in 51m 10s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/f4e45e169f8c4247a4658c69bdebc72f : SUCCESS in 1h 59m 49s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/489845da96a549eaa0392b1e62a1418a : SUCCESS in 1h 19m 47s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/f05b37703ffa48cb9d55be3f7dd3b00e : SUCCESS in 1h 13m 23s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b1c48b735c154ac2ac0f56b05181a25e : SUCCESS in 1h 07m 46s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/7929e07d1f454def8cfa040924ca0bca : SUCCESS in 1h 07m 38s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/e01d6dfefe6e49c8a7bfd523fe4179d6 : SUCCESS in 4m 24s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/0cb3f21667fb426e8102a6cb59fef764 : FAILURE in 3m 10s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8a060aa09ebe449d94120ff78063ad96 : SUCCESS in 39m 45s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/7ef382bd975e4dc68e0175c22642b9d2 : SUCCESS in 42m 41s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6588e674ec9742449a774ccae3fd8f38 : FAILURE in 52m 41s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/5415e5ec3dab4b859d49da07dc5d46b7 : SUCCESS in 1h 19m 37s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/ca38a90efcd0460eb126bc36b046962b : SUCCESS in 1h 07m 20s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5fa66c6f4c8a43ce8d3c53cc76282ff4 : SUCCESS in 1h 06m 45s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/819114d9654548e48b2c3c00297ff391 : SUCCESS in 27m 24s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/780b6063679d4e59bbc0a5821dd9b89b : SUCCESS in 32m 32s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"07f704fef82cf271a50b9e68d12de5d655734fed","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-20 15:59:30.000000000","message":"Patch Set 12:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/94/911694/12/check/DellEMC_Unity_Manila_dhss_true/fcb9c7c/DellEMC_Unity_Manila_dhss_true/964 : SUCCESS in 3h 17m 43s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/12/check/DellEMC_Unity_Manila_dhss_false/bedb373/DellEMC_Unity_Manila_dhss_false/1032 : SUCCESS in 1h 09m 49s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":12},{"id":"a671be1688e39b2add0624e08dbc784dc092d297","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-03-20 21:39:21.000000000","message":"Patch Set 12: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"fb62693de646089f40219cc550e23f44a5069092","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-22 14:15:25.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"2ebf6905a4f415d1bd93b372d35e7e11cb3ee1fb","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-22 14:16:37.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"20e21b35d35128db30fbe0e47cb8600c234c727c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 16:15:00.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f66480c2ba84c268e93959f9d716985\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d0630c6df4c4270a7ffb03526804f69 : SUCCESS in 12m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed9eb6bff220479a96884d1121c0a307 : FAILURE in 25m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bbf6dedded14200b1ec4839eb271c54 : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3d3096af39347aebde6ada327888a5d : FAILURE in 12m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4c7afea42a4746818274443068ffc2e4 : FAILURE in 9m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6a796c91f6954227aa7f42998c6fdf01 : FAILURE in 12m 39s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/82cb9d77f42e49f2a8cb24c293e397f7 : SUCCESS in 1h 12m 19s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a1c99f60c25647baa9c52ca8e3703cc2 : SUCCESS in 1h 53m 06s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0085d266e8e6439fb9a99778661e6f3f : SUCCESS in 1h 24m 13s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/81ba73ac7fe74bb9a6f1cd45d83f950f : FAILURE in 1h 35m 09s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/e3304d9a8d3b4c769b79f4cf25022ac4 : FAILURE in 1h 38m 25s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1f5aeea030134abbb4e2967d34bd666b : SUCCESS in 1h 15m 23s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/9276762cc698497785914afe0b1f32ac : SUCCESS in 5m 18s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/155c207cf1854cdc830bc42a4ce1c432 : FAILURE in 3m 20s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b87844c0c0a84420988cac3ac319429f : SUCCESS in 34m 18s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/5044b5a0d5594994ad9193bbcd26f406 : SUCCESS in 41m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/2e06b425696a4b33b1027a7e9e36c14e : FAILURE in 57m 20s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/51ca5e4962214d97b7930d304116be15 : SUCCESS in 1h 22m 24s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/d48de4896c424a5b8873f0697fd9e621 : FAILURE in 25m 38s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8ca60b7723df47328d3b205788112b55 : SUCCESS in 57m 53s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c522a701fcc94e01ac410197c14970b1 : SUCCESS in 35m 06s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/8750d5f2a3a04983bd5a723e58e44858 : SUCCESS in 37m 57s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"56b4ad413ed7004ec6d6738635ab389625cf2a98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-22 16:40:50.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"349b16631db228867ab7657a3e2d593150cf85bb","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-22 16:50:14.000000000","message":"Patch Set 13:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-13/NFS/ : SUCCESS in 1h 15m 25s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":13},{"id":"0ff3337941e51cbe6c36825fe78d6a7b91e7570e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 19:57:13.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35a7c8bd29eb4604b21f9c923a4535cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1806280d7233405f89898ebd5c9618b6 : SUCCESS in 8m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/469db7208bd34962b092901c857ca70c : SUCCESS in 26m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6948eca4ab494a73940bb2bd8ed344df : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f0342dc9cf84ce98933211abe7cc64e : SUCCESS in 12m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67371d3102b84bf2af5b9dcf8226c4f0 : SUCCESS in 9m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8240ee4581fc447d81719a5bf3e64efc : SUCCESS in 9m 26s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7c791964f86a4a5eba187580aea3e609 : SUCCESS in 1h 26m 16s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2ba2a58ea97b45488a44b6ff551a3bf5 : TIMED_OUT in 3h 02m 36s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fc8399b8c32b433294fb324f58d3f2ee : FAILURE in 2h 50m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a16e90e528f644b3802c555643a9d077 : FAILURE in 29m 08s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a7a4132d8fa44c0586b23afb188e86b5 : FAILURE in 17m 31s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9c970f8de51e4c068c433765702d9dae : FAILURE in 1h 14m 07s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/10e6d8971a3040ed9458da055c0fc476 : SUCCESS in 5m 26s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3d81c29a3f21418ebf0a55df9aaf0e5e : FAILURE in 4m 53s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b3806d7699fa4e4e9e538107bf09d759 : SUCCESS in 54m 46s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/679051db57ab4e26a71e8d39297b0396 : SUCCESS in 39m 20s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d8ec617b6a804f2892541cf312a26446 : SUCCESS in 1h 01m 20s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/373d5b86078d4d479083aab599db0129 : SUCCESS in 1h 04m 53s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/962c555a95e34bb18dcb32362f8249bd : SUCCESS in 1h 07m 46s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/169bf513b6e342179ec418f1564cd8c2 : SUCCESS in 52m 43s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/f239d58f5bdf4358bd0b5b7931b075bb : SUCCESS in 32m 32s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/35533b644fdc4fa39cedb2737d110044 : SUCCESS in 39m 21s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"5106ce1b93732dca574066e6eac1c232c6919fbc","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-22 20:33:35.000000000","message":"Patch Set 14:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/94/911694/14/upstream-check/manila-tempest-plugin-ontap-dhss/dd98aa7/ : RETRY_LIMIT in 3m 48s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/94/911694/14/upstream-check/manila-tempest-plugin-ontap-no-dhss/71d4958/ : RETRY_LIMIT in 3m 36s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":14},{"id":"07a3f750e7db5f856029662d3025689834e3a4cf","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-22 20:34:01.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/14/check/DellEMC_PowerStore_Manila_dhss_false/4474b8a/DellEMC_PowerStore_Manila_dhss_false/134 : SUCCESS in 36m 32s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":14},{"id":"341c43d7d7ebff1ccf9c4a1bee4b296a58f5ed52","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-22 20:56:23.000000000","message":"Patch Set 14:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-14/NFS/ : SUCCESS in 1h 21m 32s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":14},{"id":"b9e08b2e2310edece03f3ab20c6356842e54352f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-23 17:09:57.000000000","message":"Patch Set 14:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/14/check/DellEMC_PowerFlex_Manila_dhss_false/e457c21/DellEMC_PowerFlex_Manila_dhss_false/1065 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":14},{"id":"a063f7c0f100ecff51cfb3d4173c7be5e3f0b0a2","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-23 19:59:00.000000000","message":"Patch Set 14:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/94/911694/14/check/DellEMC_Unity_Manila_dhss_true/82dd3b4/DellEMC_Unity_Manila_dhss_true/988 : SUCCESS in 3h 01m 56s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/14/check/DellEMC_Unity_Manila_dhss_false/806fe36/DellEMC_Unity_Manila_dhss_false/1067 : SUCCESS in 59m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":14},{"id":"c7347875dcb6adff67127aa760404fcccc153cfc","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-03-25 22:52:41.000000000","message":"Patch Set 14: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"f002a9b6ce5947e9728889707c4a34ae9f652ef9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-27 06:18:35.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"122a89bf8dcd31d916e9c188558b0be1bb764f68","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-27 06:24:24.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"ef920711380acd289b5577eee1d68934f29ac98d","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-27 06:28:15.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"c724d940921207fc69940ec576d294b16d587407","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-27 07:51:08.000000000","message":"Patch Set 15:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-15/NFS/ : SUCCESS in 1h 32m 11s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":15},{"id":"b8feb6ca79ce7f4a0ce40a3a1cc0a9e0959e540f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 08:10:12.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b408c2b582e94975a2b7d5da374c1b34\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e38617f889944a687711ecf92fb396a : SUCCESS in 8m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10c70f8b29d64799a2cf916afb2bff1f : FAILURE in 16m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/449d76d0f5c74486b91bffe8515e1e1b : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d6d3ba07a4646109f77d897eed1b8f5 : FAILURE in 9m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/20c53460d3614259af2c895f6d546dda : FAILURE in 7m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/71443bcebc51471794370906c519e508 : FAILURE in 10m 08s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/076c5c52257e40118679b7d4b07fdbcb : SUCCESS in 50m 07s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/ff8384eb5e9c467092fee5b0a47a58c2 : SUCCESS in 1h 49m 08s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/17f913c5f7ec49aabe9fc794fe7d9328 : SUCCESS in 1h 24m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/0f8ea043e5034f9e8968072643715d86 : SUCCESS in 52m 08s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/5f34efde95f84ecab53ad4b797586dcb : SUCCESS in 1h 29m 24s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/843a0526f93640fabca91af172b71e2e : SUCCESS in 1h 03m 18s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/50f71a3dc7ce470e91bb62fc47b7318b : SUCCESS in 4m 15s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dcd4287f18d44ab098d7333941441fa9 : FAILURE in 3m 56s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/750149bfcbbb421ea2b3c29b8ccefd2c : SUCCESS in 39m 57s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/adb5480e51dc4d9b903d3292f3366afe : SUCCESS in 41m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6908b1502d7b412e8184488c005996ac : SUCCESS in 53m 58s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/db1b273b020c49dd978f3f197dfe6143 : FAILURE in 22m 21s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/1a426f6d5872440e8ad6d9906097221a : SUCCESS in 56m 51s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cb8ccedc87b84750a406a2c29f22a970 : SUCCESS in 1h 05m 06s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/dba1d49555d845eea88d646b4e19166a : SUCCESS in 37m 39s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7fce53be934b4c918385464c56e62fea : SUCCESS in 47m 10s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"7c491779685bb81cb70f7e84a2f1ed12e3560f7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-03-27 10:07:23.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"a78356ba60532751eceb7cfaef2ed11a48377ab7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-27 10:39:41.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/16/check/DellEMC_PowerFlex_Manila_dhss_false/3a41315/DellEMC_PowerFlex_Manila_dhss_false/722 : SUCCESS in 32m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":16},{"id":"89ef14db853cbcccad81eb335df6508ab4d516d7","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-03-27 11:32:58.000000000","message":"Patch Set 16:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-16/NFS/ : SUCCESS in 1h 25m 14s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":16},{"id":"231e4562ce90c7094941d48577466f3600b852a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-27 12:04:43.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dd387e9751d470c877c6193c5df2d02\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c2398344b07408897eec9c7b4da68cc : SUCCESS in 9m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acd54b55ef5a4e9caef283a10c7cce86 : SUCCESS in 24m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc7474320c1a4cde8ebe627a77bebddd : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4962baeb6f56466fb3d192272e538d09 : SUCCESS in 9m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8401033b313f43e59e423e2f04e69077 : SUCCESS in 8m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/892bd8f145fc441abb87ede4df0d9990 : SUCCESS in 11m 01s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/bc04e4db213d472696e778cbe7a870dd : SUCCESS in 1h 08m 27s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/64f72f04089f44f998c659bff50cb316 : SUCCESS in 1h 55m 17s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/48615f4011364edb9e00635c450f3cf3 : SUCCESS in 1h 19m 40s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/3588488b51554358bd755f1e318a5c13 : FAILURE in 1h 01m 14s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b34393f7e21741df94b3ab9ffcf6b898 : SUCCESS in 1h 25m 36s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b7782730c6d74016a4dfe2d2b8afcfbe : SUCCESS in 1h 12m 00s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5f06a7418fc94cc796bddfe591efa582 : SUCCESS in 5m 16s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/beb38e3c11ec467191a10d0ddee1d168 : FAILURE in 3m 06s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/e85feb4a4a3246ee9f0eba22c6c9bbb6 : SUCCESS in 34m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8ce7bde0fa6c4546a89bce4ab71fbecf : SUCCESS in 47m 38s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/4a66b4dace8f412b85e4505d641304b6 : SUCCESS in 46m 48s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/28eff64ca4bd402580405c86fe53d952 : FAILURE in 21m 09s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/e81a42ce180248c5870155ca19098d1b : SUCCESS in 54m 13s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/00958823667443cfa5fca4faa282627e : SUCCESS in 58m 34s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e36a3217fe514d3abfb2458d9776cad2 : SUCCESS in 42m 17s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e8cf07ac5ec449d2a5fdb2e1c809f2c0 : SUCCESS in 32m 20s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"5c3b8dc518f7c49369997bdf0c6141d88cb05dbb","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-03-27 12:15:06.000000000","message":"Patch Set 16:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/94/911694/16/upstream-check/manila-tempest-plugin-ontap-dhss/a49d099/ : RETRY_LIMIT in 3m 46s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/94/911694/16/upstream-check/manila-tempest-plugin-ontap-no-dhss/323eb1f/ : RETRY_LIMIT in 3m 48s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":16},{"id":"df7c2248b3d6ba1da694d82d19703cf735032482","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-03-27 12:35:39.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/16/check/DellEMC_PowerFlex_Manila_dhss_false/4b1629b/DellEMC_PowerFlex_Manila_dhss_false/724 : SUCCESS in 30m 40s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":16},{"id":"0df50593bb98ae565555b602b82792aaae1c3ba2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-03-27 12:41:11.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/16/check/DellEMC_PowerStore_Manila_dhss_false/dc14354/DellEMC_PowerStore_Manila_dhss_false/149 : SUCCESS in 36m 12s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":16},{"id":"6eaa167cf3e324068016fc23d7b9ef57f9b957d5","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-03-27 19:46:15.000000000","message":"Patch Set 16:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/94/911694/16/check/DellEMC_Unity_Manila_dhss_true/68ad8a8/DellEMC_Unity_Manila_dhss_true/1009 : SUCCESS in 3h 28m 45s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/16/check/DellEMC_Unity_Manila_dhss_false/5b6133f/DellEMC_Unity_Manila_dhss_false/1095 : SUCCESS in 1h 10m 21s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":16},{"id":"e0856163b69a0e2f85838008c7c01b6ece93bf14","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-04-02 23:19:21.000000000","message":"Patch Set 16: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"3d4ad95c0c90201aa763fdfb56960056028ff95a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-04-04 11:02:53.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"317ba07781a01711c4bf1f5780dd5c96c8751f95","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-04-04 11:04:53.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"6e53ce2b1fb8720d1dbbd1c05258673b551c82ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-04-04 11:24:45.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"96416e8d6841dad368d74787d4045729bf44c85a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-04 11:56:23.000000000","message":"Patch Set 18:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/18/check/DellEMC_PowerFlex_Manila_dhss_false/c9139bf/DellEMC_PowerFlex_Manila_dhss_false/797 : SUCCESS in 31m 21s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":18},{"id":"8a32f9557a62390563f930aeb2feb2826ee88a7a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-04 12:38:02.000000000","message":"Patch Set 17:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-17/NFS/ : SUCCESS in 1h 17m 28s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":17},{"id":"1eb9f187b7d33af880f8bd313de2d12ecba78d78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 13:32:08.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44c5ad3b8b594624b9e94abba2633664\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d326439575c840c7a4b80d048a4e1161 : SUCCESS in 9m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc2df6d5f11d4526bd03fe37398d9acf : SUCCESS in 28m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d457bf272ef74f9fbd8676c662955b93 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/205a79122252427686c6b737084fe66c : SUCCESS in 14m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e88c4e7708af4fd2a2f1c2fd499bcfd4 : SUCCESS in 8m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1008a055f07c4110932d842372cb9277 : SUCCESS in 13m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2335e85c49274db1b567e86e9cd99439 : SUCCESS in 6m 59s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/81fb853fbcbd4ef98faa06c97cf8ac55 : SUCCESS in 1h 13m 39s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d2c32cd359ae44eda80842bca28f2ebd : SUCCESS in 1h 56m 15s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fd927dcaf59c49cf9f7983989ff64da3 : SUCCESS in 1h 04m 21s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/17f23b690e934019a24b0162caeab2a0 : SUCCESS in 1h 13m 24s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/de1ac04517094a4a85e76d0fd6220e2e : SUCCESS in 1h 28m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/50d1d6938d094d3997ad2d0bded21266 : SUCCESS in 43m 18s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/4f1f87f957f4407686b0a506047b210c : SUCCESS in 4m 10s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9ab861f37fed420fa86a1443813eb7b8 : FAILURE in 2m 48s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/032b171455594301a1716d3e12b0f9e7 : SUCCESS in 36m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/959f536781164ccebe6de83ee4100a63 : SUCCESS in 30m 21s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9aca14fe860c4c9b88f0fb569660b59b : SUCCESS in 39m 04s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/733142d796f54576ad4a8e5f8919f757 : SUCCESS in 1h 00m 23s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/2bff09af32bb4e22bd5045dbfc9cb93e : SUCCESS in 51m 58s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f7b2eb32a0fe4b6dbdc8e8ebab130409 : SUCCESS in 49m 38s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/19b1b80a02664fd29fa249aa1ebf4797 : SUCCESS in 39m 39s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/98bd29add2b64b2d88a7f10bd2901059 : SUCCESS in 41m 01s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"27342f868dc67ce65524766f34f49419a53373f5","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-04 13:37:28.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/94/911694/18/upstream-check/manila-tempest-plugin-ontap-dhss/3f11931/ : RETRY_LIMIT in 3m 47s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/94/911694/18/upstream-check/manila-tempest-plugin-ontap-no-dhss/1e19f91/ : RETRY_LIMIT in 3m 56s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":18},{"id":"693bc1316c83df7d31f539e4c3e338616b7ccb5e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-04-04 14:06:27.000000000","message":"Patch Set 18:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/18/check/DellEMC_PowerStore_Manila_dhss_false/f7e3d5e/DellEMC_PowerStore_Manila_dhss_false/176 : SUCCESS in 34m 04s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":18},{"id":"9551e48da42097ac479fe146696e30ffa6201a68","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-04 14:15:57.000000000","message":"Patch Set 18:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-18/NFS/ : SUCCESS in 1h 55m 23s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":18},{"id":"920281cab963fdadd78a4e3100ac435f5b8e010b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-04 14:38:58.000000000","message":"Patch Set 18:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/18/check/DellEMC_PowerFlex_Manila_dhss_false/c6f446c/DellEMC_PowerFlex_Manila_dhss_false/801 : SUCCESS in 31m 05s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":18},{"id":"bc38a6f2d55c8072c4cc845fa600cee1f67d79c4","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-04-04 20:38:51.000000000","message":"Patch Set 18: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"18894b3f9b08a4ea16c13f9d58166c82cebcea0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-04-04 21:58:59.000000000","message":"Uploaded patch set 19: New patch set was added with same tree, parent tree, and commit message as Patch Set 18.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"29821bbb46f12b222965ff714e824c90b23e9ead","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-04-04 22:02:07.000000000","message":"Patch Set 18:\n\n(3 comments)","accounts_in_message":[],"_revision_number":18},{"id":"e23a4eb2876bdc64add93ac802626a5a00d895be","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-04-04 22:28:10.000000000","message":"Patch Set 19: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"938e5b12a2da9e875101cf2e550a092b0fb2a3e4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-04 22:40:57.000000000","message":"Patch Set 19:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/19/check/DellEMC_PowerFlex_Manila_dhss_false/92f4f07/DellEMC_PowerFlex_Manila_dhss_false/809 : SUCCESS in 33m 23s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":19},{"id":"a34ec5d15e94fc825852c30ebaf1a30250c81e5a","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-04 23:22:25.000000000","message":"Patch Set 19:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-19/NFS/ : SUCCESS in 1h 23m 4s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":19},{"id":"a38db9ad897c29b2f8371b3de4ce0dbed11f8894","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 00:46:59.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/755b6762d48b49bb9bbcae1125533a2c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc52aba31ec149ad882e8718abae941b : SUCCESS in 9m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee9007ae1f9848ff96ab50841d147ca2 : SUCCESS in 23m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e01f2f90874942b1ba9579fadf95768c : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3254ddec008148569023400baf06ffc9 : SUCCESS in 9m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d6811a6075a64db584656bc632492c51 : SUCCESS in 8m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f43f025f5ef140c3866e4d009c4d7f47 : SUCCESS in 11m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41ae970c5fcb4cb2b28ed30eca73ee8e : SUCCESS in 7m 48s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/9ce19276b31a48b49d2f067d80065ec3 : SUCCESS in 1h 10m 56s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/911c4f9c15614ea8a52661aed0f2859e : SUCCESS in 2h 25m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/afec3cfdf54e4b188198df3a345e9386 : SUCCESS in 1h 19m 06s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6d976b523118446ab532446e22492c4d : SUCCESS in 1h 12m 46s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/dde86e23a9284eabba3e8d2398df13e4 : SUCCESS in 1h 36m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/a1c4ee3eada441169345a58f4d89c952 : FAILURE in 1h 04m 33s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/0059133bfa974af2ae6944536a4ab203 : SUCCESS in 4m 34s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/3b954d88ced84dd68ae7a45fcc2f8960 : FAILURE in 3m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1c0f56ad48d04903b1a3b586f7edbddd : SUCCESS in 39m 02s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/51058de35a2544d8bcb575841207e6ec : SUCCESS in 43m 08s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/605058d239a742eabfe9625c986a89d8 : FAILURE in 53m 19s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/113dec13357041f5bc902e7ff4653b44 : SUCCESS in 1h 05m 52s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/0378eca0622e4808948bb83a7d17912f : FAILURE in 29m 51s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/90d374ed1abb4ea8bd363b623c70ded1 : SUCCESS in 1h 00m 01s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/237643a297a14fb791d25e391cae6606 : SUCCESS in 42m 38s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/3f9c55d6d76a4ccb9a02748bf132fbeb : SUCCESS in 41m 06s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"5a43353acbe1b1b5fa18b8fd07d8d0b965021c6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-04-05 05:14:03.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"e8b7d610090cddddc8c560cc44c00f1c2a70cd4a","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-04-05 05:37:49.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"a2bd9998476965f648355550d9a845027a39c908","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-05 05:45:58.000000000","message":"Patch Set 20:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/20/check/DellEMC_PowerFlex_Manila_dhss_false/0cbe866/DellEMC_PowerFlex_Manila_dhss_false/810 : SUCCESS in 31m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":20},{"id":"6c467bb14075f1123feb0d437f26c4e55e81eeab","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-05 06:30:59.000000000","message":"Patch Set 20:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-20/NFS/ : SUCCESS in 1h 16m 37s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":20},{"id":"890066114e1eac54f1478a44923f83e24ca70320","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 06:49:40.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35170ba2464548a48c1a553a4d60ee45\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57af9affa54a4673991948358620a38e : SUCCESS in 8m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9453da431a474bf5933115b9c7120cb0 : SUCCESS in 24m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7c6623a82ec42089e786373547870aa : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/180980ab6a0a4f61a7b11740e15fb071 : SUCCESS in 9m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47f03552409e4213b1eccf728d4b111d : SUCCESS in 6m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52dd74ba30e84793a18dd682c6e70e60 : SUCCESS in 10m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b66dd561e4f94f9c843bb1dace549310 : SUCCESS in 8m 32s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/7207c970ca2644dc850c2517b55f2c31 : SUCCESS in 1h 11m 58s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/faf59ed03fb742258e1aefe3c301fc81 : SUCCESS in 1h 26m 43s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/eb3a4c8eca6f4dc4ae3051a95b83da34 : SUCCESS in 52m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/a52ebc6aa4fa4a1e9f126fe284113cf9 : SUCCESS in 1h 00m 04s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/23990b95b5304edfbaa48ef48073fbb2 : SUCCESS in 1h 04m 25s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6a41f84cd75d432aaf954d4633b9aa3c : SUCCESS in 1h 16m 05s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/230fabbea039404fb169a1accc483515 : SUCCESS in 4m 13s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/713a250abfa44441b309a48a0e340cf5 : FAILURE in 3m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/da683c842aa04e9c9347b157315e62d5 : SUCCESS in 28m 20s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/352b3a0424b54502b36a46d3ab5d29e9 : SUCCESS in 43m 19s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/faf85da34c0f46d491c218a329831e0e : SUCCESS in 49m 45s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/b66b9aa8c1ee41558e9f59a8a6075130 : SUCCESS in 1h 14m 39s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/fa8a7ed70b7e4e88a3f3559663baa8be : SUCCESS in 1h 03m 01s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7bd8b73e6f8542fca4c34a971349f2a8 : SUCCESS in 52m 33s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d62e893dd1c34ce7b87de9e226691b63 : SUCCESS in 23m 34s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1f8af65009ba49e4b70cbe1816347404 : SUCCESS in 26m 26s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"e588dcbe3ecfe9161d90ee666402bc78e4e7ce94","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-05 07:22:22.000000000","message":"Patch Set 20:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/20/check/DellEMC_PowerFlex_Manila_dhss_false/c6a679d/DellEMC_PowerFlex_Manila_dhss_false/811 : SUCCESS in 32m 27s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":20},{"id":"e96f29d40eca8fac5eedc12fe1655b2bd622bb93","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-04-05 07:22:53.000000000","message":"Patch Set 20:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/20/check/DellEMC_PowerStore_Manila_dhss_false/3e2017e/DellEMC_PowerStore_Manila_dhss_false/181 : SUCCESS in 32m 58s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":20},{"id":"d3ad5f94de4de32ddc0cad8315f330c92f988f17","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-05 07:25:57.000000000","message":"Patch Set 20:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/94/911694/20/upstream-check/manila-tempest-plugin-ontap-dhss/3cc73f1/ : RETRY_LIMIT in 3m 18s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/94/911694/20/upstream-check/manila-tempest-plugin-ontap-no-dhss/ef68bde/ : RETRY_LIMIT in 3m 39s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":20},{"id":"85c6f58bc71e3cb43913ead18e967d7df52df09b","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-04-05 14:45:16.000000000","message":"Patch Set 20: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"cf0b3c79f307ba0c848a1e1fc0c5ab7ec75c37d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-04-05 14:52:00.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"e8ebcbddbf1f65f055ef5dc10ff69a262c054a6b","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-04-05 14:54:39.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"f57d0aef4849947145a6cd4839f6a9e7b23b6800","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-04-05 15:08:08.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"8760a2b42e2133437b73814d0fb79a359b024ef3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-04-05 15:09:09.000000000","message":"Uploaded patch set 23: New patch set was added with same tree, parent tree, and commit message as Patch Set 22.","accounts_in_message":[],"_revision_number":23},{"id":"33b3029731729bd620056cfda8c4430920bc7639","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-04-05 15:10:17.000000000","message":"Patch Set 21:\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"4b607ce66168d731de0960cbf50fd6e4c0576feb","author":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"date":"2024-04-05 15:11:56.000000000","message":"Patch Set 23:\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"286ebf64f97431328fa729277f427f8f6881a780","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-04-05 15:12:08.000000000","message":"Topic share-command removed","accounts_in_message":[],"_revision_number":23},{"id":"1cf76a7f712bbdc8e0b6f13501615c5c2e532ada","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-04-05 15:12:15.000000000","message":"Topic set to bug/1867030","accounts_in_message":[],"_revision_number":23},{"id":"40482aacae44c2dd65361449ae05e90f2e8dd6a6","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-04-05 15:12:21.000000000","message":"Patch Set 23: Code-Review+2","accounts_in_message":[],"_revision_number":23},{"id":"8c0b7a544cd4985908e3d0ed5104c47803b1095f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-04-05 15:12:35.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"c5e46bc09958051bb011b89d4c10dbcae2d95c1b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-05 16:29:15.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/23/check/DellEMC_PowerFlex_Manila_dhss_false/199f661/DellEMC_PowerFlex_Manila_dhss_false/820 : SUCCESS in 32m 49s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":23},{"id":"9b3e8367aae408e24a905c4af9073aed4ecc804f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-05 18:39:38.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eb8dbb7317c42c89608c6e0c9837b8f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fe7e1c3f5a4469c8ca7992cd49a4a41 : SUCCESS in 11m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97a43981b2144256819af30e82ce3fd8 : SUCCESS in 26m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a980cdbbb3a4c92b4331519a7f9bd4e : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9348bce522e4fa7a2a8812b3c9e91ba : SUCCESS in 11m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9b3da339b682477cbff8263044ca2adf : SUCCESS in 9m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8a37b475bc74b5bbc76db0639b067ac : SUCCESS in 12m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdf6868df4374fd5b43b8c8aeaf822df : SUCCESS in 9m 52s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/5523b77215494c809c9556c8aaeab6d5 : SUCCESS in 1h 29m 32s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5605ce28de79401d8da27f1b2e3eb8ac : SUCCESS in 1h 40m 32s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/361a1153489844cba108904d8d0b7bad : SUCCESS in 1h 16m 51s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/e8abb5213b5144ae82e1fef9216a5705 : SUCCESS in 1h 08m 51s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6fd6aaa827d74e9c90ec75f4cd4c3896 : SUCCESS in 1h 35m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/da1b5e9aa5924ce0a45c673f66bcc71b : FAILURE in 1h 13m 30s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ad157d22a2da4264920de30f9143d363 : SUCCESS in 6m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8fb3fbff96254c88bce431925eaa46d9 : FAILURE in 3m 20s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/049cf8c9874041d6bb2551d77fdcb198 : SUCCESS in 39m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/81cce4fba7794f959388e358e1eae8ad : SUCCESS in 37m 40s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/3a291789c4424aad80f133940dc49e0f : SUCCESS in 48m 16s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/e37682ab49454e30b2ea0e4e3e77c9fe : SUCCESS in 54m 55s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/bdd2efc09d1945e88d73b5955da9d21d : SUCCESS in 1h 02m 52s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/600d49aea02040908b6aead0a86bd461 : SUCCESS in 55m 18s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b16aca4af9744cedb5883ca75741b786 : SUCCESS in 40m 58s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/ec2cd30a185a4752818f283ed155fec0 : SUCCESS in 45m 49s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"568c975886daf42bc8e7a8291e9f71f2562d7c65","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-05 18:49:42.000000000","message":"Patch Set 23:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/94/911694/23/upstream-check/manila-tempest-plugin-ontap-dhss/be1c963/ : RETRY_LIMIT in 3m 33s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/94/911694/23/upstream-check/manila-tempest-plugin-ontap-no-dhss/a981705/ : RETRY_LIMIT in 3m 34s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":23},{"id":"23e2f279206170a4a6cf98f5f223dde5f913b019","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-05 19:12:36.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/23/check/DellEMC_PowerFlex_Manila_dhss_false/93a6208/DellEMC_PowerFlex_Manila_dhss_false/821 : SUCCESS in 32m 42s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":23},{"id":"1dc0f0016bffc457d6ac59a58f93f2a4e8b95dce","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-04-05 19:17:56.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/23/check/DellEMC_PowerStore_Manila_dhss_false/ef757ba/DellEMC_PowerStore_Manila_dhss_false/185 : SUCCESS in 38m 04s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":23},{"id":"4b7f2defae82b10b84792239c60f72f993042860","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-05 19:53:56.000000000","message":"Patch Set 21:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-21/NFS/ : SUCCESS in 1h 44m 5s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":21},{"id":"b830ef00845768650d693d8a70a3f10fd919862e","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-05 19:58:06.000000000","message":"Patch Set 22:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-22/NFS/ : SUCCESS in 1h 18m 15s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":22},{"id":"d09a6a39c25d1a1a9d6f0101af4e73ad8ac21b15","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-05 20:56:40.000000000","message":"Patch Set 23:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-94-911694-23/NFS/ : SUCCESS in 1h 46m 47s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":23},{"id":"ba833faca80d8e325f613e620955760009923cd1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-07 21:02:07.000000000","message":"Patch Set 23:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/94/911694/23/check/DellEMC_Unity_Manila_dhss_true/c6ada0c/DellEMC_Unity_Manila_dhss_true/1056 : SUCCESS in 3h 41m 16s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/94/911694/23/check/DellEMC_Unity_Manila_dhss_false/1cb4fbd/DellEMC_Unity_Manila_dhss_false/1145 : SUCCESS in 1h 18m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":23},{"id":"446baa7cbc0f2e97335189d91cc63568a3dd5d4b","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-04-09 16:18:43.000000000","message":"Patch Set 23: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"f91ade487040e675e6e1183d3cf7a916b99f9f7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 16:19:13.000000000","message":"Patch Set 23: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":23},{"id":"dedea3a5c780004f667b1be8b9eae9255a4777a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 17:09:35.000000000","message":"Patch Set 23: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff7caed21f50463e8357fdbf0c5d52c7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b4412bc269944fe9bd85b27238ba075 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/305f104df08b4c558ee6efe0ac8a8dae : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6823e7eb5f0e4cb5bbfe8e3574109384 : SUCCESS in 9m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/218b5377cccc4dd4bb66fe384df3f843 : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0bf2cac15ee4430b2687fcf90eb8b6e : SUCCESS in 5m 25s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6ba740825370425db35e7ec20ad5fca4 : SUCCESS in 49m 46s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/d1831d4bf2be43e7a0f6fb72be0fd299 : SUCCESS in 27m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/536621b889934aed8705ed0ba533e7d2 : SUCCESS in 29m 13s","accounts_in_message":[],"_revision_number":23},{"id":"e26b9b905a5b7a8f11edbad8664a018afced3eae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 17:09:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":23},{"id":"6066e9cb42a79da6b6922335b5ea4249c6f28fb7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 17:10:46.000000000","message":"Patch Set 23:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11a68dca98b44bbe8d32b0caf04e4409\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9ec6cafd9a14f6f8c39b1b41a4914e6 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d358d724b62a4f0383b23c923d4215fa : SUCCESS in 51s","accounts_in_message":[],"_revision_number":23}],"current_revision_number":23,"current_revision":"23de7b4b0cfb7c3ac4663f395298d19fa7a8776f","revisions":{"a7d7894d7e0e94f538b0acc389e83431d607792d":{"kind":"REWORK","_number":1,"created":"2024-03-07 11:17:43.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/1"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:17:39.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a7d7894d7e0e94f538b0acc389e83431d607792d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a7d7894d7e0e94f538b0acc389e83431d607792d"}]},"branch":"refs/heads/master"},"101c0b99a762eccbd8774dd3ccbbe0c8d85276bb":{"kind":"REWORK","_number":2,"created":"2024-03-07 12:25:58.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/2"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 12:25:42.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/101c0b99a762eccbd8774dd3ccbbe0c8d85276bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/101c0b99a762eccbd8774dd3ccbbe0c8d85276bb"}]},"branch":"refs/heads/master"},"e7e4583b9b99a73b6a7d3b9766395b576c98381d":{"kind":"REWORK","_number":3,"created":"2024-03-07 12:29:56.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/3"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 12:29:41.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e7e4583b9b99a73b6a7d3b9766395b576c98381d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e7e4583b9b99a73b6a7d3b9766395b576c98381d"}]},"branch":"refs/heads/master"},"e44a0208555fb81e7cff46b5bb526fd8864c27c1":{"kind":"REWORK","_number":4,"created":"2024-03-08 13:37:01.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/4"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-08 13:36:45.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e44a0208555fb81e7cff46b5bb526fd8864c27c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e44a0208555fb81e7cff46b5bb526fd8864c27c1"}]},"branch":"refs/heads/master"},"8abad458a7e53caafe5050f5eb023b171f972d77":{"kind":"REWORK","_number":5,"created":"2024-03-13 11:47:24.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/5"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-13 11:46:55.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8abad458a7e53caafe5050f5eb023b171f972d77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/8abad458a7e53caafe5050f5eb023b171f972d77"}]},"branch":"refs/heads/master"},"48e2f32a46bd770f87717b103857fc98d3e496e2":{"kind":"REWORK","_number":6,"created":"2024-03-13 11:56:54.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/6"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-13 11:56:39.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/48e2f32a46bd770f87717b103857fc98d3e496e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/48e2f32a46bd770f87717b103857fc98d3e496e2"}]},"branch":"refs/heads/master"},"3585f12c8139d0a3e79eb5240588ca8caeff87e3":{"kind":"REWORK","_number":7,"created":"2024-03-16 07:59:22.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/7"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-16 07:58:54.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3585f12c8139d0a3e79eb5240588ca8caeff87e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3585f12c8139d0a3e79eb5240588ca8caeff87e3"}]},"branch":"refs/heads/master"},"6bbf22c8441d259a15966608c1da293e0ce349b9":{"kind":"REWORK","_number":8,"created":"2024-03-16 09:21:20.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/8"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-16 09:20:55.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6bbf22c8441d259a15966608c1da293e0ce349b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6bbf22c8441d259a15966608c1da293e0ce349b9"}]},"branch":"refs/heads/master"},"dfe1ec47b5aa192dff955ea55ed2a4cc71cfd46c":{"kind":"REWORK","_number":9,"created":"2024-03-19 12:25:37.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/9"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-19 12:25:12.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dfe1ec47b5aa192dff955ea55ed2a4cc71cfd46c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/dfe1ec47b5aa192dff955ea55ed2a4cc71cfd46c"}]},"branch":"refs/heads/master"},"fceb19204b01dce7c030699dca2c1074e599db0d":{"kind":"REWORK","_number":10,"created":"2024-03-19 12:47:35.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/10"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-19 12:47:12.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fceb19204b01dce7c030699dca2c1074e599db0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fceb19204b01dce7c030699dca2c1074e599db0d"}]},"branch":"refs/heads/master"},"a7457735c1dbee2492590188d2545ba79ab1edd3":{"kind":"REWORK","_number":11,"created":"2024-03-20 12:38:14.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/11"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-20 12:37:57.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a7457735c1dbee2492590188d2545ba79ab1edd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a7457735c1dbee2492590188d2545ba79ab1edd3"}]},"branch":"refs/heads/master"},"fe1f6fdd3269d7c9ebb1422a841d9e89001c28de":{"kind":"REWORK","_number":12,"created":"2024-03-20 12:41:30.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/12"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-20 12:41:12.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fe1f6fdd3269d7c9ebb1422a841d9e89001c28de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fe1f6fdd3269d7c9ebb1422a841d9e89001c28de"}]},"branch":"refs/heads/master"},"4c71b383b6db1e7f12f6da7f08bc7336b2e2d4b1":{"kind":"REWORK","_number":13,"created":"2024-03-22 14:15:25.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/13"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-22 14:15:09.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4c71b383b6db1e7f12f6da7f08bc7336b2e2d4b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4c71b383b6db1e7f12f6da7f08bc7336b2e2d4b1"}]},"branch":"refs/heads/master"},"1bda25df3311c44e3ba7c57ac98a858e55a0cf84":{"kind":"REWORK","_number":14,"created":"2024-03-22 16:40:50.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/14"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-22 16:40:19.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1bda25df3311c44e3ba7c57ac98a858e55a0cf84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1bda25df3311c44e3ba7c57ac98a858e55a0cf84"}]},"branch":"refs/heads/master"},"170e5753346d89d7104d293f91160e6b2cc77fe2":{"kind":"REWORK","_number":15,"created":"2024-03-27 06:18:35.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/15"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-27 06:18:11.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/170e5753346d89d7104d293f91160e6b2cc77fe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/170e5753346d89d7104d293f91160e6b2cc77fe2"}]},"branch":"refs/heads/master"},"64fe7c3d0901e7a3281af0ed493dcc0ca8f8e98a":{"kind":"REWORK","_number":16,"created":"2024-03-27 10:07:23.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/16"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-27 10:07:00.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/64fe7c3d0901e7a3281af0ed493dcc0ca8f8e98a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/64fe7c3d0901e7a3281af0ed493dcc0ca8f8e98a"}]},"branch":"refs/heads/master"},"6cf77ccc65607c6bff3573fb6c7c9ad62224473a":{"kind":"REWORK","_number":17,"created":"2024-04-04 11:02:53.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/17"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-04-04 11:02:05.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6cf77ccc65607c6bff3573fb6c7c9ad62224473a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6cf77ccc65607c6bff3573fb6c7c9ad62224473a"}]},"branch":"refs/heads/master"},"6ec0331fc24debd0e6df2d9af7c4e2607ace3a4a":{"kind":"REWORK","_number":18,"created":"2024-04-04 11:24:45.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/18"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-04-04 11:24:25.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6ec0331fc24debd0e6df2d9af7c4e2607ace3a4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/6ec0331fc24debd0e6df2d9af7c4e2607ace3a4a"}]},"branch":"refs/heads/master"},"97336fba555a664dbc14b69f1f9ea6be1e9ed4fe":{"kind":"NO_CHANGE","_number":19,"created":"2024-04-04 21:58:59.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/19"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-04-04 21:58:18.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/97336fba555a664dbc14b69f1f9ea6be1e9ed4fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/97336fba555a664dbc14b69f1f9ea6be1e9ed4fe"}]},"branch":"refs/heads/master"},"a98d55714b5c64938d01b85af8def78fc08c784d":{"kind":"REWORK","_number":20,"created":"2024-04-05 05:14:03.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/20"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-04-05 05:13:23.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a98d55714b5c64938d01b85af8def78fc08c784d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a98d55714b5c64938d01b85af8def78fc08c784d"}]},"branch":"refs/heads/master"},"b97ae959cb6a5a6935fd1dc91ed40c2a5af80410":{"kind":"REWORK","_number":21,"created":"2024-04-05 14:52:00.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/21"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-04-05 14:51:38.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b97ae959cb6a5a6935fd1dc91ed40c2a5af80410"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b97ae959cb6a5a6935fd1dc91ed40c2a5af80410"}]},"branch":"refs/heads/master"},"85a9f98825925f6de6a8c4bdfe6cc3516bc85987":{"kind":"REWORK","_number":22,"created":"2024-04-05 15:08:08.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/22"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-04-05 15:07:53.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/85a9f98825925f6de6a8c4bdfe6cc3516bc85987"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/85a9f98825925f6de6a8c4bdfe6cc3516bc85987"}]},"branch":"refs/heads/master"},"23de7b4b0cfb7c3ac4663f395298d19fa7a8776f":{"kind":"NO_CHANGE","_number":23,"created":"2024-04-05 15:09:09.000000000","uploader":{"_account_id":36853,"name":"christian okeke","email":"eceokekechima@gmail.com","username":"ccokeke"},"ref":"refs/changes/94/911694/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/94/911694/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/94/911694/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/94/911694/23"}}},"commit":{"parents":[{"commit":"f4c77bfe7111bc000ba3f663ce5820f06a7df17b","subject":"Merge \"Add cephfs filesystem to shares metadata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f4c77bfe7111bc000ba3f663ce5820f06a7df17b"}]}],"author":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-03-07 11:15:20.000000000","tz":60},"committer":{"name":"Okeke Christian","email":"eceokekechima@gmail.com","date":"2024-04-05 15:08:55.000000000","tz":60},"subject":"Add delete CLI helper to ShareCommands","message":"Add delete CLI helper to ShareCommands\n\nThis patch adds a delete cli command to ShareCommands\nThe helper CLI command checks if service if up. If service\nis down, it deletes the share instance\n\nCloses-Bug: #1867030\nChange-Id: I6a0575c1ed86213010e50fe1b7a733cdf7fa1736\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/23de7b4b0cfb7c3ac4663f395298d19fa7a8776f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/23de7b4b0cfb7c3ac4663f395298d19fa7a8776f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
