)]}'
{"id":"openstack%2Fmanila~916608","triplet_id":"openstack%2Fmanila~master~I989e825ef160a65a5a72c58d6fd4e8099044bcd5","project":"openstack/manila","branch":"master","attention_set":{},"removed_from_attention_set":{"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2024-04-30 06:58:33.000000000","reason":"\u003cGERRIT_ACCOUNT_30407\u003e replied on the change","reason_account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-04-30 08:01:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I989e825ef160a65a5a72c58d6fd4e8099044bcd5","subject":"Ensure password values are not logged","status":"MERGED","created":"2024-04-22 06:36:50.000000000","updated":"2024-04-30 08:02:49.000000000","submitted":"2024-04-30 08:01:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916608","meta_rev_id":"e2dd8c363ac6773c16f1bec66110761d50fd87b4","_number":916608,"virtual_id_number":916608,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-30 08:01:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2024-04-22 22:47:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":2,"date":"2024-04-30 06:58:33.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":1,"date":"2024-04-22 15:16:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":1,"date":"2024-04-30 06:58:33.000000000","_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-22 07:11:48.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-22 07:52:34.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-04-22 08:35:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-22 09:05:50.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-04-22 09:12:43.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-04-22 10:28:55.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-04-22 15:16:58.000000000","updated_by":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"},{"updated":"2024-04-22 22:47:51.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-04-29 17:58:23.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"}],"messages":[{"id":"2c9ef1782e38ae2c0a13acb1cdf3f67820495178","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-04-22 06:36:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ca9344c5bb419271a22c3a571e7b8eae36ccafe","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-22 07:11:48.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/08/916608/1/check/DellEMC_PowerFlex_Manila_dhss_false/24a60c1/DellEMC_PowerFlex_Manila_dhss_false/901 : SUCCESS in 34m 42s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"d160502f2948115f6eabdb4739d0633b9f749c8d","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-04-22 07:52:34.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-08-916608-1/NFS/ : SUCCESS in 1h 15m 20s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"dc44da039e05dd92a2c7e129f14f1d5e9f36b3ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 08:35:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65ab956576c0456d8b0612bb21626d61\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dbf9cf68297417a85e226393a2dbce9 : SUCCESS in 10m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f9ddb527e4744a6992f1f99077900c7 : SUCCESS in 20m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c77641d1e174ac0b8c89a3e0fb5eb6f : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b7c79915d2a422f9d9a5cd040092544 : SUCCESS in 12m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5834b3867794d4d9b45d1c0580ddeaa : SUCCESS in 7m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a7dd7a475f4249a1b621c050ec438627 : SUCCESS in 13m 23s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/46cc476fd8b64b7da88b1bda2dada9df : SUCCESS in 55m 46s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3f34e97716cd45deb667b01fa248fa6d : SUCCESS in 1h 38m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/2f803fd2a1b149b5b13371e18b18a979 : SUCCESS in 1h 21m 11s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ecca64c5041c49a6bf701eb9a462c761 : SUCCESS in 1h 18m 19s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f42c47a938384f9994b694637261005c : SUCCESS in 1h 35m 29s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/86daba5a02a44b9396732d3235101be1 : POST_FAILURE in 43m 35s (non-voting)\n- manila-tox-py311-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/264caedb2c684942a2ea8f8258d15183 : SUCCESS in 10m 49s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/351e88d9e465443ab0cf4a7d9e8f9727 : SUCCESS in 4m 22s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b8dd139220c44f7387dbb162d68a9109 : FAILURE in 3m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/a7b3d1a8b0274d729025651b654fb460 : SUCCESS in 39m 04s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/05012fb49d0b4e5f8873c922b14ce6f3 : SUCCESS in 43m 53s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/1a4e0555125d477d970b07357660a9a9 : SUCCESS in 47m 24s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/31b1612ce0c34004a89b85c13ce970df : SUCCESS in 1h 15m 39s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/754f9cd7460b401b95aeac18bf707501 : SUCCESS in 48m 47s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/579b365d2f5d463bb69ceffd0b3d1120 : SUCCESS in 1h 07m 46s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/125fb64ad6ad4e35a981f13f8fd97a84 : SUCCESS in 42m 08s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/089eb3fad0c3472db342e84a2ab9e1d4 : SUCCESS in 45m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43817d16d11262f8df603f3e4fd383cac2b05520","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-04-22 09:05:50.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/08/916608/1/upstream-check/manila-tempest-plugin-ontap-dhss/c352693/ : RETRY_LIMIT in 3m 34s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/08/916608/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/8c39a15/ : RETRY_LIMIT in 3m 22s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"863f9dbcccead106fb08fca3ab470743318c7b38","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-04-22 09:08:14.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/08/916608/1/check/DellEMC_PowerFlex_Manila_dhss_false/cd6eff0/DellEMC_PowerFlex_Manila_dhss_false/904 : SUCCESS in 32m 57s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"504756e38bb4c6c684ca84529ddb70c1ef67d803","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-04-22 09:12:43.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/08/916608/1/check/DellEMC_PowerStore_Manila_dhss_false/085d84b/DellEMC_PowerStore_Manila_dhss_false/208 : SUCCESS in 37m 27s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"6a170f9ba223b0986d042d14e885baa2abdd7e4c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-04-22 10:28:55.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/08/916608/1/check/DellEMC_Unity_Manila_dhss_true/ba50a4f/DellEMC_Unity_Manila_dhss_true/1106 : SUCCESS in 3h 51m 48s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/08/916608/1/check/DellEMC_Unity_Manila_dhss_false/5f40cdb/DellEMC_Unity_Manila_dhss_false/1198 : SUCCESS in 1h 16m 32s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"2b77c52a7e17681421604826bca09e153bcc448d","author":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"date":"2024-04-22 15:16:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e09df5523805821303c5bd1a3d4d00d3d46654df","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-04-22 22:47:51.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c9634002524a45d6b629621b02b6b695e738eaa5","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2024-04-30 06:58:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"00932d5ff811dbbe34f548544e5bcf5b97d03359","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 06:59:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e5d30338265a2edf5eca31d801332e0da03491b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 08:01:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c934f4a9e5d4c8186e45f702143f7cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f7e011cf8ea458a90b549b3b463eb73 : SUCCESS in 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc1105cd08b648fc87568d1c213d55ed : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e49db84cff443be8725a953a43ccd06 : SUCCESS in 10m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/490bbb37dff64a80a825818b5e3914fa : SUCCESS in 8m 53s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d82f4d6a7a6e47c2a75475faad1dbd94 : SUCCESS in 49m 55s\n- manila-tox-py311-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/60519b3616ec4c22a596d98d0ddd17ab : SUCCESS in 11m 32s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/66b53051b60d4b7fb1a46772bf58c203 : SUCCESS in 40m 11s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/cc810ef7bf62472a963a1a0abefb9cb9 : SUCCESS in 33m 53s","accounts_in_message":[],"_revision_number":1},{"id":"e3cb3362818655abb823cec17df943b1a326d444","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 08:01:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e2dd8c363ac6773c16f1bec66110761d50fd87b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 08:02:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d948615a7234d60b2d98f2655cb1d4b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4376b74079b472abe2890852da023ee : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f8a53c037a1f80880a9180b218288f1766084449","revisions":{"f8a53c037a1f80880a9180b218288f1766084449":{"kind":"REWORK","_number":1,"created":"2024-04-22 06:36:50.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/08/916608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/08/916608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/08/916608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/08/916608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/08/916608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/08/916608/1"}}},"commit":{"parents":[{"commit":"a30697d3e78c49297e3a67db902d21a8254b000c","subject":"Merge \"db: Remove session arguments from AZ methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a30697d3e78c49297e3a67db902d21a8254b000c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-22 06:35:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-04-22 06:36:26.000000000","tz":540},"subject":"Ensure password values are not logged","message":"Ensure password values are not logged\n\nMake sure that all password options is defined with secret\u003dTrue so that\nthe given values do not appear in debug logs.\n\nAlso remove the redundant default\u003dNone .\n\nChange-Id: I989e825ef160a65a5a72c58d6fd4e8099044bcd5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f8a53c037a1f80880a9180b218288f1766084449"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f8a53c037a1f80880a9180b218288f1766084449"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
