)]}'
{"id":"openstack%2Fmanila~918436","triplet_id":"openstack%2Fmanila~master~Iccc028f15078a3e197eb2aedde8e289102bf6d74","project":"openstack/manila","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-07 17:34:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iccc028f15078a3e197eb2aedde8e289102bf6d74","subject":"Remove logic for Python 2 compatibility","status":"MERGED","created":"2024-05-07 12:16:49.000000000","updated":"2024-05-07 17:35:23.000000000","submitted":"2024-05-07 17:34:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918436","meta_rev_id":"1e33f521ca1a97e258ba4089c6c8390b666af1dc","_number":918436,"virtual_id_number":918436,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-07 17:34:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":2,"date":"2024-05-07 16:25:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":1,"date":"2024-05-07 16:25:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-07 13:04:32.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-05-07 13:18:00.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-07 15:26:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-07 15:46:47.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-05-07 16:25:03.000000000","updated_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-05-07 16:34:42.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"}],"messages":[{"id":"5973423ddcf853195d95a6d1ebcd34179b573f17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-07 12:16:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c5c1f7047f21a208f3cc02b4021d49b57db4937","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-07 13:04:32.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/36/918436/1/check/DellEMC_Unity_Manila_dhss_true/b2e6603/DellEMC_Unity_Manila_dhss_true/1201 : FAILURE in 24m 22s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/36/918436/1/check/DellEMC_Unity_Manila_dhss_false/12fee52/DellEMC_Unity_Manila_dhss_false/1299 : FAILURE in 19m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"1f0a10e084d124aaa05d2d0fa7c9019fddaa1323","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-07 13:18:00.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/36/918436/1/check/DellEMC_PowerFlex_Manila_dhss_false/c7df350/DellEMC_PowerFlex_Manila_dhss_false/998 : SUCCESS in 30m 35s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"3c18b8cc39ec28995df2ce5916753648ac6bc13b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 15:26:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d846f9fca9a47289fe8bc6511f7bc3f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/667c5e34844c40259b82e084c9091aaf : SUCCESS in 8m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4f7b683ee9f41458f38dac9fafde4b4 : SUCCESS in 26m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49b1127a08764cc6bbcafac0a4ab0cf9 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e9ad78af24e4dbb8f57ef4e271ad38a : SUCCESS in 9m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f9148ae81d89479aa82572b5f5f38e0a : SUCCESS in 8m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff6a76e5e4ff479cb072b7c23264b777 : SUCCESS in 9m 14s (non-voting)\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a91197ee22474a77b0000695dd45e2b5 : SUCCESS in 1h 16m 48s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/7fed160abeb0471a82713be8e28dc718 : SUCCESS in 2h 36m 54s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/de8accb054ae4a9aa9fb7d1fd69a552f : FAILURE in 3h 00m 26s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/badb1e5cda094db387e5bedefb6721be : SUCCESS in 1h 11m 54s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/429e978b1675482d932fde7c8357ed4b : FAILURE in 1h 34m 49s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b2b873ec50fc4233ab20c7d0f3fc4b19 : SUCCESS in 57m 22s (non-voting)\n- manila-tox-py311-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/8aaf1cda1f7146afb920246dc144b4d5 : SUCCESS in 9m 24s\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/c0fbd9ee89ca499592a535f8c6014f94 : SUCCESS in 4m 08s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/50e1857d872344949a05fa6f21dd7675 : FAILURE in 3m 27s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8aa8ffd649674de4a213172d51f830bb : SUCCESS in 43m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c482317140d44aa492791fb5fb464fc3 : SUCCESS in 30m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/b09992b2e71644ce83e9bb8f4dd607fa : SUCCESS in 48m 03s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/aa9d5c24975e4750891cf31428786081 : SUCCESS in 1h 07m 03s (non-voting)\n- manila-grenade https://zuul.opendev.org/t/openstack/build/6a0f72f8070d4db48f4dbcf18d38da60 : SUCCESS in 50m 48s\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7adfc99551da4a62955312bd2d00ecc5 : SUCCESS in 1h 02m 11s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/7a49e91e8da3401eafa89cacc767c228 : SUCCESS in 43m 27s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/36527c3f51344823b63b037504b6b695 : SUCCESS in 35m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c4af6e44d73295dec42f8c4a9d80d12dff3663ee","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-05-07 15:46:47.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/918436/1/upstream-check/manila-tempest-plugin-ontap-dhss/c4706f3/ : RETRY_LIMIT in 3m 46s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/36/918436/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/90093f5/ : RETRY_LIMIT in 3m 31s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"7103d56dbb74a7c9bb83b049178e4676aa483075","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-05-07 16:10:54.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/36/918436/1/check/DellEMC_Unity_Manila_dhss_true/4c289fe/DellEMC_Unity_Manila_dhss_true/1205 : FAILURE in 25m 10s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/36/918436/1/check/DellEMC_Unity_Manila_dhss_false/216b439/DellEMC_Unity_Manila_dhss_false/1303 : FAILURE in 19m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"0e8a625c4b3ec51aa53bc9966553c921b6de98bf","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-05-07 16:25:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"32bfcea5297edf71505a85aa6bcef7cd7e145a3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 16:25:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8c16c9bbf51b18f4e702c2b66f3e0634b4c61873","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-05-07 16:27:39.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/36/918436/1/check/DellEMC_PowerFlex_Manila_dhss_false/54da742/DellEMC_PowerFlex_Manila_dhss_false/1002 : SUCCESS in 34m 08s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"fe1a802920ae351bab9067162cfec9a00cc16769","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-05-07 16:34:42.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/36/918436/1/check/DellEMC_PowerStore_Manila_dhss_false/3c8fd53/DellEMC_PowerStore_Manila_dhss_false/246 : SUCCESS in 37m 10s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"171817bf88faf398b1f6dbf04731ff15cb811664","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 17:34:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cd208d4c2de46daaabcf0f50595d6d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2218372ebc1d41b497c56c9f6f393adb : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8ea88a63e89447fb525caa7c8ed9dee : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/214b454f8eb3497b90bceba708f3a639 : SUCCESS in 11m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8bca49cdabe749fcb29b8552a0bda496 : SUCCESS in 8m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/db35d74324ee482b8147cf65f55957e2 : SUCCESS in 1h 07m 17s\n- manila-tox-py311-with-sqlalchemy-2x https://zuul.opendev.org/t/openstack/build/a49666249421416b8069a2336fb2a48e : SUCCESS in 9m 39s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/faa91542a3034a5da46a56fcc3869955 : SUCCESS in 37m 34s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/77edb00fccf84c0f9806716e0abe633f : SUCCESS in 41m 07s","accounts_in_message":[],"_revision_number":1},{"id":"7648ed87c698f9492cf94d72d7d4df0ecdbe9a8c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 17:34:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1e33f521ca1a97e258ba4089c6c8390b666af1dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 17:35:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca00b9508b24450d8b60f21a1811c944\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23618a95a1964ca0a1b9096ab47e323a : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"224e304e8cce8debf410a1a8a0fa1a7e97c31f50","revisions":{"224e304e8cce8debf410a1a8a0fa1a7e97c31f50":{"kind":"REWORK","_number":1,"created":"2024-05-07 12:16:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/36/918436/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/36/918436/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/918436/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/918436/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/36/918436/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/36/918436/1"}}},"commit":{"parents":[{"commit":"842fda18e50b9099dbba8a3624b1f940c8329578","subject":"Merge \"Remove old excludes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/842fda18e50b9099dbba8a3624b1f940c8329578"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-07 12:15:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-07 12:16:25.000000000","tz":540},"subject":"Remove logic for Python 2 compatibility","message":"Remove logic for Python 2 compatibility\n\nThe collections module should/can be always used in Python 3.\n\nChange-Id: Iccc028f15078a3e197eb2aedde8e289102bf6d74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/224e304e8cce8debf410a1a8a0fa1a7e97c31f50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/224e304e8cce8debf410a1a8a0fa1a7e97c31f50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
