)]}'
{"id":"openstack%2Fmanila~921313","triplet_id":"openstack%2Fmanila~master~Ie827619d3952d1db69559206b5dfd47efb4cf7b1","project":"openstack/manila","branch":"master","topic":"bug/2068043","attention_set":{},"removed_from_attention_set":{"18816":{"account":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"last_update":"2024-09-19 19:52:40.000000000","reason":"Change was submitted"},"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-09-19 18:11:37.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-09-19 18:24:17.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2024-09-19 19:52:40.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2024-09-19 19:52:40.000000000","reason":"Change was submitted"},"30407":{"account":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"last_update":"2024-09-19 19:52:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie827619d3952d1db69559206b5dfd47efb4cf7b1","subject":"Update share/snapshot instance deferred deletion","status":"MERGED","created":"2024-06-04 17:05:33.000000000","updated":"2024-09-19 19:53:52.000000000","submitted":"2024-09-19 19:52:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":101,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921313-bug/2068043","meta_rev_id":"a40c288b24ab9ba9ad1979c98ddc42565730870b","_number":921313,"virtual_id_number":921313,"owner":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-19 19:52:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-09-19 18:11:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":2,"date":"2024-09-19 18:24:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"value":1,"date":"2024-09-19 18:24:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},{"value":0,"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-04 17:40:08.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-04 19:22:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-04 19:26:17.000000000","updated_by":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"reviewer":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"state":"CC"},{"updated":"2024-06-04 19:59:23.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-06-04 20:49:56.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-06-04 21:57:19.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-06-05 10:17:29.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"REVIEWER"},{"updated":"2024-06-05 10:17:29.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"state":"REVIEWER"},{"updated":"2024-06-05 10:17:29.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2024-06-05 10:17:29.000000000","updated_by":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-07-17 05:06:18.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"state":"REVIEWER"},{"updated":"2024-08-19 12:29:11.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"}],"messages":[{"id":"bceec58655e51eb1f4b64d05e9fbd85e93d3b5ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-06-04 17:05:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e6e8a7c24df84e88af9f989d9fcd76e53504779","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-04 17:40:08.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/1/check/DellEMC_PowerFlex_Manila_dhss_false/aeec846/DellEMC_PowerFlex_Manila_dhss_false/1124 : SUCCESS in 34m 20s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"0cb27c64fb9d1420f7d2edcf8ad0431d9f6832f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 19:22:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86ffb279211541b1acc3c230de7ce09d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9713f0fdb161482e8d793d83370732e7 : SUCCESS in 10m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5809fd6ce56d41449f415e763bd9939e : SUCCESS in 26m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f1e651d9cf942809c224e3f882e69ab : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd1dfc317a5c491ca6c587086666593c : SUCCESS in 9m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10cbe88dc94d4ce5a11685232ab3a07b : SUCCESS in 8m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/13abcfd6f609402cbd7321df644ae0e5 : SUCCESS in 12m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/354581b6c6c74d7bb86cde18aa6fb3f7 : SUCCESS in 10m 02s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/815ff1fee093458b9bb206927b9ddd33 : SUCCESS in 1h 19m 48s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d2730a329fc34a558341846a0da708c5 : SUCCESS in 2h 11m 40s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a0552bd265ba4c63bfedfa4c562d44a4 : FAILURE in 1h 12m 31s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/2cf482f4d88f432aa56d60382014b738 : SUCCESS in 1h 20m 43s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b16765b2c6a04d8cbd5c13f8fc202a44 : SUCCESS in 1h 26m 26s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9de71217323342269e0cdfae82118103 : SUCCESS in 1h 02m 00s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3b1690bf3d9a48d3b39c7b174fb44416 : SUCCESS in 4m 31s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/059219ccc4754da48702a319f980034c : FAILURE in 3m 21s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f6c73504a6204fd6ad678ae4fe1909f8 : SUCCESS in 36m 06s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3bcd5389cb114794aef8872cec51a404 : SUCCESS in 33m 00s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/205489c1f6934c38b1669ddab4ca6ee3 : FAILURE in 34m 21s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/6eb5be8812764710b13095e9134ed59b : SUCCESS in 59m 36s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e613483cb3f344cca84354061dcc5d62 : SUCCESS in 1h 02m 06s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/b6bb6c47b30e4284884b24db4fc6dd18 : SUCCESS in 28m 37s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/7fad493cc09047bdbd784a6281828aa0 : SUCCESS in 39m 02s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/2c9afbf0e01443849b27b1c94846bb93 : FAILURE in 4m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f107ef9cdb90fce1bb3e59b160d12e1edbfd8ab","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-06-04 19:26:17.000000000","message":"Patch Set 1:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-921313-1/NFS/ : SUCCESS in 2h 20m 24s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":1},{"id":"2bfb07a5f1e0ac461b372e13a103958df93e588d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-04 19:53:43.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/1/check/DellEMC_PowerFlex_Manila_dhss_false/3f25537/DellEMC_PowerFlex_Manila_dhss_false/1125 : SUCCESS in 30m 40s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"2f91fa821088d9c6949584a7ae5acf3c5dcbfefa","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-06-04 19:59:23.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/1/check/DellEMC_PowerStore_Manila_dhss_false/0468654/DellEMC_PowerStore_Manila_dhss_false/294 : SUCCESS in 36m 20s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":1},{"id":"700aba01a012644bf0754263b7d043e23ed312d6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-06-04 20:49:56.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/921313/1/upstream-check/manila-tempest-plugin-ontap-dhss/9944a7c/ : FAILURE in 1h 24m 46s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/921313/1/upstream-check/manila-tempest-plugin-ontap-no-dhss/db5bab9/ : FAILURE in 1h 26m 32s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":1},{"id":"24a7185e11863bbc1243bb786e77e133171f0a4e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-04 21:57:19.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/921313/1/check/DellEMC_Unity_Manila_dhss_true/92f11af/DellEMC_Unity_Manila_dhss_true/1289 : FAILURE in 4h 51m 22s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/1/check/DellEMC_Unity_Manila_dhss_false/83802a9/DellEMC_Unity_Manila_dhss_false/1388 : SUCCESS in 2h 05m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"0a113bcd3e827c4f0a7c075bef7b5b0a7a3567de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-06-07 08:42:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"358f48c67cbf4dcde1c84af4a7c6d8ea8072c0ce","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-06-07 10:47:36.000000000","message":"Patch Set 2:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-921313-2/NFS/ : SUCCESS in 2h 4m 42s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":2},{"id":"e960ca47fe6acc82a526bf68a1f60a3af624fffb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-06-07 11:10:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c1e68d78ec695c39202ddde4f6b943e3a4acc6b8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-06-07 13:18:27.000000000","message":"Patch Set 3:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-921313-3/NFS/ : SUCCESS in 2h 7m 43s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":3},{"id":"6b02a9b43ef66ad0c6719ee68bf11dae69fab253","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 13:44:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fb81804ce1d48f7a318a471f7b8bd08\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e390a994d9c84af4b9e1d4e3b5e2b001 : SUCCESS in 10m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93f0a7db1c7a4c328c1f5747a47b5c4b : FAILURE in 28m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abe2c71553c14cb1af6db4ac261113e1 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/256a0e1cb33644f19931ae74a3edf92f : SUCCESS in 8m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d2e99fe8b4e43508c213e7c6564ad69 : SUCCESS in 9m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/78b44044abec489ca854f8c3fdafd5c8 : SUCCESS in 11m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b43d950cfefa4c91bd8e8cb1ae622b10 : SUCCESS in 8m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/d4349d64f92c424bae9370f59d665af0 : SUCCESS in 1h 01m 21s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/2a5563f5cfd14751a0e69204adeb782a : SUCCESS in 2h 09m 44s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6350c894197444458a277759894e5a11 : FAILURE in 2h 26m 59s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/9144c6a92dab42359f3eebe4ba63e4c1 : SUCCESS in 52m 26s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/b051efa5adf24705abae189e83c48600 : FAILURE in 1h 16m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0318f0ec176d43b2b1149e7f3514063e : SUCCESS in 1h 11m 31s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/8d14c4dac535436d8f707f861bb70d82 : SUCCESS in 3m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c1b430a690ec467c8227dcce665ca93e : FAILURE in 4m 30s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/cfda4a31b5674d1b901abe675ecaf66d : SUCCESS in 32m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/9af8bb7c19344133a3306fcdba72ddcb : SUCCESS in 39m 06s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e95e6e82f78346cb97335521fff468dc : FAILURE in 49m 15s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/b1eb2870c55944c4bf52b7b8d9b5664b : SUCCESS in 56m 12s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f7251a0ddbd04d4f8868b0550cb02d59 : SUCCESS in 59m 03s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/100112b9a5aa44339cdac65866a61ce2 : SUCCESS in 42m 55s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f249259cdd94433faa08c30a2a8e39bf : SUCCESS in 30m 30s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/506ee4cf145740d3a759d0706236e402 : FAILURE in 4m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dc304d667b330eed9e99ba2392fddbd5017ba960","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-06-07 14:05:54.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/921313/3/upstream-check/manila-tempest-plugin-ontap-dhss/0157aa7/ : RETRY_LIMIT in 8m 48s\n- manila-tempest-plugin-ontap-no-dhss finger://managesf.services.openstack.netapp.com:7979/5953d6fea669404d8a05337463233aec : RETRY_LIMIT in 19m 59s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"be158cf310ed437b5707fc00876728922f715f57","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-07 14:10:40.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/3/check/DellEMC_PowerFlex_Manila_dhss_false/e2d75e9/DellEMC_PowerFlex_Manila_dhss_false/1136 : ABORTED in 3h 00m 01s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"31fb0b6d5f27324306ed9594557f78e7fcb986ca","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-06-07 14:18:44.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/3/check/DellEMC_PowerStore_Manila_dhss_false/e0fa770/DellEMC_PowerStore_Manila_dhss_false/299 : SUCCESS in 33m 50s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"bbdfee30b3966e73b40424cd96463a1d11a5b040","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-07 14:19:21.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/921313/3/check/DellEMC_Unity_Manila_dhss_true/2dc255a/DellEMC_Unity_Manila_dhss_true/1296 : SUCCESS in 3h 08m 44s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/3/check/DellEMC_Unity_Manila_dhss_false/42260a5/DellEMC_Unity_Manila_dhss_false/1394 : SUCCESS in 51m 09s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"000c46b426d4cdb0bbee3463b4d8eaa3fa28a9ba","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2024-06-12 03:09:46.000000000","message":"Patch Set 3: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0863012634401327d560abe5a952c075d5609731","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-06-12 07:13:41.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9bb3b471e50c82c8eb159a07962bd066f2cdb3fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-06-12 07:28:45.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"23ae8f4641b06b49a7f14ab5a6cba803f0526d9f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-12 08:00:27.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/4/check/DellEMC_PowerFlex_Manila_dhss_false/a7b8d72/DellEMC_PowerFlex_Manila_dhss_false/1161 : SUCCESS in 31m 26s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"dcc5f5e395c18a2e6c044ee1f20a36d63ea9723a","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2024-06-12 08:05:35.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b8b46be81fd762677dd499339c5e7b8f84fbd045","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-06-12 08:46:03.000000000","message":"Patch Set 4:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-921313-4/NFS/ : SUCCESS in 1h 16m 57s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":4},{"id":"958eb7e9d8c893e74ce4c10847eba60da2bd3fae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 09:58:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bbc6bf543594488b33fa35d70381e9f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ebf0464725544cab5eb8f4ca324ada4 : SUCCESS in 11m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/477bb6a21fec4ef1a6b8436b06d58db4 : FAILURE in 28m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51a5df9c6e9249bcbd73bdad1fa7bf61 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08270d5324b24d6da21616a7a0be373d : SUCCESS in 10m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc1cf7abe6ca4b6db832a89b27fa3130 : SUCCESS in 9m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d1899ee49fb44d19b7b656afccdd367b : SUCCESS in 13m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eba06ffe8c804138a0ca330619137442 : SUCCESS in 10m 34s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4f648873cfbc402491dc970e0bf4f468 : SUCCESS in 1h 11m 45s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5c5a632075fb45b9bc7a1c464b436a5e : SUCCESS in 2h 23m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/10ff8aa4535a48c5a8fd10a6d3ca50d7 : FAILURE in 1h 15m 35s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/6b5a51eaa07f4ca5abc3cbf9a0987cb7 : SUCCESS in 1h 14m 19s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/baa8a3efa9514588bf23493e5a06d75d : FAILURE in 1h 41m 57s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/bd85409dad05422cbf1da7ad876af7fe : SUCCESS in 1h 23m 13s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/3a3f3f31ff0b4cfaac6cfd9dbc9bda32 : SUCCESS in 5m 10s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/212a1cceee474903a14989de2abc767e : FAILURE in 3m 51s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/59ee9fd2c83f4b31a6b401bdc3360133 : SUCCESS in 36m 10s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8a183080e023426495c21febaebc1f2b : SUCCESS in 43m 28s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/c7fc0b9b6cb6436f9ba970456356f9f6 : FAILURE in 53m 22s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/23694a6cdb6346be86e9e0fc10d8f1a8 : SUCCESS in 1h 11m 55s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/518597976b9845a59c75e9df2d23ea2a : SUCCESS in 59m 51s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/d69ef9a2800248d4ab38c71bf1f00a72 : SUCCESS in 36m 42s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c24375b599ea4a25b3ffd67bab7fbfca : SUCCESS in 42m 37s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/fd26c662634e4cd29db79c6b3e4138a4 : FAILURE in 3m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"51fcce32df51b9541735be83223bbbdf4af584e8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-12 10:30:14.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/4/check/DellEMC_PowerFlex_Manila_dhss_false/15c19ad/DellEMC_PowerFlex_Manila_dhss_false/1163 : SUCCESS in 31m 45s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"1b33b7304f095cafe7d395c059d142a29857d60d","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-06-12 10:34:01.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/4/check/DellEMC_PowerStore_Manila_dhss_false/b8596c9/DellEMC_PowerStore_Manila_dhss_false/310 : SUCCESS in 35m 33s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"0a7caa45d5ba1c1000aa0436764e5dc0ba50704a","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-06-12 10:57:06.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/921313/4/upstream-check/manila-tempest-plugin-ontap-dhss/9a924f5/ : FAILURE in 57m 50s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/921313/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/dba4520/ : RETRY_LIMIT in 12m 41s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"df19bab29b68f2b217f45df34f6662d7c4a77363","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-12 13:06:06.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/921313/4/check/DellEMC_Unity_Manila_dhss_true/296f03e/DellEMC_Unity_Manila_dhss_true/1312 : SUCCESS in 1h 54m 29s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/4/check/DellEMC_Unity_Manila_dhss_false/07deb46/DellEMC_Unity_Manila_dhss_false/1410 : SUCCESS in 57m 40s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"c6a0009c285dbce10af380861c16cf5a7ce6ddb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-06-21 13:42:44.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"09659009acea18f558dacdd4e7a7e9be276e8a6e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-21 14:17:43.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/5/check/DellEMC_PowerFlex_Manila_dhss_false/027686b/DellEMC_PowerFlex_Manila_dhss_false/1195 : SUCCESS in 33m 15s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"9b5d6f1e421ac801b0e1b4b722541553fd6e8a13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-21 15:45:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e93858852994526a0137de3cf8e631b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d890153f4f8f4413816d4f779924918e : SUCCESS in 10m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bef6f7cec3ad49928eef29bc8c461b35 : FAILURE in 28m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2f7e62139444fb9b5dd394fa89bd3f7 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9fc0a33f7704f189844153fbf753a9c : SUCCESS in 8m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cbf82fab1c2c40e3a010b6abeff08dc2 : SUCCESS in 7m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e506ffbdeaf46f9a9da76aa14f9461f : TIMED_OUT in 40m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fe9c1f3c1b94fa7937aae279ecd2040 : SUCCESS in 9m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4d91f5d79b8d414a9e948270202478cc : SUCCESS in 56m 04s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/a42e061aaaf94b6a85a13f2f1fb80c10 : FAILURE in 2h 01m 26s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/d54b69db65b14606a9d0a0b7161ee7d5 : FAILURE in 44m 50s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/b7d0e943b5174b2ab8491e51f4b24762 : SUCCESS in 1h 01m 47s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/24b8825024394de2949e34791e222565 : FAILURE in 1h 27m 36s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/b9b2fb1bf07d48deac5f8d13e7af7e3c : SUCCESS in 1h 21m 12s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/1cad0f2fe31c48359a9f76b18919cd53 : SUCCESS in 6m 47s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/16385f4c6a96402c8e0e391a309ad7f8 : SUCCESS in 6m 16s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3d83277767854ace86c35834196c2bd0 : SUCCESS in 34m 21s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8a2256bb89bd48e0a0bc791dc1121b1a : SUCCESS in 39m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e7ace73bec034a1a91a0deb0fb1a3c24 : FAILURE in 44m 57s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/509a2303f5084608852d8391e01d974f : SUCCESS in 1h 09m 03s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6c8ff1542a00427487a68f0dc8f6ec6a : SUCCESS in 1h 05m 00s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/de5bf6f6bb7d4bf9b1b3eda20f95f775 : SUCCESS in 28m 58s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/35721e4e169f45fa81faa5a237b5e099 : SUCCESS in 47m 40s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/9a3264108ae3472892650b63c0684061 : FAILURE in 4m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"73f65b2cf1dc294ec2426c3f4574a5462d285412","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-06-21 15:54:46.000000000","message":"Patch Set 5:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-921313-5/NFS/ : SUCCESS in 2h 11m 40s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":5},{"id":"f209ab0e5a2b632c37df1138c961233ba6c43cb3","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-06-21 16:03:54.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/921313/5/upstream-check/manila-tempest-plugin-ontap-dhss/9245b66/ : RETRY_LIMIT in 8m 48s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/921313/5/upstream-check/manila-tempest-plugin-ontap-no-dhss/f9b60dc/ : RETRY_LIMIT in 12m 28s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"775072055ae4f4f190fafb5e2321ec6c1889df70","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-21 16:17:13.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/5/check/DellEMC_PowerFlex_Manila_dhss_false/f02d684/DellEMC_PowerFlex_Manila_dhss_false/1196 : SUCCESS in 31m 45s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"0ac7c570b7affd176e414f5a09dcc651de775f78","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-06-21 16:21:02.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/5/check/DellEMC_PowerStore_Manila_dhss_false/48d589a/DellEMC_PowerStore_Manila_dhss_false/322 : SUCCESS in 35m 44s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"5af9d1506ca6dca9fb4c3171c2891af50414b97a","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-06-21 16:51:14.000000000","message":"Patch Set 5: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f02203ce182ff250a7b6c21d2b7f004da9ba4450","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-21 18:21:43.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/921313/5/check/DellEMC_Unity_Manila_dhss_true/779b7a6/DellEMC_Unity_Manila_dhss_true/1337 : SUCCESS in 2h 46m 08s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/5/check/DellEMC_Unity_Manila_dhss_false/13a7167/DellEMC_Unity_Manila_dhss_false/1436 : SUCCESS in 51m 23s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"c7db3c24173c29f87952f229b338e0a4182c602d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-06-24 08:35:50.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"4cae491acdd6aec903e209a5512cfed99fa5af8a","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-06-24 08:35:50.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4b82eab242467179022c38334df4fd6b9f0bd92a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-24 09:07:14.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/6/check/DellEMC_PowerFlex_Manila_dhss_false/1968244/DellEMC_PowerFlex_Manila_dhss_false/1208 : SUCCESS in 31m 08s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"5fc71394a429b9f270ca59d1f69a29e59f5ed292","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-06-24 10:40:06.000000000","message":"Patch Set 6:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-921313-6/NFS/ : SUCCESS in 2h 3m 59s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":6},{"id":"3ad0044c8b98e2d6329844313d79661ae01cd989","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 10:55:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fede8bc8803492f8b5181bd33edb00d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdc230929d50448ba873b5511e5cdfc3 : SUCCESS in 8m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9b948cbb71e4b5881b61e9ee34d0c5c : FAILURE in 18m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e04200a6927a4e73b64f1d1e2a141d3d : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e19d53c5117c44779a4a1aaa2cfa044a : SUCCESS in 8m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6dfeaa721914e138e90e5d34c76da06 : SUCCESS in 8m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ab439b874d744deda31311416f0086bc : SUCCESS in 13m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e4bafa2aa81458bad1b69e324436ed8 : SUCCESS in 9m 04s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/06b134b8003d45759382bc7786030457 : SUCCESS in 47m 45s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/368a33c9058044359e27cef7f3155183 : FAILURE in 2h 14m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/909d796b5a2a46649c9bf67c3590023f : FAILURE in 52m 20s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/28fb7ab2a166447ba6d98b31e4307c72 : SUCCESS in 47m 40s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/823ae0042a854a76b2e1c4bc50c6839a : SUCCESS in 1h 30m 53s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/6ce9dcbbe2ad46e6a79cd48299e3ea71 : SUCCESS in 1h 22m 29s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ce48b1f1e338413cb8ea0ff24443b221 : SUCCESS in 3m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/2c284d4861114eba8db780b5e9e2f445 : SUCCESS in 4m 49s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/f4c922a8653b44e2a5a2c63ea78f2834 : SUCCESS in 36m 44s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6cee1ddce20049638685cda6be63beb0 : SUCCESS in 38m 45s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/26c3f6803d034791bdb018b994f09bc5 : FAILURE in 32m 54s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/66d0804d6f8e46d0893978f895b96066 : SUCCESS in 1h 08m 20s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/af3dd31bbd8c44618c1cb4894862b426 : SUCCESS in 1h 01m 05s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/66791446f04b4294977b4139146d5366 : SUCCESS in 42m 55s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/cdaf7b88dc5a40be9b72ab0f2e0c389e : SUCCESS in 44m 54s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/d771bea125824c2299f9287c3108fea3 : FAILURE in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ade0f1103ef8f894fab19d1105e07c2755690aac","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-06-24 11:11:00.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/921313/6/upstream-check/manila-tempest-plugin-ontap-dhss/e8d9176/ : RETRY_LIMIT in 9m 52s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/921313/6/upstream-check/manila-tempest-plugin-ontap-no-dhss/b68d4a2/ : RETRY_LIMIT in 14m 23s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"5b92ef0799c16000d1bcbdcb18821250b8e78ef7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-24 11:29:12.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/6/check/DellEMC_PowerFlex_Manila_dhss_false/dac670c/DellEMC_PowerFlex_Manila_dhss_false/1209 : SUCCESS in 33m 25s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"3bc6b07b819150841846734d8286be3761552101","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-06-24 11:35:52.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/6/check/DellEMC_PowerStore_Manila_dhss_false/6397f27/DellEMC_PowerStore_Manila_dhss_false/327 : SUCCESS in 40m 10s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"81837683cb2cd886d10e01a25dd987dae63b443c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-24 11:46:38.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/921313/6/check/DellEMC_Unity_Manila_dhss_true/5f3e0f6/DellEMC_Unity_Manila_dhss_true/1346 : SUCCESS in 3h 10m 31s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/6/check/DellEMC_Unity_Manila_dhss_false/c678721/DellEMC_Unity_Manila_dhss_false/1445 : SUCCESS in 52m 22s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":6},{"id":"1dfca5a8d44c2856389a0514fe042ffc8f2f8818","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2024-06-25 01:28:09.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4b89b5ce2275cf9b9e79f55f644b923074ea4070","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2024-06-25 08:00:10.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a10914d6a0bb0f5473df24ecbe7ebe8c829633b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-06-25 10:05:44.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"bf9565512a2dbd0c3a7810ae89a1c82f16688fe4","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-06-25 10:05:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"70f6efe14db4f4978392442633a666797a9c1e6e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-25 10:38:12.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/7/check/DellEMC_PowerFlex_Manila_dhss_false/5f4f0d9/DellEMC_PowerFlex_Manila_dhss_false/1213 : SUCCESS in 32m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"2e4157cce1ebd43d95f62750f65f2cd63999d7ec","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-06-25 11:28:06.000000000","message":"Patch Set 7:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-921313-7/NFS/ : SUCCESS in 1h 22m 1s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":7},{"id":"4586ea6b56265ae9521fd4742aca61ed04910b7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 12:18:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76ac62422d0044cbb4fc1bb32322b7da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4ec741264c14a4ea94db0f69e56421c : SUCCESS in 9m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8ad9d2a82ec46b9b4e75538f4f50a5f : SUCCESS in 28m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6aec3f9f3c6041a4ba902b69c702e51b : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/734f590e71f84ed3839696807c4178a6 : SUCCESS in 11m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c2e4fb665df1483095350211f665304d : SUCCESS in 9m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44b7c06b398549a39ceacc99b8d6f3da : SUCCESS in 13m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2258e71f336342cda7a934508a7e4dce : SUCCESS in 8m 09s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/156257841cff4f80bf8838ec3c698eff : SUCCESS in 1h 10m 58s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/cf5a102f2d0440649ca9c0add09d5667 : FAILURE in 2h 07m 50s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/ab5c976f55bc4efbb96643d22b5c4ee4 : FAILURE in 1h 10m 02s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/873cb085f9e64a0394628c01070fc482 : SUCCESS in 1h 14m 57s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/7de23f1ee3794542a7d04fdee5fc7963 : SUCCESS in 1h 38m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/d61d2b3831d84e72a1eefa7b8107269b : SUCCESS in 1h 09m 56s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/ed9c612269fd4a0ebfad8c00e2a1d37b : SUCCESS in 4m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/937c75854e50400588b4ed40f32ad6bc : SUCCESS in 4m 43s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/61f1e025440948128f0a0f18889409d3 : SUCCESS in 33m 40s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0dd4a1e4b64d45a2a4cda0824d7eb87d : SUCCESS in 39m 39s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7de9a496c8784193a3cb588ec0da8b9c : FAILURE in 43m 34s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/29f4802f6de242388bf47b860ef3b0aa : SUCCESS in 1h 08m 21s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/73b2bf77ab634d26aa04fe10d6c21a5d : SUCCESS in 57m 57s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c864507426824677ba2293777161264f : SUCCESS in 52m 02s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/bc00a8d396ce41c4859cabef83fe39b9 : SUCCESS in 38m 53s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/5805b08d47b943ea822d151b6a46a563 : FAILURE in 4m 16s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/b1b5b7f0bb41435ab4e432a8aa79793c : SUCCESS in 4m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fe39f9d1f11d24dc94618c64f81ad9d9fe8eb64f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-06-25 12:49:59.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/7/check/DellEMC_PowerFlex_Manila_dhss_false/2f79e40/DellEMC_PowerFlex_Manila_dhss_false/1214 : SUCCESS in 30m 47s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"aa14ab9dc38169c6dad4230c6110fe4953a6f90e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-06-25 12:56:03.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/7/check/DellEMC_PowerStore_Manila_dhss_false/14cadc3/DellEMC_PowerStore_Manila_dhss_false/330 : SUCCESS in 36m 51s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":7},{"id":"b6df8e5e8baa79fd4195286d980007f75760c68e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-06-25 13:19:08.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/921313/7/check/DellEMC_Unity_Manila_dhss_true/0011b29/DellEMC_Unity_Manila_dhss_true/1349 : SUCCESS in 3h 13m 07s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/7/check/DellEMC_Unity_Manila_dhss_false/bfd021c/DellEMC_Unity_Manila_dhss_false/1448 : SUCCESS in 1h 00m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"f9ccae5505c41937c7bea401d6caea0955965acf","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-06-25 13:46:15.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/921313/7/upstream-check/manila-tempest-plugin-ontap-dhss/35572f8/ : POST_FAILURE in 1h 20m 59s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/921313/7/upstream-check/manila-tempest-plugin-ontap-no-dhss/527013e/ : RETRY_LIMIT in 13m 21s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":7},{"id":"30b64622dfb87553db1e79df816600ac2aefc73f","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-07-10 14:39:01.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b26d699d6480e7e322adedda49dbb018cf1953be","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-07-17 05:06:18.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"60f49a73bf1f3c3282c920939c72df858b26a45f","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-07-18 22:22:45.000000000","message":"Patch Set 7: Code-Review-1\n\n(14 comments)","accounts_in_message":[],"_revision_number":7},{"id":"77e66b166b5e8db7c44f90058721f582689511bb","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2024-07-19 03:05:58.000000000","message":"Patch Set 7: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":7},{"id":"afaeb8a24b00b8d2cb1b5ba00ea33cb9b4c9435d","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-19 11:31:29.000000000","message":"Patch Set 7:\n\n(18 comments)","accounts_in_message":[],"_revision_number":7},{"id":"d309b48d1f6c42261913dcd4842ccc6d66c615b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-07-19 11:35:21.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"080e61563d98e25b4b5248dbe33c7b75dd20f0d8","author":{"_account_id":21863,"name":"Infinidat CI","email":"jenkins@infinidat.com","username":"infinidat-openstack"},"date":"2024-07-19 13:26:42.000000000","message":"Patch Set 8:\n\nBuild succeeded. For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n* Infinidat NFS http://openstack-ci-logs.aws.infinidat.com#logs/refs-changes-13-921313-8/NFS/ : SUCCESS in 1h 20m 34s\n\nLeave a comment with run-INFINIDAT to trigger a recheck. For more information about Infinidat CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/INFINIDAT_CI","accounts_in_message":[],"_revision_number":8},{"id":"61c892317d9a8ef4035115c8e88bb7dc34ccf013","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 13:31:50.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9ff6898fa3943f7bc484f6f46b6d16d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33f43299699644c4bb0c2c9cbf30a018 : SUCCESS in 9m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25dfc92cbc314dc7836bdb65dbe613c5 : FAILURE in 27m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aea766c3743340478f391392b1080e79 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96d7ec3248154863a0b9ca2444b7e09b : SUCCESS in 10m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c3c790212a63462d85a3deb3fdbae028 : SUCCESS in 6m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af746bf81ca14760990eff1b4dacc0ca : SUCCESS in 13m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edc5fc6325034efa8375d0114afcf4e9 : SUCCESS in 7m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/3d62ca914eab4fe9993f87b73de0de42 : SUCCESS in 1h 10m 30s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/bc6ac36830d24290aeb1e2831d99a964 : SUCCESS in 1h 55m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fe7c88b7c6594bf5adebafe07b514a6d : FAILURE in 59m 19s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/bdd799277c55445b92a754a1839edfb8 : FAILURE in 1h 01m 33s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/6426520d7224483e8c0d9b21292ecad3 : FAILURE in 17m 32s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/128f4bf9ba254d23ac6e9cad9ea030e8 : SUCCESS in 1h 10m 56s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d24fce7ceb394643bce07a9163ecf05d : SUCCESS in 5m 44s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/cd8e6246002146c5a9bc7566d1352a93 : SUCCESS in 5m 27s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/45bfe7464057469391161b584c3b7afb : SUCCESS in 31m 56s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/0b47ac301e7d44b89af530d92bcb0130 : SUCCESS in 44m 02s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/aa7b3d4efeee45f98fefd379ea97ddb1 : FAILURE in 48m 12s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/fe452f9a25174050801f2d0410f97c69 : SUCCESS in 1h 27m 00s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5deb1be5953349c8bf14ceae0d16a734 : SUCCESS in 1h 05m 35s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/6869fd7ef6b2477282cea50483b6870b : SUCCESS in 34m 10s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/4a4483cfa19e4cc2abf8ce9150177478 : SUCCESS in 37m 43s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/e8d7e418c7e14c22aac60ecac0191779 : FAILURE in 4m 32s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/d6a03f9b2f464d70b400af3970d473db : SUCCESS in 5m 03s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c68e401a6ebe0b8022ccc2f431cf90d2bd8377ba","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-07-19 13:51:38.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/921313/8/upstream-check/manila-tempest-plugin-ontap-dhss/0ebfe1b/ : RETRY_LIMIT in 3m 25s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/921313/8/upstream-check/manila-tempest-plugin-ontap-no-dhss/a8f9b85/ : RETRY_LIMIT in 10m 20s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":8},{"id":"69cee1211d57179ac1321ce9da4c421b281508d5","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-07-19 14:08:47.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/8/check/DellEMC_PowerStore_Manila_dhss_false/a89df4d/DellEMC_PowerStore_Manila_dhss_false/359 : SUCCESS in 36m 43s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":8},{"id":"e28d4c99c094362c28f83d7c642effdb8b5b58cf","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-07-19 19:17:30.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/921313/8/check/DellEMC_Unity_Manila_dhss_true/7f69ce7/DellEMC_Unity_Manila_dhss_true/1431 : SUCCESS in 2h 36m 49s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/8/check/DellEMC_Unity_Manila_dhss_false/d54998b/DellEMC_Unity_Manila_dhss_false/1533 : SUCCESS in 59m 16s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":8},{"id":"8b0c1343d702c4d351bd172526f67e91a9d7bddb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-07-22 11:42:12.000000000","message":"Patch Set 8:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/8/check/DellEMC_PowerFlex_Manila_dhss_false/af8772a/DellEMC_PowerFlex_Manila_dhss_false/1300 : SUCCESS in 34m 12s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"aca934ddb23d16f8997b12bd17de7d8aafb62b73","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2024-07-23 02:54:01.000000000","message":"Patch Set 8: Code-Review+2\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"0a5c534c5ec44f73823fb88b3ecbba82d7d0cd63","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2024-08-14 07:44:30.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"cd3e588a286821237358da3377e1ffc0e5869b3c","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-08-14 13:23:56.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4434995b48811697869258f036caa7ca39732531","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-08-15 04:38:31.000000000","message":"Patch Set 8: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":8},{"id":"4ab7b726197e37dda133e4a84fe0e4cafaa093be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-08-19 11:52:58.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"a3bad59d9ac578463a13d81da73532fadf96ec0d","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-08-19 11:53:00.000000000","message":"Patch Set 8:\n\n(7 comments)","accounts_in_message":[],"_revision_number":8},{"id":"cf032ccec3da8b8d6fcf319b4f771fd31e65ca0a","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-19 12:05:53.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/921313/9/check/DellEMC_Unity_Manila_dhss_true/faf5cf5/DellEMC_Unity_Manila_dhss_true/1550 : FAILURE in 12m 33s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/9/check/DellEMC_Unity_Manila_dhss_false/b29348b/DellEMC_Unity_Manila_dhss_false/1661 : FAILURE in 12m 29s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"6718938ecf0df61dd64e9bbc795b25209a80aacf","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-19 12:24:42.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/9/check/DellEMC_PowerFlex_Manila_dhss_false/9506af6/DellEMC_PowerFlex_Manila_dhss_false/1483 : SUCCESS in 31m 28s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"7477382924e6d2cdf658fcd9c16b0b61a2db1a73","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-19 12:29:11.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/59cd0960c9414621814e110ef3cf0d37\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/7c909c8411284ebb8a3909af8bf25d40 : FAILURE in 31m 15s","accounts_in_message":[],"_revision_number":9},{"id":"182d75f07b6f97c7f5acb3fe44563ae8a105fe9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 13:48:02.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b1ed6446cce42bc963988905b3f8221\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e28d38a493e471e953adbf1767dd769 : SUCCESS in 10m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28a2230dd05d4ec890b11d17437fa7c8 : FAILURE in 24m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4507c90e4ec43d8a3997040e810a597 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37629c47718041bfb6a39cf28f2d6f49 : SUCCESS in 9m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fbd32996069747cb9471edfdcc84f8fb : SUCCESS in 7m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/51bb4b5353e94a2f8505ae1610d714b2 : SUCCESS in 10m 45s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4cf0fcd7c80e482f98c4b9af1fd9b0da : SUCCESS in 9m 36s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/ef69327f60c34b2daf0061f09639707d : SUCCESS in 1h 07m 05s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/91f279d006dd4cf99fc3677f3afd7385 : SUCCESS in 1h 54m 18s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/b2b54545afef46eb9a57ae1d6ff10a90 : FAILURE in 1h 05m 24s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ce59737bc71d4b25a5dcf12bb52996da : SUCCESS in 1h 03m 42s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/2e0792e8432840a2a584c04d4841aec9 : SUCCESS in 57m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/35708650fc5244e0a98be3de583d2a25 : SUCCESS in 1h 02m 02s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/580f1cec1c6e45aba0b2d1bc195f71c2 : SUCCESS in 5m 43s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/9fbdeb19077e40bdae943ce24790a1be : SUCCESS in 5m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/174355565829425d99c2568c83fe8657 : SUCCESS in 26m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/3841893874e746eb8975f2edc5911168 : FAILURE in 30m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e8ad5a1c5270407190550daa3b614676 : FAILURE in 39m 44s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/540ee3b3d8d54049ab876bccaf47ff0f : SUCCESS in 57m 31s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ec7801cca05d41738ba4ab69081c61a8 : SUCCESS in 57m 43s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c6c6e5819304471788b3321a6c2df3a5 : SUCCESS in 29m 17s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f5b635464b6b4b338334003c35d0e103 : SUCCESS in 29m 43s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/94ab1d5059cc4fad9c81a634e9a8ed10 : FAILURE in 4m 29s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/4ad9dfd420204648b44b12fdd6368f3c : SUCCESS in 5m 18s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"ae3ee941b7783b5bbd840b81db8ba5ea22ae6130","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-08-19 16:23:28.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1f3b0abdff2fb82d31ef00a42897abc71b3bb753","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-19 16:56:26.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/559944db08974d6d8e3ae4afb14a4c67\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/1142cda600504f04b734759c66614066 : FAILURE in 27m 59s","accounts_in_message":[],"_revision_number":9},{"id":"ab7a0b7da87f3d6bab2b64e882efa1bab910d571","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 18:40:34.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b469748a75c7448391ad0643c63dd73b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/107563d1f54841439d1b59f69d1eef05 : SUCCESS in 10m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/042480b306634eb4916f774e70a552e3 : FAILURE in 27m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cf9a99681404978b11024c2a8737392 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcce4946b88f4c66a892f717f53d1a82 : SUCCESS in 18m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af837c60cf74496d9d4f42756ebdeb44 : SUCCESS in 10m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b2c549be4a24439f9e437d44e91e140f : SUCCESS in 8m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ce93dcd90c6403f9df2daebc8ca4dd2 : SUCCESS in 9m 21s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/18ec3f0e2b32489ab4e6f1cf1344966e : SUCCESS in 1h 14m 46s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/50c184b537ba4b5cafc5f5c7cacd6e89 : SUCCESS in 2h 16m 22s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/6fe33d168f0744359d8ca66e86a4cfd0 : FAILURE in 51m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/391faeb4bfed4cd2bd600e2fcd92c1cd : SUCCESS in 46m 14s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/a75eac31d5ed43ef9525806681ed64d4 : SUCCESS in 57m 10s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/cb89ac5dfb31491ab57bf03fcf69ac0f : FAILURE in 1h 20m 07s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d7a62a4a0ad04ea68568cecb2ef578e3 : SUCCESS in 4m 33s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/26443534d5fc4bb3a3e710c02e970c24 : SUCCESS in 5m 16s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4f0cff2387a344038de7be8db692d941 : SUCCESS in 39m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/85e4b3eb1428497bbd0e915278e2ee50 : SUCCESS in 42m 53s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ca646c24937c4282b59f0afdbaa823e2 : FAILURE in 55m 38s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/19d265743e3b49baa81a373c2551a89f : SUCCESS in 55m 22s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/efd9b7589f5042619ce3c6975b5f530e : SUCCESS in 47m 23s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/2d89776ca30e4443af12f15afb557c05 : SUCCESS in 22m 07s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/2351fa010d294ddf9edc817bcfe0ac09 : SUCCESS in 33m 19s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/c6afe1722e7b4c908d54af0845266ed1 : FAILURE in 4m 05s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/b3dfbd18c2a94372b614e3a56c1ec25b : SUCCESS in 6m 17s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"3f9ae690f2c49345d6b5ec7a1bb1889f6bfcca31","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-19 18:53:33.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/921313/9/upstream-check/manila-tempest-plugin-ontap-dhss/32c328b/ : RETRY_LIMIT in 4m 11s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/921313/9/upstream-check/manila-tempest-plugin-ontap-no-dhss/8d89c10/ : RETRY_LIMIT in 11m 54s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"ca3264e6cb83c502572835653a8961d0bfcf9e7b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-19 18:57:06.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/921313/9/check/DellEMC_Unity_Manila_dhss_true/987789f/DellEMC_Unity_Manila_dhss_true/1557 : FAILURE in 9m 51s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/9/check/DellEMC_Unity_Manila_dhss_false/cf246d4/DellEMC_Unity_Manila_dhss_false/1668 : FAILURE in 10m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"0512fee2dbabdee116a0ffcfdcebe068b66dc197","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-19 19:15:51.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/9/check/DellEMC_PowerStore_Manila_dhss_false/8d1185a/DellEMC_PowerStore_Manila_dhss_false/427 : SUCCESS in 35m 03s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"6d84933e8e4cf580c5282bec1b168c5dd12bc400","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-19 19:40:10.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/9/check/DellEMC_PowerFlex_Manila_dhss_false/1ae8131/DellEMC_PowerFlex_Manila_dhss_false/1489 : SUCCESS in 32m 37s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"171290066bad0e9f00a922e08f0cd1f4d89f2acd","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-08-29 20:35:33.000000000","message":"Patch Set 9: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"eb5df09d8089576c0ab05f2404b9e42278577e20","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-08-30 05:46:40.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"21699bdf8a18fae61d1996cc2428946351553815","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-08-30 06:55:27.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"abaf47978494878e583ff3a40dc3b1cb64c80563","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-08-30 07:13:47.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c7fd5ed127d5096ccd6dbeabce90773995c57960","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-09-18 21:01:08.000000000","message":"Patch Set 9: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"f989ea3c38db1f68dbf7728ef496790b25b0873d","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-09-19 11:20:55.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a41c2287e3ae316dae4eae1b91bdfe4a7242123f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-09-19 11:20:55.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"f60e79457ce019f4006e8e3dcf53e02756900e1b","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-19 11:45:37.000000000","message":"Patch Set 10:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9e37483a28f34e4b82b9f0dfd0ba399c\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/b34d9c9fa87b4d6dbf083cb3713257c3 : FAILURE in 24m 12s","accounts_in_message":[],"_revision_number":10},{"id":"32d752cdfff196c6e319bd9f30fd726b735829ad","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-19 11:54:22.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/10/check/DellEMC_PowerFlex_Manila_dhss_false/ac1b244/DellEMC_PowerFlex_Manila_dhss_false/1717 : SUCCESS in 33m 09s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"a155cf5f64390ecb067d07bde82324d45d5f1316","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-19 12:51:09.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/921313/10/check/DellEMC_Unity_Manila_dhss_true/0c7f4af/DellEMC_Unity_Manila_dhss_true/1706 : FAILURE in 14m 48s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/10/check/DellEMC_Unity_Manila_dhss_false/09b902b/DellEMC_Unity_Manila_dhss_false/1813 : SUCCESS in 1h 29m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":10},{"id":"ec97b7859e6d3ffdfc279504e44fdd1ec48fb32f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 13:49:50.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef3669e506d74dfdacef94711eebbeaf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b4624c3db664ebea8a1d70066b90866 : SUCCESS in 10m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e249f2a0b8c434cb512b572d159e36d : FAILURE in 26m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dde6cff387f943578a45afb4c274aa2b : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1cffe66757a84629bcffd38edf9c806b : SUCCESS in 9m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb5a40f036bc45ad980e736e7b192688 : SUCCESS in 8m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ee42bb925a714c1dae59e718f2c9cf65 : SUCCESS in 10m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b3db9b96317442b9661b84d5d9ac90d : SUCCESS in 9m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4081d53b6d964fbeb4d6318ead78812e : SUCCESS in 1h 13m 38s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/29c3b913cf59406d87521f31c911b03b : SUCCESS in 2h 22m 39s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/fba6060fd38b40d9820ccc5808fe2466 : FAILURE in 1h 05m 09s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/29057643503049a6a3e67c3d302cc5e9 : SUCCESS in 1h 08m 30s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d2a4e1eac0f846649f9d7c94a5d512ed : SUCCESS in 1h 12m 07s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/637ea6892ee744209eed8b590c85c647 : SUCCESS in 1h 17m 55s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/fa20cc3b4f544154a5e09fac7ce0601c : SUCCESS in 4m 25s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ccc0340f436a4fab9526d7c5eedad66b : SUCCESS in 4m 44s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c5fe6234c6244ae0a8cbbdcb187e9549 : SUCCESS in 34m 23s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/d9213c88b67b4b86b0f2bfd3120cfc16 : SUCCESS in 32m 19s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/8766ef220e7b4775acf069d12a7c361c : FAILURE in 40m 14s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/9fba9a49d82c429c8e743be8ad4db0f5 : SUCCESS in 1h 21m 15s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/09b77d84414b4615a98e12e1048a65f8 : SUCCESS in 1h 01m 25s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/288a0054853540278e73e8986312fe23 : SUCCESS in 23m 18s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/eae34bec760f48f2bfab2e85cda7e9d9 : SUCCESS in 40m 07s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/676655f2183c4e78b13ae3f8f623bad4 : FAILURE in 4m 01s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/f643beefb4424c1e9a6c1608c73a6795 : SUCCESS in 6m 37s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0bc6a8d8db76927e8ea9b80d5351ff0a9ae89955","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-19 14:03:47.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"84b56510db95a786e65b4ab9945ffc8606c79165","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-19 14:05:38.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/921313/10/upstream-check/manila-tempest-plugin-ontap-dhss/1b0cab9/ : RETRY_LIMIT in 13m 50s\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/921313/10/upstream-check/manila-tempest-plugin-ontap-no-dhss/a97f13b/ : RETRY_LIMIT in 15m 04s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":10},{"id":"bbd89a180b7917a95d24b2cf4f467daee199929d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-19 14:23:55.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/10/check/DellEMC_PowerFlex_Manila_dhss_false/f66da7e/DellEMC_PowerFlex_Manila_dhss_false/1718 : SUCCESS in 33m 40s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"4d919cb122bc7491960d6e223ce3f23e38fac3b9","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-19 14:27:06.000000000","message":"Patch Set 10:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/10/check/DellEMC_PowerStore_Manila_dhss_false/15bb60e/DellEMC_PowerStore_Manila_dhss_false/505 : SUCCESS in 36m 57s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":10},{"id":"b11dddafc994fc3125b5262a404bb848fc265670","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-09-19 15:33:14.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"9d374399c6450d508d502ec87939e5fdd964a56c","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-09-19 15:33:23.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4098aac2f1427fb5ec2ab0cfd11595de891886fe","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-19 16:04:33.000000000","message":"Patch Set 11:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/dbcdb29f5b964d0faa89ccab721cbe29\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/ad60f57ffad34760acfb1b4358acbd31 : FAILURE in 30m 48s","accounts_in_message":[],"_revision_number":11},{"id":"6d1725088e2488edf1bf7af862c558e18bbbc9f0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-19 16:06:12.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/11/check/DellEMC_PowerFlex_Manila_dhss_false/293621c/DellEMC_PowerFlex_Manila_dhss_false/1719 : SUCCESS in 32m 42s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"48d081b13b51654d2d0a1754d95fd0a40101d11b","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-19 16:51:24.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/921313/11/check/DellEMC_Unity_Manila_dhss_true/30ffeca/DellEMC_Unity_Manila_dhss_true/1708 : FAILURE in 13m 42s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/11/check/DellEMC_Unity_Manila_dhss_false/60ef1ae/DellEMC_Unity_Manila_dhss_false/1815 : SUCCESS in 1h 17m 53s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"65cd2c306e7cf0166db0e7ea18df33a03e03803c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 17:32:23.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8590c139653f4a94985eaf6bcb018430\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0b07f502a16424098053e0c1894c375 : SUCCESS in 6m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/907d52efa1dd401195a47d63906eee03 : FAILURE in 27m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9460ccbd965742ce903c13cbbf4bffba : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/294b3e8e27624f7b84842db622fe70a4 : SUCCESS in 9m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5a24c1ecc444713965c5f8e81184149 : SUCCESS in 7m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4b26028987c1459d89bea4fddd55387a : SUCCESS in 13m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f26f1a4dbce4a94b54a4e2d0922ebfb : SUCCESS in 9m 07s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/cc6abd85b97d4c00a6e3824511ce302e : SUCCESS in 1h 13m 31s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/92d9fa4dbe4f477c8ae052a536cb0baa : SUCCESS in 1h 56m 34s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3a921555260b4a108a6e35458d32acf0 : FAILURE in 1h 21m 36s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/2e4a67acb65147eeb3dce2ca39101630 : SUCCESS in 1h 02m 14s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f67a107ef0ca41c2a503d2caee09a14e : SUCCESS in 1h 14m 21s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/c6660f0d6c5b4aa4a82da2de64e4dafd : SUCCESS in 41m 48s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/d8340a474bd84ac39f014cd6ba1aecfa : SUCCESS in 5m 08s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/257cbe29804c4a47a7ba5a7b33fdd424 : SUCCESS in 6m 37s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0f27326376024eeebac16f247cd7dbf4 : SUCCESS in 32m 26s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f33c19d86fd240bc8d14df6716e6656c : SUCCESS in 43m 40s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/54ed373e8c714860bfb793d5097807ea : FAILURE in 42m 40s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/dadc19c3906f46ccbc618d89f958588f : SUCCESS in 1h 14m 12s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/96340912669f4f4c82a17a114ec17a5f : SUCCESS in 53m 34s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/e5c241e409ee4b3881fe4635cd04a90e : SUCCESS in 28m 39s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/e6225ca4c7b6464e9339731755c4195b : SUCCESS in 28m 06s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/4686918d82ce431ba90eee5312c6f2aa : FAILURE in 4m 43s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/427a4bca1e6f4e2786fa724bab728af4 : SUCCESS in 7m 09s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ac5212a9c84abadcbd15d956b854554ca45ed2f0","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-19 17:50:48.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/13/921313/11/upstream-check/manila-tempest-plugin-ontap-dhss/6c01a06/ : RETRY_LIMIT in 8m 04s\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"b523ee950d9ba876928eeb4f33b36005e8641916","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-19 18:05:18.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/11/check/DellEMC_PowerFlex_Manila_dhss_false/9d47516/DellEMC_PowerFlex_Manila_dhss_false/1720 : SUCCESS in 32m 36s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"774dc21b9d5057dba968eac339483da2e44a9ca6","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-19 18:07:27.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/11/check/DellEMC_PowerStore_Manila_dhss_false/f033000/DellEMC_PowerStore_Manila_dhss_false/506 : SUCCESS in 34m 46s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":11},{"id":"36dd6ba94b58dd3d5e1792109fb9d61aa755662c","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-09-19 18:11:37.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"32c5021dd23fed393e23ef180f513a5d55be9707","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-19 18:24:17.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"f491727e5f4b97ce0e9f4d4c00a013e01d0477e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 18:24:53.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"844f10cfcf73f341f3a7cae11f299552ac26029e","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-19 18:49:16.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/13/921313/11/check/DellEMC_Unity_Manila_dhss_true/b2ed560/DellEMC_Unity_Manila_dhss_true/1709 : FAILURE in 13m 48s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/13/921313/11/check/DellEMC_Unity_Manila_dhss_false/f8fc6a5/DellEMC_Unity_Manila_dhss_false/1816 : SUCCESS in 1h 16m 34s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"7948bc2c0c274244926defb22f0d8b34c16605f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 19:52:39.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29f5b0aa3fb94c1bbb0f771472473e54\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5738989101b3403ca696d56750b2eb37 : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9402d8ed8a041db85c10e5273700598 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7993f12f8bf14eca8f9be529bd0dbfdb : SUCCESS in 14m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f873fca5e2f24ab79dbf75efee5a15e6 : SUCCESS in 8m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7449f1b8c44c4748afe1e710a2e84932 : SUCCESS in 9m 56s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/98ea1861e0c74415b14cffde63e70174 : SUCCESS in 1h 16m 56s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/76aa3c0332934485901b36c6c19729fd : SUCCESS in 32m 24s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/4bcf18662cf44d30a49b0b951ee3d036 : SUCCESS in 31m 56s","accounts_in_message":[],"_revision_number":11},{"id":"a50009518ffd75b481aa05f3e1ed5250a0cb6f92","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 19:52:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"a40c288b24ab9ba9ad1979c98ddc42565730870b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 19:53:52.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b97ae9e544aa439a97404dcb2ff0c9ac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f73985aa0914c98ac0e9fff2998224e : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43f6c6cd9f0147be93f0c0224b36ca21 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"e8f086d0f58608a2ff04958af668538e0ed5f7be","revisions":{"b1f04854e6637a2c26d03a29871c4fb0534eed06":{"kind":"REWORK","_number":1,"created":"2024-06-04 17:05:33.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/13/921313/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/921313/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/921313/1"}}},"commit":{"parents":[{"commit":"76d032969848fa409f3b7da499cc2da3de67f6be","subject":"Fix leak of ports on share server deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/76d032969848fa409f3b7da499cc2da3de67f6be"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-06-04 16:55:39.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-06-04 16:55:39.000000000","tz":0},"subject":"Fix share instance deferred deletion race","message":"Fix share instance deferred deletion race\n\n- Make sure snapshot instance of share instance should be deleted before\n share instance deletion.\n- count_share_group_snapshot_members_in_share() should not return share\n instances in deferred deletion states as those shares are hidden from\n end-user once deferred deleted.\n\npartially-implements: bp/deferred-deletion\nCloses-bug: #2068043\nChange-Id: Ie827619d3952d1db69559206b5dfd47efb4cf7b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b1f04854e6637a2c26d03a29871c4fb0534eed06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/b1f04854e6637a2c26d03a29871c4fb0534eed06"}]},"branch":"refs/heads/master"},"cb844626bb00fbaf7e9c65543b61123c3a90c579":{"kind":"REWORK","_number":2,"created":"2024-06-07 08:42:31.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/13/921313/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/921313/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/921313/2"}}},"commit":{"parents":[{"commit":"76d032969848fa409f3b7da499cc2da3de67f6be","subject":"Fix leak of ports on share server deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/76d032969848fa409f3b7da499cc2da3de67f6be"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-06-04 16:55:39.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-06-07 08:30:42.000000000","tz":330},"subject":"Update share/snapshot instance deferred deletion","message":"Update share/snapshot instance deferred deletion\n\n- Deferred deletion would be regular deletion where quota is freed\n  before resource deletion by driver and deletion states are hidden\n  from end-user. Also, everything will be handled by Manila or admin\n  if any error occurs in deletion.\n- Any error during deferred deletion will put share/snapshot in\n  error_deferrer_deleting state and periodic task will handle it.\n- count_share_group_snapshot_members_in_share() should not return share\n  instances in deferred deletion states as those shares are hidden from\n  end-user once deferred deleted. Only actual deletion (by driver) will\n  make sure snapshot instance of share instance should be deleted before\n  share instance deletion.\n\npartially-implements: bp/deferred-deletion\nCloses-bug: #2068043\nChange-Id: Ie827619d3952d1db69559206b5dfd47efb4cf7b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cb844626bb00fbaf7e9c65543b61123c3a90c579"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/cb844626bb00fbaf7e9c65543b61123c3a90c579"}]},"branch":"refs/heads/master"},"9adb97f32feb8cfd4a432d747d3f680132743818":{"kind":"REWORK","_number":3,"created":"2024-06-07 11:10:21.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/13/921313/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/921313/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/921313/3"}}},"commit":{"parents":[{"commit":"76d032969848fa409f3b7da499cc2da3de67f6be","subject":"Fix leak of ports on share server deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/76d032969848fa409f3b7da499cc2da3de67f6be"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-06-04 16:55:39.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-06-07 11:02:27.000000000","tz":0},"subject":"Update share/snapshot instance deferred deletion","message":"Update share/snapshot instance deferred deletion\n\n- Deferred deletion would be regular deletion where quota is freed\n  before resource deletion by driver and deletion states are hidden\n  from end-user. Also, everything will be handled by Manila or admin\n  if any error occurs in deletion.\n- Any error during deferred deletion will put share/snapshot in\n  error_deferred_deleting state and periodic task will handle it.\n- count_share_group_snapshot_members_in_share() should not return share\n  instances in deferred deletion states as those shares are hidden from\n  end-user once deferred deleted. Only actual deletion (by driver) will\n  make sure snapshot instance of share instance should be deleted before\n  share instance deletion.\n\npartially-implements: bp/deferred-deletion\nCloses-bug: #2068043\nChange-Id: Ie827619d3952d1db69559206b5dfd47efb4cf7b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9adb97f32feb8cfd4a432d747d3f680132743818"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/9adb97f32feb8cfd4a432d747d3f680132743818"}]},"branch":"refs/heads/master"},"fa3f3e801ac474f8fe1903295250cc92fc6f1823":{"kind":"REWORK","_number":4,"created":"2024-06-12 07:28:45.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/13/921313/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/921313/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/921313/4"}}},"commit":{"parents":[{"commit":"d773353502cd8932c31581cfbc3b80846067fe68","subject":"Merge \"Fix backup creation errors with NetApp driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d773353502cd8932c31581cfbc3b80846067fe68"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-06-04 16:55:39.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-06-12 07:28:19.000000000","tz":330},"subject":"Update share/snapshot instance deferred deletion","message":"Update share/snapshot instance deferred deletion\n\n- Deferred deletion would be regular deletion where quota is freed\n  before resource deletion by driver and deletion states are hidden\n  from end-user. Also, everything will be handled by Manila or admin\n  if any error occurs in deletion.\n- Any error during deferred deletion will put share/snapshot in\n  error_deferrer_deleting state and periodic task will handle it.\n- count_share_group_snapshot_members_in_share() should not return share\n  instances in deferred deletion states as those shares are hidden from\n  end-user once deferred deleted. Only actual deletion (by driver) will\n  make sure snapshot instance of share instance should be deleted before\n  share instance deletion.\n\npartially-implements: bp/deferred-deletion\nCloses-bug: #2068043\nChange-Id: Ie827619d3952d1db69559206b5dfd47efb4cf7b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fa3f3e801ac474f8fe1903295250cc92fc6f1823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fa3f3e801ac474f8fe1903295250cc92fc6f1823"}]},"branch":"refs/heads/master"},"1ae65f88c1f2c62ae3b9e8307bbf377125423881":{"kind":"REWORK","_number":5,"created":"2024-06-21 13:42:44.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/13/921313/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/921313/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/921313/5"}}},"commit":{"parents":[{"commit":"d773353502cd8932c31581cfbc3b80846067fe68","subject":"Merge \"Fix backup creation errors with NetApp driver\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d773353502cd8932c31581cfbc3b80846067fe68"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-06-04 16:55:39.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-06-21 13:42:37.000000000","tz":0},"subject":"Update share/snapshot instance deferred deletion","message":"Update share/snapshot instance deferred deletion\n\n- Deferred deletion would be regular deletion where quota is freed\n  before resource deletion by driver and deletion states are hidden\n  from end-user. Also, everything will be handled by Manila or admin\n  if any error occurs in deletion.\n- Any error during deferred deletion will put share/snapshot in\n  error_deferrer_deleting state and periodic task will handle it.\n- count_share_group_snapshot_members_in_share() should not return share\n  instances in deferred deletion states as those shares are hidden from\n  end-user once deferred deleted. Only actual deletion (by driver) will\n  make sure snapshot instance of share instance should be deleted before\n  share instance deletion.\n\npartially-implements: bp/deferred-deletion\nCloses-bug: #2068043\nChange-Id: Ie827619d3952d1db69559206b5dfd47efb4cf7b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1ae65f88c1f2c62ae3b9e8307bbf377125423881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/1ae65f88c1f2c62ae3b9e8307bbf377125423881"}]},"branch":"refs/heads/master"},"2f88c6722efd8522c35f97c030dfa7f501112ca1":{"kind":"REWORK","_number":6,"created":"2024-06-24 08:35:50.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/13/921313/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/921313/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/921313/6"}}},"commit":{"parents":[{"commit":"0999857c47fca78b0d0a2f8059cd5aa1527cb422","subject":"Merge \"Bump pylint version and fix pylint issues\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0999857c47fca78b0d0a2f8059cd5aa1527cb422"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-06-04 16:55:39.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-06-24 08:34:03.000000000","tz":0},"subject":"Update share/snapshot instance deferred deletion","message":"Update share/snapshot instance deferred deletion\n\n- Deferred deletion would be regular deletion where quota is freed\n  before resource deletion by driver and deletion states are hidden\n  from end-user. Also, everything will be handled by Manila or admin\n  if any error occurs in deletion.\n- Any error during deferred deletion will put share/snapshot in\n  error_deferrer_deleting state and periodic task will handle it.\n- count_share_group_snapshot_members_in_share() should not return share\n  instances in deferred deletion states as those shares are hidden from\n  end-user once deferred deleted. Only actual deletion (by driver) will\n  make sure snapshot instance of share instance should be deleted before\n  share instance deletion.\n\npartially-implements: bp/deferred-deletion\nCloses-bug: #2068043\nChange-Id: Ie827619d3952d1db69559206b5dfd47efb4cf7b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2f88c6722efd8522c35f97c030dfa7f501112ca1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2f88c6722efd8522c35f97c030dfa7f501112ca1"}]},"branch":"refs/heads/master"},"10b885a093572a9a4a26d13bdecc6a9cbac38c14":{"kind":"REWORK","_number":7,"created":"2024-06-25 10:05:44.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/13/921313/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/921313/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/921313/7"}}},"commit":{"parents":[{"commit":"0999857c47fca78b0d0a2f8059cd5aa1527cb422","subject":"Merge \"Bump pylint version and fix pylint issues\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0999857c47fca78b0d0a2f8059cd5aa1527cb422"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-06-04 16:55:39.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-06-25 10:05:34.000000000","tz":0},"subject":"Update share/snapshot instance deferred deletion","message":"Update share/snapshot instance deferred deletion\n\n- Deferred deletion would be regular deletion where quota is freed\n  before resource deletion by driver and deletion states are hidden\n  from end-user. Also, everything will be handled by Manila or admin\n  if any error occurs in deletion.\n- Any error during deferred deletion will put share/snapshot in\n  error_deferrer_deleting state and periodic task will handle it.\n- count_share_group_snapshot_members_in_share() should not return share\n  instances in deferred deletion states as those shares are hidden from\n  end-user once deferred deleted. Only actual deletion (by driver) will\n  make sure snapshot instance of share instance should be deleted before\n  share instance deletion.\n\npartially-implements: bp/deferred-deletion\nCloses-bug: #2068043\nChange-Id: Ie827619d3952d1db69559206b5dfd47efb4cf7b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/10b885a093572a9a4a26d13bdecc6a9cbac38c14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/10b885a093572a9a4a26d13bdecc6a9cbac38c14"}]},"branch":"refs/heads/master"},"7036a5d631aa7b14397f4989e4b3728740d805e7":{"kind":"REWORK","_number":8,"created":"2024-07-19 11:35:21.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/13/921313/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/921313/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/921313/8"}}},"commit":{"parents":[{"commit":"4ba31e8c6719813864d825820b643576611dd972","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4ba31e8c6719813864d825820b643576611dd972"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-06-04 16:55:39.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-07-19 11:35:14.000000000","tz":0},"subject":"Update share/snapshot instance deferred deletion","message":"Update share/snapshot instance deferred deletion\n\n- Deferred deletion would be regular deletion where quota is freed\n  before resource deletion by driver and deletion states are hidden\n  from end-user. Also, everything will be handled by Manila or admin\n  if any error occurs in deletion.\n- Any error during deferred deletion will put share/snapshot in\n  error_deferrer_deleting state and periodic task will handle it.\n- count_share_group_snapshot_members_in_share() should not return share\n  instances in deferred deletion states as those shares are hidden from\n  end-user once deferred deleted. Only actual deletion (by driver) will\n  make sure snapshot instance of share instance should be deleted before\n  share instance deletion.\n\npartially-implements: bp/deferred-deletion\nCloses-bug: #2068043\nChange-Id: Ie827619d3952d1db69559206b5dfd47efb4cf7b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7036a5d631aa7b14397f4989e4b3728740d805e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7036a5d631aa7b14397f4989e4b3728740d805e7"}]},"branch":"refs/heads/master"},"19b87b124b54e819e84dbf4980ddf92e2ad6da4d":{"kind":"REWORK","_number":9,"created":"2024-08-19 11:52:58.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/13/921313/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/921313/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/921313/9"}}},"commit":{"parents":[{"commit":"3920cd46e40d4c2a449ae9247a2e228bb2b994c5","subject":"Merge \"Pass share metadata updates to backend drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/3920cd46e40d4c2a449ae9247a2e228bb2b994c5"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-06-04 16:55:39.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-08-19 11:52:50.000000000","tz":0},"subject":"Update share/snapshot instance deferred deletion","message":"Update share/snapshot instance deferred deletion\n\n- Deferred deletion would be regular deletion where quota is freed\n  before resource deletion by driver and deletion states are hidden\n  from end-user. Also, everything will be handled by Manila or admin\n  if any error occurs in deletion.\n- Any error during deferred deletion will put share/snapshot in\n  error_deferrer_deleting state and periodic task will handle it.\n- count_share_group_snapshot_members_in_share() should not return share\n  instances in deferred deletion states as those shares are hidden from\n  end-user once deferred deleted. Only actual deletion (by driver) will\n  make sure snapshot instance of share instance should be deleted before\n  share instance deletion.\n\npartially-implements: bp/deferred-deletion\nCloses-bug: #2068043\nChange-Id: Ie827619d3952d1db69559206b5dfd47efb4cf7b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/19b87b124b54e819e84dbf4980ddf92e2ad6da4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/19b87b124b54e819e84dbf4980ddf92e2ad6da4d"}]},"branch":"refs/heads/master"},"e3706cc4372b4f19df320d396a5731534e1c6a80":{"kind":"REWORK","_number":10,"created":"2024-09-19 11:20:55.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/13/921313/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/921313/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/921313/10"}}},"commit":{"parents":[{"commit":"03c0d21ca1cf2b4a76b8fb078a1d1b369b897251","subject":"Merge \"Fix export location metadata updates by drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/03c0d21ca1cf2b4a76b8fb078a1d1b369b897251"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-06-04 16:55:39.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-09-19 11:20:26.000000000","tz":0},"subject":"Update share/snapshot instance deferred deletion","message":"Update share/snapshot instance deferred deletion\n\n- Deferred deletion would be regular deletion where quota is freed\n  before resource deletion by driver and deletion states are hidden\n  from end-user. Also, everything will be handled by Manila or admin\n  if any error occurs in deletion.\n- Any error during deferred deletion will put share/snapshot in\n  error_deferrer_deleting state and periodic task will handle it.\n- count_share_group_snapshot_members_in_share() should not return share\n  instances in deferred deletion states as those shares are hidden from\n  end-user once deferred deleted. Only actual deletion (by driver) will\n  make sure snapshot instance of share instance should be deleted before\n  share instance deletion.\n\npartially-implements: bp/deferred-deletion\nCloses-bug: #2068043\nChange-Id: Ie827619d3952d1db69559206b5dfd47efb4cf7b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e3706cc4372b4f19df320d396a5731534e1c6a80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e3706cc4372b4f19df320d396a5731534e1c6a80"}]},"branch":"refs/heads/master"},"e8f086d0f58608a2ff04958af668538e0ed5f7be":{"kind":"REWORK","_number":11,"created":"2024-09-19 15:33:14.000000000","uploader":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"ref":"refs/changes/13/921313/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/13/921313/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/13/921313/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/13/921313/11"}}},"commit":{"parents":[{"commit":"697c9bf3d34b4eb291158c99ae859da10a0c2c59","subject":"Remove extra policy check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/697c9bf3d34b4eb291158c99ae859da10a0c2c59"}]}],"author":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-06-04 16:55:39.000000000","tz":0},"committer":{"name":"Kiran Pawar","email":"kinpaa@gmail.com","date":"2024-09-19 15:33:06.000000000","tz":0},"subject":"Update share/snapshot instance deferred deletion","message":"Update share/snapshot instance deferred deletion\n\n- Deferred deletion would be regular deletion where quota is freed\n  before resource deletion by driver and deletion states are hidden\n  from end-user. Also, everything will be handled by Manila or admin\n  if any error occurs in deletion.\n- Any error during deferred deletion will put share/snapshot in\n  error_deferrer_deleting state and periodic task will handle it.\n- count_share_group_snapshot_members_in_share() should not return share\n  instances in deferred deletion states as those shares are hidden from\n  end-user once deferred deleted. Only actual deletion (by driver) will\n  make sure snapshot instance of share instance should be deleted before\n  share instance deletion.\n\npartially-implements: bp/deferred-deletion\nCloses-bug: #2068043\nChange-Id: Ie827619d3952d1db69559206b5dfd47efb4cf7b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e8f086d0f58608a2ff04958af668538e0ed5f7be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/e8f086d0f58608a2ff04958af668538e0ed5f7be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
