)]}'
{"id":"openstack%2Fmanila~926049","triplet_id":"openstack%2Fmanila~master~I73c4d027852fbfbbdab59538cbe23ba1487da0de","project":"openstack/manila","branch":"master","topic":"bp/snaplock-support-for-manila-driver","attention_set":{},"removed_from_attention_set":{"29632":{"account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"last_update":"2024-09-05 19:02:04.000000000","reason":"\u003cGERRIT_ACCOUNT_29632\u003e replied on the change","reason_account":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}},"36178":{"account":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"last_update":"2024-09-05 18:32:00.000000000","reason":"\u003cGERRIT_ACCOUNT_36178\u003e replied on the change","reason_account":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"}},"36179":{"account":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"last_update":"2024-09-05 20:07:48.000000000","reason":"Change was submitted"},"16643":{"account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"last_update":"2024-09-05 18:59:35.000000000","reason":"\u003cGERRIT_ACCOUNT_16643\u003e replied on the change","reason_account":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},"36180":{"account":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"last_update":"2024-09-05 20:07:48.000000000","reason":"Change was submitted"},"32919":{"account":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"last_update":"2024-09-05 20:07:48.000000000","reason":"Change was submitted"},"30025":{"account":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"last_update":"2024-09-05 20:07:48.000000000","reason":"Change was submitted"},"36171":{"account":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"last_update":"2024-09-05 20:07:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I73c4d027852fbfbbdab59538cbe23ba1487da0de","subject":"[NetApp] SnapLock support for NetApp driver","status":"MERGED","created":"2024-08-09 10:36:19.000000000","updated":"2024-09-09 13:23:39.000000000","submitted":"2024-09-05 20:07:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":106,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"926049-bp/snaplock-support-for-manila-driver","meta_rev_id":"ab44a37d099995eb0caaeabc8d5333054b83f80f","_number":926049,"virtual_id_number":926049,"owner":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-05 20:07:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":2,"date":"2024-09-05 19:02:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},{"value":2,"date":"2024-09-05 18:59:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":1,"date":"2024-09-05 18:29:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"all":[{"value":1,"date":"2024-09-05 19:02:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"value":0,"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},{"value":0,"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"value":0,"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"value":0,"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"}],"reviewers":{"REVIEWER":[{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"}],"CC":[{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-09 11:35:46.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-09 12:24:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-09 16:24:06.000000000","updated_by":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"reviewer":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"state":"CC"},{"updated":"2024-08-13 12:38:00.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"state":"REVIEWER"},{"updated":"2024-08-13 12:38:00.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"state":"REVIEWER"},{"updated":"2024-08-13 12:38:00.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":36179,"name":"Saikumar Pulluri","display_name":"Saikumar Pulluri","email":"saikumar1016@gmail.com","username":"pulluri"},"state":"REVIEWER"},{"updated":"2024-08-13 12:38:00.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"state":"REVIEWER"},{"updated":"2024-08-13 12:38:00.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"state":"REVIEWER"},{"updated":"2024-08-13 12:38:00.000000000","updated_by":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"reviewer":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"state":"REVIEWER"},{"updated":"2024-08-13 14:50:57.000000000","updated_by":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"reviewer":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"state":"CC"},{"updated":"2024-08-13 14:59:12.000000000","updated_by":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"reviewer":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"state":"CC"},{"updated":"2024-08-16 17:31:29.000000000","updated_by":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"reviewer":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"state":"CC"},{"updated":"2024-08-21 08:21:36.000000000","updated_by":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"reviewer":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"state":"CC"},{"updated":"2024-08-26 18:55:09.000000000","updated_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"reviewer":{"_account_id":30025,"name":"Vida Haririan","display_name":"Vida Haririan","email":"vhariria@redhat.com","username":"vhari"},"state":"REVIEWER"}],"messages":[{"id":"d43fbb05f93fb63244b9c15f504553a4d49e2f58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-08-09 10:36:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0281117b43a4f7a2ab9a1de35e03f326a93fb043","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-09 11:35:46.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/1/check/DellEMC_PowerFlex_Manila_dhss_false/0f1d543/DellEMC_PowerFlex_Manila_dhss_false/1439 : SUCCESS in 31m 57s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"7f937eaef289a4f1b560b32bb97f90be4c6846b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 12:24:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/28f16eca19334b2ca528cd734470d73a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5de781762d90452a879cea5aafd71a00 : SUCCESS in 9m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac13d4a91a8349d98e0d759aac589dc0 : FAILURE in 20m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5210d25a712749fbaa810bf203319d3a : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2248742a5e34a6d97829b5e7bae86de : FAILURE in 10m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/296de5ec78e24a7da6b7b61668f3742c : FAILURE in 9m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fb6f0926ca5440068c125257f930b6cf : FAILURE in 14m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e0dba6e958a43329dfde838fca64486 : SUCCESS in 9m 12s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/884e6cc5b8b24a6d9eba22ccd30a3355 : SUCCESS in 1h 19m 02s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/28d30709b66f453c940fbdd1ff0d3303 : FAILURE in 1h 42m 42s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/72262f6ce10d42ba95e85a25aa4dff87 : SUCCESS in 1h 42m 43s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/7b86e3be0ce04a89a632fb35c772ebf0 : SUCCESS in 1h 12m 59s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/112c8357b9aa4aee8de66a868014e2ba : SUCCESS in 1h 01m 37s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/db5710ea3ea04d8984c202e01511feea : SUCCESS in 55m 08s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/dcf935c6d36b4e7ab8828aa5f4bf6a8a : SUCCESS in 4m 17s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/ae4d2223cbed4cadaac1e8fa8cca4ebb : SUCCESS in 4m 27s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/587d25211af04525bb8b2941d9622f89 : SUCCESS in 33m 15s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/12a54e4d29c74165aafce665f77c81b4 : SUCCESS in 32m 22s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/e8633d23789c4cce95e1c83843ac6672 : FAILURE in 32m 54s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/4646f4734f604094aba8be968e875549 : SUCCESS in 1h 03m 52s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2743f487aa204bffa7af192bd5cca5a4 : SUCCESS in 53m 02s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/c8cd5f6132334c67bff556069985fa2e : SUCCESS in 23m 51s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/69e3b842159f4f129308bb8a90df7443 : SUCCESS in 29m 20s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/ce98948911cc4fe5b5491c25f0da99ff : FAILURE in 4m 39s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/03aa794559f14c0693fe8ec7c9477127 : SUCCESS in 7m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c27b7e86e7bf630917ded066ba1708db1e4e32d1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-09 16:24:06.000000000","message":"Patch Set 1:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/926049/1/check/DellEMC_Unity_Manila_dhss_true/3d8b5a9/DellEMC_Unity_Manila_dhss_true/1518 : SUCCESS in 2h 45m 42s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/1/check/DellEMC_Unity_Manila_dhss_false/2ae8393/DellEMC_Unity_Manila_dhss_false/1629 : SUCCESS in 1h 05m 31s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":1},{"id":"2ca48409d8550f0a551a99ed46dc4559d6c48cba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-08-13 12:32:46.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"668ceb251777afa36e7fe80b5166c64be1e7fcb0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-13 12:57:08.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/2/check/DellEMC_PowerFlex_Manila_dhss_false/7a7a02f/DellEMC_PowerFlex_Manila_dhss_false/1465 : FAILURE in 24m 03s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"26bb1a28f7d9de3f4c6405aaa2a63f15b1ddef5d","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-13 13:01:54.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/926049/2/check/DellEMC_Unity_Manila_dhss_true/65898e0/DellEMC_Unity_Manila_dhss_true/1533 : FAILURE in 28m 47s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/2/check/DellEMC_Unity_Manila_dhss_false/f3cb6bd/DellEMC_Unity_Manila_dhss_false/1644 : FAILURE in 25m 28s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"1e8c8a40b9f8e4f30fd117ae383b947082c0f286","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-08-13 13:32:03.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9b7c00ba2fe37ba4c99975ffb3e86b29835a1b1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 14:25:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4087ef2dbd9462093a41ed57c84ce37\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfe562d176cf4c69b88e6b8b4c8dbe6f : SUCCESS in 9m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b718926faf944b6bbe395ac61dd61864 : FAILURE in 25m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3a78944aca842d096bfa04acdbe8b4c : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2f1f42dd7b5429e86fbabb0ea18e568 : SUCCESS in 11m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/db28b927d4754084b4c24a0062fe1f0a : SUCCESS in 10m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/80c3bb18d72f49c091afa015a736a675 : SUCCESS in 16m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d511d7143504adeb63eea7728af4dbf : SUCCESS in 8m 28s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/81700046c4a54134a69f2f5503960e67 : SUCCESS in 1h 01m 45s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/09615e38477d45dab80dc2014137f39c : SUCCESS in 1h 51m 14s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/0545e2fc6f3641938682017c8508e85c : FAILURE in 57m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/ac24efb8caff49018c2ae668e6c3e335 : SUCCESS in 1h 01m 24s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/bb06f0eedef84c20a89964e36c480646 : SUCCESS in 1h 05m 03s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/70ef99639a244dc88e8888573826cb35 : SUCCESS in 1h 24m 28s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/45c83dfd69e0491ba5d750c37569066b : SUCCESS in 4m 03s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/5a7bd8b9e43942f0aaf7b566afbb70e7 : SUCCESS in 3m 48s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/4f39be4804c640b2a9943941ae05ea74 : SUCCESS in 33m 59s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/8ffc0da943cb4a07adf0857342d97709 : SUCCESS in 32m 04s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/a5d490356ea14302a1fa37aaef557c33 : FAILURE in 49m 29s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/0fb093fe893d48228a9d4e2f8361fd0b : FAILURE in 24m 27s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/224f45206af84f259e2fc5dcbf0ce539 : SUCCESS in 55m 38s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/3446ceb9b3204bf2ad09b90fc5d547e7 : SUCCESS in 28m 31s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/19beb7725ded4d51b2ebddc1a309c2f8 : SUCCESS in 33m 59s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/80a054e64ae24bd7baa9f3c8f086d342 : FAILURE in 5m 18s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/594e91d9a99e4520aa280dca1f7e8a2c : SUCCESS in 5m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5dc82c48225e1c99fd2ce5cd07e4b5dabe658495","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-13 14:49:09.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/2/check/DellEMC_PowerFlex_Manila_dhss_false/848a100/DellEMC_PowerFlex_Manila_dhss_false/1466 : FAILURE in 22m 49s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"0cb3fc26af7a7c56763f13d477a5172c50b4128b","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-13 14:50:57.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/2/check/DellEMC_PowerStore_Manila_dhss_false/fac2c2e/DellEMC_PowerStore_Manila_dhss_false/418 : FAILURE in 24m 39s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":2},{"id":"195f6755029cfcbe281ade0c4dfbc347894028de","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-13 14:54:09.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/926049/2/check/DellEMC_Unity_Manila_dhss_true/1461205/DellEMC_Unity_Manila_dhss_true/1534 : FAILURE in 27m 49s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/2/check/DellEMC_Unity_Manila_dhss_false/0ec3022/DellEMC_Unity_Manila_dhss_false/1645 : FAILURE in 24m 03s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":2},{"id":"f535f54ff8435b7e1b94f5165f3e449c03cb87d1","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-13 14:59:12.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/2/upstream-check/manila-tempest-plugin-ontap-dhss/a7e536a/ : RETRY_LIMIT in 3m 41s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/2/upstream-check/manila-tempest-plugin-ontap-dhss-rest/47d4d27/ : RETRY_LIMIT in 3m 51s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/2/upstream-check/manila-tempest-plugin-ontap-no-dhss/3ed7c34/ : RETRY_LIMIT in 11m 05s\n- manila-tempest-plugin-ontap-no-dhss-rest manila-tempest-plugin-ontap-no-dhss-rest : NODE_FAILURE in 0s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":2},{"id":"e7aa4ccfe874e730e9ca365e6bcf2d37cd4f8662","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-08-16 16:58:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"07c7b975c6c4891c499cc2ca6527788804d3585c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-16 17:26:10.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/926049/3/check/DellEMC_Unity_Manila_dhss_true/a0947b6/DellEMC_Unity_Manila_dhss_true/1543 : FAILURE in 25m 59s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/3/check/DellEMC_Unity_Manila_dhss_false/531e94e/DellEMC_Unity_Manila_dhss_false/1654 : FAILURE in 23m 45s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"fbf0082c43f1b3750b48ab08543f69f9d93530a6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-16 17:31:09.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/3/check/DellEMC_PowerFlex_Manila_dhss_false/3efa8f3/DellEMC_PowerFlex_Manila_dhss_false/1476 : SUCCESS in 31m 04s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"567e5387f912b82ee6283b4a3bd9a192d04c31a4","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-16 17:31:29.000000000","message":"Patch Set 3:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a044a48559b14196a9911a7c740286ab\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/54ae89836b3e45699566bcf10dac9797 : FAILURE in 28m 05s","accounts_in_message":[],"_revision_number":3},{"id":"4a2f1ad18d2b5e11d4016948b5de0826807e440d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 19:44:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6673f988ecd54e138d6747294444d705\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65abf7f3fd9e435d86e28d0d91a2dcb9 : SUCCESS in 9m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38414c83d20a496ab8fe685c2860e4fb : FAILURE in 29m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8fa73ab6822423ab51c1b14e4beb13f : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/024184e7f72e4ed0b9c4843bcb3138d1 : SUCCESS in 9m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/448feb6328c64d61ab91dc3d7f961ad9 : SUCCESS in 7m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45d0eca9ffda40c483c8f26509c3b4ed : SUCCESS in 11m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/171949ed43724759bc93b366eef9b1d4 : SUCCESS in 7m 57s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/418d0fb613084bd4a2e8cf921b61890c : SUCCESS in 1h 08m 21s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/d6a4279cf7274423bdc149d8a03f9a1d : SUCCESS in 1h 27m 38s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/945249eca9114f0095c7d582144d24ff : FAILURE in 2h 32m 52s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/8feff483f0df4ceda4b55f1f0f75948a : SUCCESS in 1h 00m 47s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/49df3cad5a3b46e98d93657b186a22b2 : FAILURE in 6m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/43711a30364746e7b042804e42465b9b : SUCCESS in 1h 11m 11s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5da5336a0a4947c799e5b3afec2900e5 : SUCCESS in 4m 08s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/8ce0ef94fb424c03b516954792ded779 : SUCCESS in 4m 08s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8c78c746b5b74489a6c26b5aedef57e0 : SUCCESS in 27m 07s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/2467d98d6e944b4bbd6a49234a7ddfdd : SUCCESS in 37m 14s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/6bf6393529de4a44b2ea3954e5ebebd9 : FAILURE in 46m 02s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/bb8c13da4ee140478c8ed42c1d9e4668 : SUCCESS in 56m 25s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5f3aaec2b174491ea9bb858b26918408 : SUCCESS in 41m 10s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/747fdc6a2e434f3fa1bde553954c2e14 : SUCCESS in 21m 59s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/f3fd2a6beadb4350bb51c37352d9b183 : SUCCESS in 24m 20s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/50ee60987ad24baa80037fc18f27fd1f : FAILURE in 3m 59s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/2e14bd13f46e4ce3822c2e379caebf3e : SUCCESS in 4m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9bf9bf144bb306c6c3d16bec0123980209d2afc3","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-16 20:15:05.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/926049/3/check/DellEMC_Unity_Manila_dhss_true/2d95395/DellEMC_Unity_Manila_dhss_true/1544 : FAILURE in 29m 06s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/3/check/DellEMC_Unity_Manila_dhss_false/908cdc2/DellEMC_Unity_Manila_dhss_false/1655 : FAILURE in 27m 15s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":3},{"id":"c20ac0d0a5828a55045a82be0653333127281bb4","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-16 20:17:38.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/3/upstream-check/manila-tempest-plugin-ontap-dhss-rest/8d775d2/ : RETRY_LIMIT in 3m 47s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/3/upstream-check/manila-tempest-plugin-ontap-no-dhss/779cb69/ : RETRY_LIMIT in 3m 27s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/3/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/a29ffc8/ : RETRY_LIMIT in 3m 33s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":3},{"id":"ee74e89b167f09bc52e8655213929a150ed73c5b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-16 20:20:13.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/3/check/DellEMC_PowerFlex_Manila_dhss_false/45fbc27/DellEMC_PowerFlex_Manila_dhss_false/1477 : SUCCESS in 34m 58s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"a4a4cf35e0e6ab9af6dfbb53c70341195890ff90","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-16 20:20:47.000000000","message":"Patch Set 3:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/3/check/DellEMC_PowerStore_Manila_dhss_false/29c066a/DellEMC_PowerStore_Manila_dhss_false/424 : SUCCESS in 35m 31s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":3},{"id":"66ad181327d9f27491bfd57bd8c1f42f0141bee0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-08-17 10:59:38.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"1034818cf8fe8caba5e5b5baedaf2daa5bf7d38c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-17 11:27:37.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/926049/4/check/DellEMC_Unity_Manila_dhss_true/d495002/DellEMC_Unity_Manila_dhss_true/1547 : FAILURE in 27m 39s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/4/check/DellEMC_Unity_Manila_dhss_false/f3f71a7/DellEMC_Unity_Manila_dhss_false/1658 : FAILURE in 24m 07s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"22b2b1116404d228320983fd68226069a8f638a8","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-17 11:30:00.000000000","message":"Patch Set 4:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/37e4f351f7414dd7928168576f0b8bce\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/023c672fada241f1b4279d659d6d4112 : FAILURE in 24m 56s","accounts_in_message":[],"_revision_number":4},{"id":"a78106485ccbc00bd543bb8ac936f7a8cf7cd725","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-17 11:33:32.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/4/check/DellEMC_PowerFlex_Manila_dhss_false/b32fb0f/DellEMC_PowerFlex_Manila_dhss_false/1480 : SUCCESS in 33m 35s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"e65d21a3c83008d30a307a60b862b980f3ab3ffd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-17 12:27:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebba256404c840ccb89dc92092a8d1d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d585de4cbf34490b4a3e6f47360dbe2 : SUCCESS in 10m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d30e1b73139449dbad39bc3aeae9ba99 : SUCCESS in 27m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c82098611ca648eabcc3519aad7c3883 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30ad1107b4a742fb94d74236e70529a4 : SUCCESS in 9m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b1812c55bb334d0cb68eb5cdd4b970f0 : SUCCESS in 8m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e97dcdffb13146bf8a366bfcb0f6ea03 : SUCCESS in 9m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a125a6e75f648729a89591a276ed38e : SUCCESS in 8m 03s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2ba9bf040fc5461b8cdd45d742ed93a4 : SUCCESS in 47m 29s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/b7f6953f566c45129a2ff2ed15e26f83 : FAILURE in 1h 26m 45s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3c9a793566f64d13881b43deb97b5427 : FAILURE in 50m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/58d80a2678c54783a1a794e8155f7e70 : SUCCESS in 42m 06s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/d0653c30da5f4e35b8bd9235096865e4 : FAILURE in 6m 35s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/9332aa11788d4cebad218bf9eaac0b8a : SUCCESS in 44m 54s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/b8e77aa507ff4bb1bc870480e56a7688 : SUCCESS in 3m 58s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/082afd695c9a4727b5c9c8feee86eb34 : SUCCESS in 4m 09s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/0810388f31b44b21b4f0effb63e3a4cc : SUCCESS in 35m 27s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/e9486d8157d3423e9f62012cc968115d : SUCCESS in 38m 18s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/57ecf78dea974e04852305120eaffa1e : FAILURE in 28m 10s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/f7a0367a6fe24b5ca44477571bcca65e : SUCCESS in 44m 47s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/289d54e83cb34fbeb83391bfaf21a620 : SUCCESS in 59m 04s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/93274538b1da420a9388c500d4c7f4e3 : SUCCESS in 34m 21s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/72f732f797e045d9a453fd1424a110f4 : SUCCESS in 22m 25s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/b41839027411415f8700a75aab72ab78 : FAILURE in 3m 59s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/a4870237171c4d4cae9c9724046b5790 : SUCCESS in 5m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1350e903e3b0b944dd9910e1570729d1a9598d2d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-17 12:54:17.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/4/upstream-check/manila-tempest-plugin-ontap-dhss/7e220cd/ : RETRY_LIMIT in 4m 02s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/4/upstream-check/manila-tempest-plugin-ontap-dhss-rest/6d4c5b8/ : RETRY_LIMIT in 4m 06s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/4/upstream-check/manila-tempest-plugin-ontap-no-dhss/30bd605/ : RETRY_LIMIT in 3m 57s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/4/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/cdf423b/ : RETRY_LIMIT in 11m 02s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":4},{"id":"ce460425f642449ab3457bb3df6169dae85c38d1","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-17 12:55:28.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/926049/4/check/DellEMC_Unity_Manila_dhss_true/1ed2770/DellEMC_Unity_Manila_dhss_true/1548 : FAILURE in 27m 39s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/4/check/DellEMC_Unity_Manila_dhss_false/bbb644d/DellEMC_Unity_Manila_dhss_false/1659 : FAILURE in 23m 56s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":4},{"id":"a7ed28cda05be6fa94a9b3e9df0342f4a69e6273","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-17 13:00:53.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/4/check/DellEMC_PowerFlex_Manila_dhss_false/7775ee7/DellEMC_PowerFlex_Manila_dhss_false/1481 : SUCCESS in 33m 04s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"1b394b845811d32f9d32d8b815906e1f9b57426e","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-17 13:02:44.000000000","message":"Patch Set 4:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/4/check/DellEMC_PowerStore_Manila_dhss_false/6a794a4/DellEMC_PowerStore_Manila_dhss_false/426 : SUCCESS in 34m 57s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":4},{"id":"85f7b9fa430cf790dec7c9f4b77c27dba0ca92a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-08-20 12:56:32.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"603efcb7b4dafbce058021e0f65bb7c7c3223ac9","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-20 13:29:05.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/3329c913f61945f394368d7eb931f075\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/11c12772aca74c37b94c3e5150137fa5 : FAILURE in 26m 53s","accounts_in_message":[],"_revision_number":5},{"id":"9b19de24d1019ccf01476b469f8f51c4e51e34b8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-20 13:50:52.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/5/check/DellEMC_PowerFlex_Manila_dhss_false/b52741c/DellEMC_PowerFlex_Manila_dhss_false/1497 : SUCCESS in 32m 50s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"3b593c0322bf19c128bb953b8a7c36c7de54514e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 14:51:27.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ff66b535efc49fe8155fa5d123e95c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f758993a7a0466bab2baab93b4a2e9d : SUCCESS in 10m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc32428bb9a44c828dd78d08ca2bcbe3 : FAILURE in 26m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebb6c70497c44ad38721df5b336a8e3b : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8471e5fe1f8045b4b7f7f9dba5cd92df : SUCCESS in 10m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95daa1d99a1842b29e1daad451562f7b : SUCCESS in 8m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a55a41ef133e42f3a6281af966006491 : SUCCESS in 12m 00s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05a7c2e8a1b8409c9c2c8c687d27df19 : SUCCESS in 10m 16s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/53a247198e2d47c09b90415d0d93b452 : SUCCESS in 1h 24m 50s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/92ba855a3e5d4178a64fadb930669d41 : SUCCESS in 1h 53m 02s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/1293084ac3d14207a50444e53972abca : FAILURE in 1h 06m 17s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/def86a5b526f4e80b5905017ce586313 : SUCCESS in 1h 03m 00s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/8a6217e25b5d4b57a16c5ee4901455e4 : SUCCESS in 1h 01m 04s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1bbb3aa83dcd45358e852e61ca5ca5fd : FAILURE in 1h 00m 48s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/5085a0cd94b6498ab084e2a536dc5e7b : SUCCESS in 4m 07s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/a30ab6c6acb149d3870463fc6e51196d : SUCCESS in 5m 01s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/7317bce5a06f4cae82bbfa2c10d7367d : FAILURE in 14m 29s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/c85ddc445f4944a6afaf2b0706819e1a : SUCCESS in 46m 30s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/d40b81b63e1c497596998a60c05a07dd : FAILURE in 36m 37s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/abee418716084de09d7ec2585b27ba6d : SUCCESS in 1h 04m 58s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a6d14742e8b4df2bacf38f5439af6b1 : SUCCESS in 1h 00m 41s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/93f90435452c4d2686f37998a9aa2e3e : SUCCESS in 39m 47s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/b432262c4c194d99977e868faf378589 : SUCCESS in 29m 35s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/072911daa7d64538841aabd98b10473d : FAILURE in 4m 15s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/a7bff5cc7f094a2799bbb46958a1f10c : SUCCESS in 6m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f36a88deccca569a7cc2b0eafcb0518257a67fbd","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-20 15:17:27.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/926049/5/check/DellEMC_Unity_Manila_dhss_true/ce07686/DellEMC_Unity_Manila_dhss_true/1566 : FAILURE in 27m 21s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/5/check/DellEMC_Unity_Manila_dhss_false/ead1d73/DellEMC_Unity_Manila_dhss_false/1678 : SUCCESS in 1h 21m 51s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"9ff52548d241d622c6799a8e47c92648a518f91e","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-08-21 05:38:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"904f52bf3db9c0cd67dc079d8b9993db77f5a5cd","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-21 06:10:18.000000000","message":"Patch Set 5:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/fc5003afa3b04741877a337dacf3fab9\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/17fb4b733f9a4d05b3938a2bcc4f7fdd : FAILURE in 31m 35s","accounts_in_message":[],"_revision_number":5},{"id":"fbf6aeedbc3ac320d0bae98ad9a27dcf37ebc2ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 08:01:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20ae2ef02d2842b4806a8b4a1438ccc9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5f307aca78c495bad31f972e07dffa2 : SUCCESS in 9m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2b22c21dc3e483a83c04e2c65e02864 : FAILURE in 21m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f759a6ad05584118a0d1e3ad86768fb5 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9153986cf0945089b71cb38861c4a6d : SUCCESS in 10m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/288d1f24bc9e4e51a22f31f96e1db295 : SUCCESS in 9m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8712c4a484994d4d96848a9a48bb7279 : SUCCESS in 12m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7abba666211e48379bb7a98204ca80f5 : SUCCESS in 8m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/a4c814a3418140ef9533997ca77f7876 : SUCCESS in 1h 03m 18s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/14e0f397f2864554b003fd6c3b58a433 : SUCCESS in 2h 17m 37s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/a3d7caf7d75f47f0ad5874822d4bf705 : FAILURE in 1h 02m 07s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/519d04c4bde5407380a63cb43b85847a : SUCCESS in 52m 29s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/719cd0f4bad44e24815705f93ec06fb9 : FAILURE in 7m 15s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0365a9da0e124d6aa89ff4fa758edc1b : SUCCESS in 58m 09s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/efe14116e7624962a25c42a9b54a658d : SUCCESS in 4m 00s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/dad25a0edd6b4bddbcb8a1769846b04c : SUCCESS in 4m 04s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/83ee4e4bddb64df8a071489c1055fe65 : SUCCESS in 39m 52s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/a97252b2180146cbb8730db2ff791834 : SUCCESS in 39m 30s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/9e900d6081e54983afaacf475fd8e045 : FAILURE in 43m 27s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/84dd3a24383243f98ff61787226f92cc : SUCCESS in 1h 20m 34s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/85060f0107454570b8d77bb15614974f : SUCCESS in 58m 04s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/ae034bda87e34ffe974e0763ce820312 : SUCCESS in 38m 00s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/dc9b79262a1544bd8441899107a47568 : SUCCESS in 42m 25s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/a2f5c129c29f4271a41b4cf7ed52def2 : FAILURE in 4m 19s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/2555a1601e314995bd90c60279e77ab2 : SUCCESS in 5m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0c10187bd05f9b8640ce293aeb27d951f29655c9","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-21 08:17:34.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/5/upstream-check/manila-tempest-plugin-ontap-dhss/dcf9412/ : RETRY_LIMIT in 3m 33s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/5/upstream-check/manila-tempest-plugin-ontap-dhss-rest/7ec34cd/ : RETRY_LIMIT in 14m 30s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/5/upstream-check/manila-tempest-plugin-ontap-no-dhss/e78a475/ : RETRY_LIMIT in 4m 55s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/5/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/b10b480/ : RETRY_LIMIT in 4m 58s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":5},{"id":"c5f46310d43546485e17b94331d77f419e33cf24","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-21 08:21:17.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/5/check/DellEMC_PowerFlex_Manila_dhss_false/fb9f6ad/DellEMC_PowerFlex_Manila_dhss_false/1507 : ABORTED in 19m 10s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"847df19c0577c7e5bce89504ed3978e02aba4b51","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-21 08:21:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"87cdd53ef41a3874379c99432ddd9ee50a14a102","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-21 08:24:00.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d6598efc5121c2d507895c3fc7a6520b415b0dab","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-21 08:40:29.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/5/check/DellEMC_PowerStore_Manila_dhss_false/a82c5d3/DellEMC_PowerStore_Manila_dhss_false/431 : SUCCESS in 38m 24s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":5},{"id":"8852ef8d22ae46aa37bee9eb7955ef3d34eac6c2","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-21 08:45:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4897fd97f029578883753dd017a345a74d0ee6f9","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-21 09:18:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ccbd0e08650dcd9a3e7378f1d5a2663e1ced0b27","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-21 09:18:59.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/5/check/DellEMC_PowerFlex_Manila_dhss_false/774b3f8/DellEMC_PowerFlex_Manila_dhss_false/705 : FAILURE in 17s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"172af2bef5eedde0f19339ae224fb2d309790e29","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-08-21 11:17:40.000000000","message":"Patch Set 5:\n\nBuild succeed.\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/926049/5/check/DellEMC_Unity_Manila_dhss_true/7b9b9e4/DellEMC_Unity_Manila_dhss_true/1572 : SUCCESS in 3h 08m 07s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/5/check/DellEMC_Unity_Manila_dhss_false/7877aea/DellEMC_Unity_Manila_dhss_false/1683 : SUCCESS in 1h 22m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":5},{"id":"b489badf08186ef974644d7acffbd37bfd42a91f","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-22 02:55:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9a398977b3d865c965ef41c8089a0fdc71ea3e75","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-22 03:11:41.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/5/check/DellEMC_PowerFlex_Manila_dhss_false/adf6629/DellEMC_PowerFlex_Manila_dhss_false/1509 : ABORTED in 15m 31s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"f147e4469b8d074466f1379f416b0c7475e3ac97","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-22 03:11:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d74d510b13a3d1f7ec01a6abe03e48069f8bdf36","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-22 04:37:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"94522fbbef5e7e5e2e0c5e484c9f49f0b81d57c8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-22 04:49:51.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/5/check/DellEMC_PowerFlex_Manila_dhss_false/c39e258/DellEMC_PowerFlex_Manila_dhss_false/1511 : ABORTED in 11m 51s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"3daa607820e5891a6cddc1255c84f1067b912824","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-22 04:50:06.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5ac16da764ea3e126344a0a689e363b6d1a7da28","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-22 05:09:34.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/5/check/DellEMC_PowerFlex_Manila_dhss_false/e9a8c65/DellEMC_PowerFlex_Manila_dhss_false/1512 : FAILURE in 19m 11s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"01b7f0383fc983a8ba603502489f22fa6189e0c0","author":{"_account_id":33807,"name":"Jacob Wang","email":"jacob_wang1@dell.com","username":"jacob0522"},"date":"2024-08-22 05:24:53.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"62f1a3e9c62a8c11c641690a35212405d1cac775","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-22 05:47:18.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/5/check/DellEMC_PowerFlex_Manila_dhss_false/b6ec72f/DellEMC_PowerFlex_Manila_dhss_false/1513 : FAILURE in 22m 05s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"a56c47ff2f364d85f8c2fb130ee66a02afadf5e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-08-29 11:28:16.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"1f621fa7df50b1bf2df7c386d3e0b6ff294e7b13","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-08-29 11:37:02.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"317d515eb6a3a2e91b82a99f5301b9e7be22bc62","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-08-29 11:55:31.000000000","message":"Patch Set 6:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/f5e99fbde65e4ef788804f7a1668a8e6\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/f09eb1302fbe43929ed8bd51bb77636a : FAILURE in 26m 44s","accounts_in_message":[],"_revision_number":6},{"id":"cb06e4d2bef1436ce430a1191724e74d71434069","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-29 12:03:34.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/6/check/DellEMC_PowerFlex_Manila_dhss_false/aa90cf7/DellEMC_PowerFlex_Manila_dhss_false/1570 : SUCCESS in 34m 56s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"306624d35fe0c118bbb365fde8386ed673eacab4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 13:39:33.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f660dcb357443fab0956567afe109f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c715d62283a42b89f864b99b6fce72a : SUCCESS in 11m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48bf61568b6949bb9b99975f330a7647 : FAILURE in 27m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/375ab67769e74929a5635cb1a977b082 : SUCCESS in 8m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8250cd6eb03b482e97e23ac71cbe438b : SUCCESS in 12m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c142eb519224bf987fd0b88e1f0f36c : SUCCESS in 10m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/81a598207adf4c6eb1241b56d3b107be : SUCCESS in 13m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb011951274e41aca69ec9d9d1141944 : SUCCESS in 9m 00s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/2070de74f8f64836a724e4a64ff595b6 : SUCCESS in 1h 07m 04s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/e20f9fecd0b44b4dbf775a9b67c46ff6 : FAILURE in 2h 10m 10s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/35051d0462ab4530a779634f15ed3c06 : FAILURE in 1h 09m 04s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/61c7ef10655344479154394ca4a1614f : FAILURE in 58m 03s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/ef1cbf81244b49e49959d3ce0cf30d8f : SUCCESS in 1h 05m 41s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/0278d3f996da4ad39349a0ea77724d28 : FAILURE in 1h 17m 51s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/415cc612b1174be2941874a5d7d10e92 : SUCCESS in 4m 40s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/b674a33b6a7c42728685cdd5209f089f : SUCCESS in 4m 31s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/8259543927154909a1db8985f0bbbdfb : SUCCESS in 38m 28s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/060d7debfa784f10b1ae4206cc9a0434 : SUCCESS in 41m 15s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/ab5a6043c4f449a8af5cdd093e0932fe : FAILURE in 46m 52s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/a4de73ec7c224474a01a25e0faa74faf : SUCCESS in 1h 14m 56s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9114316d2b004cc09c04f3911af2bef0 : SUCCESS in 1h 08m 39s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/1bbc1d61f4194419ae862ec82eb5448d : SUCCESS in 25m 24s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/9e40b5e2f13d4c95a10f4fb2bfe9e9b8 : SUCCESS in 38m 56s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/fcc2d3d9e59a48b194b59e78e5d39db2 : FAILURE in 4m 15s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/4cd03e0959bf46daa266185e69d6afb0 : SUCCESS in 9m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"30821edf6afd550641f57fd52d5e5bbe960757b6","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-29 13:41:27.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"c34739d51db1ffb8947cd490a979581eb94a02e2","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-08-29 14:15:53.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/6/check/DellEMC_PowerStore_Manila_dhss_false/be2d06f/DellEMC_PowerStore_Manila_dhss_false/454 : SUCCESS in 36m 05s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":6},{"id":"7e83bf5edb0ff0eb7a1adf1d1466e9cd423c80e1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-08-29 14:50:06.000000000","message":"Patch Set 6:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/6/check/DellEMC_PowerFlex_Manila_dhss_false/ebd6639/DellEMC_PowerFlex_Manila_dhss_false/1572 : SUCCESS in 34m 46s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"318e08de3dada2c13ebd252cfca4ad819a8832de","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-08-29 21:53:36.000000000","message":"Patch Set 6: Code-Review-1\n\n(19 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b4869252e11a209480d65aefe05e166445086f92","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-08-30 00:50:17.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"25442e85c45a47ab92dda0c404653b2835527532","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-08-30 07:19:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5a0d2e05a00862df9475f43d86ecd09cad1ebd46","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-08-30 08:14:15.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/6/upstream-check/manila-tempest-plugin-ontap-dhss/a7c5827/ : POST_FAILURE in 42m 53s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/6/upstream-check/manila-tempest-plugin-ontap-dhss-rest/d0077dc/ : RETRY_LIMIT in 4m 27s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/6/upstream-check/manila-tempest-plugin-ontap-no-dhss/039222d/ : RETRY_LIMIT in 3m 50s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/6/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/c028cbf/ : RETRY_LIMIT in 4m 37s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":6},{"id":"ca56f7105eeacaa7b1f5e70a3f3d1042baa09b25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-09-03 09:34:16.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"fac47addcdfee9df287778379cf230011aa8443f","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-09-03 09:35:24.000000000","message":"Patch Set 6:\n\n(25 comments)","accounts_in_message":[],"_revision_number":6},{"id":"24b50ddcece7492bf6e297572ae81f03f3a17546","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-03 09:48:15.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/926049/7/check/DellEMC_Unity_Manila_dhss_true/c88b3e7/DellEMC_Unity_Manila_dhss_true/1612 : FAILURE in 13m 41s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/7/check/DellEMC_Unity_Manila_dhss_false/2024374/DellEMC_Unity_Manila_dhss_false/1723 : FAILURE in 12m 00s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":7},{"id":"4a7dacf721100ea56a02937d0da38d15058a36b7","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-03 10:00:56.000000000","message":"Patch Set 7:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/e6e5867803d142fab538922e7219cd4b\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/a5852e1ab02049b39655af8f0d17a8bf : FAILURE in 26m 10s","accounts_in_message":[],"_revision_number":7},{"id":"e3e814283586a7c3c8bf7d0f6d7f55d381db7650","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-03 10:06:27.000000000","message":"Patch Set 7:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/7/check/DellEMC_PowerFlex_Manila_dhss_false/6dbf965/DellEMC_PowerFlex_Manila_dhss_false/1614 : SUCCESS in 31m 53s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"1ff0a74daa6abef2ddd3beb15ecc616c7eb7272e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 11:23:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ab98194f5bd46c2af9b12ffdf564218\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fd1e58833694ef58c6ba0ca015163fa : SUCCESS in 11m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ea855c90bd147a3ae108e71ee100490 : SUCCESS in 30m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc23fa79d67741bc883303f68ec56fdf : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f2dcf2917b1481f98d2f86cd88692e2 : SUCCESS in 10m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/463bef2539194a6fa6f6792ba064c94a : SUCCESS in 8m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b20e15fae62d4201afcd63f21c546d60 : SUCCESS in 11m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbfc985441634ddb84d1dc53c8af955d : SUCCESS in 8m 40s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6e701ce5ab6546ecb3904b327e60fea5 : SUCCESS in 1h 04m 36s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/3f5f84fce1db4379914b9e6b1be2d1d3 : SUCCESS in 1h 46m 57s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/3ccae7c0bfe94fcbb54404dfb511ab4b : FAILURE in 1h 09m 29s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/c0de0eeca9304214ab5b67ee57094999 : FAILURE in 1h 01m 45s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/33229b96c8494a3a822f862c8221855a : SUCCESS in 1h 03m 07s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/086b7261dc4c4a2fb66f2e57817212df : SUCCESS in 50m 56s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/742621ce8c3c4526a8a901ec253be592 : SUCCESS in 4m 54s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/62a630c56e494370a9199a5443072d6c : SUCCESS in 4m 19s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/1bee4c0df3034068b764c8281902249b : SUCCESS in 34m 32s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/6ba9da2d13e74ee99933b01ee76d068d : SUCCESS in 37m 32s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/7d43c4015ffc4834a2a035b1d2cfcb59 : FAILURE in 42m 38s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/8ef9400f8f854474aa6af8f7b975fdf8 : FAILURE in 31m 48s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/89764c4ccc2848d0af290b982e4338f6 : SUCCESS in 1h 05m 18s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/a0747e1322a34aec9e66f1df0b6cb55b : SUCCESS in 31m 24s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/c4a8150bcede43fc855a76b6b1c29b17 : SUCCESS in 34m 35s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/c20b956da92445cbb6c099475863ba3b : FAILURE in 4m 57s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/b543120573014c57b41cba983d66b838 : SUCCESS in 6m 21s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"686f583657d2b5df0492058d74d8a63a25b67c92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-09-03 11:26:41.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"c9ea1ac36b830dda70349c805156545311500106","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-09-03 11:30:12.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"676e7e206cb76ee8c620b5ef4fd55c6e4e37468c","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-03 11:43:10.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/926049/9/check/DellEMC_Unity_Manila_dhss_true/69f9cc1/DellEMC_Unity_Manila_dhss_true/1615 : FAILURE in 12m 39s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/9/check/DellEMC_Unity_Manila_dhss_false/7fe76cf/DellEMC_Unity_Manila_dhss_false/1726 : FAILURE in 11m 12s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"a00cbe36fe1713e01f0d8b41661c2fb4a9d0c4a6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-03 12:00:17.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/9/check/DellEMC_PowerFlex_Manila_dhss_false/485f1b4/DellEMC_PowerFlex_Manila_dhss_false/1617 : SUCCESS in 29m 46s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"fcf2bdcb48ff6353193ab38b3f44db98a7fdca83","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-03 12:02:06.000000000","message":"Patch Set 9:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/a764d1f3a7be46958a3e3c3c6316ccc7\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/30152247956d43f4b79e395c6fc71fc9 : FAILURE in 26m 41s","accounts_in_message":[],"_revision_number":9},{"id":"462fec950fa107077b021b5614f3feaf625ce114","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-09-03 13:17:31.000000000","message":"Patch Set 9: Code-Review+2\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"38e4e8a8c68b7e76b6afede83bc014bcb72ecae8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 13:29:34.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d90d934f0354e74ab9382b289397bee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/038b4373c3fa4a9f9470a77f5b1e64aa : SUCCESS in 10m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f147b47f9cd04b17bfc8da9b6dd0473c : SUCCESS in 29m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5aa3457b3df249de871d06b524a8904a : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf30f0d1711542d8bc5f3cc57b12c8ea : SUCCESS in 13m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f44e0e03d3b4c5688be5db5cf02a63d : SUCCESS in 9m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/91bf8aa4ee0b49e7b41a5ef2d68cef90 : SUCCESS in 10m 15s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60be0ca1a4734112b46f456acbd52e8c : SUCCESS in 8m 11s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/6d3814b61c9c45ca8841ff38c54d1cbd : SUCCESS in 59m 31s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/735a01f2dc6f4fd4a60ed760bae893de : FAILURE in 1h 57m 20s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/5822eeb6bbbe40819fa79db797ba8df8 : FAILURE in 35m 19s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/5f993943d8934ea282ab66d92972b430 : SUCCESS in 48m 19s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/f5b0a1c406874ae2b3d238f20582ddde : FAILURE in 7m 44s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/1af9a601238f41c189d675548f51ac77 : SUCCESS in 57m 34s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/6a37f7e08efd41e28752f8302e35967a : SUCCESS in 4m 33s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/c9882eda6bd84570a1d0a0a809ae44e3 : SUCCESS in 5m 13s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/3fff67b2628d428a8f1ca37cb1fac31d : SUCCESS in 44m 25s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/deb67c57442d4d3096c34fa75b2e74ec : SUCCESS in 30m 47s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/af1a8301b13643fdb06cad88b69653cd : FAILURE in 35m 02s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/918da6d33cab4ecba30db9d7a8bbc2bf : FAILURE in 16m 37s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b9d497fd3f0c457381e8bfa26a6e4ae9 : SUCCESS in 55m 21s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/9b56360e3acd4c34a572b8e4a98740c9 : SUCCESS in 25m 53s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/5369d9568ab44274a616dbbe11a659a3 : SUCCESS in 28m 35s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/b57087a3e10844df946a27e86397c506 : FAILURE in 3m 57s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/2a1e38f9efe3461b93b18393c605ea5e : SUCCESS in 6m 36s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8f8ce950b6221040bb5f08a85694905f580f57fc","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-03 13:43:36.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/926049/9/check/DellEMC_Unity_Manila_dhss_true/b66bcd0/DellEMC_Unity_Manila_dhss_true/1619 : FAILURE in 13m 43s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/9/check/DellEMC_Unity_Manila_dhss_false/603ce7d/DellEMC_Unity_Manila_dhss_false/1730 : FAILURE in 12m 02s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":9},{"id":"80805b31a419502fc025fbaf5485bc1a9deec58d","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-03 13:46:07.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/9/upstream-check/manila-tempest-plugin-ontap-dhss/b6c50cd/ : RETRY_LIMIT in 14m 20s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/9/upstream-check/manila-tempest-plugin-ontap-dhss-rest/34efc5b/ : RETRY_LIMIT in 14m 20s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss manila-tempest-plugin-ontap-no-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-no-dhss-rest manila-tempest-plugin-ontap-no-dhss-rest : NODE_FAILURE in 0s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":9},{"id":"99e7b32d9659c7a429d0899901cfa39543195136","author":{"_account_id":32919,"name":"kiran pawar","display_name":"Kiran Pawar","email":"kinpaa@gmail.com","username":"kpdev"},"date":"2024-09-03 13:46:33.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6aeee41acbd45b3c40a2adc60b25b823b40a16fd","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-03 14:00:49.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/9/check/DellEMC_PowerStore_Manila_dhss_false/b10799f/DellEMC_PowerStore_Manila_dhss_false/471 : SUCCESS in 31m 00s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":9},{"id":"41a131c4c756e8ac90a799a2c33244be3fabe045","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-03 14:03:34.000000000","message":"Patch Set 9:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/9/check/DellEMC_PowerFlex_Manila_dhss_false/8e60a78/DellEMC_PowerFlex_Manila_dhss_false/1621 : SUCCESS in 33m 43s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":9},{"id":"cfc8454c1f16dd28ddc8d81cd97c74ddb3947c39","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-09-03 15:31:16.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"56ab1016f9c32f2c608e5566244afd2de11616e0","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2024-09-03 16:16:44.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"79e6fb41b24e456c12037b2c725a4890b5109d3d","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-04 06:10:34.000000000","message":"Patch Set 9: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a2ee2b4c442a3a80f1d9861c563e3244de49fcb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-09-05 14:33:15.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"8b351c1cafa74892fd48b4b08ebb4868307c6895","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2024-09-05 14:54:57.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"86f54975887df874176884e42db92e0293f82c8e","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2024-09-05 14:55:18.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"660b0b6b314da16f8766fa7e18532cfaeec2591b","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2024-09-05 14:56:21.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"8ef1fc3a19d9f5bc0f4a102fa276f144821af0e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-09-05 14:56:38.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":11},{"id":"a2bb2b968c494ff7c3c35088713b9072b9296453","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-09-05 14:59:54.000000000","message":"Patch Set 11:\n\n(5 comments)","accounts_in_message":[],"_revision_number":11},{"id":"e2ceb62bf7254150671e59f522dbab1b10a69667","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-09-05 15:17:58.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"f42b965d5d2a5edca5ed196e386feb414e3cc6e1","author":{"_account_id":32912,"name":"Pure Storage Third-Party CI","display_name":"Pure Storage Third-Party CI","email":"openstack-third-party-ci@purestorage.com","username":"purestorage-thirdparty-ci","status":"Doing CI things"},"date":"2024-09-05 15:22:36.000000000","message":"Patch Set 11:\n\nBuild failed.\nhttps://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/buildset/9f21743f29ce4d1eaffeab9e1cff5756\n\n- pure-devstack-manila-tempest-aio-jammy https://cinderci-sf.dev.purestorage.com/zuul/t/third-party-ci/build/d891da26421f4c02a3554062717965ad : FAILURE in 25m 27s","accounts_in_message":[],"_revision_number":11},{"id":"ec5b2846131dd3c3fa6cda5a39842a4cfee4e842","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-05 15:29:43.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/11/check/DellEMC_PowerFlex_Manila_dhss_false/c4cbd32/DellEMC_PowerFlex_Manila_dhss_false/1633 : SUCCESS in 32m 46s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"caf5b085b9116c450ffb828fd08d6af38e1baefe","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2024-09-05 15:35:08.000000000","message":"Patch Set 11: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"3172f5346234feeef8db875095a988a6389cef1f","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-09-05 15:48:10.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"790176c0a5d394ace777c5462f780671da58c21f","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-05 16:18:38.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/926049/11/check/DellEMC_Unity_Manila_dhss_true/31f0184/DellEMC_Unity_Manila_dhss_true/1631 : FAILURE in 13m 22s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/11/check/DellEMC_Unity_Manila_dhss_false/8e9608f/DellEMC_Unity_Manila_dhss_false/1740 : SUCCESS in 1h 21m 41s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"65270082827a5bf8fec19ad9cced4655f250c779","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2024-09-05 17:04:24.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"3cb3edde512519afc5b0e887dda4342f569a4f7a","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-09-05 17:16:45.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e8310413428d5b865a0930945137cae364ea86dc","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2024-09-05 17:21:05.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"2a8833b9d463df60fc5c37587a42b4106dff82df","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-09-05 17:32:53.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"30490b87ca8873e27475084675f2105fc006e4d3","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2024-09-05 17:39:14.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"a010bdeb9fd3bcf4690095e9f37b6103178f6970","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 17:51:09.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc11d72a310f4988be459cf567ab3058\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb3129da7dae45f6b6bc063608cfa8f9 : SUCCESS in 9m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4166b4307ef44fea5665ea4f6c47d29 : FAILURE in 25m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6002f8bcebe44c06b95db4efac996364 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4acbf51cd33844c3ac2ca756349d377a : SUCCESS in 9m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f59482a57904b3e8461769a8500c222 : SUCCESS in 8m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ad56630e40f48b88a9f87d9e40671da : SUCCESS in 13m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/efb2640c97f54ac782cf1505ba018911 : SUCCESS in 8m 26s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/0ec4fec9f4de45a3bafddf560c2efe32 : SUCCESS in 1h 12m 50s\n- manila-tempest-plugin-generic-scenario https://zuul.opendev.org/t/openstack/build/5faf542ce58e417fab8ed8a58db23d07 : FAILURE in 2h 44m 04s (non-voting)\n- manila-tempest-plugin-generic https://zuul.opendev.org/t/openstack/build/319ef462087340c0932eaa5ce1a1f551 : FAILURE in 1h 10m 10s (non-voting)\n- manila-tempest-plugin-cephfs-native-cephadm https://zuul.opendev.org/t/openstack/build/2bfc09504ba24301970c39851751182d : FAILURE in 1h 02m 36s (non-voting)\n- manila-tempest-plugin-multinode-cephfs-nfs-cephadm https://zuul.opendev.org/t/openstack/build/659b66bf8e2e4f4c895f1e12504d9a4e : SUCCESS in 58m 10s (non-voting)\n- manila-tempest-plugin-zfsonlinux https://zuul.opendev.org/t/openstack/build/4e501375e4de4ac49b8a1f3486c4a41f : SUCCESS in 1h 18m 13s (non-voting)\n- manila-tox-genconfig https://zuul.opendev.org/t/openstack/build/f892abd4269144909dc22826b0b15413 : SUCCESS in 4m 36s\n- openstack-tox-pylint https://zuul.opendev.org/t/openstack/build/867981f6f8394d88bbf4e23a83a57f9c : SUCCESS in 4m 22s (non-voting)\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/b5051957fa5e44adabe3d93e243959fb : SUCCESS in 30m 36s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/36f14f38d86a4b1f8f7f03c8bd30d748 : SUCCESS in 28m 55s\n- manila-tempest-plugin-container https://zuul.opendev.org/t/openstack/build/54c667e1db1a4f0da379a4cbdaadd7b0 : FAILURE in 42m 53s (non-voting)\n- manila-tempest-plugin-lvm-fips https://zuul.opendev.org/t/openstack/build/b4ddead398174107a585d0a6b7ab8b81 : SUCCESS in 1h 18m 20s (non-voting)\n- manila-grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f60e0a21ea9b40c1b41d1b7768df663c : SUCCESS in 56m 56s\n- manila-rally-no-ss https://zuul.opendev.org/t/openstack/build/462fd5d0f7d44c998f223c229f147d71 : SUCCESS in 33m 52s (non-voting)\n- manila-rally-ss https://zuul.opendev.org/t/openstack/build/1a8a0cc1a3e24dbeae74428a761c7cbd : SUCCESS in 36m 27s (non-voting)\n- manila-tox-bandit https://zuul.opendev.org/t/openstack/build/f4dd8006e4ed499aa725d2c191e10545 : FAILURE in 4m 49s (non-voting)\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/0a846a4598714d62bf83179e2e103476 : RETRY_LIMIT in 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"f1a34db578809473f033d97ca07289080cc67d45","author":{"_account_id":34853,"name":"NetApp-CI","email":"ng-openstack-ci@netapp.com","username":"netapp-ontap-ci"},"date":"2024-09-05 18:09:14.000000000","message":"Patch Set 11:\n\nBuild failed.\n\n- manila-tempest-plugin-ontap-dhss manila-tempest-plugin-ontap-dhss : NODE_FAILURE in 0s\n- manila-tempest-plugin-ontap-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/11/upstream-check/manila-tempest-plugin-ontap-dhss-rest/3b154fd/ : RETRY_LIMIT in 9m 30s (non-voting)\n- manila-tempest-plugin-ontap-no-dhss https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/11/upstream-check/manila-tempest-plugin-ontap-no-dhss/62b3238/ : RETRY_LIMIT in 11m 59s\n- manila-tempest-plugin-ontap-no-dhss-rest https://netappopenstacklogserver.s3.us-west-1.amazonaws.com/index.html#logs/49/926049/11/upstream-check/manila-tempest-plugin-ontap-no-dhss-rest/9075ff2/ : RETRY_LIMIT in 11m 04s (non-voting)\n\nComment \"run-NetApp CI\" to rerun these tests. https://wiki.openstack.org/wiki/ThirdPartySystems/NetApp_CI","accounts_in_message":[],"_revision_number":11},{"id":"72023d62a3f685ff68613e61581f77994c030fe9","author":{"_account_id":32159,"name":"DellEMC PowerStore CI","email":"dellemc.powerstore.ci@dell.com","username":"dellemc-powerstore-ci"},"date":"2024-09-05 18:24:27.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerStore_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/11/check/DellEMC_PowerStore_Manila_dhss_false/27614c9/DellEMC_PowerStore_Manila_dhss_false/477 : SUCCESS in 33m 03s\n\nLeave a comment with \u0027run-DellEMC PowerStore CI\u0027 to trigger a recheck. For more information about this CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerStore_CI","accounts_in_message":[],"_revision_number":11},{"id":"b7562c27f99df6d856066bc8de2c90ab25dcd294","author":{"_account_id":36171,"name":"jayaanand borra","display_name":"jayaanand borra","email":"jayaanand.borra@netapp.com","username":"jayaanan","status":"netapp"},"date":"2024-09-05 18:29:09.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"7232256718c7345150f33d9c61c3bf11ff816bb6","author":{"_account_id":36178,"name":"Saravanan Manickam","display_name":"msaravan","email":"manicsaran@gmail.com","username":"msaravan"},"date":"2024-09-05 18:32:00.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5a6636a21bf7e961f09334dfc2d413fe402dbaa7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2024-09-05 18:47:41.000000000","message":"Patch Set 11:\n\nBuild succeed.\n\n- DellEMC_PowerFlex_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/11/check/DellEMC_PowerFlex_Manila_dhss_false/0d269a7/DellEMC_PowerFlex_Manila_dhss_false/1639 : SUCCESS in 33m 39s\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":11},{"id":"36485ae53dbdc0bd471edf1133528a5b6cd86bc7","author":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"},"date":"2024-09-05 18:59:35.000000000","message":"Patch Set 11: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":11},{"id":"ea6526b4e6b66549df97932985ec0c8426db0b93","author":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"},"date":"2024-09-05 19:02:04.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d53e55cb4c20b5d291fb2376017754663911649d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 19:02:34.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"7de0bc4ca8beb7f8d28a6558ac96778202489300","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 20:07:47.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30125dc8a0604048b834567d0001f291\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b75b22ae1baf40569edf5779a1092c7d : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c36cf35d8b5f43cd91a779e253cf293f : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52e8ee89fb2c4936a9160a7b1c3b89ab : SUCCESS in 9m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c9f19f6fcd0947c3a5317629b6adeea5 : SUCCESS in 9m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83afb74d54354db8b4a86a654afd8a2d : SUCCESS in 9m 46s\n- manila-tempest-plugin-lvm https://zuul.opendev.org/t/openstack/build/4a1fd845fd2346f293b45dce0315bf1e : SUCCESS in 1h 03m 51s\n- manila-tempest-plugin-dummy-no-dhss https://zuul.opendev.org/t/openstack/build/c5b5aa800be8493bb720689058a0af41 : SUCCESS in 27m 43s\n- manila-tempest-plugin-dummy-dhss https://zuul.opendev.org/t/openstack/build/f31e80e448e74f45b1adb6e388495832 : SUCCESS in 30m 21s","accounts_in_message":[],"_revision_number":11},{"id":"ca816de98bfd13cc4ba253702bf6cddf8897c79e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 20:07:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"3e6598b86460dba619e5d38525e80456d152bde8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 20:08:55.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1a74ad3173c43ff9f8d2005486fd3e5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c9f6693a1c84a07aa81ee5e8a7f8e7a : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41c9777f738d42b6ac76ac839c1db8b7 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":11},{"id":"0ba51a45d88afb347dec158e5e1af951de8649a9","author":{"_account_id":21884,"name":"DellEMC Unity CI","email":"emc.unity.ci@emc.com","username":"emc-unity-ci"},"date":"2024-09-06 00:14:59.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- DellEMC_Unity_Manila_dhss_true https://elab-os-logsrv.delllabs.net/49/926049/11/check/DellEMC_Unity_Manila_dhss_true/b22cca1/DellEMC_Unity_Manila_dhss_true/1638 : FAILURE in 13m 30s\n- DellEMC_Unity_Manila_dhss_false https://elab-os-logsrv.delllabs.net/49/926049/11/check/DellEMC_Unity_Manila_dhss_false/9bb41bc/DellEMC_Unity_Manila_dhss_false/1746 : SUCCESS in 1h 30m 44s\n\nLeave a comment with \u0027run-DellEMC Unity CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Unity_CI","accounts_in_message":[],"_revision_number":11},{"id":"ab44a37d099995eb0caaeabc8d5333054b83f80f","author":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"date":"2024-09-09 13:23:39.000000000","message":"Patch Set 11:\n\n(4 comments)","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"0892de04f67cc34bb205151238fedb0e3908b99a","revisions":{"2c9674f91b55c474fe0afb5ea38903f66399211d":{"kind":"REWORK","_number":1,"created":"2024-08-09 10:36:19.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/49/926049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/926049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/926049/1"}}},"commit":{"parents":[{"commit":"14ef89161130a41bdc6e916de4fab9b08ae439f6","subject":"update requirements for VAST manila driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/14ef89161130a41bdc6e916de4fab9b08ae439f6"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-08-09 10:05:22.000000000","tz":330},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-08-09 10:30:43.000000000","tz":330},"subject":"[NetApp] SnapLock support for NetApp driver","message":"[NetApp] SnapLock support for NetApp driver\n\nThe NetApp ONTAP driver now supports the SnapLock feature. A new\nextra spec, \u0027netapp:snaplock_type\u0027, has been added to create SnapLock\nvolumes. Additionally, extra specs related to the SnapLock retention\nperiod have been introduced to set the retention period on SnapLock\nvolumes. Along with this, the \u0027worm_support_type\u0027 capability has been\nadded to the backend pool to schedule shares on the required\nWORM-supported pool.\n\nDepends-On: Ibd8fca324020a6e2efc7f0bea7dde83e4572f5c7\nChange-Id: I73c4d027852fbfbbdab59538cbe23ba1487da0de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2c9674f91b55c474fe0afb5ea38903f66399211d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2c9674f91b55c474fe0afb5ea38903f66399211d"}]},"branch":"refs/heads/master"},"a934318aee3060983cf398ef38bc5adcb30d30cd":{"kind":"REWORK","_number":2,"created":"2024-08-13 12:32:46.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/49/926049/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/926049/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/926049/2"}}},"commit":{"parents":[{"commit":"14ef89161130a41bdc6e916de4fab9b08ae439f6","subject":"update requirements for VAST manila driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/14ef89161130a41bdc6e916de4fab9b08ae439f6"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-08-09 10:05:22.000000000","tz":330},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-08-13 12:32:04.000000000","tz":330},"subject":"[NetApp] SnapLock support for NetApp driver","message":"[NetApp] SnapLock support for NetApp driver\n\nThe NetApp ONTAP driver now supports the SnapLock feature. A new\nextra spec, \u0027netapp:snaplock_type\u0027, has been added to create SnapLock\nvolumes. Additionally, extra specs related to the SnapLock retention\nperiod have been introduced to set the retention period on SnapLock\nvolumes. Along with this, the \u0027worm_support_type\u0027 capability has been\nadded to the backend pool to schedule shares on the required\nWORM-supported pool.\n\nDepends-On: Ibd8fca324020a6e2efc7f0bea7dde83e4572f5c7\nChange-Id: I73c4d027852fbfbbdab59538cbe23ba1487da0de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a934318aee3060983cf398ef38bc5adcb30d30cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/a934318aee3060983cf398ef38bc5adcb30d30cd"}]},"branch":"refs/heads/master"},"f1fb6655510f6e3633754d225ecec950cfc098bc":{"kind":"REWORK","_number":3,"created":"2024-08-16 16:58:06.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/49/926049/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/926049/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/926049/3"}}},"commit":{"parents":[{"commit":"14ef89161130a41bdc6e916de4fab9b08ae439f6","subject":"update requirements for VAST manila driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/14ef89161130a41bdc6e916de4fab9b08ae439f6"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-08-09 10:05:22.000000000","tz":330},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-08-16 16:56:38.000000000","tz":330},"subject":"[NetApp] SnapLock support for NetApp driver","message":"[NetApp] SnapLock support for NetApp driver\n\nThe NetApp ONTAP driver now supports the SnapLock feature. A new\nextra spec, \u0027netapp:snaplock_type\u0027, has been added to create SnapLock\nvolumes. Additionally, extra specs related to the SnapLock retention\nperiod have been introduced to set the retention period on SnapLock\nvolumes. Along with this, the \u0027worm_support_type\u0027 capability has been\nadded to the backend pool to schedule shares on the required\nWORM-supported pool.\n\nImplements: blueprint snaplock-support-for-manila-driver\nDepends-On: Ibd8fca324020a6e2efc7f0bea7dde83e4572f5c7\nChange-Id: I73c4d027852fbfbbdab59538cbe23ba1487da0de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f1fb6655510f6e3633754d225ecec950cfc098bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/f1fb6655510f6e3633754d225ecec950cfc098bc"}]},"branch":"refs/heads/master"},"bf4b614b19f4d97e84f206c0a05e08254ce21500":{"kind":"REWORK","_number":4,"created":"2024-08-17 10:59:38.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/49/926049/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/926049/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/926049/4"}}},"commit":{"parents":[{"commit":"14ef89161130a41bdc6e916de4fab9b08ae439f6","subject":"update requirements for VAST manila driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/14ef89161130a41bdc6e916de4fab9b08ae439f6"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-08-09 10:05:22.000000000","tz":330},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-08-17 10:57:28.000000000","tz":330},"subject":"[NetApp] SnapLock support for NetApp driver","message":"[NetApp] SnapLock support for NetApp driver\n\nThe NetApp ONTAP driver now supports the SnapLock feature. A new\nextra spec, \u0027netapp:snaplock_type\u0027, has been added to create SnapLock\nvolumes. Additionally, extra specs related to the SnapLock retention\nperiod have been introduced to set the retention period on SnapLock\nvolumes. Along with this, the \u0027worm_support_type\u0027 capability has been\nadded to the backend pool to schedule shares on the required\nWORM-supported pool.\n\nImplements: blueprint snaplock-support-for-manila-driver\nDepends-On: Ibd8fca324020a6e2efc7f0bea7dde83e4572f5c7\nChange-Id: I73c4d027852fbfbbdab59538cbe23ba1487da0de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bf4b614b19f4d97e84f206c0a05e08254ce21500"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/bf4b614b19f4d97e84f206c0a05e08254ce21500"}]},"branch":"refs/heads/master"},"5f79a3e53a4a0a9aaacb63991cf888e7389d8613":{"kind":"REWORK","_number":5,"created":"2024-08-20 12:56:32.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/49/926049/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/926049/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/926049/5"}}},"commit":{"parents":[{"commit":"14ef89161130a41bdc6e916de4fab9b08ae439f6","subject":"update requirements for VAST manila driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/14ef89161130a41bdc6e916de4fab9b08ae439f6"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-08-09 10:05:22.000000000","tz":330},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-08-20 12:55:37.000000000","tz":330},"subject":"[NetApp] SnapLock support for NetApp driver","message":"[NetApp] SnapLock support for NetApp driver\n\nThe NetApp ONTAP driver now supports the SnapLock feature. A new\nextra spec, \u0027netapp:snaplock_type\u0027, has been added to create SnapLock\nvolumes. Additionally, extra specs related to the SnapLock retention\nperiod have been introduced to set the retention period on SnapLock\nvolumes. Along with this, the \u0027worm_support_type\u0027 capability has been\nadded to the backend pool to schedule shares on the required\nWORM-supported pool.\n\nImplements: blueprint snaplock-support-for-manila-driver\nDepends-On: Ibd8fca324020a6e2efc7f0bea7dde83e4572f5c7\nChange-Id: I73c4d027852fbfbbdab59538cbe23ba1487da0de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5f79a3e53a4a0a9aaacb63991cf888e7389d8613"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/5f79a3e53a4a0a9aaacb63991cf888e7389d8613"}]},"branch":"refs/heads/master"},"fa86d2267ec5d8603c43eacfdc96b01fe392ed6b":{"kind":"REWORK","_number":6,"created":"2024-08-29 11:28:16.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/49/926049/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/926049/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/926049/6"}}},"commit":{"parents":[{"commit":"2d987cf5859d9db72827b810c2fc3006743d9c1c","subject":"Merge \"NetApp: don\u0027t check kerberos config on mgmt LIF.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2d987cf5859d9db72827b810c2fc3006743d9c1c"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-08-09 10:05:22.000000000","tz":330},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-08-29 11:27:51.000000000","tz":330},"subject":"[NetApp] SnapLock support for NetApp driver","message":"[NetApp] SnapLock support for NetApp driver\n\nThe NetApp ONTAP driver now supports the SnapLock feature. A new\nextra spec, \u0027netapp:snaplock_type\u0027, has been added to create SnapLock\nvolumes. Additionally, extra specs related to the SnapLock retention\nperiod have been introduced to set the retention period on SnapLock\nvolumes. Along with this, the \u0027worm_support_type\u0027 capability has been\nadded to the backend pool to schedule shares on the required\nWORM-supported pool.\n\nImplements: blueprint snaplock-support-for-manila-driver\nChange-Id: I73c4d027852fbfbbdab59538cbe23ba1487da0de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fa86d2267ec5d8603c43eacfdc96b01fe392ed6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/fa86d2267ec5d8603c43eacfdc96b01fe392ed6b"}]},"branch":"refs/heads/master"},"7d74197fec347dc9a22a6c700a41168bbe8bc50e":{"kind":"REWORK","_number":7,"created":"2024-09-03 09:34:16.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/49/926049/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/926049/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/926049/7"}}},"commit":{"parents":[{"commit":"2d987cf5859d9db72827b810c2fc3006743d9c1c","subject":"Merge \"NetApp: don\u0027t check kerberos config on mgmt LIF.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2d987cf5859d9db72827b810c2fc3006743d9c1c"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-08-09 10:05:22.000000000","tz":330},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-09-03 09:32:30.000000000","tz":330},"subject":"[NetApp] SnapLock support for NetApp driver","message":"[NetApp] SnapLock support for NetApp driver\n\nThe NetApp ONTAP driver now supports the SnapLock feature. A new\nextra spec, \u0027netapp:snaplock_type\u0027, has been added to create SnapLock\nvolumes. Additionally, extra specs related to the SnapLock retention\nperiod have been introduced to set the retention period on SnapLock\nvolumes. Along with this, the \u0027netapp_snaplock_type\u0027 capability has been\nadded to the backend pool to schedule shares on the required\nWORM-supported pool.\n\nImplements: blueprint snaplock-support-for-manila-driver\nChange-Id: I73c4d027852fbfbbdab59538cbe23ba1487da0de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7d74197fec347dc9a22a6c700a41168bbe8bc50e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/7d74197fec347dc9a22a6c700a41168bbe8bc50e"}]},"branch":"refs/heads/master"},"4a26064383cc0a563010a787d4b73335c9b59cb8":{"kind":"REWORK","_number":8,"created":"2024-09-03 11:26:41.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/49/926049/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/926049/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/926049/8"}}},"commit":{"parents":[{"commit":"2d987cf5859d9db72827b810c2fc3006743d9c1c","subject":"Merge \"NetApp: don\u0027t check kerberos config on mgmt LIF.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2d987cf5859d9db72827b810c2fc3006743d9c1c"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-08-09 10:05:22.000000000","tz":330},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-09-03 11:26:15.000000000","tz":330},"subject":"[NetApp] SnapLock support for NetApp driver","message":"[NetApp] SnapLock support for NetApp driver\n\nThe NetApp ONTAP driver now supports the SnapLock feature. A new\nextra spec, \u0027netapp:snaplock_type\u0027, has been added to create SnapLock\nvolumes. Additionally, extra specs related to the SnapLock retention\nperiod have been introduced to set the retention period on SnapLock\nvolumes. Along with this, the \u0027netapp_snaplock_type\u0027 capability has been\nadded to the backend pool to schedule shares on the required\nWORM-supported pool.\n\nImplements: blueprint snaplock-support-for-manila-driver\nChange-Id: I73c4d027852fbfbbdab59538cbe23ba1487da0de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4a26064383cc0a563010a787d4b73335c9b59cb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/4a26064383cc0a563010a787d4b73335c9b59cb8"}]},"branch":"refs/heads/master"},"2db360d39a14b6f56a3fcda824d0c0d6d4eb1b1a":{"kind":"REWORK","_number":9,"created":"2024-09-03 11:30:12.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/49/926049/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/926049/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/926049/9"}}},"commit":{"parents":[{"commit":"2d987cf5859d9db72827b810c2fc3006743d9c1c","subject":"Merge \"NetApp: don\u0027t check kerberos config on mgmt LIF.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2d987cf5859d9db72827b810c2fc3006743d9c1c"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-08-09 10:05:22.000000000","tz":330},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-09-03 11:29:56.000000000","tz":330},"subject":"[NetApp] SnapLock support for NetApp driver","message":"[NetApp] SnapLock support for NetApp driver\n\nThe NetApp ONTAP driver now supports the SnapLock feature. A new\nextra spec, \u0027netapp:snaplock_type\u0027, has been added to create SnapLock\nvolumes. Additionally, extra specs related to the SnapLock retention\nperiod have been introduced to set the retention period on SnapLock\nvolumes. Along with this, the \u0027netapp_snaplock_type\u0027 capability has been\nadded to the backend pool to schedule shares on the required\nWORM-supported pool.\n\nImplements: blueprint snaplock-support-for-manila-driver\nChange-Id: I73c4d027852fbfbbdab59538cbe23ba1487da0de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2db360d39a14b6f56a3fcda824d0c0d6d4eb1b1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/2db360d39a14b6f56a3fcda824d0c0d6d4eb1b1a"}]},"branch":"refs/heads/master"},"c7b9c6b5a48a9de1d62c3f4f1e4bf814ea9ebf3f":{"kind":"REWORK","_number":10,"created":"2024-09-05 14:33:15.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/49/926049/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/926049/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/926049/10"}}},"commit":{"parents":[{"commit":"d643c2601d17a51f636e94ab7a9566860faae967","subject":"Merge \"[NetApp] Custom storage efficiency policy support for Manila share\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d643c2601d17a51f636e94ab7a9566860faae967"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-08-09 10:05:22.000000000","tz":330},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-09-05 14:06:01.000000000","tz":330},"subject":"[NetApp] SnapLock support for NetApp driver","message":"[NetApp] SnapLock support for NetApp driver\n\nThe NetApp ONTAP driver now supports the SnapLock feature. New\ncapability \u0027netapp_snaplock_type\u0027 has been added to the backend pool\nto create and schedule shares on the required WORM-supported pool.\nExtra spec, \u0027netapp_snaplock_type\u0027, can be used to create SnapLock\nvolumes. Additionally, extra specs related to the SnapLock retention\nperiod have been introduced to set the retention period on SnapLock\nvolumes.\n\nImplements: blueprint snaplock-support-for-manila-driver\nChange-Id: I73c4d027852fbfbbdab59538cbe23ba1487da0de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c7b9c6b5a48a9de1d62c3f4f1e4bf814ea9ebf3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/c7b9c6b5a48a9de1d62c3f4f1e4bf814ea9ebf3f"}]},"branch":"refs/heads/master"},"0892de04f67cc34bb205151238fedb0e3908b99a":{"kind":"REWORK","_number":11,"created":"2024-09-05 14:56:38.000000000","uploader":{"_account_id":36180,"name":"Gireesh Awasthi","display_name":"Gireesh","email":"gawasthi2010@gmail.com","username":"agireesh","status":"NetApp"},"ref":"refs/changes/49/926049/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/manila","ref":"refs/changes/49/926049/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/manila refs/changes/49/926049/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/manila refs/changes/49/926049/11"}}},"commit":{"parents":[{"commit":"d643c2601d17a51f636e94ab7a9566860faae967","subject":"Merge \"[NetApp] Custom storage efficiency policy support for Manila share\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/d643c2601d17a51f636e94ab7a9566860faae967"}]}],"author":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-08-09 10:05:22.000000000","tz":330},"committer":{"name":"agireesh","email":"gawasthi2010@gmail.com","date":"2024-09-05 14:56:15.000000000","tz":330},"subject":"[NetApp] SnapLock support for NetApp driver","message":"[NetApp] SnapLock support for NetApp driver\n\nThe NetApp ONTAP driver now supports the SnapLock feature. New\ncapability \u0027netapp_snaplock_type\u0027 has been added to the backend pool\nto create and schedule shares on the required WORM-supported pool.\nExtra spec, \u0027netapp_snaplock_type\u0027, can be used to create SnapLock\nvolumes. Additionally, extra specs related to the SnapLock retention\nperiod have been introduced to set the retention period on SnapLock\nvolumes.\n\nImplements: blueprint snaplock-support-for-manila-driver\nChange-Id: I73c4d027852fbfbbdab59538cbe23ba1487da0de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0892de04f67cc34bb205151238fedb0e3908b99a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/manila/commit/0892de04f67cc34bb205151238fedb0e3908b99a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16643,"name":"Goutham Pacha Ravi","email":"gouthampravi@gmail.com","username":"gouthamr"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29632,"name":"Carlos Eduardo","email":"ces.eduardo98@gmail.com","username":"silvacarlos"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
